)]}'
{"id":"openstack%2Fcinder~881233","triplet_id":"openstack%2Fcinder~master~I98403fe9f8e5f28fa43724b167ed67987b456647","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-04-21 15:48:19.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I98403fe9f8e5f28fa43724b167ed67987b456647","subject":"WIP: RBD: Log librbd version at startup","status":"NEW","created":"2023-04-21 15:01:52.000000000","updated":"2023-04-24 03:33:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7cdc5e425ab14a4093abb1f8b2db22e67bd78a6d","_number":881233,"virtual_id_number":881233,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-21 16:57:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":-1,"date":"2023-04-21 15:01:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-21 15:29:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 15:48:19.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-21 15:51:21.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 16:08:44.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-04-21 16:20:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-04-21 16:38:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 16:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-21 17:15:00.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-21 17:39:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-21 18:35:04.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-23 00:49:17.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-23 10:00:19.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-04-24 03:33:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"6c3e351dfbd1da8e2d2ab7d9b8b225f2eae4ae56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-21 15:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b23fb59d71d796a3f31f3de7f2c77006112f1cf1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-21 15:01:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"663899d6f7c66f2f6846d64c3e56062bc199163d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 15:29:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcd96f561b554774a265652ff290a862\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a586bbf75e64919ac9f2b52c820b4b2 : FAILURE in 25m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f7684e04ca8a4028a68b6c30c8341728 : FAILURE in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7c6923ff9ec23701f58f4ce90802046417d4b94","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-21 15:48:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/300446cdb81c4281922538a8214edfde : POST_FAILURE in 45m 39s","accounts_in_message":[],"_revision_number":1},{"id":"12751757f67ea7a31187b0aab1444f02d965609d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-21 15:51:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/33/881233/1/ : FAILURE in 41m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e21d0c0c59c47fffb24d66c40661e1a13377b2c0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-21 16:08:44.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/926d4870a33a48b9afb9dc67e0a08f4f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fb2cc4e4089f4d21952593b72ebb1869 : POST_FAILURE in 1h 04m 11s","accounts_in_message":[],"_revision_number":1},{"id":"837b831115bfb98e2202b6646bde3de116f09af5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-21 16:20:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/33/881233/1/check/iscsi-driver-alletra-6k/ced3590/ced3590.html : FAILURE in 1h 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7a3d74ad89353499096f358f558965e521c1a25f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-21 16:38:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881233 : SUCCESS in 1h 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"de8282e3ffdb027aff3ab84f7ffe4804ef327cd0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-21 16:39:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21f603263efe14aeca8be7f9830a1c4dc59bc1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 16:57:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a5d5ef833b3426a9f89ce3e427af369\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d48fe5c0e44ae38af419ba96083809 : SUCCESS in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77430ab4fcfd4ce0ab86e07350bfc099 : FAILURE in 12m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e31ff12e73e4a9d8b0269f563b5e809 : FAILURE in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2fe9b3f70e4b6486d363c8a46e38a0 : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/fa447728b6d84a3ea6912dea90e67e7f : SUCCESS in 53m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5dd5aa6964764ec7a10f5c36bc383476 : SUCCESS in 1h 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca75f4f665674d269b3dbccb229113b8 : SUCCESS in 1h 01m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e43f84efb124e0b95caa71872f84878 : FAILURE in 11m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/325caaaaf95e4bd8a97ff33470999f43 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a55f703fe2543e39800cb425f71e8f8 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/564aa02b117f4612952c9a4a094de525 : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5950b882f4b4e10920a8c2529bc0e97 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba2d06ba698b4c92949152bff9c199c4 : SUCCESS in 34m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a9ef4605cd2441888e3545000b3fc99 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e871e03b3f2a4a4a91525d4a58b54bd7 : SUCCESS in 1h 32m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/906cdef65b5b4dcea7906a32e06972a8 : FAILURE in 1h 14m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9a7510fcabc3493097900fcbaebad43f : SUCCESS in 1h 26m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5e367b794c74430d9c1bb67b70d51040 : SUCCESS in 1h 48m 51s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/51b2ba704119447398ad295b19d37a6a : SUCCESS in 1h 10m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ba132ce35e74a3998d248489794b391 : SUCCESS in 1h 30m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fee96f5609f147e691e54410b7bd124e : SUCCESS in 40m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c18f6f25d46a4c21b60b61cbda7d2070 : SUCCESS in 1h 04m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fef22c09acd640d78652fd00964ce41c : SUCCESS in 1h 09m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b24aaae3a9c47318ffa8c956f3d5b8f : SUCCESS in 1h 49m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ec35f7b30464cfcac30b582f3468a4b : SUCCESS in 55m 32s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/113efac7cb83425da7e2a5222e899991 : SUCCESS in 1h 23m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d60cbbdec8b8ba6af4c6857644e0983e8d709892","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-21 17:15:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-881233-1/FC/ : SUCCESS in 2h 12m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-881233-1/iSCSI/ : SUCCESS in 1h 46m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e4333896e9b807a05c8b562f88d994fb811b313","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-21 17:28:16.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/79155453adc046aca62ae025f6e8f72d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4d43c4af86df442b9468844f2b79a16e : POST_FAILURE in 46m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bd43b34328eb9461e8fb6aa1298824a994ec430f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-21 17:39:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/33/881233/1/check/DellEMC_Unity_FC/4dcc4a5/DellEMC_Unity_FC/59 : SUCCESS in 1h 31m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/33/881233/1/check/DellEMC_Unity_iSCSI/0924636/DellEMC_Unity_iSCSI/61 : SUCCESS in 2h 04m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"3599d10128439c48cb4ad31f2a6e4d3ff5e0b04c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-21 18:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-33-881233-1/ns5_nfs : SUCCESS in 3h 32m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-33-881233-1/ns5_iscsi : SUCCESS in 3h 29m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7c23920ca6b230251f64d4ef1572085d32e2e234","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-23 00:49:17.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1e7d1f3685844b1fab9162998f5bfd38 : RETRY_LIMIT in 1m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/aa186b8ea57046de8103e729ecbfb455 : RETRY_LIMIT in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1f71ee18e9a6b0b768ae7bd9391dc04315622cd5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-23 10:00:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/33/881233/1/20230423131521/NEC-V-ISCSI : SUCCESS in 2h 17m 14s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/33/881233/1/20230423131521/NEC-V-FC : SUCCESS in 5h 44m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7cdc5e425ab14a4093abb1f8b2db22e67bd78a6d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-24 03:33:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/33/881233/1/20230424084321/NEC-ISCSI : SUCCESS in 3h 16m 11s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/33/881233/1/20230424084321/NEC-FC : SUCCESS in 3h 50m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"226bf0678fe5972060be1f173c5b7f2ea8d68672","revisions":{"226bf0678fe5972060be1f173c5b7f2ea8d68672":{"kind":"REWORK","_number":1,"created":"2023-04-21 15:01:52.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/33/881233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/881233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/881233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/881233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/881233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/881233/1"}}},"commit":{"parents":[{"commit":"2e8aff660b18d1e8f961d6de979d541a8d06f1c1","subject":"Merge \"Add fips check jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e8aff660b18d1e8f961d6de979d541a8d06f1c1"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 14:51:22.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 15:01:28.000000000","tz":-240},"subject":"WIP: RBD: Log librbd version at startup","message":"WIP: RBD: Log librbd version at startup\n\nTODO: figure out unit tests\n\nLog the version of librbd at RBD startup, to\nassist with field debugging.\n\nChange-Id: I98403fe9f8e5f28fa43724b167ed67987b456647\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/226bf0678fe5972060be1f173c5b7f2ea8d68672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/226bf0678fe5972060be1f173c5b7f2ea8d68672"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
