)]}'
{"id":"openstack%2Fcinder~881470","triplet_id":"openstack%2Fcinder~master~I697e8c685d67250e2d016964c20bb2189ac384a4","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-07-17 23:55:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I697e8c685d67250e2d016964c20bb2189ac384a4","subject":"Cleanup image_utils chown check","status":"MERGED","created":"2023-04-25 12:07:46.000000000","updated":"2023-07-17 23:56:21.000000000","submitted":"2023-07-17 23:55:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881470","meta_rev_id":"7a22834110e15e232ec0fa0c95188d5cb4e7ba21","_number":881470,"virtual_id_number":881470,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 23:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-07-15 12:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-07-11 18:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2023-07-17 20:56:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-12 07:07:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":1,"date":"2023-07-17 20:56:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-25 12:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-25 12:24:57.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-25 12:25:03.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-25 12:38:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-25 12:42:22.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-04-25 12:46:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-25 13:17:05.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-04-25 14:58:25.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-04-25 15:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-25 15:07:50.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-25 17:18:27.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-25 17:23:37.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2023-04-25 21:24:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-26 18:30:14.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-18 16:00:29.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-18 16:25:07.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-24 13:50:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-06-21 04:13:00.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-04 11:02:07.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2023-07-06 12:54:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-11 20:57:15.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-07-11 22:11:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-11 22:13:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-07-12 02:15:00.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-12 07:07:31.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-07-12 08:01:09.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-07-12 10:46:52.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-07-14 09:25:44.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2023-07-14 11:11:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-15 12:10:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-07-17 20:56:51.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"155e161a79a23d8ac1d0cc1127a58dc0a0059d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-25 12:07:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ae7671f0d6a4bb32120db53a53ee09c10e0c4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:18:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73285e8fef4b4767bad211e6a12c7a00\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/050e39f7a7d14923830366a1eebce571 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59286607948b4ff7bdb7570dc1a5916e : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ea1e9bc76f77bc0cb95fae39d1f231d759a8051","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-25 12:24:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/111150c17ca24465a51b34410064a09d : POST_FAILURE in 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"37cbfcb8f3a399ba8eeff19950113a92814cd7e9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-25 12:25:03.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-881470-1/ns5_nfs : FAILURE in 16m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-881470-1/ns5_iscsi : FAILURE in 16m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f17fd148038289918fb63d7ccd530802b63fb791","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-25 12:38:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/881470/1/check/DellEMC_Unity_FC/09fcd24/DellEMC_Unity_FC/66 : FAILURE in 29m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/881470/1/check/DellEMC_Unity_iSCSI/21f4a97/DellEMC_Unity_iSCSI/69 : FAILURE in 12m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8f9ad80bbd1eaa08e28ac383e5a33ad34524d61c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-25 12:42:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/881470/1/check/iscsi-driver-alletra-6k/29c318b/29c318b.html : FAILURE in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1fc5aefdfb3f181e8e1695f0fa1c06081f6162ea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-25 12:46:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881470 : FAILURE in 36m 55s","accounts_in_message":[],"_revision_number":1},{"id":"74eb76770fd48237a9c3a636d8051b7b9599ef28","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-25 13:17:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/19a794af8e204ddbb95c28a1ed9d08e2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fd641dd276054da1bfd23987a13af6c1 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":1},{"id":"093255bdd74f24ea5b19c7bbc09bb73b96e0a04c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-25 14:58:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/1/20230425210749/NEC-V-ISCSI : SUCCESS in 2h 22m 23s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/1/20230425210749/NEC-V-FC : SUCCESS in 2h 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"03b483b23399f7dccf4c41b07a2564c2fff8d0b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 15:02:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53d6eaaf60fc498eb9a778fae28ce963\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfad076755ce4936915792731dbc36d8 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67964978cca4afb9f5ad6ad56269529 : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff4e6ba01f474870be87bfeb93221dc9 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465c619ba1f947fd91e33cf470043662 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/994dea9ac5fd425ab9c62c0fc4f6353b : FAILURE in 17m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50849222086b446bae111840f46d66ec : SUCCESS in 1h 45m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a08320a8a6bf490d92be9aa532a655f7 : POST_FAILURE in 16m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a728612ddc054cdeb1b741750c41284a : SUCCESS in 16m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2304d346f98e4092822e4fd75cc4ab49 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b09741a9a12461286d52ca3e77aabd4 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f99a09c003f43fbb988be99898c8c58 : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d6795073ede948a9929bd0f11effb87e : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f431f1070af647ae967fef9bd8a3d509 : SUCCESS in 38m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d0b30069c90f4fa7886d69688cba71f8 : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e127a5093f1c43ae87bf7d10bcb4e5b8 : POST_FAILURE in 19m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a14ce2add47449c84bf904b35f2fc53 : POST_FAILURE in 18m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f3057b3fab44c2e876ad660f838c0f6 : SUCCESS in 1h 52m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a9fefd53d2c446268ae0fddc6f759ab1 : FAILURE in 2h 07m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c5a826fe362848bfb8735f83dd14ce8b : SUCCESS in 1h 02m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bc55daa4f7d14f389a6417d7df4bfcc1 : FAILURE in 17m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/944355a0526942d9a6a857ebb9f2dc89 : SUCCESS in 28m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c7e3ea90d694c7296a92526cb5f4915 : FAILURE in 16m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/78d74173e2bf4566a5e1bd10de44a357 : FAILURE in 13m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f9fe79128eb41c3816cacd9daa0c451 : SUCCESS in 2h 48m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb8e907e3e3a4c5f87a6f6dc26131e3b : FAILURE in 19m 54s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4efef2503c9245d79a5e5410381f6a00 : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a0da4183af617b5be893def8e5e0078a580335e1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-25 15:07:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/881470/1/ : POST_FAILURE in 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3da2a11622f16b1a355d518a53bc0e038e1aea35","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-25 17:18:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/1/20230425225642/NEC-ISCSI : SUCCESS in 3h 15m 15s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/1/20230425225642/NEC-FC : SUCCESS in 3h 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a1975e05aa917979f8b582f880f9282121c54e81","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-25 17:23:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55994564eeb829eb36de16bcec6bf3597ef8868f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-25 21:24:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-1/FC/ : SUCCESS in 2h 6m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-1/iSCSI/ : SUCCESS in 2h 41m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c51f910952bfa7f6fa80f7ddeb087c52c93d6904","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-26 18:30:14.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/0f7467ba4b884c5bbf9c30d37d185d2e : RETRY_LIMIT in 1m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2e0e1728d88c4d5292743f99c37695e5 : RETRY_LIMIT in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7178f655969c8952d7bc4a3ddcc5ae9393181ec4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-10 09:28:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"471c8a93e6ad4ee2a9ca3b69461408d78b8437aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-05-18 15:58:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"10947309717dc56a74bab4effdaebe9aa5a6d538","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-18 16:00:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2338ed5563474833bde1ab0218d406a6 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7b5cc98dcc38ab8563e4c0fca6ab1e3af73f5955","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 16:11:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4785bd8fae3c4b6fb99b7a735195f6eb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/323f24ea53b64654ae8c74eac3e5cbe5 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2beebb82c0904096b23c58930a5eb9ba : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7460df861f74bd997353637477853f5 : SUCCESS in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b60665e3814b4b7c5f0d330bb614b158071f1c0d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-18 16:25:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/881470/2/check/fujitsu-eternus-dx-iscsi/6a20cb2 : FAILURE in 25m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/881470/2/check/fujitsu-eternus-dx-fc/d9c449c : FAILURE in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"52cedd75bda5ae9d128b19d419022bfcc7ba0e8f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-18 16:47:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/9abf4b360cda41b68d834babf67e1310 : POST_FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e3fa6aa6e3956d8854d15b0fb53ae845755cdf4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-18 17:04:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881470 : FAILURE in 1h 03m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5b28e0982e5da0b78fcbbac7a59446b4b170b7f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-18 17:35:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/7098bfab7c88442e96720fcfa2d58ac6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ef249f909141466386cd3311e13480a3 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d14e85fcfcab3a00757d43ab6535782fad3db05d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-18 18:38:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-2/FC/ : SUCCESS in 2h 39m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-2/iSCSI/ : SUCCESS in 1h 46m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"9db2717a62542e2d59a494750457816c573c60c4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-18 18:41:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/881470/2/check/iscsi-driver-alletra-6k/1dc7cfe/1dc7cfe.html : FAILURE in 1h 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b79ab605ebb86769cd1da8397ddfb027fe3d3538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 18:42:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ecf86593694b838173ce1319485f1d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89c903b437f46759a374b0fdf487c67 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb22ed39356489f9023822bfcb20ad8 : SUCCESS in 13m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f11b80fdd1f4c48a8bdc0ad3911e52b : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce9a4bb965ed4f539d08581eb04fd0ee : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7216c317069a4b80b12ebf57c3216890 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/e1daf25c80ef4008bc18449a01e5875e : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/679e3db064904d2e88bc935f4a8884f2 : FAILURE in 1h 16m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a53001ba00341e4af9f418d86e2a8bf : SUCCESS in 55m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d3b568c4f8614b79a0e005644dfcb518 : SUCCESS in 15m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/137414b28b1c499f97affef0508d7c55 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c35cae8480584a5a8173964c014cccbd : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51e55ceedc7e46c4809057b33cb7d123 : SUCCESS in 5m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d0be866e3b14b85ac3be7eafea6ce45 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/24620cb7e6ff4499a3a02fea618416fb : SUCCESS in 35m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6aba881c5545440da20ab48b289cb4ba : SUCCESS in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7f1290bb4f324ec0898e4f5fb0771083 : SUCCESS in 1h 55m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/971b26d0f7c14f118ec9d4ea4cc97303 : FAILURE in 2h 04m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19eb4bf8d31242f0b726e21ae7b3760c : SUCCESS in 1h 37m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/28f83fe8a34d4db890537728867a6c30 : POST_FAILURE in 18m 26s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/967c891a61d74885a3790a4d6c3f6327 : SUCCESS in 52m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c6336c5fcdde40bd8354125cc8c3e9c2 : SUCCESS in 1h 32m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b6f6f27b2d404ba194aa47119f3e87fa : SUCCESS in 39m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a048571ca76b421a993472edd71c4510 : SUCCESS in 1h 19m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/231860099a914740a180936ce2ff1f19 : FAILURE in 18m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f484f9e63da409fbf71aa385922f0fc : SUCCESS in 2h 36m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de89c08731f44517aafa900160ae5697 : SUCCESS in 42m 06s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1adcb2801c3a43ca928c48d8971ee501 : SUCCESS in 1h 16m 08s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/c4e7e82a13ba4293a9aad42ff9125528 : SUCCESS in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a55eaf333d96b84288889afff06642c2ddb46e1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-18 19:46:37.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-881470-2/ns5_nfs : SUCCESS in 3h 47m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-881470-2/ns5_iscsi : SUCCESS in 3h 45m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"285c4f24ad8696d6d3a19a53b246263d817c9e35","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-18 21:04:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/881470/2/check/DellEMC_Unity_FC/fdaf694/DellEMC_Unity_FC/158 : SUCCESS in 1h 53m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/881470/2/check/DellEMC_Unity_iSCSI/090516e/DellEMC_Unity_iSCSI/161 : SUCCESS in 2h 21m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5700e21ffaa25fb7764396d246924bd7d0c6fd4e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-19 07:22:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/2/20230519131519/NEC-V-ISCSI : SUCCESS in 2h 33m 05s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/2/20230519131519/NEC-V-FC : SUCCESS in 3h 07m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6041737ac21102d5af4a67c4d97da9453ea22fc9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-19 12:52:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/2/20230519182012/NEC-FC : SUCCESS in 3h 32m 34s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/2/20230519182012/NEC-ISCSI : SUCCESS in 3h 32m 42s","accounts_in_message":[],"_revision_number":2},{"id":"64bab6c54858e37311898d71f6528afd4767ca48","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-20 08:29:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/881470/2/ : RETRY_LIMIT in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"507102eff7abf20ee52d1f5564598086ac1055ce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-20 14:52:50.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/70/881470/2/check/tempest-dsvm-full-focal-py3/5a8759b/ : FAILURE in 1h 22m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/56c2606d02a04c8db2e4f5eaa76436e8 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"593334e785da764919d0846b3839f056436380c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-05-23 17:10:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"08022c268cb6ee8d1653e3ed0dd1292b6f131c5a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-23 17:11:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"f4ab1ec4c3caadb85ae5910c36b55a5d456c58cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:32:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef1b56e818e04265b96ad42c81fdafcb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36cff0648b2f480da916b557eaff034f : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2422e8f3c18d4442a78a204de9b38154 : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d8c145b827148a48f1ab3519ff32c9c : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"342bbdfd4ef6c45d14732dbc443d2cd1e00af48c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-23 17:43:16.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1bc4225cbced4e868abe46faba462a50\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8f299e8110914efda40ce0534d2650cf : FAILURE in 29m 31s","accounts_in_message":[],"_revision_number":3},{"id":"19b403f35698fd8a0358ec777c9e52d023a81fcc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-23 18:30:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8d4a94e105cd48a3882b47b12809656d : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1746206894449ad2b5b7b3f4a04db812a44ef161","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-23 19:03:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/881470/3/check/iscsi-driver-alletra-6k/6c3e135/6c3e135.html : FAILURE in 1h 51m 16s","accounts_in_message":[],"_revision_number":3},{"id":"644a38e7bc5cc21d89b7a5c613a3dbad84da06b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 19:39:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b7b0ec322234232ba1ee868dfd90c57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acfeed34cd604f9ab7b6b2f9520200ac : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1c90c5b406a43ecb9da7f09f558621c : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9df307f56b494b46a589153e594e4c02 : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0703cc3c202c45199d124345e711011d : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315b8e2d19c14fb7866c3ffc78141f9f : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0ede955c1a0b4572be7d8633935e85ef : SUCCESS in 44m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb29630b895049959cbd8804296cb150 : SUCCESS in 1h 23m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0090761015e42bb9f3facd59a1d306b : SUCCESS in 1h 00m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77e1660c8c2b4ab88d883a07106fa96e : SUCCESS in 16m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7927ee2bee294230af0c59ffdeac5884 : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/665f4a4a42db4945a386c9149c487d1d : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/239658d44d9544659da1d5ad0fef83a3 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/71ac8bd76411423791faab16bf495faa : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf5fe28bbc4147ce94f43bb83ed458d6 : SUCCESS in 50m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c58c4f75a09442ea09e2817adf7a24a : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3739e5891c64d3182eecf3b75f83e10 : TIMED_OUT in 2h 03m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e2a285aced8c466d957c7b645a3c674c : FAILURE in 1h 42m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1cea09d04c234329b18fe05066553cb4 : SUCCESS in 1h 54m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e1a3567895244f86a69031a6293f6451 : POST_FAILURE in 9m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1af36f67915745b99f203c7003f614fb : SUCCESS in 1h 01m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85a79cfd074046b588ca7dc783c9f37f : SUCCESS in 1h 27m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/683148244cbe4c979c28339f6baceaa4 : SUCCESS in 40m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19ea32f7c1b34c09b9ad5478407440e6 : SUCCESS in 1h 23m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0e40c3c40f4d4442ba62765127833c27 : FAILURE in 8m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21287e20b9d342d585ef2d6ad2113eb3 : FAILURE in 2h 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d89553e5cdc41869724ca3a1db0f6be : SUCCESS in 50m 02s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7416655617724865886857d0a4fc0ae3 : SUCCESS in 1h 55m 38s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/b98a956fc460436e969a6ee06bfaced2 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b59e92835c913205d733d8bdb53b63437024e197","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-23 19:47:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-3/FC/ : SUCCESS in 2h 36m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-3/iSCSI/ : SUCCESS in 2h 10m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"1cab2fac6ecc5f309f799e42ad36c643e4cda752","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-23 20:45:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/c664215f32b04cb1b441794986c094ed : POST_FAILURE in 50m 28s","accounts_in_message":[],"_revision_number":3},{"id":"18d15175cb2e9038125e0c5863cc9dd4a4122b1d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-23 21:04:53.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-881470-3/ns5_nfs : SUCCESS in 3h 53m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-881470-3/ns5_iscsi : SUCCESS in 3h 49m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"0aed3d32f15109c002623cdce6cc3b3f53aaf005","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-24 06:32:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/881470/3/check/DellEMC_Unity_FC/a159297/DellEMC_Unity_FC/174 : FAILURE in 2h 04m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/881470/3/check/DellEMC_Unity_iSCSI/020e009/DellEMC_Unity_iSCSI/174 : FAILURE in 2h 10m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"67eb6c39167a005b87815985333501dd8dc5d9ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-05-24 13:50:59.000000000","message":"Patch Set 3:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1373/","accounts_in_message":[],"_revision_number":3},{"id":"201255cb24344ec0a54563ee9c44910df72a4da7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-25 02:28:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/3/20230525050052/NEC-V-ISCSI : SUCCESS in 2h 46m 45s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/3/20230525050052/NEC-V-FC : SUCCESS in 6h 27m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f3171cb3bb36118a719d56c31162f7406f23a6e7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-25 05:34:44.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/881470/3/ : RETRY_LIMIT in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9233348a4e1202d6bd19eebed3a423bb5ab966af","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-25 11:00:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/3/20230525121353/NEC-ISCSI : SUCCESS in 3h 57m 54s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/3/20230525121353/NEC-FC : SUCCESS in 7h 46m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ec4837a825d451658e99ad533cf2a28cd34365a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-05 14:16:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1e95f7712220f8a64aad9e3c8a5794190a774b9e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-05 14:17:11.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"79e49c847a44ad41a478baf75bbfe9843a337070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 14:41:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a57fcbfd5c4d528a17c102adcbe6bc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40d330d932f84d3a800a064a27bfa702 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef9eb026af5a441d9df34aac50bb87a4 : FAILURE in 23m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e647e3caf64b4a619ac47fdbb5258c94 : SUCCESS in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"866b6ca52863853080e7fa8787d0cca62770fb66","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-05 16:20:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/b54ae3c9029d401c92dac7c7925d6ca0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7015097940394673853b590c3566994a : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":4},{"id":"30128a05637bd62895db0512fee5f3396215e396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 17:07:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df88c383709749a2ade8d13625a31882\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6faae7bc48a04186aee6768b008f2bde : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cefc18379e04c92859e963f9cf86643 : SUCCESS in 15m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1cb634ab84f4c94bb6c49adc9b641ba : SUCCESS in 12m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bebda13dbaac4b2f8fa9ef3a7914e09b : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dabac48d1ab64a3187a7df3b8ad5f779 : SUCCESS in 12m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/0a6f25f7d95e46c3a2364f30ea645a55 : SUCCESS in 53m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7862c767d8744fd28a088c3fd145552f : SUCCESS in 1h 23m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87a20a6f230e4dd8935804db7af3f66f : SUCCESS in 1h 04m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/76b95a88251b436f818045cf222a2ce6 : SUCCESS in 15m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64ce4223d4c74bdcb8d343f3ddd29ea9 : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75217d3c4062404d859e6895dced632d : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3fe7f859430472bbacbefd22446c091 : SUCCESS in 6m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cee7e35eada84761b74f1d46ca451539 : SUCCESS in 5m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/03e8ded6dd824c3baf6e1e57d667e754 : SUCCESS in 40m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/724edc8f49a94e0c9d7b90d5d5ec0171 : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f1198a997ce420e90e3cb6983286212 : SUCCESS in 1h 45m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b069c54a5c447a98ac409d5df753ea0 : SUCCESS in 1h 47m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c79bac71c0ae4cab905d78b2a2534e18 : SUCCESS in 2h 04m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b93af357f41a464597274b943a8bfbe2 : SUCCESS in 2h 17m 44s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c4ee0f0b3dc94aa4ace9d0f9875c98b1 : SUCCESS in 2h 02m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a919203e385145e0aa838333a3895985 : SUCCESS in 1h 42m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/840b40dc1d5e4c28aaa843f1eca29ff8 : SUCCESS in 48m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5623ccd720474b18afcf4eeb091f8162 : FAILURE in 2h 29m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6e3f564f175f4e5fabe5ce85c1bb9fd4 : SUCCESS in 1h 35m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4a73e64437c472aa8b06e8dff301100 : FAILURE in 2h 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c423da24eecf4881aa9f1d2f87069dad : SUCCESS in 1h 37m 00s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/143ef8781f15457f8d8dbe8d2d0c02ae : SUCCESS in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"678e6b6a45c0933fd2202585df92870f93c525a3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-05 18:16:16.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-4/FC/ : UNSTABLE in 27m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-4/iSCSI/ : FAILURE in 3h 47m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"5fe8ebf633a50055cfb158ed8abe594d52f7f937","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-05 18:26:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/881470/4/check/iscsi-driver-alletra-6k/3589a92/3589a92.html : FAILURE in 1h 10m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8531eb61536e82d10d73ada9f73bf075e1c74613","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-05 19:56:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881470 : FAILURE in 1h 36m 30s","accounts_in_message":[],"_revision_number":4},{"id":"011ecf49c05361f3355118938bab630ff22368af","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-05 19:57:09.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/881470/4/check/DellEMC_Unity_FC/0d1d9dd/DellEMC_Unity_FC/219 : SUCCESS in 2h 06m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/881470/4/check/DellEMC_Unity_iSCSI/7cefbe7/DellEMC_Unity_iSCSI/219 : SUCCESS in 2h 11m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b7b2e9ec912f57fa103391f7c8fced352bf8a31a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-14 16:19:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/4/20230614213727/NEC-ISCSI : SUCCESS in 3h 22m 19s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/4/20230614213727/NEC-FC : SUCCESS in 3h 42m 16s","accounts_in_message":[],"_revision_number":4},{"id":"1edc4611d21a14290822c11b07f3ec485f12a159","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-14 17:34:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c4f5caff526002b8b2a24ec864047a426fc8b88","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-14 17:35:02.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e1c858e65edcf7003707e0c764782b2305e23ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 17:48:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4130af953ae74ae58602ce28d502f105\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/221285f98dc54c34832a2953f6edebd1 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3296394c891a405bb5b67d3d63b85997 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7d59bf9dcbf45e3953bc61c55c93dce : SUCCESS in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"11638a4478d988c31d1104f78e31b2367e802527","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-14 19:06:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/dc0725eb074f4deabca006c9c66e83fe\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2b218cfa6fee4dffb34efb0502482f9d : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":4},{"id":"c01e95e639fd8766f5cb6b378c7c20181500db28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 20:42:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/169fe2e680144daea2cf8d5290c32de2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0afea3b7c3694f75892a8087df7f25df : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef9a46be6184b0bb1280ca29a716ac8 : SUCCESS in 11m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eec2308d5e048faa58b3bc17ee120a4 : SUCCESS in 13m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b3cbad84a94c259a5eeb4930426ca8 : SUCCESS in 14m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595500581902483693f1b1eae7272e58 : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4dd02dcf7871412997a4319af4252826 : SUCCESS in 43m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c84a88991e444458dba2772a4f7ec38 : SUCCESS in 1h 24m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13974e72a0bd4bdc90532bfe8dad4ae8 : SUCCESS in 1h 04m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32ad8fa2802243b1bf61d2867379da49 : SUCCESS in 15m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/44d93670a5664a49bc97d1a4826b3d0c : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dabc9a2990b84440a530646f95e37bea : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5aba496253924f10bc23f98225806c0a : SUCCESS in 6m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/716a496d909f4b36a53364d940a72fae : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/75f8f0e6f2c74236aebb898cd72d4ce4 : SUCCESS in 52m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/052e7a65b53d4ef99f1a46e4bdd29c8d : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a254cf93cc7458394e7ccb902ebe256 : TIMED_OUT in 3h 07m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/217a1e8f32a44839912d45b3d651546f : FAILURE in 1h 50m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/abfe8a41b9694d0882724a5984023316 : SUCCESS in 1h 51m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/04a37b9f4b6e494c94c1289ab79352ff : SUCCESS in 1h 44m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5ab7cb819725473480b35ff3b654e854 : SUCCESS in 1h 08m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d08232592f8e4b39bdd3e95b9e67609f : SUCCESS in 1h 08m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f29248c73d644d9ea6daffda9f2d5ee1 : SUCCESS in 40m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/86ae5937de3c43849da02e0ba1843161 : SUCCESS in 1h 10m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f6e8702574cc4cbea14d98dac2e383a1 : SUCCESS in 1h 18m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3a1c3d76b3448ecb1d097b436f1c70e : SUCCESS in 2h 11m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f2991a3cd7340c388eaea04d7865f93 : SUCCESS in 1h 04m 28s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/17b38d4821634b41854133e60ac4d383 : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8b15a98511cc66dba5b5a8e88bbd4be41226136","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-06-19 13:46:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/881470/4/ : RETRY_LIMIT in 3m 06s","accounts_in_message":[],"_revision_number":4},{"id":"27f9a04287574a6783dcd9c52bc3326021c617f7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-20 14:34:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"375a4f3266f83bc7a532795cba36f08de19b95e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-20 14:35:45.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"52301ea928a695dedf1ff69d50eb01412df16449","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 14:45:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a829027b954e4b508d4fab8f994f10bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2a8b1b3fe2c44bb80faf1932d9404ff : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3aad9690f06f44dea1855670fb0a7123 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d953806941745d1be85b243731c5da9 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"464e5531e8c05bc0047dcdcee396cce8a9595972","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-20 15:52:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/6d1778f1826f484db1c4b9f9269e9ca4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/043b5791006a47dc9a0ecef3800e2df9 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c433edc71327443e7511d0ea083664cb38f6a4be","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-20 16:39:11.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/881470/4/check/fujitsu-eternus-dx-iscsi/2d75450 : SUCCESS in 2h 02m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/881470/4/check/fujitsu-eternus-dx-fc/459fb36 : SUCCESS in 2h 02m 19s","accounts_in_message":[],"_revision_number":4},{"id":"e0caf1c9a542670d131550dd69086112c2b8327c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 17:49:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7578c90279644f1ac8ab859df472660\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56bb4a20553548a5a70603a1b54663a6 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691628d7b89846189aea48de519e59d6 : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bbfddfb83314191abe84b3d0a1e46ae : SUCCESS in 14m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92f354be710444ad97eb3e322d5caabd : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb75b40db944291a94ae9f5adf3137f : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3fb1234d83844af59e2d7752052854d9 : SUCCESS in 1h 10m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e10265dfe6c7409f93cd487ae8c816e0 : SUCCESS in 1h 31m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fca32c372344595862cbc7d90dab469 : SUCCESS in 56m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d5b01b14e3e46cf90b85d5b28dc016b : SUCCESS in 18m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cdd4b065b9644e45afcb0d6ff0807cd1 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed487dac05294cb99763d90bbb3f7008 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bbd4edd0479843568abbe28a05d26a1b : SUCCESS in 9m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4eb1bd5d9e7c4af9b34dc140a282a7cf : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a8d077d9bf1843a797ac66baf7c41728 : SUCCESS in 48m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c9331cd687f4464b834e927bd733548 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b86ba1899d8d407eb76122eccc5d2c83 : SUCCESS in 2h 00m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae2441683101497297aefc1e54840145 : FAILURE in 2h 02m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6097c6edda9d428ca0ea7b1f1dcb97bc : SUCCESS in 2h 06m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/73c1c97f55a7461f9c836cb88d337e58 : FAILURE in 2h 42m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/41cc59405ac94c1ab6b2e9ef20173eb4 : SUCCESS in 1h 07m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a1da9e2a8de74a729e25b663dd435b58 : SUCCESS in 1h 31m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9dd31b4cafe54105be4c18f36f5c58fd : SUCCESS in 45m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4bdfaec568eb45d9b139f64c0eb70145 : SUCCESS in 1h 17m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3fc5eeed45e4453883878a3445d4602e : POST_FAILURE in 2h 31m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6819f8a88c9b480782904ec26764dee0 : TIMED_OUT in 3h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a648940ebb304586b5a1b0411a333f5d : SUCCESS in 59m 26s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/d9a4f7c4b7814e7f91167ecdcedf4766 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"80832d5010f0794cb906a4541a25f51e1ae2bc64","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:13:00.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/881470/4/fc : SUCCESS in 0h 31m 33s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/881470/4/iscsi : SUCCESS in 0h 47m 49s","accounts_in_message":[],"_revision_number":4},{"id":"d013b0ccfd05341e5b905748cff54b45659bb161","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-26 12:05:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"58ae42e3d514f949ada57728093442080e29ed0c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-26 12:06:11.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c9378bdb083cdd2687d50d0f79f473eb381467cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 12:24:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ebc724c102f4d06869b84279aa06b63\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01b43bf0f969458191211ab5f55b7677 : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75ac92b212574ffc910be0ee442480a4 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27b234dfc0924db98b86cc91bbf975bf : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99b544420e455e48ae8582f9ee5333d5c2a4e086","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-26 13:21:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/c6aed33d19424b3b8e99a2859f84dca1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/92cb3787a5314c9ba0923e3963142444 : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":4},{"id":"400d9d35204f081f3d703aff4b21fe226211f402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 17:44:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef62e66ae39249f9932d5df8e0cd8595\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c5199865ce4f4f82ef4ab59f65144d : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71e7c09ca8214d8c9835d5daa3270a72 : FAILURE in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a25a997bd8c0461e976054cc10abd568 : SUCCESS in 12m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6812cfebb75543beb6f3d08bbe33cd6b : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45759415f3ce41108c64ee9fb6794561 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/f3c1aeb674e3473dbe0f456691c8dc60 : SUCCESS in 1h 17m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fc530a9fe754623a5352bd8b58d82e1 : TIMED_OUT in 2h 03m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2738f91a02c42f48e1675f95b1aff9c : SUCCESS in 1h 18m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1f097235261240a1b52f5936a2622466 : SUCCESS in 17m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5670d10fd9a24d17b476c86df88cf6e5 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a9c6d04e5354cbeb6052ae498597699 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf7a4399de2d4533b53bef5fdc52c543 : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1e6baacb35944e8af24b133c985f18f : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a94baca62bde499aad1c47944a25d91e : SUCCESS in 51m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db9f57f624f04956aa3a3acc28a66812 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ccfdc2e76dc47de80a920364e98747a : SUCCESS in 2h 01m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a116d06ffd9c4025addd4419ed2da352 : FAILURE in 2h 17m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb21758872de4b6386397027d7dcffdd : SUCCESS in 1h 58m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c85683fdcab644c58d5094a1e9d773db : SUCCESS in 2h 33m 53s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ce96306a42b475b9d72d0cc02112c59 : SUCCESS in 1h 14m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/766a8648f6c84911ba2dbafa2d04a129 : SUCCESS in 2h 07m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/791f9ac6aebb499e99edbd10d9baf3db : FAILURE in 1h 17m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6eea158accb2493a96153b09ca07857f : SUCCESS in 1h 51m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/207cf8de03dd4ed89338e7deff83dfd1 : SUCCESS in 1h 57m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/784003d4eaf94e12afe74bc11acaccec : FAILURE in 2h 51m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acc0c93a345c4c72a77f910b0323b978 : SUCCESS in 1h 42m 35s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/c72c635367f840c489c5bdb6081964af : SUCCESS in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"79181efb4dbdb8530a017ab503f8d76aca4fdb28","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-30 12:32:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29a3e8a5cea6ef4e758e61a52f5f069ff6c59cd4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-30 12:33:44.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"82ce7fed555e55610f009f69f87c0e76a7c19935","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 12:52:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d418371c6c63446c9a56fd2a4ac85ae0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a5acc4b061b4f2f9b8a4fcaf40d89db : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d9b7522fdd34e96b0616e28564299f9 : SUCCESS in 16m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/856e023448cd41a9922b8960bf46ad59 : SUCCESS in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d72deae37c9564a9f4f5f3df0ef1a63521cb57a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-30 13:49:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/a0cd2f3b5c80496eb5616c871a1bbc82\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cbcca9dd09844316971e53721477a7eb : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":4},{"id":"fea8e9e795b6856765162a82ff54e67e41365d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 15:45:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2296c00421048fe995370c9928430ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98722c2ba1a445c8a22c89267e1bf1f6 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0e3168b3dd94d91bc323a049abff7d4 : SUCCESS in 12m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/792425417c0543878035306f5f040961 : SUCCESS in 11m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b849f6b8aed6495c806e5dad1822f0e8 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e62a269c5ef4a12bb348a3225f454bb : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/c4b15135c1a14eb9bb8a18aaeff38e52 : SUCCESS in 48m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e70d2b03c3204278b9e40c31250f505a : SUCCESS in 1h 26m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1e84ebea31744f8b03a302c78dfe2ac : SUCCESS in 1h 04m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/583f0f0827ff4b3996f12fa007657b0f : SUCCESS in 15m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f292ae351e24dd79c426d1c4f9ae0c1 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f79526ae6ae247b8bc351a26943c0066 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7de59da99b9e4a039b69031cfe53120e : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f12466de3ede477da710dfdb198dc494 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b9d59d0bf03f4c52983f692042a70e11 : SUCCESS in 42m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d0bba0732904ccbaffd8c9db5b3314f : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d906fde6f0074eaaa76af24d4d6287ff : SUCCESS in 1h 52m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/95039a7ff1084c81883e3f191fefdba3 : SUCCESS in 1h 41m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf3a12b6241c406eac4e086e3e7287bd : SUCCESS in 2h 17m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7efeb68535b447639ccb5f3a37b4da8c : SUCCESS in 3h 04m 59s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/352743e3931f4a9e972d50b2a3190301 : FAILURE in 1h 37m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/391f2cd6ea834c7eadba10ca24d1292d : SUCCESS in 1h 53m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/70564a11a287406e9153a88bdf4e8cd1 : SUCCESS in 38m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf84228941bf4c1685040c0ab708ffad : SUCCESS in 1h 33m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2222583aec4545849b7a5d60acb4f6a3 : POST_FAILURE in 2h 31m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86bbff10cea24cf5835ec0bfc75af843 : TIMED_OUT in 3h 03m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87e457c6d30e4e648e65cf82937894f1 : SUCCESS in 47m 28s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/32f709297ae84595abbd12a2517c534f : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00991eac3ac35bf7171541e46875c11bb5b4d11f","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-07-04 11:02:07.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f5ca529aeabff4665c4ec81569ddeaf57a5f431b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-06 12:42:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"569e97088df37d27ee633b090ef1e9e320170a36","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-06 12:42:56.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4217db9cca32f671c2adccb4f64c1b57477206c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-06 12:54:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/d22f2857cfb944b8903c08bcc876143f : RETRY_LIMIT in 1m 13s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/13dcbb2e77c94879a09f047da6f3f2f0 : RETRY_LIMIT in 1m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/9d85c8b1a7f44584995bf3ab19e48de3 : RETRY_LIMIT in 1m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/82debac66da542d28d743463e761c2c5 : RETRY_LIMIT in 1m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7d73013b2e395f5635957a5f8f6a2a1b686b84ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 12:58:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f00b944644413a8c01f939deb2c05a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/660a276048014904925696c1c9773e8d : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55aa7be27d274848be6a90c1d3642e77 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/143186a7516448738e9c8fd7634a0806 : SUCCESS in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c752e8d6f16e31bc903f7b3ffe7191b90785e0b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-06 14:01:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/dae10eb45e5c4f0db4cee2c999a7cd66\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e3511347012e4f4086bea9e1ae5ae111 : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":4},{"id":"a5820888feb7f9cc48405f3a782e03c6d8792123","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-07-06 14:49:58.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/881470/4/check/fujitsu-eternus-dx-iscsi/5592fc0 : SUCCESS in 2h 07m 20s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/881470/4/check/fujitsu-eternus-dx-fc/82381cc : SUCCESS in 2h 00m 21s","accounts_in_message":[],"_revision_number":4},{"id":"cde48d21d48c466307a5dd30835881c70ad7082d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 15:59:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0798c28af0734df3bc2d17f7de922d34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf34b8b73a324210bf67d66a904cb31a : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45bde4e1d7c44fd59a7a401eec833e93 : SUCCESS in 12m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/532a441ff1964af4858170f12d5acdea : SUCCESS in 14m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/321bc09d0b38490a91df44b40e02cc77 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148c7eb8cec34b2f8a0f58585236819d : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/49f1358568e34be99daaa63285bf47d7 : SUCCESS in 50m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff93f1d8492b4a3785e2d2331115611d : SUCCESS in 1h 51m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6855baecc8c4c59b4ecb963b9193a04 : SUCCESS in 1h 04m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/740cd39570794f4cb71d6f347bd31f03 : SUCCESS in 19m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e388c83413040c2841be7197612df4c : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e18ac2e6b73948d5a1c33daef8ad4dce : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3955c6fd93b344c4b99b25ed0b7eaba1 : SUCCESS in 12m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/481821a3b05442d698c2cf8e2ff97a0d : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b1298468a73d450b836d74aa6670e041 : SUCCESS in 1h 08m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c5e585876b9c49299396a3a4bcd52b8e : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/215edcd4ea264bddbfa01f6eb7c57a1a : SUCCESS in 1h 55m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ac4985fec1bf4da1ac5d184223d0c125 : FAILURE in 2h 16m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/216a3238278248c4b4aa849fa0193e06 : SUCCESS in 1h 58m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dc231d86c9ed475cb7be422c477903b1 : SUCCESS in 2h 36m 40s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2178116ccadd41aab551531c6e6b7c6d : SUCCESS in 1h 12m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fdd0a77cce7e46b6aceb72f097a3eb27 : SUCCESS in 1h 38m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a4c05af063954223a47029b120ce5a3c : SUCCESS in 38m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dcc38d6228f84f13a7779af50719b53d : SUCCESS in 1h 13m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2fe5ac89c0694f42b72696283bce3af6 : POST_FAILURE in 1h 50m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/554e660769fd44ff9efc2ceaba68e962 : TIMED_OUT in 3h 03m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e408d823c1c44f2af76fa2f29661690 : SUCCESS in 49m 33s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/589e124089674d1893fada192becdfda : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ae07db2b8e52d215e9f4397326d0db935cd28af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-11 18:27:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"148d621bd46812d12981aeb67d045da54b2652ba","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-11 18:29:15.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-881470-5/ns5_nfs : FAILURE in 1m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-881470-5/ns5_iscsi : FAILURE in 1m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"5f4ab222d5452e66f276e69630ef077f5eaba024","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-11 18:32:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3ebb54e511274382a6a7a87fd223d267 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/97d08669176f4d4bb3a9007958e24d1b : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"d6d8cdddc6b0726774b2e232e26655d8f011e221","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 18:38:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbd2ba7d9a424b9f97b0ef3dc73e1a45\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/afd8515505564bd9a820316b5d843120 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e65d0c211b24ba9b4891bd9191aed7f : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe4c0a7d7ddb4996bff08c1188fde717 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"be346615194af47f04fe31c7e86d3fb23a0923fb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-11 19:47:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/ca9da5aaaf30470099e4df4201faab32\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a1c0f0d95ad94f77b4b8dfb1ed6af5b3 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":5},{"id":"c67e4decc8bdeda9340726a1ced39caaccb74b8a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-11 20:02:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881470 : FAILURE in 1h 33m 49s","accounts_in_message":[],"_revision_number":5},{"id":"47377e57c6c4787afb80566e0cf3a2d017c727b5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-11 20:37:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/70/881470/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2244417/ : SUCCESS in 2h 03m 45s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/70/881470/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/3656998/ : SUCCESS in 1h 53m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/70/881470/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/ba13b5e/ : FAILURE in 2h 08m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/70/881470/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/7e36266/ : SUCCESS in 1h 49m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3562bf4bdb9cfcaf27d2319c0aac6e916f0ef770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 20:50:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9bee5f0f4b4762980b16d6bd18ca03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6e7559495b4da79140a723685e3329 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9610607499ad4853b97f16364432d4f6 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1daf901c33bb4ee492d0a74d7b8e2cf2 : SUCCESS in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba3ca0bb26524a5b879ce643b95fda2a : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691e88b2c1a3462cb04293d6d32a47b0 : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/589998daa4ed43b19dd23134c803e365 : SUCCESS in 47m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76a091f63d864b2c8ec4de0749b733a3 : SUCCESS in 1h 17m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9505ae1f1454bb994b7638960cec8e7 : SUCCESS in 40m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e202d6ac9f946f7a2218f92139bee28 : SUCCESS in 14m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a1abbd13fc1446b88aa0ae7edfd4aa9 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2afcb97badc43ac948cb5c0f14f342e : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e320445f5d3d448ebf4acf8e8ce0323b : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b46bdbe419b0462897960c53b225828e : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f367f9fe97f4acb97f58fe8ef4661da : SUCCESS in 36m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b4210d2596e546b4a17731deb1432bd5 : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb71271f14934e72b99afefd1cd9b44a : SUCCESS in 2h 05m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/425082aab8744e919ff487f29a9e3edd : SUCCESS in 1h 41m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b146b84e6ac048478b07415c09c973d5 : SUCCESS in 1h 31m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4267c6c202c04bca889dc932a1464b70 : SUCCESS in 2h 17m 37s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34944f4745b4438d88798d5fa0d9814c : SUCCESS in 1h 17m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2568a22fa0f947f2b13ae0e91083ccb4 : FAILURE in 2h 12m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4549bd9d72ce4b1cafbb58ce7cdbabc1 : SUCCESS in 33m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/20ae7dea01b2463fb1e3031b59a05e15 : SUCCESS in 1h 08m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cccf54ae53a745f4bef63b97ed775d0d : SUCCESS in 1h 11m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd0a890fb9a842b7899514606c4a75bd : SUCCESS in 1h 07m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13036b0dbccc495182e1f313738450a2 : SUCCESS in 1h 06m 27s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/2513905fc81e45c99e72c98ba2afb69a : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"acb3bf63ff255fea256cfe5b827e0a7c65f3d862","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-11 20:57:15.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_07_2023_20_50_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"92d8d2935b719d96155fef46437ef882b24748bc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-11 21:00:08.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/881470/5/fc : SUCCESS in 0h 32m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/881470/5/iscsi : SUCCESS in 0h 44m 33s","accounts_in_message":[],"_revision_number":5},{"id":"4c6b6d05a55dfe7827afb5ff456139ac7f431cd2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-11 21:10:16.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-5/FC/ : SUCCESS in 2h 42m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-881470-5/iSCSI/ : SUCCESS in 2h 8m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"67cf85005b75b4b7ad61d5620f38fc7cd07ca7c8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-11 22:01:02.000000000","message":"Patch Set 5:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1568/","accounts_in_message":[],"_revision_number":5},{"id":"f3605939be4b95f22138140407055a29f75b8479","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-11 22:11:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/881470/5/upstream-check/cinder-tempest-plugin-ontap-fcp/47b8e0e/ : RETRY_LIMIT in 8m 24s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/881470/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/8781dda/ : SUCCESS in 1h 17m 53s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/881470/5/upstream-check/cinder-tempest-plugin-ontap-nfs/31c0e1e/ : SUCCESS in 47m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/881470/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7f61466/ : SUCCESS in 1h 20m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"f70e2af550bc66dc3f1351c5b2d92c9f3442d1e4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-11 22:13:30.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_PowerStore_iSCSI/040a569/DellEMC_PowerStore_iSCSI/153 : SUCCESS in 53m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_PowerStore_FC/523cc05/DellEMC_PowerStore_FC/161 : SUCCESS in 1h 22m 58s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_PowerStore_NFS/7f3ef75/DellEMC_PowerStore_NFS/155 : FAILURE in 30m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_PowerStore_NVMeTCP/7931577/DellEMC_PowerStore_NVMeTCP/153 : SUCCESS in 1h 14m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"682fdaf0700b77e020ab49598c9214fb996236b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-11 22:37:34.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_Unity_FC/65d8221/DellEMC_Unity_FC/407 : SUCCESS in 1h 55m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_Unity_iSCSI/af34fc2/DellEMC_Unity_iSCSI/409 : SUCCESS in 2h 01m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"3b52ca4751e923e1e0fb97427bbe83f10e275000","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-12 02:15:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/70/881470/5/check/tyds_test/9d18c13 : SUCCESS in 5h 52m 14s","accounts_in_message":[],"_revision_number":5},{"id":"37e8c1ff71c9a80094c5921f285f3be7c4b1aa99","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-07-12 07:07:31.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"884911143e1e003aac27a233b373e0ba6628c6ea","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-12 08:01:09.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"c650c1855c6da9f2d7e0936fca65d506849b02b9","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-12 10:46:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/881470-5/HPMSA-iSCSI/ : SUCCESS in 1:51:46\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/881470-5/HPMSA-FC/ : FAILURE in 1:28:26","accounts_in_message":[],"_revision_number":5},{"id":"1668d056beb986650ce6f8373a918d69c956b68c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-12 20:42:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/5/20230712232405/NEC-V-ISCSI : SUCCESS in 2h 58m 31s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/5/20230712232405/NEC-V-FC : SUCCESS in 6h 18m 39s","accounts_in_message":[],"_revision_number":5},{"id":"b44d67ad1a2c5977e0bb4375c05379c6bb5533ed","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-13 00:44:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/5/20230713053714/NEC-ISCSI : SUCCESS in 3h 56m 11s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/881470/5/20230713053714/NEC-FC : SUCCESS in 4h 07m 24s","accounts_in_message":[],"_revision_number":5},{"id":"ff9fdfee255f8b13a250e489b814e95241f82955","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2023-07-14 09:25:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"998b29efaa13548952cc330a19323bdbbe954245","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-14 11:11:38.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/70/881470/5/check/DellEMC_PowerFlex/23a1701/DellEMC_PowerFlex/274 : SUCCESS in 1h 45m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"a543d388f4103978ec9e6675a40cd1a1ffda73cb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-07-15 12:10:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"32c41e58f4363a57bb25ffa8f1bea913cc9d740f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-17 20:56:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"016166d0eb868bb35e208db4e854fda1898a5a19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 20:57:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"66521fa9b82af2eaeb6dc868ec49b1d111695ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 23:55:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699ff140cf8c4acc89d1a2bea3557abe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69562c4a47d043e2aafef9af9bdc4b68 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c41fa84b39794627b447a3e04b8c6c2b : SUCCESS in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc5f043c23ae4edda16be0322194caaf : SUCCESS in 9m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8db198a953ae49b5a7b39019ca507014 : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c891624566f4efab30065821df43912 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/044ae8c59c184b4cacd356d9caae3f1f : SUCCESS in 44m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad415b7d3ff54ff997c3f578b1015c51 : SUCCESS in 1h 51m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44327e74210b431a813a461893f7323b : SUCCESS in 46m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba2f03593a804de386995ea5b7dc1015 : SUCCESS in 1h 10m 40s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d18b34a2a8346db9c18123ed427a8e7 : SUCCESS in 2h 09m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc13ae1dc75141b9a47d5b0edb4233f3 : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":5},{"id":"b98a5a6381804dc9796ca4992a92b65371d56bb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 23:55:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"7a22834110e15e232ec0fa0c95188d5cb4e7ba21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 23:56:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29b78e4ab934803bf1fab6999ddc9e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5ad5a94df44a8aa535788566c56c7e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"540ff2e7158d1bdfb1adede178c7a13d3cec460c","revisions":{"19e807fa60623205629fa507d904152cb1100099":{"kind":"REWORK","_number":1,"created":"2023-04-25 12:07:46.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/70/881470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/881470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/881470/1"}}},"commit":{"parents":[{"commit":"2e8aff660b18d1e8f961d6de979d541a8d06f1c1","subject":"Merge \"Add fips check jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e8aff660b18d1e8f961d6de979d541a8d06f1c1"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 15:27:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-25 12:07:30.000000000","tz":-240},"subject":"Cleanup image_utils chown check","message":"Cleanup image_utils chown check\n\nRemove duplicated code.\n\nChange-Id: I697e8c685d67250e2d016964c20bb2189ac384a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19e807fa60623205629fa507d904152cb1100099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19e807fa60623205629fa507d904152cb1100099"}]},"branch":"refs/heads/master"},"76b9f4d7cb01977ec3cde4c609a937b8d0264826":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-18 15:58:51.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/70/881470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/881470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/881470/2"}}},"commit":{"parents":[{"commit":"456b6399bece9a9ac6274e56130cff6680d99096","subject":"Merge \"Reject unsafe delete attachment calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/456b6399bece9a9ac6274e56130cff6680d99096"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 15:27:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-05-18 15:58:51.000000000","tz":0},"subject":"Cleanup image_utils chown check","message":"Cleanup image_utils chown check\n\nRemove duplicated code.\n\nChange-Id: I697e8c685d67250e2d016964c20bb2189ac384a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/76b9f4d7cb01977ec3cde4c609a937b8d0264826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/76b9f4d7cb01977ec3cde4c609a937b8d0264826"}]},"branch":"refs/heads/master","description":"Rebase"},"2001089942db278354bc3b9b9b4ac02ad25b27cf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-23 17:10:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/70/881470/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/881470/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/881470/3"}}},"commit":{"parents":[{"commit":"753b1cd318eb732d34c3ca9528e93fec4cb4a44d","subject":"Merge \"doc: Trivial typo fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/753b1cd318eb732d34c3ca9528e93fec4cb4a44d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 15:27:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-05-23 17:10:47.000000000","tz":0},"subject":"Cleanup image_utils chown check","message":"Cleanup image_utils chown check\n\nRemove duplicated code.\n\nChange-Id: I697e8c685d67250e2d016964c20bb2189ac384a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2001089942db278354bc3b9b9b4ac02ad25b27cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2001089942db278354bc3b9b9b4ac02ad25b27cf"}]},"branch":"refs/heads/master","description":"Rebase"},"df5816ad67ba20a392f564a59ea09582ea963c4b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-05 14:16:14.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/70/881470/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/881470/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/881470/4"}}},"commit":{"parents":[{"commit":"ac31b9e2b3d29f9fb840abc8ead3532e047042cd","subject":"Merge \"Fix glance metadata properties filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac31b9e2b3d29f9fb840abc8ead3532e047042cd"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 15:27:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-06-05 14:16:14.000000000","tz":0},"subject":"Cleanup image_utils chown check","message":"Cleanup image_utils chown check\n\nRemove duplicated code.\n\nChange-Id: I697e8c685d67250e2d016964c20bb2189ac384a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df5816ad67ba20a392f564a59ea09582ea963c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df5816ad67ba20a392f564a59ea09582ea963c4b"}]},"branch":"refs/heads/master","description":"Rebase"},"540ff2e7158d1bdfb1adede178c7a13d3cec460c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-07-11 18:27:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/70/881470/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/881470/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/881470/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/881470/5"}}},"commit":{"parents":[{"commit":"7fa5561eb5ac9100f26be5a9c75ed070ecb728ac","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fa5561eb5ac9100f26be5a9c75ed070ecb728ac"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-21 15:27:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-07-11 18:27:23.000000000","tz":0},"subject":"Cleanup image_utils chown check","message":"Cleanup image_utils chown check\n\nRemove duplicated code.\n\nChange-Id: I697e8c685d67250e2d016964c20bb2189ac384a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/540ff2e7158d1bdfb1adede178c7a13d3cec460c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/540ff2e7158d1bdfb1adede178c7a13d3cec460c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
