)]}'
{"id":"openstack%2Fcinder~882781","triplet_id":"openstack%2Fcinder~stable%2F2023.1~I98e021594c59dbad1723597e38aeb4c0cec6934b","project":"openstack/cinder","branch":"stable/2023.1","topic":"I98e021594c59dbad1723597e38aeb4c0cec6934b-stable/2023.1","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-05-29 14:24:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I98e021594c59dbad1723597e38aeb4c0cec6934b","subject":"HPE 3PAR: Few issues with new WSAPI (of 2023)","status":"MERGED","created":"2023-05-11 05:09:22.000000000","updated":"2023-05-30 06:44:35.000000000","submitted":"2023-05-29 19:51:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882781-I98e021594c59dbad1723597e38aeb4c0cec6934b-stable/2023.1","cherry_pick_of_change":879836,"cherry_pick_of_patch_set":5,"meta_rev_id":"24d6e7f0ae01c183378a028e613b82c7c724936a","_number":882781,"virtual_id_number":882781,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-29 19:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-29 13:50:27.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-05-25 15:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-29 13:50:27.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 05:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-11 05:28:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-05-11 05:41:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-11 05:48:44.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-05-11 07:06:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-05-11 08:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-15 18:50:07.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 04:37:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-05-25 11:41:08.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-25 15:04:59.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-05-29 13:50:27.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"79631105a43aca06a0b4e540a7fe25a8b250e69d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-11 05:09:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* cinder/volume/drivers/hpe/hpe_3par_common.py","accounts_in_message":[],"_revision_number":1},{"id":"709d97d4b72a9fa3c8c47d307982a8689303462c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 05:17:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c620093310548e08b86c0566b0600a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91e400e8169b49a7aa9e97ff1c2d8228 : FAILURE in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e713b3630674c5fb3524c8215fd1a69 : FAILURE in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"009804a6425df7d9943ca842b7cd3f702b245a24","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-11 05:28:34.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9288fd057eff45069f953d23e83978a9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e37458fbeefa4bbea866b484a9e56338 : POST_FAILURE in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4ebd0d46f65803a75e274bda56eb09a6028cafcf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-11 05:41:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/882781 : FAILURE in 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"21462f40ae0462f2ca2be40761d6ff1fffb93310","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-11 05:48:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/882781/1/check/iscsi-driver-alletra-6k/01f2416/01f2416.html : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"63df3ea493a1fd2537fc92e056c034c2de0a270d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-05-11 07:06:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/81/882781/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/327c8a1/ : SUCCESS in 1h 47m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/81/882781/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/f4345dd/ : SUCCESS in 1h 41m 07s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/81/882781/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/4f87871/ : SUCCESS in 1h 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d20bd899f8aabb7830e28ec0afe6ac76b94e52b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 08:38:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e0b1a250f124c5e9ad193bedfde4dd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/982b97b9cdb443fcb43ecfd06c616f93 : FAILURE in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8f5b93dfd1f4d0d96b9e92e2cdd7b42 : FAILURE in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04f8535d313a4f8c81125c272cdb8e00 : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409417f9b26c46c18daae0c016e3bdba : FAILURE in 6m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/4d78bdf8747c4a6eb0870c6c4e5095dd : SUCCESS in 43m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f780c2f29634d37b3bc93ed104c5432 : FAILURE in 42m 33s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b4592eed4424bdba8dab1d1effdd573 : SUCCESS in 1h 15m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93868b6f43d94074b3a50c72bb09781d : SUCCESS in 1h 02m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96828d0fa72b4696b76e861a77e144c3 : SUCCESS in 8m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd3396efaa6449b29e2ba9480ffcc106 : FAILURE in 4m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bb545c27912f48edbdaad57b23ec54c1 : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60ae569ac60745e0b4cc86c29b738171 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0405246d8ae74f3e9a215649ebd5c9f4 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11fdfabc0e2a42eebfff345342ed870d : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9370481396348e0bb43af9dfaff43f7 : SUCCESS in 34m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b953dc86c2ed48fbb23d24541e0ee2d2 : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/89601811df104edbb8e1e201c5321e87 : SUCCESS in 1h 20m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/43e71e50d58f422d8ac539049c3a337f : TIMED_OUT in 3h 05m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8814a33156174f919bd77a4c9e765029 : SUCCESS in 1h 25m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ac6ebb7fc1e4c1faca572dea32b2a8a : SUCCESS in 1h 09m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94df6a328c9344b0ab0a73a3fae2be93 : SUCCESS in 1h 24m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0297b1d9613e4244986642e32eada324 : SUCCESS in 32m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/20f1475fab324ae5b381f80e551ae92b : SUCCESS in 1h 10m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/06505360029d4b3d96eca078379d7e5b : SUCCESS in 2h 02m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edd3ce84c131410c8cc2f646a16e419c : SUCCESS in 44m 42s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d268032725e48e69f344c44db0133f6 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b70f3db40154219810a748bf9beffd356585ee90","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-15 18:50:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/81/882781/1/ : RETRY_LIMIT in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c479642039b4ced2af6c439c82f53ca578d901dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-24 10:15:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"882096c5412e07d6b789acf90b80f2f16d7d52f0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-24 10:20:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c8cd8ec26b6f05a5ce3bfc0b4b1014372be46f7e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-24 10:35:45.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a0d3d0e2297844a789fa1422c1ec6230\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ac83d5279c784dc882d36ebbb1940667 : POST_FAILURE in 17m 53s","accounts_in_message":[],"_revision_number":2},{"id":"73c5a7b00a3403a246bf705a52f081d6c032263f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 10:40:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89f51de124984a368b5bff6d96963382\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/add784543ceb4523aa3ea01cf0f8b923 : SUCCESS in 21m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/60758a6eecca4743b919b35654d764ed : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac13f1cd50bf222ce952cbedaf674491e9111186","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-24 12:00:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/882781/2/check/iscsi-driver-alletra-6k/ae1fdca/ae1fdca.html : FAILURE in 1h 43m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e375781c292bbbde6c77926912d591d306f297c8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-05-24 12:09:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/81/882781/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b8ea2aa/ : SUCCESS in 1h 52m 29s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/81/882781/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/513e9c0/ : SUCCESS in 1h 52m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/81/882781/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/123a8fc/ : SUCCESS in 1h 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f5f61bba3f74e5581554e4f32477c220796f314c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 12:39:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e92a346f2cc4c328b7c93633818c9c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b36d9ed9b567474c87e60206cd0c69c1 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f72880115462481ea4ef86057d648484 : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de7c6acae6104ad18085cdb1561b9789 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eda98ef676c4659aa3f434a0e618497 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3a12daebcf314fc5a8a9bb07ce886c39 : SUCCESS in 56m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/73ceeaf6236a44b6a6d5508e49399773 : FAILURE in 31m 13s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e91e6ed75ab746d7abcf18f6547249af : SUCCESS in 1h 33m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1c5b8130e2a4a6fbffc160bfab7c5f9 : SUCCESS in 1h 02m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73782174ad334db28da38410aba5ffe1 : SUCCESS in 7m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f869cc38ee34b4a8454b3d38489d5aa : SUCCESS in 15m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/456458e8516c41a1963d0a271bd795de : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d32fbb1748b24849ba8b9ce939f872ba : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e6006b6b1bff4dd4bfd94aed28a789f1 : SUCCESS in 4m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1d6cd2181d14af9b614115fd269a157 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/141ab419d6c54d9bb1373c4d67217def : SUCCESS in 57m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b23a0e1a0fde4f06815bc8500e313c93 : FAILURE in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbb07469fd9a4870aff9eb755323fe08 : FAILURE in 1h 21m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/93d4c76d2373421c9e3c0146a93925d9 : FAILURE in 1h 24m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a3cd6b6fd1f4655a0ba85cb8328f04f : SUCCESS in 1h 47m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/073f8f726af84f8a82242db8515bb3ee : SUCCESS in 1h 13m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7612c1a86b6f4feda8763b553692d777 : SUCCESS in 1h 19m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e48e3d12a28d4a3f873081eadcc5c696 : SUCCESS in 36m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d17d474c08c14a208ec1a048d49bf80e : SUCCESS in 1h 06m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d69b4c31ebf84217837a5a1a374cfc9f : SUCCESS in 2h 19m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec3a129386f24d9eaa1a2f1f64f4e704 : SUCCESS in 50m 12s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77a4c7331daa40c6bc47b177e1add648 : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"56b074666b3212858e4b2126f68f1c874a60647c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-24 12:56:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5170174dfd79730d6063f08d7e5b5f70aade88fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 13:11:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c8a9ab802343adb91b9cfdb9123f49\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78e0fa5b77a4494fa18d92ca68616d9b : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be1b0da0dfc2421ea94e54a8e55e3d8e : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d4a105fa6887f38eec88c6f70ca791b43ba5659","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-24 14:06:23.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/043d366d83f343288322ad26abcb6a19\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/aae40199dafa462d90619f3e1aae271a : POST_FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2431fda7bf41d75edccf85cbe8704482e772c3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:07:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/659430edba4c438da26c79cd7f367134\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07a49abb1724b059ac24d6b2cd66f08 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a0a92784a74b23b15befd1d44f02c0 : SUCCESS in 9m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10007efc10534b96917ee5a3014a7644 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1f470cdf6149f4aa1ab457f3838df7 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/6d4397b8b1154b28a09477dc5dd4b74f : SUCCESS in 51m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b405d61aad644352a884f356b8810ef4 : FAILURE in 30m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea541d5b25b74d788081c7efbfc0fba2 : SUCCESS in 1h 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/130d6a2d72714300b6a1f81aa0ab4463 : SUCCESS in 1h 04m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23c4ee3553a74baeb3ad1c72c0f2d175 : SUCCESS in 7m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8f9409fab5db418c842cab66f36ddd41 : SUCCESS in 14m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/808f841a826948f89b6e04872f8ab4f1 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73e00534c3224a70b7ddeb02f7ba37e7 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68eb58fbb1784e01aa8a6ac66da2695b : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a848b96209c94894b5d8e6b330f1a706 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a87345287fea4cf4ab6e9928f5e2e8bd : SUCCESS in 36m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d58849b0bdc44f9c83bed55f81089652 : FAILURE in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13697807e5084ef69c2f28149a50b2c6 : SUCCESS in 1h 23m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/02d8a19d4c164d07b1f452c5659bd625 : FAILURE in 1h 19m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/38ee117452174c508a7451344a88de74 : FAILURE in 1h 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6a604d3a7914f78b3815089bd3c1b4f : SUCCESS in 1h 02m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f0a47a7658d34d13a76a8478718fa3fa : SUCCESS in 1h 38m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/67a95bf415aa4b36b045fb0f0002466b : SUCCESS in 48m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b5095d784cf4377a6abdcfdfea9d84d : SUCCESS in 1h 11m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/64971b67cd7f4c34886111ae0424e9f5 : SUCCESS in 2h 08m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102c5405347640ddb0e46faa254356a4 : SUCCESS in 55m 11s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1a20d82ad094feea1c958ccd4e5faee : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9aea081b8f987c74ba2d2fe2c75fe18d1a00049e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-05-25 04:37:06.000000000","message":"Patch Set 2:\n\nBuild unstable. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1378/","accounts_in_message":[],"_revision_number":2},{"id":"3a9161ad61ababf1fcdc25e1178281850038743d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-25 09:08:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b935f63316a1aaeb560d0268aa663120ff36362","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:19:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d40bb2ff20d4ff89d99df017a877014\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d367fd8e3e34ae6b72fba2e8fdbfb79 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41bbe84f839647dd88bad467d7644b89 : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52af2b76bdaa20b00e6cef5cf0bab67e8f9d536e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-25 09:30:01.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1942e2ad316e4bafb6b993c2ed3c34a8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0088808320db40f3b8ab57ecf8e6651a : POST_FAILURE in 18m 36s","accounts_in_message":[],"_revision_number":2},{"id":"657e9aa40d23214da136e4b037721e44754c21eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 11:32:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4df499b4e83f4a0cac0813a5447b5aa2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ccbbb679554514b57733d05265a26a : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9677b082e3344eea1c77a4a86ba77c7 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/099bb1c43e8c4935a5a40bfb43f05d7b : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0b3bdc053942d1ba30128ce3a1f187 : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/6f219ef017b9402fb7ffee74bca45d4f : SUCCESS in 56m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bb719e9adc1643d98b1c992727899f69 : FAILURE in 32m 56s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/058e7ace75704f60a59bbbaf9b98462f : SUCCESS in 1h 19m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0d50bb5958148bda90d98e3a0e8f7cc : SUCCESS in 1h 01m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b03725d9c10149a8b5d0d90c0ffb4fa2 : SUCCESS in 7m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3aa3c81b7cc949a08aa207359897d64d : SUCCESS in 14m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/466feae3098048e6a79c2dc1d3762873 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b886ccccd68461ab40e8c36538dbb43 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2978bee68c4644f0bdb4e45f61802094 : SUCCESS in 6m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95384b75d89b45e190366b25746ab6b9 : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a63ba46f06834b4ab6eb6c66c1682cbd : SUCCESS in 39m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73a89b532d6647f4a8e344ac06a08bc8 : FAILURE in 8m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f3817a1188648d28824f9e72cd68ef3 : SUCCESS in 1h 20m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6882e358ca87407f99239294c8510443 : FAILURE in 1h 25m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57560f5b875e48f0b5c2e12789b9e037 : SUCCESS in 1h 57m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/717da2cb007d456ea1a8810944330a3e : SUCCESS in 1h 02m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/907d73dc40364a04a1f34ba5ed81b482 : SUCCESS in 1h 25m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bcf2c7539e6e44408a49568a4de0d859 : SUCCESS in 42m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce2a885dc526449fafd2b24e7e2c8b17 : SUCCESS in 1h 17m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/056af4c2cd414a959b41e11926498afa : SUCCESS in 2h 18m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5d6528e94c24ffabf1cbbaebdfd4a77 : SUCCESS in 52m 00s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b27383ae061426a9b361c20d6d69170 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"70952efe679e1b03d4713c7d40de30d3e6d21c84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-25 11:41:08.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_05_2023_11_32_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"954618253dd40a9a5c98d06ae8f307d3d6dae981","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-05-25 15:04:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5906cddf3d8030a56291919092cdecdc8b3e4738","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-26 11:47:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/81/882781/2/ : RETRY_LIMIT in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"754f82d297f0bfcca1d3da03c070238f88b1ca35","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-05-29 13:50:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad04c04dca788744d72821a2af628f4ed4587039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 13:51:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9049244cc46b22b9fae286c3d1abe61a153187d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-29 14:24:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ae8f59fc4d58f1fac69f10c4b6dbd0d0a652292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 19:51:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1280913d024ef3813386036d3896be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/932b30eb84af4094ad44fa9b32dab50a : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdf0f1479a2e4b11877828af3081f502 : SUCCESS in 12m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82451db36ffb4715ba7d6f5965543782 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce0b8d7e9364d6295befae27638efd2 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/1fe923e630794472a4264f91883ba785 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d363aa68dc834dc48937340d4b2d5ed6 : SUCCESS in 1h 33m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83ad2bb94caa46c8b1f0643786ca83f6 : SUCCESS in 48m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f1ffe6eaa3645c7873691916d4bff54 : SUCCESS in 8m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0cf7a1d13214753b91130938c33dfb6 : SUCCESS in 54m 22s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/82a0e3c9e90e475db3fa8cd80b6e8f8f : SUCCESS in 1h 10m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd98518084164a558dd6f0545d4345ea : SUCCESS in 58m 56s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7965ab236154d1b96a8cbdc77c914fe : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6fc5b4acc880c88e3dadd002171b72a3c43acd11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 19:51:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5a7480c4160dbcd7ec465a748d8d0cae2d2e679d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 19:52:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/180a005939ff4e4ebb64fe98c383d29c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6c7f2835e44c36b6a011e74a1a8197 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02d4c57392d24b6697d81459f26ce302 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2},{"id":"24d6e7f0ae01c183378a028e613b82c7c724936a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-30 06:44:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ced3bcbe173b0d62a05a12f319edefef542db1c4","revisions":{"ebde295e6863197b453e8bf672e1b6e93ce91339":{"kind":"REWORK","_number":1,"created":"2023-05-11 05:09:22.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/81/882781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/882781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/882781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/882781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/882781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/882781/1"}}},"commit":{"parents":[{"commit":"5975376ba951d6abb1e31cece2748e25823327e6","subject":"[Pure Storage] Add check for new error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5975376ba951d6abb1e31cece2748e25823327e6"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-04-10 10:53:32.000000000","tz":0},"committer":{"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-05-11 05:09:22.000000000","tz":0},"subject":"HPE 3PAR: Few issues with new WSAPI (of 2023)","message":"HPE 3PAR: Few issues with new WSAPI (of 2023)\n\nIn recent HPE 3PAR WSAPI, observed issues with parameters\nlike snapCPG, localSnapCPG, etc.\nDue to this, few operations (eg. create volume, create\nremote copy group [RCG]) fail.\nIn order to work with new wsapi (of 2023), performed minor\ncode changes in 3par driver code.\n\nCloses-Bug: #2015746\nChange-Id: I98e021594c59dbad1723597e38aeb4c0cec6934b\n(cherry picked from commit b2c8bc3590eb98a4539e3a647d18092039437436)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebde295e6863197b453e8bf672e1b6e93ce91339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebde295e6863197b453e8bf672e1b6e93ce91339"}]},"branch":"refs/heads/stable/2023.1"},"ced3bcbe173b0d62a05a12f319edefef542db1c4":{"kind":"REWORK","_number":2,"created":"2023-05-24 10:15:09.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/81/882781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/882781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/882781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/882781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/882781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/882781/2"}}},"commit":{"parents":[{"commit":"5975376ba951d6abb1e31cece2748e25823327e6","subject":"[Pure Storage] Add check for new error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5975376ba951d6abb1e31cece2748e25823327e6"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-04-10 10:53:32.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-05-24 10:32:12.000000000","tz":0},"subject":"HPE 3PAR: Few issues with new WSAPI (of 2023)","message":"HPE 3PAR: Few issues with new WSAPI (of 2023)\n\nIn recent HPE 3PAR WSAPI, observed issues with parameters\nlike snapCPG, localSnapCPG, etc.\nDue to this, few operations (eg. create volume, create\nremote copy group [RCG]) fail.\nIn order to work with new wsapi (of 2023), performed minor\ncode changes in 3par driver code.\n\nCloses-Bug: #2015746\nChange-Id: I98e021594c59dbad1723597e38aeb4c0cec6934b\n(cherry picked from commit b2c8bc3590eb98a4539e3a647d18092039437436)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ced3bcbe173b0d62a05a12f319edefef542db1c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ced3bcbe173b0d62a05a12f319edefef542db1c4"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
