)]}'
{"id":"openstack%2Fcinder~884315","triplet_id":"openstack%2Fcinder~stable%2Fzed~I91c81502f72a47a4584c300e2d918ee2b376fbfa","project":"openstack/cinder","branch":"stable/zed","topic":"lightos/zed/active-active","attention_set":{},"removed_from_attention_set":{"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2024-04-29 11:21:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I91c81502f72a47a4584c300e2d918ee2b376fbfa","subject":"Allow lightos driver to run as active-active","status":"ABANDONED","created":"2023-05-25 06:10:06.000000000","updated":"2024-04-29 11:21:12.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a1f36c9b16d5cb1ac156e84fea417543d9922cd","_number":884315,"virtual_id_number":884315,"owner":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-25 08:47:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"disliked":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":1,"date":"2023-07-05 10:55:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-07-19 14:58:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 06:12:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 06:18:35.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"state":"CC"},{"updated":"2023-05-25 06:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 06:39:49.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-05-25 07:24:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-05-25 07:59:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-05-25 08:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 08:57:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-26 06:26:12.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-05-27 04:06:38.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-27 05:40:36.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-05 10:55:22.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-07-19 14:58:42.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-04-29 11:21:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"13b75ca1f6a680a6d000b0e7e0be39c8b701d174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2023-05-25 06:10:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"521e670c87eae5358f8362db1da583076fbf98d7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-25 06:12:07.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-884315-1/ns5_nfs : FAILURE in 1m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-884315-1/ns5_iscsi : FAILURE in 1m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"44d331905431ed9728462bbe8d3b182dcd7965c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 06:20:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12fd0fa9aed94592940fdef00ce245f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cdc66c9e71947519075fea5bc458b93 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9ef757ce39547a9be4b6eb3910a1d69 : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cd3840a1f48229b123f8e896adc0c0155afbda2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-25 06:39:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/15/884315/1/check/iscsi-driver-alletra-6k/e523b9a/e523b9a.html : FAILURE in 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e5a5a27789ebfa3eb7681477db2daaf9f0eb0bce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-25 07:24:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/c2026a82c5b547b59f680bf9ddc6adeb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1f04b311ea6e483ca34e447bfb1826cc : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4f70a3b86096f9eee89c61151bd6837ca97c310f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-05-25 07:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/15/884315/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a8932f4/ : SUCCESS in 1h 48m 18s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/15/884315/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/15f293c/ : SUCCESS in 1h 26m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/15/884315/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7e4aa18/ : SUCCESS in 1h 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"68cc74dd7674f3b38c9250218e1c5175c6e16a4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 08:47:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755e0a17d6594e8a80ad348be8063f5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ae9a4bead1441e8162a23738cbd7cf : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4779d0025ca04ab688e66e9f6101becc : SUCCESS in 12m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fffe06d8d3354c318ea812f03504d4cb : SUCCESS in 10m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b1bc8f23e5d4e51a36cdaa62249a8e7 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a01a43cc02414ab7da7ac8efecc8dd : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/d51af6411b0b41029ca4bc0e50743143 : SUCCESS in 43m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6fc0a1fe4644b44a647ba467a50e390 : SUCCESS in 1h 37m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12aa09c966724d51823a9a711fec399d : SUCCESS in 47m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1059c0fc7c14e3b844925924c1cd03a : SUCCESS in 13m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a858a5dffb794ecb84930f4ac93c6ce6 : SUCCESS in 4m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0107c0a6d61f42a2b271c14ce6f3a511 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42e7223070314a859f4ad0aacfbfe377 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad9250a3d18846ef8d9909b262d99aeb : SUCCESS in 8m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1915f7931320447b94a57b1e5b423851 : SUCCESS in 31m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a324be696d484d89adc9a4325ab37b4e : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd1533ea463b4451aa1a89772a2e53c6 : SUCCESS in 1h 32m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/13358f1292a34e5eb82cff9ace81b3bb : FAILURE in 1h 40m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f99045e633b4cd1a66a1a9f3035eddf : SUCCESS in 1h 45m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1ddd3fb08234cf88e2f826222ce9943 : SUCCESS in 1h 02m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1145be266b08401e822a52a2f5c77e7c : SUCCESS in 1h 28m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/476219a835bc4be9b9ffb9f293efe527 : SUCCESS in 34m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa65f73b5a004886b2116043fc7eb61e : SUCCESS in 1h 12m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be6ff125761149cc9e623369ba405284 : SUCCESS in 2h 31m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01772f3a2a424fd48f1abc5858f97859 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1dc7ee5d94809babe720932460584ee2e75cd42e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-25 08:57:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_05_2023_08_47_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"8d9fbfe9562d4481c37a6b4ea3bac842c863cd4f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-26 06:26:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/884315/1/20230526120956/NEC-V-ISCSI : SUCCESS in 2h 46m 30s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/884315/1/20230526120956/NEC-V-FC : SUCCESS in 3h 16m 13s","accounts_in_message":[],"_revision_number":1},{"id":"44493743e425041530d45bf7c10d230ceb2085e2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-27 04:06:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/884315/1/20230527055427/NEC-ISCSI : SUCCESS in 3h 12m 45s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/884315/1/20230527055427/NEC-FC : SUCCESS in 7h 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3a9b4034ae042fafdc1d12b8b8b729bc94e890db","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-27 05:40:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/15/884315/1/ : RETRY_LIMIT in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"391fd68129238753d8f24ab05d84a01e5f10a941","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-07-05 10:55:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"708a15ea8c5e1c4a74724411ffdab2e0ea0fcbeb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-07-19 14:58:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a1f36c9b16d5cb1ac156e84fea417543d9922cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-29 11:21:12.000000000","message":"Abandoned\n\nstable/zed branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/zed if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"563b8a02186c9a889e3a9e10e03eb37b944cf4cb","revisions":{"563b8a02186c9a889e3a9e10e03eb37b944cf4cb":{"kind":"REWORK","_number":1,"created":"2023-05-25 06:10:06.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/15/884315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/884315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/884315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/884315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/884315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/884315/1"}}},"commit":{"parents":[{"commit":"51dbd1e8061012bee4baca8aa60bf16a459279f9","subject":"Merge \"Reject unsafe delete attachment calls\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51dbd1e8061012bee4baca8aa60bf16a459279f9"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2023-05-22 10:27:27.000000000","tz":180},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2023-05-25 06:09:51.000000000","tz":180},"subject":"Allow lightos driver to run as active-active","message":"Allow lightos driver to run as active-active\n\nChange-Id: I91c81502f72a47a4584c300e2d918ee2b376fbfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/563b8a02186c9a889e3a9e10e03eb37b944cf4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/563b8a02186c9a889e3a9e10e03eb37b944cf4cb"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
