)]}'
{"id":"openstack%2Fcinder~884316","triplet_id":"openstack%2Fcinder~stable%2Fyoga~I91c81502f72a47a4584c300e2d918ee2b376fbfa","project":"openstack/cinder","branch":"stable/yoga","topic":"lightos/yoga/active/active","attention_set":{},"removed_from_attention_set":{"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2024-01-26 20:35:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I91c81502f72a47a4584c300e2d918ee2b376fbfa","subject":"Allow lightos driver to run as active-active","status":"ABANDONED","created":"2023-05-25 06:11:03.000000000","updated":"2024-01-26 20:35:31.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"daec6f475965a9195910c15d0b812778387dbba8","_number":884316,"virtual_id_number":884316,"owner":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-25 08:35:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"all":[{"value":-1,"date":"2023-07-03 10:37:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}],"CC":[{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 06:13:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-05-25 06:18:24.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"state":"CC"},{"updated":"2023-05-25 06:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 06:36:43.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-05-25 06:46:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-05-25 08:35:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 08:40:00.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-26 13:11:36.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-05-27 05:44:20.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-27 11:24:59.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-03 10:37:36.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"}],"messages":[{"id":"0264ff858a6db9246f6f8f7791307d81216fadf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2023-05-25 06:11:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db3f9573df739c6053d6f198e47b03af00a9db69","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-05-25 06:13:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"81fb4572f2b38f3eedeeca650880dba22d7bccc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 06:23:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a56be6b453c24c9784d05ab69cc37fa7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c7bfb543afd48da8321b4579464b7ee : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cace94c202e04475891631879be22012 : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66e3c3fc7652cf4aeae0c7048505b6ab47248ef0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-25 06:36:43.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/dadcfce3e3bf4a31824e5f7b399009c5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4897fae01bc4476d967dd034ac749104 : FAILURE in 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a2745ef795177931ab0a42b125943571a090d7f5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-25 06:46:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/16/884316/1/check/iscsi-driver-alletra-6k/5bb4540/5bb4540.html : FAILURE in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4c31ce16b87d136dd68c95459962de81ae427e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 08:35:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ddb526dc1b4a269ada5b1681153676\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdbb2918581d4124969741f8d957c6ec : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1548fe7a0c2e43249776207d478f712f : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/614735380e5349349214d009cc93ee31 : SUCCESS in 13m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fae6b664404b01a830256181214c5a : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/21d4a08a9f7b47fc8878f6fda0e52117 : SUCCESS in 1h 03m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c90e524f540b4144a2ecde4996c1f4f3 : SUCCESS in 1h 22m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4e815cfd75b4c2ea745494866217ea2 : SUCCESS in 42m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ced7310bd01243e2adb80fa3b30b73ac : SUCCESS in 13m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b99dfbb9e0a4b23913406b9e7e53305 : SUCCESS in 4m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e29fc39adf44653a9906dceaf6da957 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b95dce53a4d343db8675bfe9d2d4f664 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2f24e1466ad41068e411d92bbc922cf : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/72a39f45de944db0b8bfb1ca92abbfae : SUCCESS in 32m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee82be8e1a1e409db4cd7ff85b18c47e : SUCCESS in 4m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ef161f9c088f46bab6624d21b07b6b00 : SUCCESS in 1h 29m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5761b943e544dce9a532af7457e3f1f : FAILURE in 1h 36m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a996283768d492da6dfe92e565fe2d1 : SUCCESS in 2h 16m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89b132cef7c8467990b1e99f925c2790 : SUCCESS in 52m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3cb9771211b1483bb17705da729e785c : SUCCESS in 1h 25m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/90bc91b8c5ec4599997a92f55ba1c2b5 : SUCCESS in 40m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ee2160e3a3594b588f4863f2d7f6b682 : SUCCESS in 1h 01m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a086f4dec3b34e08b76b8ac1634dd13b : SUCCESS in 1h 55m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9694fd7282ed429681d11f16df692533 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3efc82be09c9f00490dab1b74ef4cfd399ce38b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-25 08:40:00.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/25_05_2023_08_35_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"65aeff181abf73708d9d3d0394162c8fe66d85ba","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2023-05-25 09:34:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"993b609d1181d72a878b791c0197c2d08d230194","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-25 09:38:45.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/25_05_2023_09_34_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"bb24dec1cb49ecd3c1b70ad9121f6876fe141570","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2023-05-25 13:17:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a8da8b7dec500221e680709700f6fab7290e7e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-25 13:21:32.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/25_05_2023_13_17_20\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"3bc08ff1eb15e847d0f1da09dce05b4659b5f1a3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-26 13:11:36.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/884316/1/20230526152615/NEC-V-ISCSI : SUCCESS in 2h 56m 38s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/884316/1/20230526152615/NEC-V-FC : FAILURE in 6h 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2e42fb2a655d6776449d28ae7f7d043ebe295711","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-27 05:44:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/16/884316/1/ : RETRY_LIMIT in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ed0c2680519b6341e3d7b0cebb5b69c25828a8f3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-27 11:24:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/884316/1/20230527130642/NEC-ISCSI : SUCCESS in 3h 12m 50s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/884316/1/20230527130642/NEC-FC : SUCCESS in 7h 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"dfd1d5a2016637f8cc07a2ede2650c9cfb91f804","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-07-03 10:37:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daec6f475965a9195910c15d0b812778387dbba8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-26 20:35:31.000000000","message":"Abandoned\n\nThe stable/yoga branch is now in Unmaintained status, and is in the process of being tagged end-of-maintenance and deleted by change I99a8120217aa. If you want to continue with this patch, please propose it to the unmaintained/yoga branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa43a4e4e1385c4f072e2e6cdb4b2cd5c88bdd7d","revisions":{"fa43a4e4e1385c4f072e2e6cdb4b2cd5c88bdd7d":{"kind":"REWORK","_number":1,"created":"2023-05-25 06:11:03.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/16/884316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/884316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/884316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/884316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/884316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/884316/1"}}},"commit":{"parents":[{"commit":"92e8a3df46e77878d79575b67dc145004c4fb06a","subject":"Merge \"RBD: Fix total_capacity\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92e8a3df46e77878d79575b67dc145004c4fb06a"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2023-05-22 10:27:27.000000000","tz":180},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2023-05-25 06:10:43.000000000","tz":180},"subject":"Allow lightos driver to run as active-active","message":"Allow lightos driver to run as active-active\n\nChange-Id: I91c81502f72a47a4584c300e2d918ee2b376fbfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa43a4e4e1385c4f072e2e6cdb4b2cd5c88bdd7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa43a4e4e1385c4f072e2e6cdb4b2cd5c88bdd7d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
