)]}'
{"id":"openstack%2Fcinder~884671","triplet_id":"openstack%2Fcinder~master~I080e846c34ad91c279362f16b3e9de045776fb05","project":"openstack/cinder","branch":"master","topic":"remove-focal-job","attention_set":{},"removed_from_attention_set":{"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2023-06-01 08:29:11.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-06-01 04:06:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I080e846c34ad91c279362f16b3e9de045776fb05","subject":"Revert \"Add tempest integrated storage job to run on ubuntu focal\"","status":"MERGED","created":"2023-05-30 09:47:40.000000000","updated":"2023-06-02 17:35:08.000000000","submitted":"2023-06-01 08:29:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884671-remove-focal-job","meta_rev_id":"d26a903587e5c88b463819819505eb3ac7f72281","_number":884671,"virtual_id_number":884671,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 08:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":1,"date":"2023-05-30 10:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-31 11:33:24.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-05-30 13:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-31 11:33:24.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-05-30 13:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 10:10:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 10:13:59.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-05-30 10:25:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 10:29:18.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 10:55:19.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-05-30 11:07:21.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-05-30 12:23:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 12:30:50.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-30 12:48:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 12:58:33.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-05-30 12:58:35.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-05-30 12:59:45.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-30 13:00:00.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-05-30 13:06:27.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-05-30 13:14:49.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 13:19:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-05-30 13:32:22.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-05-30 13:39:52.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-05-30 17:55:12.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-30 22:37:52.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 11:33:24.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-05-31 17:16:56.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-31 17:50:57.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-05-31 22:35:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-01 09:16:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-06-01 20:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"89abc9567cd7bc167339b71324e38896c6cdad56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-30 09:47:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a87bcf6e3b507bcafb5425f942cbaa8c1fa5097e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:10:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d210ab36b047d1b8869cdb48879cb5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69fc3d81c1c5422e8221b32b6cb5c088 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba0a4bc692844057abe125e82f8d8f22 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58edadb93430423eb31af44292eac2d5 : SUCCESS in 21m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ec5e61d3557bc2ea801135f68d0ffa0c37ed07a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-30 10:13:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"626d87ec476808a28fbba1c2090ad499719e4861","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-30 10:25:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/42dd836b09d24456b4ac7c3641a371bd : FAILURE in 36m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b25deb1406e2ea357e7fc56646dd1870724a0bd5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-30 10:29:18.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-71-884671-1/ns5_nfs : FAILURE in 40m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-71-884671-1/ns5_iscsi : FAILURE in 41m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"99bf24b84a54038360038edeb0931c39afa8ea7f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-30 10:55:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/71/884671/1/check/iscsi-driver-alletra-6k/6cd4598/6cd4598.html : FAILURE in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d058a5c40f36cde30c0b70a336c6d9c5c278019b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-30 11:07:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/75fa28b7d29c43ad8fa6323e5b0c80eb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dcc14631bd3148c6a66048f86394505c : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"527021ab00cb8615a8b0fbe6465a1be54d95e095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:23:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972cc2d6ee9e4aadb88e8508a55aa715\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670da3631d2947e4ab4a0ad67a822c57 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b06b28037f6843e3b5b115eba9ee90a5 : SUCCESS in 12m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76a8120971174e2ebda7003e802111cd : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93509834c8c64fc480a6f83297672ccd : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f3309a3fbbe4a6f9db3eb36e2aae4ac : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/962a534dc015494dae1b885d3cd30b8a : SUCCESS in 1h 03m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d47e514c98743018022a386d9a6f3bc : SUCCESS in 1h 33m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c1c93a94684dacafbc975efa1e0a5f : SUCCESS in 56m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/31e351f5a309417c8ec52b00e1e0ff6b : SUCCESS in 16m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/25c98bdbab96480c80630858b3512edb : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c12e15c16fcd4cbfa96a929e2cdc042d : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e55d881774a64f4384d2b79827831ffb : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/576e7b87d1684045be050bc3015a30ae : SUCCESS in 5m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f921623d5cbc4736894b08086b54f78f : SUCCESS in 39m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a698a1c508ea417b9539c59f5d1a179b : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ff33a3201d44afc8594d2aaa32c9e83 : SUCCESS in 1h 53m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/71b2528ca21b4636a222e91f421d94ce : FAILURE in 2h 10m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e52584a3c29840d8bbaa76e48335e5c5 : SUCCESS in 2h 24m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f181e67b8fcd45d7afa42f861a9e2dbb : SUCCESS in 2h 15m 08s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/786dafb022c145d2a09833e3e9e36f24 : SUCCESS in 1h 21m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f759cb8e8bd4dad825c99fe603236af : SUCCESS in 1h 33m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2ca9725346ba4c2fafca37e3e88b6803 : SUCCESS in 34m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/207eefeb85d4489887d118401e124abf : SUCCESS in 1h 40m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9db0377e38c849e58e8103ab4b86b1d8 : SUCCESS in 1h 27m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2c278c7100d417989a477dbdb558269 : SUCCESS in 2h 32m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fde496ea462f48c7b89b673e4640e7fa : SUCCESS in 1h 02m 47s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/9bb622b73a15413cbeeb85640279a35e : SUCCESS in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3db7fa921735a977b88b152fd3b5ce41d88d2ab3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-30 12:30:50.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_05_2023_12_24_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"af19bcf75822972c42b390e0e81903a8b70cff8f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-30 12:48:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerFlex/744fa96/DellEMC_PowerFlex/111 : FAILURE in 24m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"987ce59c961cd643b91482cd6155c6c8aa9354a4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-30 12:58:33.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"25d212b44f30fcaf46a3dd53d811129bceddd411","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-30 12:58:35.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"7da7d8ee500fe1878fe37a9d547b3c310ab0d701","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-30 12:59:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-884671-1/FC/ : SUCCESS in 3h 11m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-884671-1/iSCSI/ : SUCCESS in 2h 2m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2df74b4bdf4d91f0a007e6771dda5f954f110356","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-30 13:00:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_iSCSI/490ae75/DellEMC_PowerStore_iSCSI/66 : FAILURE in 27m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_FC/d6c2320/DellEMC_PowerStore_FC/71 : FAILURE in 35m 45s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_NFS/345a3d3/DellEMC_PowerStore_NFS/65 : FAILURE in 29m 12s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_NVMeTCP/2a27ac2/DellEMC_PowerStore_NVMeTCP/66 : FAILURE in 26m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"59bc371e599fb5ea8492ca442c91b33920c50963","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-30 13:14:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerMAX_FC/e52c509/DellEMC_PowerMAX_FC/88 : FAILURE in 50m 32s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerMAX_iSCSI/1631676/DellEMC_PowerMAX_iSCSI/84 : FAILURE in 45m 55s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"bf4f49a660255211c74805f571f9542c32b599d1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-30 13:19:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-fcp/2730d62/ : RETRY_LIMIT in 8m 34s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/3b2e497/ : FAILURE in 53m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-nfs/d25dd33/ : FAILURE in 54m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4115a7e/ : FAILURE in 53m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe1ea3f7286a236d66bddcfc3eaa0061f294e547","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-30 13:32:22.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"d6fb33470d038fd9af81eb9055a33b4c2871f160","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-05-30 13:39:52.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41f70ae77de1c45c5334b5afa87938f57b7207ae","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-30 17:55:12.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/71/884671/1/20230530184741/NEC-FC : FAILURE in 8h 06m 49s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/71/884671/1/20230530184741/NEC-ISCSI : FAILURE in 8h 07m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3a530609a17d31de6df8cd0f53425e031a5f63a9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-30 22:37:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/71/884671/1/ : RETRY_LIMIT in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b7cdb2feae8beed99b1c717b98b295baf9162523","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-05-31 11:33:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cdf08b9301a0e2d5955acf2a17e785ee1d6f618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 11:34:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f880fb2c192137a2160834ab34131237400165e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 13:43:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a1b8afacdac4c2eb67642ed3581f6bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808945911bca4329b3a20968bbcedd85 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5be6a83db2f41f0bfc7bc06d9fcc001 : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/335c956126be4bf798c5bde39936c0a8 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/addf099783c04789a1027fb6cbd5e7c9 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16473717ee274592b2548060308ea6cb : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/876f7340440342f2833686663a76f7ba : SUCCESS in 1h 09m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8fbbdbec33348d9a641e2b9dc34d3f6 : SUCCESS in 1h 32m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61d7ce0438e24be1b50146a892a395b9 : SUCCESS in 1h 01m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7c718097bf942b3a72d7a9cfdb9631c : SUCCESS in 1h 04m 02s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64a7fed1186e4b42bb33351513ed9eb6 : TIMED_OUT in 2h 04m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d72be27142942d0999b25d4288a403d : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a44ec0bd5f652dff4c0bcb42aaa0997bd857db4c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-31 13:52:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb5c13c385ea1fa3aad0f11744033c5434bfb221","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 14:17:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fbfa589578d4ddb8a68a231f9a28760\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/866cf8c033f54b468faebbf71b18a2d3 : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a234c7700f04406b3e8f64ca9198166 : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/594054386a554f51b4a4ee41a5e48556 : SUCCESS in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89fe78dbd2682f6f1a85fa0efe9815dd9c3aa6c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-31 14:31:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b66bd2e2a8484e96a50839665a39fdff : FAILURE in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c4dde6fe8a073e63ea6612fc5db01fd738e260ea","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-31 15:47:16.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ec8d430905894bb4a7d24556b2b3fddf\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1fccd877db394cd2a431eaabb1cffd50 : FAILURE in 1h 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"57ff6c3c675f2a24e6976ecbe6716573f9b21ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 17:00:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aadcb129bf6a4e0193724ac2b72d099a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98b1c2efad24fff8a492b1976ed92c3 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b71af4858bb44db5b12162939ee22d19 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5249938576405c87a0ae0cd068b835 : SUCCESS in 11m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/794510aae98745c99bc9b6d188b27040 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e090e02a25bd45db9a604a404784da83 : SUCCESS in 15m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/c112411e4be947309625e0b357320a15 : SUCCESS in 50m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f08d0d8778841528633654bb445417c : SUCCESS in 2h 00m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49245c882a4147d2b5bfe814643ac0d7 : SUCCESS in 56m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/48e32b9313b44243a849b6acf8c07b79 : SUCCESS in 17m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48ef44413b5f4be5b17dd23329ae4774 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/49ef09204bdf4a0dbbb81f8edd1c3ebb : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/526d368543c1424ca4a4b40fb449be85 : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/989c178c18784f24aa90d6cf52e2b489 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/807ebed448054a8baaec9aae698f1c5f : SUCCESS in 34m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e27d2a63dcf147ae8772d3f0b9d329da : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6a307f95d73b4d84b5ef8c0c130c8dc6 : SUCCESS in 1h 57m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b72a1ead9c634d0da094de656293d589 : FAILURE in 1h 52m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df4cb9925e444a0ab94b960249912fb1 : SUCCESS in 1h 44m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3db0d388ef9a44459a8fe6508fcb8b3d : SUCCESS in 3h 00m 01s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/135b8228f2a14ea09432bbadddcfb57e : SUCCESS in 1h 12m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17dd251b5cc341d18e81fcd719bedfb6 : SUCCESS in 1h 37m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2af396cb66b940cc839e265d17eecd65 : SUCCESS in 32m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/72fbea48c8344facb2be6462adda5878 : SUCCESS in 1h 08m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/956bb27a6b834005beb73402040083a2 : SUCCESS in 1h 22m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/553994f267bb4d2c815a9fa7bf68d354 : SUCCESS in 2h 43m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c6ce09c77204a7383c9ac40c13d35ef : SUCCESS in 1h 11m 01s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/117db0dff6384b1b8a09988509222d6f : SUCCESS in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63e13bbc1ea9a52aef5393c0a0ceb1fec6255aa5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-31 17:00:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerMAX_FC/b73bc72/DellEMC_PowerMAX_FC/89 : FAILURE in 5s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerMAX_iSCSI/539d220/DellEMC_PowerMAX_iSCSI/85 : FAILURE in 5s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"4b1c098c036bdc3c1e9251d0f33e0b2d8ca3b7cb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-31 17:01:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_iSCSI/54d6eea/DellEMC_PowerStore_iSCSI/67 : FAILURE in 6s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_FC/bdb90e7/DellEMC_PowerStore_FC/72 : FAILURE in 5s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_NFS/70ed5d9/DellEMC_PowerStore_NFS/66 : FAILURE in 5s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_NVMeTCP/206d0e5/DellEMC_PowerStore_NVMeTCP/67 : FAILURE in 5s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"b0e472837b859d15b23f584a10bc43aafbe6106f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 17:01:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a56d75ef3a9fcce5e81833f1d54674a4405a1be6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-31 17:04:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerFlex/f73251a/DellEMC_PowerFlex/112 : FAILURE in 3m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"e2cb1f05b8e6db6b39446358347ae09b1d7a0ed4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-31 17:10:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_05_2023_17_02_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"e17963c6612db85f811db30f3dfd3d50b66d4eea","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-31 17:11:58.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"8ae2231870cd80e90e1d4fff928b0f28227051e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-31 17:16:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_Unity_FC/b85d655/DellEMC_Unity_FC/199 : FAILURE in 16m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_Unity_iSCSI/efbaeb2/DellEMC_Unity_iSCSI/198 : FAILURE in 9m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8113d7009f9ec7ec33ac9ccb81ee7d5da5a30bd8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-31 17:45:11.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"90329faf5c86e18c24c572e53ef636abc75361fb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-05-31 17:50:57.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"9c77eb84a630b6a9d8333f3c4ce5d0b2a987558a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-31 17:57:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-fcp/54b0535/ : RETRY_LIMIT in 9m 31s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/8deda35/ : FAILURE in 56m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-nfs/53b3d61/ : FAILURE in 56m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/66a57d3/ : FAILURE in 55m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ce0af81759ecb974a122c2791ea2913d9270c29b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 19:05:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/faea9a25b26341d58cecd706ab84d8ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b138adbd8c44c29ac9a77f3598b3dcd : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/563d4d6721f44e3d810ae688a205c20f : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc813672dddd45859eee3510bf807b8f : SUCCESS in 11m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae31cf329a6c47999cd7bd3b5766e02f : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a768f3a1ef064dfeafcd7048da739609 : SUCCESS in 16m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/5868a7db50254b12a2946ac7077e69bf : SUCCESS in 1h 11m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58d853fcd40c4559adfc0cbf3fb5e089 : SUCCESS in 1h 58m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f869f762796a43f897575f12d2f01674 : POST_FAILURE in 1h 03m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aab52d17c31d4fd2955f5c118e5fc3f8 : SUCCESS in 1h 00m 30s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c68ca873c7c545bc86379dd058228855 : SUCCESS in 1h 51m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74a9c0584e7149e4bf105d28f6012182 : FAILURE in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c09112ccbb5fb5f7f207a1fc5ed8ae3a344ae2a1","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-31 19:53:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e16db2f38e280f4531c4849f73773a319a7a9fcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 20:19:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70694e81ed8450eb05f4dd5920833df\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55916bb116994f8082316dcf141eb716 : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04a9b35a455649308b0a75121145d6a5 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49a8614c02ce47e1816b59cead718f7b : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57a6f009200baef332f1c8eb4e0af4a24ce9e905","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-31 21:21:42.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/450130de5ede46fdad8261c2b407cf19\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ec7bd2fa1ef14a9ca3092331356606aa : FAILURE in 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"786216bc26e83e126a8cfdcb214b40a02a36e006","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-31 21:27:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ad5589143fcd4f7bac401f2820ce2a6c : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6ef27c277f5fe226f4f032c1f415fff07605e432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 22:27:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd1abf192d014c30bf3b182afa5f1c25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9452139547b4af58cd0af241e6282ed : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ec9cc388444b76a6f92d8c50986d47 : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53e12af23d034e95bc8f9368708e8adc : SUCCESS in 15m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/895f4269db8a46b088acaf9a4cacd939 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca39eeb22fa4879b926b19f17b6e9b7 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/fbe9cc02e25d4105be349f9beaf63b62 : SUCCESS in 43m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/209c651d0b144f29a343b541260570df : SUCCESS in 1h 24m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74e0eeb3e55747fcb3233a4cd139ec79 : SUCCESS in 50m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/070388333f33499fad4adb107d379da0 : SUCCESS in 15m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6c2036a1c404d5aa9cff418ffccc249 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6dc3cce6ced744a99624479cca879bce : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8addb11323e44d2ebbd21ac826bf614e : SUCCESS in 6m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a5c5c466c2a4ca69b343d0d19885a5b : SUCCESS in 10m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3992f6c3218542bfb9857c343ede6c0a : SUCCESS in 41m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d3add934f7de47f3bdd746ddbdf6efb9 : SUCCESS in 9m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/198ebcebc54c4665b79ee9382c9878d1 : SUCCESS in 1h 53m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d1fb8693f4f4452cacb6764f480bb091 : SUCCESS in 1h 52m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e0af8e7820504aaca144877639c5a24c : FAILURE in 2h 04m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/52c1ec63eeeb4b5eab8a525d3db54a0f : SUCCESS in 1h 39m 00s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f886d5daab224daeab6cd1733e9cfc88 : SUCCESS in 1h 08m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de762bd277ff4f598839f1b9b7e5fb9a : SUCCESS in 1h 28m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bc5ee356ed814e50b12df7cde5c21445 : SUCCESS in 49m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/71f63934061b48c4927d228fd5b2e520 : SUCCESS in 1h 34m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6de161e0917b4dbdb4b0e93bc597a36e : SUCCESS in 1h 22m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a88d5a12519b456d993b4e822a4127c4 : SUCCESS in 2h 30m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f32b3fd151f4fd2bd2f1ade266fd025 : SUCCESS in 1h 06m 32s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/97f46f2bdb4a4f9996be488c972ff392 : SUCCESS in 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"564e74d08b1b4a5b36bc35bc9632300373812eb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-31 22:35:41.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/71/884671/1/check/tempest-dsvm-full-focal-py3/e7ad344/ : FAILURE in 1h 21m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/71/884671/1/check/devstack-plugin-ceph-dsvm-focal-py3/83f1f04/ : FAILURE in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5b220271be7c4e183b9f0ba2d868b294d8517c22","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-31 22:39:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79f59351e0b1bb5d489214008a6e5294e38687be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 23:05:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbe762f36634574816a9d3128494d8f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cacfb5cf218c43d8a9ba0ee5cb494d0a : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c0d266295194628b72114689812b07d : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1761fb488c44397860262626aa68c8a : SUCCESS in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a2501eda2aec5d2ca0e796c35243797af206839","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-31 23:11:39.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fa27f026e9fd44f7904e0537d5c0ff19\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ebf8a6edb8bb412b83bfc9052ed1c742 : FAILURE in 28m 49s","accounts_in_message":[],"_revision_number":1},{"id":"afb50bcb06a85de818be14d5a5f6621930e099b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-01 00:03:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/10ba27482f7445b481313b72fb8a45c7 : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"aba47efd3c54e48fc25d706cef8a008633d45a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 01:00:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f3e6dfd9545407bb9d8a49a68b7b7be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc21b75ff0b4340b60b8f0c8b4b9ab2 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6017e56d1b704aa093132d47e67776ab : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f95c93a20cb47ceb98431e566ad6746 : SUCCESS in 12m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f4fc671313a4b0d841558b52f2b28b6 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1081e43c914dfbaf767b46d905c19a : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e85deed6b853460eba130ecb1caa138e : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/082d8a486959405481b68af3c6c8f4c5 : SUCCESS in 1h 35m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93ecace4343f4353bff7a9220cea3cd8 : SUCCESS in 39m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/38c075832e104a3c82cb4d5bd77edf10 : SUCCESS in 13m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bc79558c8b704c5a81abd30c6c322a3f : SUCCESS in 11m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6860294ea6874c3aa494016bd6f4edde : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7d8d7c131964f50bd4dc28dc13c6b4e : SUCCESS in 5m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d8f5691c34b46b4bce431027496664e : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67c18ea3277543e2a099e9dca867cbfc : SUCCESS in 31m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/504e9a90734a477a93b50b18fac701ce : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae476b660eb94da4bb01da4817276dd1 : FAILURE in 1h 30m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/da4c55dd0f3f476a808cb7823cf3f50c : FAILURE in 1h 55m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/23e1bffab66d4f7bb9bf389599201c9c : SUCCESS in 2h 05m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f6b230dcd2c04b40bc3d48066d101f8e : SUCCESS in 2h 17m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/59c67aaf9c7341b5a99debaec1725238 : SUCCESS in 51m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/394b551f5c794fd3aeeffdb4582dbed7 : SUCCESS in 1h 44m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6988981787bf47d39ed0492fd3577ec4 : SUCCESS in 38m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fe754b03aab640db81d80107593027eb : SUCCESS in 1h 04m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/221a825224724109ac4310f261594055 : SUCCESS in 1h 30m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91152ae7b31a49c49deb23be3337e989 : SUCCESS in 2h 14m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6f3061416e448afb6211ada6e819972 : SUCCESS in 1h 00m 28s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/4aa7fa3d138a44e280bddb52853f529b : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09a24c14a40f92440074b9abf5b439fe0c933b99","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-01 04:06:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2454b8f045b886f734dce5aa0a8ee05d81561c7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 04:32:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/831c29d0401544b4ae74a4b20b982f3b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7314261d2e64b668202257e50e609f8 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/230adfee8b7649e1a843ac2cce4db122 : FAILURE in 24m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e62bceaee7d046498ca45e0fac41b0e7 : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b672ef92b311082d22a8121e041a89347c7bcce6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-01 05:03:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6075b70be78b4e18829d0190dd96d0f4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6ddd09c0fc05420a96162094a4dd119d : FAILURE in 54m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c6e7a99f997add4d529758b399466f440969e3bc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-01 05:25:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7fd68e8805b940f9aa61c7781fdffaab : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":1},{"id":"684eb2206235102d67f173eb010e433ed269379d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 06:28:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c855914b1a479298f74418ad5e2fcd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b36d451e8c45019cd5dea1674222dd : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc350eca39d043dbaef51bec96f95df3 : SUCCESS in 11m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2640173a2784595bb58f6a04231621f : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f8fa211931344dc81c7304ac7ab2c5f : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dba8ea739124201b3d987db898402c8 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/0b408e112f2044fbbb873a7a510a0666 : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b364b626d4ae4c98b49932cceee508c7 : SUCCESS in 1h 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe30988818ab4f1c8d531de90456b805 : SUCCESS in 45m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f338889e01714c7b837e22255f33b82f : SUCCESS in 15m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7cb13159b6348f09b00f9033cc864a4 : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ddf47c36b4a24371acad37d260b4eb3b : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d56a06b617e49838ed8843f864c6dd2 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d19deb0dbac9416682437c16c1881810 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2758528bfb645e6b426c4cfc579b909 : SUCCESS in 36m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db1945fea7bf48669ca81697e96214c9 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9618946fc904981b3e15e1270ac3170 : SUCCESS in 1h 29m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/16dbdc12b3ed47baae29d58f7fcc4f0c : FAILURE in 1h 37m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b50ca13c1814fc58630c868e03ea4d3 : SUCCESS in 1h 37m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7682919ac4c246399c2d1da0efbae58f : SUCCESS in 1h 31m 34s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8d3101ebbd84d12b45f1da82796e3b3 : SUCCESS in 59m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/57f876fa09bb4d08a68dbd042a75d7ec : SUCCESS in 1h 23m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c9fc59295d4496fb5d03a1b17887aa5 : SUCCESS in 30m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/74542bd069cb4e1297357626c4dc809e : SUCCESS in 1h 17m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e5ca11742d064d4988fcb7125b4af708 : POST_FAILURE in 2h 00m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a24ee34ccc49464d8c19da33cf490fac : SUCCESS in 2h 20m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/553f2bb204804ad699d7d9bb0cc0996f : SUCCESS in 47m 47s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/98d17a9014c44c7f91a099373705ad82 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5617475e1ebd272ef5d1ba49ab9dc6f4ea1f8d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 06:29:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7006297634e176ea59a080b9c7a3aa94f38df473","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-01 06:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_06_2023_06_29_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"35bcf711323864208f65e9c0df4a2ce9cdc36c01","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-01 07:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-fcp/d7623dd/ : RETRY_LIMIT in 8m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/64096a7/ : SUCCESS in 46m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-nfs/0cf6b1d/ : SUCCESS in 1h 23m 32s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/884671/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9ee1293/ : FAILURE in 1h 20m 24s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"73f4b28b40cc3720f2ba83259eb8225441b81d47","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-01 08:14:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_Unity_FC/5888e88/DellEMC_Unity_FC/205 : SUCCESS in 1h 45m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_Unity_iSCSI/3a93fb1/DellEMC_Unity_iSCSI/204 : FAILURE in 16m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"502faa843d507a60f58d192cbe4c6c6f03121842","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-01 08:15:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerFlex/2897697/DellEMC_PowerFlex/114 : SUCCESS in 1h 46m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9f6f3e69753c54a72b12bc70f65ec1d61ce5185e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 08:29:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8ae3b4f4ae4d0880746449eec5aede\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec5827c7ccd4fc99e7a84f496606cf1 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37ce902568c749e7b3e58394587a7164 : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff63b77c6ddb42209a5f5b59b981e0bf : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08df8d48107e4cc68a77330f8286a9d0 : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2002bc1f54244021bec172b25d48b22d : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/ed3fb575bb3a4be097751ceaa2a06f70 : SUCCESS in 47m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ed21593e2634442b2b982b01b4501e5 : SUCCESS in 1h 37m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c30c610da1f41cf81fbf3a4eb316c70 : SUCCESS in 58m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c23744a2087e4207a38f135bffe01bcf : SUCCESS in 1h 07m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2c98fbd48a241d898c0b4c520ae72b4 : SUCCESS in 1h 59m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f150b6da7645448e95732b2cf831f492 : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d6b6d025e47f8ec72b03b11fdd3f3ae013ca0996","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 08:29:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c7c988eae8d65335a4395570bdcd7cc7b90428c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 08:31:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8a08283c9749448d6b773717359c7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984cdefcbacf43b299523825948d37d7 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5df50e99d02a064553c42c060a39bef9fade7599","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-06-01 08:34:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/884671-1/Lenovo-iSCSI/ : FAILURE in 36:58.74\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/884671-1/Lenovo-FC/ : FAILURE in 36:46.82","accounts_in_message":[],"_revision_number":1},{"id":"7496e3cc42c0fe349c3873fdecfbe1499e158aed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-06-01 09:16:17.000000000","message":"Patch Set 1:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1410/","accounts_in_message":[],"_revision_number":1},{"id":"5d9099f41a32cf8c70735c7aa6629fadb42bb2a2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-01 09:29:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_iSCSI/ee2e7af/DellEMC_PowerStore_iSCSI/69 : SUCCESS in 52m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_FC/0bf201a/DellEMC_PowerStore_FC/74 : SUCCESS in 1h 18m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_NFS/3962bbd/DellEMC_PowerStore_NFS/68 : SUCCESS in 58m 24s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerStore_NVMeTCP/fe4c8ad/DellEMC_PowerStore_NVMeTCP/69 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"1996617a2d95a34fc2b11f0256b47a79199fe1d3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-01 11:03:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/884671-1/HPMSA-iSCSI/ : SUCCESS in 1:49:15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/884671-1/HPMSA-FC/ : SUCCESS in 1:28:21","accounts_in_message":[],"_revision_number":1},{"id":"3448a1d181e294726318a26bc7a589dc052618d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-01 12:30:39.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/2d23a3598bae488a9b8172abfff320d8 : FAILURE in 1h 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8fadeba53069bf30975db9fd517b0c05a0a3fa19","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-01 13:00:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/884671-1/PVME-iSCSI/ : FAILURE in 1:59:17\n- PVME-FC https://os-logs.seagate.com/dell-ci/884671-1/PVME-FC/ : SUCCESS in 1:36:55","accounts_in_message":[],"_revision_number":1},{"id":"ce17df1877aa120907c753424772d775a7fb4690","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-01 13:25:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerMAX_FC/35ecfe6/DellEMC_PowerMAX_FC/91 : SUCCESS in 6h 56m 35s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/71/884671/1/check/DellEMC_PowerMAX_iSCSI/6b106f0/DellEMC_PowerMAX_iSCSI/87 : SUCCESS in 6h 02m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"7b24de94e04f5e577faddbf94e520d082c8fe4ac","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-01 20:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/71/884671/1/fc : SUCCESS in 0h 26m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/71/884671/1/iscsi : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d26a903587e5c88b463819819505eb3ac7f72281","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-06-02 17:35:08.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/71/884671/1/check/tempest-dsvm-full-focal-py3/43034e0/ : POST_FAILURE in 1h 30m 42s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/71/884671/1/check/devstack-plugin-ceph-dsvm-focal-py3/25af6e7/ : FAILURE in 1h 39m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8fee1813b4f39926bae20ee801aaa8d11d6044d","revisions":{"a8fee1813b4f39926bae20ee801aaa8d11d6044d":{"kind":"REWORK","_number":1,"created":"2023-05-30 09:47:40.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/71/884671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/71/884671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/884671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/884671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/71/884671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/71/884671/1"}}},"commit":{"parents":[{"commit":"77b6035e4fd8931f40370d0534245100128275d9","subject":"Merge \"Automate generation of backups api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77b6035e4fd8931f40370d0534245100128275d9"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2023-05-30 09:27:39.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2023-05-30 09:27:39.000000000","tz":120},"subject":"Revert \"Add tempest integrated storage job to run on ubuntu focal\"","message":"Revert \"Add tempest integrated storage job to run on ubuntu focal\"\n\nTesting with focal was required as part of the testing runtime\nfor the 2023.1/Antelope release, which is a SLURP relase [1].\nNow that Antelope is released, the development moves towards\n2023.2/Bobcat where the old LTS ubuntu release is not among\nthe required testing runtimes[2].\n\nMoreover, this job started failing consistenly in master.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2023.1.html\n[2] https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nThis reverts commit b3e7d363ca317acab86ae9950ed9f524815cbf8c.\n\nChange-Id: I080e846c34ad91c279362f16b3e9de045776fb05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8fee1813b4f39926bae20ee801aaa8d11d6044d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8fee1813b4f39926bae20ee801aaa8d11d6044d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
