)]}'
{"id":"openstack%2Fcinder~884896","triplet_id":"openstack%2Fcinder~master~I130a4fd6f2edd5628208774aa3f3b70ac494e476","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-06-01 13:43:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I130a4fd6f2edd5628208774aa3f3b70ac494e476","subject":"zuul: Extend c-p-ceph-tempest, make integrated-storage-focal nv","status":"ABANDONED","created":"2023-05-31 15:02:23.000000000","updated":"2023-06-01 13:43:30.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"789af5095c81e2ff403e818ba5423218325fd877","_number":884896,"virtual_id_number":884896,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-31 18:14:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-06-01 06:24:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2023-05-31 15:12:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-06-01 10:24:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-31 15:12:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-05-31 15:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 15:24:15.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-31 15:41:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 16:08:49.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 16:52:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-05-31 17:23:51.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-05-31 18:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 03:14:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-06-01 06:24:33.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-06-01 09:07:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-01 10:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-01 10:24:07.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"51dac5ff39f21c0ac8ceb62e6a9bd12a9389e2b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-05-31 15:02:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9852ff536f8bb4a3af8d3bf068e8e24408240bce","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-05-31 15:03:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdc055c94d3ea50e37e355a48b8ba61ecd265457","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-31 15:12:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ec651932d4bca4cd8edc1a582a1f12a3d514315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 15:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861a16e1f7c44e3b9f6bdc7c132e5837\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/403e08dbd2f24046861c4c84741b19bc : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6209782d9e0b436aae4506fe3b4ebdfa : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d2285d1fb0d4e21afea917d3f8dfd9a : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ab0c3c4e533c0f1cd675e41a98914acabfb5e6c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-31 15:24:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/96/884896/1/check/DellEMC_Unity_FC/32ffb6e/DellEMC_Unity_FC/197 : FAILURE in 21m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/96/884896/1/check/DellEMC_Unity_iSCSI/a36cc1b/DellEMC_Unity_iSCSI/196 : FAILURE in 15m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d64cbabd9c9fc61827989c3e3d03f691162e2746","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-31 15:41:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a5b93e6fd4ba44d3b830838e0c18090d : FAILURE in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"218e27c396409de369087c4e18640ec5b6a0a236","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-31 16:08:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/96/884896/1/check/fujitsu-eternus-dx-iscsi/89a2c0a : FAILURE in 1h 05m 48s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/96/884896/1/check/fujitsu-eternus-dx-fc/9a38c22 : FAILURE in 1h 03m 36s","accounts_in_message":[],"_revision_number":1},{"id":"acaee8daa8e107383a41055ce92969fd878422e9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-31 16:52:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/96/884896/1/check/iscsi-driver-alletra-6k/455e17d/455e17d.html : FAILURE in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b0458e5ceb34b618e88b50d4d719b4ffa7cee642","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-31 17:23:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/83b224f016d848299867eba7b86439d3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/11e73b2acbec489eb5de0b5876029a3b : SUCCESS in 1h 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d5eafbafc7ec6f06cd2878f4932c3004ca5c08a5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-31 17:55:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41d26efad4182fbea63748b874fc3927bcff52b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 18:14:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78d766f4f8324944be6da54b0effad02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786e6223cbfa4ae5b8474b1f8b0614f5 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00aca8e0de98432ebfbee141a66d83f5 : SUCCESS in 12m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca24ab4fcdb242f3be12b7f3205e4ec1 : SUCCESS in 14m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5854c3d34f564311a7a9fbc9fa5256e4 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff93c775c8734c4d96d9c1fb92c172fe : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/b7b6d4e4a6774de582419adb6c93fe59 : SUCCESS in 1h 00m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7675c0ec56b64235b7e7d59f9495a158 : TIMED_OUT in 2h 06m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba62487134e94ed595da504ecaf46b03 : SUCCESS in 56m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ca7f71d07a2498cbaec99b46cdae759 : SUCCESS in 20m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3effaece33884cc687c03b1599639134 : SUCCESS in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3452d8479dc414e9059bd4d7dfaaf47 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b466f08318ab4fc7883e80237a625104 : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14f32a4b41be47a79f2682713a0d78fe : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b069caa4990e4ac18f6741c8e92cd0b9 : FAILURE in 1h 10m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/911f22c179bf473d9cc007a57070027d : SUCCESS in 7m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd4127075aa7421e81395959452a95ed : SUCCESS in 2h 22m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1d2a783a9a7e491da1647519bac92b96 : FAILURE in 1h 51m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f8eb432d2bc4468e8d07973dc5397d42 : SUCCESS in 2h 34m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5b4620d7d95a45629c96e1deba4b92e5 : SUCCESS in 2h 47m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e61e668fac994284a22738fe6487db23 : SUCCESS in 1h 26m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/95c2c605bbbf488eaaad526b5d8305d7 : SUCCESS in 1h 35m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c59895bcbc84437b04dc2bef7b68fad : SUCCESS in 52m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b1d6f02ebe1a45738ac03b420a864f4c : SUCCESS in 1h 49m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5363a7bc4a2a4e02a30850c983497e7b : SUCCESS in 1h 54m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2adc76c2b31f43eca4d9a16abd4833fc : TIMED_OUT in 3h 03m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/229c6f8461d44a50a46f55df72264625 : SUCCESS in 1h 07m 18s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/801a2f9353174cfca6ef58836d670227 : FAILURE in 41m 31s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/47bd70c33a374e1bb0afe1f4d1207569 : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0957e103707c5d4b76d45ad3247cec06996366ef","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-31 19:31:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"113026fdb7827ffa1a47af2a1fe4fa39a890c346","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 19:57:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89fed7ab24be4292b6903aa81d203e57\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8766b6214964c38ae23818e6c4f1e7f : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41c7417fa65c43ff9493eba5fca71142 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c5d32bcb8e84c2480dec788f1b09c4f : SUCCESS in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a42685e5f507ab8d18e2ec18097fc01cfa364dd7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-31 20:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/8092303794cb4867a482bd8fa2e9aad9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/35f63c7cc4e04b8e8b400a9249b94e79 : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3921af135323055942842810bda4afcf2d33bbed","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-31 20:58:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a92705497dc14db69296bb9a9338004e : SUCCESS in 1h 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"14b135d26c5e8fd882dd7fb3009605e636fd1f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 22:43:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd54cf46d21c455ebc2bf894462ee8cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3395e4ba8c43cdae4cd4f6c619e680 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7fc3812fd40498cabf9ea26beed0c25 : SUCCESS in 10m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0971d68743d843d78c2693ec4e4d79ad : SUCCESS in 15m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299d30dd28d4466b865eb135e7ee8b27 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaec9163d5e6407587ad316fd1dfb1c6 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/7b3f1fbb51374151ad8d1b5201a18303 : SUCCESS in 1h 22m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7596faac58b1499495d77351b5f8d9b3 : FAILURE in 1h 22m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fe726d3b10a4615820bb1154d6f3ce5 : SUCCESS in 1h 09m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/60f00c72634441649101b7aff4f41ed4 : SUCCESS in 16m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9870ccea84dd4c5e94abfaae5807b28f : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/14aba41f65974a0595d71b0f5f2447ee : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/315bd16cfb584fd09a93aacdfd7b1423 : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe283183e559406781157ef2758a9863 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b165196214774197b219d5afa166e38c : SUCCESS in 52m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e9cbdde344e4fb2861f45d3d1eaa1c3 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/842b7d10034f4e22a7a102e8f50416b5 : TIMED_OUT in 3h 04m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/84e4c3876c5e4cd5a356b33bf6c22f7c : FAILURE in 2h 13m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a55b10418b3b40e7b1d64dc14aebdd98 : SUCCESS in 1h 55m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9f489d4325484d4397c81c353abe3fe8 : SUCCESS in 2h 15m 12s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3da0dde24eee4bd5bcb364730fc52b9e : SUCCESS in 1h 30m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/712d9f258f6447709caa2c12ed3b1d20 : SUCCESS in 1h 43m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4dba20fce92c4d0889b9120b543452e3 : SUCCESS in 31m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/128be09923de4afc9c948d98632ada26 : SUCCESS in 1h 14m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/091fc1b1190c460c8c2d72dd3b2acc98 : SUCCESS in 1h 55m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c89fed3dfe53465f84f2e6ccf122d976 : SUCCESS in 2h 46m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90a0eb97fd134d66bb57f9aeae4b5c76 : SUCCESS in 47m 18s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a765e75eadfe4866afabe1ac86e540ed : SUCCESS in 1h 26m 18s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/3c7da32e7ec943af9521cd19333da5a5 : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a308983f8202a1d621bed7bd3da7ac58eca87a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-06-01 03:14:06.000000000","message":"Patch Set 1:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1407/","accounts_in_message":[],"_revision_number":1},{"id":"2ed067b178ea0dfba00cab14cf3749bdc8a36566","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-01 06:24:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18226e5944086a77885e3ea6c2807f33eb3979ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-01 09:07:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/884896 : FAILURE in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"89ae72231e2a4b6f779a98522d3bf1d1a8687cde","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-01 10:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/96/884896/1/fc : SUCCESS in 0h 38m 59s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/96/884896/1/iscsi : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4f952574ec39637ea8c26d5797f320482d6a4742","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-01 10:24:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"789af5095c81e2ff403e818ba5423218325fd877","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-01 13:43:30.000000000","message":"Abandoned\n\nWill be handled by https://review.opendev.org/c/openstack/cinder/+/884194","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"814c72f4034685ab62f2dac32770f4190718c203","revisions":{"814c72f4034685ab62f2dac32770f4190718c203":{"kind":"REWORK","_number":1,"created":"2023-05-31 15:02:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/96/884896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/96/884896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/884896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/884896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/96/884896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/96/884896/1"}}},"commit":{"parents":[{"commit":"77b6035e4fd8931f40370d0534245100128275d9","subject":"Merge \"Automate generation of backups api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77b6035e4fd8931f40370d0534245100128275d9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-05-24 12:41:33.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-05-31 15:01:46.000000000","tz":-240},"subject":"zuul: Extend c-p-ceph-tempest, make integrated-storage-focal nv","message":"zuul: Extend c-p-ceph-tempest, make integrated-storage-focal nv\n\nThis job is somewhat regularly bumping into the\n2h time limit, bump it to 3 hours.\n\nhttps://zuul.opendev.org/t/openstack/builds?job_name\u003dcinder-plugin-ceph-tempest\u0026result\u003dTIMED_OUT\u0026skip\u003d0\n\nMake tempest-integrated-storage-ubuntu-focal non-voting in\nthe check queue and remove it from the gate queue, since it\ndoesn\u0027t work and is preventing us from merging gate fixes.\n\nChange-Id: I130a4fd6f2edd5628208774aa3f3b70ac494e476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/814c72f4034685ab62f2dac32770f4190718c203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/814c72f4034685ab62f2dac32770f4190718c203"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
