)]}'
{"id":"openstack%2Fcinder~885428","triplet_id":"openstack%2Fcinder~master~I28839e2a97c96b8d71d28dbeb9baa049423fa1c4","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-06-13 19:55:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-01-22 13:07:49.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-06-24 23:25:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-06-25 00:55:36.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-06-25 00:55:36.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-06-25 00:55:36.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-06-25 00:55:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28839e2a97c96b8d71d28dbeb9baa049423fa1c4","subject":"Add GMR to cinder wsgi","status":"MERGED","created":"2023-06-07 07:58:58.000000000","updated":"2024-06-25 00:57:06.000000000","submitted":"2024-06-25 00:55:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885428","meta_rev_id":"13ab2d2040702822eb5ddb44767316d35cb48275","_number":885428,"virtual_id_number":885428,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 00:55:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-06-13 19:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":2,"date":"2024-06-24 23:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2024-01-23 01:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":1,"date":"2024-06-24 23:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 07:59:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2023-06-07 08:01:38.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-06-07 08:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-07 08:33:33.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-07 08:34:03.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-07 09:40:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-07 10:01:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-07 10:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-07 10:26:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-07 11:40:41.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-07 11:43:19.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-07 12:36:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-07 14:08:33.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-06-07 18:08:30.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-07 19:09:17.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-07 22:29:02.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-08 04:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-12 10:23:58.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2023-06-15 12:51:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-16 14:47:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 14:58:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 15:26:52.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-26 15:54:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-26 18:31:26.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-09-26 20:16:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-09-27 04:47:55.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-10-05 20:12:57.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-30 01:36:52.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-01-22 16:08:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-22 16:22:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ba31ea7646a047580ac6e99fc66104209a4a58f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-07 07:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4876c81b7acea9f0e0aafb0059a45f57e4e467d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-07 07:59:56.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2db5d979feccef3677afffea9402cb4adfa85ba8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 08:20:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f572d1f50fb4434a3b7b5ad33c2248b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9cdba32b3b44bdc9cc06041216569d9 : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf23be9ec57a4e66b5ff80c1521b5daf : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/afa66244db8848d4a9dc6f58eebbefab : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf877f8d225a966666378f449f81e74bc7709217","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-07 08:33:33.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/79b2c3660fed4b77ba39523660146492\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8b671eb5fc084c2bab0054de04fd311c : FAILURE in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"812d74683fdb9fe47b885a083981c13fd5f07c27","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-07 08:34:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/28/885428/1/check/fujitsu-eternus-dx-iscsi/def3598 : FAILURE in 34m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/28/885428/1/check/fujitsu-eternus-dx-fc/be9539b : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"66719e0d71d295d78f0b5f780336f299e2f7b725","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-07 09:40:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_Unity_FC/eaef13a/DellEMC_Unity_FC/235 : SUCCESS in 1h 32m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_Unity_iSCSI/e719cd5/DellEMC_Unity_iSCSI/235 : SUCCESS in 1h 40m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8777fe9b6473409009abdaf858be3a99c4b78dfd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-07 10:01:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/28/885428/1/check/iscsi-driver-alletra-6k/1381d54/1381d54.html : FAILURE in 2h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c97cdad0d2e024db3e70399e67c9f629f3dfb8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 10:19:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a6eb360e3d4e80ac8d2639c1ac3afc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2cea008c31f459583ef4dd1b038d61d : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50105d86f25c4374842a98b8e8819fa5 : SUCCESS in 9m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca585b2cd33c425188d62fda46479fee : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2f3240afa9140299978e88bf2f24f63 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d447abdee2443fbf700fc5f0b82cb4 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/989a5edb80a9456bb55bd7c43513c028 : SUCCESS in 1h 04m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c9185a4e632489d866676c77808e6b6 : SUCCESS in 1h 23m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0893e9e93d54dfdb179d12db27a06d5 : SUCCESS in 48m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad947e88ee994622a3489a3b143d432c : SUCCESS in 17m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ee09568bee394d1e8857e3ec0f0ead3f : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb569db6dd2c403f90aed7dc31aedb87 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/845c4849eb0140aca110c4dc31e47064 : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f5c2da4e0ff474ea1a3b0966f7d6e77 : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c7d05bba88ea40ffa484591141bd60a5 : SUCCESS in 34m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/90a82663b48e48069f67d5c1bba3a82c : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b05564a03614794a414791dfffd9a4a : SUCCESS in 2h 05m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/86b2331abc2e48fe82f5c8168b9abf13 : FAILURE in 1h 57m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f21b5c4d1f8d46cb9998d1fc70681615 : SUCCESS in 2h 02m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/be4d1d32a3524e479afa853f312a4f8a : SUCCESS in 2h 19m 20s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31294b22ab8043c79b3b93902faf3d5b : SUCCESS in 1h 11m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1fc28c0048d24727bec9804cdc51db22 : FAILURE in 1h 54m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c866b7aea66f4e95a5ba5b404a1a2ea9 : SUCCESS in 32m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/22411efc5a5b4d7583398a52bbaceee5 : SUCCESS in 1h 39m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/104f94d93c9e47a3b9ffb5898f196e95 : POST_FAILURE in 1h 56m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73753b62985943ed92a0fc7b05c860f5 : SUCCESS in 2h 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bd3636149fe4187b283fe6fec3d3d63 : SUCCESS in 1h 02m 26s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/fbae424148744c079fb497d3646e8764 : SUCCESS in 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e2c0971b772d1a3ace544a1af827c28fad562af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-07 10:26:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_06_2023_10_19_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"64ca0d92c7e0566e09d92a0d1efda1011535a83a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-07 11:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/1/upstream-check/cinder-tempest-plugin-ontap-fcp/dd78242/ : RETRY_LIMIT in 9m 00s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/b404a65/ : SUCCESS in 49m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/1/upstream-check/cinder-tempest-plugin-ontap-nfs/bf54fa6/ : SUCCESS in 1h 18m 27s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9912bb0/ : FAILURE in 1h 19m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e2c05386fe5aaffaadc0776f157cee0a85eea573","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-07 11:43:19.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerStore_iSCSI/a55217a/DellEMC_PowerStore_iSCSI/82 : SUCCESS in 51m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerStore_FC/c6800ef/DellEMC_PowerStore_FC/87 : SUCCESS in 1h 23m 30s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerStore_NFS/33e37d1/DellEMC_PowerStore_NFS/81 : SUCCESS in 1h 01m 17s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerStore_NVMeTCP/fd838d8/DellEMC_PowerStore_NVMeTCP/82 : SUCCESS in 1h 11m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f250954db49ef0b9b825e9c3c04d45342323f7cd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-07 12:36:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerFlex/a38bd20/DellEMC_PowerFlex/130 : FAILURE in 2h 17m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a69da25e82a0023db838aa6d20f4bac755ad893d","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-06-07 14:08:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885428-1/Lenovo-iSCSI/ : FAILURE in 35:59.77\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885428-1/Lenovo-FC/ : FAILURE in 35:51.94","accounts_in_message":[],"_revision_number":1},{"id":"c4c3ef9ef98f3a9f18bceef321cd1e2522661551","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-07 15:21:59.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_Unity_FC/91c234f/DellEMC_Unity_FC/238 : SUCCESS in 1h 41m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_Unity_iSCSI/f8cb6fa/DellEMC_Unity_iSCSI/238 : SUCCESS in 2h 02m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"668bc1b4a1ea433b3070645a4a5cebfb3ad2c8e4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-07 18:08:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885428-1/HPMSA-iSCSI/ : SUCCESS in 1:46:03\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885428-1/HPMSA-FC/ : SUCCESS in 1:26:41","accounts_in_message":[],"_revision_number":1},{"id":"54069d19e0cc25a1d0aa0fa2a61701c154ff14e9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-07 19:09:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerMAX_FC/04f07a5/DellEMC_PowerMAX_FC/105 : SUCCESS in 7h 06m 09s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/1/check/DellEMC_PowerMAX_iSCSI/2affae7/DellEMC_PowerMAX_iSCSI/100 : SUCCESS in 6h 13m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"747ec02a0c9ba6cf898a67ad0863d987efbc17cd","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-07 22:29:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885428-1/PVME-iSCSI/ : SUCCESS in 2:01:28\n- PVME-FC https://os-logs.seagate.com/dell-ci/885428-1/PVME-FC/ : SUCCESS in 1:36:47","accounts_in_message":[],"_revision_number":1},{"id":"edcb7bce833334245c206aa284925eac84c5fe46","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 04:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/1/fc : FAILURE in 0h 40m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/1/iscsi : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ae24118fd4300c924a0ced0f8630f776dc9ab90b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-06-12 08:24:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f1a551efe92a7f333ada4228907cd363d9efc40","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-12 10:23:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d067041833e87fbeb3c2d8b50d46623b2573780c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-15 12:51:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/1/20230615145413/NEC-V-ISCSI : SUCCESS in 2h 48m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/1/20230615145413/NEC-V-FC : SUCCESS in 6h 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e62fe2c6a9c8d6ceb107dabdd55e4f54a2e5e4c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-16 13:59:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"44eb489eb8bfc8cfad81594ddbe1c2f1fa1e426a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-06-16 13:59:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f88ec57d240bef68e92b68e9fd3aad36340440f9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 14:00:08.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8479e486d394b2e440e999684ab3b6a3346b9323","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 14:12:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823594447e4247d9a0f8a638b5527d04\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba9a059e123845ffb3de2cccaa250c00 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1399bc38e2c145e09f88c5ef513f0c57 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b44600ec70524a31b3ccd7344e5c18a6 : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58f0fa8741c5a0dfa079b08079cf9eda1aa704f7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-16 14:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/1/20230616160827/NEC-FC : SUCCESS in 3h 24m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/1/20230616160827/NEC-ISCSI : SUCCESS in 7h 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0b2a3ae4c4f3449c4d8a204758775d1800e89e01","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 14:58:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885428 : FAILURE in 57m 13s","accounts_in_message":[],"_revision_number":2},{"id":"ff64940317d7096e800466835e08da75165d2c73","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 14:58:59.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3f31993f9bbc41e39b048ec63eba4c90\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a20bd5d863e74bf5996339b0ac205d31 : FAILURE in 56m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7379944c75eb3def55c2327677a5b35958699c14","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 15:26:52.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-28-885428-2/ns5_nfs : FAILURE in 1h 27m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-28-885428-2/ns5_iscsi : FAILURE in 1h 26m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4cf15e2e3c03f2b8558dfb62a1092b2d40fff5a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 15:45:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/28/885428/2/check/iscsi-driver-alletra-6k/88af4e2/88af4e2.html : FAILURE in 1h 44m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1b2bd33927fa37cee39f077ab28b4adf65fef7bb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-16 16:16:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86b2b91e833cf456bf2c7bd8ff4bae4d3cadfd63","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 16:48:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/28/885428/2/check/iscsi-driver-nimble/a44c83e/a44c83e.html : FAILURE in 57m 40s","accounts_in_message":[],"_revision_number":2},{"id":"3bf4d3d9c58b57d9d89177b17bbb5b55dd44c5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 17:07:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e80ecd4394f9476995f808336cb70741\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ecbe6b44cd9442b95a3681f863830d1 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/850f97ed68564cdba62d1697b2de2fe7 : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/565fbabdae014da39f02ea994cc5d638 : SUCCESS in 12m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2685f7878374159883f68550cfaaebd : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2724eef7d65f4c408431efaf35db6ede : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/2aa3a11adc6e42af8e3d7c02dc72078f : SUCCESS in 53m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b07efc484dc4f55bed45817c88fbdeb : SUCCESS in 1h 36m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57889199c865471f9ad1223a2adbd1b6 : SUCCESS in 57m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d1177288c7b41c394b746e7422361be : SUCCESS in 8m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a1866f407aa94511a82953a2c53ec01e : SUCCESS in 15m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e5663071bf44e73aecbb9f76372694d : SUCCESS in 6m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7f28ac910de4668bb02c5f2c46baeaf : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b58712b691840088f687592ce4c9021 : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8405e0f34920464087f933f3a8b8cf3a : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce16e42091a64edeaf1fe249a965209c : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a004431730fc446bafe147003ab85af7 : SUCCESS in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b7c99e9e62943e9baf1e955cfe40415 : SUCCESS in 2h 10m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/24d078b99e0140b7907e337bc53fc09a : FAILURE in 1h 35m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/86f0ff5d112b4ef3b366676062345bd3 : FAILURE in 1h 59m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a67a3bd510b242a397ae2ab063a3ba62 : SUCCESS in 2h 25m 59s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88d8fa62d73744988d6693f595b60af3 : SUCCESS in 1h 13m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/536d80596b0e4e3a8f61127e950a02a9 : SUCCESS in 1h 43m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f9c66467082c4d92a352d1669495fa23 : SUCCESS in 34m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c9b8b7b53654028af0609a1e097dcb1 : SUCCESS in 1h 31m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/35492af6d26e4980925b33557f2e0b96 : POST_FAILURE in 2h 21m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c9b1d821a764567b6d7003e6b3f09f2 : TIMED_OUT in 3h 03m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcd5d0d097fb4b66a058ff1b641c37b0 : SUCCESS in 55m 41s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/33fc31b3ec394d388e33fdeee4d24c50 : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4b3b99fb11b48322e104fcc1e349b2b53e91162","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 20:26:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/885428/2/check/DellEMC_Unity_FC/fd0b1be/DellEMC_Unity_FC/290 : SUCCESS in 1h 51m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/2/check/DellEMC_Unity_iSCSI/f79c480/DellEMC_Unity_iSCSI/291 : FAILURE in 29m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2024142b7e5b404a2aabff14505a708f859dba64","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:54.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/2/fc : SUCCESS in 0h 46m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/2/iscsi : SUCCESS in 0h 47m 55s","accounts_in_message":[],"_revision_number":2},{"id":"537b088a570e97e0434c3eb39dcf5a31190b6367","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-21 17:49:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/2/20230621234312/NEC-V-ISCSI : SUCCESS in 2h 53m 11s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/2/20230621234312/NEC-V-FC : SUCCESS in 3h 06m 25s","accounts_in_message":[],"_revision_number":2},{"id":"272233edae7180ead16ccace2342b4a6a7e4b151","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-22 19:06:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/2/20230623002917/NEC-ISCSI : SUCCESS in 3h 35m 23s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/2/20230623002917/NEC-FC : SUCCESS in 3h 36m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6f48b11cb7c36972b652fd939cf2a3e63595ae1e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-07-13 15:45:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e3ece1e4761c4f96fbbfac72baee2d29db109dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 15:51:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0f71e30ee9e74a7481d1b2299a003fdb : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5e38da3429834ae394e5181c168a700e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"d4b219513b9db476c11ef6d3e6747c63db22da3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 16:00:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f404ad12de44185b10cdf3e913e8087\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a76ddeb864a424c83a47a0cec33fd82 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c107b1aa41c45d9b08c53c395bdd42a : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7451611783144cf9a799f2be59f66c4c : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74a75ea8e8d02f76aa1e27db3c9ed8574bfa3c49","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-13 18:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/5e580c0ea6124984bda348b0d2e133ea\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/800b528a8595481fbd3fbf02644dace9 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":2},{"id":"af89ec107004174b3440a7f8619f3cb427d3ecbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:46:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da7e1ef1df68413aa878006309f41d88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf239b12bd39484bb20470343420c8bd : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79603fc2d5674b4e80420da198df16ac : TIMED_OUT in 40m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2f300902bb240e89bdd169961d07e44 : SUCCESS in 12m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14fda4129a354d7ebe307e01599b26d1 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9a9fea200a49f69769a61d096f0ff9 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/a501c0ee43c942319df86362f2aaf9bf : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f54e92b0856e4596bb531e93a50dbb7d : SUCCESS in 1h 29m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e43b216a93364257ba05c6b576fe277a : SUCCESS in 57m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e68d7eac81d647559f7407de6e21cd54 : SUCCESS in 8m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f3ba97e743e645ab945e2d73475241d1 : SUCCESS in 17m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4061a1a1d37940538b8195cf532347e0 : SUCCESS in 6m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c67583b51684b04af426e9d5713d5de : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2cf0818672540148f8b3609b082c505 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d732a6850d5844e18e5151a757604fd0 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c4cdfda5ad5450fb4d3bfb4cdbc4321 : SUCCESS in 41m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc69febe07684414b6e58dd14ac1c03e : SUCCESS in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c5dfdc352e04bee92e3142deb22b04d : SUCCESS in 2h 34m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b19b9e961b14460a6c90715bcdb2b58 : SUCCESS in 1h 53m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f66e7d07b414aa8bca0ee71054011ea : FAILURE in 2h 25m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/31347bf4fe4b46069eacf07a41c90ae2 : SUCCESS in 2h 21m 51s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e7914c6ac0d49809f6d02230a1d7fb8 : SUCCESS in 58m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e751d7997494d9cb664ee6c33069576 : FAILURE in 2h 06m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a2df1e095b54e27af9a6ff0e6ac7a29 : SUCCESS in 38m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/addd82ba22b840d7b3c69d4f58c1c39d : SUCCESS in 1h 29m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d21dbc5cabee4bd5b7fa547253906a4f : SUCCESS in 1h 23m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb2c954dd73a490e94e4db7081ffa837 : FAILURE in 1h 47m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edba9c120ae44096bf3fd76e9636ed50 : SUCCESS in 1h 21m 54s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/a0b3b621ccb74c1aa8e5662155cb2861 : SUCCESS in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9821543a78d41cf1c46479c8fe79427221c4b503","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-26 15:00:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a2b585480ee5deafa116fdbc85475f8535561feb","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-26 15:00:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9face882d0efc423d24e3620f35510735e6f176","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-09-26 15:01:57.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"6cd14ee8bdcca2aaf1bf104c598d7d78f8d55bc4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-26 15:02:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/73150db428b045ff919a4de062f9e9e9 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"f4c7cb540a3f2abf4a7be29bdc8aa7658fff6cd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-26 15:06:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a0404641faca5df62efa9616bb07a7144b0e7916","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-26 15:54:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy pure-devstack-cinder-tempest-iscsi-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/0c14254174af4a64b3c287f76606af70 : RETRY_LIMIT in 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/5734cc8548af4705885641a706a8d671 : RETRY_LIMIT in 48s","accounts_in_message":[],"_revision_number":3},{"id":"c57776712d22e33cf1fb00eac32803dd9d8613a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 17:09:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d98863f77b4c428b706a518a8e0c7c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/144b3772546640e7a7a027a72922b678 : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf93c71eab8f40c690b7cb25505b4022 : FAILURE in 34m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6dca7c33959445108e64aa422c8e0a62 : SUCCESS in 31m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/43486ba0d04d4ab7a474f920a3403635 : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e04a6d403e21017d46fb39533e702557d66a8bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 17:26:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d39924e11e401c84120ec4937c6a09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de72b61165574499b5c91ac77f56b528 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/491912aac6a84fc8856660d75c75f763 : SUCCESS in 12m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceeab92c9db94a25a202c54a629d394f : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3154748745354ed2bbc51350377021bd : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/759804a417534dc2be10f5efe5710e5f : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816af859eef945e39aa3381d3104226f : SUCCESS in 17m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/534a4541fa81478d8fa41c11f7d03bcc : SUCCESS in 1h 00m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf453d035c6f4321b496e6ee505163e7 : SUCCESS in 1h 16m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af947df35e384c2ab4606620d159936e : SUCCESS in 50m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37327308d634bf190bb4c6697a8a9ee : SUCCESS in 8m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac65481a32124f468e1f65632117c0a8 : SUCCESS in 17m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21ede5914d76451abee4c8a62ecedaee : SUCCESS in 4m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb13a449a5714bf5a12168c6c270d5ac : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e66fc64dd43e45728a440577b5d47fac : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13910d00929240a2ae30952a824ee9c1 : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b34852da17d24085bff0c76477c835bf : SUCCESS in 39m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d0684d155fb84dada356ec2e5b9638b4 : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/066f69b8d9eb4762ba94fd6c8b88bfcd : SUCCESS in 1h 56m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/27c76f010cdc48ca9ffa2e2cd65b21a9 : FAILURE in 2h 01m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d7f83e726534154b9f7293866457fbe : SUCCESS in 1h 51m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7c108d616f634466aae3df2a72e77d4b : SUCCESS in 2h 09m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3b3d98109ea480ababe12e0c786c7f1 : SUCCESS in 1h 17m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/19b116b027234071ad31a2e378eca5ca : SUCCESS in 1h 40m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b2cc4877139480aa7f31df02c8b40b6 : SUCCESS in 56m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/185fc14f31ce497c8d754d0be78ed6fb : SUCCESS in 1h 13m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3a71f673e18c490bb61b5e3e70fa3297 : POST_FAILURE in 2h 18m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa2044b13ac24612813a9e491e18a547 : SUCCESS in 1h 38m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d95b168686040219bdc53845571c3b5 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":3},{"id":"6bbe017891db614ecbd42fbe52bb4650f356a74f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-26 18:07:59.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/82f8d4366a434f729b79f10e0c6cd1fb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a815adc822e84a93af28200b2bbfc9a5 : FAILURE in 1h 18m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f2a3ab4933086355df1e44be01e5b617fb653ff5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-26 18:31:26.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-885428-3/FC/ : FAILURE in 2h 17m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-885428-3/iSCSI/ : FAILURE in 1h 33m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"d0408bf7e0e621a0ee4eff60f335f7ef0e433f9b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-26 18:41:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_PowerStore_iSCSI/d08599d/DellEMC_PowerStore_iSCSI/326 : SUCCESS in 50m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_PowerStore_FC/9e3adea/DellEMC_PowerStore_FC/333 : SUCCESS in 1h 14m 38s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_PowerStore_NFS/c043542/DellEMC_PowerStore_NFS/332 : SUCCESS in 51m 55s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_PowerStore_NVMeTCP/1920625/DellEMC_PowerStore_NVMeTCP/323 : SUCCESS in 1h 07m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"5c168b91f73bd431309d72cce986ddf1c86c59e6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-26 19:00:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/3/upstream-check/cinder-tempest-plugin-ontap-fcp/2b22555/ : RETRY_LIMIT in 8m 43s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/a9a7ae0/ : SUCCESS in 1h 32m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/3/upstream-check/cinder-tempest-plugin-ontap-nfs/b856138/ : SUCCESS in 45m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/725097f/ : SUCCESS in 41m 00s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"ac004d12ddf78a67e2ca162f58cd334c6196f0bc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-26 20:00:06.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/3/fc : SUCCESS in 0h 43m 07s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/3/iscsi : SUCCESS in 0h 48m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7d663568f9d71b5f2f4a912ced7da121e1633d82","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-26 20:16:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1922/","accounts_in_message":[],"_revision_number":3},{"id":"2cd113c9b7edea074a2ced5ea9fddd5086ffb51f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-26 21:16:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_PowerFlex/0f1e36b/DellEMC_PowerFlex/677 : FAILURE in 1h 08m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"51d97952346581d36be6402e8fff6f5f384a4d8d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-09-27 02:39:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885428-3/HPMSA-iSCSI/ : SUCCESS in 1:44:12\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885428-3/HPMSA-FC/ : SUCCESS in 1:23:12","accounts_in_message":[],"_revision_number":3},{"id":"d24a5b12aa402fc538e99301687551a69d5c39d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-27 04:42:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_Unity_FC/00f24c4/DellEMC_Unity_FC/852 : SUCCESS in 1h 44m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/3/check/DellEMC_Unity_iSCSI/7d64a2c/DellEMC_Unity_iSCSI/861 : FAILURE in 2h 24m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6c9b70564d94b4b5e13e22f562951a9da6238fec","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-27 04:47:55.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/28/885428/3/check/tyds_test/ec1a097 : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"6b259e21fb126a696de315e8ca4d4af8f525f78a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-09-27 07:27:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885428-3/Lenovo-iSCSI/ : FAILURE in 2:29:13\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885428-3/Lenovo-FC/ : SUCCESS in 1:44:20","accounts_in_message":[],"_revision_number":3},{"id":"e39eed8c79de67fd7b0585d0f9606b963eb3bd27","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-09-27 07:31:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885428-3/PVME-iSCSI/ : SUCCESS in 2:12:16\n- PVME-FC https://os-logs.seagate.com/dell-ci/885428-3/PVME-FC/ : SUCCESS in 1:44:56","accounts_in_message":[],"_revision_number":3},{"id":"5910142d7defaca394784a4b0c9aecd2790e3d33","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-09-28 14:35:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/3/20230928164618/NEC-V-ISCSI : SUCCESS in 3h 09m 22s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/3/20230928164618/NEC-V-FC : SUCCESS in 6h 49m 08s","accounts_in_message":[],"_revision_number":3},{"id":"e12e9adf341e5e6d876c3ba07cb5a832c45e2187","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-09-28 16:29:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/3/20230928173444/NEC-ISCSI : SUCCESS in 3h 50m 53s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/3/20230928173444/NEC-FC : SUCCESS in 7h 54m 55s","accounts_in_message":[],"_revision_number":3},{"id":"73d74c15407160c22b7b2a4b77230decb5109591","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-10-05 20:12:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/28/885428/3/ : RETRY_LIMIT in 2m 06s","accounts_in_message":[],"_revision_number":3},{"id":"b29874db938c02057529880f5b7d8a3533490972","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:36:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8a1555aeca89b18f4d4edb8000b599675a44af89","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-11-03 18:06:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"65011c846675e742ca6043ebb3a444d9c308bf6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 18:06:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"993b6bb78f866f94a194393df0e8f16f18de0685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 21:30:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13cf70c1bb174b1da0bef230dadc411a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae04b9b2d364b688451b4a745a1c854 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c1e1a7b0436445699c01c493faedf4f : SUCCESS in 11m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c612286869634b8a835888bd2ca65d71 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809fabf1e3bf4cb8851e341b7448e075 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/4897c3ebca134b04a967e02e71378216 : SUCCESS in 43m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3e1ab83a8e9242bfb72fdc689abc37cd : SUCCESS in 48m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f77ad3f389c44f9849aa23a74336a8f : FAILURE in 1h 44m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18cb7f5c108b4cb1bca57341f6c6102b : SUCCESS in 1h 00m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23a62c726d5b47d1951e9e8207a414c7 : SUCCESS in 6m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5de8eff0dd444719a8d5a7d1fac673da : FAILURE in 1h 01m 19s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6b8bedf6765407699086887b6df723f : SUCCESS in 2h 04m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e2af315f36e4a0d8d23a77e158ed5e1 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":3},{"id":"79cdb0206f4bf99c2ed434fd039bd4b0be4d195b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-09 21:58:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e31a91dca59183995c1397e61ac92de72ecdf69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 22:10:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de6f9e53ee04a0d8a0ff33abb9da84d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcb964896e264ce69b8ead19e685fa08 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3577245811d84b2183d79cb15e204206 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2ca7710302e480db1d067103e682338 : SUCCESS in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"058811eaffcbbd03986a1dc9319dedb8102bdc36","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-09 23:14:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9212c029facd4d33b0c40fec35cfa43f : SUCCESS in 1h 10m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9187ed4227294b48b02e635581e280c6 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":3},{"id":"9afdc0d5aca35283fa1e50bb4b6d1518093b8159","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-09 23:28:47.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f519fbf83c2e4e1194af5f9a762c52ab\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/93c8e7ed43904bd289ed966b3ac5a66c : FAILURE in 1h 22m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ba625f09e6fdf41680e9bbb3cde58a374499a85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-10 00:29:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b13ca64cd2444492b4e2828d70ea1bc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5acdd697524b928709c91b7a0205a4 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e1731e502e4d90adaadd2f1398d8e2 : SUCCESS in 11m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/075837ba6849479fb8d4342b981fd088 : SUCCESS in 14m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d34064fd7f439fac216c489035999b : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/549347ac73c84c4b9a0741f2cbf36c28 : SUCCESS in 50m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2c6eb4a7fe914ca3988700f7ce2c0407 : SUCCESS in 39m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d0245f434884d4fb4d6aab469b4d8ba : SUCCESS in 1h 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/822a52dbc5d647f296d86315b656e8fb : SUCCESS in 45m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f946568070c242b6bd50bdbabddd5bfe : SUCCESS in 7m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/39b3a41f108a4c2e9a52ce07557574be : SUCCESS in 12m 50s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fd45120eb02443469f911e29cafe6888 : SUCCESS in 14m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4b7f6d08d164745b44e7c65bea53ddd : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4514001c01eb4a0db325a0c85a1b1f10 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ceda9a425263472493377318ab545ae5 : SUCCESS in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d845a68b189347e28361173a47e8a644 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f666f5204f44d8bb67d4ed7993b63f6 : SUCCESS in 57m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d040487e7d0c4a2286f92802ae358f53 : SUCCESS in 4m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04524d6fc7264c70b1ca26b805d47bb0 : SUCCESS in 1h 20m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aeec1b27eb724c1298abac5ba85a511c : FAILURE in 2h 24m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3c429c37d6e74fa3912ee6ac82b5925f : SUCCESS in 2h 00m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e73bffacd55e464890c5cc8c2d586532 : SUCCESS in 2h 13m 29s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f8d073520b24f32ac482b22ee4b3bb9 : SUCCESS in 46m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6727b0e4fcee4596a6fd81f7823df495 : SUCCESS in 1h 48m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/991c73e37659476f94c5fb848c25bfb3 : SUCCESS in 44m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/434849a841974108be5f76c4ad8f5691 : SUCCESS in 1h 25m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/23ff11ec14ea475092adc9794e5a0d66 : SUCCESS in 1h 01m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd0e2dd5f4e04e03bc10a30eedf206bd : FAILURE in 1h 09m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3852f123b2364e51b132a10ca3390e40 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":3},{"id":"96c61f2c321842df4767715d807378216523fee4","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-01-22 13:07:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff464f9d6d99f5e2d6f49d557fac0a69d70d2b77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 14:26:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/377bf8a130f64e35a1fdf4fb71955e4a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4659bf9e464f48ae950a7b065655c7c0 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1dac9bcdee14941a9f697705fbc12f8 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/028ed8603f43443d8a992776e670cdd7 : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9208f0dddde876155fb0bbb0aa3c20de86eba2f5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-22 14:42:05.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4547145474d54f7286e3c22bd36a50b1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9f74b74a79f24499a839c262ea36ea04 : FAILURE in 1h 21m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e2fe44ae7e312dff5410ca73ec55c867af611a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 15:52:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e8c6fc97d2844d8932c416e0defec02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6495cb99f5c94142ae18d2ae1f1e0715 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49d1c510bd474b4281c104e0cb2919a3 : SUCCESS in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/177b5555b40647e1add318075ec8f8db : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b63bc2b4abb4e00a62681cd3008e7fa : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/04f7f14ba30b4ad08e2b4f051e0af479 : SUCCESS in 56m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/35fe3b0d2462426c9b50cc75e036f211 : SUCCESS in 1h 10m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5066dca7aa9744f6a4260f51f6033dc9 : SUCCESS in 1h 30m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1ae78dd890341edbae99efe79898080 : SUCCESS in 42m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de28a411ed894e3baf77faf5e30ca844 : SUCCESS in 8m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ae50eb0f4bd44c4d9f279adfcd9c7174 : SUCCESS in 17m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2536b50b65d942a28e0e832cb53aec1b : SUCCESS in 11m 43s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cbf85a3d04634a52a47d2153a9a0a30a : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b14e03966a04c38aba07a902504753b : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3964a4ca2d22442ab11d0d4924c521c6 : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f9f57e484644a868ccaef2a76ebaebd : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f1054593332e4725bde183519d531fdf : SUCCESS in 48m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3bef28e8652340679e9d5b6f0d368e1b : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4fab7a2b141b476694d072fe3e7b15bc : SUCCESS in 2h 11m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/51ed6e7fc09c481bb56f92794bb111fb : FAILURE in 2h 33m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b974c99cd53142ed8a306ce038f992d5 : SUCCESS in 1h 35m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/662c02bb0a344f959aed7820a08544eb : SUCCESS in 2h 13m 43s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1b53538fc73349f196286f9636a553ef : SUCCESS in 1h 28m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a51f527b5a14d0ea4845880df76f77c : SUCCESS in 1h 33m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/07583235fd3f4041b2412c55c580b582 : SUCCESS in 48m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9d45331d9f4d45bcbaa5a451cd9be6f5 : SUCCESS in 1h 23m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/29058e113a43433a9b8852deb4da1948 : SUCCESS in 1h 34m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20b10eee5bea45dab5e053f81f899821 : SUCCESS in 1h 28m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4e0ff18a98541ae969783efcace53e0 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":3},{"id":"91b587b2aa1012babe21bad857dee887c8c8cebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 15:53:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"51e8a97329fafa562904fe8dc4cad0aea372411b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 16:08:25.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1cc5d40e6a6d565cd69dc4a1a9cf9e9ae2447e7a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-22 16:18:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0452f9417f0c8c85483319197a8b2c484deae85d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-22 16:22:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"00d80212983c16ec1dca8af0129e4a83bb5e1e4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-22 16:22:41.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16392/","accounts_in_message":[],"_revision_number":4},{"id":"4ebd2d457a8ce3c31cb894b548e6268314bbc65c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-22 16:22:44.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16392/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"de8a4ecef28803d79820157bc10c609ddc36ef14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:06:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e601d76dd8844cf7a9bf3ce74c1efb16\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c5e6b1677804bc5a5c538c1f9c60ee2 : SUCCESS in 16m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/50e9221acbfe407c80bf5c05c836b48a : FAILURE in 34m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/986d7c1ee0df46d6b989f99dbaac6297 : SUCCESS in 18m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d6a14b4af4ef10e0d7931f18e16dab426c83e194","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-22 17:46:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/476/.","accounts_in_message":[],"_revision_number":4},{"id":"7e4525277d5a14da6af4a4e299d20d161f97b229","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-22 17:58:00.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/474/.","accounts_in_message":[],"_revision_number":4},{"id":"2da337ba908a6631a07133fed41b23cd02e66f4a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-22 17:59:32.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/66225e5f68404ad1b53cd17421a941d5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fee6729fed9b497a804282083b7784f9 : FAILURE in 1h 24m 27s","accounts_in_message":[],"_revision_number":4},{"id":"7f3ae7b08f1233bdf65da4f28f6fcb067de43e8f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-22 18:03:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885428 : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":4},{"id":"707724020db5456f3b232b6c27fe82865a41b3af","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-22 18:54:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/789e2c3bc771449889ccac22d9809bf9 : FAILURE in 1h 19m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f001451c2d24822983505a018ee5734 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ab35d30f28cc5db3de72dd49dd09c8e9d8643261","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-22 19:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/4/fc : SUCCESS in 0h 36m 05s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/885428/4/iscsi : SUCCESS in 0h 54m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b5af9921c169dff3ed9ad9d81906286d44f5fae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 19:39:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c700e9e4a0714bd5832d452d51418586\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e6ee858fb34f98ab5311331733109d : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/869a2215013849b388ae40a2d2dd249b : SUCCESS in 11m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ee8984c45124984b31c5310bd3fc42c : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5df5b8786b44595a40c191b370337ec : SUCCESS in 12m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d12a7c2c1b20428fbf2d358d095734e5 : SUCCESS in 2m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/3bd95b15d9584083b5835b2080a23997 : SUCCESS in 53m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d1668e0bd67d481495ee0e4c56505648 : SUCCESS in 48m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1b4563b03374ae8a3af3c1bee8cf3dc : SUCCESS in 1h 17m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f0b2174fcd043548fb9dd045c2eaf0f : SUCCESS in 25m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12df364c8f374df2adf68c5bc6717600 : SUCCESS in 8m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dbf92b96c5d64e69a05592f1a25d31c9 : SUCCESS in 15m 24s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/147f94de51584c40925ffc066062b481 : SUCCESS in 12m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/79b94d119aa142808e1fce06e0af09de : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4eee44c91bc445658585fde9a3ac924a : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/551aeefedaee4b34b1f2af7c3236796e : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf33ebc88fb340479a8ed5d4091de49e : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/64a639f0df67447dbeb3aa3815d5a55a : SUCCESS in 57m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c10e6e4f762a44378b8eb67ba2d03b65 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0238914159f143b09f9499ce2f7febe9 : SUCCESS in 1h 57m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/94d032931aad44fc95c325985526a8ea : FAILURE in 2h 10m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1dae63aee9e43c985f260cd4b3a3354 : SUCCESS in 1h 22m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bde5093b28374fe18a816d748ebeb053 : TIMED_OUT in 3h 07m 47s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a79fa24af114d3d9f10f200f94652e3 : SUCCESS in 59m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47641f6d8e734a0b825e3c12d8c93cd0 : SUCCESS in 1h 33m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/452199b6e9204bc0bc31a9993f4a20ab : SUCCESS in 45m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf71d05f2e6d42e4b8de5c81c0d4ea12 : SUCCESS in 1h 23m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/67a17c478e8f47c0bf3c091d7d475945 : SUCCESS in 1h 31m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a80b9ce2433d4d4585176ca62f3c7b24 : SUCCESS in 1h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e8cf16e941c481995633bc02a232a2b : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":4},{"id":"865ad560930be1449a1a49ed50ec5d5f68367630","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-22 21:01:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/4/upstream-check/cinder-tempest-plugin-ontap-fcp/25783d4/ : SUCCESS in 1h 06m 17s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/1e4d957/ : SUCCESS in 46m 11s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/bd72bf3/ : SUCCESS in 1h 19m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"efd8a3eb62c9fe12d493a4e9374ce5babca5b784","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-22 21:19:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/28/885428/4/check/DellEMC_PowerFlex_3.6/caa2f09/DellEMC_PowerFlex_3.6/1268 : SUCCESS in 1h 10m 38s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/28/885428/4/check/DellEMC_PowerFlex_4.5/d081066/DellEMC_PowerFlex_4.5/238 : FAILURE in 1h 57m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"454d09267bc3344b6020f21edf6b75f3ef221585","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-22 22:03:50.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/4/check/DellEMC_PowerStore_iSCSI/2be6fbe/DellEMC_PowerStore_iSCSI/970 : SUCCESS in 1h 17m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/28/885428/4/check/DellEMC_PowerStore_FC/1e6be8d/DellEMC_PowerStore_FC/1126 : SUCCESS in 1h 28m 54s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/28/885428/4/check/DellEMC_PowerStore_NFS/0344b4e/DellEMC_PowerStore_NFS/656 : SUCCESS in 1h 03m 10s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/28/885428/4/check/DellEMC_PowerStore_NVMeTCP/b811fa5/DellEMC_PowerStore_NVMeTCP/522 : SUCCESS in 2h 17m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"646fc95686cb42f49dbefa4dcd1ab03a1bd30b1b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-01-22 23:57:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885428-4/HPMSA-iSCSI/ : SUCCESS in 2:05:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885428-4/HPMSA-FC/ : SUCCESS in 1:35:49","accounts_in_message":[],"_revision_number":4},{"id":"b1d2525188d412be035685ea823aa50cde247665","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-01-23 00:15:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885428-4/PVME-iSCSI/ : SUCCESS in 2:05:59\n- PVME-FC https://os-logs.seagate.com/dell-ci/885428-4/PVME-FC/ : SUCCESS in 1:54:27","accounts_in_message":[],"_revision_number":4},{"id":"476d38457467ae12737cd0e038228394e3a63461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 01:11:40.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6a8c6c7cee46189f73d8a56f2d3319c0973f0b37","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-23 01:11:47.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16395/","accounts_in_message":[],"_revision_number":5},{"id":"141b095e6bcb1077cbb3cf4ecb7541756fc551c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-23 01:11:49.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16395/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"433082f0e84cb8da9783f08e1c9cebb6bc7e6be1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 01:12:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ef4bb05f6d14250ec8dffc3790d1c98d2fac02a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 01:37:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adae197c3a5a4b17a05a71d907a0a6fc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06b5385ef4924c0fbf7a97f3f9e7f844 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52d29d6ca95a464dae90f9a96fd61bd5 : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/24d5e33d348546aabdc2a5f81301b06b : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7f556c3524f5059ed6a7605001865bbacbb3383e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-23 02:06:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885428 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":5},{"id":"6f15c7dc4f0edb082cd2a03ff89cdff2968cff9b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-23 02:26:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ea3e2ef0b94f4b66958106c105f43d61 : FAILURE in 1h 12m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4fbf97605ad6460db6256da707e1898d : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":5},{"id":"7efb0fc7b92dda97fc246337cd2e4b423cf8b9f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-23 02:33:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/477/.","accounts_in_message":[],"_revision_number":5},{"id":"755dd25f8e5d124bfea1e6c9c6adaa8a9aabce37","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-23 02:33:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/479/.","accounts_in_message":[],"_revision_number":5},{"id":"751b9b3cc84a86b516063c3240e49c476cc23a11","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-23 02:38:33.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b6fe49bd697d493d82b60dd4de9a5c66\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a7191167e2844791a520712be38f2145 : FAILURE in 1h 23m 35s","accounts_in_message":[],"_revision_number":5},{"id":"2a659438c4b9f7c297a890bbef3dc5946e196999","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-23 03:19:33.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerFlex_3.6/ee903e4/DellEMC_PowerFlex_3.6/1271 : SUCCESS in 1h 15m 30s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerFlex_4.5/bad3dd5/DellEMC_PowerFlex_4.5/241 : SUCCESS in 2h 07m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"ddf8b4cba4b87facc533c43df6383f26a67cbd01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 03:42:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae61e692407d46799dce6b293be6bd5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee2cd37221a4090a52a4a935846b2c7 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e0fac832e1a41358679e34d812e0c38 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ad0e61f480748e79878ee8c36f426ea : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbc6a620d2e42afa161c5c798d84853 : SUCCESS in 12m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/adb6a98aeecd4b2da59afaa8757fa96c : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/76fefae5dd714370ac150828042c007d : SUCCESS in 1h 10m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f986bb6fad57404fa6a2aade99f86b8f : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4bc606272e64c4c8e127beafbf130eb : SUCCESS in 1h 28m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c3b228a9cf9430db9df37eb5107b8eb : SUCCESS in 20m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29a661d9935e4f538cff18ced803d996 : SUCCESS in 8m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/470ca9294a024a21b07f5cb320eb7e0e : SUCCESS in 13m 24s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6efd509feee64252bb1a8e9cdb077ac1 : SUCCESS in 15m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f948a5fe19394480aa6266344d7b1348 : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf5d7fcad2b04debb16f10d674a68a6b : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e826c06544b4ff9841f16bb8954fed9 : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f26b78ddebd49a0b0e225efd9fa4ba9 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3e8da07c1c6d40879c553bfc0343d79d : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0a7503d771e4fc6ab204bb2f6717391 : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6450da1235540819dcffd2effec1b49 : SUCCESS in 2h 02m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f0480cd026b74729b01fd13e52615642 : FAILURE in 2h 21m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5cc45e9759ab45efbf6ccb831d1d99c6 : SUCCESS in 1h 44m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/49054bb7ca1544709e1c8b56c62e3dda : SUCCESS in 2h 30m 21s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f1e7c636ce24a7d9bd50a86d7f74a60 : SUCCESS in 1h 21m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0dd009f3d46b4ba7a29298db10c8bad8 : SUCCESS in 1h 30m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/61292c4b05d94b46be5607ce67c88fe2 : SUCCESS in 45m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/82a80413d26046fb8153b258e2fa4ecc : SUCCESS in 1h 24m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b77274a60bf74fecaf14130a84b302e5 : SUCCESS in 2h 23m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/18c399b2e8d04c5693f63d6820430b48 : SUCCESS in 1h 35m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/708a7d363e0541c78511f21c38848308 : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":5},{"id":"1186e0fc993d31737306e46e8412b53c97fec49c","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-01-23 04:30:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885428-4/Lenovo-iSCSI/ : SUCCESS in 2:04:02\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885428-4/Lenovo-FC/ : SUCCESS in 1:54:54","accounts_in_message":[],"_revision_number":4},{"id":"7fdcc4855ab01de71bdcd82e1dbd9937bf2d5355","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-23 05:20:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/5/upstream-check/cinder-tempest-plugin-ontap-fcp/7668488/ : FAILURE in 1h 04m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/e6ec2e2/ : SUCCESS in 1h 18m 58s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/5/upstream-check/cinder-tempest-plugin-ontap-nfs/5bd1565/ : SUCCESS in 48m 13s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/885428/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6dd1827/ : SUCCESS in 44m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"ecba0b144072a03c7b2073ed2774198698f4d625","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-23 05:39:01.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerStore_iSCSI/97fbbd6/DellEMC_PowerStore_iSCSI/971 : SUCCESS in 1h 06m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerStore_FC/dd5771b/DellEMC_PowerStore_FC/1127 : SUCCESS in 1h 36m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerStore_NFS/e0a2566/DellEMC_PowerStore_NFS/657 : SUCCESS in 1h 01m 32s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerStore_NVMeTCP/920b087/DellEMC_PowerStore_NVMeTCP/523 : SUCCESS in 1h 55m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"09b51b65d1d5d68649c288c59b8bcba5ac00c44e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-23 05:50:33.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerFlex_3.6/b385362/DellEMC_PowerFlex_3.6/1272 : SUCCESS in 1h 35m 22s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerFlex_4.5/9e7c11d/DellEMC_PowerFlex_4.5/242 : SUCCESS in 2h 07m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"040ee0c8aebb843effc527da3c9803fa2215de60","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-01-23 08:19:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885428-5/PVME-iSCSI/ : SUCCESS in 2:06:44\n- PVME-FC https://os-logs.seagate.com/dell-ci/885428-5/PVME-FC/ : SUCCESS in 1:54:34","accounts_in_message":[],"_revision_number":5},{"id":"4d369e1e7c3ac08efdd43f1e44b8b5e9d0408fea","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-01-23 08:46:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885428-5/HPMSA-iSCSI/ : SUCCESS in 2:03:33\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885428-5/HPMSA-FC/ : SUCCESS in 1:37:56","accounts_in_message":[],"_revision_number":5},{"id":"62b8abd5e6db90bc19a6bfc3331842e99c1088f1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-01-23 09:20:28.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerMAX_FC/83bdd03/DellEMC_PowerMAX_FC/310 : SUCCESS in 4h 31m 15s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/28/885428/5/check/DellEMC_PowerMAX_iSCSI/89fd044/DellEMC_PowerMAX_iSCSI/332 : SUCCESS in 5h 37m 15s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"9d0f0b9ee01e925f675cc67b06388df801e482f3","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-01-23 12:53:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885428-5/Lenovo-iSCSI/ : SUCCESS in 2:04:03\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885428-5/Lenovo-FC/ : SUCCESS in 1:54:46","accounts_in_message":[],"_revision_number":5},{"id":"80af21585ae8adaa075c282e2e21f0202d15cf27","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-04 04:54:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/5/20240204093009/NEC-FC : SUCCESS in 3h 55m 07s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/5/20240204093009/NEC-ISCSI : SUCCESS in 4h 24m 34s","accounts_in_message":[],"_revision_number":5},{"id":"dfb0d3a126c4443e754249e507f05c20c7935e83","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-25 14:50:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/5/20240225203741/NEC-V-FC : SUCCESS in 2h 51m 25s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/885428/5/20240225203741/NEC-V-ISCSI : SUCCESS in 3h 12m 38s","accounts_in_message":[],"_revision_number":5},{"id":"e9dd6fe7ab9a81497564a605eea747ee32e15e73","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-03-20 15:56:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"07543a92d0c0b2a65b82dacec251bf9ad0cacb57","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-06-13 19:55:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5a07d0cc0daea9cced1cff53990632d35fb9ac6a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-06-24 23:25:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bcc0e593d7682a6061d684ae3223b2a15ff67fed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 23:26:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5d4e95ed30fd54c76de6dda4e438d92b063a53ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:55:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b8cea8e4e8a47988b4f2082a3d0b578\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a660e273edf4dc89de914997783ff86 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc0c4df9d363450db48c851982c096e5 : SUCCESS in 10m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df08c166005543daa11a574e9e23e61f : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea9204d9bac4f7cacfd4c3935f82fea : SUCCESS in 9m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1a6431b7bd5424b8d31c5ba4088ac85 : SUCCESS in 2m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b54bf98611f74973bf21427598ae4ad2 : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca02427466ed4f259a18d4708541d7ba : SUCCESS in 1h 20m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cb17603b3a24c479144dcc5dd3c245b : SUCCESS in 1h 05m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c50aa946d0cf4ae0b188f16850397a88 : SUCCESS in 18m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed85f12aa10e4ff7a69e4246c20e7c86 : SUCCESS in 1h 14m 56s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a8db460e18147a4aa3a2eef45abe0a8 : SUCCESS in 1h 28m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67ff08b69e704bdfa9f2647c9f7fa349 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":5},{"id":"6bc5efd3e56213436677c0fa1128c63e417e4f87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:55:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"13ab2d2040702822eb5ddb44767316d35cb48275","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:57:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d6fb2364d95434196bad4716d05a613\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7b3fdd2eed4ebd97420708453412c0 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b23bac7909b148bebb124d915b07e283 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"92b9c25d82e634269018153cfa7ff20efca1fbde","revisions":{"2a1ef7455ecc42b868aa11321086c9e482d0d576":{"kind":"REWORK","_number":1,"created":"2023-06-07 07:58:58.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/28/885428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/885428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/885428/1"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-07 07:50:55.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-07 07:50:55.000000000","tz":120},"subject":"Add GMR to cinder wsgi","message":"Add GMR to cinder wsgi\n\nWhen running cinder under WSGI, we might want to have Guru Meditation\nReports as well as when running outside of WSGI.\n\nNote that, under WSGI, the GMR can only be trigger via a \"touch\" on a\nfile, the \"signal\" was is not working because apache is registering the\nsignal handlers instead of cinder.\n\nChange-Id: I28839e2a97c96b8d71d28dbeb9baa049423fa1c4\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a1ef7455ecc42b868aa11321086c9e482d0d576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a1ef7455ecc42b868aa11321086c9e482d0d576"}]},"branch":"refs/heads/master"},"25cfc3a4d3df830018cefab73e9bc1aee71049a3":{"kind":"REWORK","_number":2,"created":"2023-06-16 13:59:18.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/28/885428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/885428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/885428/2"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-07 07:50:55.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-16 13:59:08.000000000","tz":120},"subject":"Add GMR to cinder wsgi","message":"Add GMR to cinder wsgi\n\nWhen running cinder under WSGI, we might want to have Guru Meditation\nReports as well as when running outside of WSGI.\n\nNote that, under WSGI, the GMR can only be trigger via a \"touch\" on a\nfile, the \"signal\" way is not working because apache is registering the\nsignal handlers instead of cinder.\n\nChange-Id: I28839e2a97c96b8d71d28dbeb9baa049423fa1c4\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25cfc3a4d3df830018cefab73e9bc1aee71049a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25cfc3a4d3df830018cefab73e9bc1aee71049a3"}]},"branch":"refs/heads/master"},"4555d14905e08829d318a3e3d0a89dfca21f29c6":{"kind":"REWORK","_number":3,"created":"2023-09-26 15:00:32.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/28/885428/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/885428/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/885428/3"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-07 07:50:55.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-26 15:00:05.000000000","tz":120},"subject":"Add GMR to cinder wsgi","message":"Add GMR to cinder wsgi\n\nWhen running cinder under WSGI, we might want to have Guru Meditation\nReports as well as when running outside of WSGI.\n\nNote that, under WSGI, the GMR can only be trigger via a \"touch\" on a\nfile, the \"signal\" way is not working because apache is registering the\nsignal handlers instead of cinder.\n\nChange-Id: I28839e2a97c96b8d71d28dbeb9baa049423fa1c4\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4555d14905e08829d318a3e3d0a89dfca21f29c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4555d14905e08829d318a3e3d0a89dfca21f29c6"}]},"branch":"refs/heads/master"},"5ac4a64b3db9d0d0f23f1fe0795b4d1d17a577f7":{"kind":"REWORK","_number":4,"created":"2024-01-22 16:22:29.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/28/885428/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/885428/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/885428/4"}}},"commit":{"parents":[{"commit":"b8cd101fe41066431b3ccf715780e9cc8b3c5d1f","subject":"Merge \"[docs] quotas apply to image-volume-cache owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8cd101fe41066431b3ccf715780e9cc8b3c5d1f"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-07 07:50:55.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-01-22 16:21:17.000000000","tz":60},"subject":"Add GMR to cinder wsgi","message":"Add GMR to cinder wsgi\n\nWhen running cinder under WSGI, we might want to have Guru Meditation\nReports as well as when running outside of WSGI.\n\nNote that, under WSGI, the GMR can only be trigger via a \"touch\" on a\nfile, the \"signal\" way is not working because apache is registering the\nsignal handlers instead of cinder, so it\u0027s disabled.\n\nChange-Id: I28839e2a97c96b8d71d28dbeb9baa049423fa1c4\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ac4a64b3db9d0d0f23f1fe0795b4d1d17a577f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ac4a64b3db9d0d0f23f1fe0795b4d1d17a577f7"}]},"branch":"refs/heads/master"},"92b9c25d82e634269018153cfa7ff20efca1fbde":{"kind":"REWORK","_number":5,"created":"2024-01-23 01:11:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/885428/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/885428/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/885428/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/885428/5"}}},"commit":{"parents":[{"commit":"b8cd101fe41066431b3ccf715780e9cc8b3c5d1f","subject":"Merge \"[docs] quotas apply to image-volume-cache owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8cd101fe41066431b3ccf715780e9cc8b3c5d1f"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-06-07 07:50:55.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 01:11:39.000000000","tz":0},"subject":"Add GMR to cinder wsgi","message":"Add GMR to cinder wsgi\n\nWhen running cinder under WSGI, we might want to have Guru Meditation\nReports as well as when running outside of WSGI.\n\nNote that, under WSGI, the GMR can only be trigger via a \"touch\" on a\nfile, the \"signal\" way is not working because apache is registering the\nsignal handlers instead of cinder, so it\u0027s disabled.\n\nChange-Id: I28839e2a97c96b8d71d28dbeb9baa049423fa1c4\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92b9c25d82e634269018153cfa7ff20efca1fbde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92b9c25d82e634269018153cfa7ff20efca1fbde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
