)]}'
{"id":"openstack%2Fcinder~885825","triplet_id":"openstack%2Fcinder~master~Ifdd99433563ab72d8d43a374af0377213834d46e","project":"openstack/cinder","branch":"master","topic":"volume_transfers_show_accept","attention_set":{},"removed_from_attention_set":{"35851":{"account":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"last_update":"2023-07-17 12:34:14.000000000","reason":"Change was submitted"},"35836":{"account":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"last_update":"2023-07-17 12:34:14.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-07-13 14:54:38.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-07-17 10:56:53.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ifdd99433563ab72d8d43a374af0377213834d46e","subject":"Automate generation of volume transfer api-ref samples","status":"MERGED","created":"2023-06-10 22:00:52.000000000","updated":"2023-07-17 12:35:52.000000000","submitted":"2023-07-17 12:34:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"885825-volume_transfers_show_accept","meta_rev_id":"d88574f675b360af0cd709274a32396396f93f5e","_number":885825,"virtual_id_number":885825,"owner":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 12:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":2,"date":"2023-07-13 14:54:38.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-07-17 10:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2023-07-17 10:56:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-10 22:01:43.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-10 22:04:02.000000000","updated_by":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"reviewer":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"state":"CC"},{"updated":"2023-06-10 22:04:02.000000000","updated_by":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-06-10 22:04:02.000000000","updated_by":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-06-10 22:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-10 22:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-11 00:22:05.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-11 00:43:48.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-11 01:06:25.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-11 01:14:04.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-11 01:37:18.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-11 02:09:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-11 02:20:53.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-11 02:22:35.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-11 03:00:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-11 03:03:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-11 03:07:51.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-06-11 05:28:47.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-11 08:08:09.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-13 02:06:58.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-17 15:14:08.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 04:11:24.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-22 22:39:43.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 14:30:22.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"state":"REVIEWER"},{"updated":"2023-07-07 22:10:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-08 02:54:51.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-08 07:52:50.000000000","updated_by":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"reviewer":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"state":"CC"}],"messages":[{"id":"fd44ae070d16df945f9c643fedd2fa19ec20dc4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-10 22:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfead6d5e0dff2c4f142a07ae70cb1af8e9c33af","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-10 22:01:43.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0f0b14cee72d3418dde45fa49234ed82d11afd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 22:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75bbb93f915e4b54bd72712094c430d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23269b4def864f8faf155bad8ca34a83 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/217ef2d50c88452dbe5ce4d531e83a67 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc96e6de3b1e4e799d8ebfad21c78f5c : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b30e339ae29ed448252d4341467c0e627db19b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 22:42:45.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab167631d94b444c955e1595b852f563\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9ab7736ebf1a4d7a8e8b757038720c63 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d45dfbd75604b9aa1930b276798d8c5 : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/177e478ca0b043f6abd6168b3f6a2b6e : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdb533a1e0e0480a87bebf71eb346744 : SUCCESS in 12m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98b2b673231a4b928a147c589caf7004 : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b126f1ddf64f97964f3aea5c6010e9 : SUCCESS in 13m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2f7048cce3d647c8a8864c58651b78af : SUCCESS in 14m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9db6e13b02f44dd9871ef1da38ffe6ac : SUCCESS in 4m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d56c58a731754a7992716556d6fd8c59 : FAILURE in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4fdec5f35f640fc9abd33f7139f6956 : FAILURE in 5m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/09909ac74d14436789a70d680a3938cb : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/135a9301dcc14d16ba9c8752b55b02f1 : SUCCESS in 6m 17s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/f5a7ed78291b4f30964b429c59f5e6ea : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4069866970bda3c6680f3dddaf15626118150cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-10 23:06:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f8dfd425d1efe3bcf82bb70f4d3260864804c72","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-10 23:07:23.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c3d25ba92f5a7afb85503ccf4a7c9670e64a332a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 23:19:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee8375403ee43bfa226474d878c8bfc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2fce293f81e41958d4418f25deca753 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/923f8654d0b1401aa75a083fd755eb93 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c84bd8a47234eb4942b93ab4d6e13c8 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9aff9105972b8ee10efd4fe8f7933e6479a5a7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 23:41:56.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9497dcbb85146de921d5fadb4ceb549\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4acce91a39904897aae6fa6c2623eb0a : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcdce37a400747f78459d055dc3ab52a : FAILURE in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05956d9f4c004739bfcccd67913d3431 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23c6b81eba0440fd8dd4927917970991 : SUCCESS in 10m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18fbe968054d4ff58501bf61a6d61afb : FAILURE in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0695e43ea0bb4e1fa6b6668073fa86c3 : SUCCESS in 12m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/70a61ebaefbc42509dffbb17eeef151c : SUCCESS in 14m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7d6f31351a841baaf6ba8ef8c7cf9a5 : SUCCESS in 4m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af48abebea304d1e8ff4f0fc11c3c17d : FAILURE in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c63bd70e78744a928e61153647ea4a73 : FAILURE in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a69a38ceb2a3468c92a77ec78ed5bf11 : SUCCESS in 34m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/45a41b0fb6e44837be4df27e682e3976 : SUCCESS in 5m 48s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/b3cfaa8f8e2847588981815c632e556d : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f60fac6eefebebddc1a1d60c770ad908674cc4d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-11 00:22:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/24613276613e4efe9355cbc86eef1696\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/02da21bf09084f87b7d3fa7e29aace48 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c2468bc02bb3ce11e2852e84a4303cdbb0a69f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-11 00:31:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"518dc426a5c8f44dd1dc41c640fd092d2404da68","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-11 00:32:37.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"622d2bbf81634edf70f3a04f80acc5859fc7ab6f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-11 00:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-1/FC/ : SUCCESS in 2h 21m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-1/iSCSI/ : SUCCESS in 2h 42m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"27d00288333f75b95c89c54dad31bf95e177f860","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-11 00:45:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27cc04921e954cc38cae789f773a73ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ae81bec28254fb4a786f83336261786 : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d228250519e84c668d4e827757a825b4 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab0e324e6db54c729e19a147adc7a15e : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e27f137a5607b2d0f96472fee6d51d1581dca933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-11 01:04:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026017d76c37469cb61dfed12df9f2fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/053e924f3fac4bb9a7237b32b9c599ab : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30006a5b8504422ab4c275c1059d81e0 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e3fcfa6f16645d6a37863580034bb7b : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee31610f44464733a613096263816acf : SUCCESS in 12m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca107625d43b45c58e6ea0f19d7ed2f7 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7541cb410d8e4680a87148c77eb79898 : SUCCESS in 13m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad1481098e3c422e8c75fc5ed4c7ae7e : SUCCESS in 16m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c386155473b64af08b043216cb932d23 : SUCCESS in 5m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c160dc21251e4942bc7299d7475d5fd4 : SUCCESS in 6m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a62966805d5434098728c48b2dff09f : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/64bcb00b48f94a298dda08f127378f3c : SUCCESS in 31m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/896a91c83cd44e2b8acfc2b6ef26c89b : SUCCESS in 6m 13s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/c75d9528df674a13a9c07d1725314262 : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"efb9940fc662d85fdf2cc9748911175ac2c3eba9","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-11 01:06:25.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"2ae5a9dcf8500f3889c7c511213bc64b2d6d5bb9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-11 01:14:04.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_06_2023_01_04_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"cc4eca2f9a1c295d357ab925f57ecb2d69677ae6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-11 01:36:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-2/FC/ : SUCCESS in 2h 19m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-2/iSCSI/ : SUCCESS in 1h 59m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"196239918d8d11dcda8832f81ca59af11d465dd0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-11 01:37:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/885825/3/check/iscsi-driver-alletra-6k/b909892/b909892.html : FAILURE in 1h 04m 15s","accounts_in_message":[],"_revision_number":3},{"id":"63b5e3c9ab2f048de6b4be941088d58bb52f3919","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-11 01:46:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/958b63ed2c464005962378755d48ad4e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/913b502874524cc3be5e0fbb95c45a0b : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":3},{"id":"fa0f2927ab3729121693e3e70e99b71c45f0db33","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-11 01:53:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/885825/3/check/iscsi-driver-nimble/724dd06/724dd06.html : FAILURE in 1h 17m 47s","accounts_in_message":[],"_revision_number":3},{"id":"9b3fc8a559c573fdd3647c8e2672c6e76053dbab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-11 02:09:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 1h 36m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f16014fee92c3fee63f476261b7cad384320f9da","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-11 02:20:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/3/upstream-check/cinder-tempest-plugin-ontap-fcp/9774c38/ : RETRY_LIMIT in 8m 47s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/2a3a93b/ : SUCCESS in 46m 09s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/3/upstream-check/cinder-tempest-plugin-ontap-nfs/2ff30e2/ : SUCCESS in 1h 14m 34s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/0a5f38e/ : FAILURE in 46m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"7fc50d3a9a53af857103770877e17d8389bed5ca","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-11 02:22:35.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerStore_iSCSI/946bf04/DellEMC_PowerStore_iSCSI/94 : SUCCESS in 54m 16s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerStore_FC/a525c04/DellEMC_PowerStore_FC/99 : SUCCESS in 1h 17m 37s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerStore_NFS/ea1e511/DellEMC_PowerStore_NFS/93 : SUCCESS in 1h 03m 34s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerStore_NVMeTCP/869ad38/DellEMC_PowerStore_NVMeTCP/94 : SUCCESS in 1h 09m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"17ad3e1c3d143ea0338333457e959da10b6a4dee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-11 03:00:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerFlex/8727d14/DellEMC_PowerFlex/141 : FAILURE in 1h 55m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"338de999acaceb9f97f851aa1020833926b8e10f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-11 03:03:22.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_Unity_FC/ea10726/DellEMC_Unity_FC/255 : SUCCESS in 1h 38m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_Unity_iSCSI/e685c38/DellEMC_Unity_iSCSI/255 : SUCCESS in 2h 14m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bd21011193c4875bb647d4eefe24a63e17200344","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-06-11 03:07:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885825-3/Lenovo-iSCSI/ : FAILURE in 36:58.74\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885825-3/Lenovo-FC/ : FAILURE in 37:07.46","accounts_in_message":[],"_revision_number":3},{"id":"c95aadba50c7248568cd0914db9d384d07c88fdc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-11 03:10:31.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-3/FC/ : SUCCESS in 2h 38m 21s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-3/iSCSI/ : SUCCESS in 1h 59m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"191186b2f35527c5c19eb4cec838a7c4b157b83c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-11 05:28:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-3/PVME-iSCSI/ : FAILURE in 1:59:03\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-3/PVME-FC/ : SUCCESS in 1:35:49","accounts_in_message":[],"_revision_number":3},{"id":"fe7847dafb1af0312bbf94838fbd6b9b2e15746a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-11 08:08:09.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerMAX_FC/a19e7f2/DellEMC_PowerMAX_FC/115 : SUCCESS in 7h 03m 14s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/3/check/DellEMC_PowerMAX_iSCSI/be82838/DellEMC_PowerMAX_iSCSI/110 : SUCCESS in 6h 43m 06s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"878dbe01fadb48d6ec4d4d606f48c5b0ea3e57cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-11 19:10:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"dbd42d19ae7e6b282cecfcd6f805ec36f5bd2828","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-11 19:11:19.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6a1a59d430a5c232ff386448fd3500f137653738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-11 19:17:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f71583eddbe155319d2f36cc01c5dbbe45c3e871","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-11 19:18:05.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a5307183c6562f0fe7e36099995cbd14b464f91f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-11 19:30:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4a4daa893d54fdf99682cf1568099cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/786df5065dd94648b61384df64eea0da : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13aee02b8b09471a8f2fb59ee5fbac3a : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a3175ea00034f1991a00b27d8c8ff92 : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7bf034b4c064613277e3703977abfa1193d19ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-11 19:50:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957146c3452547b597870859d62832da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cded03ec7a3a42d0a18ccb46457ef02a : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ac2ccb341f4f23afca00db3bbb6747 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2294c0f8ba8e4322999f747a6b71d6bc : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6c78950a5214f20a1779613a5d0efad : SUCCESS in 9m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18a1c2e0edad41b1893429f352bb73ef : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1414e8680b74e249e90632611299171 : SUCCESS in 9m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c5a0f82166e8428e86cb2a91cac7a957 : SUCCESS in 13m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23238801ec604d118ba89777e6f3aa95 : SUCCESS in 4m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bb4af53e54643e5a0a61fcc5fb6683f : SUCCESS in 5m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3d87a233d4f434db6f006b582e900c3 : SUCCESS in 5m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eebdf49432114a08ae254f8579de8bed : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f8d6258ebc44998a05728ef896ebe19 : SUCCESS in 5m 11s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/3d96b07cdcbe42a8b74ee6c235d8ea7b : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d18e0fb5a2b9ae77211d193af474cef5c4c43b46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-11 19:58:25.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_06_2023_19_50_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"0ca3af8c2673469500bc584b662d7dd0996f7251","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-11 20:33:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/98af74d13ab14c75bb76f6ad8e9db0cc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/67b65c1d98374157a4b372820e4c8370 : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":5},{"id":"e8c4fc55154ced60b189ca9daf2ab63292f934ed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-11 20:40:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/885825/5/check/iscsi-driver-alletra-6k/7c38ba5/7c38ba5.html : FAILURE in 1h 21m 57s","accounts_in_message":[],"_revision_number":5},{"id":"9a2b7447777c4de4b8cd7ba25be426c922ed6af9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-11 20:45:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/885825/5/check/iscsi-driver-nimble/60e5e86/60e5e86.html : FAILURE in 1h 24m 28s","accounts_in_message":[],"_revision_number":5},{"id":"e16faf6834af93c02abae6d84f13391832ed7b2d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-11 20:55:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 1h 36m 59s","accounts_in_message":[],"_revision_number":5},{"id":"2ddbf25a8701d38f1fcdc24d1978f3cfb4a348bc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-11 21:04:46.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_Unity_FC/47c2ccb/DellEMC_Unity_FC/258 : SUCCESS in 1h 29m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_Unity_iSCSI/4894e8b/DellEMC_Unity_iSCSI/258 : SUCCESS in 1h 47m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a12f0a0f38b9ae04fc18536c234697f92ed03175","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-11 21:08:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/5/upstream-check/cinder-tempest-plugin-ontap-fcp/dc507ce/ : RETRY_LIMIT in 8m 57s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/cd2fc7c/ : SUCCESS in 47m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/5/upstream-check/cinder-tempest-plugin-ontap-nfs/6ad8960/ : SUCCESS in 1h 11m 11s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/124507f/ : FAILURE in 45m 19s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"4a5c997ffa5bfe7924b0d294ae39ecb74b932ec9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-11 21:17:02.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerStore_iSCSI/0c29803/DellEMC_PowerStore_iSCSI/96 : SUCCESS in 51m 28s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerStore_FC/05091de/DellEMC_PowerStore_FC/101 : SUCCESS in 1h 25m 49s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerStore_NFS/6c26592/DellEMC_PowerStore_NFS/95 : SUCCESS in 55m 41s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerStore_NVMeTCP/f005939/DellEMC_PowerStore_NVMeTCP/96 : SUCCESS in 1h 16m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"6028b8b3308e0254ac331a72d79365b6c18d0a3b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-11 21:29:00.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-4/FC/ : SUCCESS in 2h 18m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-4/iSCSI/ : SUCCESS in 1h 56m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"2c005e74bd7a75fcbf39287a0da13a1ec86dd7f7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-11 21:32:33.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerFlex/cf459c0/DellEMC_PowerFlex/143 : SUCCESS in 1h 41m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"3451632077112dc89fd00871f212c9518b0c84fb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-06-11 21:51:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885825-5/Lenovo-iSCSI/ : FAILURE in 36:28.04\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885825-5/Lenovo-FC/ : FAILURE in 36:33.32","accounts_in_message":[],"_revision_number":5},{"id":"903524305fa94a4fd316ed66aec6f1968cc0e5f2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-11 21:56:41.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-5/FC/ : SUCCESS in 2h 30m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-5/iSCSI/ : SUCCESS in 1h 52m 21s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"363700a8e95987ba73d92d55316ecf36d738eb30","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-12 00:21:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-5/PVME-iSCSI/ : SUCCESS in 2:02:41\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-5/PVME-FC/ : SUCCESS in 1:38:23","accounts_in_message":[],"_revision_number":5},{"id":"394bfa86c7376fa05c9af24e9f2d83f664e678a2","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-12 01:52:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885825-5/HPMSA-iSCSI/ : SUCCESS in 1:45:16\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885825-5/HPMSA-FC/ : SUCCESS in 1:26:11","accounts_in_message":[],"_revision_number":5},{"id":"18cbe9babf527a8042973747da09b9f7a8d7725c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-12 11:46:20.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerMAX_FC/45f4f16/DellEMC_PowerMAX_FC/118 : FAILURE in 8h 11m 42s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/5/check/DellEMC_PowerMAX_iSCSI/28e1839/DellEMC_PowerMAX_iSCSI/113 : FAILURE in 10h 03m 10s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"ef0ca83cd1d5000ead6d39df449b9771e68206f1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-12 12:22:22.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2299fe8a78877aebcb8b991bcfb876d77c1723ff","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-12 12:23:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"096a5f26a50a1e975ce495a80e5d40996c1b99ec","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-12 12:25:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ad494aff9bb4986f3349c92be833056e182351c9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-06-12 12:29:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3346f51b59d43c3c3bf1fa9ec379a3efef66a0e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-12 22:11:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"629136c765972e0de6b707b663cd521364d53d8d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-12 22:12:47.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a64a3552108f869769dc3a0132e677ff1b4facbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 22:15:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80c794e1d55a4910aceee41c01c35ce2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93cd1d1ae2a14fda9f5b74ce808eb545 : NODE_FAILURE Node request 300-0021447674 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/590f9ba946aa433984a73156ba89f333 : NODE_FAILURE Node request 300-0021447675 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/026598843cd245278a56a87a941a3cf2 : NODE_FAILURE Node request 300-0021447676 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0fd893cb57b9948a5341fe988f44ac82ac41bb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 22:43:04.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ae4e0f396f848ba848b78ebfd4b6219\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d9510e49ef8474c87fe96f70956269d : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82cf64c93b72404db238646d0c54b339 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db3092b73ce447198bde90a60e858cd6 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f36b9e186a4457688177d8cb74f6919 : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00b6eac31c6f4db7a3f02e2f6d2bd41c : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd7abed8e8940d4a765b77552dd7d9e : SUCCESS in 11m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0dea4f1a8f6c4da2832abf0d53c147fe : SUCCESS in 17m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/89a3aff99e5a441e8a872e58a078c43c : SUCCESS in 4m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9dfa45cc253c46e788fce80fd6a5cc24 : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a747a32507f42c889d2fc0db4ec85dc : SUCCESS in 4m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c02d67ba2ac438793de0fe3cb73ffbc : SUCCESS in 30m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9aae501aab774d89a1441879f292b6b6 : SUCCESS in 4m 56s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/6e1cfe7ec1f546ffba52648db508d040 : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5db63d3646d9ba7abe23b3b6efed304aba0067ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-12 23:26:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/885825/6/check/iscsi-driver-alletra-6k/4eb19e5/4eb19e5.html : FAILURE in 1h 12m 31s","accounts_in_message":[],"_revision_number":6},{"id":"eb29fc470c8f4eaa5176f9a0256037b0cf9b0a7b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-12 23:28:32.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4498d711f85b431796987272659ed77e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/58e35bf65f244e908a4bf7aa14ea1ca5 : FAILURE in 1h 13m 53s","accounts_in_message":[],"_revision_number":6},{"id":"94ab355050d19f5d9ba966c0d3d94b1a6ea9e8b1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-12 23:46:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 1h 32m 43s","accounts_in_message":[],"_revision_number":6},{"id":"929330af4bf236128af26484aa0d5d13a56ccd97","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 00:13:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/885825/6/check/iscsi-driver-nimble/74838f1/74838f1.html : FAILURE in 1h 04m 42s","accounts_in_message":[],"_revision_number":6},{"id":"d7009837b6f258499f115b33767281dd4b5698c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-13 00:57:58.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/6/check/DellEMC_Unity_FC/d788076/DellEMC_Unity_FC/263 : SUCCESS in 1h 43m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/6/check/DellEMC_Unity_iSCSI/26bcbb8/DellEMC_Unity_iSCSI/263 : SUCCESS in 2h 04m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"48a1b4de221181823e38d7c24c724742113e3ed3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-13 01:38:47.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-6/FC/ : SUCCESS in 3h 26m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-6/iSCSI/ : SUCCESS in 2h 46m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"01504098addd91056bc46401177e5cea9abb9c68","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-13 02:06:58.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-6/ns5_nfs : SUCCESS in 3h 54m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-6/ns5_iscsi : SUCCESS in 3h 54m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"84e6530657bc2a59330132ed1b9d6b72278bfe52","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-13 06:30:26.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8733ab5f9d4343f9cacb9ed0b635c42b5bb789bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-13 07:02:36.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c65c1838ca48a3a7d7f9751c6f625f79258ca4d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-13 07:03:32.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bbf48727157aa2b3bffb7c1478537875a8304254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 07:14:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41eb6fa566ca4f4b9358b970cd84697e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f496906236f47998f95290bab8abaef : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c39910a871847d9a2a4ed8026fe2a27 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/380e26a315a44f43a5ea416b417052b3 : SUCCESS in 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6eab1095cef06a4667ae6e8c39818e75be521655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 07:38:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ea0e34d8e64f49af517fd3b45dd430\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e1fa49459a024523875cac3a58b33445 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b985dc27b7a744fe80daa6c75bf21c31 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e24916e01cb54cb39797fc70bc0e320d : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c31ada70fab04bfebbd4d988016dcdbc : SUCCESS in 10m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/169c6abd6f5946ddaab38185944a6162 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2872c6b8a94a41b899df3b1bf7d97c : SUCCESS in 10m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b205364dd6c24e1186a38a5023cc879f : SUCCESS in 12m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11eb29b8e16a4b06a7f84391af0f1704 : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/149014c2da744f5798e429f0897ec301 : SUCCESS in 6m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/07053aec89b347e9b7e2ba74458b25f7 : SUCCESS in 6m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aa3e855ce2b146eca08607faa43994d6 : SUCCESS in 34m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29bb29ea7e0d40dc8b00282c185b7814 : SUCCESS in 5m 53s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/66598c630be4484f89ced27a660d978f : SUCCESS in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"da0a6386591d74c34083954e7a88b0b14e1e6cc0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-13 07:45:52.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_06_2023_07_38_52\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"02d832ede04c3f16c817bd44b840fac434cda8ed","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-13 08:07:36.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerMAX_FC/775b2e8/DellEMC_PowerMAX_FC/121 : FAILURE in 28m 36s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerMAX_iSCSI/033de67/DellEMC_PowerMAX_iSCSI/116 : FAILURE in 26m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"2904fe207a3bb3fdf258c352481a0dd1d421c1f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-13 08:23:38.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/d760a9aa112f472aaab929e9997cd21d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/37b657a35fd24e7c9ccb8cd5ea42a08f : SUCCESS in 1h 17m 53s","accounts_in_message":[],"_revision_number":7},{"id":"9402a365fa066db8df5e1a34bd5088cfcde9e96c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 08:26:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/885825/7/check/iscsi-driver-nimble/a48f3b5/a48f3b5.html : FAILURE in 1h 22m 56s","accounts_in_message":[],"_revision_number":7},{"id":"ed3d63fcdd3d5f577c174f177b2b3f1608da3f8f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-13 08:27:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/7/upstream-check/cinder-tempest-plugin-ontap-fcp/3872afe/ : RETRY_LIMIT in 8m 13s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/ae91446/ : SUCCESS in 48m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/7/upstream-check/cinder-tempest-plugin-ontap-nfs/1daa224/ : SUCCESS in 43m 53s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/67d8e48/ : SUCCESS in 44m 42s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"ac7f28da2d3831ad2d7ef161021bf42bfa6626fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-13 08:34:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 1h 30m 39s","accounts_in_message":[],"_revision_number":7},{"id":"052587c0954c6b1cabdb367f151bba5053cdc1f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 08:56:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/885825/7/check/iscsi-driver-alletra-6k/fb6fba4/fb6fba4.html : FAILURE in 1h 52m 27s","accounts_in_message":[],"_revision_number":7},{"id":"9adc5413ccee29bb22ad940f083b0f2a43124121","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-13 09:01:43.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerStore_iSCSI/ebd3cda/DellEMC_PowerStore_iSCSI/99 : SUCCESS in 51m 39s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerStore_FC/47e15b5/DellEMC_PowerStore_FC/104 : SUCCESS in 1h 22m 40s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerStore_NFS/169ad7e/DellEMC_PowerStore_NFS/98 : SUCCESS in 59m 23s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerStore_NVMeTCP/d7206d9/DellEMC_PowerStore_NVMeTCP/99 : SUCCESS in 1h 12m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"b00b83b6f18125a5ffc401ccde13335260ac5fd9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-13 09:07:39.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_Unity_FC/23dcfd9/DellEMC_Unity_FC/264 : SUCCESS in 1h 43m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_Unity_iSCSI/b82470a/DellEMC_Unity_iSCSI/264 : SUCCESS in 2h 04m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"3ea9bc2b6075c2624a9b2194419f875d4d6213a1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-13 09:10:30.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/7/check/DellEMC_PowerFlex/50a7fe9/DellEMC_PowerFlex/146 : SUCCESS in 1h 31m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b0160116c1fc1038f76d3079a98f292d4b55f29f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-06-13 09:40:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/885825-7/Lenovo-iSCSI/ : FAILURE in 36:33.08\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/885825-7/Lenovo-FC/ : FAILURE in 36:08.82","accounts_in_message":[],"_revision_number":7},{"id":"8db009c51b623704466ca42d4cb38dc77400ceab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-13 10:43:21.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-7/FC/ : SUCCESS in 2h 32m 44s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-7/iSCSI/ : FAILURE in 3h 40m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"d413f10ba4c37a93622eab7bb75e1ef52fd26100","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-13 10:58:38.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-7/ns5_nfs : SUCCESS in 3h 52m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-7/ns5_iscsi : SUCCESS in 3h 55m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"9fcd94cb3b647a5e8280af07e407f9ad72d7e200","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-13 11:44:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885825-7/HPMSA-iSCSI/ : FAILURE in 1:47:32\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885825-7/HPMSA-FC/ : SUCCESS in 1:28:01","accounts_in_message":[],"_revision_number":7},{"id":"8b5af13165abb75a634e06184b4929b82200fcae","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-13 14:03:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-7/PVME-iSCSI/ : SUCCESS in 1:57:11\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-7/PVME-FC/ : SUCCESS in 1:37:48","accounts_in_message":[],"_revision_number":7},{"id":"fb24c3abf609739e747cf0356fddf7407cb08fed","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-14 09:34:16.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"18e25da00e74d1549d86d208bcd03705b09ffe60","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-17 15:14:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/7/20230617211439/NEC-V-ISCSI : SUCCESS in 2h 41m 14s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/7/20230617211439/NEC-V-FC : SUCCESS in 2h 59m 27s","accounts_in_message":[],"_revision_number":7},{"id":"639cf9bff32d60f20f670442d48dcd791c672997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-17 18:10:45.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8899e4dcd3238129e993d6f3eab88d2a76eab757","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-17 18:11:45.000000000","message":"Patch Set 8:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":8},{"id":"437ffe2e46680b29cc36e3bf8ab221718f74ec15","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-17 18:12:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"56c2a7e85273a9ad1332dcaead32eaa71e17d4c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 18:26:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c306a69601741beaf2ba4c0126083a5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/301571fa831e4edd9df682ab5c1806c4 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cffaeb602dc41c4a12ceebd718cf14f : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e0ce679c25ab479986720557c199432c : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b192db7792cbc60f7c28d9583a88e8eefa557d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 18:44:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c98fb46df24d968a6cc4741a9c1ec0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf458918501746cf9116785f31ed8bc7 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca4b3a1b99154cbf9f07ddc978705a84 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d510a13d011646368f857ab96db56a89 : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad2fc9ffec944326a7623fdea48f1b87 : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1a555f68db0492bae4e244a872da196 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db59537b006f46db8fe4deb3c41a4428 : SUCCESS in 11m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/11a16615ad194338adcc0b4eefcbfe46 : SUCCESS in 17m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f09e349c758f47a9a3dd19c8291dd20f : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7195cf56f09a453c97ca012fb86a2ff3 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3fb4961ec861455abf3774b4432cf84e : SUCCESS in 6m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/266689e7d3d24aa8870865f29935d969 : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e576119b6f7048539045359ec74df184 : SUCCESS in 5m 22s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/f74d5f5503154fd7bf22c9c28323d7b8 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8a6abde531ac452fcb70d4d80c7bdd21672ee73c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-17 18:51:36.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_06_2023_18_44_31\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"8d80dacdb7bfdc247c0f987245720342e4bad266","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-17 19:25:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/885825/8/check/iscsi-driver-alletra-6k/db73dee/db73dee.html : FAILURE in 1h 12m 32s","accounts_in_message":[],"_revision_number":8},{"id":"f8ca9f5e5b9880e8f22a33b7f6f9659ac3b28197","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-17 19:27:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/c7e8a6eb67dd41b4848ec452a5561687\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0e8d9e6e29f8464c963caa4137a7c7f3 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":8},{"id":"f203ef18df642d1c21dcb06e6829ba926c691aa5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-17 19:30:24.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerMAX_FC/b3c1260/DellEMC_PowerMAX_FC/137 : FAILURE in 45m 45s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerMAX_iSCSI/62409af/DellEMC_PowerMAX_iSCSI/131 : FAILURE in 28m 02s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"3c7282539f89cd7fbfb110866f249a0d13c00682","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-17 19:56:40.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_Unity_FC/6da6c6a/DellEMC_Unity_FC/298 : SUCCESS in 1h 45m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_Unity_iSCSI/4dae854/DellEMC_Unity_iSCSI/300 : FAILURE in 18m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"2e4b85c683edc3f5514f961bbb2bb2fcccbc045d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-17 20:09:19.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerStore_iSCSI/f0bf9d9/DellEMC_PowerStore_iSCSI/113 : SUCCESS in 56m 51s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerStore_FC/7e81365/DellEMC_PowerStore_FC/118 : SUCCESS in 1h 24m 37s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerStore_NFS/b358481/DellEMC_PowerStore_NFS/112 : SUCCESS in 57m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerStore_NVMeTCP/10a3d62/DellEMC_PowerStore_NVMeTCP/113 : SUCCESS in 1h 10m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"8dc8101e999f918f6394b98ceee125815e5624b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-17 20:26:30.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/8/check/DellEMC_PowerFlex/b6fe539/DellEMC_PowerFlex/154 : SUCCESS in 1h 41m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"b3d16ad8bdeeac023c8304504fdaa1098c149305","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-17 20:31:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/8/upstream-check/cinder-tempest-plugin-ontap-fcp/46e19c5/ : RETRY_LIMIT in 8m 00s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/d760fd9/ : SUCCESS in 1h 36m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/8/upstream-check/cinder-tempest-plugin-ontap-nfs/fa65b72/ : SUCCESS in 1h 39m 37s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8f6d3ab/ : FAILURE in 1h 43m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"8125fab63e0689c7b3694c550a6918c4ef3c6700","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-17 22:00:25.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-8/ns5_nfs : SUCCESS in 3h 47m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-8/ns5_iscsi : SUCCESS in 3h 49m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"bbe59e38d055f14df8958cd9cfe9c55ef505f507","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-17 22:24:21.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-8/FC/ : SUCCESS in 2h 25m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-8/iSCSI/ : FAILURE in 4h 13m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"9d61f83f73703d1e6514252763ed0f743224fccf","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-17 22:49:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885825-8/HPMSA-iSCSI/ : SUCCESS in 1:46:44\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885825-8/HPMSA-FC/ : SUCCESS in 1:28:12","accounts_in_message":[],"_revision_number":8},{"id":"b145baf119732bb801eb64f88e9b715e0d0d8409","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-18 00:36:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-8/PVME-iSCSI/ : SUCCESS in 1:57:46\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-8/PVME-FC/ : SUCCESS in 1:35:49","accounts_in_message":[],"_revision_number":8},{"id":"805d92a206115060c8d475446665ce5f5e0b2f8d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-20 11:32:02.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 1h 15m 38s","accounts_in_message":[],"_revision_number":8},{"id":"ba1cc036379f7bb70c41e79ecfe3437cb9d1e648","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:24.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/7/fc : SUCCESS in 0h 36m 51s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/7/iscsi : SUCCESS in 0h 43m 51s","accounts_in_message":[],"_revision_number":7},{"id":"db93fa652f9ceaca908774317624804b3cda9a70","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:28.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/8/fc : SUCCESS in 0h 32m 54s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/8/iscsi : SUCCESS in 0h 43m 25s","accounts_in_message":[],"_revision_number":8},{"id":"523da0420ae90662ff58872a280745411a23bcfd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-21 20:40:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/8/20230622024944/NEC-V-ISCSI : SUCCESS in 2h 50m 22s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/8/20230622024944/NEC-V-FC : SUCCESS in 2h 50m 45s","accounts_in_message":[],"_revision_number":8},{"id":"f0a5b6f913812da027ade8ec7e5d0073b234b75f","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-22 06:52:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fba643dca704af76cf7a4d3853240099aee0973d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-22 22:39:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/8/20230623040606/NEC-FC : SUCCESS in 3h 33m 22s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/8/20230623040606/NEC-ISCSI : SUCCESS in 3h 33m 33s","accounts_in_message":[],"_revision_number":8},{"id":"816b506914c55edeca3f946c42f0f3df3920bdc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-06-23 03:44:57.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"adb6bb97d970d913655bf1a3c0d4ec6eff29617f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 03:45:50.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4383f78ec610cc1463a0a6c349d8d518749071bb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 03:47:02.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-9/ns5_nfs : FAILURE in 1m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-9/ns5_iscsi : FAILURE in 1m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"0f1f72ef1a0f650c493617ee249fb4fc96ebc11e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 03:58:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8febabc5324de0ad21db60a5970616\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d87b72bd16a8450e9763271b11ffe837 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18a809201f7742a0abfc168ca76d1fe9 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81e1d2b0baf14186b0c2269c0b288de3 : SUCCESS in 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7b18e051afbb51bf6fb1d1cb5c1f60edd62379f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 05:02:21.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 1h 13m 48s","accounts_in_message":[],"_revision_number":9},{"id":"e6217cae2a13baceeb9b15ac61487172427dc3ed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 05:02:26.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/885825/9/check/iscsi-driver-alletra-6k/2f7eedb/2f7eedb.html : FAILURE in 1h 13m 14s","accounts_in_message":[],"_revision_number":9},{"id":"fde846a1795ec1c340d4409f50cca15ca0a685df","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 05:16:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/09578320ff7b4df58419ad9e01f4234a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/16cfd428e3c94eca895d789ec5838178 : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":9},{"id":"94831df1e092d752cfc99e193fbc83f186fde6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 05:48:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be24ad1cc5364da4af2635fd54321dc4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ff701891f2db4b3dbe4ebba3ba8bdaba : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f77ae3ef9840d58057654a51012950 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3085c49bb2e4901a68b5880d1f8e532 : SUCCESS in 12m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/238b3cdcc4ad404a95742ba99de96839 : SUCCESS in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59dba5f17f4642a2946eb3c2ff9fda42 : SUCCESS in 15m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577235a28c624487afe0ad2d04e5a2a0 : SUCCESS in 13m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0480c1d670694ba1bdb3ab6e8f794adf : SUCCESS in 17m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/173963f54f17445987a309829a5b736a : SUCCESS in 6m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d44705b22834a65893fa0cdcbba1f76 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8f8d88cc51b49e692a5228076c39da4 : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/998e62cf4ce34a69a4da46a818b13bea : SUCCESS in 1h 07m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d76b3123255f4c8693d0d34262563a93 : SUCCESS in 5m 36s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/93c423ffc4e94b1ab8c07f7945c67226 : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"df6f48b6e987078fed2a51e0eb23527b29aea973","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-23 05:55:47.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_06_2023_05_48_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"75a449e045760b2676e55bf7b3e3289c7e99501b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-23 07:12:43.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerStore_iSCSI/59dc6a3/DellEMC_PowerStore_iSCSI/122 : FAILURE in 58m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerStore_FC/388e436/DellEMC_PowerStore_FC/127 : FAILURE in 1h 23m 45s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerStore_NFS/9b4b118/DellEMC_PowerStore_NFS/121 : FAILURE in 1h 01m 44s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerStore_NVMeTCP/be5ef00/DellEMC_PowerStore_NVMeTCP/122 : FAILURE in 1h 14m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"013813a21f4073b368f86213647b4d64b9ebb824","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 07:27:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/885825/9/check/iscsi-driver-nimble/c87d2d1/c87d2d1.html : FAILURE in 1h 04m 39s","accounts_in_message":[],"_revision_number":9},{"id":"1db876298596e90fcc319518c5761f745a3d231b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-23 08:29:25.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-9/FC/ : FAILURE in 4h 36m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-9/iSCSI/ : SUCCESS in 3h 8m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"b192b79cde17e78d9cb8beb647bfcdffeacd823c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-23 10:03:37.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885825-9/HPMSA-iSCSI/ : SUCCESS in 1:53:56\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885825-9/HPMSA-FC/ : SUCCESS in 1:29:53","accounts_in_message":[],"_revision_number":9},{"id":"eefbdfdcd2fb56aef23b76991e9186014947d56c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-23 10:34:56.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-9/PVME-iSCSI/ : FAILURE in 2:15:37\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-9/PVME-FC/ : SUCCESS in 1:40:15","accounts_in_message":[],"_revision_number":9},{"id":"49cd819e4a3f4a4270fe7bf11ef957cbbd55f525","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 12:59:30.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerMAX_FC/ca94158/DellEMC_PowerMAX_FC/152 : FAILURE in 7h 10m 37s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerMAX_iSCSI/5e3281c/DellEMC_PowerMAX_iSCSI/145 : FAILURE in 6h 19m 57s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":9},{"id":"e5f38c2a0bce1d9bb23ba1c07406f7a832b981dd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 13:16:42.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/9/check/DellEMC_PowerFlex/6d4d87b/DellEMC_PowerFlex/193 : SUCCESS in 1h 50m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"0fd8ffcb916be212c880f6f518fe559961b67567","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-23 14:30:22.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0cac9bb0dd8866711ffad6fe1d321dbae37e37e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-25 02:00:08.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/9/fc : SUCCESS in 0h 34m 44s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/9/iscsi : SUCCESS in 0h 51m 41s","accounts_in_message":[],"_revision_number":9},{"id":"e2b82092c00fdcf44a8810309c525ef777b970df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-27 13:39:52.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"14bca4004e6f5c15774c8ce394e4275aad20e677","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-27 14:05:26.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"82a3a238f2f149cbe25688a6a0df9d2e0166be4f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-30 08:59:14.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/9/20230630100807/NEC-FC : SUCCESS in 3h 31m 05s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/9/20230630100807/NEC-ISCSI : SUCCESS in 7h 51m 02s","accounts_in_message":[],"_revision_number":9},{"id":"73c6ea99078d665324f55445c56a350edd4be8c7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-01 23:13:44.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/9/20230702051244/NEC-V-ISCSI : SUCCESS in 2h 35m 10s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/9/20230702051244/NEC-V-FC : SUCCESS in 3h 00m 57s","accounts_in_message":[],"_revision_number":9},{"id":"94e6b1b6366f8a4618d40f5e3b5b16cee1345a69","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-03 12:10:25.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-9/FC/ : SUCCESS in 3h 2m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-9/iSCSI/ : SUCCESS in 2h 14m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"3b7bb348aef5aeb712dc0c42ee62211e18b414ea","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-07 11:22:38.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6e9ca68b1618c6146b37bd7183f01e7d7924365e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-07-07 14:37:08.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"435ba6589388c94f21d1f2abe52311366369d268","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-07 14:38:56.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-10/ns5_nfs : FAILURE in 1m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-10/ns5_iscsi : FAILURE in 1m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"d06bf8ec0214103758e375e6b3053e7050e37fe1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-07 14:42:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5756d27e756b4e3c84b629ca4e239245 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a472bb3dbab046a6b0b87dbb0cb207f0 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"4002fe1311bf38c9ef92f142ae1cb111a8b8157b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 14:51:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca8f4569edf744d7880ab4565f8996ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f79ce558a522466bad3b41725240a2fe : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2c030adf75749788b08fe0836dd9a75 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca442bf405044bd99795a8c2870dcece : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7fcdd1297b5f44174d11067afaa419ac194a77c3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-07 15:16:01.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885825 : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":10},{"id":"85777a75d40ece064090952e78e0b5ef5533a0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:35:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20224fcef9cb45e1a3e6b6305b33ec9c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6295b0a8f75c4e01b30db3db0ed4e187 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bfd5b6b8a54415a39e6ccacefec7af : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5517ef042b8c4179898374e335102fef : SUCCESS in 12m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee9d60d6408847488715537a5e2b0e6f : SUCCESS in 13m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fedaec29e47493c841fc64e2093bd70 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e58e71edd7d8448ebaae344f5f60fd88 : SUCCESS in 12m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3da678f7dfe74608b37f51d7d564871c : SUCCESS in 15m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/289bef8e09514eb09fda853c8706e5ae : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05ac8cd3cc98457faae6f55ea53f448b : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d0defc97aa08458382e8291f7106142e : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4e5c94e1f15846cdbb0b7508302df086 : SUCCESS in 51m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/31f3d88e27c449d6aa79c5b9f1ea1e5f : SUCCESS in 7m 23s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/bde2da67f46244baaba04aaa6ef99719 : SUCCESS in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c18e03d05b3da02685fda600b5163284ed951085","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-07 15:42:45.000000000","message":"Patch Set 10:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/07_07_2023_15_35_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"092b1593dbfe2f4dde41de2563828c3d77bcf97b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-07 16:18:06.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_PowerFlex/83d48b4/DellEMC_PowerFlex/265 : SUCCESS in 1h 40m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"9615456dd361747f5b24ad0b1ec41a9fde6d1fb6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-07 16:29:45.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_Unity_FC/f516de7/DellEMC_Unity_FC/394 : SUCCESS in 1h 40m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_Unity_iSCSI/8c188cd/DellEMC_Unity_iSCSI/396 : SUCCESS in 1h 52m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"ba26f0d8105c40c6142dc7790359d73ca751bb61","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-07 16:42:19.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2c4c5392897842f1a215abb1e6f8f7e8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7b10218b3dfa4530b51702e24f6de019 : FAILURE in 1h 13m 25s","accounts_in_message":[],"_revision_number":10},{"id":"9868aaaa5cd9a3f44e94be009034c5adfa32399a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-07 16:53:52.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/10/upstream-check/cinder-tempest-plugin-ontap-fcp/96702c5/ : RETRY_LIMIT in 8m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/3c2b3e1/ : SUCCESS in 44m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/10/upstream-check/cinder-tempest-plugin-ontap-nfs/c6288fc/ : SUCCESS in 1h 16m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/207524b/ : FAILURE in 44m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"d7f33c72a40dd7d7b012c8a75d11d96c52e66811","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-07 16:54:33.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_PowerStore_iSCSI/3cb21f4/DellEMC_PowerStore_iSCSI/143 : SUCCESS in 54m 10s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_PowerStore_FC/a6f2298/DellEMC_PowerStore_FC/151 : SUCCESS in 1h 19m 00s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_PowerStore_NFS/27ac357/DellEMC_PowerStore_NFS/143 : FAILURE in 34m 23s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/10/check/DellEMC_PowerStore_NVMeTCP/145a9a5/DellEMC_PowerStore_NVMeTCP/143 : SUCCESS in 1h 10m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"959781792a0bad0d2142281636983a75be731ea4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-07 17:00:08.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/10/fc : SUCCESS in 0h 43m 35s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/10/iscsi : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":10},{"id":"efd4a8f6f048c96ffc21ffcb6fe233ca25b99f17","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-07 17:17:31.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-10/FC/ : SUCCESS in 2h 40m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-10/iSCSI/ : SUCCESS in 1h 49m 50s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"6372e4640bd70b8baf890e9c61042d4c84fdab27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-07-07 17:52:56.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"1241643a9edcac99a3ff8e06929088da89a47df5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-07 17:54:47.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-11/ns5_nfs : FAILURE in 1m 23s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-11/ns5_iscsi : FAILURE in 1m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"c59380ba8114be9e95e58b685cedf35d9ecc05fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-07 17:58:23.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a9df9b5c3c6646d99f0b96421475bcf1 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/de4eaf1ca3604aa8b35bf930a15da155 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"e993278f3c4a7b348a07dbeccdccf16d54197df7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 18:03:57.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc9f79ec32f4e0289adda9ded517234\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a7c80b4e41b437492ea0ed78280f160 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c6cf448640c4f818f24527bb78cf5a2 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03bf6027a875475c964be240483ea705 : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"826e9d1e2939fab2e5788dbb39674003a835f6dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 18:38:23.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0319f59dc01f408d80e22d7555e57d05\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/28ff0383cb0b42eb8d92bee0ee808a1c : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b360e553c64e31b40567c0c260469c : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98c9ab595edd4aec82c356537ab52f68 : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21a29753391c449b80e6dc89faf5200a : SUCCESS in 9m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bf9871cc3a940e294fe097382eae101 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b543902fd21543b98ba969e05d27b94f : SUCCESS in 13m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ff4ef72cbdb438592e9b68e998c6f69 : SUCCESS in 16m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d222567c07a04a379947c7c713938f30 : SUCCESS in 4m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a24aed942ad04637a4b4d41879e8f6f9 : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fea5354ec9364db799f6f1de7bc7aaab : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6c7f814fb22d4f10af7ebd8dc4ffd270 : SUCCESS in 44m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f46baf9d0e754f3daf8a71123e619848 : SUCCESS in 5m 09s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/c68b28ae8f654e5fb7b55c4c633b76c4 : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a880a46a3387e064b54d50481ed6dcc6838a688d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-07 19:09:07.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/43448297de65440ea89504de3a2ca5b8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2b2d0e5b2b2e4ab79ab82b2772e3105d : FAILURE in 1h 13m 19s","accounts_in_message":[],"_revision_number":11},{"id":"c52b19de28804e27483a833d668133e4d555217b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-07 19:49:52.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885825-10/HPMSA-iSCSI/ : SUCCESS in 1:53:31\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885825-10/HPMSA-FC/ : SUCCESS in 1:29:08","accounts_in_message":[],"_revision_number":10},{"id":"68ce72fbb5716cae1710f0fbfc8c66234c5c89de","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-07 20:00:08.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/11/fc : SUCCESS in 0h 30m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/11/iscsi : SUCCESS in 0h 41m 47s","accounts_in_message":[],"_revision_number":11},{"id":"e49bc5c426e14b6008fa0a39eb993e722073f15f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-07 20:09:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-10/PVME-iSCSI/ : SUCCESS in 2:01:49\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-10/PVME-FC/ : SUCCESS in 1:40:52","accounts_in_message":[],"_revision_number":10},{"id":"a77c56684ada3f56f47ff84a540ef73087b53147","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-07 20:35:56.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-11/FC/ : SUCCESS in 2h 42m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-11/iSCSI/ : SUCCESS in 2h 2m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"be8fe2e0269fa0202558060044be37cd89a51ebf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-07 20:48:59.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/11/check/DellEMC_PowerFlex/17816c2/DellEMC_PowerFlex/267 : SUCCESS in 1h 47m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"8c8208c4d47cdcff96492e2580400e08a184c277","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-07 21:28:26.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/11/check/DellEMC_Unity_FC/6b7601b/DellEMC_Unity_FC/396 : SUCCESS in 1h 44m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/11/check/DellEMC_Unity_iSCSI/dd5baf0/DellEMC_Unity_iSCSI/398 : SUCCESS in 2h 00m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"6929480f2c2eddb753dbdfc1ab5da7cf7373d2d2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-07 22:10:23.000000000","message":"Patch Set 11:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1558/","accounts_in_message":[],"_revision_number":11},{"id":"3efbc60ffa23d6fe3ca3c2e61e2677405fbc954c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-08 02:54:51.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/25/885825/11/check/tyds_test/373e986 : FAILURE in 4h 49m 45s","accounts_in_message":[],"_revision_number":11},{"id":"6fbf48353d040756a997e57eea38ffbac7abe844","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2023-07-08 07:52:50.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"82b4e181ae786074bb0c47b5754a2d66ebf6c4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-07-08 08:02:55.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6f3a430066772e3c04ff36437fad5ef37ccf039f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-08 08:04:44.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-885825-12/ns5_nfs : FAILURE in 1m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-885825-12/ns5_iscsi : FAILURE in 1m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"5fdd3a7a8c35273a9d3bef30eb3c18baa07215f5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-08 08:08:29.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/86d50aba58a9487597622db3e1d07a10 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1af9d8aa9ca74a6192d1102e566e3331 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"46f23a88659e7b912ccc5a2b28b186fc1d34764d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-08 08:19:23.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b588f415665497a83820dca707c96a6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2641dbec80c440e81ce348ac29b31f6 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57a0a2dd075144c5b4121c66dce80489 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4973ad2e891a4b19bb771897fb69c2ec : SUCCESS in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"eb99a0d3968c878f2ef4541a54e37fb893bf4327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-08 09:09:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e418a278ec74491bd0acb6325e08116\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a0e7197985a44ec8a46a379d2d35c07f : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50c411f95e44cc19ae78d103056532a : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d937b412c6e44e6b665eaa7454058f5 : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42693074d19e428986c1bf527f83d846 : SUCCESS in 11m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7711b17926fc46959d5500918a0e1750 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79db98e7251472cae26d68dab6f98af : SUCCESS in 13m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/87af24a9b9064c5ea65812447cd4f0ad : SUCCESS in 16m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8dcf7d0f3a774edf85ed6af6d5409d79 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/516afe4b25a643fcad48f6e5217dfb8f : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe1882c874dc4723bdb5b9f1022c546a : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ddd69df1f3446c2acc4c3a4ea3fe59d : SUCCESS in 50m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08fb66bf972440958177a8feb07a4cdc : SUCCESS in 6m 12s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5f79ca07986845098b1617ad640b82dc : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dda1e0a23f177cd894d78cce19867d320ced8c2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-08 09:16:13.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_07_2023_09_09_22\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"84b7ef0c212e0328a2b5e9d150629a1efc661261","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-08 09:19:47.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/546376160ce84f9c891097c6927c79d0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/81a0352a3aa144f1a762e3114c098eb4 : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":12},{"id":"ee4d38bd01f1c407dba9f9ec107062effefdbfe4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-08 09:57:15.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_Unity_FC/d1a3527/DellEMC_Unity_FC/398 : SUCCESS in 1h 43m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_Unity_iSCSI/8f67e10/DellEMC_Unity_iSCSI/400 : SUCCESS in 1h 53m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"2be129e6d71edc460a00c7a8e55da1680d44e1b1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-08 10:04:47.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_PowerFlex/78c0204/DellEMC_PowerFlex/269 : FAILURE in 2h 01m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"e9924f9b30238ab6ab76a3771c0bec1fcd39b5e6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-08 10:32:20.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_PowerStore_iSCSI/12fe8d2/DellEMC_PowerStore_iSCSI/146 : SUCCESS in 52m 05s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_PowerStore_FC/06b17c7/DellEMC_PowerStore_FC/154 : SUCCESS in 1h 22m 43s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_PowerStore_NFS/7159bfe/DellEMC_PowerStore_NFS/146 : FAILURE in 34m 25s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/885825/12/check/DellEMC_PowerStore_NVMeTCP/2c19ac9/DellEMC_PowerStore_NVMeTCP/146 : SUCCESS in 1h 07m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"c54baad5b4fa91c2bed61aa55672581cdd080be4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-08 10:37:35.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/12/upstream-check/cinder-tempest-plugin-ontap-fcp/9ddf832/ : RETRY_LIMIT in 8m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/ab5af04/ : FAILURE in 1h 27m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/12/upstream-check/cinder-tempest-plugin-ontap-nfs/359913a/ : SUCCESS in 1h 24m 03s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/25/885825/12/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/bfa6aa4/ : SUCCESS in 1h 23m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"7988341ce8756353f89d5229e6c7f7fde4c93d3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-08 10:46:21.000000000","message":"Patch Set 12:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1560/","accounts_in_message":[],"_revision_number":12},{"id":"f7a745bb418ef123b89254c73ccca1a9a7c4ce7c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-08 11:00:07.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/12/fc : SUCCESS in 0h 38m 47s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/885825/12/iscsi : SUCCESS in 0h 50m 02s","accounts_in_message":[],"_revision_number":12},{"id":"1fbcb73f2ec945aa7a93512758151a7821e6c1cd","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-08 13:25:03.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/885825-12/HPMSA-iSCSI/ : SUCCESS in 1:53:35\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/885825-12/HPMSA-FC/ : SUCCESS in 1:30:48","accounts_in_message":[],"_revision_number":12},{"id":"0117ca3e2a45c4278d5e143145cff0b70487c08f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-08 13:49:20.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/885825-12/PVME-iSCSI/ : SUCCESS in 2:05:50\n- PVME-FC https://os-logs.seagate.com/dell-ci/885825-12/PVME-FC/ : SUCCESS in 1:42:06","accounts_in_message":[],"_revision_number":12},{"id":"f6a7bba40ecf13df6515c66009d09f97cc12ed16","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-08 15:19:18.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-12/FC/ : FAILURE in 7h 16m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-12/iSCSI/ : SUCCESS in 1h 50m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"f4bb82de18a89fc977b058d06bcb23016dfb4b99","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-08 23:50:18.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/12/20230709054740/NEC-V-ISCSI : SUCCESS in 2h 39m 27s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/12/20230709054740/NEC-V-FC : SUCCESS in 3h 02m 34s","accounts_in_message":[],"_revision_number":12},{"id":"373668783b06f994d01d9487c9a7a1b0d56a5fc8","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-09 01:14:06.000000000","message":"Patch Set 12:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/25/885825/12/check/tyds_test/14c988a : FAILURE in 6h 49m 37s","accounts_in_message":[],"_revision_number":12},{"id":"4cccb5c1d3fb39aa213c4b7fe7e3442ebcc1a2f8","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2023-07-09 03:31:55.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"07ecdd5928fd2ee00e630e08917334a68ecbc699","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-09 07:04:03.000000000","message":"Patch Set 12:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/25/885825/12/check/tyds_test/8da1696 : SUCCESS in 3h 31m 50s","accounts_in_message":[],"_revision_number":12},{"id":"342471c21c5643a0f467ef72aadbe8f5444b8940","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-10 07:34:39.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3af0d9147ec94d4eaf1d2e0955b21cf6 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e6459124f4e54742954cb7f88d3a7d08 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"23e43a87b279f811b3f216f9682f68abd0f38b9e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-10 07:44:38.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/982e54bb83134cbc915da2212bf3952f : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c1889086f41045b7aedaae0ed94b882f : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"7981924121d7cfb483371e5b2c6e6919a93be0fa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-10 13:44:54.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-12/FC/ : SUCCESS in 3h 35m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-885825-12/iSCSI/ : SUCCESS in 2h 40m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"b7f8867d72bdb4e7bfaf5f67eed01045daf411a5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-10 21:48:45.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/12/20230711024730/NEC-ISCSI : SUCCESS in 3h 49m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/885825/12/20230711024730/NEC-FC : SUCCESS in 4h 01m 12s","accounts_in_message":[],"_revision_number":12},{"id":"8a66caf4b3534986272cad89e0a87a2ecaa2b5b6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-13 14:54:38.000000000","message":"Patch Set 12: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0b594dc5ae39a799dd53e8a4ff9b2928b6967d44","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-17 10:56:53.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1d044d804953d9996ff5671cbdd39a6fc8ed6dc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 10:57:18.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"0fa9c508704744e1e80cb339b0b9804c46d3c01d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:34:14.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f64a74ef60040d094fa855d5aa17c2b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/17e329041b5846ce85f5e07af42c82da : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4fae02c80344818983fa75386636b0 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f14b7878a18445a18e77e77f747a4d8a : SUCCESS in 13m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c299e1fbd64d4c228caa75ee277be14f : SUCCESS in 10m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3eeb43f731bd420d815279d769bf1fb7 : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ecd8105eb74404ac3e31b2739f1b5e : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":12},{"id":"0e86595df37584d43527b81453506519508e2186","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:34:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"d88574f675b360af0cd709274a32396396f93f5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:35:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0277f0f3cd3642229b5da67940da1eb6\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/15ff831e3d9a47c7878ad98810074de6 : SUCCESS in 55s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e380c54bfd243abae42ed8c5407ca39 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"79a9210c52bca5fd633ac4cfa429e640a78c7c2c","revisions":{"897e766c32a83e00b65e307cefebea134cc78167":{"kind":"REWORK","_number":1,"created":"2023-06-10 22:00:52.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/1"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:26:32.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/897e766c32a83e00b65e307cefebea134cc78167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/897e766c32a83e00b65e307cefebea134cc78167"}]},"branch":"refs/heads/master"},"eb01691812a818918c68990286ee5654bc168dff":{"kind":"REWORK","_number":2,"created":"2023-06-10 23:06:34.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/2"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 23:04:56.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb01691812a818918c68990286ee5654bc168dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb01691812a818918c68990286ee5654bc168dff"}]},"branch":"refs/heads/master"},"1847fba4eeed1e6b4641e450ea52ebaeb9c1b504":{"kind":"REWORK","_number":3,"created":"2023-06-11 00:31:49.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/3"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-11 00:10:36.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1847fba4eeed1e6b4641e450ea52ebaeb9c1b504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1847fba4eeed1e6b4641e450ea52ebaeb9c1b504"}]},"branch":"refs/heads/master"},"8d212d53e2c90454c71c4573ecb05c6baa29573e":{"kind":"REWORK","_number":4,"created":"2023-06-11 19:10:26.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/4"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-11 19:09:51.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d212d53e2c90454c71c4573ecb05c6baa29573e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d212d53e2c90454c71c4573ecb05c6baa29573e"}]},"branch":"refs/heads/master"},"851b2d76eb48da0acf8020ef281d74c2a70f3fb3":{"kind":"REWORK","_number":5,"created":"2023-06-11 19:17:14.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/5"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-11 19:16:21.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/851b2d76eb48da0acf8020ef281d74c2a70f3fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/851b2d76eb48da0acf8020ef281d74c2a70f3fb3"}]},"branch":"refs/heads/master"},"89d1b276759213fb10f21270d1d5ec323a2cb9f3":{"kind":"REWORK","_number":6,"created":"2023-06-12 22:11:51.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/6"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-12 22:10:30.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89d1b276759213fb10f21270d1d5ec323a2cb9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89d1b276759213fb10f21270d1d5ec323a2cb9f3"}]},"branch":"refs/heads/master"},"ae1c63b875f1fe6739170e91cbaf8d8f0ae7367c":{"kind":"REWORK","_number":7,"created":"2023-06-13 07:02:36.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/7"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-13 06:56:59.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1c63b875f1fe6739170e91cbaf8d8f0ae7367c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae1c63b875f1fe6739170e91cbaf8d8f0ae7367c"}]},"branch":"refs/heads/master"},"dfcd4ec35fde646192969e306909c062015c262a":{"kind":"REWORK","_number":8,"created":"2023-06-17 18:10:45.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/8"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Barine Desire","email":"desirebarine16@gmail.com","date":"2023-06-17 17:57:35.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer create for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfcd4ec35fde646192969e306909c062015c262a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfcd4ec35fde646192969e306909c062015c262a"}]},"branch":"refs/heads/master"},"a0d5d820f3eb80c75ca1017b67fc3dcbc0a12b47":{"kind":"REWORK","_number":9,"created":"2023-06-23 03:44:57.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/9"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Barine Desire","email":"desirebarine16@gmail.com","date":"2023-06-23 03:42:33.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0d5d820f3eb80c75ca1017b67fc3dcbc0a12b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0d5d820f3eb80c75ca1017b67fc3dcbc0a12b47"}]},"branch":"refs/heads/master"},"9852c23dae651c08c4d0ac21791200af7c35b556":{"kind":"REWORK","_number":10,"created":"2023-07-07 14:37:08.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/10"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Barine Desire","email":"desirebarine16@gmail.com","date":"2023-07-07 14:30:30.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9852c23dae651c08c4d0ac21791200af7c35b556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9852c23dae651c08c4d0ac21791200af7c35b556"}]},"branch":"refs/heads/master"},"8ba8ec6c63d88dce274d8b46fdae3f5016d4bbb3":{"kind":"REWORK","_number":11,"created":"2023-07-07 17:52:56.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/11"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Barine Desire","email":"desirebarine16@gmail.com","date":"2023-07-07 17:50:35.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ba8ec6c63d88dce274d8b46fdae3f5016d4bbb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ba8ec6c63d88dce274d8b46fdae3f5016d4bbb3"}]},"branch":"refs/heads/master"},"79a9210c52bca5fd633ac4cfa429e640a78c7c2c":{"kind":"REWORK","_number":12,"created":"2023-07-08 08:02:55.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/25/885825/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/885825/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/885825/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/885825/12"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-06-10 21:08:43.000000000","tz":0},"committer":{"name":"Barine Desire","email":"desirebarine16@gmail.com","date":"2023-07-08 07:56:50.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer for microversions 3.55 and above\n\nPartially-Implements: blueprint generating-sample-files\n\nSigned-off-by: Desire Barine \u003cdesirebarine16@gmail.com\u003e\nChange-Id: Ifdd99433563ab72d8d43a374af0377213834d46e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79a9210c52bca5fd633ac4cfa429e640a78c7c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79a9210c52bca5fd633ac4cfa429e640a78c7c2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
