)]}'
{"id":"openstack%2Fcinder~885917","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~Ie2b7a163ee3a83121c04a21808ef437d740426d5","project":"openstack/cinder","branch":"stable/victoria","topic":"bug/1981354-stable/yoga-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-03-06 11:24:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie2b7a163ee3a83121c04a21808ef437d740426d5","subject":"Fix Infinidat driver to return all iSCSI portals","status":"ABANDONED","created":"2023-06-12 21:42:39.000000000","updated":"2024-03-06 11:24:41.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":871906,"cherry_pick_of_patch_set":3,"meta_rev_id":"0570ccbc6fee8a35406df36cf416e46e30af8bf0","_number":885917,"virtual_id_number":885917,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-25 12:55:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-12 22:04:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-12 22:49:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-12 22:51:26.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-12 23:30:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 01:10:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-19 01:34:26.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-06 11:24:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"b82568b615b8a3d1ce0c93c0f9da5573c04de97f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-06-12 21:42:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5693912f68ef25c49b6e8f21f500b6f719e0c9cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-12 22:04:29.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3c8307ca60994652aa1e386e92c6cbc3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/61d0aa1fe71144fcba64a8b511ede115 : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3903a5c2343a761ff7e12af511731a6ae705a020","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-12 22:49:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885917 : FAILURE in 1h 05m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f037414bd8922260b158cba9224a06b14c514b91","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-12 22:51:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/885917/1/check/iscsi-driver-alletra-6k/02d7a50/02d7a50.html : FAILURE in 1h 07m 44s","accounts_in_message":[],"_revision_number":1},{"id":"35c9420cccd77719ebe063386aa210570ea9ffd0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-12 23:05:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/885917/1/check/iscsi-driver-nimble/55279db/55279db.html : FAILURE in 1h 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e72cd83a5b8afa14afc88362906912d0c86f8698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 23:30:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6a887bcd05d4cd0b8189066422784a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9fcdc31f8f843e4be6b4867532e591f : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56d813441fea458d90085a6e2586e971 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d96749ce5a6540149ea9d5ee37a49c25 : TIMED_OUT in 40m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ce6921e5064aab9456eb4c6b541f76 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b19f39e4a90d4f1b8655786e77f95342 : SUCCESS in 7m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/75cb7f3fbdab49f89f3bfa9554c9912d : SUCCESS in 10m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa12f424b73247b98a3b2714c26fdcfc : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bbffa770bf694c34aa8e41380f64f02d : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/084e903ea56c415e937c23838f099458 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/edece6eaa20d47a9b86f28b2ed00b1ba : SUCCESS in 46m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9bcd48b2d1246cfafae15585213531b : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1d97c69dca5845ffb988bb1a49f7d2da : SUCCESS in 1h 00m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b0acbef0a1f4150b2ca49732e54aac8 : FAILURE in 35m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1545de35ebf04d6f913794df79e96b62 : SUCCESS in 1h 04m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eea7cd405a3047118dd049ffa0f2599a : SUCCESS in 1h 23m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e0d17063e274287a432d8ffd45efe60 : SUCCESS in 1h 30m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab05202791c846c88227b8eb4637029a : SUCCESS in 1h 41m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bb1c1893482464f8a5a1cf53a49e28f : SUCCESS in 1h 28m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed1701c5bda448919bc703bfc73b34b4 : SUCCESS in 41m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a581fe1c83e7464bac4cbcf5a5b2c07b : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":1},{"id":"eae7568bd4927f759ab314904d26996a16117b69","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-13 01:10:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-17-885917-1/ns5_nfs : SUCCESS in 3h 27m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-17-885917-1/ns5_iscsi : SUCCESS in 3h 23m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"62f3c26f909974192d32264434a1b94584854e59","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-06-14 15:37:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"386ca9bc066f33a23841601f21b241e32e734b84","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-14 15:59:26.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5fef72dc98924dc4ad018e51223257a4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/084f1df9f9ce41e1b4c23aee6a925825 : FAILURE in 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e1481d4b88c9ac69eac7e5f5559dea74bf2cc626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 18:12:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d2ffe2de8ed433da90f17097c44ebb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6c5d4e8e0d4ad8be25bfe60d420c99 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9999df2e02f47c889327775217ebf97 : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60b0519709e045ad83cf09c4c79fec20 : FAILURE in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69b063422344113a16a52029cebc886 : TIMED_OUT in 32m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c96e9b114324fd3ba1ffdbbcd97d2fe : SUCCESS in 7m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/24aec450f5f74e25b3f34623ac2ebaf9 : SUCCESS in 12m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15f80021be524214bb59ada1bd17d279 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b94cb9ac2404fa6b768940b9bd1087b : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af7fca2a9bba425e9ff1b9a61059fe07 : FAILURE in 21m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d5c6acfd17b649ef80da3189da3f0089 : SUCCESS in 47m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa7c56ab94a0490ea85ebfd34b921d38 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd0d46a239654087ba479491c40dd01a : SUCCESS in 1h 04m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f809bed7577446caf6c1294d21105c8 : FAILURE in 1h 25m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f653a7db16ed4c86bb5a8fddbacc13b1 : SUCCESS in 1h 35m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/842715e379cf402dbba419510aadd319 : SUCCESS in 1h 22m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1cd57ad6eda244ceb47050e13427491e : SUCCESS in 1h 28m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3379dea24b7c48458f4ee819a846d4da : SUCCESS in 2h 27m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c352256d79b475888e4dfd349772d70 : SUCCESS in 1h 16m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6008d2755df49a99622c36565ae5d7f : SUCCESS in 1h 03m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04241099d2e54759970d85545effd09a : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d1211d51322fc54407ef0a5405efd5ea6404c24a","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-06-14 18:27:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f96a7c722254bb7181f6d562570c2da86fe5d373","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-14 19:26:53.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/70ddf84516174b1397dc13f1c7157cbe\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3002e98fd38544a982c009448b16ac27 : FAILURE in 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"645c5f2460464762ddc1d03d54d338bc8294e858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 20:43:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/607501df1ff3411eae7b56080ce3dec8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfc14217f6c4cde9acc5c38dc0acf9c : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6caaa0d44a3416aaeabcbaaf1b0a66d : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c1a44470a1f48f5845d2ecb1be56dc1 : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b476a3233584ae6ad2500ab15948c84 : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35dd4538bcf54674b35851f8e4ed3c09 : SUCCESS in 7m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8e44bd256d04592987c1bb20f38f582 : SUCCESS in 12m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15f88d1253bc4f2bb42cd15d6d495624 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4aa37244c52342df8db31fbbe7b4b71b : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ffa430ee8f6487d9c57fa41d6a8592b : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0fc9819c6ef34fe9ae1b80ebf89451bb : SUCCESS in 39m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/978a882d6a3b4424be641b12ef8ecb43 : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1fbb0e0929be4ebd8b2f8b9899f21d8f : FAILURE in 1h 27m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fcbd46f22bec4c91a8f7d6d326cb4caf : FAILURE in 40m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ba6088844e441fcb8dd2cae07f697dd : SUCCESS in 1h 32m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/067e81ff305b4b289a032ea7ee14ccc8 : SUCCESS in 1h 15m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/743fee52683a4ceb8a649ebfed6a5ed0 : SUCCESS in 1h 30m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bfdcc2d4e04540f983af89d1d72a2384 : SUCCESS in 2h 14m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f36934f6c30e4a4c9dca4b8a8be745cc : SUCCESS in 1h 22m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/320bfbcb36144a79a3d5420d6a1a4257 : SUCCESS in 1h 01m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1462920dff40471096d9587ed450352a : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1341f61f9262f5049ed3b04c882a7c63447f28a9","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-13 09:29:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"593ce03bb411f45546f6a50ca21040d88c108ee4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-13 09:51:45.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/174517d93a5d4789af88273d5b1e4def\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7c4cbc8a2b7f4cc3aaeda537699540e1 : FAILURE in 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"04ae7ee659c8cd05f7840212c41330038569722a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 11:11:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b3cc37679cd413f8548b02426198b50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80802190c3e64c7eb21156f4e2156dae : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857aefb5f1b14edaad3b049c23598a79 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de055b8e84d54d8f837eec89b26c5c64 : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f066f3722dcc41268f1f3e7576417f81 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dd38df45d4a4fe8a274adb41213a4e0 : SUCCESS in 8m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33e66ededaaf4981911e396d7e6321af : SUCCESS in 12m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06d3e5b88c414435b3a7a5627f2779de : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eca086fd130146a3ad1b877282a94e70 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa0d6c00fd0d481eb03f55ceb7197b73 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b7f6167cfba423c9d7b9a27fd0df67a : SUCCESS in 37m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29133695e13a49f99401c7c3956fa948 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7bb7235b21f7448cb0f0a1ae28d44189 : FAILURE in 1h 20m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6c35288c7bc94322a3f8827efb98850e : FAILURE in 31m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2122a7a33aea4dffa61905478f2bc10a : SUCCESS in 1h 23m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8419b4dd55e34e11829641e949ebba1d : SUCCESS in 1h 09m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9dc98b8c3a3c45fdad652c6c3110ede6 : SUCCESS in 1h 12m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6ba97b30c2f42b881e1aba56a26351a : FAILURE in 40m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6f83710411243e7aa5d815572fb7219 : SUCCESS in 56m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f29598fed4cc4068b713a9e621d94b7c : SUCCESS in 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/831517a840c24d52bd9f19cf19b2b0a7 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4f97f7be3dc67f550592838c07fa57fd421fb033","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-19 10:59:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9bc64f710cd78805ccc67c4eab6f6ded73f4f6ff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-19 11:22:10.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9693c285cb684b91a3d9c1cf2bb3c95c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4f97ca0586664e69b234181d3d3df42b : FAILURE in 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"83379721501844f97ec5070742e7b041d9db706a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:15:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64d6c1adf67d4bb195d0b41d404c373b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e74a7f428542adb9f17e22a4f2386a : FAILURE in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7648a1d3a8ab45329bef91f80f961bbd : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0e8ebf8bc341e594e7ef0bdba952e3 : FAILURE in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f900d4596b418ca3ca03a9ebc79766 : FAILURE in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9019c3c095d946a3a6816b456d9780d3 : SUCCESS in 8m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b72f7bda7d84dcc816d4607df5aa418 : FAILURE in 5m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/189d56e30efe444480506f07cda61855 : FAILURE in 4m 44s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/594638ca982a4e2dbfec1afada6461bd : SUCCESS in 5m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94fb69a19f2b4ca4826f45bb34ca795a : FAILURE in 5m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2387078fc0b4d418b01728478839ece : SUCCESS in 32m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/078712df17a14123b376f3cd2981f948 : SUCCESS in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/46a78534d1af4443928ff43088d9bb9c : SUCCESS in 1h 06m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0552c70170374887af30b830950fd031 : FAILURE in 48m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/413e7a2c22bc4850874224067b34778a : SUCCESS in 1h 27m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac75d7e0733d4e1a861379f90d0c8b5d : SUCCESS in 1h 04m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9fbf6186e0dc4057b854d7a44172bbbd : SUCCESS in 1h 10m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c90058b61c049ef951e4270a6099a98 : SUCCESS in 2h 15m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0335fe404c77493a9b11a4174d971328 : SUCCESS in 58m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/610fb870e981449589c091e1dee762e0 : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b0fa6b19a234438bdb06a0a6f20f6d9 : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2f2c8a8e03456d4a1e6e6b1758ded21651401f3b","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-19 16:30:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"914044eeac7a4acdbae290db650a938ec427ab55","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-19 17:41:52.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/dab9dfbdd4f34feb84d2ee40c7d582c1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/783b8ac8c8874889b17c90ef1e6d3ad7 : FAILURE in 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"883c47e6b0215e9072294f608abac9b15965d77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:50:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3da230db62c4b908bbd5245f69632a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c474a9573ad414d9b800ef41e9cfc2f : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d57d0fd26b43dd8f5d740a4f285c6a : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b18c7a9a89384119bd55c8cb55df3dda : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577a1910fc154669b8538767cd788b06 : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad52027d8d784f458647d333a56f5bc5 : SUCCESS in 9m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8947017af744dc5a6e003dbab369fc4 : SUCCESS in 11m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b86548c9b9a6495085305c6c890a080c : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d0a03a3a2b74e15b45436ac6a41cb47 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8244b7bfeab4392b8ddc11df2376801 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a939368cba6b44eaa35d105dab9b8b90 : SUCCESS in 42m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f998ca56c19d40fa9c22c87db4fbe4e3 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f67c58c3d483442f9973dd84febbb325 : SUCCESS in 1h 16m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc857128efd648769611b84180465221 : FAILURE in 42m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9537da00c78741af82a9a775ca02ecff : SUCCESS in 1h 24m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/159d9b9b6b04400897103b857eb671ea : SUCCESS in 1h 05m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1326c95490a44cb78aea07024b7198c2 : SUCCESS in 1h 10m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58836927f5c84c5eac28317e2e3b4acf : SUCCESS in 2h 01m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50bb686c8f744681a0ec8f0ed2e2bba9 : SUCCESS in 1h 18m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8973f755fd6e454191e3ad5933415e99 : SUCCESS in 1h 08m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fe7cfb1a96841e79ccb3a3de033b119 : POST_FAILURE in 37m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e1f197dea85c811a720a9eedaffd752f0e46f532","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-19 18:52:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d2e45680d23b9409203dd9b51c31ec232a3650e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-19 19:14:33.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6f85c9208df64dc0a3c4363d30cde7da\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d346c50607d646d7b962dcc7b7afc326 : FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d2a7faed9751ee16dea272182695bf522e1f6b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 20:40:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df735aac5d4d47e2b91d1ef54629abfd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507a8ce131c448e99b37cd8ab2d7f5f1 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45112dc2131d403192bb95e7553f8d66 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2be1f79bfb84adf9d55e44ff84206c0 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3081aa5176401389caa5c7acc28d9b : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a329043c63d46a09213b15a684efe68 : SUCCESS in 7m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/377da732ef114d7eb25f5f55b86ad0bf : SUCCESS in 13m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1fdd52f0fc194af2925f93edfa8a730f : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d1b3956014140459e526fe147f8e65c : SUCCESS in 5m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae156a3eaaf44a25afc5c1c0004c676a : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/993e615007df44b8924211d5880f8d42 : SUCCESS in 48m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9679fe26529f4790bd6127fc081ef526 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f0bb5372ced4d42ac4f7542438c1a1d : FAILURE in 1h 26m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3515a28378114fc88cb570d625875959 : FAILURE in 32m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9e7ab817f0347b9846d90d473e9806d : SUCCESS in 1h 35m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65c96ab225184054835f556ca9e22242 : SUCCESS in 1h 19m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7b52539ef5eb4159ba6e0e766e0827a6 : SUCCESS in 1h 00m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0d401efa5fc45c38dddab7d96340530 : SUCCESS in 1h 46m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/702392989ad0444590d1beefeb00f9f6 : SUCCESS in 1h 02m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abf20dfce83145e49393bca2485e5dc6 : SUCCESS in 44m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a883ef47f2db44fd9e3c6cd84699c7e1 : FAILURE in 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9e1384b0102bb27f468ae77c6a863a3eae6e7301","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-20 09:38:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ece6b52311418e9b28e2031a77efe21ee71a1904","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-20 10:29:04.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ecb8a04f438845568360b06eeff8eaba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7e366edbf5c5416c8c70a16b05290d1b : FAILURE in 19m 29s","accounts_in_message":[],"_revision_number":1},{"id":"82ff5123226ef2c646d5050160536da1c061b4d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:10:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ff22743b82647d4a603b386dc4114f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56615efc26684465926ff2e7b3e1612e : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e0938dc2a644a2bf13170d1d01e545 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3cb9e0489c64b7386455fcfb6d47386 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d5bed80b3b4f7b8ae572a10700dcfa : SUCCESS in 14m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e497f4e1b3ad42788a0fd97c9820d9de : SUCCESS in 8m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f5b42274e524a5a8db1a5923f66383d : SUCCESS in 15m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e51f0ff405de40d1ab782a00b798ebaa : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1387c9ebfe2f4ae594b4c95bd10cf88a : SUCCESS in 6m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5a53277443445b2934ca042779b6baf : SUCCESS in 4m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e976e15d945f4e6bb1af586bf9b5e87f : SUCCESS in 57m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a052088d53b44f19ad5c5c5d66da3277 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6732457cc6ce4ff7a1eb9ddabd37c070 : SUCCESS in 1h 29m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5f79242d8a2e46cca0d74d6e692d5de8 : FAILURE in 44m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6aec23c1b611458d9d3239f9796347e2 : SUCCESS in 1h 35m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe39c880e9024ff09583d984ebb1f413 : SUCCESS in 1h 25m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fe75f9685ab54f45a90d01607651c6f2 : SUCCESS in 1h 41m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd5dccf9d501433684e6ff5bb4fbc387 : SUCCESS in 3h 03m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31448ca7838143ac946554bf1949bb26 : SUCCESS in 1h 06m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a57f5656bd94416ca18288435066bcd2 : SUCCESS in 54m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f33bdf22d2e44d0a565e923a5f697f4 : FAILURE in 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4a2e3cbe9836d0810362d5fd59fb24f39162afdd","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-07-25 10:42:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c199cd0a416c1990106ac82067d9bc5eee196623","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-25 11:05:56.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/45966d950ac74bffa3e6ff8fe741ab39\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/eb677316af4141d290a670b08d2651aa : FAILURE in 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0c48942742bc038618da7cd7761b1560600f7b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 12:55:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd1130830e1c4100a73f715992d763f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2637509f0f2848e59dc42b3b0a9cf2cc : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/323c55c0ef8f4c08a3c7ab3c3fe2135b : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e4e726fc0c7417fade1fbe68d158e3f : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86537216f8dd469a98a9f7e99665a9c2 : SUCCESS in 18m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222653d9d266486ea79108d2af10e275 : SUCCESS in 11m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6bae9ab3794e43a58fc32ca838682554 : SUCCESS in 12m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/274507daf3454394bba488e7be54fbc9 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e797720ed362422fa528083453167d0c : SUCCESS in 7m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c20d6db5f0c441391b98e0222422f1d : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/744f8ad5e53d43eba787695129b0ada2 : SUCCESS in 33m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/346026eb6f914d9483d0fbb3e2b6f4a0 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5acdc6c483314255b15707f2eff252e8 : SUCCESS in 1h 05m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cc6e8bfa9c344336958d00e66ebf8667 : FAILURE in 40m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff8a513468624393aefaa86323215c57 : SUCCESS in 1h 35m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8b0507835cc4ad693df6782cfc44360 : SUCCESS in 1h 20m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db65f0a6d2ea4149a324ada12d3f3701 : SUCCESS in 1h 42m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a415271b7c54860ac2b3b50830ceae7 : SUCCESS in 2h 03m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d66ec3bc868349009838f52b918de1f1 : SUCCESS in 1h 19m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d218e84896984c47b702b3ba1135cdbe : SUCCESS in 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cfac26022bb40069b256d60ab77c29b : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3497c4e9a698f6abc5840d69b4fd531693fee273","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-19 01:34:26.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-885917-1/FC/ : SUCCESS in 1h 47m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-17-885917-1/iSCSI/ : SUCCESS in 1h 16m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"0570ccbc6fee8a35406df36cf416e46e30af8bf0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 11:24:41.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c451111d4ebd6090808c15bc7dde6548b043a41b","revisions":{"c451111d4ebd6090808c15bc7dde6548b043a41b":{"kind":"REWORK","_number":1,"created":"2023-06-12 21:42:39.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/17/885917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/885917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/885917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/885917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/885917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/885917/1"}}},"commit":{"parents":[{"commit":"63d7848a9548180d283a833beb7c5718e0ad0bdb","subject":"[stable-em-only] Add CVE-2023-2088 warning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63d7848a9548180d283a833beb7c5718e0ad0bdb"}]}],"author":{"name":"Alon Zeltser","email":"alonz@infinidat.com","date":"2021-02-04 13:16:14.000000000","tz":120},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2023-06-12 21:42:39.000000000","tz":0},"subject":"Fix Infinidat driver to return all iSCSI portals","message":"Fix Infinidat driver to return all iSCSI portals\n\nInfinidat Cinder driver truncates the list of configured iSCSI\nportals and returns only the first IP for a given network space.\nAnd in case of network path failure we lose access to the data.\n\nTo fix this issue, we need to return all configured and enabled\niSCSI portals for a given configured network space.\n\nCloses-bug: #1981354\nChange-Id: Ie2b7a163ee3a83121c04a21808ef437d740426d5\nCo-authored-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\n(cherry picked from commit a25dcc85180124f3f8db836ec49c0ee5b5741b62)\n(cherry picked from commit bb510b64a5c77c60950ee13ce3a05c39ec667f01)\n(cherry picked from commit 4e2b049555e701e3742af6239c1bb991514b05aa)\n(cherry picked from commit f6b44fc201c3e679d70367fd588304d7a8cbf420)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c451111d4ebd6090808c15bc7dde6548b043a41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c451111d4ebd6090808c15bc7dde6548b043a41b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
