)]}'
{"id":"openstack%2Fcinder~885919","triplet_id":"openstack%2Fcinder~stable%2Fyoga~I609a4f056d19f67aa7b8113bb38b9e9083f24124","project":"openstack/cinder","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"35002":{"account":{"_account_id":35002,"deleted":true},"last_update":"2023-07-26 18:46:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I609a4f056d19f67aa7b8113bb38b9e9083f24124","subject":"Fix NetApp NFS driver to never spawn a native thread again","status":"MERGED","created":"2023-06-13 12:45:23.000000000","updated":"2023-07-26 18:47:31.000000000","submitted":"2023-07-26 18:46:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885919","cherry_pick_of_change":878540,"cherry_pick_of_patch_set":1,"meta_rev_id":"9dcd0081a1e5939fa567f0bf5f0fb8a8fb163f0d","_number":885919,"virtual_id_number":885919,"owner":{"_account_id":35002,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-26 18:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"all":[{"value":1,"date":"2023-07-19 18:22:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":1,"date":"2023-07-21 12:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":2,"date":"2023-07-25 12:53:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-19 15:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":1,"date":"2023-07-25 12:53:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 13:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-13 13:07:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-13 13:51:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-13 13:57:00.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-13 15:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 15:08:24.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-14 16:01:01.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-17 18:12:27.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-20 00:29:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-19 15:09:45.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-07-19 18:22:19.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2023-07-21 12:46:39.000000000","updated_by":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"reviewer":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"state":"REVIEWER"},{"updated":"2023-07-25 12:53:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"c4818eee9db6d1c89b2eef7a9f278641f8fbaccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35002,"deleted":true},"date":"2023-06-13 12:45:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"ba658b15310695992d9b8dc031263003dc41358d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 13:03:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24ad38dddd504140a3638f91c2c5130d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd417ce5695a43f29f8d363a4c65d701 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ef55291e3b745a2b5a6d2ad07563d41 : SUCCESS in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e4886b2daec19d148152f1d558503e9991aef73","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-13 13:07:29.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e37b31671051444e8aa9e2e73406b61d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b3b60af2a7ff4723b7b42b313c9a052e : FAILURE in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a77ac458e14b7f144392608c6ef94f06ff5b8a7d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-13 13:51:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/885919 : FAILURE in 1h 04m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3d6f4c301fe443abd9f008955856160fb6a5a1ac","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 13:57:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/19/885919/1/check/iscsi-driver-alletra-6k/b990b61/b990b61.html : FAILURE in 1h 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f9f81380fc59566da3573776f359c50660df7bf8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 14:08:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/19/885919/1/check/iscsi-driver-nimble/8cced37/8cced37.html : FAILURE in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3c73d5f9887be35c1af2209eb772ea24a4012a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 15:03:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d2044d223f47b297a1b5a55014412a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb53a854758045ce9f9ea503ba64abe9 : SUCCESS in 8m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d02ddb6b8b1148b3adfdc2604bfbd67b : SUCCESS in 15m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4a631a8dcbc4d2cbeed5a71a623af6e : SUCCESS in 13m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64eaab59b7c7473faba3b321832dd0a2 : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/28c03720f57f42a994cf6112b8f19f0b : SUCCESS in 43m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f04785fb9c644769ae39e89441ed7283 : SUCCESS in 1h 38m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e132128f3b24203a3355c475ca3398b : SUCCESS in 55m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65955653d60043a09cc984154b1be1f6 : SUCCESS in 14m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/15338857903a453094ef059546ae59b2 : SUCCESS in 17m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/87b836da97c344dca7878f808bd9fee2 : SUCCESS in 4m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e127a872f78c4bb596b3ae0e670d4245 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77518687b943489e941e1a27bb8b90b9 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3c521f00a6c04aee948e029354f8a3fa : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a4b39e8487b40eabc8c4badcc068213 : SUCCESS in 33m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/249393fa29ca49a88b24d10e477d13dc : FAILURE in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f3e4256cbb145559ad5769fccdc20eb : SUCCESS in 1h 26m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b017e625edd410bb608fa457480eca4 : FAILURE in 1h 15m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d85b39d270a54766b93f74fac122a66d : SUCCESS in 2h 01m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ed30b6d828f474e8a4673eaff2bb211 : SUCCESS in 46m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/31c3b266cc1447fca9e46ac21ff082b6 : SUCCESS in 1h 24m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8b53f6ddd08445d2bb697eaf5c504554 : SUCCESS in 51m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c140be8f0cea4f4f88e94f3f3344d741 : SUCCESS in 1h 15m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2225ca0e807741dab09e6dd5a051e1bb : SUCCESS in 2h 00m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83950be696914f33bebae756fac8d053 : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c9982daf1033cb168bb9534aa600d23d38ee4ccf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-13 15:08:24.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/13_06_2023_15_03_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"e5607839aee30f953f8c22d4f530eb0cdb55d094","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-14 16:01:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-885919-1/FC/ : SUCCESS in 2h 15m 21s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-885919-1/iSCSI/ : SUCCESS in 1h 45m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"89c6eb19adb9b6d75635b13127f8038076da917e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-14 22:51:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-885919-1/FC/ : SUCCESS in 2h 7m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-885919-1/iSCSI/ : SUCCESS in 1h 52m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"7b8db03e748ca8e64344207283d026941786ad2b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-17 18:12:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/885919/1/20230618001412/NEC-V-ISCSI : SUCCESS in 2h 53m 10s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/885919/1/20230618001412/NEC-V-FC : SUCCESS in 2h 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"965e31e8cbf30f46811969068a92b84a364cadeb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-20 00:29:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/885919/1/20230620014029/NEC-FC : SUCCESS in 3h 41m 44s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/885919/1/20230620014029/NEC-ISCSI : SUCCESS in 7h 48m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5c130a68916885cba467f97f0c5d10635ec3de7f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-07-19 15:09:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbac8a456e54b25247440342a8ed582f51ad0a3b","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-19 18:22:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1482de3546ada0025ee354fd18d3a955ad888e75","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2023-07-21 12:46:39.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f51340b7a4da2472cc4a78440060d6deb104a6de","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-25 12:53:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b69a06d351ce09489c4db573bc7ba79d06ed584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 12:53:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ded46da70dabf31f4020c7aea5310d615b355ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 14:45:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef80f61ad29a413992fb9f9a4b32a7b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f193a1e2ca31449c95ee6aa078a62036 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed48ce37ef746a4aae55ebbcd0f83a2 : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3abf5101fb4841c7bf92c728bac319bb : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/506eb5e5d4944411b90eec92aa2d480c : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/2640494a5b5e47b197dcbd33a29ccf1c : SUCCESS in 1h 08m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebf45f44b7a64901b7f263f89ee7c533 : SUCCESS in 1h 16m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b73b175f1e0e4d41a68644beba01d69d : SUCCESS in 1h 02m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71f5ec7b7e7346e496358b068e77d2c1 : SUCCESS in 7m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c517027a96ef4a44bf2a63f441f534b5 : FAILURE in 51m 41s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c752b50ad9246cdb63b7660d0e7907d : SUCCESS in 1h 09m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78a8f731eb774097a3e7f384e8f71466 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0a7e4daa6ffdcd1d203f2e555e5d3591f6bffa49","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-25 15:36:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a462643ed3cfc26423a3b921a42530c1d2bdcd2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f747146bca24cce8e21983414422363\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec514e3d78074505b78b2f4678e9d93e : FAILURE in 27m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa4138cc55f64db297bb75f65c31cd6e : FAILURE in 32m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c30bb72d709df1c3b8f52f8f49ef325d757a5ef","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-25 16:53:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9554abfe7ea94803b6a34d0c0c2bae8c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9acead0e34e24a679d70bf9b8f1fc60b : FAILURE in 20m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9b40c991e2b55ebfd43fa48f27f5df8ae683149c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 19:11:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70ca7734ea54ecba9816943ead82017\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13dedeed36e4fe896dc575cb428aaa7 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/075517a26b564f59970b47cea16b3464 : SUCCESS in 12m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f771f3587d4af0a9cba325a827922a : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491ee3829d674a92826cdb8354503599 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ccdfa2a636a844979c5c9ffb9a3c11c7 : SUCCESS in 54m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c04fc624b7345b1875994bd5d1982ab : SUCCESS in 1h 25m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d1254c77da9482dab72e3f82b818a07 : SUCCESS in 54m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc2c77784ae54480bb77cd008017d262 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59af3d7d04f14d7a866af1f7e9ce03f4 : SUCCESS in 20m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd859e66b6484c6390a2a83e28e89d2f : FAILURE in 5m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b55989a43d7042edb3668b908fdb0190 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/309d438e1afc4a8d814e73ad2a473385 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eedb44733f304a1785249f2dfde1e767 : SUCCESS in 9m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ddca3f2b39c48d6bccc256897830ff2 : SUCCESS in 41m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d2ac40431814ed3ba73bf5feca31cd9 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7238189c38d243bd8fbee9f7f0837c4b : SUCCESS in 1h 10m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c2ace625c15a42c9a395e61e311fd2ca : SUCCESS in 1h 43m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f6d6b635691f4ce685b83f747cb21916 : SUCCESS in 2h 16m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f48e5ffb1c0941ff899ef01bacdd6a2b : SUCCESS in 1h 20m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbb42a0ca78848219761796ffe24b6b7 : SUCCESS in 1h 26m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7895c57c9a2744d798db8b4e28f896fa : SUCCESS in 36m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/00ecccc33fe14af482342445dafbfcba : SUCCESS in 59m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4814ab0fcc954054802842eddd0da935 : SUCCESS in 1h 10m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/944c608ac3e34131a51c9dd0d6bbaa18 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":1},{"id":"785c498736616601e1cff7d434147fab2c81b386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 19:11:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eecbd7044e208918131693008a5f5980c6b7e3b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 21:57:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5973f245848d453d887a5c4d4586fcfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d722edbf59941eb94b901ea9d81d9bd : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/686dde8164084701a272352adf3c3507 : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1aa69d643c4f6ab6952218d35435c1 : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2434ab386a4975914ccf4286b7c0a3 : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/66f2d3679f744b6984e1ed8b57020b20 : FAILURE in 2h 39m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6443da38bd6a4d9aafdfa7ed8c2494cf : SUCCESS in 1h 24m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/216f6ca9693a41c7910e5cd4155c66b5 : SUCCESS in 1h 01m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45f130f8ba814c4cb59173a672880d81 : SUCCESS in 8m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a726252b63a24a2fb3bf06646f1a37fd : SUCCESS in 55m 44s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44d08216870348668170fb8897a260ab : SUCCESS in 58m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/082bbae3d0fc4171bf26d1c4bb3119a2 : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":1},{"id":"286d452ba78d956046b40045aba49cad0b7ef932","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-26 01:18:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bab8c46dbf6ab547d77a126940e558c166b2b60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 01:31:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/187e64b6519242fcaff11bca63ba3432\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dd7fe292ae34d238ce64389b860c276 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/571ba036574f431f9aa2b6c54b87137b : SUCCESS in 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58eb6a8c48282e69b30651a30442274a5a48f4ec","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-26 01:41:06.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1d93668ec7ef426da782d003ddcbdff0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/49319a2a982c4e2da4124e41e1fe0fa7 : FAILURE in 20m 06s","accounts_in_message":[],"_revision_number":1},{"id":"12eacbd78243b8ec91f77d35874d3a6ade64afc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 03:25:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/000dda7684eb4ddaad4dd3f6c73f9006\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43de824834d74af59fcb9499b127a8c0 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/263a07faa51644fc93ea368ae65376a4 : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26708839ce834c9bb49241ca1ca790d4 : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cab620f2bba4fbda445bb60ea74e817 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/47e2656addbe4d2484136f978a6d42c2 : SUCCESS in 1h 01m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d6284fbf6f04ad8af01df40ac28c8b3 : FAILURE in 1h 09m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ab9373f43b481d8295d7074db4b9ca : SUCCESS in 57m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09441d98ef464539858deac10dab62df : SUCCESS in 6m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/66ebd0c7c80048d8ad13103f6f526770 : SUCCESS in 11m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1504d4915661445688554468d2952f63 : FAILURE in 5m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2bd82aa3a08c474ea0557e7bd8c4e18e : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3f93212123740d8923c2117cfdf4fc2 : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e566dfbf39b246828e45472ecb78fcd8 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/150200f0834245eebee57ed1e7bc7086 : SUCCESS in 40m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6e83d1ad607843b9a22d0bf63381fcfc : FAILURE in 5m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ba8b69234094e6aa20b98038684ab5c : SUCCESS in 1h 29m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/725ec650a47a404da2a9ba2dab4b06c8 : FAILURE in 1h 31m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5769d3ee138940418408103d084e4131 : SUCCESS in 2h 01m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/038ca949f7514c768e9a94e34b9f5baa : SUCCESS in 1h 10m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cec05c40491845298a062f7d619d3266 : SUCCESS in 1h 44m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f4f22d8c28844887af478b945713fc42 : SUCCESS in 44m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/70c27b6374fb4c40b4c1803607f9f1db : SUCCESS in 1h 18m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/976ef422dc7342b289b169a8299fff3d : SUCCESS in 1h 24m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/562da6fc695f4ce18f8b73a208b9094d : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b747708471b4bc83f719e139ed219363b861b7fd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-26 14:04:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3095197af5d95f77f3188f51fdacae880d8d3655","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-26 14:27:20.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/03413b53c86f47baa1590bd888ed10a5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c25101895e95420cbe64b71755af0b33 : FAILURE in 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"415f7a0111b770bbadc02cbc21a8f414231b8279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 14:39:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7224ce9649c74262b459b403e5150115\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17351b80908245eb83a58762fd99bf0a : FAILURE in 32m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72ec3163bb374beb8fa0cbf815ee8926 : FAILURE in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"735bc1fe27c1b81118d1db0547203ec5fe20c05b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 16:40:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b853ed8f575143e185fa37501868cd35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb5b375fdde4f9b80cda7cf6d978feb : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3cc7bc2c2647009f1267bd73a9e1f6 : SUCCESS in 14m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9599a9a649bf4711b7003a738dac3c5d : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740784c2ad0f46abad6aa0e979322985 : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/6600b3e1633d4cde9d9830cc2476f5fd : SUCCESS in 1h 06m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c81ebdcb30e4a15b09ccd308e9186a5 : SUCCESS in 56m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/355a0feefd124732ac8234352955c1ba : SUCCESS in 58m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/495fd8c873d44b4c8fb63714e4eeea6f : SUCCESS in 10m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1c5128ae56354ce7a0c4968c574da7a9 : SUCCESS in 15m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b622e7daf9ad466ebf365cd03e90e328 : FAILURE in 5m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e2d710160d944ed3a5c6951f98b96d37 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4c0223cd8084abf89b0ef51cb66a617 : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8056dfc1fce94103bfa80191d965e014 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b18ba6a07e7041cbbbfbb2ea5d8b10b0 : SUCCESS in 31m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd89ed0f97294dd38c7872df39913144 : FAILURE in 7m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/963eab86c4614ec4a709866574534ab2 : SUCCESS in 1h 03m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7c2f4e5565af40b69b55e02a5807e431 : FAILURE in 1h 15m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/12f3afcf71ae4278ac7fd3873ce3bb05 : SUCCESS in 2h 04m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d868064222c54090b04416c521646999 : SUCCESS in 1h 19m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6e9a930a471646478ecd53ffaad57834 : SUCCESS in 1h 26m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4cfa7849c0d6431d851b90f59925bfe8 : SUCCESS in 1h 04m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f75e7e1fb0f041c8812c54f5c97ee573 : SUCCESS in 1h 21m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fbd9a8854584e4eb65a7ec021d229cf : SUCCESS in 1h 37m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c58d10db2ef4ad3910537230be7488e : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a233d89efb8b1e9e73d4e07aedbf2b7d0ba30d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 16:40:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42dc114495a762a1b239b48bfc3f53013028dea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 18:46:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec0676d72b2843f38bde8aaa3c9132ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3bad49c546540c0b89e73eb38a41cfc : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fa585f99bae4c5b9882361008fb43bd : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1801b38ea02f44a689a3f7f2612cae88 : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970d3a83ebb14c17b11f0fde76b3ce3c : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f657718d92ab45c0b0a08b4bcb428518 : SUCCESS in 1h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05ba240920ee4596a80c07018760376f : SUCCESS in 1h 09m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ce4a67c690e4babb331fd665f6c5830 : SUCCESS in 1h 01m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4893ad7babdc4a53828e75c7f22f03e4 : SUCCESS in 9m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3d0a6a5e4ee4cecaa3e5653db933a78 : SUCCESS in 1h 06m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eafee31d3aef44be970a382668d5c158 : SUCCESS in 1h 18m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f4dcda26ad446978df36e0e93ea1e45 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ff7739bc983b44fd3329bb12b3eaff2af70ef0a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 18:46:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9dcd0081a1e5939fa567f0bf5f0fb8a8fb163f0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 18:47:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45762798b611465f8336e556e9fd4b69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee3ea1098f844ae9261d057dc68a7a0 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6db90b9063444b8292568470cb0c52c9 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f80d022ca2c9892191e82b76cb3feba96a182a8","revisions":{"0f80d022ca2c9892191e82b76cb3feba96a182a8":{"kind":"REWORK","_number":1,"created":"2023-06-13 12:45:23.000000000","uploader":{"_account_id":35002,"deleted":true},"ref":"refs/changes/19/885919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/885919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/885919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/885919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/885919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/885919/1"}}},"commit":{"parents":[{"commit":"92e8a3df46e77878d79575b67dc145004c4fb06a","subject":"Merge \"RBD: Fix total_capacity\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92e8a3df46e77878d79575b67dc145004c4fb06a"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-08-16 03:48:43.000000000","tz":0},"committer":{"name":"Luiz Felipe Looze dos Santos","email":"luizfelipe.looze@gmail.com","date":"2023-06-13 12:45:23.000000000","tz":0},"subject":"Fix NetApp NFS driver to never spawn a native thread again","message":"Fix NetApp NFS driver to never spawn a native thread again\n\nThe NetApp NFS driver is spawning a Python native thread to\nschedule image cache cleanups, which can lead to unpredictable\nbehavior such as thread starvation due to its hunger to preempt\nother Cinder\u0027s fellow threads.\n\nThis patch fixes it by reimplementing the NFS image cache cleanup\ntask as a green thread with a customized periodic interval.\n\nThe patch adds the configuration\noption ´netapp_nfs_image_cache_cleanup_interval´ to allow\noperators to set an interval in seconds between calls to the image\ncleanup task.\n\nCloses-Bug: #2008017\nChange-Id: I609a4f056d19f67aa7b8113bb38b9e9083f24124\n(cherry picked from commit a869c8e222a68af731eb72e9f8ec5f0755db0326)\n(cherry picked from commit 3dde6b7d99956b4643796025905d1150123f8d3f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f80d022ca2c9892191e82b76cb3feba96a182a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f80d022ca2c9892191e82b76cb3feba96a182a8"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
