)]}'
{"id":"openstack%2Fcinder~886204","triplet_id":"openstack%2Fcinder~master~I33991337441647358bebc96de8d81992fd247de9","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"33986":{"account":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"last_update":"2026-01-02 16:31:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I33991337441647358bebc96de8d81992fd247de9","subject":"DNM - Test the Netapp CI for cinder wallaby","status":"ABANDONED","created":"2023-06-15 19:13:03.000000000","updated":"2026-01-02 16:31:00.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"509feb16cd5cd8ace41c9f3a7099ec66909b9f98","_number":886204,"virtual_id_number":886204,"owner":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-15 22:45:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 19:13:54.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 19:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 20:27:51.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-15 20:35:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 21:34:41.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-15 22:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 23:16:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 03:28:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-19 04:18:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 16:10:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"89ce08b9902ca0448fe7e7c0109d6d27c7a2015e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2023-06-15 19:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4350f1fb7de329ced0da52559d19721c65091ca6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 19:13:54.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98d451357d531c17041eacc130477b07fec80b5d","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2023-06-15 19:14:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"443ce24345d05fb3a09ee98f51dd97bd4c1a9441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 19:24:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d13eefebb47c4b0da81fbd6433ed0ca8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a80cc4493aec48fa8bb33399e2965cc0 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b41e99efd643456c88e2f0524b9189e9 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30ab137817d64dfc9d00267df5a3d960 : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a9630f6df0f91f91d7456d0616b437225ba3ad9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 20:27:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/04/886204/1/check/iscsi-driver-alletra-6k/59eade2/59eade2.html : FAILURE in 1h 08m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0d437f3f5ab5642c0eb7dadbdaa020eef38f3f51","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-15 20:35:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886204 : FAILURE in 1h 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3af5ce97ddfb6ddd66444a04267324fc79341e27","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-15 21:34:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/f71b01d29814404a92a71411dfecca23\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cdefb413a406447fa4ff792b1cabef57 : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fd7e55f130a4a13013bed4f11724812a504975be","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 21:56:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/04/886204/1/check/iscsi-driver-nimble/b727efe/b727efe.html : FAILURE in 1h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1e24d6f080dc05a5d52c3f16082307182ddd5043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 22:45:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fb9c37847424ede8bed0eab7ea2e89f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e941aed45fdd42ab9f5f0c55e3ba32e6 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcd3b1ee443a487e8f44d8b8a4c10596 : SUCCESS in 12m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4ff2ccd395d408193c853f0c5fc80e9 : TIMED_OUT in 41m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc9b1f23875d4f0292b321733320fe94 : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d758344b5fd94727a873d045644c242e : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/c991e14a374e403994b62f2c1e8a13ce : SUCCESS in 51m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99ce3e0d25c54bdfae99429051f38b26 : TIMED_OUT in 2h 04m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97a923def0a9456bbe229cfbdab9657a : SUCCESS in 1h 04m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54d194ffff944e11b7238082d724d31d : SUCCESS in 14m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f86a7bbbabe459a8c88c780b3678433 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7d61a62b6bd41de938cb39207dfdc5b : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/836ad3101ccb47508d190bbbf2fa0b29 : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37e0d4d559e64d24bbaf0ec2ef4b86f6 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1635946a239048b796cc888a9c196fd8 : SUCCESS in 42m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9d70dcf115af4a009ca6251ac9bacddf : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35020ff1b7c7488ead86b712fcf4626a : SUCCESS in 2h 13m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18155c9db07e4771811bbd89eba65430 : FAILURE in 2h 25m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8cb4c25f732d4bf69895f7f6d91703cd : FAILURE in 2h 36m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6ea01e17456b403b92a1d464a6c29aba : SUCCESS in 2h 29m 40s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/273e1dfd5a3f42ed9845b393cf512d4a : SUCCESS in 1h 16m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/35aa2b1252304b04915cae0ea5d4d13a : SUCCESS in 1h 27m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0492c9aa87ba4d529bedd4f418001973 : FAILURE in 1h 54m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ff4f02811ed4d6f85f76f7f6bd7585d : SUCCESS in 1h 26m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e0124350a7a040f4beea4491fd4e5dc3 : TIMED_OUT in 3h 31m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3419775a12c04092948a99d1185a69b3 : SUCCESS in 2h 24m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7e26ea810444795bce8405296c83e99 : SUCCESS in 53m 42s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/9751e71734194d57b3f43243490612b8 : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"daac7fbb90ba94fa56b50f0a0c956627f02d7254","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-15 23:16:31.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-04-886204-1/ns5_nfs : SUCCESS in 3h 54m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-04-886204-1/ns5_iscsi : SUCCESS in 4h 2m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"04cb45ce954f7e237178656aa5c9a4a60dc8421d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 03:28:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/04/886204/1/check/DellEMC_Unity_FC/da0019b/DellEMC_Unity_FC/277 : SUCCESS in 1h 43m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/04/886204/1/check/DellEMC_Unity_iSCSI/4f7956a/DellEMC_Unity_iSCSI/277 : SUCCESS in 1h 50m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"92fe2c32bccbeaeb0db04377468076b9b18617a6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-19 04:18:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/886204/1/20230619101811/NEC-V-FC : SUCCESS in 2h 48m 33s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/886204/1/20230619101811/NEC-V-ISCSI : SUCCESS in 2h 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"69ce360f65a4e7042637e938f1fb69eaad645f85","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-21 16:10:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/04/886204/1/20230621174242/NEC-ISCSI : SUCCESS in 3h 35m 39s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/04/886204/1/20230621174242/NEC-FC : SUCCESS in 7h 27m 41s","accounts_in_message":[],"_revision_number":1},{"id":"509feb16cd5cd8ace41c9f3a7099ec66909b9f98","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-02 16:31:00.000000000","message":"Abandoned\n\nCleaning up.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98f5f64543ecce99ee60b42b701e784187d1c425","revisions":{"98f5f64543ecce99ee60b42b701e784187d1c425":{"kind":"REWORK","_number":1,"created":"2023-06-15 19:13:03.000000000","uploader":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"ref":"refs/changes/04/886204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/04/886204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/886204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/886204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/04/886204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/04/886204/1"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Lucas Oliveira","email":"lucasmoliveira059@gmail.com","date":"2023-06-15 19:09:57.000000000","tz":-180},"committer":{"name":"Lucas Oliveira","email":"lucasmoliveira059@gmail.com","date":"2023-06-15 19:09:57.000000000","tz":-180},"subject":"DNM - Test the Netapp CI for cinder wallaby","message":"DNM - Test the Netapp CI for cinder wallaby\n\nChange-Id: I33991337441647358bebc96de8d81992fd247de9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98f5f64543ecce99ee60b42b701e784187d1c425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98f5f64543ecce99ee60b42b701e784187d1c425"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
