)]}'
{"id":"openstack%2Fcinder~886219","triplet_id":"openstack%2Fcinder~master~I83d8c7b1ef93851e2d4dc4b3da9886c90ee87f54","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-06-22 18:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-25 15:39:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I83d8c7b1ef93851e2d4dc4b3da9886c90ee87f54","subject":"Remove all occurrences of six.text_type","status":"ABANDONED","created":"2023-06-15 21:39:00.000000000","updated":"2024-06-25 15:39:41.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"fae4dfa41adc1df73549d1dddf53ceb7627e948a","_number":886219,"virtual_id_number":886219,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 19:02:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"all":[{"value":2,"date":"2023-06-22 18:15:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-24 05:56:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 21:43:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 21:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 22:54:23.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-15 22:56:05.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 23:32:19.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 00:18:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-16 00:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 00:43:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-16 01:26:43.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 02:13:01.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-16 03:08:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-16 08:07:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-16 11:33:30.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-16 13:31:57.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-16 18:39:36.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-19 22:50:58.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 20:22:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-06-22 07:46:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 18:15:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-06-22 19:27:51.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-24 00:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-24 05:56:06.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"31187a0a6c774f62102f6005efe6fbea4bb8e549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-15 21:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cba3190d5e66c3c6583c2908d8c032d9c0835ad6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 21:43:12.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"252bb968e2294a729c6360ea999bdf1d64541048","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:51:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f56fe56350b43a6a3b5b623122d43aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff6b48869c5e4c9396c8b76e42fcec61 : NODE_FAILURE Node request 300-0021468906 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc99c14e520e40c2907dd97a2504dc16 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2038c25b46dc402c9fe9f5a3f91f8717 : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"401a48e36532aac75f2e33f0de65e008dbb56d41","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-15 22:33:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"208daf8220366952d4bf472a7dad4bb0fdfa6b00","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 22:54:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/19/886219/1/check/iscsi-driver-alletra-6k/c0d824b/c0d824b.html : FAILURE in 1h 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"30dccfdb222d92928df5a41a5730087bc452bbb1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-15 22:56:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886219 : FAILURE in 1h 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b778c9651fa29d6867798369eda86af71ab28873","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-15 23:32:19.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-19-886219-1/ns5_nfs : FAILURE in 1h 30m 13s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-19-886219-1/ns5_iscsi : FAILURE in 1h 30m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3224d65e618db25c8ec8e91f89f3366e5f57ee42","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 00:18:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/194bb3e8c9c349718dcef94ee451bf67\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e9331ff3b37a47d4935622de664f8823 : SUCCESS in 1h 14m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7578a9c1d889483420e42e35954d296d9806c67e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 00:35:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752e1278eb8e469e8d4abc1df6fd2730\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9483aa3e054cf69e2e6ac625156ef7 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50534a787d5f4f5bb59bfce5c082afc7 : SUCCESS in 12m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/760d8c1f766c4ab69ff62fdaf5b2d456 : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77b6ab0d6fc3470693522dd2f7ba98d4 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368d982214e34ace9a48fe398f3c5482 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/13896f03454746faaba084bc7eebb2c8 : SUCCESS in 58m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6409a073c6e54017a07204f6bd7c5981 : SUCCESS in 1h 26m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/460cd9bf8f1241da817d2789c310619b : SUCCESS in 1h 07m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/640613b9205346b8b26ccb047661c988 : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/16dfbad3d0de48b0aab924923cf539ff : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c10457cdd744499b9b564c5f4fde48df : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9be46884836488f81b14bbadb72e577 : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/446da29305234887b779ba9469d969db : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b8e23cd9cbf4c53833fd757c2252910 : SUCCESS in 55m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2729a5b4ec7346df8688daf615db0d7f : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22b2bf7b97b74d3e9f681f0bc7ac7804 : SUCCESS in 2h 04m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a1d3f60311a044ce84080f532f1d4e70 : FAILURE in 2h 14m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e749869f0a74ccbac08d8c1fabc3ca1 : SUCCESS in 1h 48m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5b1f9eadb60f46b8891d7683cc1c252e : SUCCESS in 2h 12m 34s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e01dee9ce07f4265a8f62b143564d59e : SUCCESS in 1h 17m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af9e89955ae744b99ec5cd67f653c552 : SUCCESS in 1h 45m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2276b3d5310040ba8b9f06055e235824 : SUCCESS in 47m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78e6506a549144739ad90f0e3be389e5 : SUCCESS in 1h 21m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a09257f2e6d74750a61bcb2ef0ed67b0 : POST_FAILURE in 1h 33m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b651f4bf7c7403a9268c6d588f3ac7b : SUCCESS in 2h 30m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc358cb6ff91427c8e300eb218c778a6 : SUCCESS in 1h 01m 17s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/34df1736fcb7423281dd91700b615332 : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cdad8883b7b0c6446dcba9927d3eebb6e2905bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-16 00:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_06_2023_00_36_30\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"770d19d6cfbedaf45029346527e230896e4b9e1a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-16 01:26:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_PowerMAX_FC/fa817ea/DellEMC_PowerMAX_FC/129 : FAILURE in 29m 06s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_PowerMAX_iSCSI/5df2b6b/DellEMC_PowerMAX_iSCSI/123 : FAILURE in 27m 40s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e61a9f20cf449cbe98242cf92053e92c9234d3ec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 01:59:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/19/886219/1/check/iscsi-driver-nimble/b8697ec/b8697ec.html : FAILURE in 1h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fae434ad23dc08cbe2553af7afb22cc198ede6d7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-16 02:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/upstream-check/cinder-tempest-plugin-ontap-fcp/a3b8477/ : RETRY_LIMIT in 8m 56s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/f608bb3/ : SUCCESS in 42m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/upstream-check/cinder-tempest-plugin-ontap-nfs/dae9f88/ : SUCCESS in 45m 29s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8bc8d4b/ : SUCCESS in 48m 41s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/5fda3e3/ : SUCCESS in 46m 40s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/c4b302d/ : SUCCESS in 44m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"003c040db144bdb0d7a8077ba37a228e918022bf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-16 03:08:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_PowerStore_iSCSI/aa64589/DellEMC_PowerStore_iSCSI/105 : SUCCESS in 53m 35s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_PowerStore_FC/913cfff/DellEMC_PowerStore_FC/110 : SUCCESS in 1h 21m 24s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_PowerStore_NFS/90eeb76/DellEMC_PowerStore_NFS/104 : SUCCESS in 1h 00m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_PowerStore_NVMeTCP/5e9604c/DellEMC_PowerStore_NVMeTCP/105 : SUCCESS in 1h 10m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"a60cd79ba80496c89bd539dc537244191f56ac45","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-16 03:26:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/experimental-upstream-cinder-wallaby/cinder-tempest-plugin-ontap-fcp/f1c6708/ : RETRY_LIMIT in 8m 39s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/experimental-upstream-cinder-wallaby/cinder-tempest-plugin-ontap-iscsi/3008dd9/ : SUCCESS in 47m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/886219/1/experimental-upstream-cinder-wallaby/cinder-tempest-plugin-ontap-nfs/5e47f77/ : SUCCESS in 46m 20s\n\nComment \"run experimental wallaby\" to rerun these tests.","accounts_in_message":[],"_revision_number":1},{"id":"872ab7b255d8895443c93c4534fc6fc6cc032833","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 08:07:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_Unity_FC/8d90ff8/DellEMC_Unity_FC/280 : SUCCESS in 1h 44m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/19/886219/1/check/DellEMC_Unity_iSCSI/2e4265e/DellEMC_Unity_iSCSI/280 : FAILURE in 30m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0b96b3f6ed158f84c0ea1208820ea7e05d632738","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-16 11:33:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886219-1/HPMSA-iSCSI/ : SUCCESS in 1:48:27\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886219-1/HPMSA-FC/ : SUCCESS in 1:27:03","accounts_in_message":[],"_revision_number":1},{"id":"af0771aa151ff3d87f0d48683b1fb1d1a533da9b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-16 13:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886219-1/PVME-iSCSI/ : SUCCESS in 2:02:27\n- PVME-FC https://os-logs.seagate.com/dell-ci/886219-1/PVME-FC/ : SUCCESS in 1:37:14","accounts_in_message":[],"_revision_number":1},{"id":"340bf3fc2a8747df8972b7c5e4b11a0f98aa3425","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 18:39:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-886219-1/FC/ : SUCCESS in 2h 57m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-886219-1/iSCSI/ : SUCCESS in 1h 50m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"3bffb3ebca20c4981661b830393a3d4b35f83a61","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-19 22:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/1/20230620015146/NEC-V-ISCSI : SUCCESS in 3h 00m 26s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/1/20230620015146/NEC-V-FC : SUCCESS in 5h 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b5a48af0794c38338380035a9a390dbaf178546f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-21 20:22:24.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dc4fce9099ffa458eca76f2eaa4c390cb0e1a3ad","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-22 07:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/1/20230622130251/NEC-FC : SUCCESS in 3h 41m 09s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/1/20230622130251/NEC-ISCSI : SUCCESS in 3h 43m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d23dacead1ac50d7431c8a7a1abbf19a61a7dbbd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 15:17:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58e6642e73864ec9545e27e34209ed18acfd420c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 15:17:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"46992b6651df5d9ad619e81dec3e558cb0e1f57f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-22 15:24:03.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc82f143e38b56475c07522401153736d6acf2da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:32:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed5a4788febf44e5baf92bc22e610f13\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/893e8b5cbed14d41b01161f4a56c5fdb : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/130ddba95c7743eb83d8c5f350a592a6 : NODE_FAILURE Node request 300-0021519034 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db95eb2a2b73462885b3cbb5be908226 : NODE_FAILURE Node request 300-0021519035 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a389611a4fd810e7d5562673587d885918afe041","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 16:35:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/19/886219/2/check/iscsi-driver-alletra-6k/072dd72/072dd72.html : FAILURE in 1h 13m 52s","accounts_in_message":[],"_revision_number":2},{"id":"66aefc73b5ff51df237cd998b167f587e72db5da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-22 16:39:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886219 : FAILURE in 1h 17m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1d5b6842c9816defbf5ea6113102ccd4083b0a2b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-22 17:09:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/8d06f9670eb340cdb77fc75cfb7e35c9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/721a9aeacd314c91b795ae74a63561e6 : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ce80efa2502c88ea6b37e8a48e73a59c40812013","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 17:26:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/19/886219/2/check/iscsi-driver-nimble/bbebed3/bbebed3.html : FAILURE in 1h 03m 00s","accounts_in_message":[],"_revision_number":2},{"id":"beb114fd4457a4356f5a208ea32171dac3c220d5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-22 18:00:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-886219-2/FC/ : SUCCESS in 2h 27m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-886219-2/iSCSI/ : SUCCESS in 1h 50m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"d1c212622ad74d83d052a9c3a243abeb664766a8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-22 18:15:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a64bd4ac4716ef4b807f7c6cde83f553c6967d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:02:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b46e6486f9545be8f2ba37c8dbe2657\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391426417e6841f9adbc2893fbfc2380 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38c3375296f741d39f9c342e9d4d5c98 : SUCCESS in 12m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e03ab0ef27c94348aa75f4cf0a0ba66f : SUCCESS in 13m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8c25894c5424641be467c1c5b8532c0 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d035047c4f4855bfac23f2128eaa7b : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/eeef40dc67e0497da3620c061e5c62e2 : SUCCESS in 1h 27m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da69ea66e9114de6986eccc3ab6bf38f : SUCCESS in 1h 36m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30c368533d9842878f21baa78aacd199 : SUCCESS in 1h 38m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/773e1e6861e343a0a23feb716d8a9fb3 : SUCCESS in 15m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/861c447afff7434bbd91b0038212df1a : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2a47718efdb4b1cab16dda34e33e1a2 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fcd93959af6645b68d1f1ac85b1c75a2 : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a72cc42fc11e48068e3498c7ee01a051 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51ae77d8589b481a82ceb6586089bdbc : POST_FAILURE in 56m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e09e243602984954a1b508ecdfe2f707 : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49fa7b0e84e4492299f9de4856c3e10e : SUCCESS in 2h 23m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/32f5f500115940128fb28b740b287efd : FAILURE in 2h 41m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/58f98832742f44ef946720f16d39b132 : FAILURE in 2h 30m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/66492cd1e26c4b8785ce09911ccea6b5 : FAILURE in 2h 26m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5cbe6764b864c9895bea21f696ada18 : SUCCESS in 1h 12m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a24c8f66c26545f1b9f1fa61f2cedf64 : FAILURE in 2h 08m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f086097a3fb40c7820148f7d034e8ba : FAILURE in 1h 02m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78831137a8aa4cf7bb1fe33aa65bef93 : SUCCESS in 1h 33m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0737db146a0a45b18cba7995cbea8b6b : SUCCESS in 1h 41m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e0a9dee769e47c4b6fe03b105ba1996 : SUCCESS in 2h 49m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8777c4ed982842ab80f8adc178b350d9 : SUCCESS in 1h 25m 30s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/1304ccd782a24f8385ec7df7e57aad02 : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8d91b46f8b30c34f73b4d87397efd98daf40099","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-22 19:27:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/19/886219/2/check/DellEMC_PowerFlex/ae6d236/DellEMC_PowerFlex/183 : FAILURE in 1h 45m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"67e8d668239bfb63e5a920c5cc1c316eb4377017","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 04:53:13.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-19-886219-2/ns5_nfs : FAILURE in 12h 50m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-19-886219-2/ns5_iscsi : FAILURE in 13h 4m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"eeac29e085e9fc1cdbb5cfe03b330e6532dd03f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-23 11:30:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae0b0dbbaf001a8989d4cdc2d3e910339737de41","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 11:31:10.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca3bedd77e004a65c5e43ab00078692730cb263a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:54:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f40e9a060da9451f94884d0734d8843e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37ddf646af7e4377a89949562c561144 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eed9c273510e4958a4f3881f14bb869e : SUCCESS in 22m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fef49be92f4f4263ba2a832b07ca9ab8 : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad726db82db7df6b36e773f076ad5c5a74d81108","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 12:46:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/ae0e0d98052049b0b3e3474b48f6b393\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8e79b474d8264dde8f2c3a157dba7db9 : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ffaca0515424dc35bee4999d617417295ac40f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 14:45:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5b55e840bc64493b523bed45a0454b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec78da5e2da64b1aa2230516fd0a7c15 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03bc1ea7c89b44ecb7d1e7093aa5ac2f : SUCCESS in 13m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f8dc6113b0a45298dd25dee9d263495 : SUCCESS in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27a49d03512d420c8dc10d808652c101 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f9a76d3e8444069082963df07f29b4 : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/f50732701a2640a6baf960ee10ebb53e : SUCCESS in 1h 05m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68e67ba6e0e242e59fac6edb1ae34a14 : SUCCESS in 1h 18m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe81bfcaac0c4bac828e75c2b6693713 : SUCCESS in 1h 04m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1cfa5c2ca122496fb3624ddf401522d6 : SUCCESS in 14m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04f74f5bae0e478aa5b6a2e14e340b81 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/910375ef293c4b09809f7bdb244170b9 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07524c9c77b44f0d8a974c8256a17d71 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6969af22a529407680a53b13b75a38aa : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8fce57ff5dba4c2c8c502a6f42599486 : SUCCESS in 57m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0a381616d63d4ba7b435b449062d22b9 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ea8d6ab48f7492bb128a63d123fb755 : FAILURE in 2h 31m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80a2bf36b7394b47932ba57a8feaefd6 : SUCCESS in 2h 27m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e85f3996b15b45348182640f23b24a8c : SUCCESS in 2h 05m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5fdaed66589d49a6a96a380c19ec8636 : SUCCESS in 1h 47m 29s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3efde9c71c8f4ba38a438f913b59827e : SUCCESS in 1h 13m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/292ed7076ee84961a42306260206d86d : SUCCESS in 1h 48m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8716785af8374513b63c1846facc079b : FAILURE in 1h 09m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab45f3e6731349f99ca8d22b86a00f81 : SUCCESS in 1h 28m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43050e217b7c4d2893d05454894ee68f : SUCCESS in 2h 13m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/adfaed81f49741b1abe8b959a753c957 : TIMED_OUT in 3h 03m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1c41d290806493093c15290a85be0b5 : SUCCESS in 1h 15m 28s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/a9e3b9b8f3e34dd5860ae13fa6e4f2b3 : SUCCESS in 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"03c0e2782f27ea71f4100ac0a0cfbb9c34c85380","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-24 00:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/19/886219/2/fc : SUCCESS in 0h 28m 44s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/19/886219/2/iscsi : SUCCESS in 0h 44m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3bc6e8d1778facc887c4179c4b0cbc35e0ed1b40","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-24 05:56:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6d02fc5bd3b34cbeba920caa1975646ec122d820","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-27 06:52:30.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/2/20230627073146/NEC-ISCSI : SUCCESS in 8h 14m 07s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/2/20230627073146/NEC-FC : FAILURE in 8h 20m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b9a30d41c6cda3bf303ca3203e4a68dd152ab79e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-30 13:23:29.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/2/20230630165357/NEC-V-FC : SUCCESS in 2h 40m 15s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/886219/2/20230630165357/NEC-V-ISCSI : FAILURE in 5h 29m 29s","accounts_in_message":[],"_revision_number":2},{"id":"091c34c907fef4cbd48e74fc77cf6bb6be0a9a7a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-03 18:03:04.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-886219-2/FC/ : SUCCESS in 2h 36m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-886219-2/iSCSI/ : SUCCESS in 3h 24m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"fae4dfa41adc1df73549d1dddf53ceb7627e948a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-06-25 15:39:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05c050da34ffb03cb1d67b607144eb596749be63","revisions":{"7718be8b72edd7fe6d78c8bc14ceceeb183831e5":{"kind":"REWORK","_number":1,"created":"2023-06-15 21:39:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/19/886219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/886219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/886219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/886219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/886219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/886219/1"}}},"commit":{"parents":[{"commit":"e37366888251422a467039d47ee6e5642e5db7de","subject":"Remove all occurrences of six.moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e37366888251422a467039d47ee6e5642e5db7de"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 02:01:19.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 18:04:59.000000000","tz":120},"subject":"Remove all occurrences of six.text_type","message":"Remove all occurrences of six.text_type\n\nChange-Id: I83d8c7b1ef93851e2d4dc4b3da9886c90ee87f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7718be8b72edd7fe6d78c8bc14ceceeb183831e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7718be8b72edd7fe6d78c8bc14ceceeb183831e5"}]},"branch":"refs/heads/master"},"05c050da34ffb03cb1d67b607144eb596749be63":{"kind":"REWORK","_number":2,"created":"2023-06-22 15:17:53.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/19/886219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/886219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/886219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/886219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/886219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/886219/2"}}},"commit":{"parents":[{"commit":"e37366888251422a467039d47ee6e5642e5db7de","subject":"Remove all occurrences of six.moves","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e37366888251422a467039d47ee6e5642e5db7de"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 02:01:19.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 15:15:20.000000000","tz":120},"subject":"Remove all occurrences of six.text_type","message":"Remove all occurrences of six.text_type\n\nChange-Id: I83d8c7b1ef93851e2d4dc4b3da9886c90ee87f54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05c050da34ffb03cb1d67b607144eb596749be63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05c050da34ffb03cb1d67b607144eb596749be63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
