)]}'
{"id":"openstack%2Fcinder~886220","triplet_id":"openstack%2Fcinder~master~I8b6d25206335a87fb4633d8ffa4f920c089f6b99","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-25 15:39:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8b6d25206335a87fb4633d8ffa4f920c089f6b99","subject":"Remove all occurrences of six.add_metaclass","status":"ABANDONED","created":"2023-06-15 21:39:00.000000000","updated":"2024-06-25 15:39:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7d4e5a24bdbb324d2eb5869e1627fdda9e65d84","_number":886220,"virtual_id_number":886220,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 19:26:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 21:45:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 21:53:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 23:11:31.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-15 23:55:35.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 00:08:34.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 00:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 00:12:26.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-16 03:09:30.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-16 09:45:46.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-16 22:07:58.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-16 22:53:25.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-17 00:27:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-17 00:36:28.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-17 01:35:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-17 02:07:12.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-17 06:53:04.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-20 01:41:22.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 04:11:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-22 15:29:13.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4680cd043ceab9ad96ef4f143b3b9f14e65a2f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-15 21:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b27baa6f41d9cdf5893ecaa3fc52a04e5c4d231e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 21:45:56.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14206eb85723c01bd5e6a568ebcd0b7a1c4ce96b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac18957018f847a99f62e989cf03b1de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2601dff57ca4d0aaed6f175cce80999 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d70c4def0584b649949ae6ab557a796 : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/15a62fb1fac3419d947e4f0754c81fd9 : SUCCESS in 11m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d92ada92afdff42f548e0e30d93a39c0976d216","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-15 23:11:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/23d2aeefbc5a4e79bf7abd69eccddb7f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d618242adafa4a5eac1dd999bc002d70 : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6243dab229911472806d9a3aef732ab8430f053c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-15 23:55:35.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-20-886220-1/ns5_nfs : FAILURE in 1h 30m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-20-886220-1/ns5_iscsi : FAILURE in 1h 30m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"15fdb602d7b40beef5299ae66dfc7907ebd01538","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 00:08:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886220 : FAILURE in 1h 06m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0bb798eb5e8c0f67501a8d657062605a840e9a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 00:11:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43dd517e0f7344cbb381dd792444d8b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d4d0108c0a47b69f99522fe9dabe27 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55c8be3dfb3e4fbeb55faba90b06ec69 : SUCCESS in 11m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/404865f8dd284fbe8b71c6b195608b91 : SUCCESS in 13m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3648b97194d4f63b8bb39c4d24c17e2 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0480feb8ef84fc48c1f8798244ebb83 : SUCCESS in 13m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/440f247a462f4831a03803463c285c20 : SUCCESS in 49m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4c4c533b7244443a5b6a1aea2857f22 : FAILURE in 1h 17m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef79807d43984443af40f3290673ba8d : SUCCESS in 1h 02m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9bcc8e719474b6b9eadad6ce8f28f0a : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7ee56c3f9a3f40f094afe01c35625529 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/afca563baded4eb396c27510e937006e : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1008eda5370840fab84f40293fc705a4 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3b3593361ae417f8c2399e4b14f91ed : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5801b8edf7804eab9b587d1e0fff54a1 : FAILURE in 1h 06m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d35376dd277646b6ac768aa2f94895bd : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c5f31bb3c5d14e298a6d1059da304ea1 : SUCCESS in 1h 40m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2275396f5086402f844f61554605caa5 : FAILURE in 1h 40m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/830a6b568d4e425f880e05f6d040fd9f : SUCCESS in 1h 42m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3e60172f591c4fcfa2a4adcb18947701 : SUCCESS in 2h 00m 35s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/35c6bbf801484197adc248fcf9b24545 : SUCCESS in 56m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c029ad077a54836944360ea71e960b8 : SUCCESS in 1h 41m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2250b9bb41504eeb8b26b5d733cd4e67 : SUCCESS in 39m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e8f3f43787841e988c0b7ae26585dda : SUCCESS in 1h 22m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fb4cb33ee48d46bc8ed507cf663ae09a : SUCCESS in 1h 17m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57679e6dc97e4814b206a31ad30c34c1 : SUCCESS in 2h 29m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f46204078cbf4a2ba8769ea3b9ea1f21 : SUCCESS in 1h 00m 12s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/6dbab90766574de3be59fbb80ffb4468 : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf4cfbd9e8a0fe369bc6f71ec406580fed361db3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 00:12:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/20/886220/1/check/iscsi-driver-alletra-6k/34f7fe2/34f7fe2.html : FAILURE in 1h 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"65ab4036bfb68d9b741b9aed6fb2ca5def4cd526","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 03:09:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-886220-1/FC/ : SUCCESS in 4h 15m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-886220-1/iSCSI/ : SUCCESS in 2h 55m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"16d0d94fa79501bce37239df4453d1a76953220c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 03:23:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/20/886220/1/check/iscsi-driver-nimble/72f1a81/72f1a81.html : FAILURE in 1h 19m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4137e7c8760e1ee2375da319d7a63b5808e07d9e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 09:45:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_Unity_FC/066ac1d/DellEMC_Unity_FC/281 : SUCCESS in 1h 47m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_Unity_iSCSI/02337c5/DellEMC_Unity_iSCSI/281 : FAILURE in 30m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7cf151ce3313d591772db7f3f0524afc228222b5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-16 19:15:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"859d154eee7bdd644ddeff8cbec8a5853272e52b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 19:20:39.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86a6d1f09df6f19e38bc44a63f0a898e81003da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 19:27:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05dd7a3f68314690bbe0eca4a28efe8b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4b96ed186554bc7b0b7d89a9ae28f41 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3ace215b556486abc07b219b8cc74ae : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2a8973e1b6a418882b420b9ff48f052 : SUCCESS in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"946e9397da2d4b0b53269af0a839a0e55f81e638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 22:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d39b30de1c5e4784b0ae2423e10505bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1d05dfae404da9be1b993669a9bf95 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39e9be22643a4dd4b06317e0d3fa3d8c : SUCCESS in 10m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62a74fa048a74bb7b5a3e1a2460fef08 : SUCCESS in 13m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ba72f2f4c7d41fb9ba1f7525a8fca93 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb8d691175b4f9fab20bfe781b7d264 : SUCCESS in 15m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/02d49a17dbf049f089c86abccd01d6f4 : SUCCESS in 40m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4abe95b3f75641ff9406f075e34ac371 : SUCCESS in 1h 24m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cddfde859dc45b895984ad4f6cebb26 : SUCCESS in 45m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b3a7f3d2838460d9268895e173c61d8 : SUCCESS in 16m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b17843db0d414636bc55b00e84e0d611 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c78aeeb9a20f424fa31028677ac1cf04 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec5e791377cd423cb7717f85d86f12fc : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58f11de761e748cdb846d6bb1cfee707 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fea94108605f42dd8de3b2a8110c6119 : SUCCESS in 49m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d4db8ed1bdcf470f839f39c66d58ee5d : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e00443f20e254330864f2d7e9d6ffb67 : SUCCESS in 1h 30m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd662b2b31ba46cd818964f2eb79b9af : FAILURE in 2h 13m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8b6dea90d9b4d1390bfe28ead70f6e1 : SUCCESS in 1h 46m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f903c2f63c3542fda02c689ec22877f4 : SUCCESS in 1h 47m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5aca5f0c05944afeb470f55dc3ead2e5 : SUCCESS in 49m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/665d09f01cb545b7ae48f8222236f3c9 : SUCCESS in 1h 32m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de66394df9f649cd836de621da7cf9ea : SUCCESS in 37m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c4c310d229194e4b80a208e0a6f2a6ef : SUCCESS in 1h 42m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/84ef713abaa9429dafdc893f508a2a55 : SUCCESS in 1h 08m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79992aa35a3d45f99f2eb62c728ae666 : SUCCESS in 2h 39m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cc71ce07f174d9bb0e3d3ac09fb8cc2 : SUCCESS in 1h 06m 24s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/f57e3f2373b142eb83bd952c82d86b27 : SUCCESS in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f88171dbee5e41485085a8498a441b57344ee8e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-16 22:07:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_06_2023_22_01_04\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"0bdaf26f57e89fc4636d983c1adf57b808c05534","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-16 22:53:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerMAX_FC/e4f6afa/DellEMC_PowerMAX_FC/133 : FAILURE in 27m 05s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerMAX_iSCSI/615f459/DellEMC_PowerMAX_iSCSI/127 : FAILURE in 26m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"391cbbafcc750bb9b12eeb4c75e5daf8375fff23","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-17 00:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/c6d4cde47c7b4cc7bbbac03716f44446\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/40f83259420541938b179282374778e8 : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"faed2d038121e25807e3983ca76cfff87174852b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-17 00:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/886220/1/upstream-check/cinder-tempest-plugin-ontap-fcp/09d9ed9/ : RETRY_LIMIT in 9m 52s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/886220/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/b95ca80/ : SUCCESS in 2h 13m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/886220/1/upstream-check/cinder-tempest-plugin-ontap-nfs/637645b/ : SUCCESS in 2h 15m 22s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/886220/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1927830/ : FAILURE in 2h 15m 21s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/886220/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/3486b15/ : SUCCESS in 49m 39s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/886220/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/b8910fe/ : SUCCESS in 2h 26m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8c1cbc964b4c92d225261b72b6672f4f88e74a8c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-17 00:36:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerStore_iSCSI/92dace5/DellEMC_PowerStore_iSCSI/109 : SUCCESS in 53m 37s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerStore_FC/9035f3a/DellEMC_PowerStore_FC/114 : SUCCESS in 1h 23m 10s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerStore_NFS/bd30252/DellEMC_PowerStore_NFS/108 : SUCCESS in 51m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerStore_NVMeTCP/b8ab68e/DellEMC_PowerStore_NVMeTCP/109 : SUCCESS in 1h 07m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"a1e732e50361808a5204757d3e5b524c793ea9ce","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-17 01:35:27.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_PowerFlex/9f89f65/DellEMC_PowerFlex/150 : FAILURE in 1h 50m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"0844ae8688c5348629149d3fd23fcc856b809b11","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-17 02:07:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886220-1/HPMSA-iSCSI/ : SUCCESS in 1:50:11\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886220-1/HPMSA-FC/ : SUCCESS in 1:25:42","accounts_in_message":[],"_revision_number":1},{"id":"dfa73f07d2cea0d83fc7cd2241851311b8a070d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-17 03:39:35.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_Unity_FC/e9b3776/DellEMC_Unity_FC/294 : SUCCESS in 1h 57m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/886220/1/check/DellEMC_Unity_iSCSI/6ab6696/DellEMC_Unity_iSCSI/296 : SUCCESS in 1h 49m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc99860362ab669cd6c809915a0c724826552e9a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-17 06:53:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886220-1/PVME-iSCSI/ : SUCCESS in 2:00:20\n- PVME-FC https://os-logs.seagate.com/dell-ci/886220-1/PVME-FC/ : SUCCESS in 1:36:58","accounts_in_message":[],"_revision_number":1},{"id":"a929972c042f16025873469375cca03f1c3ede92","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-20 01:41:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/1/20230620075101/NEC-V-ISCSI : SUCCESS in 2h 45m 36s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/1/20230620075101/NEC-V-FC : SUCCESS in 2h 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6534ddec58ce3b35dfdc7be69f07c908501b4db5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:11.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/886220/1/fc : SUCCESS in 0h 32m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/886220/1/iscsi : SUCCESS in 0h 42m 46s","accounts_in_message":[],"_revision_number":1},{"id":"754159e2cc2693b437e0b7017b687f92bceb0bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 15:17:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"75c706d9f2d9b132badc5eb37ee177633fa40be9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-22 15:24:30.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e69ea67c162dfca5075a5a06b995a0b05f670a2e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-22 15:29:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/1/20230622164614/NEC-FC : SUCCESS in 3h 34m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/1/20230622164614/NEC-ISCSI : SUCCESS in 7h 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6f850bf08750f54542588492e828222f3fc84c26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:34:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905afc64fb5545719737a2323728933a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a6442706027407585ae1018b842bcda : NODE_FAILURE Node request 300-0021519036 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82f422034a244be5be24cc3903699a9f : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1022fd63a6fd46b39066d87cd682efdd : SUCCESS in 12m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd28dfbc6e26e395a0534f253f795d11ef2a4fc4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-22 17:52:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886220 : FAILURE in 2h 05m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b0c81c669b340988146b253363ef33e40976b122","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 17:52:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/20/886220/2/check/iscsi-driver-alletra-6k/b4ebd54/b4ebd54.html : FAILURE in 1h 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e7aca15cf3d2f5ce0663752657531b542af0c68e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-22 18:25:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/1aa95e0835dd45f781d933edbe83301a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/75e92ee59aea453b96330b7337fb9f78 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b73e9577ee5e801b8030bc0cb19b86fc4cd8ef65","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-22 18:27:15.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-886220-2/FC/ : SUCCESS in 2h 38m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-886220-2/iSCSI/ : SUCCESS in 2h 39m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"108b95598e1ae2d5a6c94b8fbea51419fb331b8a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 19:01:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/20/886220/2/check/iscsi-driver-nimble/af6e4c5/af6e4c5.html : FAILURE in 1h 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"8c02aea1f8f2b90d94291434fc11d46176893df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:26:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cca1de72853d478ab8d6029568758cd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a167b87bc46f4a118962aa071679d771 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32225c72222b45e6a55de9467b69efd0 : SUCCESS in 12m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d2a11a5b87457eb0ffae4cf6532869 : SUCCESS in 12m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/631fcf855ca94d00b63c91cc58faace3 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b95aab701d45ba8bd548833b3c28f8 : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/23f722319e904dec99394735ae74fc99 : FAILURE in 1h 07m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6d2310bdaf54bcabec48e9616676493 : SUCCESS in 1h 57m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/526ca4b8f5e9483bbc931c959e891575 : SUCCESS in 1h 11m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd67c01d7c204f50b853e6445c76f80f : SUCCESS in 20m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4d59dca04602496683ff453fc0b5a63d : SUCCESS in 7m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87d7da3072f44e64888ca2a3919f6470 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4563b748a1464f818dbf2f4e9405ec95 : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62f2b03d5a624540ac0b682812c4868b : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1a70e78a7daa4371abd9f7780bf6c023 : SUCCESS in 52m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/460497a64cc34e5ebd3500052ffd9b43 : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/993f9dbf9a324ff2a1a2db231aca2155 : SUCCESS in 2h 08m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e99dae9d20c14b919105410962ba8e39 : FAILURE in 2h 21m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fb2e51293ce94c9ca6d5c907dfb6a04e : FAILURE in 2h 18m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1a20040b856544428c7c8f83a676fcd2 : SUCCESS in 2h 49m 17s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/292d7fa118e542818826309ff14bd4ca : FAILURE in 1h 06m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f2b97feb164f4fee93994be3d2d931f2 : SUCCESS in 2h 26m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a0b5bfc0b664e3d933d18dcec0c6631 : FAILURE in 1h 04m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5ab7e6e0c7d4ed398a4fca6b8fd5105 : SUCCESS in 1h 25m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1ba4f12e392f4e6aa451aaab5144fa37 : POST_FAILURE in 2h 08m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2f0a16f73cea4c93add68bed68d6bf1a : TIMED_OUT in 3h 03m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd7cc1e7938d441ca5f8f34b2fa1f6ec : SUCCESS in 1h 23m 23s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/63b525e34d28454c83fa311d9b2356a1 : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a24eef4f67e454bc3315b031a0b031e69a1d29ac","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-22 21:19:04.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/20/886220/2/check/DellEMC_PowerFlex/f897c3e/DellEMC_PowerFlex/184 : FAILURE in 1h 51m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6425b2e6c408ebc4b509c5b9bca1fb0b097a73db","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 04:56:47.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-20-886220-2/ns5_nfs : FAILURE in 12h 38m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-20-886220-2/ns5_iscsi : FAILURE in 12h 45m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"c7cdaf6edb0f3489262053b153b5780d20e74e79","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-24 02:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/886220/2/fc : SUCCESS in 0h 28m 01s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/886220/2/iscsi : SUCCESS in 0h 44m 56s","accounts_in_message":[],"_revision_number":2},{"id":"92221e36f5e03ed2ebe691deee308cd6d98a3ded","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-27 14:34:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/2/20230627155234/NEC-ISCSI : SUCCESS in 3h 30m 15s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/2/20230627155234/NEC-FC : SUCCESS in 7h 41m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e701bbb4b11b7471d28b0a8fd93662e4751846a6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-30 16:06:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/2/20230630222333/NEC-V-ISCSI : SUCCESS in 2h 41m 27s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/886220/2/20230630222333/NEC-V-FC : SUCCESS in 2h 43m 05s","accounts_in_message":[],"_revision_number":2},{"id":"38d09fde8b5893f9e49cfb6dd540bbf6f2d7653d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-03 17:37:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-886220-2/FC/ : SUCCESS in 2h 28m 47s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-886220-2/iSCSI/ : SUCCESS in 2h 43m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"c7d4e5a24bdbb324d2eb5869e1627fdda9e65d84","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-06-25 15:39:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf5e7ebdfbc62ca3a030d993f91650571b0f7eb0","revisions":{"b759d9465289b6f6a48a210089283314ea6632fe":{"kind":"REWORK","_number":1,"created":"2023-06-15 21:39:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/20/886220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/886220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/886220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/886220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/886220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/886220/1"}}},"commit":{"parents":[{"commit":"7718be8b72edd7fe6d78c8bc14ceceeb183831e5","subject":"Remove all occurrences of six.text_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7718be8b72edd7fe6d78c8bc14ceceeb183831e5"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 03:11:38.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 18:04:59.000000000","tz":120},"subject":"Remove all occurrences of six.add_metaclass","message":"Remove all occurrences of six.add_metaclass\n\nChange-Id: I8b6d25206335a87fb4633d8ffa4f920c089f6b99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b759d9465289b6f6a48a210089283314ea6632fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b759d9465289b6f6a48a210089283314ea6632fe"}]},"branch":"refs/heads/master"},"bf5e7ebdfbc62ca3a030d993f91650571b0f7eb0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-22 15:17:53.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/20/886220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/886220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/886220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/886220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/886220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/886220/2"}}},"commit":{"parents":[{"commit":"05c050da34ffb03cb1d67b607144eb596749be63","subject":"Remove all occurrences of six.text_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05c050da34ffb03cb1d67b607144eb596749be63"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 03:11:38.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 15:15:28.000000000","tz":120},"subject":"Remove all occurrences of six.add_metaclass","message":"Remove all occurrences of six.add_metaclass\n\nChange-Id: I8b6d25206335a87fb4633d8ffa4f920c089f6b99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf5e7ebdfbc62ca3a030d993f91650571b0f7eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf5e7ebdfbc62ca3a030d993f91650571b0f7eb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
