)]}'
{"id":"openstack%2Fcinder~886223","triplet_id":"openstack%2Fcinder~master~I283a509e2693253066e54de085a0ade2b5758a55","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-25 15:37:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I283a509e2693253066e54de085a0ade2b5758a55","subject":"Remove all occurrences of six.PY3","status":"ABANDONED","created":"2023-06-15 21:39:00.000000000","updated":"2024-06-25 15:37:18.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"780d2e841c018baa9259f8c54434efce55781b33","_number":886223,"virtual_id_number":886223,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 19:43:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2023-06-22 15:17:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":-1,"date":"2024-06-25 15:37:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 21:47:34.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 21:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 00:29:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 00:36:28.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-16 00:57:35.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 01:03:47.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 01:37:37.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-16 01:38:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-16 01:46:37.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-16 02:05:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 02:43:35.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-16 04:46:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-16 07:28:20.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-16 08:48:35.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-16 12:26:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-16 15:32:04.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-06-20 10:46:10.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 04:11:17.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-23 02:42:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-28 09:46:01.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 15:37:15.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"ca8d08e31587ef49b35099c2b95320dfedc07bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-15 21:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e22afe618432c8d0704bd846de7d0b608f96db9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 21:47:34.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6a4bc968ef02f98f3677638c46713d0cb1c6e92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:55:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/087bfe0b69984143bd2ef7daffe80b81\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb85c2abe9a94b0bb084f2655ef9117b : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0d27121283c4700a106e9d2a9f04b6b : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6935ed9eed8945b19201d140520c3248 : NODE_FAILURE Node request 300-0021468957 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"925a6e6639ff930fb67a352dc39facc6a722892e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 00:29:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2dfd2b126db4c7a9bfeab0c404430f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcce2580edda4b50a8d892497926d4f7 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a63f3fa42523470d839ddce85a1d7d74 : SUCCESS in 10m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/243b1e99f2bf4b7e83095c3bddb54a8f : SUCCESS in 10m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7807d29fa374f63b704cc63501dba22 : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c30ca9a46c842dd90d8fa1856c11487 : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/8c734079e94648f9ae3be9b637210a54 : SUCCESS in 45m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abdd5ac7350d4e8bb8895cf25b12040a : SUCCESS in 1h 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94139540d0434fbb8113576b7fd9291e : SUCCESS in 54m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d7847900c624abca157c15411736494 : SUCCESS in 15m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/322d8fd414f449bfaa831b14ea9f0ea1 : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e7714b628ad4fdcb8335a2dcefa48f9 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/025039bbe9ef4d83bde6770c3ceda444 : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01c456b970154dd2bc718f38d715ada7 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8288e454dc3a45aa8240723b74f0fada : SUCCESS in 48m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89b1102ec54d45359367a8534c887e3c : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f766e49c9f2a4caa9cb4a7c587329cc0 : SUCCESS in 2h 16m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/69d00c4043a647b5aff6cad0e0417aa5 : FAILURE in 2h 25m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ba58645f2ae4fae9c93180377a7758c : SUCCESS in 2h 18m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c4f20c9aa8434449afa9f4519ddc631c : SUCCESS in 2h 26m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d538488b71354bea9cf5cd8e14508ee4 : SUCCESS in 1h 06m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ee40de87affc4e5a8a0b2fdcbfbc60a7 : FAILURE in 2h 03m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69d7cdafbc9c4720aef91499eb8eb160 : SUCCESS in 45m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/76a8c60e88c7481487a6e30149f290ab : SUCCESS in 1h 27m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1ac796ef559841a5be87df91cb495859 : POST_FAILURE in 2h 25m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45bd50a20bf347e48432a51f37bf749d : SUCCESS in 2h 20m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/644df30815e44effa8df56e52be7ccc5 : SUCCESS in 51m 47s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/423303732670489da9d58e3f4004300d : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3e9c6eebc9498eef0d2e05e90d32c9f51a9ec8b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-16 00:36:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_06_2023_00_29_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"735222fc5da0febb042228c16525d3ccad1bbbc6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-16 00:57:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_PowerMAX_FC/4057dd5/DellEMC_PowerMAX_FC/128 : FAILURE in 27m 34s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_PowerMAX_iSCSI/8db1380/DellEMC_PowerMAX_iSCSI/122 : FAILURE in 27m 38s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"6250af7bf6283d9dae47a2e29fc8850a7e437553","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 01:03:47.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-23-886223-1/ns5_nfs : FAILURE in 1h 31m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-23-886223-1/ns5_iscsi : FAILURE in 1h 31m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9ed853b94923d2839989eaa27b044354095796b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 01:37:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/886223/1/check/iscsi-driver-alletra-6k/b7757d4/b7757d4.html : FAILURE in 1h 15m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5f73eb7260a24b97697c8376c5daf52ffa20b444","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 01:38:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/b04af3bbfbda4eecb47a403c7201926c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c358a3f8df334f8e93abf5921aa49caf : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c01719d4f5a8d56425d4524ffc8d6518987214c0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-16 01:46:37.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_PowerStore_iSCSI/d6939c5/DellEMC_PowerStore_iSCSI/104 : SUCCESS in 51m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_PowerStore_FC/faf7d66/DellEMC_PowerStore_FC/109 : SUCCESS in 1h 16m 23s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_PowerStore_NFS/9a8656a/DellEMC_PowerStore_NFS/103 : SUCCESS in 58m 21s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_PowerStore_NVMeTCP/3469759/DellEMC_PowerStore_NVMeTCP/104 : SUCCESS in 1h 08m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f094243a1a5c5beb154b0faa250cb4afc7dd3d7e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 02:05:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886223 : FAILURE in 1h 36m 55s","accounts_in_message":[],"_revision_number":1},{"id":"33057d20a49182123ab8042b9b0a94e4e2c62fb6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-16 02:43:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/experimental-upstream-cinder-wallaby/cinder-tempest-plugin-ontap-fcp/6cce313/ : RETRY_LIMIT in 11m 35s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/experimental-upstream-cinder-wallaby/cinder-tempest-plugin-ontap-iscsi/c74eb06/ : SUCCESS in 2h 04m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/experimental-upstream-cinder-wallaby/cinder-tempest-plugin-ontap-nfs/e016b12/ : SUCCESS in 2h 10m 19s\n\nComment \"run experimental wallaby\" to rerun these tests.","accounts_in_message":[],"_revision_number":1},{"id":"75cf4c966665e9565442d460010ff9ad3de06727","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-16 02:44:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/upstream-check/cinder-tempest-plugin-ontap-fcp/ca238c4/ : RETRY_LIMIT in 8m 29s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/49e6c06/ : FAILURE in 2h 13m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/upstream-check/cinder-tempest-plugin-ontap-nfs/804c8bf/ : SUCCESS in 2h 03m 05s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/784bc5a/ : SUCCESS in 46m 30s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/56ffa25/ : SUCCESS in 2h 07m 31s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/886223/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/d11bb99/ : FAILURE in 2h 04m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"74f0e48fb1ac4db8116d5e5973397e801fe5cfaf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 04:46:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-1/FC/ : SUCCESS in 4h 25m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-1/iSCSI/ : SUCCESS in 4h 58m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"797b16ab859f56b430d593a60e32728783cef12e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-16 07:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886223-1/HPMSA-iSCSI/ : SUCCESS in 1:48:08\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886223-1/HPMSA-FC/ : SUCCESS in 1:27:44","accounts_in_message":[],"_revision_number":1},{"id":"519ec86cb631ac4a4bf8f402ce38a6bd716ee1e5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 08:31:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/886223/1/check/iscsi-driver-nimble/f73ff2d/f73ff2d.html : FAILURE in 2h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"606543148137a2f6ad547ccabe6d0c76122d5b6b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-16 08:48:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886223-1/PVME-iSCSI/ : SUCCESS in 2:00:02\n- PVME-FC https://os-logs.seagate.com/dell-ci/886223-1/PVME-FC/ : SUCCESS in 1:36:53","accounts_in_message":[],"_revision_number":1},{"id":"59a5952c1db8cb9fabe9799a4a55c3b2f3cdc716","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 12:26:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_Unity_FC/676a847/DellEMC_Unity_FC/284 : FAILURE in 53m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/23/886223/1/check/DellEMC_Unity_iSCSI/db3652a/DellEMC_Unity_iSCSI/284 : FAILURE in 29m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"bf9a82fd8ba93332420f97ffb34a3d0434ed6e9e","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-06-16 15:32:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1f2b34b96b2f16dd7695d43b9c8d29b1707448d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-20 10:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/886223/1/20230620164246/NEC-V-ISCSI : SUCCESS in 2h 37m 10s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/886223/1/20230620164246/NEC-V-FC : SUCCESS in 3h 03m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4e631ddc31070e3adb7d40659d6240a1fb58197a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:17.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/886223/1/fc : SUCCESS in 0h 29m 19s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/886223/1/iscsi : SUCCESS in 0h 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"14b60112ee0c502b3461ccb75e1785b5b30ea412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 15:17:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4e1b1914923b2a042e778ceeb1c18092e38b1418","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-22 15:25:52.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf03d5b36db69ae0418bf66ceeedc6c8b5b65b73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:33:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09e8ebdf99b54cd48f1ae6d32769b245\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38072ea49d40451584ed594a79ed034b : NODE_FAILURE Node request 300-0021519045 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7aad47b99d6d4f769f0137b53b0cd2e1 : NODE_FAILURE Node request 300-0021519046 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3db0def447c84666ac8d16a45a5db16b : SUCCESS in 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3053935beb27c0ca31cc586a226192efd6acb31d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-22 19:26:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886223 : FAILURE in 1h 18m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5d93c339f64bddb0fa4cc7d675d16ece2a579ff4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-22 19:38:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/9f48885916054c5a8cb8ee9d6276de3a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/442e3db3bf8449f6a7cb47c5dd3177f7 : SUCCESS in 1h 11m 50s","accounts_in_message":[],"_revision_number":2},{"id":"19b2a2e74b5af0cd3842579e5f95001d17120581","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 19:40:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/886223/2/check/iscsi-driver-alletra-6k/0dd8b2b/0dd8b2b.html : FAILURE in 1h 16m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3c6fa73ce65f96ac055b9960544cb2860d0fc644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:43:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dd719957f3c441aa3198d899ac98cdc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7544c3ae2a1d406ca025c5442ee833a4 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b96ac3c27494136b6fac11dae090969 : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b7b23e50fcc46a9a0a7923efa0dd02d : SUCCESS in 13m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4144633bfa9429bae2eccdef21908bb : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7477f0ad34a489baff92f9029bcda5f : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/71ea854b817748159170fb4673d0c945 : SUCCESS in 1h 12m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a51c74f6b554da09744ca9e736c8c4f : SUCCESS in 1h 34m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcc23b06b14f4e53ad926469052e3f10 : SUCCESS in 1h 02m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3b33f3b56d004018b91084d40584ca57 : SUCCESS in 19m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ec0e26b900640b099a39b7728055612 : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4219683536ce419f9f5f5ea2b1a8c7a8 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22166dcba2a348af8c3500475cef8829 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/67b8a1ec95a847d2857b9cd446dc90f9 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36342aa196f344b287e074a51080a0e8 : SUCCESS in 51m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/99e7471fb0eb44be80e9d83effa3f172 : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/051b19d0680646e3a786e9a5f56d901b : SUCCESS in 1h 52m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/94db0d8df44b480d9c0fedb6f28aa008 : FAILURE in 2h 11m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/767c142df2634a57acb50916b8b8850f : SUCCESS in 1h 54m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3acd15934ab14b608e4a0228932c9675 : FAILURE in 2h 55m 44s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e0570a78c1f34ba2ac7314d9b9bc8599 : SUCCESS in 1h 23m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/846fcfb3c1d4484280da7fc542d49f74 : SUCCESS in 1h 33m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fadde58ff669481ea05057688bbfab6f : FAILURE in 1h 10m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b38538f0f8fa42ef90e60f77ed30668a : SUCCESS in 2h 04m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fe6a8048030a4f7eb04e88121d680ebb : SUCCESS in 1h 38m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/034ff7d7187545e9925a80380f67b04d : TIMED_OUT in 3h 03m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9995528c4b8b4f7ea5397cd17a5afabc : SUCCESS in 1h 00m 51s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/53723de3aaa94286bab4513c8001c591 : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54391aa8d143d543a9b4e42eeb3741245687eac1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-22 22:13:09.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-2/FC/ : SUCCESS in 3h 13m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-2/iSCSI/ : SUCCESS in 5h 36m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"fb8af51b04d1ba45dd55e052bd61117b81823d48","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 22:49:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/886223/2/check/iscsi-driver-nimble/81300fe/81300fe.html : FAILURE in 1h 06m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e833b18dfec1c9ed0c46467bca905d2fde143d54","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 02:42:22.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/23/886223/2/check/DellEMC_PowerFlex/b9b4912/DellEMC_PowerFlex/187 : SUCCESS in 1h 43m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8af3fd464d7fcaafaf06a3e45509223c9e71a7f2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 05:49:54.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-23-886223-2/ns5_nfs : FAILURE in 12h 31m 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-23-886223-2/ns5_iscsi : FAILURE in 10h 45m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"45abf49cf2d36d9fdea2f4b4d6c8329c3fef95b0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-24 07:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/886223/2/fc : SUCCESS in 0h 37m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/886223/2/iscsi : SUCCESS in 0h 44m 57s","accounts_in_message":[],"_revision_number":2},{"id":"0876f2aaaa2db6abab4c755d460170c6243e7c2a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-28 09:46:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/886223/2/20230628104916/NEC-ISCSI : SUCCESS in 3h 42m 46s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/886223/2/20230628104916/NEC-FC : SUCCESS in 7h 56m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4f43e9b65b5fe46e505f3be246447c1bd344257e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-28 15:23:47.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-2/FC/ : FAILURE in 36h 21m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-2/iSCSI/ : UNSTABLE in 33h 36m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"d07984f944591bac94c85a02946c91c7e4e477c9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-01 05:57:26.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/886223/2/20230701091013/NEC-V-FC : SUCCESS in 2h 54m 07s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/886223/2/20230701091013/NEC-V-ISCSI : FAILURE in 5h 47m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a5bc30c8bac2f1d4ab5b5059e9260dfa47b00d82","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-03 19:47:54.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-2/FC/ : SUCCESS in 3h 47m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-23-886223-2/iSCSI/ : SUCCESS in 2h 48m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"86fba3187c2718c5b34b781b7ffc141ea99171b6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-06-25 15:37:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"780d2e841c018baa9259f8c54434efce55781b33","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-06-25 15:37:18.000000000","message":"Abandoned\n\nThis has been completed as of Ie27f31bbfb.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc335a8c9188755c9b680f5d8e384bd24061ac91","revisions":{"da40c793ede4cb40a08410ec6e2d4152aa9a994e":{"kind":"REWORK","_number":1,"created":"2023-06-15 21:39:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/23/886223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/886223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/886223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/886223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/886223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/886223/1"}}},"commit":{"parents":[{"commit":"8fdda5db23f7ed61c56ac5ce4a333e26587dd6e7","subject":"Remove all occurrences of six.iteritems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fdda5db23f7ed61c56ac5ce4a333e26587dd6e7"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 17:09:14.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 18:04:59.000000000","tz":120},"subject":"Remove all occurrences of six.PY3","message":"Remove all occurrences of six.PY3\n\nChange-Id: I283a509e2693253066e54de085a0ade2b5758a55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da40c793ede4cb40a08410ec6e2d4152aa9a994e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da40c793ede4cb40a08410ec6e2d4152aa9a994e"}]},"branch":"refs/heads/master"},"fc335a8c9188755c9b680f5d8e384bd24061ac91":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-22 15:17:53.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/23/886223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/886223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/886223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/886223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/886223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/886223/2"}}},"commit":{"parents":[{"commit":"f26448a89cda40b1c0e9842ed525881fad2e6711","subject":"Remove all occurrences of six.iteritems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f26448a89cda40b1c0e9842ed525881fad2e6711"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 17:09:14.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 15:15:28.000000000","tz":120},"subject":"Remove all occurrences of six.PY3","message":"Remove all occurrences of six.PY3\n\nChange-Id: I283a509e2693253066e54de085a0ade2b5758a55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc335a8c9188755c9b680f5d8e384bd24061ac91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc335a8c9188755c9b680f5d8e384bd24061ac91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
