)]}'
{"id":"openstack%2Fcinder~886225","triplet_id":"openstack%2Fcinder~master~Id5ac092d76461f951622e481bb87daf3d66bf18e","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-25 15:39:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id5ac092d76461f951622e481bb87daf3d66bf18e","subject":"Remove all occurrences of six.{string,integer}_types","status":"ABANDONED","created":"2023-06-15 21:39:00.000000000","updated":"2024-06-25 15:39:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26f903b728006c2ed4f39e86439e92e265e40316","_number":886225,"virtual_id_number":886225,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 19:56:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 21:48:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 21:53:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 00:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 01:48:50.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 02:36:19.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 02:54:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-16 05:28:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-16 14:15:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-20 23:31:53.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 04:11:31.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-22 23:38:46.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-23 06:15:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-29 01:04:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"2d0507ee4e0a8020f84cf3a0dfe8ac45d5fe7b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-15 21:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d57bcdc7d9beddbd98786063ea238d2c33adde7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 21:48:31.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25fa38e8d097c006b0e7380a2b9b97edd48823d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:53:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1cada5d7cae49748f796a0da0666146\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a61445782914855bf734e8f863cf79d : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ffbdbbeb8f644008aa798a5baf63405 : NODE_FAILURE Node request 300-0021469016 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34765d9b2e254ee2b44571046f93c331 : NODE_FAILURE Node request 300-0021469017 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bd26f16fcb34b91f0ea8343d2311f19fa78586f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 00:52:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ecf6186b6634287a50fecdcc6901771\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c0cf7cd8314e5699d60d9f3a16662a : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35aa98a35c02437b97257c16b70e0819 : SUCCESS in 12m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14f636b5abc44d82a46e068d304bed4a : SUCCESS in 12m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7532046ccd074e3c92d01009d993de8d : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845e5ac57fec40cfa5c4bf31101c9b26 : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/c905007734d74ca49c0b368efe9f7771 : SUCCESS in 54m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba4e864814be49b08c2c06bc2d89bef1 : SUCCESS in 1h 36m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d024cf176b634c62991f8090837d534c : SUCCESS in 1h 15m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/30e7b5982dbc40c28be4f032dd26998d : SUCCESS in 16m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2f6cdfa9cab4601be7b4840385245d3 : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9df82e599c71411fb85200d9a7537980 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48c70a97dc6442e097fb97008fb4c375 : SUCCESS in 6m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/29785d1ca21c456ab5517e7fd62236d6 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2d2d07eed5c48a6958484cff6567f2b : SUCCESS in 1h 03m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26d57f63897940b99bb50254b0063df8 : SUCCESS in 4m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f053023957954f65b3d961514d171bab : SUCCESS in 2h 04m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb59261bac2f49348ec9b44d8d0cdd5f : FAILURE in 2h 02m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3353ce77d7c440f2a6b8d5e7cfbcb972 : SUCCESS in 2h 12m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/37dabf3c2c2f41e5ac518e7bbc53368f : SUCCESS in 2h 19m 06s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99b7fd10cb864307a6bfa1f7c832f4af : SUCCESS in 1h 04m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9bef14e1d77240ab86bd48b8327d2f3b : SUCCESS in 1h 56m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0d960ffb0bbf42ccbc38f106e4aa2256 : SUCCESS in 50m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f7de1a2a2124b1197eaaa4efbdf1ec7 : SUCCESS in 1h 30m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/435ec3a4d5a8491b9b77a6cc6b6d91f5 : POST_FAILURE in 2h 11m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3fa50d6908074b3caceacbd49fad78c0 : TIMED_OUT in 3h 03m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/274b4f2e6a2e4f4b93bc9157c711202d : SUCCESS in 1h 01m 00s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/1e8a8dce9bbf4978a30b55dddc9d7bdb : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5338aa948a8c1902b044a4ba7aa46e647e65e0f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 01:48:50.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-886225-1/ns5_nfs : FAILURE in 1h 31m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-886225-1/ns5_iscsi : FAILURE in 1h 31m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ecc469c775761aeec969307234c0df6cc374ec17","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 02:36:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886225 : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":1},{"id":"95acf2fd165776d100f32c754a0f33dbab1be581","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 02:54:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/886225/1/check/iscsi-driver-alletra-6k/96ab80c/96ab80c.html : FAILURE in 1h 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2af0088a53728f12bc97a238c24c056e6ad5fe58","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 05:28:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/e7c25a39b7f24f779c85ebea592663f8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d04c2398bbcc42e59ec09888b0083ed7 : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1362da3f45134d15209fadaed440a7a0c793091c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 11:35:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/886225/1/check/iscsi-driver-nimble/0a39049/0a39049.html : FAILURE in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"789aa02d0da54ea0b803da0998969ebd609fd5fe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 14:15:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/886225/1/check/DellEMC_Unity_FC/14f0369/DellEMC_Unity_FC/286 : FAILURE in 54m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/886225/1/check/DellEMC_Unity_iSCSI/9814522/DellEMC_Unity_iSCSI/286 : FAILURE in 29m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"18e251112af9d671af3bc797997a861562c47979","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-16 19:15:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bddeb434d793de184b9c2e56a87bc75c66d1d6e2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 19:19:07.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3acfb12e6c9f9c3c6950edff8efd83c936841ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 19:30:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10d735dcfad494796df3e6a22d6a4b0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b510b39d055349929c1e5b1ccafd29b6 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c883908f47da4a069356dd86380e95be : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03af47ef399c440483fbb1be6b4d5941 : SUCCESS in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ed3f906d78673d597c3b93756ef2935c526420d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 22:19:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/90bf20cf523642498807b724a0fcdba7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8ce465afb4b048dd8a064913dd84252c : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"674745a41579a1d0ea621de3d9e85b03a4e0f1a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 22:28:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad80b6a257ff48b4b692611e8572bfce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a279c699197945be9f4f75e2e32b55be : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd19103ec324d5f9d1cf193a5dd7d06 : SUCCESS in 10m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73799829ead6414bb594f5400354b676 : SUCCESS in 10m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0729ab63c4a4eba8d6bd164a1fdb851 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003177ef606f458f86947e950f042192 : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/5a50e6e39ce24ea793fc676540ec1ae0 : FAILURE in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7dabe5a3a9eb4d7d9c6c4bcaf545e6c2 : SUCCESS in 1h 36m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6efd7624a3204b248e7c3e2474a2088f : SUCCESS in 1h 05m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95332ec0c7484bfbbaec77bca3432b19 : SUCCESS in 17m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf9d256e51ac426c9d65ca3214d61fae : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f3ac9da3c05c4aa38967c50ae09aa02a : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22f6b1542b9346378c938041c6beadb7 : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/42836da3ded54dca8f8d79edb0399cd0 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f8e40f0d7ec74247ac9eff0695499648 : SUCCESS in 50m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/16e9926c1b58477d960c09555efb27b9 : SUCCESS in 4m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3fd3f44b1b584aac93f7d8ef63e079ff : SUCCESS in 2h 03m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d6451761a7894b07bf2f8d90dd97a41f : FAILURE in 2h 03m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3b281d4303cd4595acdaf623513e9714 : SUCCESS in 1h 57m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e8ff4fcd02b0493ca59e919bbc2e2cbc : SUCCESS in 2h 19m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/50fe355578d047abb2fe7914d857a61f : SUCCESS in 1h 00m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbcb613520ac472a8306cf529db6f968 : SUCCESS in 1h 40m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ce946483c09a42c8b0a04601444b65ee : SUCCESS in 52m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5e51efe8e80f4563aa5d1e513cb00f8e : SUCCESS in 1h 08m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a5b0e666d51d4023a69544d9b6ac0648 : POST_FAILURE in 1h 34m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e578678c1ea34cc0a6f39dac56f125fe : TIMED_OUT in 3h 03m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fac6703207be4a29b321fae3286409bf : SUCCESS in 58m 02s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/54c55db1c462422188d556fa9b553e2e : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44b4fcd015ee294e088a86495ad3c5d867946a9a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-20 23:31:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/886225/1/20230621052147/NEC-V-ISCSI : SUCCESS in 2h 41m 10s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/886225/1/20230621052147/NEC-V-FC : SUCCESS in 3h 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5265c709b95986e1b8a86e4526f936a5de644c9c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:31.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/886225/1/fc : SUCCESS in 0h 32m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/886225/1/iscsi : SUCCESS in 0h 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"50dfa7549e4c09933725962057f7587921c4160f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 15:17:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f72c76151cde1636ba4cf64145d3a564311e9688","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-22 15:27:03.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87cb4ac0445b05dcf0bf07563944a1cd82dcd458","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:36:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d46ff1c1a3c747eea4059b071467dd2b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af01788876324e8cafaead117f47139f : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4e2402e99614c29bcd16b0e0e9b7649 : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d4ecafcb9d9411faf48c8f6f96ab90e : NODE_FAILURE Node request 300-0021519061 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5342ae30c7839589e7c235ee861ec0f9b3d2c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:56:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7bd7a56f11a40a5a22a33d7fc3fb936\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e3437fe6b940108053f65003b97f1c : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efdcd18f1b7e4945b5439d312e65f42f : SUCCESS in 13m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eb71c4b3f4c4b21ad54622f77396f8b : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b9a830c544a4ee3a156a8c54d7905d9 : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088219b18fb24e26b3ff976790b939b5 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/5521a5298c934c96af8efd206b419579 : SUCCESS in 1h 04m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2cf8d20085a94840aac77edfa60d797b : SUCCESS in 1h 30m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/106a9f0c7506486aac24dd7d092f4af7 : SUCCESS in 58m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8710553ba3214a99987b6ab22366d035 : SUCCESS in 16m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ea718acbd86f4c62985f963807bab8e2 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff475370ef784afb92f01428c4f5fed6 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/387324edea7840289a4cc20baa92b5a9 : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b740c4cf151d4befbd358ac94a3f459e : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b7223f443df47c884b3bc732dc5757d : SUCCESS in 56m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8692ddd7ffd64d06aed94f95a4cdd162 : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/56c0832018d7449290e2548456788634 : SUCCESS in 1h 59m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d80e10f245040a2900d57ecb5489b86 : FAILURE in 2h 48m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/feb067c3570449ba9fe47339abc88ee2 : SUCCESS in 2h 44m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a7025d9a19414e66814c4d9b0e08aac7 : FAILURE in 2h 09m 53s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/87709ac2ff9048b28abafd2740294cb8 : SUCCESS in 1h 29m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/89366a83f609497a846c2d7fb3746bf1 : SUCCESS in 2h 12m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8698ac61014c49e8af649f6e3551e3f1 : FAILURE in 1h 06m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3571c9702ad54a3c9c0fec3d4618e0b7 : SUCCESS in 1h 50m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4fe2cb0ffdac4122a69c1c2f4325b37a : SUCCESS in 2h 07m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad31b2de2e0a4990a3ccada51d2723f2 : TIMED_OUT in 3h 03m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf0bebc1d16543e59cd2a68e8af7f74e : SUCCESS in 1h 07m 32s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/96a1d7dde14d4a69a6fbf0c87e43e555 : SUCCESS in 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02be1ea432df993c6a310649fda2b9d503075884","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-22 20:54:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/871e42c27726474cb1275aa4fb1fa27d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7bf08d36c12141b2bda4d67779026401 : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6e8ca7a9fe0b955457868280995174ecf99bd8a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-22 21:20:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886225 : FAILURE in 1h 47m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f9c533f1a8321b418ebcf1806939faf8adb84555","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 21:29:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/25/886225/2/check/iscsi-driver-alletra-6k/d48a642/d48a642.html : FAILURE in 1h 49m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1414367491c1b29a5468dde83099dfd191b180c5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-22 23:38:46.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-886225-2/FC/ : FAILURE in 5h 16m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-886225-2/iSCSI/ : FAILURE in 5h 38m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"0df6f0fcc89fd25ba993223804b41e4dd676b684","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 01:42:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/25/886225/2/check/iscsi-driver-nimble/2549181/2549181.html : FAILURE in 1h 20m 42s","accounts_in_message":[],"_revision_number":2},{"id":"7fedbb309e6f795cff56caa6f0f0ebd5c170ea94","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 03:57:22.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-886225-2/ns5_nfs : FAILURE in 9h 53m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-886225-2/ns5_iscsi : FAILURE in 9h 53m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"855792b0a6667c864c8f5157488aa808915332dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 06:15:12.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/886225/2/check/DellEMC_PowerFlex/3f78239/DellEMC_PowerFlex/189 : SUCCESS in 1h 46m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"81f2d44e117a5edde87988d524189b458a5ed94e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-24 11:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/886225/2/fc : SUCCESS in 0h 45m 28s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/25/886225/2/iscsi : SUCCESS in 0h 50m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2e9147fae840c102b3b2fde2d448cc8dad61ac8f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-28 15:23:28.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-886225-2/FC/ : FAILURE in 40h 43m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-886225-2/iSCSI/ : FAILURE in 39h 54m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ac7b83d19df86a7fc9d8fc5d0505126ed2e9b62","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-29 01:04:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/886225/2/20230629021446/NEC-FC : SUCCESS in 4h 02m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/886225/2/20230629021446/NEC-ISCSI : SUCCESS in 7h 50m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3c38a698bc01872890016cd34bc6605286cb8b99","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-01 11:30:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/886225/2/20230701174109/NEC-V-ISCSI : SUCCESS in 2h 47m 21s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/886225/2/20230701174109/NEC-V-FC : SUCCESS in 2h 49m 24s","accounts_in_message":[],"_revision_number":2},{"id":"bc4c51a08534da608a3b59c568a8b797b71dc31d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-03 22:31:08.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-886225-2/FC/ : SUCCESS in 3h 35m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-25-886225-2/iSCSI/ : SUCCESS in 2h 14m 49s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"26f903b728006c2ed4f39e86439e92e265e40316","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-06-25 15:39:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f680e8f93b9d83ac167b93c723e9c90112585b73","revisions":{"8b256a3b5ae5c0fda834d05169bef0ab6b8e9710":{"kind":"REWORK","_number":1,"created":"2023-06-15 21:39:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/25/886225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/886225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/886225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/886225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/886225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/886225/1"}}},"commit":{"parents":[{"commit":"e26947dee106aca77ad6ea07c5302c9b71a4b378","subject":"Remove all occurrences of six.wraps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e26947dee106aca77ad6ea07c5302c9b71a4b378"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 18:00:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 20:54:14.000000000","tz":120},"subject":"Remove all occurrences of six.{string,integer}_types","message":"Remove all occurrences of six.{string,integer}_types\n\nChange-Id: Id5ac092d76461f951622e481bb87daf3d66bf18e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b256a3b5ae5c0fda834d05169bef0ab6b8e9710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b256a3b5ae5c0fda834d05169bef0ab6b8e9710"}]},"branch":"refs/heads/master"},"f680e8f93b9d83ac167b93c723e9c90112585b73":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-22 15:17:53.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/25/886225/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/886225/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/886225/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/886225/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/886225/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/886225/2"}}},"commit":{"parents":[{"commit":"5e94d2ae2964bbd4528811fe833634b3c4d0f399","subject":"Remove all occurrences of six.wraps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e94d2ae2964bbd4528811fe833634b3c4d0f399"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 18:00:51.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 15:15:28.000000000","tz":120},"subject":"Remove all occurrences of six.{string,integer}_types","message":"Remove all occurrences of six.{string,integer}_types\n\nChange-Id: Id5ac092d76461f951622e481bb87daf3d66bf18e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f680e8f93b9d83ac167b93c723e9c90112585b73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f680e8f93b9d83ac167b93c723e9c90112585b73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
