)]}'
{"id":"openstack%2Fcinder~886228","triplet_id":"openstack%2Fcinder~master~Id7741dc14da8610f0cea4ab53e27354e4ac8eedf","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-06-25 15:39:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id7741dc14da8610f0cea4ab53e27354e4ac8eedf","subject":"Remove all occurrences of six.reraise","status":"ABANDONED","created":"2023-06-15 21:39:00.000000000","updated":"2024-06-25 15:39:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc485e638c5ebcc04a0bbbd53430655224b35a88","_number":886228,"virtual_id_number":886228,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 20:12:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 21:45:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 21:50:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 00:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 02:54:09.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 04:13:03.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-16 04:49:23.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-16 05:37:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-16 06:18:46.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-16 18:34:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-16 22:15:15.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-16 23:19:03.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-17 00:24:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-17 01:58:24.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-17 03:24:15.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-17 06:11:57.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-17 11:16:29.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-21 04:11:58.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-21 14:43:08.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-30 01:08:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"49f543451da8c7db313a3bb3844103f565bd1167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-15 21:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68b341c2b61297e1ec13ddc6c637e4172c8fedb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:45:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8abd0e1eb841079acfedbe6972a963\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cf4e5db788c4d9684f5bc9d453db681 : NODE_FAILURE Node request 300-0021469051 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef18b46d6e2748809d589ad3167ac2cc : NODE_FAILURE Node request 300-0021469052 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18c67928db5f43209d922382b74be8bc : NODE_FAILURE Node request 300-0021469053 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc4867d3118aa4e02749854be4c21292f9ada742","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 21:50:01.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8a36eb21c3c71305599293403eead417d06fe32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 00:54:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b144ed6815104b6daee9c46436a8a4e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3c26bad51a48609f3d0a42ea59d4bd : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac11a3cd693b4fa3b256c3af68b3760b : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d3067f160d423aa954d3078a53b01b : SUCCESS in 11m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93800235b099403fb397a47c61489bed : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0943642b9a7d438e9e22cc6c069a3d42 : SUCCESS in 14m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/872022ef67b7496b9cf0a67fbe3f6ac6 : SUCCESS in 2m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/5ff4e6f814fe46c58ce089da5452a9ff : SUCCESS in 48m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40d570953fdb4f7a883b7a2060be76ae : SUCCESS in 1h 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ac02b630d104106921a5e9696ce20ed : SUCCESS in 1h 08m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac0050bc91894d48a6b8f6a98c4ff371 : SUCCESS in 15m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a344d5ffe74a4c77a38845aefa57b28c : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17d85c22582b42278bce6218fcc96c15 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38b709da32414fe796edd568da5bc9f4 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/edbc9ee291c54c3994717b592439c343 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5f4560bddc7440a0b1906ce45a85d28f : SUCCESS in 57m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0b661e64be64d7aaa9f786dbed372de : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b50fef81b3ec4729b924c3e6f1616e68 : SUCCESS in 1h 55m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aac7ec72d9da44b7a12725c33579c3e4 : FAILURE in 2h 14m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4c0af117802642a0b48945980a60f4aa : FAILURE in 2h 47m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b23d477239d54cfb9ed7c4a5c36f27f9 : SUCCESS in 2h 26m 39s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/884fc72a01154b5a8c738af02e0331aa : SUCCESS in 1h 29m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4101c40edb074a8794dd031682626339 : FAILURE in 2h 07m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/635a1a8b22f149f897b0ad8105104f00 : SUCCESS in 33m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eaf7d13a1cbc42edb605e7431628287f : SUCCESS in 1h 23m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/530787e041174d9b90a00be40310cd85 : SUCCESS in 1h 42m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e52a012a3c6449eb490cf4be7866d81 : TIMED_OUT in 3h 03m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/368147e9bb0847a4acf16db3c94dc2e9 : SUCCESS in 1h 04m 24s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/9663c93da7f94d96aaaddf0ef78c406d : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25dbff7ab6cf6597260726d6741209e525ea776b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 02:54:09.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-28-886228-1/ns5_nfs : FAILURE in 1h 28m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-28-886228-1/ns5_iscsi : FAILURE in 1h 28m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"582e1a305f706eeb6fb301990993bcbf5626a7a8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 04:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/266f4c96033d4ff39c75d6814feb43b8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8749db3d2fc245ee89c67db5037c70d7 : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":1},{"id":"758c215391e27f794da6b9c351ba4235f10d9328","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 04:49:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/28/886228/1/check/iscsi-driver-alletra-6k/0e99321/0e99321.html : FAILURE in 1h 03m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bcd5d993319547b9156428c3ac84e41989bcf5f8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 05:37:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886228 : FAILURE in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d2032cf5597944de86164041c37c0e1259c25ac8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 06:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-1/FC/ : SUCCESS in 4h 20m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-1/iSCSI/ : SUCCESS in 4h 11m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f63487a9effa03bbce103b537fa0948425cd81aa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 14:43:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/28/886228/1/check/iscsi-driver-nimble/fa055b4/fa055b4.html : FAILURE in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"eea0b55a30c6b509bbede8ebd70acaa483c676fc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 18:34:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_Unity_FC/e28ae0c/DellEMC_Unity_FC/289 : SUCCESS in 1h 33m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_Unity_iSCSI/f1c7ce0/DellEMC_Unity_iSCSI/289 : FAILURE in 28m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa03bb7fc6d3ada99f3a5b77b87f50afce4a7e3c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-16 19:11:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"829770381a077d31fe456125f36a20108eb0edd3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 19:13:34.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74d420522c2c2021bf193e4b9bcaa24f8f049b17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 19:23:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d3682944904686baef587f10225694\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e475d7c73774fdf98955061f410c568 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02368e22abc34151ae3246265b3b5bcd : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa590b8ff2bb4dd9a282c7ad459dd445 : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53c574076c5f6479e3393bc3e4f57bfea6bd0426","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 20:34:24.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8a017bf51e9c44599e4d8641d1ec7a2a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1098269a814046ed870e8614f4cc3adf : FAILURE in 1h 15m 37s","accounts_in_message":[],"_revision_number":1},{"id":"057eea31921f20d4c6c9a79b6b5d24a15c35d86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 22:05:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dda1dc1fa70e4edcbe0df13907808da8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a209becdfd45fbb0af92e5f142af15 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b54f865e9334c1dbef5844fd0f4fa1b : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b8d44a9a42f4007b5f57d3395772c72 : SUCCESS in 12m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b6f26fd8b44444080b044806ad42ed6 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c0489dc4614143b93ed9e7c628596c : SUCCESS in 11m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/79157057882b458089d75e56d64186ff : SUCCESS in 3m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/fbb448c689da4627bc10bccfe0d00f8c : SUCCESS in 47m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40f7f110fdc34f76995b41a50ec3f9c1 : SUCCESS in 1h 38m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/066f03f5ff1b4b75a175193b5fa1f833 : SUCCESS in 1h 01m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/61208064c72e4b6b905fc5416e8a896c : SUCCESS in 17m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/233a6caa6d9a4b34a67c768c15036211 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cad2b91c46024edcbf7a941d065f1b60 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/14a77cdd5e7f436da8c1b34f70b03835 : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e90d8e2ee3f420d9e5cbb860cb852e1 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9e9d60200de442a2a9c379b228b5ab98 : SUCCESS in 40m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b09f118c64144cc184afd4414198044d : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a1e09df99ce3495b92ebb43f457db5e2 : SUCCESS in 2h 14m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01ba1736cfd04389a3598031eaade3cd : FAILURE in 2h 00m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0482fb3112c4b98b56fdc66c1a5c919 : SUCCESS in 2h 39m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6a6a66a958d44d0a8b60a40a43b9481a : SUCCESS in 2h 07m 49s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3751347192c4872ba7157a5c424d7ff : SUCCESS in 58m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/57be26b75c724fbf86dc6e1f9a4e0c69 : FAILURE in 2h 00m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ba380c5f10747cf98b3caef832b1a6b : SUCCESS in 43m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/662db136daa14885bb304356fdaf5ef2 : SUCCESS in 1h 43m 25s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/aafacaa6e2ed4d6f84a74848412d05d1 : SUCCESS in 1h 28m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02ef531d4cc74a299f5b5ab5e3d6fac4 : SUCCESS in 2h 16m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/998e8df2e7134a26835509a21008c5a1 : SUCCESS in 55m 09s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/99e6c371921446b3b428064a07709d92 : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72aaa58ad9677515f4a6bf91487cf1ef834a7411","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-16 22:15:15.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_06_2023_22_08_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"2177e9bbf0c2e3bcfddee3c8375ca839a6e4fe5f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-16 23:19:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerMAX_FC/ed6c00c/DellEMC_PowerMAX_FC/134 : FAILURE in 27m 50s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerMAX_iSCSI/5cc9452/DellEMC_PowerMAX_iSCSI/128 : FAILURE in 25m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"30e3e1e88951f0c6787c19e788357d46df5a8f02","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-17 00:24:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/886228/1/upstream-check/cinder-tempest-plugin-ontap-fcp/45303b2/ : RETRY_LIMIT in 10m 13s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/886228/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/c11541a/ : SUCCESS in 49m 06s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/886228/1/upstream-check/cinder-tempest-plugin-ontap-nfs/55594d7/ : SUCCESS in 2h 17m 00s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/886228/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c029717/ : FAILURE in 47m 25s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/886228/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/7eb509c/ : SUCCESS in 47m 27s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/28/886228/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/16ac1a8/ : SUCCESS in 48m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"db8ce0009989d15e32e4608b01889040d75a515b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-17 01:58:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerStore_iSCSI/7505369/DellEMC_PowerStore_iSCSI/110 : SUCCESS in 54m 26s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerStore_FC/fbf11a3/DellEMC_PowerStore_FC/115 : SUCCESS in 1h 21m 55s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerStore_NFS/7ce2b1c/DellEMC_PowerStore_NFS/109 : SUCCESS in 55m 47s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerStore_NVMeTCP/aff2a57/DellEMC_PowerStore_NVMeTCP/110 : SUCCESS in 1h 32m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"b04bb85b3742978a41aeec2c2aa45c3dbcec439a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-17 03:24:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_PowerFlex/4bd87cc/DellEMC_PowerFlex/151 : FAILURE in 1h 48m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3819631d9155ba717cdf5e24d424c5153848752b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-17 05:41:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_Unity_FC/7892c98/DellEMC_Unity_FC/295 : SUCCESS in 2h 01m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/28/886228/1/check/DellEMC_Unity_iSCSI/01646e5/DellEMC_Unity_iSCSI/297 : SUCCESS in 1h 53m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a92cbe35e2b06ec856c32205b12a1f84322dd260","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-17 06:11:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886228-1/HPMSA-iSCSI/ : SUCCESS in 1:47:43\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886228-1/HPMSA-FC/ : SUCCESS in 1:26:23","accounts_in_message":[],"_revision_number":1},{"id":"3fdb6084efb9b2557379724f009d228c86cc9b6b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-17 11:16:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886228-1/PVME-iSCSI/ : SUCCESS in 1:57:59\n- PVME-FC https://os-logs.seagate.com/dell-ci/886228-1/PVME-FC/ : SUCCESS in 1:36:12","accounts_in_message":[],"_revision_number":1},{"id":"bf1bd5c606d0982572e0eae9bde56673f8efe1dc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:11:58.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/886228/1/fc : SUCCESS in 0h 31m 52s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/886228/1/iscsi : SUCCESS in 0h 41m 34s","accounts_in_message":[],"_revision_number":1},{"id":"30c61e9f087c10f062b731cc57d88fae8e556224","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-21 14:43:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/886228/1/20230621203611/NEC-V-FC : SUCCESS in 2h 58m 15s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/886228/1/20230621203611/NEC-V-ISCSI : SUCCESS in 3h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5e0aaf834c5f4a2c9ffd732212d97a6efb5d1f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 15:17:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fb62e857c6d8d86760070996a17ae32060d08805","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:23:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f86a922a130544749edfa2a7a3e02041\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32e107f16ead4494bc5fd59347dbc02b : NODE_FAILURE Node request 300-0021519068 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/008f8de785d148fcba68f5621ce25c60 : NODE_FAILURE Node request 300-0021519069 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbbb1e80b58d4e77bd1a16f425c86988 : NODE_FAILURE Node request 300-0021519070 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"03f9cbfb55716d5a2e2246ccce046891638c19cd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-22 15:28:23.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d79f2086fa37b741edbd6efa77fa5b623f154c00","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-22 19:22:27.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-28-886228-2/ns5_nfs : FAILURE in 10m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-28-886228-2/ns5_iscsi : FAILURE in 11m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"6a1537df88e97a2d22f87ffd80fff31cfb4bcdfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:12:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbf0c5e985134cc5844abecac051d76d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093acdc53daf4fe09e95a28f959befc3 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc687e99b32641a2a91795ae612a3e60 : SUCCESS in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee3d723db6ac4b2c9b38adc338ed0f6b : SUCCESS in 12m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a844a0c8e1d49458109c10640779947 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6210ee47f8404397734015e3da34da : SUCCESS in 14m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c94eeeea726048fbac17961abe785058 : SUCCESS in 3m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/144eff2056044c14accd0a477d6b274d : SUCCESS in 1h 12m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b230b81d27b84ef9bfe5dab4abc73cc9 : SUCCESS in 2h 03m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434230951ad04689a45d51ca1a824d12 : SUCCESS in 1h 15m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29db0c00bbe843ee9c34f4397aa9d763 : SUCCESS in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fbfa529904b54ad5a8b272537e353c91 : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07fca5978b104aa88dee09cf4a8ac7de : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/80e442fba8174017a95ba0b9aa6b1a98 : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f07291bee114184bc4d57e9eb53e4e1 : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/35d1fe4e21964bc29fd94200b257d152 : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/090e07f632d54eefba0838a55fd0d465 : SUCCESS in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b787147b590c479198a42b0aed45258e : FAILURE in 1h 29m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5321d675360645f3b6829ef6e4bac6a7 : FAILURE in 2h 34m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35c9983dfec648a5bcc783b8b3272a1a : TIMED_OUT in 3h 03m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1342f931bec847709f3aa7481f05d05b : FAILURE in 2h 25m 19s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/498dcef911b840329536d123398c7ae4 : SUCCESS in 1h 19m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c28b42209d4d48d2842e8e187c7728f5 : FAILURE in 2h 18m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52ee16e88ad74a6fade42639d065fa33 : SUCCESS in 53m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c98b751cf8cf46f69b0e9d74fb3a4d78 : SUCCESS in 1h 45m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8468cba2358f4a47859b11597b3a51e6 : POST_FAILURE in 2h 05m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38b578030be647459109bdab081225ad : SUCCESS in 2h 57m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73efba866eb14620a4192ac87508b26b : SUCCESS in 1h 29m 29s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/8c22d8d73bf04f4eaaabfbcd3b704157 : SUCCESS in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c00c4c9ac424e877177cf111425f1bbcc5dfc06","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-22 22:56:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/28/886228/2/check/iscsi-driver-alletra-6k/de68a83/de68a83.html : FAILURE in 1h 21m 12s","accounts_in_message":[],"_revision_number":2},{"id":"48632ccb6d0164d3c68d48b92a7c7c9225f348dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-22 23:01:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886228 : FAILURE in 1h 26m 42s","accounts_in_message":[],"_revision_number":2},{"id":"52e72cad61c55de90abac593c7725eaf6c4d4b9f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-22 23:06:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/e880c5f28b3149c4b706f7eecb2bb423\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8870cf2279aa464285db65c33dfbe597 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":2},{"id":"89644e9dff57df46c41c6ccba40d0e612e265178","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-22 23:36:08.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-2/FC/ : FAILURE in 5h 8m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-2/iSCSI/ : FAILURE in 4h 55m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"b198e947ff7f74d261fda0b45531aa1e167368c4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 06:19:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/28/886228/2/check/iscsi-driver-nimble/ef00924/ef00924.html : FAILURE in 1h 22m 00s","accounts_in_message":[],"_revision_number":2},{"id":"cdf217225483b08a6f7d9e6aa9e9910f9901bccd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 11:25:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/28/886228/2/check/DellEMC_PowerFlex/ea5f271/DellEMC_PowerFlex/192 : FAILURE in 1h 34m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"32c7ba84dff0046262f32ced8c32072c4dad558e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-24 18:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/886228/2/fc : SUCCESS in 0h 38m 49s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/28/886228/2/iscsi : SUCCESS in 0h 46m 09s","accounts_in_message":[],"_revision_number":2},{"id":"37663302d1efdfe77a3c13a582629f37d0b613c3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-28 15:11:17.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-2/FC/ : FAILURE in 42h 49m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-2/iSCSI/ : SUCCESS in 40h 49m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"d945d30627ee8493ecb1b44dcf17d28fb14a3ba6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-30 01:08:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/886228/2/20230630022829/NEC-FC : SUCCESS in 3h 50m 25s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/886228/2/20230630022829/NEC-ISCSI : SUCCESS in 7h 39m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f675de2edbb7b8831a054c030d3863f1b2c35637","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-01 20:12:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/28/886228/2/20230702022046/NEC-V-ISCSI : SUCCESS in 2h 34m 14s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/28/886228/2/20230702022046/NEC-V-FC : SUCCESS in 2h 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6e098fc969ce36e36cd43b529d7b163201133379","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-03 23:29:24.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-2/FC/ : SUCCESS in 2h 21m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-28-886228-2/iSCSI/ : SUCCESS in 3h 41m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"fc485e638c5ebcc04a0bbbd53430655224b35a88","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-06-25 15:39:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19836bd338b3258a32318abfdb03225c42041c5c","revisions":{"e23cba65efe3af2760c3d8c124c1b5218281fda4":{"kind":"REWORK","_number":1,"created":"2023-06-15 21:39:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/28/886228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/886228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/886228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/886228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/886228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/886228/1"}}},"commit":{"parents":[{"commit":"733e568245e921899a921afbea7c9b8319a425f9","subject":"Remove all occurrences of six.b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/733e568245e921899a921afbea7c9b8319a425f9"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 19:52:59.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 21:36:42.000000000","tz":120},"subject":"Remove all occurrences of six.reraise","message":"Remove all occurrences of six.reraise\n\nRemove six from requirements.txt since it is no longer used anywhere in\nthe code.\n\nChange-Id: Id7741dc14da8610f0cea4ab53e27354e4ac8eedf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23cba65efe3af2760c3d8c124c1b5218281fda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23cba65efe3af2760c3d8c124c1b5218281fda4"}]},"branch":"refs/heads/master"},"19836bd338b3258a32318abfdb03225c42041c5c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-22 15:17:53.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/28/886228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/28/886228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/886228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/886228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/28/886228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/28/886228/2"}}},"commit":{"parents":[{"commit":"19c36d5cb41a1016725f9aa7e48697fb1501fd4a","subject":"Remove all occurrences of six.b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19c36d5cb41a1016725f9aa7e48697fb1501fd4a"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-15 19:52:59.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 15:15:28.000000000","tz":120},"subject":"Remove all occurrences of six.reraise","message":"Remove all occurrences of six.reraise\n\nRemove six from requirements.txt since it is no longer used anywhere in\nthe code.\n\nChange-Id: Id7741dc14da8610f0cea4ab53e27354e4ac8eedf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19836bd338b3258a32318abfdb03225c42041c5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19836bd338b3258a32318abfdb03225c42041c5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
