)]}'
{"id":"openstack%2Fcinder~886333","triplet_id":"openstack%2Fcinder~master~Ide8f124c3a7dddc9fb45f11673684bd705131a38","project":"openstack/cinder","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ide8f124c3a7dddc9fb45f11673684bd705131a38","subject":"Imported Translations from Zanata","status":"MERGED","created":"2023-06-18 03:57:55.000000000","updated":"2023-06-29 13:23:23.000000000","submitted":"2023-06-29 03:35:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886333-zanata/translations","meta_rev_id":"b9b2cec5b66c36ac66e682a2412d6b76d3369864","_number":886333,"virtual_id_number":886333,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 03:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-06-29 02:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-06-29 02:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-18 04:09:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-18 04:17:16.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-18 04:41:49.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-18 05:13:59.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-18 05:18:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-18 05:30:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-18 05:44:57.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-18 05:55:15.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-18 06:00:15.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-18 07:02:19.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-18 07:54:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-18 08:15:56.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-18 08:40:48.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-20 11:31:03.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 04:12:19.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-26 02:04:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-26 14:01:49.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-06-28 06:37:08.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-06-29 02:06:26.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"ae5d2c36c8fd07c2106a7bd0a90a839f7cfaf0b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-18 03:57:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db57e16e60b99da944d5294596e910929f00f45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-18 04:09:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953bdf38888743e1911142a6752f0db2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41c672363394cbebd10d9351d3e8c8b : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7107645532cb4088bf5e92f3a546bebd : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0af67f8154784d57a3c493315c414e97 : SUCCESS in 7m 37s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe3bce1eb4a84b958ce8d75ca7a500f7 : SUCCESS in 5m 21s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/3a7817828c3b4222b8f3167fc3a8a55e : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"154b73d03bca7d10661238bd9b4448d271ba2a1f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-18 04:17:16.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/18_06_2023_04_09_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"eb330c652710f3225bb8787db46145dc6757e07b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-18 04:41:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerMAX_FC/7f001d5/DellEMC_PowerMAX_FC/138 : FAILURE in 32m 04s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerMAX_iSCSI/38f3928/DellEMC_PowerMAX_iSCSI/132 : FAILURE in 26m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"506a41aefb6e7a7e6234ece594f141f1c849c13a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-18 05:13:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/33/886333/1/check/iscsi-driver-alletra-6k/3d6e78f/3d6e78f.html : FAILURE in 1h 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c03a8bc80a04e738eaff00f69418202e9a519b35","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-18 05:18:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/2da2d5e1555a4d0da3b2296d2081cdeb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/50a5d6c14b994d7182dfa3cfa79674c4 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43f99af3b87c2852d4448aba5afc78fba71eadca","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-18 05:30:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerStore_iSCSI/7e47927/DellEMC_PowerStore_iSCSI/114 : SUCCESS in 57m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerStore_FC/24af2e4/DellEMC_PowerStore_FC/119 : SUCCESS in 1h 20m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerStore_NFS/f01340d/DellEMC_PowerStore_NFS/113 : SUCCESS in 1h 01m 57s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerStore_NVMeTCP/05b8cef/DellEMC_PowerStore_NVMeTCP/114 : SUCCESS in 1h 08m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ebf1f05017879890157544dcf14b5a0e6aec0a0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-18 05:44:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/1/upstream-check/cinder-tempest-plugin-ontap-fcp/1fe0916/ : RETRY_LIMIT in 7m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/da97811/ : SUCCESS in 1h 33m 41s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/1/upstream-check/cinder-tempest-plugin-ontap-nfs/68b9c35/ : SUCCESS in 1h 32m 54s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/cd7b3f1/ : FAILURE in 1h 28m 30s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7570fe66500dd0b3ed30c6be45f9944172cfeabe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-18 05:55:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_PowerFlex/34208a0/DellEMC_PowerFlex/155 : FAILURE in 1h 45m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1379ada6f1ff444d4fc2138a70684b2eb743700f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-18 06:00:15.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_Unity_FC/27ceb42/DellEMC_Unity_FC/299 : SUCCESS in 1h 42m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/1/check/DellEMC_Unity_iSCSI/17f6222/DellEMC_Unity_iSCSI/301 : SUCCESS in 2h 02m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a6e1b0fae5ae769ed0e8f7303aed5bcc82ce1656","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-18 07:02:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-1/FC/ : SUCCESS in 3h 4m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-1/iSCSI/ : SUCCESS in 2h 46m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e741236a16db84285ff027c3b509a9326a231359","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-18 07:54:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-33-886333-1/ns5_nfs : SUCCESS in 3h 49m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-33-886333-1/ns5_iscsi : SUCCESS in 3h 55m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"74340ea0d93ca3747dfd9663abcf55a7e487711d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-18 08:15:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886333-1/HPMSA-iSCSI/ : SUCCESS in 1:48:55\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886333-1/HPMSA-FC/ : SUCCESS in 1:28:02","accounts_in_message":[],"_revision_number":1},{"id":"3c2a942d2e3855ad58ec8685288892a59d9164a8","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-18 08:40:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886333-1/PVME-iSCSI/ : SUCCESS in 2:03:29\n- PVME-FC https://os-logs.seagate.com/dell-ci/886333-1/PVME-FC/ : SUCCESS in 1:38:10","accounts_in_message":[],"_revision_number":1},{"id":"fa1656dd78d6866cd64e77b931de0b6c39d05eb9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-20 11:31:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886333 : FAILURE in 1h 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3480d3090b2ee232457ab5fbbbf758674e007773","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:12:19.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/33/886333/1/fc : SUCCESS in 0h 35m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/33/886333/1/iscsi : SUCCESS in 0h 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e8fa6859fca0814d3e3280ab72a86f0c05dd230f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-21 04:35:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e7a5bc4737cb256795d0e30f49814e09b2ab0d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 05:02:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53691144f30a42b598fb17275657d182\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c3795d77ff4f8cbf06de55e9906eda : SUCCESS in 23m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d304de34a548ca8b801dbcf957a0de : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/938994adfefa4ae59ef3366d7fd62aba : SUCCESS in 8m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a71b5bc71f7749d19b45dec7ab1dbe10 : POST_FAILURE in 6m 03s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/0bcde7e4d3384fceb18cda145a9b45bb : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc7a928fbbd249b48cb0d6140d916282bfc80314","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-21 05:46:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/33/886333/2/check/iscsi-driver-nimble/74617ff/74617ff.html : FAILURE in 1h 09m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f27dfea4a7d97f49fb3fe74fd9fe0d000afeb495","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-21 05:48:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/33/886333/2/check/iscsi-driver-alletra-6k/30f5f6e/30f5f6e.html : FAILURE in 1h 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"14418950e754b82774c594b4ce10530958723707","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-21 05:51:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/d2df1c2686c942ea8163e7202a88b07f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a78415fd2b90482b83fec021b670afbd : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d1f50ed217816c47468bf93faeb42c59a4c3c878","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-21 05:55:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886333 : FAILURE in 1h 16m 58s","accounts_in_message":[],"_revision_number":2},{"id":"232234d8d93d7666d645627bd9a3616d8925cb33","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-21 06:36:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/33/886333/2/check/DellEMC_PowerFlex/c7d3fa4/DellEMC_PowerFlex/173 : SUCCESS in 1h 39m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b69e41dd0df8f5b5445b59e8360e5a6b0e273636","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-21 07:05:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/33/886333/2/check/DellEMC_Unity_FC/4c17d82/DellEMC_Unity_FC/318 : SUCCESS in 1h 57m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/2/check/DellEMC_Unity_iSCSI/933396e/DellEMC_Unity_iSCSI/319 : FAILURE in 1h 58m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"047005ebc141ecd11d2e713dd910c8f4d9bf9588","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-21 07:20:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-2/FC/ : SUCCESS in 2h 18m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-2/iSCSI/ : SUCCESS in 2h 44m 13s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"fe142ee958ce4aba9a443be396d0648a2898d4bf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-21 08:33:16.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-33-886333-2/ns5_nfs : SUCCESS in 3h 53m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-33-886333-2/ns5_iscsi : SUCCESS in 3h 57m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5845bf64970e27f1a988ceae8867b17dfde04531","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-26 02:04:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/33/886333/2/20230626073117/NEC-ISCSI : SUCCESS in 3h 30m 35s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/33/886333/2/20230626073117/NEC-FC : SUCCESS in 3h 33m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d787725c073fd49efe1344170510ec4a56b7dff8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-26 14:01:49.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/33/886333/2/20230626170024/NEC-V-ISCSI : FAILURE in 5h 53m 05s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/33/886333/2/20230626170024/NEC-V-FC : SUCCESS in 6h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"70afaeaacfc4bdb2847d549cb05f7b97825ea557","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-26 20:33:19.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-2/FC/ : UNSTABLE in 2h 22m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-2/iSCSI/ : FAILURE in 9h 21m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"76ce88fb491d50d9f5449496768215e35e500b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-06-28 03:43:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"47d8307c7fa3880c4f7978164ce3d377e5dbe21a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-28 04:09:33.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-33-886333-3/ns5_nfs : FAILURE in 25m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-33-886333-3/ns5_iscsi : FAILURE in 25m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"2b118c864fb268df3301be4a9e83c12662e9990d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 04:37:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0ba003f6dd4444b6f9d8235502aa97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20c82d380534de2b2bc09d2df46a329 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dad2a80b62c412a988a70edf2d68d6b : SUCCESS in 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27a569ce59ae4b418af9a1300597c399 : SUCCESS in 7m 46s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6016e7ed42f43499b0663ede5001e8b : SUCCESS in 5m 51s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/95b00afebc7645b092e715b6710f8370 : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"989818c526a40160c44e8d9bdc0f2e9ec8e4921f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-28 04:45:42.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/28_06_2023_04_38_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"34a9ac8dd83364326a0a960456cede1f73136671","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-28 04:55:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/33/886333/3/check/iscsi-driver-nimble/bb1f7f4/bb1f7f4.html : FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4e151edeb7192a7dd632daa97bc751d99603cf3f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-28 04:57:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/5cf6400758084a2ba72bfab4b3abdab6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/155fc6c6387e496fb237d702ae8b14e7 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"56c9b447ed68ccf5ac1df4f933c66804769289c5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-28 05:25:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/2cf463b/ : RETRY_LIMIT in 7m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/3/upstream-check/cinder-tempest-plugin-ontap-nfs/44a6922/ : RETRY_LIMIT in 7m 13s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/886333/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/abfd470/ : SUCCESS in 46m 32s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c2053c05e035e658b43512121cb8d33224f7ec48","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-28 05:29:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerFlex/8aeecf9/DellEMC_PowerFlex/220 : FAILURE in 1h 45m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"47ddc0d7366cd145f5d15c0d41ecd5ea6cdbebfc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-28 05:39:36.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_Unity_FC/15512bb/DellEMC_Unity_FC/341 : FAILURE in 1h 55m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_Unity_iSCSI/9d6e4a9/DellEMC_Unity_iSCSI/340 : FAILURE in 1h 55m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5234e4b3b5265fd453c51237014ef73baf884fe6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-28 06:07:39.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerStore_iSCSI/9916e22/DellEMC_PowerStore_iSCSI/129 : FAILURE in 56m 19s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerStore_FC/036e3c5/DellEMC_PowerStore_FC/136 : FAILURE in 1h 29m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerStore_NFS/2bb063f/DellEMC_PowerStore_NFS/128 : FAILURE in 59m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerStore_NVMeTCP/a829f87/DellEMC_PowerStore_NVMeTCP/129 : FAILURE in 1h 09m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"75d2986b9fd6cd2038a617915d011151c32f0521","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-06-28 06:37:08.000000000","message":"Patch Set 3:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1515/","accounts_in_message":[],"_revision_number":3},{"id":"c4b6bf3c1e5b275427fafa2fcc263db5b8e8ae40","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-28 08:51:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886333-3/HPMSA-iSCSI/ : SUCCESS in 1:53:02\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886333-3/HPMSA-FC/ : SUCCESS in 1:29:21","accounts_in_message":[],"_revision_number":3},{"id":"b69646964c40bb20724d743a1ad0bea0630361ad","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-28 09:00:12.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/33/886333/3/fc : SUCCESS in 0h 48m 23s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/33/886333/3/iscsi : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9a377ac3f60c0331ae614591eec1e66c8b18bef6","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-28 09:58:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886333-3/PVME-iSCSI/ : FAILURE in 2:13:56\n- PVME-FC https://os-logs.seagate.com/dell-ci/886333-3/PVME-FC/ : FAILURE in 1:50:49","accounts_in_message":[],"_revision_number":3},{"id":"0c81ba129bc891e3095fff61e87c7ac30eaf0aab","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-28 20:26:06.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerMAX_FC/7a04fad/DellEMC_PowerMAX_FC/160 : FAILURE in 6h 26m 39s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/33/886333/3/check/DellEMC_PowerMAX_iSCSI/366190a/DellEMC_PowerMAX_iSCSI/154 : FAILURE in 7h 18m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"94df0a75f451bc1e88509e5567432a6d9bbf89f6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-06-29 02:06:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"447bce572f9aa1f9d4184d9e84b66ec5556fa319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 02:12:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"11d42d274c97da9a6cf6323ad977977178c2420d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 03:35:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/684ca371f07b478bbc4cf78d758c307b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b91a923e0e4c2895994e3a41b9f2da : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da77e6d66c7e485a90ecb93d43810726 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8583d061246d4c7bb0e8db7b0c7b7d4a : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"da75104ee46e090c67157cbf84e143de727275da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 03:35:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b602f9087dd7074b96a3d18de1855dbb11d6dfc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 03:37:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b01c015cdba4e3e90e46b1f0f6cdcbc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25239ebfe17a4bfcb2101cfce46f8ad6 : SUCCESS in 2m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b09a57773aa4be191d3128312dbf4aa : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b9b2cec5b66c36ac66e682a2412d6b76d3369864","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-29 13:23:23.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-3/FC/ : SUCCESS in 2h 38m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-33-886333-3/iSCSI/ : SUCCESS in 2h 17m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f1569abd66bdccf5536e6cb258b5b7bbc2f30b59","revisions":{"6afc3ba55aae47d4e2638704d9c2b8a0fa8029a2":{"kind":"REWORK","_number":1,"created":"2023-06-18 03:57:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/886333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/886333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/886333/1"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-18 03:57:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-18 03:57:53.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ide8f124c3a7dddc9fb45f11673684bd705131a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afc3ba55aae47d4e2638704d9c2b8a0fa8029a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6afc3ba55aae47d4e2638704d9c2b8a0fa8029a2"}]},"branch":"refs/heads/master"},"59177c9d13c810b7c00f1abb02be66c12c1f627b":{"kind":"REWORK","_number":2,"created":"2023-06-21 04:35:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/886333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/886333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/886333/2"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 04:35:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-21 04:35:37.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ide8f124c3a7dddc9fb45f11673684bd705131a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59177c9d13c810b7c00f1abb02be66c12c1f627b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59177c9d13c810b7c00f1abb02be66c12c1f627b"}]},"branch":"refs/heads/master"},"f1569abd66bdccf5536e6cb258b5b7bbc2f30b59":{"kind":"REWORK","_number":3,"created":"2023-06-28 03:43:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/886333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/33/886333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/33/886333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/33/886333/3"}}},"commit":{"parents":[{"commit":"e673bcc368d3a24ec21713adbd83b4ab6cbcae18","subject":"Merge \"Remove six from solidfire driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e673bcc368d3a24ec21713adbd83b4ab6cbcae18"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-28 03:43:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-06-28 03:43:26.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ide8f124c3a7dddc9fb45f11673684bd705131a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1569abd66bdccf5536e6cb258b5b7bbc2f30b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1569abd66bdccf5536e6cb258b5b7bbc2f30b59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
