)]}'
{"id":"openstack%2Fcinder~886609","triplet_id":"openstack%2Fcinder~master~I5446d3d4fa7c76a90ca64ef6b61f90cdece1a88e","project":"openstack/cinder","branch":"master","attention_set":{"35836":{"account":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"last_update":"2023-07-27 05:48:03.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-06-21 13:04:38.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5446d3d4fa7c76a90ca64ef6b61f90cdece1a88e","subject":"Add api-ref functional tests for backup actions","status":"NEW","created":"2023-06-21 11:50:54.000000000","updated":"2023-07-27 05:48:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"21f76b7ac638f5a91da15f65483a69921d86d4f8","_number":886609,"virtual_id_number":886609,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-13 16:23:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":-1,"date":"2023-07-27 05:48:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 11:51:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 12:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 13:04:38.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-21 13:06:59.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-21 13:11:04.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 13:47:18.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 14:02:13.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"state":"CC"},{"updated":"2023-06-21 14:02:13.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"state":"CC"},{"updated":"2023-06-21 14:02:13.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2023-06-21 14:42:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 15:48:50.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 15:55:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-21 17:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 17:17:50.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"state":"REVIEWER"},{"updated":"2023-07-01 07:44:01.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-01 10:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-02 16:31:16.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-13 23:33:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-14 07:09:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-27 05:48:03.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"ce1dc8a5a9dde22f7316ba6d80f790fed9dece0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-21 11:50:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"359b90ff8db6e302ec7d43e128cbfd25bee8e1b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-21 11:51:44.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18cb3357d7c6d37c593e346e32a946013d6abd96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 12:08:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4464220621c841ed963b799b5475ae09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4804d898fc3e436d89721468e2dc1570 : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07746d2697df4287b83a9f332f343479 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b667c2d28de841baa1569a269405f954 : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3075dd804f33f80513f63eaf3c1889f8fcfe9f14","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-21 13:04:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/886609/1/check/iscsi-driver-alletra-6k/af6836f/af6836f.html : FAILURE in 1h 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8bac74d82d623acc9ae46ce56910eea17d05609b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-21 13:06:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/81efcc84b5264a3c931db936f084e1f1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e71ee1b2abfa446789c88c19fc675435 : SUCCESS in 1h 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5f36f996bc59c7a72585ea30f7356cb0a67587de","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-21 13:11:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886609 : FAILURE in 1h 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b8cd051e27adc2ced96e0384597723fdbf337dae","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-21 13:47:18.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/09/886609/1/check/fujitsu-eternus-dx-iscsi/611849e : SUCCESS in 1h 54m 45s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/09/886609/1/check/fujitsu-eternus-dx-fc/0f304c5 : SUCCESS in 1h 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"00e8f927738f8693b5000308fac345c6b31e75b1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-21 13:58:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/09/886609/1/check/iscsi-driver-nimble/aa286cc/aa286cc.html : FAILURE in 1h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8a49243781107f4ba8599c76c6135227b79ab452","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-21 14:42:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/09/886609/1/check/DellEMC_PowerFlex/8814bfd/DellEMC_PowerFlex/177 : SUCCESS in 1h 35m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b5d9f170c3834e672b499cb11142c444d8c44635","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-21 15:48:50.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-886609-1/ns5_nfs : SUCCESS in 3h 57m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-886609-1/ns5_iscsi : SUCCESS in 3h 53m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"373bbddd78a4163a5fd7ee9d5ec35da37ec98721","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-21 15:55:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-09-886609-1/FC/ : SUCCESS in 4h 4m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-09-886609-1/iSCSI/ : SUCCESS in 1h 53m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe2634178a3be12b250398bb57176ae04c485cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:08:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b490b180c4ea468b94189ebf85f3f470\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/75ad9e2b33504c08ba21fba46e7e3847 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9bf2edd57ed4a84a15017b81ceddefb : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74928185cef5450c8cb12f8bdcc7c187 : SUCCESS in 11m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0760bd4608fc47f48a0412b82d365334 : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72ae67ec9942429d9ec47895183e905c : TIMED_OUT in 41m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab35597a2ba045b99487dc5797b76ad4 : SUCCESS in 15m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/80b6a45c5dd041c191f68c3c2d86c828 : SUCCESS in 15m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ceef753305e2465d8882612daa40dfa4 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b11a130b71cf4286aed4fabe786ef5da : FAILURE in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/47ed444b79e24fd08ee01062249bdeae : FAILURE in 6m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/df16555c369d46ee907a4d5971aa94a0 : SUCCESS in 1h 29m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8eb06676592b45958582612a83c1647c : SUCCESS in 6m 29s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/cbacc382a8f244099f37a7629902dff9 : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f3790c145377ceba3f32e70a8cfd7df9f34ae8b","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-21 17:17:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"75e0fc77c322e58e82bd5137efc043f0c3b9b509","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-21 17:18:16.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"d50937ef99ed31046e788cb8b385b72b312b8c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-23 19:40:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc6d8bd18a1c0237b127a779d80f7036220190df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 19:41:28.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45d7cbb193628c97da3a9d5864d62b5e4f094bde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:51:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb3a25f466f483e947bcbe159dfd6bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee9b21ea5cd24297b9c999ef3ed5d689 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb95681984354904a859028542bea974 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e131e94f876f427fada27d3be40333ed : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec63772342bb3dd357004e7d9d94a332edc715ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 20:47:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7f0004526954ee58c2ac9e725e41d57\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/89112fefe96d4425828d0edaf47b913d : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255e1534fdba42149ce4cac933ee58ee : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec4a46883b14daf864d35297b80405b : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dce2da207de3429a890dc489d00bd307 : SUCCESS in 11m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc61ef4b809f4ecca7eeec2fe1a892d2 : TIMED_OUT in 41m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af34805bc53485bbe9761346f6d05f9 : SUCCESS in 11m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29b01c648540463ca015beeecc472f68 : SUCCESS in 16m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4bd0a58171b34a2a8e4af9ebe6c2cddd : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c84726bfc7eb4f4e90c50afa123c0e6e : FAILURE in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b0dd4f7605342bfaf0201aa87212c30 : FAILURE in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b36397811af04f50a74b020959c80412 : SUCCESS in 1h 01m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e865a24222e24f47921c46243d837642 : SUCCESS in 6m 03s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/975c9473656f478fad3133818388db0a : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c93fbab49f186c2fea6d69dcf92c0cc0effdfd7d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 20:58:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/5a7362ef6a6a4d3993a66024943c5880\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/31d8cb5a5f8245ea9f14a62e8783bb90 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2849e92528ce9f4d75104e6c290ae74ee9c93599","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 21:00:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/886609/2/check/iscsi-driver-alletra-6k/663398e/663398e.html : FAILURE in 1h 14m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6d9ae6929f12af3fb2073970b33a20003a4dd519","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 21:37:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886609 : FAILURE in 1h 53m 29s","accounts_in_message":[],"_revision_number":2},{"id":"bffc9d27cd72fa463411cb629e1638c5c22e7fe2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 22:38:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/09/886609/2/check/iscsi-driver-nimble/4c8a232/4c8a232.html : FAILURE in 1h 31m 43s","accounts_in_message":[],"_revision_number":2},{"id":"211ea3501334ab1f206ec9384035ae5625a657e6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 23:57:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-886609-2/ns5_nfs : SUCCESS in 4h 7m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-886609-2/ns5_iscsi : SUCCESS in 4h 5m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"7b76b9f779d492205a78b08f18c447730faaf171","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-24 00:12:17.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-09-886609-2/FC/ : FAILURE in 4h 30m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-09-886609-2/iSCSI/ : SUCCESS in 2h 30m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"1f3f39bf8f506f735641a731921267366a9af483","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-24 03:12:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/09/886609/2/check/DellEMC_PowerFlex/cb6cd55/DellEMC_PowerFlex/201 : FAILURE in 1h 35m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6e9fe31b39a5ab1ee589be078f9dbf9a61141689","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-01 07:44:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/2/20230701130513/NEC-FC : SUCCESS in 3h 30m 17s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/2/20230701130513/NEC-ISCSI : SUCCESS in 3h 38m 45s","accounts_in_message":[],"_revision_number":2},{"id":"29274e154be49434715ad2d46cd6ca57b6551a33","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-01 10:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/09/886609/2/fc : SUCCESS in 0h 36m 44s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/09/886609/2/iscsi : SUCCESS in 0h 43m 24s","accounts_in_message":[],"_revision_number":2},{"id":"449aa0ab1c67bd34680529b4b86acf822e7e6fe0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-02 16:31:16.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/2/20230702195515/NEC-V-FC : SUCCESS in 2h 48m 00s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/2/20230702195515/NEC-V-ISCSI : FAILURE in 5h 35m 58s","accounts_in_message":[],"_revision_number":2},{"id":"87a4ad4bef2a0d6406e32efbf035ef46665ceb5e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-04 23:14:10.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-09-886609-2/FC/ : FAILURE in 7h 7m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-09-886609-2/iSCSI/ : SUCCESS in 7h 48m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"3521a17cf6cbdafb7b4975a9ab0fa9a5ea1617e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-13 15:16:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8126f9ce4a99360d61eb88facf51943d03f67db","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 15:21:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e09d1cbb4028407fa6e0f749fd5357e7 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/26469812cf55410d8ba7c5ed711158e6 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"a060fc8bd75e9105184870f9c2d25a1cb30b5c13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 15:40:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd0de73766894d919c19674b46a04330\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df23e57e73d845d8b8c299ab4b99f5b5 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3d226c8fc51451cbf0e636cf6a9c2ef : SUCCESS in 22m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/911ff5edc1c04889bfb872f8a8678f65 : SUCCESS in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbc693711d6b45fefc60ba6db5031b2f26ecb9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 16:23:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f74721c649ad47e38d0b53c41ae42421\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f933ddab21df47e68b8e36192a126c97 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff9d23ecc84b46d8bc1312bc63dc324c : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd730b1a5d2494c9341fe5fb299bd31 : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc4a48e41c754eb480c283d2adfb4ec5 : SUCCESS in 10m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57c0733d67c54bb6b54b56a84d7c25e0 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1b855bb00d4bdf8538d9eebb68550a : SUCCESS in 12m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b32bb471fc3f405a9ca7978f60b05ac3 : SUCCESS in 17m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/146525290de44aee955663672490dff9 : SUCCESS in 5m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb310b93069849f395fda66a8f192a7c : FAILURE in 7m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1e391b529284c24bef9c98f1979b56f : FAILURE in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f2038255345545ec9859a1ae180c1ea5 : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04a231c5c2104da7bc5fad4608ff9d79 : SUCCESS in 5m 48s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/b9711e6a7bbb4156b41c108db893278f : SUCCESS in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a771fd755d309a8b6660d2c20647e76235bcfe5a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-13 17:53:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886609 : FAILURE in 1h 39m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7e9b7304c1103e02e99d41ba0a7a085b19f7e2d6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-13 18:02:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/b918879747d9434baa6e8de95fcf9e46\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/04a21036578a47c1b063a3eea03fbea2 : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":3},{"id":"27778ef6d21aaaa7620a4fe3dd93442cb565821b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-13 19:26:11.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-886609-3/ns5_nfs : SUCCESS in 3h 55m 41s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-886609-3/ns5_iscsi : SUCCESS in 4h 2m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"64d54776d648a15ab1047047d51b14a07887a4f9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-13 20:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/09/886609/3/fc : SUCCESS in 0h 36m 34s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/09/886609/3/iscsi : SUCCESS in 0h 49m 48s","accounts_in_message":[],"_revision_number":3},{"id":"00d351fb25fe30c1966215a2731c2114deddb4aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-13 23:33:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/886609/3/check/DellEMC_Unity_FC/8568e2d/DellEMC_Unity_FC/425 : SUCCESS in 1h 45m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/886609/3/check/DellEMC_Unity_iSCSI/4e4c762/DellEMC_Unity_iSCSI/427 : FAILURE in 2h 21m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"46047ed6edd9d3e7ad6a6b1b217b9342b3631bec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-14 07:09:34.000000000","message":"Patch Set 3:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1584/","accounts_in_message":[],"_revision_number":3},{"id":"525429fb6b04429c516a85470ccb7977448720c3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-15 02:55:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/3/20230715062858/NEC-V-ISCSI : SUCCESS in 5h 25m 45s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/3/20230715062858/NEC-V-FC : SUCCESS in 5h 26m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6a678fd9c33b35beee416c30e279ffc7122f62bb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-15 08:25:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/3/20230715131354/NEC-ISCSI : SUCCESS in 3h 49m 13s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/886609/3/20230715131354/NEC-FC : SUCCESS in 4h 11m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4d6ada06c6aed1418aa511ed770f19cc2bdddb66","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-24 11:11:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"21f76b7ac638f5a91da15f65483a69921d86d4f8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-27 05:48:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1c0a4fa26c08c77f1b6a6a6603413887e7c1d910","revisions":{"a7d25e8d2ea2fa29dc702c258afe74fd043208c6":{"kind":"REWORK","_number":1,"created":"2023-06-21 11:50:54.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/09/886609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/886609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/886609/1"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-21 11:49:55.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-21 11:49:55.000000000","tz":0},"subject":"Add api-ref functional tests for backup actions","message":"Add api-ref functional tests for backup actions\n\nThis patch adds the functional tests to generate api-ref request\nresponse for attachments API.\n\nChange-Id: I5446d3d4fa7c76a90ca64ef6b61f90cdece1a88e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7d25e8d2ea2fa29dc702c258afe74fd043208c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7d25e8d2ea2fa29dc702c258afe74fd043208c6"}]},"branch":"refs/heads/master"},"0e943fed46c18f285b774afcecdc54e26a5fa047":{"kind":"REWORK","_number":2,"created":"2023-06-23 19:40:40.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/09/886609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/886609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/886609/2"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-21 11:49:55.000000000","tz":0},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 19:40:12.000000000","tz":60},"subject":"Add api-ref functional tests for backup actions","message":"Add api-ref functional tests for backup actions\n\nThis patch adds the functional tests to generate api-ref request\nresponse for attachments API.\n\nChange-Id: I5446d3d4fa7c76a90ca64ef6b61f90cdece1a88e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e943fed46c18f285b774afcecdc54e26a5fa047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0e943fed46c18f285b774afcecdc54e26a5fa047"}]},"branch":"refs/heads/master"},"1c0a4fa26c08c77f1b6a6a6603413887e7c1d910":{"kind":"REWORK","_number":3,"created":"2023-07-13 15:16:05.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/09/886609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/886609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/886609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/886609/3"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-21 11:49:55.000000000","tz":0},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-07-13 15:15:24.000000000","tz":60},"subject":"Add api-ref functional tests for backup actions","message":"Add api-ref functional tests for backup actions\n\nThis patch adds the functional tests to generate api-ref request\nresponse for attachments API.\n\nChange-Id: I5446d3d4fa7c76a90ca64ef6b61f90cdece1a88e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c0a4fa26c08c77f1b6a6a6603413887e7c1d910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1c0a4fa26c08c77f1b6a6a6603413887e7c1d910"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
