)]}'
{"id":"openstack%2Fcinder~886818","triplet_id":"openstack%2Fcinder~master~I9bbfab93fb138932243d880e440c9524b5f08217","project":"openstack/cinder","branch":"master","topic":"test_list_detail_backup","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-07-27 08:52:53.000000000","reason":"Change was submitted"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-07-27 08:52:53.000000000","reason":"Change was submitted"},"35851":{"account":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"last_update":"2023-07-27 08:52:53.000000000","reason":"Change was submitted"},"35836":{"account":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"last_update":"2023-07-27 08:52:53.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-07-27 08:28:53.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-07-27 05:54:31.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I9bbfab93fb138932243d880e440c9524b5f08217","subject":"Automate generation of backups api-ref samples","status":"MERGED","created":"2023-06-23 12:34:34.000000000","updated":"2023-07-27 08:54:14.000000000","submitted":"2023-07-27 08:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"886818-test_list_detail_backup","meta_rev_id":"aab4ae6bcce6e1f35c3c3e772583623a395c315e","_number":886818,"virtual_id_number":886818,"owner":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-27 08:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-27 08:28:53.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-07-27 05:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-27 08:28:53.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 12:35:21.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 12:37:35.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-06-23 12:37:35.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"state":"REVIEWER"},{"updated":"2023-06-23 12:37:35.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-06-23 12:37:35.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-06-23 12:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 13:12:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-23 13:41:32.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-23 13:54:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 13:59:25.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-23 16:34:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-06-23 16:37:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-23 17:02:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 17:07:17.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-23 17:51:11.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-06-23 22:07:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 22:47:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-06-24 00:52:26.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-06-24 11:19:02.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-25 17:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-01 04:05:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-02 10:55:11.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-05 04:17:25.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-07-10 20:17:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-10 20:19:54.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-10 20:32:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"d96667c1b7fdc258427a9e5b837b8bb7429c0672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-23 12:34:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c5b6846ef8b0c6bccb4f15fdeeaada39fde374f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 12:35:21.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d8024571188db49039ee6daff06cbb0ab56fb0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 12:57:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2705ae657794467937dcbf3ee9ae19c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f6dd1560f9c437ea3cf044fa90d9212 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee3c97f0404c452da9a32ac4afa6b837 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04c7b906790f46849a3ddef95b89cd72 : SUCCESS in 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b127c93a5d1a0d2a88d7d208a5b61d561256cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:12:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b08674609b3416e95a8d96401524e6a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/38db204a7411430b8d1f010380ac614b : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30de45967d714ad0b62f0ef096680c52 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7242cace7841468ca760fe003d8dc9d1 : SUCCESS in 11m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79d8074edb31431287a53bee204c602d : SUCCESS in 9m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ece89534b557465c815aaeab8da999f7 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc590b50d42a49bfb42062d0832abe8e : SUCCESS in 12m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/340f6893cff74f0b89fc7bec7e9252ce : SUCCESS in 13m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4d16ad13c668413cad98fd6b07ded7c1 : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8afa40a63cf4981bc0a60a68e2e462b : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b96c3af9c40e4d5988219bea16ac6375 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/61e466653f0a41aaa76bb6eeeb9da37c : SUCCESS in 37m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5e77243ce7aa44efb5af28c51709a525 : SUCCESS in 5m 28s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/f783887dd90c45698141b537617fa690 : SUCCESS in 10m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file api-ref/source/v3/ext-backups.inc","accounts_in_message":[],"_revision_number":1},{"id":"95a936a41001a99531a87f5870ed15fcbee5737a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 13:41:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/886818/1/check/iscsi-driver-alletra-6k/7cf24e2/7cf24e2.html : FAILURE in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"be911915499e9a42622f4a9049548d55e6717fc3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 13:54:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : FAILURE in 1h 16m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d50929660aca28c7977dbc654683266fe461c853","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 13:59:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a7035d6273549539a141a331552cf2c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9be6a3fd4e67491daa94d421061b62a5 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e7f481253ace69efdcc1dc5cab8e2f1cb0ddc0e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-23 14:13:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a41976f4d0ae6fdfd577d4789ca8ab649ab71fde","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 14:14:51.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebf94af7a8205f02bf8c34f1732a26716c5e7c76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 14:28:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013e7648fe964ad09dbc03e472ada9d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1640510a158c471387a9e367c7dd836d : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e767ca479b504be6b4343f99c57f0b8d : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd4ca46d70ff4e95bd1742d32d41dfac : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8a710a521223e992805ba7987c5cb2b46fcde1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 15:15:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0944fc7c28a450abb2fdf9711421f61\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8953f5d98c59477693b9c583b3ba39ff : FAILURE in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a7de358fb24551ae8f3fbb44b7a5e4 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/affc9b9070154816996ecf7dd73f8907 : SUCCESS in 13m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b65fc0259104471c83cc27f86489dd60 : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5a183f0f8f14c9ab44e6ff44ad48041 : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a837ff115cfc4621b8cc6da609a238a1 : SUCCESS in 12m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bcc1c77b123d43cc8e27f1a350abeb4d : SUCCESS in 17m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/440f1d1ccecf416689683a1309a87e46 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb0cc9dc71c64eabb925813726550130 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a777f754e6df49ae8a4b2d4e98804070 : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2f72773d6c3141d9b40372b04c718121 : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2981ad41658d408b9a596b00fe8bdb58 : SUCCESS in 26m 21s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/12e4053b617542b18f85429416afc8bb : SUCCESS in 10m 58s (non-voting)\n\nWarning:\n  Comments left for invalid file api-ref/source/v3/ext-backups.inc","accounts_in_message":[],"_revision_number":2},{"id":"bab421d8dadd94aa86c6f528d13d1a50d62eb8b0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 15:28:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/886818/2/check/iscsi-driver-nimble/1ee8e68/1ee8e68.html : FAILURE in 1h 10m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8ddb3c2276701efe2275b0a273ee04768ec18f15","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 15:31:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/886818/2/check/iscsi-driver-alletra-6k/6e66c33/6e66c33.html : FAILURE in 1h 15m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ec0203f48f81c419a4cadaf040ab4f04d25b82ba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 15:32:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : FAILURE in 1h 16m 06s","accounts_in_message":[],"_revision_number":2},{"id":"bb2131c585d6bf2dd8c4147537c18a1ebe4ce900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-06-23 15:38:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3a36c515aa785e1e91a888ec892308fd71266f40","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 15:39:33.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"657feb8da5ef70c2300e6f892359ee2deb7a4996","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 15:50:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9d5780da066448eb7aaf85a433426d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcc04f92964841c39d96fd8c3fdfcad6 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/488c4d8cd8ce49baa13c894c11316b18 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78c9fb70227f461681e8fc3232edb7de : SUCCESS in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"015f981e713057468cf67d6ff9a8384be86df1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 16:27:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1685d685234d928676d3280ac36317\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd7c56cd9adb471389c46df8f876ddaa : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6216c4477f248878fa522b173b1b3d5 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/641fc26026844d80be98f7a7a9d05686 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e0a2089ed514b4c985247da31dc256f : SUCCESS in 11m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e57357bb8d7649b39d0044e38cf7abb4 : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e404de6e464adea587f7d56a65ddd8 : SUCCESS in 14m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8d35d9deb9d4f38993acce59d96dc90 : SUCCESS in 15m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3f5786e65525453789e2deb5cfeaf576 : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/582adb5567d645ea947087cd825dfc17 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e6c9fd7ab54c431d83c859f80b2b5730 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36b18b6892d746049e528abf4d79fe03 : SUCCESS in 47m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57d52a2f01a748b680f05f526eafde9a : SUCCESS in 5m 54s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/048420ae69be4d78bdc0798d15ae894d : SUCCESS in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"382bb59ac619fe9869d1d0c9176337e253430126","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-23 16:34:44.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_06_2023_16_27_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ecf61dd77e6ef8d5bd35c057d0f381e548102f1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-23 16:37:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/e6881ee/ : RETRY_LIMIT in 7m 55s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/3/upstream-check/cinder-tempest-plugin-ontap-nfs/9d3a0a2/ : RETRY_LIMIT in 7m 29s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/65004db/ : RETRY_LIMIT in 8m 00s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"6ae74c8bb80d48cea4b5c7c76db24fbe3caa8101","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 17:00:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : FAILURE in 1h 17m 25s","accounts_in_message":[],"_revision_number":3},{"id":"02c9a0e4e361c6e14a528f07beec4914f9cd17e2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 17:02:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-886818-1/ns5_nfs : SUCCESS in 4h 10m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-886818-1/ns5_iscsi : SUCCESS in 4h 13m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"96074750189fc3a21d1225ccf19346ae462ae5ab","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 17:02:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/886818/3/check/iscsi-driver-alletra-6k/34805b7/34805b7.html : FAILURE in 1h 22m 13s","accounts_in_message":[],"_revision_number":3},{"id":"4ffebe478eaf87a961f9bf4b7abbe67897cfefa1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-23 17:07:17.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-1/FC/ : FAILURE in 4h 26m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-1/iSCSI/ : SUCCESS in 1h 56m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6bea94856d948c0cd375472ad9a7dbcea2ad36d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 17:20:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/616f013a4ab54f3e9e87c963b0c49140\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/647cb283037448c9bf72533a3bd415ba : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":3},{"id":"a4ed0b4d85437ff7f75fe502a2c74d8d647b020b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-23 17:51:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerStore_iSCSI/31228f7/DellEMC_PowerStore_iSCSI/125 : FAILURE in 54m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerStore_FC/872ed52/DellEMC_PowerStore_FC/130 : FAILURE in 1h 23m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerStore_NFS/cf5c516/DellEMC_PowerStore_NFS/124 : FAILURE in 56m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerStore_NVMeTCP/65637b4/DellEMC_PowerStore_NVMeTCP/125 : FAILURE in 1h 12m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"98d75546ad19ea5267dfa74deedd487aef71f617","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 18:20:15.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-886818-2/ns5_nfs : SUCCESS in 4h 1m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-886818-2/ns5_iscsi : SUCCESS in 4h 5m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d9aa011bc445bc858693985fe6befa24f75ef914","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 18:35:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/886818/3/check/iscsi-driver-nimble/86b3901/86b3901.html : FAILURE in 1h 41m 35s","accounts_in_message":[],"_revision_number":3},{"id":"702f8594f9b472c30e68bb1d0d21b74b9f6c7329","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 19:43:29.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-886818-3/ns5_nfs : SUCCESS in 3h 58m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-886818-3/ns5_iscsi : SUCCESS in 4h 1m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"cb430e3af77a16d06d95b8e94624adf699a4584c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-23 19:48:18.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-2/FC/ : FAILURE in 4h 55m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-2/iSCSI/ : FAILURE in 5h 33m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"56184cdc4c82dedc44f6ad843bbd848919925ff0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-23 21:33:23.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-3/FC/ : FAILURE in 5h 38m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-3/iSCSI/ : UNSTABLE in 1h 3m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"4fcc46560dd69925b4a4fd80872165799ab161fc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 22:07:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerFlex/11fa087/DellEMC_PowerFlex/198 : FAILURE in 1h 54m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"486ec57c9817a62990cdeac9bcd108a72f96f143","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-23 22:47:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886818-3/HPMSA-iSCSI/ : SUCCESS in 1:51:15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886818-3/HPMSA-FC/ : FAILURE in 1:39:18","accounts_in_message":[],"_revision_number":3},{"id":"dbe238b31fdc775616827220661be7bbc12d2073","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-24 00:52:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886818-3/PVME-iSCSI/ : SUCCESS in 2:05:34\n- PVME-FC https://os-logs.seagate.com/dell-ci/886818-3/PVME-FC/ : SUCCESS in 1:40:30","accounts_in_message":[],"_revision_number":3},{"id":"cdbf0d42f5daeb84d0fbd855ec5c0f97ca0980cb","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-24 11:19:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerMAX_FC/5017e3b/DellEMC_PowerMAX_FC/155 : FAILURE in 7h 11m 58s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/3/check/DellEMC_PowerMAX_iSCSI/af63727/DellEMC_PowerMAX_iSCSI/148 : FAILURE in 6h 15m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"7d9e9eea153f077bc6533ebf795f25c565ec330a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-25 17:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/1/fc : SUCCESS in 0h 36m 07s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/1/iscsi : SUCCESS in 0h 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f2cfa26c0ac97cc4b21a125a41329fe51d27c810","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-25 19:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/2/fc : SUCCESS in 0h 27m 02s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/2/iscsi : SUCCESS in 0h 45m 52s","accounts_in_message":[],"_revision_number":2},{"id":"411c93ef11f6bd87d659fa641ee3d5175d598a25","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-01 04:05:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/3/20230701053148/NEC-FC : SUCCESS in 3h 51m 07s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/3/20230701053148/NEC-ISCSI : SUCCESS in 7h 33m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e99617b4d30e2dca9b483191da85097c070cbfed","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-01 07:00:08.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/3/fc : SUCCESS in 0h 39m 12s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/3/iscsi : SUCCESS in 0h 42m 28s","accounts_in_message":[],"_revision_number":3},{"id":"c4e216dd89d1d0ecc918127661b353f42ce7d97c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-02 10:55:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/3/20230702170450/NEC-V-ISCSI : SUCCESS in 2h 46m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/3/20230702170450/NEC-V-FC : SUCCESS in 2h 50m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a60bb6553fd9f9794acf2eb2b06389fb6fecabd3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-05 03:39:58.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-3/FC/ : SUCCESS in 9h 26m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-3/iSCSI/ : FAILURE in 10h 21m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"863c60d929b2dbdc85a0d321cf4230cfc91e2b99","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-07-05 04:17:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5fb4138bee5ffd589b40080aa461f89e70ddbfb5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-10 07:32:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"49cefa9dee6ea684e064515dd15b42ee26dc42f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-10 16:20:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"75ac44b508dd97e82238564e469a35abc5460240","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-10 16:22:53.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-886818-4/ns5_nfs : FAILURE in 1m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-886818-4/ns5_iscsi : FAILURE in 1m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"2a7421fc07da68b66c615b0a06aa5f280ff58c2d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-10 16:26:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aa89211cd83d435399d9e2c1524f6e62 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0014cf18b3004aaaa1ca9d9e065c58bf : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"2f6377c7da8a7e1b0460a3c9f70b06ecc2c67de5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:38:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e84bee4538f243e097eb096a00e78d99\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdd014bd82184125b81f1ebafd971e19 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90c8d80c03a144b2a79057cdd9129c98 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b735d18c25d4bb3910b2b664e46f3af : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cf7f1ad38ba9864ce31ebeb59ee33e91f6073f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 17:21:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72634b66bbad4e3e8c25b034813dc18c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/30484a26a6964157aec0198faf97eddd : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628f9283eb8d44dd8de9c573fd67c99f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/458e6aaa13e040eaa67e358fe0e2c70b : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ff152bb01e439594aa57cf8d2887d0 : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b904abbc78844f3abc62c75a04a1c39 : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336262059e3c415f9c620b98f043e716 : SUCCESS in 11m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb61d8003c7a4b0bb573fe6c2d54e7b7 : SUCCESS in 16m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3463d55eb2ad450f85b6862a47a88d72 : SUCCESS in 4m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/213117da650e41d8be73a6bba9ed8e76 : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fffd581d279c411d862de23721502163 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b8a3f867376d44449ab28fffcee6782b : SUCCESS in 54m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d27b7ef29494243976c09161cb289bc : SUCCESS in 4m 56s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/07c576ce3f80487196f6f20589592576 : SUCCESS in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6b539f5c1aebf000683f5bcbad230e6a05958b44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-10 17:28:24.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_07_2023_17_21_28\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"55e0ca95c18d9e407900c89430d705a6a08a7b0a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-10 17:39:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/0f61bbd59b55450ebbbd17a00f28f71b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d8252d114bf44c92aeb93404f987d31c : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":4},{"id":"66c2a26b0db05d7f03ba0ebc8fb3f722a6f678c3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-10 17:53:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":4},{"id":"59eccd8de480a0a0205f245ac7cac29681127057","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-10 17:54:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : FAILURE in 1h 31m 33s","accounts_in_message":[],"_revision_number":4},{"id":"9ab8aed88b41c5a1aab44443f2fcb0e5b592fafd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-10 18:09:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/4/upstream-check/cinder-tempest-plugin-ontap-fcp/a785cd4/ : RETRY_LIMIT in 7m 22s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/930a568/ : SUCCESS in 46m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/4/upstream-check/cinder-tempest-plugin-ontap-nfs/0f85b4e/ : SUCCESS in 46m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ba9c696/ : SUCCESS in 47m 34s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8e1032b9d5eeba6accc2763a1a80cf622aefa495","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-10 18:10:06.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_PowerFlex/e9e6e02/DellEMC_PowerFlex/272 : SUCCESS in 1h 36m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"9cfea34d5c79a5d8afc8d97d0dcb696da87a7c38","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-10 18:44:23.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_PowerStore_iSCSI/bcb3d2a/DellEMC_PowerStore_iSCSI/148 : SUCCESS in 59m 39s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_PowerStore_FC/4f3ab59/DellEMC_PowerStore_FC/156 : SUCCESS in 1h 21m 23s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_PowerStore_NFS/0439eae/DellEMC_PowerStore_NFS/148 : FAILURE in 34m 40s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_PowerStore_NVMeTCP/056acd2/DellEMC_PowerStore_NVMeTCP/148 : SUCCESS in 1h 22m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"637304a3e703ddfc3e9bad6c94ea59cfbafa2b21","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-10 19:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/4/fc : SUCCESS in 0h 26m 13s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/4/iscsi : SUCCESS in 0h 55m 52s","accounts_in_message":[],"_revision_number":4},{"id":"0cc909854bd93f86f882fcffaca025b3c11493c3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-10 19:13:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/4/20230711012101/NEC-V-ISCSI : SUCCESS in 2h 48m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/4/20230711012101/NEC-V-FC : SUCCESS in 2h 52m 00s","accounts_in_message":[],"_revision_number":4},{"id":"98774c7862ff57ea580dd082e705f612730d0403","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-10 20:17:19.000000000","message":"Patch Set 4:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1564/","accounts_in_message":[],"_revision_number":4},{"id":"a169546896a37a5ff46e33a4e5d3e713846295c7","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-10 20:19:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/18/886818/4/check/tyds_test/e158755 : SUCCESS in 3h 58m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2145988fbc66fd258d598ba1704ade4bd83826fe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-10 20:26:36.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-4/FC/ : SUCCESS in 4h 5m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-4/iSCSI/ : SUCCESS in 2h 51m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"e1cbf287253b17318d031d0ccb52046a869fead0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-10 20:32:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_Unity_FC/b65683f/DellEMC_Unity_FC/401 : SUCCESS in 1h 50m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/4/check/DellEMC_Unity_iSCSI/15d3ce1/DellEMC_Unity_iSCSI/403 : SUCCESS in 2h 13m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"ec8f34ba3b166000418e4b3db9803dc3409a5370","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-10 23:10:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886818-4/HPMSA-iSCSI/ : SUCCESS in 1:50:53\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886818-4/HPMSA-FC/ : SUCCESS in 1:29:40","accounts_in_message":[],"_revision_number":4},{"id":"94f9c529a5cbc8fb6bf995b224367cf548fe8371","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-10 23:40:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886818-4/PVME-iSCSI/ : SUCCESS in 1:59:57\n- PVME-FC https://os-logs.seagate.com/dell-ci/886818-4/PVME-FC/ : SUCCESS in 1:39:13","accounts_in_message":[],"_revision_number":4},{"id":"21d762e2eaeea317a08277671538e4de16a5d888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-11 11:48:05.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7a2ad1506303997361fca7cf2f59899860de2843","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-11 11:53:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/58e84d571c2d4052b4b7cb4d13c4d904 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9d39771a071d440982498d554a03c13a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"caef3dbefddf7cce71ed623dd1be15675c95b02f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 12:03:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5cc92fdfd348b3bae98fdb23187456\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ddfcdc293ec4d0db050a23a7d1d37a8 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e4d390c147949fda7ec1a7cfe4aa3d8 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6e46bf6d3a3471cae4620d91d3e75ae : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f8bb12466220d90ffeec5e73f71824cac1e4e84f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-11 12:09:42.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-886818-5/ns5_nfs : FAILURE in 1m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-886818-5/ns5_iscsi : FAILURE in 1m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"992cf04a7d92917062396521015cbf56f2888c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 12:46:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f6c06f1dfc047bfac82bbe907d795cd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f2d8dfa0c51e44c7a6c9954c0ef361cc : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e5d370a4a64c259f80efee9ef262a6 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0483f0b15da343d89c318097df34c20c : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e022798d04bf4579a2c42d7c5f00cbe0 : SUCCESS in 11m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a22c7130a70452eb89e2d71d27c28e6 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5027632f7c41c58d6b41e2c3ccabeb : SUCCESS in 12m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/96ff5c1566a94ba39c7c11fe046a26b1 : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a75434af290940b3b386eb40f3d27b92 : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1a536dc552b4ebaa7d6d8c41f131b91 : SUCCESS in 6m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ac5f6cc6cc74142be6bac833bc1f920 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a2449129cf3e440899ece9e7b1bd76b2 : SUCCESS in 50m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/898ea58f8e574cb4bfba0f6d8d92a639 : SUCCESS in 5m 46s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/01cf76a8b081434fac7c4309ac1402df : SUCCESS in 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b10d70431dd91bb4915045a81a3dadfc52b965dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-11 12:47:49.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : FAILURE in 57m 49s","accounts_in_message":[],"_revision_number":5},{"id":"d85e360e9ddbc00cf4986ea19995ae4a77ba50cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-11 12:53:36.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_07_2023_12_46_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"caf210bc8cd5983cb618548f466c9e963b70bffc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-11 13:34:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/5/upstream-check/cinder-tempest-plugin-ontap-fcp/59dda45/ : RETRY_LIMIT in 8m 29s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/7d41b5d/ : SUCCESS in 45m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/5/upstream-check/cinder-tempest-plugin-ontap-nfs/630f3ad/ : SUCCESS in 46m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7f3aec1/ : FAILURE in 45m 54s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"9eb6bbe51eb701052e0bc41a6ac5845c109905bf","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-11 14:00:08.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/5/fc : SUCCESS in 0h 38m 06s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/5/iscsi : SUCCESS in 0h 55m 42s","accounts_in_message":[],"_revision_number":5},{"id":"fb025127efd6546cb50875d81547c03b72535ed3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-11 14:04:39.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/886818/5/check/DellEMC_Unity_FC/e036050/DellEMC_Unity_FC/404 : SUCCESS in 1h 48m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/5/check/DellEMC_Unity_iSCSI/bdf3939/DellEMC_Unity_iSCSI/406 : SUCCESS in 1h 52m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"26a7ba056301f20853868d070234140545bec729","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-11 14:06:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/ba4f4ed4d01342ef9b4a9413570026cb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bc9c86d7978540768a66969d4805a31e : SUCCESS in 1h 14m 36s","accounts_in_message":[],"_revision_number":5},{"id":"b64a34226fe22ca0400218633960e4cd1eb0117b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-11 14:06:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/5/check/DellEMC_PowerStore_iSCSI/972d6af/DellEMC_PowerStore_iSCSI/150 : SUCCESS in 1h 03m 01s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/886818/5/check/DellEMC_PowerStore_FC/ad9fc41/DellEMC_PowerStore_FC/158 : SUCCESS in 1h 20m 07s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/886818/5/check/DellEMC_PowerStore_NFS/8cbb450/DellEMC_PowerStore_NFS/152 : FAILURE in 30m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/886818/5/check/DellEMC_PowerStore_NVMeTCP/ebc3ded/DellEMC_PowerStore_NVMeTCP/150 : SUCCESS in 1h 12m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"91464f0c89d151194f8b351f076667ec8c065106","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-11 14:43:38.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-5/FC/ : SUCCESS in 2h 30m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-5/iSCSI/ : SUCCESS in 2h 43m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"270ccde33878c6f590c9f9c9d62758c25facfeb1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-11 14:44:42.000000000","message":"Patch Set 5:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1566/","accounts_in_message":[],"_revision_number":5},{"id":"6f94a07e75b2dfee041f8b721b9e36023632048b","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-11 15:12:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/18/886818/5/check/tyds_test/56f38b7 : SUCCESS in 3h 22m 47s","accounts_in_message":[],"_revision_number":5},{"id":"871bc83dd77059c9e5bc21d1cb51cf95acbe26d5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-11 18:40:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886818-5/HPMSA-iSCSI/ : SUCCESS in 1:53:13\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886818-5/HPMSA-FC/ : SUCCESS in 1:29:59","accounts_in_message":[],"_revision_number":5},{"id":"2bbe94ee5eaaccf2c0a7a441d6ed264d841793d8","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-11 19:21:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886818-5/PVME-iSCSI/ : SUCCESS in 2:03:51\n- PVME-FC https://os-logs.seagate.com/dell-ci/886818-5/PVME-FC/ : FAILURE in 1:40:24","accounts_in_message":[],"_revision_number":5},{"id":"bbd41fdc50a25013d3b069a77f4a5ee6d4bc23da","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-12 08:21:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/5/20230712142851/NEC-V-ISCSI : SUCCESS in 2h 43m 34s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/5/20230712142851/NEC-V-FC : SUCCESS in 2h 52m 09s","accounts_in_message":[],"_revision_number":5},{"id":"c67758b00a1da37edf7a459fd3f0487f4f7f235d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-12 12:37:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/5/20230712132202/NEC-FC : SUCCESS in 3h 44m 49s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/5/20230712132202/NEC-ISCSI : SUCCESS in 8h 15m 16s","accounts_in_message":[],"_revision_number":5},{"id":"85d2ffe80a06e41e572f88e83bdb34e522d8d06e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-13 13:26:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"398ecfe2c8ff92f1bfd65f4dca4bbebb2430c958","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 13:32:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4c85588aa05c45b4ba13b5aa34000843 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1303f510c44b476fab40c4c10e661e64 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"8dff81925747ca40f1f51222cba5805ad3b03b54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 13:53:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/549a047a18c64e22aee9f4258b22a753\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3664c5d7a83e43d280fe1f5b8a806a91 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26649906cf07437da6bfe6277d8d287d : FAILURE in 25m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c096fea56af347658fb4bc4281010ce0 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8d6e664f7d42313741622c6545f0bd874f4ff30a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-13 14:06:44.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-886818-6/ns5_nfs : FAILURE in 39m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-886818-6/ns5_iscsi : FAILURE in 39m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"87c21315821c35e41bf618a08d3f71a88e40df24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 14:37:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b1ece0a6fad43adb46ba9a938eec85e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/925d005d20324e0ab225f7ce3caca23d : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7d27de8c364fce9abd28218e7980ac : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd0e2786203340ab9696b480804cdd53 : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d41d7f4d6aae427783847831ddf68ad3 : SUCCESS in 11m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9d1a21afb7444d69132544eacd2ff08 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4401b872b3c54d8381e2367b52a8a0a5 : SUCCESS in 12m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/736f8c9372e440b38dc9e5875676f712 : SUCCESS in 15m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9700554c98eb47a0bef467307cddd108 : SUCCESS in 5m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c785ec1d062494f8534ca79806ecc80 : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/084830626d154aa6aef5e31a39828606 : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d62e4d003dfc4516a2ea9ee73fe5c5d3 : SUCCESS in 1h 04m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b825b7f3d0eb4d79b458a8e01acbf6ea : SUCCESS in 6m 04s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/3f8666a8e8a54184890905ae6518edbf : SUCCESS in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"11e71feb7f6e60717449ed39a1dd8760b5651d6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-13 14:45:18.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_07_2023_14_38_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"51aba5cea5f7e007e201bdd108c4944e7d4f4358","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-13 14:46:43.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1efa6fe352fdcf7a6e1c10cd6e26f391e16abe89","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-13 14:51:59.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/46e5cb1c2e6a45e18a254fd64b5ee500\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0b32758dd85f4699916e54e657774eb6 : FAILURE in 1h 15m 17s","accounts_in_message":[],"_revision_number":6},{"id":"3a550d0a50bb42424b05af09ebcf4366eb0d9564","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-13 15:19:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886818 : FAILURE in 1h 49m 41s","accounts_in_message":[],"_revision_number":6},{"id":"35087f02f3a6bbee4f6c069fd0d2f3244aae7873","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-13 15:25:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/6/upstream-check/cinder-tempest-plugin-ontap-fcp/79260de/ : RETRY_LIMIT in 8m 09s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/a17845f/ : SUCCESS in 43m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/6/upstream-check/cinder-tempest-plugin-ontap-nfs/97c2687/ : SUCCESS in 46m 17s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/886818/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f223a4a/ : FAILURE in 44m 32s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"c824a19be42c529a9ddf26be31b3c72f9b157431","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-13 17:04:04.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/886818/6/check/DellEMC_Unity_FC/0f6db93/DellEMC_Unity_FC/422 : FAILURE in 1h 52m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/886818/6/check/DellEMC_Unity_iSCSI/df8cb82/DellEMC_Unity_iSCSI/424 : SUCCESS in 1h 46m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"1056f3492b0ce8ce0b08eef394475e1cf898de8b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-13 18:55:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886818-6/HPMSA-iSCSI/ : FAILURE in 1:54:30\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886818-6/HPMSA-FC/ : SUCCESS in 1:31:28","accounts_in_message":[],"_revision_number":6},{"id":"0e54b4d99842aba3b7368fd6ca09c6360293b110","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-13 19:16:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886818-6/PVME-iSCSI/ : SUCCESS in 2:06:20\n- PVME-FC https://os-logs.seagate.com/dell-ci/886818-6/PVME-FC/ : SUCCESS in 1:40:58","accounts_in_message":[],"_revision_number":6},{"id":"6fc74dac6ffff851b331a11a9245485f485439f3","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-13 20:59:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/18/886818/6/check/tyds_test/a30dfcb : SUCCESS in 3h 54m 16s","accounts_in_message":[],"_revision_number":6},{"id":"8404457cadee8e2a3c8fbc2930770e88a4cc772b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-13 21:50:36.000000000","message":"Patch Set 6:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1581/","accounts_in_message":[],"_revision_number":6},{"id":"bbc0aa20849d1f91b4c0e481479882ab861297a5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-13 22:00:08.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/6/fc : SUCCESS in 0h 39m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/886818/6/iscsi : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":6},{"id":"bb40479b8361977fd4724990a7d63eb187d67d2b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-14 11:51:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/6/20230714180301/NEC-V-ISCSI : SUCCESS in 2h 38m 22s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/6/20230714180301/NEC-V-FC : SUCCESS in 2h 48m 55s","accounts_in_message":[],"_revision_number":6},{"id":"2fa74d263191cfe19b492a622cebfd0047eb09cb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-14 14:12:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/6/20230714191436/NEC-FC : SUCCESS in 3h 51m 12s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/886818/6/20230714191436/NEC-ISCSI : SUCCESS in 3h 57m 57s","accounts_in_message":[],"_revision_number":6},{"id":"24606a2527bd200dcec164a4b72e1adab77742ea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-15 22:43:10.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-6/FC/ : SUCCESS in 2h 41m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-886818-6/iSCSI/ : SUCCESS in 4h 1m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"da6ea2db7283625fdef1ca383c08c02092e0a478","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-17 10:55:17.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6993f28763b4ce2a1658cd03a6cc06fb788eb766","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-24 10:58:30.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"d7b257c321b85e4fe59bb65a69e04973b9e7d690","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-24 11:07:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"01024a30866ee4eaca8b3eec2fd4bf54554fd647","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-24 11:08:23.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7e416c29e5e9d3d0077bb8125cba912346c61b5e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-27 05:54:31.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"000925e99329f56736a87c5fabb2a5c67b6e239b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-27 08:28:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b3d612c38ace28624d9c00a5e09edf0159b970cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:29:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c8bea37bd86c3af4a128f117911be80162430eeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:52:52.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29068f64282440518b5fd8bb93a91b87\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/231bad0a6cfc454b8b823cfb3fc8611d : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf391dc9cad9432dbe9a11a6d8eb1022 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/670db1f1e0364e789fcae9d5e5af7c02 : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/370b8312eb324ea8892c376fc096076c : SUCCESS in 14m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da07ff870b7647cd9e1f13200d01a906 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf25ceb1aac5470089c1114d7744efe5 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":6},{"id":"e74b23837289dcc1fa89f69c17cbf4c1c0239a09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:52:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"aab4ae6bcce6e1f35c3c3e772583623a395c315e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 08:54:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19db9046d8e74539aa76b02f120f6aad\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/3272fc5e3f284e9e9aa62829ee6a5de7 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7994f1a43bfd403b8f2d63770d944e93 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6172d2a9a0ad6cb35e2218ee6c2f57588ac97981","revisions":{"4d4b302b7e050f81e9110f8bcfa52587c0cf7e05":{"kind":"REWORK","_number":1,"created":"2023-06-23 12:34:34.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/18/886818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/886818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/886818/1"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) List backup with details\n2) Show backup\n3) Update backup\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I9bbfab93fb138932243d880e440c9524b5f08217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d4b302b7e050f81e9110f8bcfa52587c0cf7e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d4b302b7e050f81e9110f8bcfa52587c0cf7e05"}]},"branch":"refs/heads/master"},"f23565d062d97dff4bdc3aba53352e576f5926cf":{"kind":"REWORK","_number":2,"created":"2023-06-23 14:13:57.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/18/886818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/886818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/886818/2"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 14:13:05.000000000","tz":60},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) List backup with details\n2) Show backup\n3) Update backup\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I9bbfab93fb138932243d880e440c9524b5f08217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f23565d062d97dff4bdc3aba53352e576f5926cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f23565d062d97dff4bdc3aba53352e576f5926cf"}]},"branch":"refs/heads/master"},"254d45157d1e6c92a16f7c1bdc3bf8f4afc34ac2":{"kind":"REWORK","_number":3,"created":"2023-06-23 15:38:46.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/18/886818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/886818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/886818/3"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 15:38:11.000000000","tz":60},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) List backup with details\n2) Show backup\n3) Update backup\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I9bbfab93fb138932243d880e440c9524b5f08217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/254d45157d1e6c92a16f7c1bdc3bf8f4afc34ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/254d45157d1e6c92a16f7c1bdc3bf8f4afc34ac2"}]},"branch":"refs/heads/master"},"f34fcb8dce7797ec6bca8ee8d069a2acb8f85e84":{"kind":"REWORK","_number":4,"created":"2023-07-10 16:20:59.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/18/886818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/886818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/886818/4"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-07-10 16:18:26.000000000","tz":60},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) List backup with details\n2) Show backup\n3) Update backup\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I9bbfab93fb138932243d880e440c9524b5f08217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f34fcb8dce7797ec6bca8ee8d069a2acb8f85e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f34fcb8dce7797ec6bca8ee8d069a2acb8f85e84"}]},"branch":"refs/heads/master"},"c7027607befd8ae0f0d546a89f7e1be2e6541f80":{"kind":"REWORK","_number":5,"created":"2023-07-11 11:48:05.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/18/886818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/886818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/886818/5"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-07-11 11:46:22.000000000","tz":60},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) List backup with details\n2) Show backup\n3) Update backup\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I9bbfab93fb138932243d880e440c9524b5f08217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7027607befd8ae0f0d546a89f7e1be2e6541f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7027607befd8ae0f0d546a89f7e1be2e6541f80"}]},"branch":"refs/heads/master"},"6172d2a9a0ad6cb35e2218ee6c2f57588ac97981":{"kind":"REWORK","_number":6,"created":"2023-07-13 13:26:44.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/18/886818/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/886818/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/886818/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/886818/6"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-07-13 13:25:52.000000000","tz":60},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) List backup with details\n2) Show backup\n3) Update backup\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I9bbfab93fb138932243d880e440c9524b5f08217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6172d2a9a0ad6cb35e2218ee6c2f57588ac97981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6172d2a9a0ad6cb35e2218ee6c2f57588ac97981"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
