)]}'
{"id":"openstack%2Fcinder~886869","triplet_id":"openstack%2Fcinder~master~I401ca89440d44e04f460741981cf24a42e5264a0","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2023-09-11 18:37:33.000000000","reason":"Change was submitted"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-08-01 13:10:05.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"35677":{"account":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"last_update":"2023-07-29 09:14:50.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I401ca89440d44e04f460741981cf24a42e5264a0","subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","status":"MERGED","created":"2023-06-23 19:26:45.000000000","updated":"2023-09-12 21:06:45.000000000","submitted":"2023-09-11 18:37:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886869","meta_rev_id":"645d9dd1227e6ad9e8c7d87c7a1b8f2c0eb6cd3c","_number":886869,"virtual_id_number":886869,"owner":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 18:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2023-08-01 13:10:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2023-08-03 13:17:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":1,"date":"2023-08-04 14:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":1,"date":"2023-08-10 11:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":1,"date":"2023-09-11 05:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":2,"date":"2023-08-16 14:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2023-08-09 12:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2023-09-09 08:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-23 19:27:41.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 19:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 22:18:46.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-06-23 22:50:56.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-06-23 23:34:07.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-23 23:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-23 23:50:39.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-24 00:00:17.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-24 06:21:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-28 21:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-01 14:59:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-03 01:15:21.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-20 18:38:11.000000000","updated_by":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"reviewer":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"state":"REVIEWER"},{"updated":"2023-07-21 12:42:48.000000000","updated_by":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"reviewer":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"state":"CC"},{"updated":"2023-07-21 15:08:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-24 22:31:02.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-24 23:12:56.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 05:20:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-25 10:43:27.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 17:38:04.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-25 18:54:12.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-07-25 20:04:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-25 20:09:37.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-07-25 23:07:36.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-07-26 08:25:46.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-28 06:51:31.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2023-08-01 13:10:05.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-08-03 03:47:18.000000000","updated_by":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"CC"},{"updated":"2023-08-03 13:17:03.000000000","updated_by":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"reviewer":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"state":"REVIEWER"},{"updated":"2023-08-04 14:03:04.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2023-08-09 12:41:15.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-08-10 11:55:18.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2023-08-16 14:53:08.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-09-11 05:56:30.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"f641b89ec396bff194484d48a29c202f9be3b5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-06-23 19:26:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1e5263393fdf6b3a6be4cba3bc58e3090a0ed9b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 19:27:41.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a2a20269fd61425fe054df996c63c3e35cfc8e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8127773156f542d3896e14718784bf0e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2d53615ce9d49e9a169284d61e2fcb2 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a124ee7f3024b81898d0e4f6b179e09 : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e65ba8f3048f48528d7e6607536447df : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f69cf3ed89b93d4d847d4fb91add9b4b12dff8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-06-23 20:19:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"be09d95f089a7eb937dbed6bd05cae5459c9ba4c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 20:21:21.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4559030bb52b3e93eeb9a2802a0a2579fd7d764","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 20:34:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa15d021aa504667a058353c2d9148b1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df7c297884be44468f9e0b06921d9003 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efa1b7ed0f0a4d23827a42bc95f2e5cc : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf48386a9b12455db138d76d1909c59d : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17334c12207c80b99704d7ab4ad65ae207e3d922","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 22:18:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/886869/2/check/iscsi-driver-alletra-6k/c23fb63/c23fb63.html : FAILURE in 1h 14m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f43ed2a6264ff6ae39c57f84698e109c21a1ea1c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 22:50:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/0efd1db140f941629e84ea6f0b7aa29a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/95ad2626af054c1c8dbfff08c8f17658 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4fac031c0ebd5106682d701d68526b46f8b02fcb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 23:34:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886869 : FAILURE in 1h 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"423d19a5e599569dff8b7d882d147b4eaf210e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 23:34:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0b670f996d14d178cdc9e8fd16039c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b872ff3ceacf41f3b85687aa01c27458 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b3c276e2abf4e51acd12a1a2a49dfc0 : SUCCESS in 13m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daa48386e44f4606a270beed238b3381 : SUCCESS in 12m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13fadaeed2984745ac7298a49754033e : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44c80491f01453a980509dc41b98c7f : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/64c86cc4ae56437ab4b520d1635971bd : SUCCESS in 58m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78958babaf224f76b4df59fae8c1647c : SUCCESS in 1h 27m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17c02f16e9244940b8e6004024729c74 : SUCCESS in 1h 16m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb507e75bd83420887f171c6721475fc : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8564288fce842bea406c2a6eb27d49d : SUCCESS in 14m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04ebe73d33064e3b9ed342dade9e2186 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/697f8949fa314262a96bd1b75e64b81f : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/546f49172a9c4afc921c9d8b55ee94de : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5ca863563644f50a14b24d2e278e2cd : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/df0223daf57349d79c8823cf813afed5 : SUCCESS in 57m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e1e10dd16ee484fb335583a2b83bbd0 : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3ec81ed265d4500be6b53fa1e1e7d21 : FAILURE in 1h 48m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/449d61b25de241dd9efb1b4fdab423ad : FAILURE in 1h 53m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6719ba5c499c4610aad279731f8cb1f0 : SUCCESS in 1h 53m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5e6930076ffc43a5ad6c02207ec8605d : SUCCESS in 2h 11m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/391adc95b0234462806b954e032362d6 : SUCCESS in 1h 09m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/388dbeb15c9d4d099c11171b0a728d77 : SUCCESS in 1h 41m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de8ff8b640f040928a15454057e071d1 : SUCCESS in 33m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5e6e4f1c1c8f4c45b2161dd337b35927 : SUCCESS in 1h 32m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/68c4824f16254bf79fb24217cf215c32 : SUCCESS in 1h 48m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e081cec0f33a44e4a054b923f8b527a9 : TIMED_OUT in 3h 03m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36c9145009e14a57b9cb4e63be326395 : SUCCESS in 1h 02m 57s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/66e2149a9f9d4676b6a0c291e83517d6 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca8dd53746aa5721730c544301ee6d58f98f0e59","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 23:50:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-886869-1/ns5_nfs : SUCCESS in 4h 18m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-886869-1/ns5_iscsi : SUCCESS in 4h 23m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"2bbbbcd8a15322d725c540d627dcc7228ef9826e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-24 00:00:17.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-1/FC/ : FAILURE in 4h 33m 12s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-1/iSCSI/ : SUCCESS in 3h 5m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea18536254420de3bba5d40754b507f27726f353","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-24 00:46:38.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-886869-2/ns5_nfs : SUCCESS in 3h 58m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-886869-2/ns5_iscsi : SUCCESS in 4h 3m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"b6000aa2e9ca4c1a22344d800b64d2ec10ebc027","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-24 01:25:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/69/886869/2/check/iscsi-driver-nimble/4ff10f0/4ff10f0.html : FAILURE in 1h 20m 22s","accounts_in_message":[],"_revision_number":2},{"id":"54c5823835edafcf62d8315a9025362f1405c23d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-24 02:22:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-2/FC/ : SUCCESS in 5h 46m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-2/iSCSI/ : SUCCESS in 4h 38m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"76de8d02c0029172c65f029eac75f85ed59dd320","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-24 06:21:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/2/check/DellEMC_PowerFlex/d13e83d/DellEMC_PowerFlex/203 : SUCCESS in 1h 37m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"99c8f79d335f08ad1b9944893467a031fc62f350","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-28 15:24:34.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-2/FC/ : FAILURE in 33h 47m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-2/iSCSI/ : FAILURE in 33h 50m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"e53d4009a3b726c125cce4bb5f5f5ab9f6088ffc","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-28 21:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/2/fc : SUCCESS in 0h 38m 21s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/2/iscsi : SUCCESS in 0h 51m 17s","accounts_in_message":[],"_revision_number":2},{"id":"945f8d0a731dae7de3ce433174da8a6a98c80f3f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-01 14:59:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/2/20230701201820/NEC-ISCSI : SUCCESS in 3h 36m 20s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/2/20230701201820/NEC-FC : SUCCESS in 3h 41m 16s","accounts_in_message":[],"_revision_number":2},{"id":"93b19f329b471b1421289854a17cbab422f7a6a6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-03 01:15:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/2/20230703072711/NEC-V-ISCSI : SUCCESS in 2h 40m 42s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/2/20230703072711/NEC-V-FC : SUCCESS in 2h 48m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e6c805fc8ce6a1fe2e41879ca1dd0a6cdb9759fd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-05 17:25:12.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-2/FC/ : SUCCESS in 2h 51m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-2/iSCSI/ : FAILURE in 5h 17m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"63d972ee69e0025b415c1b5072cdc6c6fec1af8b","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-19 13:32:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9351889f34448781f89f060f1e2768315c20ae9","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-19 13:33:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"127dfcb45be051216f1754987d73efa62dc34377","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:36:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f29a7c304064d0887f35f67b7fef536\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a444116d885d407b9c5d50509707ad0d : NODE_FAILURE Node request 300-0021725842 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6f446be73654aa78f93b540645184e8 : NODE_FAILURE Node request 300-0021725843 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58b953d9595b4caca39ab496af0687d0 : NODE_FAILURE Node request 300-0021725844 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6205e6d5797d758cbe18532e2b486a39d839e6a8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-19 14:33:10.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/eb7906731d944d788a10f95dc7fdc29b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8ffc3074049e4d7ab28270ed3512a030 : FAILURE in 56m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4da444092473f35ebff7115eb5cce571b5c0b137","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-19 14:55:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/db9cc92a67714e22bfef4bb1c4077f32 : SUCCESS in 1h 19m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b52fff59c5814d1dbc3b576c5fe3b900 : RETRY_LIMIT in 1h 04m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fe601bc8af6830473fe672823825de5a8d7dc79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 16:29:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f96dd29c8d941868ea8b06fa772e23d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca328c3500bc4a989205c43db8b4e692 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca2041897d7f40a0a8ee528a47f0c8ee : SUCCESS in 12m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f5ea2a0e0414ecd8be09d71843f4870 : SUCCESS in 11m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae98e5961814f3b9d991354f0b304e1 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4511ae086f3540fcbc0d698a8fe423ab : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/87e89bf44c9648f78dbb7e7d28378222 : SUCCESS in 50m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/346026ca171b4de0abe481893e776bb0 : SUCCESS in 1h 37m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/142caef41a0544b7847b3430106bfcf0 : SUCCESS in 1h 05m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d724ea48bc624c33810b9818d5ea2bdb : SUCCESS in 9m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ae27dc1f74d4c9c8f37ccfca2496e7b : SUCCESS in 16m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dbef2e8700af48afb643d9794b5a45c8 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c7e1cad32472423da6704b1963d8d314 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/838b7e0e8ab04722b6d15b368719d092 : SUCCESS in 6m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd9b3dd21bb743b18adb0994e0df987d : SUCCESS in 7m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f63422fc658f49b0bebc1773622c10d6 : SUCCESS in 39m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9656770a33584758bc6664ad4f37629b : SUCCESS in 11m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa3b8bc396764362932c922921b80f53 : FAILURE in 2h 09m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aefc591fc1eb4b02a377b8709dfb0731 : FAILURE in 2h 23m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcb97682f6864c19aabc3dc416dd8adc : FAILURE in 2h 43m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6833193f200948e1bd48101c31900678 : FAILURE in 2h 08m 08s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be96c0a16e3f410888442c4f4f0ec422 : SUCCESS in 57m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d23d398a7965463bbe1a82fc113151d3 : FAILURE in 1h 32m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3aadc666dbc84294b6360fc9b25bfe71 : FAILURE in 1h 12m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1abceb0b6424b269acef2dd65e14d6d : SUCCESS in 1h 52m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/57b0b68ee6a94ab2be7076fd8ca792d6 : POST_FAILURE in 2h 45m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/561a36e81cb9485bb9d5387ef3dbcf39 : SUCCESS in 1h 30m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/786be2422dcf4b3aa25de13394defa01 : SUCCESS in 1h 06m 02s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/8e58874bad2d45d894e4ccb7dea0eeaf : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ac66f241d351989060467cf2ee6bee4fb77e1d2","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-07-20 18:38:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"95521ba1fa75cdda3a64b51482f765a290b2502e","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-20 20:24:53.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7872a673710f8745d68f5bc2716ed55a48829aa0","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-20 20:25:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eadea8ab6e61cc9c08e3ce7d375905a915224dbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 20:49:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a8a3b623c1842879a57732ec321f0c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/555051aa351b493ea5290a21f8a6ec8c : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05e27c7535ee4f3ab1353f357f74cdd9 : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28f819dfee994dafbe3a112923ecee56 : SUCCESS in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62d048b95674b672fe8330ce5a20a964b1b67463","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-07-20 20:56:38.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fe3e01b1fc3dfb4cf10add705cf243b3a38eca2a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-20 21:44:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/3403242af72b4f4389bc734976f20563\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b939c14406f64e2ea16c2c6a6cfaa6b0 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":2},{"id":"24a9ce20a23455bd761eca9a16e870e9d7e7e8f2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-20 21:48:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/83a47e07313746108fca78b20ad74ed2 : SUCCESS in 1h 20m 03s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/13f5ed634e3b4ab79c4437ab5dfd7c37 : RETRY_LIMIT in 49m 25s","accounts_in_message":[],"_revision_number":2},{"id":"83f3614358b4576edfe0a7e56bf6805edecbb061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 23:15:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d879bf33f7b47319bf204225f8a843c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96188f1318c4b3f97d989e490aad352 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e9f0fced4b4763932130f237d314b4 : SUCCESS in 12m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0443a902a58940a99f6ba214cf2dad4d : SUCCESS in 12m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c312d90a0c14408a822bd9c13a9060c0 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642170a4b06d4f12bb7c0154540843bf : SUCCESS in 16m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/949287234d8b40a787055a2265909b24 : SUCCESS in 50m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bf93236b36d46df9fc07d4026f88630 : SUCCESS in 1h 32m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/439b45d052b04473afae3a64e737ce5b : SUCCESS in 53m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78c675c9e9fa45aa91ce4db8d194e818 : SUCCESS in 6m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/730cf3887ddd4a35ace23f02d87a369d : SUCCESS in 17m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d817a94efa0447aca67c9b5df5a1a891 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ecc9680248744008444806bed1f61b9 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b2fb1d071e3438ab58ccd9d8db5a867 : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4dd50793bce649be9b40167e7ba8bba8 : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee2dcf3ace2d48d0b9b8c20995e6f683 : SUCCESS in 40m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27c215042ee440a09f10ccfcad4b379b : SUCCESS in 4m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/afdc33dcac5d497080e7f3dc1ac84509 : SUCCESS in 2h 25m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4003dc1461e342d59e5a33334d1bc1d8 : SUCCESS in 1h 50m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e628434749294c7b9323368ecceebe6b : FAILURE in 2h 33m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3efce44441184dbbb9ccd92bad3ad4f4 : FAILURE in 2h 42m 30s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fb60ab76508546c79e752e9a326305ed : FAILURE in 1h 07m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/59b0edd7997746f487c2de609330792a : SUCCESS in 1h 48m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f2df8a3bece149e28a3ff651682d30a5 : SUCCESS in 47m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b3e7be7da164cfa997431d28027ba8d : SUCCESS in 1h 34m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1c0cbe53b5e64912ad012156216d7d37 : POST_FAILURE in 2h 42m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86e37178ca044641ade2f38ca53375b4 : FAILURE in 1h 49m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c904c539deec488b978162ca7ba47a12 : SUCCESS in 1h 01m 00s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/bfe14b6575244c569e2b806401dce7b6 : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"487cd7ee17e7bf4b806ce979d0144b76876a12f7","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2023-07-21 12:42:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3117596f231460dc7a6b073e8a6cfc770e560070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 12:55:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0abcbece9444ac59f73eb7e0454c599\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b68caa608d774498a294463336ed3ea6 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/760b45957278445ab3f0bb2029d0a21c : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4324a11e2eb84158a0768af62b7b11ee : SUCCESS in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8eb9c3ff57227916685621d3d7108ad214c2685","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-21 13:58:21.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d912676b7a9847b999391ce79678221f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6632b24757c64c6ab18d0fcba4acdacb : FAILURE in 1h 12m 30s","accounts_in_message":[],"_revision_number":2},{"id":"36250cea8c7ae22d3eb6d9a260e824ecb9ec04fb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-21 15:08:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/69/886869/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/94f8200/ : FAILURE in 2h 23m 48s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/69/886869/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/23bf76f/ : SUCCESS in 2h 13m 04s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/69/886869/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/80a50b6/ : SUCCESS in 2h 09m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/69/886869/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/4578902/ : SUCCESS in 2h 04m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5149244645ff2e593e7141b677640a9dcbf61722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 15:11:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea8e91515eeb465fbf06a4673f68bb01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9747ae0d23e4da489ff438c0e019b5b : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab1854723abc467b82f331fad92b3363 : SUCCESS in 11m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5d9e71cc40d4c4aad6436f2a6cd793e : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30b4e013ec324f25b6886ec1671ff492 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a61daffeadb47e693eacc2805a4ad09 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/7d338fa98d2343eca3f279c17e78d63f : SUCCESS in 54m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6143700f831428ab1d5997ef5bff475 : SUCCESS in 1h 43m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aff755fb7a8241fca2bbc07d430a9dcb : SUCCESS in 1h 01m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65a4938fa9ec4dd19a6e9629c276d1dc : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34c2cb83a3a742d0816eb50b4b59c500 : SUCCESS in 15m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d34a50fc14e9485fb1628ece01e85506 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5bfaad1a95c4abd892eb1da80594b22 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/738557835f26484482c095170c4032e8 : SUCCESS in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/198db2b7a3b540209276a0bd48ba9705 : SUCCESS in 7m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc06082c9a1644b793ad379dc28add3a : SUCCESS in 40m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09ddfae952f24d20a04d1485201d25be : SUCCESS in 8m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2243add59b2f421fa4a47ff2d90771db : FAILURE in 1h 45m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f199156af9d40c0b7cb972e5a395162 : FAILURE in 2h 08m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb5cd43b516e44ff9caea0559ebf0a7a : SUCCESS in 1h 56m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/630c2c16873549b2845ad9c098b3560a : SUCCESS in 2h 12m 35s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce7b99045561471292748fa773467c2e : SUCCESS in 1h 01m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0090b86e29347d5b069aeeda66aef5d : SUCCESS in 1h 50m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a9ef4b27c8049c0a14a06d8aafb1194 : FAILURE in 41m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/27d63f7920f749d69d4a4f541b962a9b : SUCCESS in 1h 34m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ea0cfe37c66b40f1888364470b0c4305 : SUCCESS in 1h 28m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9b5023684334dfd8140ff0bfad70797 : SUCCESS in 1h 15m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08439b31d0b74ca9a671e45f11d598e7 : SUCCESS in 53m 56s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5412f99a03e64e52b867fcecd44db446 : SUCCESS in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b785789771b8a1782091aca86f866c4b93e57f9d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-21 15:18:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1e17f426d2f1424b99d6412b5628d3c0 : SUCCESS in 1h 28m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/37e13b7dc833467c8dbcc63f94548ff6 : RETRY_LIMIT in 1h 05m 55s","accounts_in_message":[],"_revision_number":2},{"id":"551433784346382a1a829b0060c3c5c7b7dcb31f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-24 19:42:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"04153cbe2316a7d6d2739477ac29df38957fb843","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:55:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/557758c1c02f44b898a76323a49508e9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be24942ee880415dbeb5f563b00f087c : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a98fcc65e6c14df4a80480094de01b90 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9607974a79844706b0429c65a0f2209c : SUCCESS in 9m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cf3b44af4f5c9c784a5fcaf563202452b8855779","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-24 20:30:38.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-886869-3/ns5_nfs : FAILURE in 47m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-886869-3/ns5_iscsi : FAILURE in 47m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"36559057dc5238815e40d2a36a57e89e6c9a78bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-24 20:33:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886869 : FAILURE in 49m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d181fb907c7676ea0e1e39713f40af600f31447f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-24 21:00:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/14808ba3ab7946acb71be2ce8a13d718\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a151cccedcf447ae90fc0ecab7fc8387 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":3},{"id":"833156100f9eadd9e10e21dd42dfb128814199b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-24 21:16:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c00cebf9f96549e1bca0a6afc18c97e5 : SUCCESS in 1h 24m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0317dc8354894deea0b1a95030ab3d70 : RETRY_LIMIT in 58m 04s","accounts_in_message":[],"_revision_number":3},{"id":"45de3f4319628f619b13b60e5517014e808f0b09","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-24 21:22:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/886869/3/check/iscsi-driver-alletra-6k/7233dab/7233dab.html : FAILURE in 1h 38m 05s","accounts_in_message":[],"_revision_number":3},{"id":"b6a9e76e8f94a8f7d62c36eda57f6d4b57383401","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-24 21:41:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/36d8cc5/ : SUCCESS in 1h 54m 23s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/4debe46/ : SUCCESS in 1h 51m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e913c9b/ : SUCCESS in 1h 56m 38s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/bf8cf42/ : SUCCESS in 1h 53m 07s","accounts_in_message":[],"_revision_number":3},{"id":"881bb192281e678c43ad90c229052a5a6041da4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:20:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b28f9bf2d155401e99c0fd9c4000fbe9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5896f9e4e8e141e2b7ba22c6733e522a : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bb98278a6d245319ba5f361ffbbba6a : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70677861735940918b38e52330956b86 : SUCCESS in 11m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79c3803a3ba247d58f0b60a01b55584c : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c446b6b330454f3b942a18f3f34454da : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/a26b5d4d89d940dc931cdec5ed8b58b4 : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd6958c2b994447aa6d03e150d8402db : SUCCESS in 1h 17m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e76d8b206c749feb917b597b4562ecd : SUCCESS in 1h 29m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d891f2401a34ed9acc24527818b6fd8 : SUCCESS in 8m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6522cb2fc6524aa482d3830c4539dc0b : SUCCESS in 15m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/482307779310464ab932a1f19768f6c3 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/68f32669a6354ecead146a97ebcb1754 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e038ecd02ea3426d8dd194b369649872 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0da4bb5a0ab949d9956584ef40514cda : SUCCESS in 7m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d6a956ef235d4ed793f0a269f19fcfa4 : SUCCESS in 40m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8912698a334b462faafb1328db62ffc4 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91344344c11b4e9ca08b80fc81e9b4aa : SUCCESS in 1h 25m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/29d097fdc4df4503a2dcf5098f1fda87 : FAILURE in 1h 44m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b51eb3405af4fdfae68fc13d073050f : FAILURE in 2h 30m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/09fa61bd9a0b477ab2c82d65b2b79dc9 : SUCCESS in 2h 05m 26s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d907de5a15de42d683c98ea7046d4c4c : SUCCESS in 1h 06m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b41032ed4084400ba68c0d3cef829f3c : SUCCESS in 1h 23m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ded7809f027747a7b6eaa4ca596f85f8 : SUCCESS in 38m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/abb1a95dbc9f4ba58cce13469a109c0f : SUCCESS in 1h 13m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/afd61eeeab7344c6ac278ffaf86ebec0 : SUCCESS in 1h 35m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed0dc19f818845dc9e23d2e35e91166a : FAILURE in 2h 03m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b159e9c20a4241bfcd1817d874d16a : SUCCESS in 1h 11m 30s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/939337435b3245149fad7f2659e94810 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94d4876e517eb755b8be05db2114647fbec4b85e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-24 22:31:02.000000000","message":"Patch Set 3:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/398/","accounts_in_message":[],"_revision_number":3},{"id":"4766d9f0c705660f773c026c68445bfef478302f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-24 22:35:03.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-3/FC/ : SUCCESS in 2h 37m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-3/iSCSI/ : SUCCESS in 2h 52m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"b2b4aa1bb949f50e689d9226fc82d56c0b603dca","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-07-24 23:12:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/69/886869/3/check/fujitsu-eternus-dx-iscsi/d98e027 : FAILURE in 3h 29m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/69/886869/3/check/fujitsu-eternus-dx-fc/4854696 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"be630ae5517eb15c6eeb77947f7b8528aca1e200","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-25 02:21:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1646/","accounts_in_message":[],"_revision_number":3},{"id":"5124b15a53ce3fb238b08d4d6f64d69492447d50","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-25 02:41:29.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerFlex/54447e5/DellEMC_PowerFlex/317 : FAILURE in 1h 41m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"556a36dd218e1c30b3aacc8e64acd65a679e223b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-25 05:20:14.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_Unity_FC/842f7ab/DellEMC_Unity_FC/471 : SUCCESS in 1h 40m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_Unity_iSCSI/143ec53/DellEMC_Unity_iSCSI/471 : SUCCESS in 1h 58m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a3d4b56320cda24d937a0820b9ade8b4ad5ea0da","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-25 10:43:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/886869/3/ : FAILURE in 1h 05m 53s","accounts_in_message":[],"_revision_number":3},{"id":"7c2ee20e9aa9f83221445f6e1995330585ec4b02","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-25 15:03:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4f5468a080188ed73d5015d94aa0ee9fa1893dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:15:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942b36cfdf6448748bb14839f6902612\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f53903f4617843fd8811e8514bc13ca1 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2371f111f434455db1725475412c4801 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6fd9b9119f546f8af043ef28380c717 : SUCCESS in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"481f1bc92907f26b8e5f7fefee5ded27dac5461e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-25 16:26:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/3/20230725222454/NEC-V-ISCSI : SUCCESS in 2h 48m 04s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/3/20230725222454/NEC-V-FC : SUCCESS in 3h 01m 33s","accounts_in_message":[],"_revision_number":3},{"id":"cdf22697dd9377bef559d2ff5fc6030d28f1dfc9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-25 16:31:50.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c00786b57eb349089a9423ddcf8c4737\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a95cb9f2cfb9458892916bbceceeeb96 : FAILURE in 1h 11m 16s","accounts_in_message":[],"_revision_number":3},{"id":"39a51e14b57eed13613e0a0d071a09fcb00d5432","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-25 17:04:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/1edda9c/ : SUCCESS in 1h 46m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/3dc3b87/ : SUCCESS in 1h 47m 04s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/883360e/ : SUCCESS in 1h 57m 22s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/69/886869/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/e78be41/ : FAILURE in 1h 59m 05s","accounts_in_message":[],"_revision_number":3},{"id":"d90cf9352fdc133ffc42acb6917c23fce5d15042","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-25 17:38:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/69/886869/3/check/tyds_test/4766cdd : SUCCESS in 4h 37m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f885dfaa6ebb3277bfec3eb49c64266008e03c37","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-25 17:47:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1c6bee0f44ae4d008e2a92a6b9d9d095 : SUCCESS in 1h 36m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2c1814f5db3b42a1b5555038be0310b3 : SUCCESS in 1h 25m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b0406305f09b20af8f1782e8901dd2315a10751a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 18:51:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803934b0f18242c18b0f54c154ebf77a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be200f9b35f14899b5ab5ef6e02cb257 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b8668abbc3f44f8a277ccf74a7ef3c3 : SUCCESS in 16m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ecd15bff21f486abd463c9828375bc7 : SUCCESS in 12m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c6a2592b3704ed98602d966a326f869 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b61419480ef465b82bc0a7fbf462e63 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/7d01de26dc654f039bfc7d704c8c5572 : SUCCESS in 46m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8c4a66ebb634e21933848113be91b38 : SUCCESS in 1h 36m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/837a05c836984c26b92dc41cb8120ab7 : SUCCESS in 1h 09m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f2e7ecc46d845f28f82aaad990dc662 : SUCCESS in 10m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea9fe227edd74d96b56bf6f5f48777be : SUCCESS in 16m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf3bd397ba2a4b5dbccc4a5f2fd97732 : SUCCESS in 5m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a37a9c5fceb249719d4a7ffcd5498862 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/173445c1868d40eea1723596b3fc4cc8 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5977a58675c4e9688139527a8bb1149 : SUCCESS in 7m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e88c245219804529a6489b7c8912bdeb : SUCCESS in 1h 07m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38194fbf8af241428be4ef3ffc89dd49 : SUCCESS in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6a5fefe717234ebf9c12c0c40ec9a5cb : SUCCESS in 1h 53m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/de44a423c6ee4448bf6fc41b99df611c : FAILURE in 2h 15m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/50a28e8f51c241428ed1da8f3766b950 : SUCCESS in 2h 09m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/898b26608a404145b786aa3872e3d1a1 : FAILURE in 2h 19m 00s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/25f7ea4c58cd450a9a4a09129296d25a : SUCCESS in 1h 02m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dc509ba3316840f2aac6ae6cf3712889 : SUCCESS in 2h 06m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54095a30d7ca4030a16bb9a473c5607a : SUCCESS in 1h 11m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f19993fe9f5f4dc2a49e4ffd91826b8d : SUCCESS in 1h 30m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b937fae45f3f46e7bfd2fb9003df51cb : POST_FAILURE in 2h 45m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aafbb8f78254477581bc233936ada508 : SUCCESS in 1h 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fb1e167383e4a0eb25d729e94a7da58 : SUCCESS in 1h 15m 53s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5092639a2c964ed59f8307d87f2f7d1f : SUCCESS in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cb2ac43546db598376bd7eb48e18f5accb1e8f7","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-25 18:54:12.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"41f08acc214708de691fc387f52f114f8125eda4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-25 20:04:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/3/upstream-check/cinder-tempest-plugin-ontap-fcp/7e53b92/ : SUCCESS in 1h 05m 33s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/3655a90/ : SUCCESS in 46m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/3/upstream-check/cinder-tempest-plugin-ontap-nfs/cdf2970/ : SUCCESS in 47m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/2e36b06/ : SUCCESS in 45m 52s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/3/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/37828f0/ : SUCCESS in 49m 43s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/3/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/d7ac841/ : SUCCESS in 1h 11m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"36e8f7b5b9a157a88db0ea3e2bb8ff4fed56a5a3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-25 20:09:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerStore_iSCSI/e844297/DellEMC_PowerStore_iSCSI/None : NOT_REGISTERED\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerStore_FC/46e269c/DellEMC_PowerStore_FC/183 : SUCCESS in 1h 17m 18s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerStore_NFS/0b9454a/DellEMC_PowerStore_NFS/184 : SUCCESS in 54m 22s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerStore_NVMeTCP/b52e4d2/DellEMC_PowerStore_NVMeTCP/175 : SUCCESS in 1h 13m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"c8b200f5ea11416916209d6f520373c520244606","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-25 20:33:02.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerFlex/6939307/DellEMC_PowerFlex/320 : SUCCESS in 1h 40m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"3c51037fc0cee34288809828641b69dd9ab32a7c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-25 20:57:30.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_Unity_FC/652fbb7/DellEMC_Unity_FC/474 : SUCCESS in 1h 45m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_Unity_iSCSI/4239a9a/DellEMC_Unity_iSCSI/474 : SUCCESS in 1h 58m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bc4dc13d391c1eeb7de0380356ca9a43ce0d6c21","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-25 21:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/3/fc : SUCCESS in 0h 40m 50s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/3/iscsi : SUCCESS in 0h 59m 36s","accounts_in_message":[],"_revision_number":3},{"id":"476911f51d62b6575d5ce67db4c353dab91de6fc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-25 23:07:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886869-3/HPMSA-iSCSI/ : SUCCESS in 1:55:39\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886869-3/HPMSA-FC/ : SUCCESS in 1:28:57","accounts_in_message":[],"_revision_number":3},{"id":"cce169fb97153464bc015162f8db498809a66e28","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-26 03:06:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/3/20230726081204/NEC-FC : SUCCESS in 3h 48m 43s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/3/20230726081204/NEC-ISCSI : SUCCESS in 3h 54m 11s","accounts_in_message":[],"_revision_number":3},{"id":"4511ed88b7399aa08c415568155a28367c6e2c76","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-07-26 08:25:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerMAX_FC/77229cd/DellEMC_PowerMAX_FC/187 : SUCCESS in 8h 00m 00s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/3/check/DellEMC_PowerMAX_iSCSI/92fe10c/DellEMC_PowerMAX_iSCSI/184 : FAILURE in 7h 29m 40s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"bc29b4486c6edad2e1d15602519c84add6653747","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-07-26 20:31:30.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5579b351f49cb693306acc3aa6a5e9cb516627d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-27 16:02:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d67c4792ae15cd414ea2a4b3f531495ee3b0df1a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-27 16:08:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aea8cb5c234042b8970137c69f9b1167 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9aa04636513f48458b840650307c69c8 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4a5b83dd803745ca8a931033ca201bf3019570fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:27:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2715ba862a5744408bb261f8809e12da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0306791a2194e5eafbfcf78f3e928a8 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/512b400d9d6b44c09713fd2c8fd29115 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d6f9ab0acec49b289b835374dd39801 : SUCCESS in 23m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a48897c89a2ddc7d2617dcc8ed2a9dab41ab6338","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-27 17:17:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/886869/4/check/iscsi-driver-alletra-6k/b6f006e/b6f006e.html : FAILURE in 1h 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"da9e976e4182b6c56ff5424def285ce1e3122403","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-27 17:23:48.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/aa50730508604bc683680391d5e7b8f8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c4c37bc85409428b9cf9b0cef5cc5d42 : FAILURE in 1h 17m 14s","accounts_in_message":[],"_revision_number":4},{"id":"4bfb93f72adabed9f350f3fcac50f6bfd57ba424","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-27 17:29:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/886869/4/ : FAILURE in 1h 17m 40s","accounts_in_message":[],"_revision_number":4},{"id":"2355ceeb1b16af2f91745a4f0b354916d35f7712","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-27 17:34:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886869 : FAILURE in 1h 29m 16s","accounts_in_message":[],"_revision_number":4},{"id":"01713f86e7de0ccc6a0c742f0a0ccefca4403d32","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-27 17:46:59.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/4/check/DellEMC_PowerFlex/43cd7e3/DellEMC_PowerFlex/325 : FAILURE in 1h 43m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"e22f68f62c85b3a4653954a69215c9a14450c44c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-27 18:15:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/69/886869/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/149a2e3/ : SUCCESS in 1h 53m 54s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/69/886869/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/a122784/ : SUCCESS in 1h 52m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/69/886869/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/7f7679f/ : FAILURE in 2h 10m 14s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/69/886869/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/dd071fe/ : SUCCESS in 1h 57m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ac45e0d3fb27e4607cefc9a91d9098fe7527ed02","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-27 18:22:01.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/4/check/DellEMC_Unity_FC/1a64726/DellEMC_Unity_FC/480 : SUCCESS in 1h 49m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/4/check/DellEMC_Unity_iSCSI/fc9a977/DellEMC_Unity_iSCSI/480 : SUCCESS in 2h 18m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"2ef032143a3e9a2ffa221f39b8acd36d4d4903c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 18:37:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/989b0bc77b0643ab87094710e76aead0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3987f65e5f48368b73a758c9b79985 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b0bd02cecfc44f69bbeff713d03f4c6 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/863b731d81984f3fb83837131f4ad610 : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce610e1e3d0c4181a42269b7126d2c40 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b890910c8148478e84ab833aafd8a3 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/74bc2376bdee468bb83df1814e69ec6e : SUCCESS in 1h 06m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/334406e9ef5a44b6a7b1d5e8ba10ea13 : SUCCESS in 1h 25m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/431752eef24c4da7b0bb580778d0233c : SUCCESS in 43m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c30d74d394647299b2da83eb3700198 : SUCCESS in 8m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc06c320399c4aba837100f7a9bb4c23 : SUCCESS in 18m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/daf814747ca540d2849161340624869e : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ab1542ae5ab4073a1d424601ce59277 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/877ba3b632a14a599889c3247e783fe0 : SUCCESS in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ad3d7ca026a42f5b8a06c151d2fd429 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f95afc674ed4ca6ac666d6deb9b5be5 : SUCCESS in 50m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d53b07b3d0c4bac9ae06b7f409f29a8 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e8277ed744947f2b2696f4b1c86c8b9 : SUCCESS in 2h 02m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e3db6cc3edb645c9ab3d6d5c0dfaa3aa : FAILURE in 1h 26m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7abacafeab2c4c0ea000cc5b6f593cf0 : SUCCESS in 1h 52m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6e4a6101c6a24ca7b52a6e4561d3380a : SUCCESS in 2h 26m 21s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4115dd01acc148d88f4bce37ff2721a3 : SUCCESS in 1h 15m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e37151a1a6614459bc77d27c57a6251a : SUCCESS in 1h 58m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9c47a58681eb4c148813707a26fa7090 : SUCCESS in 32m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/faaf1637a7da43e5a97f48eba68c0069 : SUCCESS in 1h 00m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5a63fcc03b5f4f70ba18a4337ab5fc38 : POST_FAILURE in 2h 10m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e6b6b8572b745db826bb8a194a66fdc : FAILURE in 1h 46m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/022601da534a400b8f8421a5a13eb97c : SUCCESS in 56m 00s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/71f17c13e7d742058ad7065561a509ba : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"778ee7f130746cf84e6ddb7130cb26a5ea2e22fc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-27 19:13:37.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-4/FC/ : SUCCESS in 2h 56m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-4/iSCSI/ : SUCCESS in 3h 10m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"09c9092bdb3e54f01cb34fbf4d33a272480a555d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-27 20:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/4/fc : SUCCESS in 0h 38m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/4/iscsi : SUCCESS in 0h 56m 50s","accounts_in_message":[],"_revision_number":4},{"id":"e263290a37474c93bb9b81cbf70ffed17535846f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-27 20:04:56.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-886869-4/ns5_nfs : SUCCESS in 3h 56m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-886869-4/ns5_iscsi : SUCCESS in 4h 1m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"78aa8f8340d055722a4b207dec8b31e61476c5e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-27 23:53:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/4/20230728060918/NEC-V-ISCSI : SUCCESS in 2h 41m 11s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/4/20230728060918/NEC-V-FC : SUCCESS in 2h 43m 46s","accounts_in_message":[],"_revision_number":4},{"id":"0f56b2e2c7822df26f0675b0c3f0f7134cd9c0f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-28 00:17:38.000000000","message":"Patch Set 4:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/422/","accounts_in_message":[],"_revision_number":4},{"id":"4db9a0927725c36f0ae0ed8d28acf59a2c95ea28","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-28 06:51:31.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cb7aad19e21ee6ce7714c281e3343000de39d1cb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 08:43:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/4/20230728093808/NEC-ISCSI : SUCCESS in 3h 39m 01s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/4/20230728093808/NEC-FC : SUCCESS in 8h 05m 31s","accounts_in_message":[],"_revision_number":4},{"id":"6ab9359339f0e148106153451a5e382303609838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-28 10:48:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"832b94e343f54b576b6439bd0d5bc5148e7616f3","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-28 10:50:29.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1624ea495121e7818eefe8e54b31659a81d673a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 10:59:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/527b3cb9d795417dab4c6ebf7fbc771f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4d21f4c488d4e619c4743dd4f8521b0 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7afd27215da49b1ad51cd751dbdf760 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74cbc71693b549c5bfb9e74987a0a59c : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aafbfaef713e7132df491ab785e1866cbf26747e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-28 12:01:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/886869/5/check/iscsi-driver-alletra-6k/bc6c6ef/bc6c6ef.html : FAILURE in 1h 11m 16s","accounts_in_message":[],"_revision_number":5},{"id":"309a85fd396d070ecdfa5042f6cd84fa4f2b5ae2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-28 12:14:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/886869/5/ : FAILURE in 1h 16m 40s","accounts_in_message":[],"_revision_number":5},{"id":"a8c75d2f1e23a9d244ebc88d7ce8626a67d5104b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-28 12:20:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886869 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":5},{"id":"456754904aba956cec44a209334d3613ff9c8cc6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-28 12:22:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/78ab65c82772411ba2ae90c4554ef31e : SUCCESS in 1h 31m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3fe9976c1e7a4cb4a9d42287d631cba8 : RETRY_LIMIT in 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"e19c426d3d8a492cbe41bc285889e51581ba0251","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 12:40:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/5/check/DellEMC_PowerFlex/0fb7ad6/DellEMC_PowerFlex/328 : FAILURE in 1h 51m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"2dea269cd534b7928456fa5212c5661db31e05b6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-07-28 12:57:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/69/886869/5/check/fujitsu-eternus-dx-iscsi/0dfab95 : FAILURE in 2h 07m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/69/886869/5/check/fujitsu-eternus-dx-fc/5f5a2a9 : SUCCESS in 2h 02m 03s","accounts_in_message":[],"_revision_number":5},{"id":"0ecf85f720d48a4d8fd68584b6a99f77813725a7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-28 13:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/5/fc : SUCCESS in 0h 31m 47s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/5/iscsi : SUCCESS in 0h 45m 30s","accounts_in_message":[],"_revision_number":5},{"id":"60c88b6965379935d9da80c1f15aa45cc1441a65","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-28 13:04:46.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/5/check/DellEMC_Unity_FC/fe10f9c/DellEMC_Unity_FC/483 : FAILURE in 1h 58m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/5/check/DellEMC_Unity_iSCSI/3608f4b/DellEMC_Unity_iSCSI/483 : FAILURE in 2h 15m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"474df5f7221de5bca304b616222dac8d2542d38d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-28 13:19:39.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9f68484fbdcd4fda91cf56244a773fc5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c9baa743979a40b3964328a45ccde681 : FAILURE in 1h 14m 47s","accounts_in_message":[],"_revision_number":5},{"id":"5b843325d2ef1720153fe2a32f44aae8fd5ebb51","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-28 13:53:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/69/886869/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/c7c0f10/ : FAILURE in 1h 45m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/69/886869/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/e0cf49a/ : SUCCESS in 1h 49m 12s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/69/886869/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/12ff361/ : SUCCESS in 2h 00m 43s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/69/886869/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/cf6d8bc/ : SUCCESS in 1h 43m 54s","accounts_in_message":[],"_revision_number":5},{"id":"5e35d1b0858f1c4dabcce5aaabac0ae424b8a661","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-28 14:24:09.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-5/FC/ : SUCCESS in 3h 35m 12s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-5/iSCSI/ : SUCCESS in 3h 8m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"2dd704545c874ec5332d90e56a897ee2e90e8843","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-28 14:40:07.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-886869-5/ns5_nfs : SUCCESS in 3h 49m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-886869-5/ns5_iscsi : SUCCESS in 3h 51m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"31e0086ee0400a13c599d9bb6bc4e1c9aec60495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 16:42:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e74e4e3e927e4a43be163a31ddea607a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4dbc2c5da70408d9f8d36e167d55f57 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cdbdfe97d884b8495eb139b9bb4b8d0 : SUCCESS in 15m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e33402fe8cd4e09a7d04627e96af252 : SUCCESS in 12m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3248f59380b94bce85f7c1615b4141b9 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8041098622ae4b09b3420fd2872987cb : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/105f930a6bf14b169083c3a98a61c27d : SUCCESS in 59m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0da170b2fb84078a4be3fa00f6a31f6 : SUCCESS in 1h 40m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3713d22411d64505ae596437832c33c9 : SUCCESS in 1h 01m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3ab4564216f4122b067de0e3e1ff957 : SUCCESS in 9m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/36718427881c4c269ca91fd049fcb872 : SUCCESS in 19m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ff06241952b94c04b77a61a420232041 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aef6cd37af0e48f39281c68075b369b9 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ccc303827cc043ed97fef1149093f6f8 : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/355be0e3fa804cd2996c9dde64c13766 : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1314cedcaf4d4a6caee8d397090f45f3 : SUCCESS in 48m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78cf1f00b8b246b484f66e452af787cb : SUCCESS in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55acd0637ac745df90002622b0f59426 : SUCCESS in 2h 03m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3d10486a88d4206aaa8315ee90d48b3 : TIMED_OUT in 3h 04m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a73b371f9a394783a595c3a965a6b910 : SUCCESS in 2h 22m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/294306ce1355488bbc671e8457672cde : SUCCESS in 2h 31m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98a7419be9374b4785e6334f4ca0d6c1 : SUCCESS in 1h 03m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8aea52a79a3c441d8c28d48b8efd3b84 : SUCCESS in 1h 50m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e74a928d178541dbabc9d0cf9a6b5daa : FAILURE in 1h 01m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1bce99bd45384b529e44a0ebf12ed9b3 : SUCCESS in 1h 34m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/32b3025b7cc14f5f8b4a1a827fa4a16a : POST_FAILURE in 1h 50m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad2a8127ceea4298b297e68bf276a47e : SUCCESS in 1h 25m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8c9b7e7d5cf41b7ae037ec221fddc43 : SUCCESS in 1h 15m 44s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/02cb935d02c84404a90fc91fd91a11c7 : SUCCESS in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70e1ee70222368a1c98536b22872bbefab3ad86d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-28 16:45:04.000000000","message":"Patch Set 5:\n\nBuild failed.","accounts_in_message":[],"_revision_number":5},{"id":"5fe2c9030cdb4ca9afa734f529806540ae455dd9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-28 18:09:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/5/upstream-check/cinder-tempest-plugin-ontap-fcp/478211d/ : SUCCESS in 1h 04m 40s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/dc12d89/ : SUCCESS in 1h 26m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/5/upstream-check/cinder-tempest-plugin-ontap-nfs/ae73f40/ : SUCCESS in 44m 52s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/11913e8/ : FAILURE in 1h 23m 22s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/951db6d/ : SUCCESS in 50m 02s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/6dfc58c/ : FAILURE in 46m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"19ce49f43bd72cb94d04db4bcfe1f67302f16818","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-28 18:11:13.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/5/20230728202258/NEC-V-ISCSI : FAILURE in 6h 31m 43s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/5/20230728202258/NEC-V-FC : SUCCESS in 6h 48m 12s","accounts_in_message":[],"_revision_number":5},{"id":"bf88afa6b1b5a6220909167fcf47cf44acc36e4b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-28 18:19:45.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/886869/5/ : RETRY_LIMIT in 2m 36s","accounts_in_message":[],"_revision_number":5},{"id":"8a3f64cad13ead23527361ebc44c75c82b9b2630","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 18:26:23.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/5/check/DellEMC_PowerFlex/bbfb746/DellEMC_PowerFlex/330 : FAILURE in 1h 43m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"f9488e60e0c0cdd1b368db5deecb13a7b48681ab","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-28 18:44:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/5/check/DellEMC_Unity_FC/19cf377/DellEMC_Unity_FC/485 : FAILURE in 2h 01m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/5/check/DellEMC_Unity_iSCSI/5712717/DellEMC_Unity_iSCSI/485 : FAILURE in 1h 52m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"7a5f70e5d36c01f51f624d0480c1e92a588d042e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-28 20:56:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886869-5/HPMSA-iSCSI/ : SUCCESS in 1:54:03\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886869-5/HPMSA-FC/ : SUCCESS in 1:29:19","accounts_in_message":[],"_revision_number":5},{"id":"13f6d582064be252498263d6c58e812f1fd6502e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-29 08:30:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/5/20230729092250/NEC-ISCSI : SUCCESS in 3h 57m 06s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/5/20230729092250/NEC-FC : SUCCESS in 8h 07m 28s","accounts_in_message":[],"_revision_number":5},{"id":"07d46f518a55597cec9fbe17ec4c6d4daf2cd575","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-29 09:14:50.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9d922b742775c07c102a8fd583dc8eee0454d6fa","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-29 14:10:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/886869/5/ : FAILURE in 1h 00m 14s","accounts_in_message":[],"_revision_number":5},{"id":"33862fc55dd7c240214109ea25a19dce45720f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-31 11:39:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"877e9af2a98246a0ace9a638f0e67030afe45c53","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-31 11:40:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab2c8f5e932710ea1777e015fc7a051d5de93cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 12:02:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a3c2c7c3c384e6eb177aee62d990c50\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79c2ad0dc45943dd85bdb69f225a2ddb : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75dcb4b181a342068b633f15b9561b76 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a10ea9298e74a7eb850b7eab5383db8 : SUCCESS in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ee61028f90a4236dab5f42fc1eb8cc050e55a70e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-31 12:37:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/886869 : FAILURE in 55m 30s","accounts_in_message":[],"_revision_number":6},{"id":"3a3eb0d4296f9c158676fa3dcfbb417d0a9ddb53","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-31 12:56:23.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/830a4a106c304fb3b44ae550307a7dd9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/38deccc1483c47d0858d18762609304d : FAILURE in 1h 13m 48s","accounts_in_message":[],"_revision_number":6},{"id":"9be295d7a44f71b1c1fbebfd6015e974d8f76cbc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-31 13:06:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/886869/6/ : FAILURE in 1h 18m 07s","accounts_in_message":[],"_revision_number":6},{"id":"e5a4b2b3bf5f0b54ebf140b29f5a9e7be1664378","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-31 13:10:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/76f22fbf98f447c3ad02e0c6a754ce25 : FAILURE in 1h 28m 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/10ae1510ea204bc8a4cf46b2dd70c4c1 : RETRY_LIMIT in 49m 44s","accounts_in_message":[],"_revision_number":6},{"id":"998d3e5b48e158acd756ca29577465388a06f36b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-31 13:11:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/886869/6/check/iscsi-driver-alletra-6k/7b91929/7b91929.html : FAILURE in 1h 30m 17s","accounts_in_message":[],"_revision_number":6},{"id":"43d31e96d5f32971564f7153996a2b79ce5310d9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-31 13:26:10.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerFlex/0b5beae/DellEMC_PowerFlex/336 : FAILURE in 1h 46m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"4c9a44eddf43da11490175f2db3e09ab636c323a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-31 13:41:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/69/886869/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/8cf2e89/ : SUCCESS in 1h 48m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/69/886869/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/324f6fe/ : SUCCESS in 1h 53m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/69/886869/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/d9b1a40/ : SUCCESS in 1h 51m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/69/886869/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/a99bf2a/ : SUCCESS in 1h 59m 55s","accounts_in_message":[],"_revision_number":6},{"id":"65b20a5db237b8d77ef9c48cae37a65cddaa5a1a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-31 13:44:23.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_Unity_FC/60ee4be/DellEMC_Unity_FC/491 : FAILURE in 2h 04m 41s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_Unity_iSCSI/4b8ae01/DellEMC_Unity_iSCSI/491 : FAILURE in 1h 56m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"fb9d9ba5f42f956bec6158b14f7ca02fe8c3bf20","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-31 13:58:04.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-6/FC/ : SUCCESS in 2h 18m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-886869-6/iSCSI/ : SUCCESS in 2h 5m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"87af15ca0222d3fbcb676ed7d9da2e7ba669a0e4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-31 14:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/6/fc : SUCCESS in 0h 39m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/69/886869/6/iscsi : SUCCESS in 0h 57m 29s","accounts_in_message":[],"_revision_number":6},{"id":"df1387fb0eea350433f69e1b110fa751a7ed2509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 14:26:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82a4574dbc30433387ce4564b97e8447\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1a3d6244f8246bfb6c6c8da2b7889e0 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6a9a005d3964552a2c337dc4b47f375 : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59bb36658f3246a5b17b367490cb5787 : SUCCESS in 14m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d12b5303ef9748b0b4737c6acf53adb0 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46478b610cbd45b392b07f8f2fcfc3bf : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/33b57f2e2ce84d92a500aa849ab2674b : SUCCESS in 51m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5406a1bac11440358c042510f337afed : FAILURE in 1h 06m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89d686b882f24072bd8ded92d0995878 : FAILURE in 1h 02m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98a26a9ff324471e9e78bca8cc1976b9 : SUCCESS in 10m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4511d04c7fd643bdb3794c5d958de6cc : SUCCESS in 18m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92f8da306efe4cc087430df88ce5cffb : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/794533a7002f4c7386bff757d021b3ca : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4fc5690313584d13b96efeacd75fc2f4 : SUCCESS in 5m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ab3ef881935404b8834bdf851d90b1a : SUCCESS in 6m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3de753bb627249318e4dfbfc52d14f27 : SUCCESS in 39m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7fd7660eb05d4f698aacc48e66678aa2 : SUCCESS in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce27122cd8c843e08e3842628255f97d : SUCCESS in 2h 15m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/70138db494b24893b850ca177597a1d3 : SUCCESS in 1h 54m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/938ae2c635c2464c822e041182878de2 : SUCCESS in 1h 38m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e80e471aec374f53816eff94d110608b : FAILURE in 48m 01s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c454d76e39da477eb4485291db70ec1c : SUCCESS in 1h 27m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9859d69c882b4476b9b2e0660f11d8b6 : SUCCESS in 1h 49m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8a26b67a03894c5789f796da68a3a1cb : SUCCESS in 53m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/423406b11e6441689508a64b9ef65e9f : SUCCESS in 1h 18m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eb760a8e417a4b82b84c7fb75538b448 : FAILURE in 29m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0471dc5f8d7643aba0062efd794380c2 : SUCCESS in 1h 00m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8354f42b89ae41699b857f7eefd5b8d1 : SUCCESS in 55m 04s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/ffe8a51b3d294819ad8f1e9df4a9a23f : SUCCESS in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"39c9a94490360f0ba3551d55c576be21690a1d0e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-31 14:49:04.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/6/20230731210241/NEC-V-ISCSI : SUCCESS in 2h 39m 01s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/6/20230731210241/NEC-V-FC : SUCCESS in 2h 46m 19s","accounts_in_message":[],"_revision_number":6},{"id":"33f47dbca2f42147aa1cdf2a71b7c25f50e18a3a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-31 15:32:32.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-886869-6/ns5_nfs : SUCCESS in 3h 48m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-886869-6/ns5_iscsi : SUCCESS in 3h 52m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"de0b4be92cecb6123769bae1d453f8b8d5e7b8f6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-31 17:00:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/6/20230731221425/NEC-ISCSI : SUCCESS in 3h 42m 00s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/886869/6/20230731221425/NEC-FC : SUCCESS in 3h 46m 31s","accounts_in_message":[],"_revision_number":6},{"id":"d90b24b4cdbd70f96d20625f8e433229f70f58a8","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-08-01 13:10:05.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4427a66672b11668044fc273b7525cb0ad068963","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-08-01 13:10:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"07a26b54487b9cde14c03d6e286b98dd1b61d17a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:34:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f24d1e26cab43298a5706ec5ee0a2e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15023a8cfecb43f2aff8e0046a59e425 : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd97abdcf1914e898120a40b1ff97b1b : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea46e6b240384841ade0fedf475cfedf : SUCCESS in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2d4fc3439af54d1c626342172a1423b99d16e774","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-08-01 14:05:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6bda178960dbc3a072c9916467fd38c2fe860743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 14:16:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab7845ba0d4462aa3cc3e1ded546167\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea58a05f5ccb428ea3f0d6b5a652c0f0 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c204f75891e417497823e6d7c844fed : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7abda2a7f3aa4a41b4b6f57c6d3943a5 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1e4355e6bc13c7a1bd9784ed065f891e23bc448a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-01 14:30:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/191cc5b534f348a79783a76b8e8693e1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/66490b9086ae43e5afe03fa8d8efaf92 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":6},{"id":"b3b27c8d6afdb5b7f714b8ad7cf72ffd396c90be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-01 15:04:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/009258fa85324e618958fd48875efb47 : SUCCESS in 1h 43m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/48b76e924a8d4e7cb81bf2d8831ca243 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":6},{"id":"8cba4317596c07688f75b1b260682bbe6d4d8ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 16:29:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d13948ca951345f1a0062db576a508f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269c1cc2ecfd474b80999a8dd7cc9048 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c4d93cd801d48f98254e682e9bfd76b : SUCCESS in 13m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aeb22058c88c42fb84feb4d9215b306b : SUCCESS in 12m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2b2d5211ac742d9a9bcad71015f7be3 : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8bf61196b046c5a91b4c25243aa744 : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/f865c89c97cf442da6beb6701edef499 : SUCCESS in 56m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3508bcaf4b144a1b00cd15d5e6b0c9e : SUCCESS in 1h 38m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/296012f5d89444b185151d0e87b71235 : SUCCESS in 59m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3aabd4642a47fa9417d05317573e93 : SUCCESS in 9m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6c7ad2c57d6045389706c7bbeb3b4276 : SUCCESS in 18m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f8521a8b14984b43a88387acef9cea2b : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6492b655f8714e218d89e6a71b957781 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46c88afef0e8479c94454be82d426d7b : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/237298e87c3b42c58cd53661335028f5 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4488a8ad84dd41c69601acf883e772c0 : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8276c5debb3b49c7ae0822a22fdc5637 : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/838fe98ec5914e99b9b972d8e213a088 : SUCCESS in 1h 56m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b574c7bcb4841d7a0ce460caff565b7 : FAILURE in 1h 29m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/41885044a736447dad1747ddc05fc908 : TIMED_OUT in 3h 16m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/64195f8020b8427fbad97726ae838cd8 : FAILURE in 26m 02s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d44a7d3d0e3848d8b24f39530fea93a2 : SUCCESS in 49m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/985481e406954c32938f002c413d8547 : FAILURE in 1h 58m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8f197310c1c4560a66a09280501c104 : FAILURE in 52m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f2dd42d08bd24fe2997afeaef64e2fe2 : SUCCESS in 1h 23m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7292e07cd1bd45dd83fc599f29a3ab7f : FAILURE in 47m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ffb425c6d1f465e93bf95eee8ec84d7 : FAILURE in 1h 25m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e60ba51007c34c5eb08821b51d248d3b : SUCCESS in 1h 11m 13s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/baa7ac9c66874909bac9832ddb2446fe : FAILURE in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"576c40f2529d0589e48598d0838d9be31bb1e347","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-08-03 02:31:14.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/69/886869/6/check/tyds_test/817d51a : FAILURE in 9m 52s","accounts_in_message":[],"_revision_number":6},{"id":"f4870566682188dfa8168946d0ceedf9f305a26e","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-08-03 03:47:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"47c26bbf9f98bed0b2b60fb40cc9951d3aa84618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 03:59:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/010f34aa9e5c42bc8737842777db94bd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13e3d6089a4e4d95aefb2b782f5f2c4c : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06fbca2f38c540a08dda0ceaa9c2f8d9 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8184d9ec86bd4c4fad98a5adfb7f5409 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca65255dfef26cb0458c02d52e46077d99beccfc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-03 05:05:40.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/31369eaa63984288bd345031b7e6a397\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/eb18af6bd4e144ad99ef3ba6c83f3d8c : FAILURE in 1h 14m 53s","accounts_in_message":[],"_revision_number":6},{"id":"6145a04f1815640ab1c9fb7d902c3b8d99071b1e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-03 05:28:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0c58fb7f1ef74e81b198a23e4dc87a4e : FAILURE in 1h 36m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1cf477f3274242f6ba01fb5c7ab2e584 : SUCCESS in 1h 38m 56s","accounts_in_message":[],"_revision_number":6},{"id":"8ed112cf0e80c66e2dc9e1ac0b37e05dd93a99c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 06:57:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6cc674d54f14a8d8364fded177e2439\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d468893a02e942bc826994db1c478be1 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5cea0803129481b8b085c4dd8fcf57a : SUCCESS in 14m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3e6b4eb5bea4efaabc1b27a3053c988 : SUCCESS in 10m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8120958774d45afa563502cb5fe6055 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af9b38b73014587beb5cd4dbe7e77dc : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b9da81c889064949be02b765dbdfb6c6 : SUCCESS in 1h 03m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c9a7ed603b44a2a883f3dadf5246072 : SUCCESS in 1h 24m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf2419419d3642bebb9492d5045cfc87 : SUCCESS in 51m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f1e7c1a68d44f48b5518bf905f3cda2 : SUCCESS in 8m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/697f40d52dcb4dcd822a01fcc3646589 : SUCCESS in 16m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4703e0bfc75442b3a4337644a142137e : SUCCESS in 6m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ab70ace2f974acbae14154b5e8ec334 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3dc887327390470c98ccd95a7fa58d02 : SUCCESS in 6m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db1d9b2d0b0845a39140e431c827a853 : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b03b67e41ad4903844732bda507c76a : SUCCESS in 54m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d9bcee83b8a43fb83b61ae0a5c8ebd9 : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9023f72cf4184acf9a901c2f4ac83872 : SUCCESS in 2h 04m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/52cb35ba628747bd840d2a56f59e3626 : TIMED_OUT in 3h 06m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a51459263eb42f0a36aef4a75d76ac0 : SUCCESS in 2h 33m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e77a998dda134bf5b75bf2e8d8d6b145 : FAILURE in 38m 12s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/95cc6657b2454357b014ed026b3f7e3c : SUCCESS in 1h 19m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/78e5db6605f5409392aba9da2557efe8 : SUCCESS in 1h 23m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8a89e04ba86e4a1e9138ae3048f012e7 : FAILURE in 59m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e02644d08d74bb7b418916202858194 : SUCCESS in 1h 06m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/eedf7c814a2f47a3854ff15aaca84983 : FAILURE in 45m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/807a4e7d56394b699421dfb3d5603afa : SUCCESS in 1h 39m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e3e8640153c46bbbcf6f1c39731ddb9 : SUCCESS in 46m 54s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/21b2bcdf3c584897824f2fca6f7a357c : SUCCESS in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b3eae7af17f7116749cb46f14af782b15a938eed","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-08-03 06:59:04.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"2adf76b2b838c711346c11a4d7f1313214116cc0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-03 08:19:17.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-fcp/1c064c6/ : SUCCESS in 1h 06m 09s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/78198fc/ : SUCCESS in 1h 21m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-nfs/81ecd5e/ : SUCCESS in 51m 22s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ff38cc0/ : FAILURE in 1h 17m 12s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/af4c02e/ : SUCCESS in 46m 55s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/4ede5fb/ : SUCCESS in 47m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"7ac10998de7420e0b894e93b7fe2176f76045911","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-03 08:41:34.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_iSCSI/ae95bb5/DellEMC_PowerStore_iSCSI/188 : SUCCESS in 56m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_FC/5a043ab/DellEMC_PowerStore_FC/197 : SUCCESS in 1h 16m 44s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_NFS/a0227e6/DellEMC_PowerStore_NFS/198 : SUCCESS in 50m 39s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_NVMeTCP/2107d73/DellEMC_PowerStore_NVMeTCP/189 : SUCCESS in 1h 09m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"032061b01c92b77f04fa6fab7d923f526fa0dbb4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-03 08:45:15.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerFlex/f816f40/DellEMC_PowerFlex/352 : SUCCESS in 1h 38m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"a40924885ba2c4df7dfba12e4e0cfa0a9fca579d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-03 09:13:53.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_Unity_FC/ac73b14/DellEMC_Unity_FC/505 : SUCCESS in 1h 48m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_Unity_iSCSI/cb42fdb/DellEMC_Unity_iSCSI/505 : SUCCESS in 2h 16m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"740d49a63133b4d88ff9b4248390837aeb9bcc41","author":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"},"date":"2023-08-03 13:17:03.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9bac8514b0b56071389b182cf9acc35e28fa0403","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-08-03 13:49:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886869-6/PVME-iSCSI/ : SUCCESS in 2:08:39\n- PVME-FC https://os-logs.seagate.com/dell-ci/886869-6/PVME-FC/ : FAILURE in 1:39:01","accounts_in_message":[],"_revision_number":6},{"id":"a69a7c5dbebbfe52286e7fa41a8517cede3081aa","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-08-03 21:53:22.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerMAX_FC/d08e5c0/DellEMC_PowerMAX_FC/193 : FAILURE in 8h 10m 59s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerMAX_iSCSI/6a0aef1/DellEMC_PowerMAX_iSCSI/190 : SUCCESS in 8h 21m 19s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"dba3b5e6739d239952146809b63b262dff246e52","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-08-04 14:03:04.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7185fbd159e15c18ca844d801c8571d4a2b8ce8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-09 12:41:15.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"de493a9682f0363d056e559d5adba3746b9d0c1a","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-08-10 11:55:18.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b9b2c114696bd65e1b12a9a50ed406e5ad0fb28","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-08-16 14:53:08.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"05495a81732d5dc53e2d7088fcc54d47ab659b6a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-09 08:16:44.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"38e8f19232440b500ddc36b3a8e2634ebd8a0028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:26:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1c9d3901220cd9b22f9c42fb3a2cd3d5260542bc","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-09-11 05:56:30.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1120b883e7a2d949f55747fd6cda785157d0b03b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:59:57.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f6e9d44fd6b4ccf8032d9f0c78f03af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f179b5c5894491086d2844e34de833e : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3b1ff67405244f4adaf054683be0272 : SUCCESS in 11m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc34f820a12c485db26d6396dcef14e9 : SUCCESS in 13m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffa6943cc4934fbfbd9db756477283d9 : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a00bd2703b4919bd74a7cf0c5a3296 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c0f5e5c0c93b4131b9d51f0923485206 : SUCCESS in 1h 08m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad1e7f6a731847e680f7503c60fcf4c2 : FAILURE in 53m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6030a35dedd849849a512ad4c56281b2 : SUCCESS in 53m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93d5e02eccda49dcb9934e1f58827a9e : SUCCESS in 7m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1e0621232fad4bc1bc462463502a96e3 : SUCCESS in 1h 14m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51d0fc8d44e6455684a5e7ea701c02a6 : FAILURE in 2h 26m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/674174aefe7a4771af7dfce638171a1c : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":6},{"id":"6edac891083d63be98e3f932504102e75adec73e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-11 11:00:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"932fc982e4f9e9919cddc92fa26eda748dbe020a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-11 11:09:03.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8578bb32d6aa47448d5f843bb5543443 : FAILURE in 7m 01s","accounts_in_message":[],"_revision_number":6},{"id":"38057274e8ff86cd6c43eaf7e48ac9b897c0591d","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-11 11:52:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2530389687866c248de7ed371dd6caa7787421c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-11 12:01:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1e5346f000b1458292dcaabf87685418 : FAILURE in 7m 06s","accounts_in_message":[],"_revision_number":6},{"id":"33b90f48a9ec9015af5c2dff4d949d374524313a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 13:36:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76349b09b5be4a899a9df2952dda5fdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1029f0a496cb460d8c955b15200bd69f : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43b9b754cfdf4f2094086f4ef1e49f64 : SUCCESS in 11m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6a3d427b0c64591984270453cac149f : SUCCESS in 11m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b096785e4a5b461482b10d46b861ecd5 : SUCCESS in 13m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6647250aad184e36890ccc2ebfbc35ec : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4c627a54aa4bcf84eea6d1bcd1e453 : SUCCESS in 15m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f48afebb0d6b44609b24feb9ca455391 : SUCCESS in 47m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee0b6a0d96c04e49a6e7dde04165042e : SUCCESS in 1h 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0de2f96ecde475ab9d434924077e296 : SUCCESS in 1h 03m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dff6898bfecd41fd809a0ebb517b0727 : SUCCESS in 7m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/efe6edf190204618a4653baf6f1190b9 : SUCCESS in 18m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f85d690c0549429bbb8ecf013c4599d8 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29c92e0ea53e47639b1495c262987eae : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f2545d7a40645929b56115fc824b288 : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b057e7c709e42459f5b04c36e8c571f : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1709cca671b7485cb6d4a7d7fbf8e082 : SUCCESS in 31m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab4e0de9524542e3ae36305f6af754ad : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/baaf3256255f4cc28b5b17c42a1f622f : SUCCESS in 1h 41m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/152483a0e4154219b880c41905b85a67 : SUCCESS in 1h 47m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/be78f32181f8454cb53cd04fad78c47d : FAILURE in 2h 28m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a2f74019592a42e798f7444d444416ee : SUCCESS in 2h 22m 17s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3ca1ab257354357a520f8f06bb98cf7 : FAILURE in 1h 01m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5704de44b9240abb81248d1bab9bf4f : SUCCESS in 1h 43m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/053b9b594c7849bc9c7371d88519c453 : SUCCESS in 58m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a775a211175f443f98127dc941c75a71 : SUCCESS in 1h 25m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3030308552c244348453a07889f1d866 : POST_FAILURE in 2h 18m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1ce95f5442a4d84b699245beda04ff0 : SUCCESS in 1h 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77ed14d50df24654bb825a96cdad3913 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":6},{"id":"4738fc2c77ef1e3593cfa6e90b5ffda99cc69a75","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-09-11 13:52:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f0f11e29f4469f0bdbb3b0330a2942478561034a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-11 14:01:34.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/72d12b998ebf4d21ae487dac74df60e9 : FAILURE in 7m 08s","accounts_in_message":[],"_revision_number":6},{"id":"8852a70c5a6400f28c01f1e3c88a80caf782e691","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 14:34:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56ea9f4b364d4c73b7a1b709b6ba06e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d6f2f438e19403d8a2429edfd097bcb : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fab95e56be8a448daee27f7dd10092a7 : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/535a18c338bc4029a5480a00441aab5b : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47ee9daf83ef418ab8c3889067404fcf : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"22dee93dc43336c4f49b51d717dfcff555c90462","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-11 15:54:31.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1d7e7a2ffb68455fad1ecbc9e34b765a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/be11a5561a414e12b2422764013ccb86 : FAILURE in 1h 18m 13s","accounts_in_message":[],"_revision_number":6},{"id":"069d546916fb41c087200f2f933ceaaee99e3d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 16:33:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16eef5d37c34d44bd53476097ecb736\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd98542e453e4d909a3571ec8eeca727 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/250c891d070e4fbc8ad6fb6938371321 : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f55c121019084370b39e3a5574347abd : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ac87490e3944a9f8fccfcedf75214b4 : SUCCESS in 11m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cae067bb5fa1421bbb4a8ca82b701036 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce0d04c01c2494193860625b2569456 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/1abbe8e0bc9d410aba76232d89e1667d : SUCCESS in 51m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6e94d00fb7d47649bc791ccde1bb7f0 : SUCCESS in 1h 26m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d167ef606a64667a8307bcc6c928a2e : SUCCESS in 1h 06m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87aaedeb99294149877a8deb00abf010 : SUCCESS in 7m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d1c7951fde34c73ae4c882c9cb93038 : SUCCESS in 14m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a26c38b35cd844239b7fdb66d95b342d : SUCCESS in 6m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/585b7964cbe6482793c086fd1e159a74 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4013558a1651481293113dcb2e9baf53 : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09f7f428bf2b4e94a39b95855571ee51 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f0e6f463c38845dfafe3bb8768d4862c : SUCCESS in 52m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b1e42f2fb50e43f79cc92af3fd8b910b : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee9da090becd45a19c77be1456e75d96 : SUCCESS in 2h 02m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4cf555c4007b45fba4287118eb6e86b2 : FAILURE in 2h 04m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb654b8e38ab4e96b50403601ff6ef76 : SUCCESS in 1h 34m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/57184b6b2d964b5b9da7608901f41510 : SUCCESS in 1h 54m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/33c12bd6d0d84e60913c078ab9c68954 : SUCCESS in 1h 00m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f4283d2a0de94dd9bf59d61ae1e2d671 : SUCCESS in 1h 48m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e87c7f67da7543e2a1cb20db7e3a7d0f : SUCCESS in 45m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c918a07e8176453899dd95ae6521834f : SUCCESS in 1h 29m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9212044f33474ef58039e3b0a437ccfb : POST_FAILURE in 2h 36m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bfa58d3f3e7b418f953d2478e3dfdd66 : SUCCESS in 1h 37m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f80d6cd70b324761bf36ff6227dc4ccb : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":6},{"id":"cb8067f74c80f171c73552c5bd1839aad044bee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 16:34:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"61cde5003b5215c53d59cdb728ba9f502da1ac17","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-11 16:55:30.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerFlex/cfad7bc/DellEMC_PowerFlex/585 : FAILURE in 21m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"e2f6405f4bacf0d4686a825a1216f33bf1a1576d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:37:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e4378bbbcd94817ae8cd54b9e5731c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949809a1c1b840af9d87a94ff1f40b0f : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b82d73dc23a4b1a9bafea947e01c849 : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61054b5ef6fa460eaff43a7e56c5e3ba : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fdc4a976e964dde9290415ec95e1e10 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/272c0222c1e04c9989fc442805237c8e : SUCCESS in 15m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/406142d3dff340a19de1dc3feb916f33 : SUCCESS in 47m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04cbdc2993fb48c4af85e0a568231710 : SUCCESS in 1h 29m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/389603ca15aa40a5bf0197fbc542990b : SUCCESS in 1h 03m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2db69d65888943a691d7dbe8f32a26f1 : SUCCESS in 9m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c56eaf4090f408aa62bcee2d5a3ee51 : SUCCESS in 1h 16m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fff06698dc6c411bae790f345767a0b5 : SUCCESS in 2h 03m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82325de64ffe422087e485c06fa1c902 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":6},{"id":"13d8375004698759115b6bdc67c529b30dabcb67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:37:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"2fe088d8082828cfb892c22b3ff0764f5ccacda6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 18:39:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee07fa001b44c59a19a5b8395215f2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b44c61fb4846dbb077da097a47ca41 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c8b6574b1c641338122af5dc6466cf3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":6},{"id":"156608db5803e8d17f2fb458da5616e6c6330af5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-11 18:43:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-fcp/55a1c05/ : SUCCESS in 1h 04m 05s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/09e7888/ : SUCCESS in 2h 09m 12s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-nfs/117d000/ : SUCCESS in 45m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/851d53a/ : SUCCESS in 43m 34s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/b73e1f4/ : SUCCESS in 42m 19s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/886869/6/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/aaba97e/ : SUCCESS in 2h 06m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"7419b74c18b6bdbfc50c480e12145de9a483c7eb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-11 19:31:57.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_iSCSI/df8fe8f/DellEMC_PowerStore_iSCSI/280 : FAILURE in 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_FC/b8b72da/DellEMC_PowerStore_FC/292 : FAILURE in 1h 12m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_NFS/fad988c/DellEMC_PowerStore_NFS/289 : FAILURE in 21s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_PowerStore_NVMeTCP/d74f7cc/DellEMC_PowerStore_NVMeTCP/279 : FAILURE in 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"c1136cd69bae48bb80a67383ac8f77c7e2b17bcc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-12 03:24:45.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_Unity_FC/ca9280b/DellEMC_Unity_FC/775 : SUCCESS in 1h 45m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/886869/6/check/DellEMC_Unity_iSCSI/a23c4c8/DellEMC_Unity_iSCSI/784 : SUCCESS in 2h 11m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"2ab134e0399e5ff314c4cb0eb98023f53ac645b0","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-09-12 15:06:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/886869-6/HPMSA-iSCSI/ : SUCCESS in 1:42:18\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/886869-6/HPMSA-FC/ : SUCCESS in 1:24:09","accounts_in_message":[],"_revision_number":6},{"id":"645d9dd1227e6ad9e8c7d87c7a1b8f2c0eb6cd3c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-09-12 21:06:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/886869-6/PVME-iSCSI/ : SUCCESS in 2:01:52\n- PVME-FC https://os-logs.seagate.com/dell-ci/886869-6/PVME-FC/ : SUCCESS in 1:41:33","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"45263d54dc1638a3f942d61f9660808e1d5a982e","revisions":{"bdfcaa27952375f0c651a2b7e62a2e507b562430":{"kind":"REWORK","_number":1,"created":"2023-06-23 19:26:45.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/69/886869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/886869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/886869/1"}}},"commit":{"parents":[{"commit":"2b0567b0e5b28f62c28cadfe619d8a209c61de3c","subject":"Merge \"Automate generation of volume transfer api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b0567b0e5b28f62c28cadfe619d8a209c61de3c"}]}],"author":{"name":"Nahim  Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-04-19 12:22:30.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-06-23 19:26:44.000000000","tz":0},"subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","message":"NetApp ONTAP: Added support to Active/Active mode in NFS driver\n\nThis patch enables Active/Active support in NetApp NFS driver.\n\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I401ca89440d44e04f460741981cf24a42e5264a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdfcaa27952375f0c651a2b7e62a2e507b562430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdfcaa27952375f0c651a2b7e62a2e507b562430"}]},"branch":"refs/heads/master"},"299ee4ba4cfaf74de7b4759c2b67c2bee5fec90c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-23 20:19:50.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/69/886869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/886869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/886869/2"}}},"commit":{"parents":[{"commit":"0dc8d75229fec860c6fd82ecd772c242839e4e87","subject":"NetApp ONTAP: Fix create FlexVol pool replica","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dc8d75229fec860c6fd82ecd772c242839e4e87"}]}],"author":{"name":"Nahim  Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-04-19 12:22:30.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-06-23 20:11:54.000000000","tz":0},"subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","message":"NetApp ONTAP: Added support to Active/Active mode in NFS driver\n\nThis patch enables Active/Active support in NetApp NFS driver.\n\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I401ca89440d44e04f460741981cf24a42e5264a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/299ee4ba4cfaf74de7b4759c2b67c2bee5fec90c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/299ee4ba4cfaf74de7b4759c2b67c2bee5fec90c"}]},"branch":"refs/heads/master"},"e3d88c459972a1003db85eb3e220d15432f11c62":{"kind":"REWORK","_number":3,"created":"2023-07-24 19:42:35.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/69/886869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/886869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/886869/3"}}},"commit":{"parents":[{"commit":"0dc8d75229fec860c6fd82ecd772c242839e4e87","subject":"NetApp ONTAP: Fix create FlexVol pool replica","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dc8d75229fec860c6fd82ecd772c242839e4e87"}]}],"author":{"name":"Nahim  Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-04-19 12:22:30.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-07-24 19:38:21.000000000","tz":0},"subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","message":"NetApp ONTAP: Added support to Active/Active mode in NFS driver\n\nThis patch enables Active/Active support in NetApp NFS driver.\n\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I401ca89440d44e04f460741981cf24a42e5264a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3d88c459972a1003db85eb3e220d15432f11c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3d88c459972a1003db85eb3e220d15432f11c62"}]},"branch":"refs/heads/master"},"97184a93e5ac646f7592a63ff8935f7eb2c42350":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-07-27 16:02:50.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/69/886869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/886869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/886869/4"}}},"commit":{"parents":[{"commit":"64b5543d925d9b176595ab3787e4964fd40e5e91","subject":"NetApp ONTAP: Fixed errors on failover-host operation with REST API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b5543d925d9b176595ab3787e4964fd40e5e91"}]}],"author":{"name":"Nahim  Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-04-19 12:22:30.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-07-27 16:01:37.000000000","tz":0},"subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","message":"NetApp ONTAP: Added support to Active/Active mode in NFS driver\n\nThis patch enables Active/Active support in NetApp NFS driver.\n\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I401ca89440d44e04f460741981cf24a42e5264a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97184a93e5ac646f7592a63ff8935f7eb2c42350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97184a93e5ac646f7592a63ff8935f7eb2c42350"}]},"branch":"refs/heads/master"},"db813c41fd484a2678c388eff74265acc25eb375":{"kind":"REWORK","_number":5,"created":"2023-07-28 10:48:34.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/69/886869/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/886869/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/886869/5"}}},"commit":{"parents":[{"commit":"64b5543d925d9b176595ab3787e4964fd40e5e91","subject":"NetApp ONTAP: Fixed errors on failover-host operation with REST API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b5543d925d9b176595ab3787e4964fd40e5e91"}]}],"author":{"name":"Nahim  Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-04-19 12:22:30.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-07-28 10:47:03.000000000","tz":0},"subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","message":"NetApp ONTAP: Added support to Active/Active mode in NFS driver\n\nThis patch enables Active/Active support in NetApp NFS driver.\n\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I401ca89440d44e04f460741981cf24a42e5264a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db813c41fd484a2678c388eff74265acc25eb375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db813c41fd484a2678c388eff74265acc25eb375"}]},"branch":"refs/heads/master"},"45263d54dc1638a3f942d61f9660808e1d5a982e":{"kind":"REWORK","_number":6,"created":"2023-07-31 11:39:15.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/69/886869/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/886869/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/886869/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/886869/6"}}},"commit":{"parents":[{"commit":"64b5543d925d9b176595ab3787e4964fd40e5e91","subject":"NetApp ONTAP: Fixed errors on failover-host operation with REST API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b5543d925d9b176595ab3787e4964fd40e5e91"}]}],"author":{"name":"Nahim  Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-04-19 12:22:30.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-07-31 11:38:13.000000000","tz":0},"subject":"NetApp ONTAP: Added support to Active/Active mode in NFS driver","message":"NetApp ONTAP: Added support to Active/Active mode in NFS driver\n\nThis patch enables Active/Active support in NetApp NFS driver.\n\nTo support replication in A/A mode, the method failover_host was\nsplitted in two phases (failover and failover_completed) as\nrequired by the spec [1].\n\n[1] https://specs.openstack.org/openstack/cinder-specs/specs/ocata/ha-aa-replication.html\n\nChange-Id: I401ca89440d44e04f460741981cf24a42e5264a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45263d54dc1638a3f942d61f9660808e1d5a982e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45263d54dc1638a3f942d61f9660808e1d5a982e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33986,"name":"Lucas Morais de oliveira","email":"lucasmoliveira059@gmail.com","username":"lucasmoliveira059"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
