)]}'
{"id":"openstack%2Fcinder~887558","triplet_id":"openstack%2Fcinder~master~Ia4981d36a0051c5ebd2fc86b5c68a4ae509f6d23","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-11-16 04:37:12.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"Ia4981d36a0051c5ebd2fc86b5c68a4ae509f6d23","subject":"DNM: internal test","status":"ABANDONED","created":"2023-07-07 06:03:44.000000000","updated":"2023-11-16 04:37:28.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94c2e11dbb00274b1792957fc0011f58901ea28b","_number":887558,"virtual_id_number":887558,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-24 12:56:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":-2,"date":"2023-11-15 17:11:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":-1,"date":"2023-11-15 17:11:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 06:04:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-07 06:05:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-07 06:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-07 07:19:39.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-07 07:33:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-07 08:29:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-07 08:49:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 09:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-07 10:04:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-07 10:24:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-08 14:59:19.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-10 00:39:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 06:49:15.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-24 07:04:37.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-24 07:26:32.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 07:28:26.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-07-24 09:17:35.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"REVIEWER"},{"updated":"2023-07-24 09:25:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 09:49:58.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 12:59:04.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-07-24 14:42:35.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-25 09:39:15.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-07-25 14:08:10.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-15 17:11:34.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"9a4bf9ae41867c77513c448a6f7142d3bc686b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-07 06:03:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfc3c84eedd413074bc4ec3c8c03039c02d693c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-07 06:04:39.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"160e75453a0bbc9749bfdae3de9de5918d2a4450","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-07 06:05:36.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-887558-1/ns5_nfs : FAILURE in 1m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-887558-1/ns5_iscsi : FAILURE in 1m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d5f6b2955aeae0bb50e3a4a8697bd5cee33cad06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 06:16:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21d51b3f71540b889194b56b1bbecf0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99550a6f07284b0b970116d464ed1c97 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40874d49b57b4a0fa7cc2051829825a8 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e18899c4ccb643afbc769bda565284c6 : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba186b115631d9ef09d7332fe5a1b03500e01b68","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-07 07:19:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/8b5f029acf3f4c54a32a596c0b2e0ba0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/26f29042a7c8443fac4c35e369e1499f : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"55e31e3fa23ea392895d09c36301a29b105ba533","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-07 07:33:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/887558 : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fdfc14fe646a4fd7fa67ed034201ef3c80cd3763","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-07 08:29:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/58/887558/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/dbc43f8/ : SUCCESS in 2h 05m 07s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/58/887558/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/451ab49/ : SUCCESS in 2h 05m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/58/887558/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/236209c/ : SUCCESS in 1h 50m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/58/887558/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/02d3b19/ : SUCCESS in 1h 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b35e2724e49bbf478b413b18ea4632f05e9a539b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 08:49:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4adf85a4d10427f82bdb2ff356c9008\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4d9a59cfe74276b9217ccf6ca092d5 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38d3c2ce9efb4533885b818367170742 : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8b33b72bca144b2a31255e53d027a9c : SUCCESS in 12m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e57ecb987504b378946534efd71ab09 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a5a17488214505b0969822de562d4b : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/203c5efaf6a9409498eb4a0d4fe0760e : SUCCESS in 1h 14m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3311b13211d9470781555e7a245960d8 : SUCCESS in 1h 58m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18fade2d67b34ce4827c79c57693f031 : SUCCESS in 39m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8095b50ab3e64becbae3d9fd0d787cb2 : SUCCESS in 16m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/854f16ccc6dc47bc932a0c82dbbc91ac : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1a51685fb10420c80ce916ae7d6abc4 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb7604028bbd49fe8ddf3aed594829d0 : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89d8a0baf7684a898176550efd1c9b2a : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/14028c00a45d43439226acaebb3098f6 : SUCCESS in 41m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e76af3daac4c42fd9310392a3bf7e713 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/155b1c97c9564e32a2052a168041c5e4 : SUCCESS in 2h 19m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc51ba075f694e74b7be07594f58762b : FAILURE in 2h 14m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf0bd46debed411dabd1c1e29a1b74f6 : SUCCESS in 2h 07m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/415b263aa1df4141bf0e9487fc281b4f : SUCCESS in 2h 30m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8f47891075244d69eb15d2e16375ed5 : SUCCESS in 1h 12m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1663d07958454191a40a4708eccafd00 : FAILURE in 2h 16m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/83260428849149bca855602050aae49f : SUCCESS in 36m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ad4410fef1d4793972368fd6429be95 : SUCCESS in 1h 30m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0d1533b7533c40f18ea375f1317cb3eb : SUCCESS in 1h 30m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d52f25afc3c442bc9c71604f927272c6 : SUCCESS in 2h 42m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/619f13da2dca46f8aa7d958cc10d9ba3 : FAILURE in 50m 23s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/cf60c99d72cb443db798e7b144111e5c : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64caeb61a3769548961d2e7b0f3b6154bcddbe32","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-07 09:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/887558/1/fc : SUCCESS in 0h 43m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/887558/1/iscsi : SUCCESS in 0h 49m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0805e0573da6ef2922c65a7f4264a2f623bdc7e7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-07 10:04:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-887558-1/FC/ : SUCCESS in 4h 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-887558-1/iSCSI/ : SUCCESS in 1h 53m 21s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b7e6583832f0bc91656f98701d6e939f6c4c9d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-07 10:24:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/887558/1/check/DellEMC_Unity_FC/9feaf74/DellEMC_Unity_FC/391 : FAILURE in 1h 41m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/887558/1/check/DellEMC_Unity_iSCSI/1908d0e/DellEMC_Unity_iSCSI/393 : SUCCESS in 1h 53m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7c598a63b26be55894b501a54aa3ecb6cb98fda","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-08 14:59:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/1/20230708205749/NEC-V-FC : SUCCESS in 2h 57m 31s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/1/20230708205749/NEC-V-ISCSI : SUCCESS in 3h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4bce6f8b259a45089bee427675ecd1e20444f4d0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-10 00:39:22.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/1/20230710005907/NEC-ISCSI : SUCCESS in 8h 22m 36s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/1/20230710005907/NEC-FC : FAILURE in 8h 40m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f69658185c479773165246f5fda226b1bdf27fc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-24 06:18:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"81a53ba0f151dbdd144667499fc1e6bf3422725c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 06:29:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76633b779de34e97b9c17e0ee914fe6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9126669c5964c36b9a0a8f06794df85 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a073e250e976415b9e6dd6453acea9c7 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dff4a9eb955a4ff8ba04d6313eff17c9 : SUCCESS in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98f3458902813925517d737abe4ad729db897e0d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-24 06:49:15.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/58/887558/2/check/tyds_test/a248b47 : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":2},{"id":"4656b16cb72a1647b2122c7aba0af84b62b142c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-24 07:04:37.000000000","message":"Patch Set 2:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/387/","accounts_in_message":[],"_revision_number":2},{"id":"f67a0404561054e180971a53efdaeedf8cfadba4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-24 07:21:54.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-58-887558-2/ns5_nfs : FAILURE in 50m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-58-887558-2/ns5_iscsi : FAILURE in 49m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"bb6d743cad8241d1f11533a516da65c8e44a8505","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-24 07:26:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/887558/2/ : FAILURE in 30m 21s","accounts_in_message":[],"_revision_number":2},{"id":"41a9fef954a1b37509649466f434bf8f8d9ff23c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-24 07:28:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/58/887558/2/check/iscsi-driver-alletra-6k/0378166/0378166.html : FAILURE in 1h 08m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a922890b65de0f9d64185fe34aa738f7c90a3ac8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-24 07:41:21.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fb3d558f9f4c49e1a1897f52e5e12c87\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/be6e835a0d124c0db516f353de584d85 : FAILURE in 1h 15m 55s","accounts_in_message":[],"_revision_number":2},{"id":"0d9d44742eac9de83ad0961b7bf90a622b213f2c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-24 07:47:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/887558 : FAILURE in 1h 27m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0f2f5f19a7def5877683dc85f8a076489ba3ad1b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-24 08:24:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/58/887558/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/74af493/ : FAILURE in 2h 04m 55s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/58/887558/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/49e3e4b/ : SUCCESS in 1h 57m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/58/887558/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/8df5a4b/ : FAILURE in 2h 04m 25s","accounts_in_message":[],"_revision_number":2},{"id":"fee75b286163f6048100344c26e6e35d4eac68fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-24 08:25:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4fbc20d360a74885bd477de0c97d8d2d : SUCCESS in 1h 22m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/29fbac24ac684789a8c80efa0c2cd3fd : RETRY_LIMIT in 54m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b33e3c4b75409aa4736458981b380f44bae636c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:10:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba2a0f1f00d04d499ed10bcf11f7bd61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8a94fb8f73445f866db5cc4a11ffd7 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f55ac5e3a6bf4115bcb55a1365319570 : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b93349443674dec98efc0783cc96ddd : SUCCESS in 9m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712861726d924d9aa192d43fe6f2035e : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b73870ef6f492daf11cb147d63febb : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/8cc782da3cdc408ea13ddf7ac4316517 : SUCCESS in 48m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d26556aa82c4052be073e10c11fcea3 : SUCCESS in 1h 36m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2797bea44ba44a78013d52d9c1086a3 : SUCCESS in 49m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/851b10607c884f799271e95b1e9c06ae : SUCCESS in 15m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f9451f1ffafa4d218a79dd79faf56040 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67e56852b406428d83deca0fd6b15a7e : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9d188db3a0244979fa1caddb9407103 : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87b117d5c50d4ff58b25ef7c0a5e440f : SUCCESS in 5m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2a39aaec35564a969ec7b5f12b137a4e : SUCCESS in 36m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c06de935d784d9e8927c32739b332f8 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9174967203b74a5ca363ba551818cab8 : SUCCESS in 2h 03m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5576ed27c65e4db2bd9968912f90a185 : SUCCESS in 1h 54m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/33b5cd1ea4154e2ba3cf9fb2c54921d2 : SUCCESS in 2h 11m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/da32cf36ff504b098756dc0fe146b616 : SUCCESS in 2h 45m 08s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c67a046815bd4888b57385e1f21c8e0c : FAILURE in 1h 10m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4414c56a7e4d4f6dba98456ba1a25692 : SUCCESS in 1h 22m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b0a3fe2bfbd54e9198aea57b7848e5e4 : SUCCESS in 38m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1e1713a033404b82bce311d0f163d8b3 : SUCCESS in 1h 35m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6fa5afa11d034f5c882028ae1c672b94 : SUCCESS in 1h 18m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ddd0329d83824089bc1654a4f44c6976 : FAILURE in 1h 29m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/423c03a73164484db3af86b0f37841ca : SUCCESS in 37m 23s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/cc5a8f16b7d54f4aa72b7f2d858dfdb2 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53c9d81c7e439cc72fb97f54e0721eca0909e460","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-24 09:12:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-887558-2/FC/ : SUCCESS in 2h 48m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-58-887558-2/iSCSI/ : SUCCESS in 2h 21m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"64b2fb8af527788af63aa72e041b81edc3e9206e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-24 09:17:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"346cefb3ee09fcf9e37dca6e192f4e5fc0ff0c27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-24 09:25:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_PowerFlex/ebb543c/DellEMC_PowerFlex/308 : FAILURE in 1h 48m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"a4034c7b3a43380b02c0f2567060a1c101c1a12e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:42:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f6432be70534a39bb16d12a4da8fcb5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f43d40c6166845e7a3be3ce3662589ea : FAILURE in 22m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c5cbf7110bc4d5b91bcd4535036b903 : FAILURE in 22m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b27b3f3004646daa08915bfaa6d8550 : FAILURE in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b052004997415ce09a281605dc4b58e6d03d2126","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-07-24 09:49:58.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/58/887558/2/check/fujitsu-eternus-dx-iscsi/ce74653 : SUCCESS in 3h 30m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/58/887558/2/check/fujitsu-eternus-dx-fc/9854ac4 : SUCCESS in 3h 30m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ff624756f702c44d2579b28af781f2ca264eea5a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-24 10:18:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_Unity_FC/b44abb0/DellEMC_Unity_FC/462 : SUCCESS in 1h 48m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_Unity_iSCSI/b03fb16/DellEMC_Unity_iSCSI/462 : SUCCESS in 2h 03m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"246a7d2c6840f3d4b6bc03d807f9bf2976e0991a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-24 11:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/56b8876aefac4d1b94f3f71b787efd0a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/abd49f1d157349b0a95ffdef1f898741 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3d66d8c3355278e4daa26ba3c6b675d3958d3b01","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-24 11:51:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1640/","accounts_in_message":[],"_revision_number":2},{"id":"83069a92cb17bc3e93bd95f7fc420a76f3c5bc20","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-24 12:11:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/2/20230724180545/NEC-V-ISCSI : SUCCESS in 2h 39m 17s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/2/20230724180545/NEC-V-FC : SUCCESS in 3h 06m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d64f7e23f96dd98aa37dfb7efbad3fc9158103ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:56:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58bbaa6ce5b942caad525a004cbca89a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c152a7058f842b3be67579e68a24178 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b8caa5c228447e588192e02292be15e : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/635420b05a8d4ac0a8edc4fd85f81f3f : SUCCESS in 11m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9489a58373c149ab97b2d5d2e25ea41d : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da072d932f94c169be40cd53f699532 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/12ffbd5aa2dc40c898fd4001a0910721 : SUCCESS in 44m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/edb62203022d427ea4b019ea3932fa64 : SUCCESS in 1h 58m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/685bc91346e84613b8ae3e7098dea6fa : SUCCESS in 58m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fdb92aa2d07b47f4ba7997cadae7d2e7 : SUCCESS in 13m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da1439be6d064b389c86bd7491d7b81d : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23b0a9aad9114533b4843d60e996cc63 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/469eaea3c9214af8a5fe7a35bd1fe28c : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea098cc6f6374b99bd1f243aea0c1132 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d5d735e99dca46fe85a7075fe7935722 : SUCCESS in 40m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fcd0e1c747945638a2842285f7334f2 : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/410f2a60739b48adbb89748e18b3876a : SUCCESS in 1h 43m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb8dcdaf47bc40a48b02f34069ae3155 : TIMED_OUT in 3h 07m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4c1fe0faed1d4a6dadf26952fae57f5d : SUCCESS in 1h 59m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c7b58de1d4e24f35bdf8f85031761497 : SUCCESS in 1h 58m 06s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a9177441f15438f928d1e2fdee2fcc5 : SUCCESS in 51m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/461a86eeffe240a9877be18efdb0ad75 : FAILURE in 1h 57m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8ace48b62a9447ceb870f9b0206ac8a6 : SUCCESS in 36m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/891c0ba89bec49c9bc9fbdaef3b22e9d : SUCCESS in 1h 12m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b4c7620fa56c4b7687fd58c01bb00197 : SUCCESS in 1h 38m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f09a35707c041eca5a4d58403dc5e03 : SUCCESS in 1h 08m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e91b8ca4b3e432aa644a57b57c72e12 : SUCCESS in 49m 26s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/69eb24afff80496aa92ddc9348566624 : SUCCESS in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"236f1cac6d9a34cf47a03e2f8a50d695b2c8944b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-24 12:59:04.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"615cc56c6ee091df9724b6f9d2d8bcb05c659fb0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-24 13:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/887558/2/fc : SUCCESS in 0h 33m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/58/887558/2/iscsi : SUCCESS in 0h 52m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7124ffeffed84cc6a71d90823b2ffad9b0b927f9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-24 14:42:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/887558/2/upstream-check/cinder-tempest-plugin-ontap-fcp/5b20445/ : FAILURE in 1h 27m 06s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/887558/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/8deb5fa/ : SUCCESS in 1h 43m 06s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/887558/2/upstream-check/cinder-tempest-plugin-ontap-nfs/a0c4a34/ : SUCCESS in 1h 44m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/887558/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d54e8a6/ : SUCCESS in 1h 42m 06s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0637e5680dd387a819264fd3061f07ff788aa16a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-24 15:18:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/23b8d43639c947b0ad1c64b5f7d238ff : SUCCESS in 1h 32m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3bfea6a84e6a4523bcb94368a2518c07 : RETRY_LIMIT in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d7456a16b5aa63739b0d779935ee191f8c3f065c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-24 18:48:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/2/20230724234604/NEC-FC : SUCCESS in 3h 41m 17s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/887558/2/20230724234604/NEC-ISCSI : SUCCESS in 4h 02m 38s","accounts_in_message":[],"_revision_number":2},{"id":"57b6b2adbff802e35173ae4c58ea45694d0f5b10","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-24 21:38:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_PowerFlex/0009d65/DellEMC_PowerFlex/314 : FAILURE in 1h 42m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"a60eea0c8dd5cd8c73fedbc48280a7e9eebfd22f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-24 22:51:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_Unity_FC/df7e099/DellEMC_Unity_FC/468 : SUCCESS in 1h 51m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_Unity_iSCSI/2592e52/DellEMC_Unity_iSCSI/468 : SUCCESS in 2h 08m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cfe9546f01f9e80eb76975e6d35d47aaec43ebac","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-25 09:39:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/887558-2/HPMSA-iSCSI/ : FAILURE in 1:57:52\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/887558-2/HPMSA-FC/ : SUCCESS in 1:30:31","accounts_in_message":[],"_revision_number":2},{"id":"1d9775f03b0e983c85172fc687063f32c6a69945","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-07-25 14:08:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_PowerMAX_FC/4bb6bac/DellEMC_PowerMAX_FC/185 : SUCCESS in 7h 42m 03s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/58/887558/2/check/DellEMC_PowerMAX_iSCSI/be26ab5/DellEMC_PowerMAX_iSCSI/182 : SUCCESS in 7h 06m 26s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"9305f7436af91a0b2f2b01175b67b2a717255bd3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-15 17:11:34.000000000","message":"Patch Set 2: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b325ddcfc58b2b2eaad1a153c4ace7f2f584b719","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-11-16 04:37:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94c2e11dbb00274b1792957fc0011f58901ea28b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-11-16 04:37:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe6105bb8a1580ded9f51db9ed91103ed8a87a5d","revisions":{"631e62043bca268b29fd564314011089449e8099":{"kind":"REWORK","_number":1,"created":"2023-07-07 06:03:44.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/58/887558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/887558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/887558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/887558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/887558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/887558/1"}}},"commit":{"parents":[{"commit":"25455c476f2d918d550ddf6f209cdfc05a2cf420","subject":"Merge \"db: Remove the legacy \u0027migration_version\u0027 table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25455c476f2d918d550ddf6f209cdfc05a2cf420"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-07-07 06:17:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-07-07 06:19:24.000000000","tz":0},"subject":"DNM: internal test","message":"DNM: internal test\n\ndo not merge.\nThis patch can be ignored.\n\nChange-Id: Ia4981d36a0051c5ebd2fc86b5c68a4ae509f6d23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/631e62043bca268b29fd564314011089449e8099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/631e62043bca268b29fd564314011089449e8099"}]},"branch":"refs/heads/master"},"fe6105bb8a1580ded9f51db9ed91103ed8a87a5d":{"kind":"REWORK","_number":2,"created":"2023-07-24 06:18:28.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/58/887558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/887558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/887558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/887558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/887558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/887558/2"}}},"commit":{"parents":[{"commit":"25455c476f2d918d550ddf6f209cdfc05a2cf420","subject":"Merge \"db: Remove the legacy \u0027migration_version\u0027 table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25455c476f2d918d550ddf6f209cdfc05a2cf420"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-07-07 06:17:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2023-07-24 06:33:59.000000000","tz":0},"subject":"DNM: internal test","message":"DNM: internal test\n\ndo not merge.\nThis patch can be ignored.\n\nChange-Id: Ia4981d36a0051c5ebd2fc86b5c68a4ae509f6d23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe6105bb8a1580ded9f51db9ed91103ed8a87a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe6105bb8a1580ded9f51db9ed91103ed8a87a5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
