)]}'
{"id":"openstack%2Fcinder~887851","triplet_id":"openstack%2Fcinder~master~I9068b6bd32309f1c9a32c1aee57f700a4a5865b2","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-07-06 17:30:48.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-12-15 20:02:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9068b6bd32309f1c9a32c1aee57f700a4a5865b2","subject":"mypy: Add backup/rpcapi.py","status":"MERGED","created":"2023-07-06 14:05:10.000000000","updated":"2023-12-15 20:03:59.000000000","submitted":"2023-12-15 20:02:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887851","meta_rev_id":"5dbc5f50e8a40a9ac17a9986b2c42cd7bcd0b5dd","_number":887851,"virtual_id_number":887851,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-15 20:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2023-07-06 17:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-15 16:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2023-11-17 14:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-15 16:08:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 14:05:25.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-07-06 14:05:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-06 14:06:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-06 14:07:01.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-06 14:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-06 14:50:05.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-06 15:23:13.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-06 16:02:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-06 16:04:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-06 17:19:22.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-06 17:55:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 14:58:45.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-08 10:49:26.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 04:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-08-17 18:35:25.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-08-17 18:50:22.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-17 20:05:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-18 03:13:01.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-11-17 14:59:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-12-15 16:08:35.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"fd4da244bdf50dad20897a3b93121d40b793027b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-06 14:05:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"338fbdf2b0d4c2fa2f7a92e26610c205f2c03c28","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-07-06 14:05:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"50975169f3f72348f7bcb35e6d3bc9f87ac240f5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-06 14:06:06.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a24ff91907fdf3b49d43fb23b693a5f49d3ab4a5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-06 14:07:01.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-51-887851-1/ns5_nfs : FAILURE in 1m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-51-887851-1/ns5_iscsi : FAILURE in 1m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"0b2e30fdfaeda4aca57a4600cbfe671c4506ada4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 14:19:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd38d06bf344626a6acf1854162869d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e75e26582a704dcf9b35264e1c2a7e35 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ada9aecbbd6d4822be8fedce221d9479 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d26b21a1a5174982ae2ec6650c993ed0 : SUCCESS in 9m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57c560e15b20103a8bb3e54a342236931c51a9ac","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-06 14:50:05.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-887851-1/FC/ : UNSTABLE in 3m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-887851-1/iSCSI/ : FAILURE in 44m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e6a67bec4fc2020f4dc0b261f57b0052ecbaf29","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-06 15:23:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/72402850256245dcb68b91348fbad662\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5b26133033bb4425928c4055ef0ab88b : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"53e34a559d62273630aafc75383458fd6ce8eb3a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-06 16:02:25.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_Unity_FC/cdc9254/DellEMC_Unity_FC/384 : SUCCESS in 1h 56m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_Unity_iSCSI/b3054ce/DellEMC_Unity_iSCSI/386 : SUCCESS in 1h 51m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"83ab94ff8a69d00dbf5643cd361413a3254cbb8f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-06 16:04:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/51/887851/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b966cf0/ : SUCCESS in 1h 57m 52s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/51/887851/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/3e9a69a/ : SUCCESS in 1h 49m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/51/887851/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e70dc15/ : SUCCESS in 1h 53m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/51/887851/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/9c3c43e/ : SUCCESS in 1h 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3a531a1e2af05566599578124d08e5afb132663","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-06 17:19:22.000000000","message":"Patch Set 1:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1549/","accounts_in_message":[],"_revision_number":1},{"id":"27d68bec85de9f6e1daaec5e641c9d86d6ebc96c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-07-06 17:30:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"af16882dd8a1bbd61f055b4b9f5897e79aae0442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 17:55:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed5e557d9428408dbeee9273e3c31121\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9a9435755b241feb5dd846880212af6 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b1ddb4ca3cd4397b7eeeaec1ec4f2fe : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f66fd341354563b704aba3ceeb2c0a : SUCCESS in 11m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16af12251ced49959357a4f807dfb7c6 : SUCCESS in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fa7de9e14a4320bce8eab63a558a82 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/b1f0afb044e54a209d8b8511e905f111 : SUCCESS in 46m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c166be8e9864605ad283ebc09ecd096 : SUCCESS in 1h 50m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6844649491846f987ee260fd9225577 : SUCCESS in 46m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f099a8e270354d48b76054d515b2d240 : SUCCESS in 13m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/97eff4fa17ee4e9dac64639f6af0ddcb : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/afbfadcd40db4daab6f4a205261406d5 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bcb703f37d24147abeaf8765171ab6c : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e18bd60b6dcb49fe833d0e59128f8c0c : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9065ec7ae274d6c9f0090b567615793 : SUCCESS in 32m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4112f241479f4d24b821209ecbf6324d : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7fa3e070299549fab65069740062faf8 : SUCCESS in 2h 00m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a79b631d097a4b5282cd69e8e93bddab : TIMED_OUT in 3h 06m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5fa9027bf4c34a3d97f4bcda3a2b8c1e : SUCCESS in 2h 29m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0206c73f16fe4d23b461f707e62fef1a : SUCCESS in 1h 50m 31s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9fe48332c4ac44618d17d9b74005a309 : SUCCESS in 1h 04m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/456086d4b0834bfcb232bd94083c79e6 : FAILURE in 2h 08m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/679df1dfe3a4477e8782395dcccbef27 : FAILURE in 1h 04m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3cf4134aae9d42e4bda04458fdb8cd4d : SUCCESS in 1h 24m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43502cecac794c8692557e4bf0af2d80 : POST_FAILURE in 2h 26m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0129489049b44b02a8c6ec7511d12e67 : TIMED_OUT in 3h 08m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e499ce26487487eb115c379a37711d0 : SUCCESS in 1h 20m 02s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/ef30457f3eb54aa6a2b1d54844d5794c : SUCCESS in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0a97c250998c0cdc33b1641131bf7d57a38fd0a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-06 18:29:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-887851-1/FC/ : SUCCESS in 2h 32m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-887851-1/iSCSI/ : SUCCESS in 3h 17m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"ab739f6178fc251eabb9a5be357f1f4cb8c2d890","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-07 14:58:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/887851/1/20230707210217/NEC-V-ISCSI : SUCCESS in 2h 42m 00s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/887851/1/20230707210217/NEC-V-FC : SUCCESS in 2h 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8c6a4bad2a4be5cacfd787f781dbd29076c32ae1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-08 10:49:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/887851/1/20230708155718/NEC-ISCSI : SUCCESS in 3h 45m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/887851/1/20230708155718/NEC-FC : SUCCESS in 3h 52m 05s","accounts_in_message":[],"_revision_number":1},{"id":"db5083bd48b1932067b9f266bd6460cb1402fbb2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-19 14:44:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a510c6e36fdb8214f26380d8f4b7b829b696854","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:14:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99a28098d3b64b1d97954aec293cf29f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2466ad21c8f44634a4da36bd2b4ec370 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ac70deb377b43279595a1bc1d3704dd : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06e2e192b8bf4f63b1cd10b9f77214fa : FAILURE in 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2091cb173d657aecb00434835931d79e39bf067","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-19 15:44:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/713f92b6e6034cdcbc352571fe19e8a8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2acc470588de4683a2e229d6a214ee7b : FAILURE in 56m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8567050e63261aa4ebac8a413e7e34c9c0f51aec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-19 16:35:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d6a7078dcc83490294c697efc5afd0c1 : SUCCESS in 1h 21m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2570f9eefa9e46b4ae63b35468bea892 : RETRY_LIMIT in 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"23cebd44d2e2064087e59f685c0d11967ad1f504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 17:57:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5da17c8141449f48eb519d1c6cfabc8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f2522395ab45bc8d52c58bc6297349 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9d9a9eeb664c2fb806b7a49daf2ca1 : FAILURE in 12m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22c997228ba74832be30a87d78271c8e : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54d181917a424e33aaf2c75ea954cc98 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d318da7c6f64c918c1fd198fc28e69c : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/dfc0af303e9043b49d5f3172a96c7cba : SUCCESS in 55m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80a424bffec042688f09b496e26f663d : SUCCESS in 1h 41m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44abb3e104b54855b438382792bdf8dd : SUCCESS in 50m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/48e725dc030e48c186d901e3cee7ecc0 : SUCCESS in 20m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6afcc22ea3d43868847739110b3ebd6 : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3006651617de486ebf3c9859c96ca87a : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a1abe757db504fdfb92f07dbbb633221 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/975e1a0e109d4a4b8b269f084bb08eda : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d8fcd651f046405c84cd1db264e2ef9a : SUCCESS in 1h 17m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a714dfd066e24ac182b2577e254fa9ab : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64d4aa9f53494fce908b077ffcf391c6 : FAILURE in 34m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cef63081183641da9f9248f41ec39345 : TIMED_OUT in 3h 05m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e603780acd84c03a38fa71f746ac9c5 : SUCCESS in 2h 02m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ce51f498f2634e7cb4c1d88f323edef3 : TIMED_OUT in 3h 05m 19s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/854c5b5a91df438b85129e1caa756c1f : SUCCESS in 1h 27m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4965fbf444ed4c92b54c5c368c8ca5ef : FAILURE in 2h 14m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5a691331edc8448485b12b82bf44c153 : SUCCESS in 1h 11m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5302e25ef71749cc9ea98129e23ef885 : FAILURE in 1h 39m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/75db9ac16c2742a691a9438b2da74a08 : SUCCESS in 2h 12m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8720f1e7993439c8058e8dd0c240833 : FAILURE in 1h 57m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66f092a6e8be40e9b91936b904819bf8 : SUCCESS in 1h 10m 49s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/d7194aaeb5be4adcaf0cac97f0aa8f6c : SUCCESS in 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15409c906d8693e7758e26b954ad0c7db0a3c4cb","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-20 15:51:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b706cd8a5a87dbfda58e11c6a0fa668f5792b513","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 16:18:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574a320f67a74bd78d3509b6cdfe7a7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcf55d0a83514b9485b490782183034c : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/840196167acd4915a24d2c20eb12520d : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/132ee1b71dc44977b465e5cde05c62e4 : FAILURE in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1fcaaacbf96d30934ebfd7efe86324e8b0db536","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-20 17:09:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/366a6e782afd4b98a6b315e0d51dad80\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7976ca211a7e42ea8a255ccf8808b065 : SUCCESS in 1h 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cfc7f9b73e0724bb2d21caa54adc9cea8a588dd8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-20 18:16:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e2bf2aa10a9f4c549809333ef82f9589 : SUCCESS in 1h 19m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/790987801c5848f4a9fa9597969a0e90 : RETRY_LIMIT in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ed8fa73cf732d34761aa57dc2b7aacb03ce24fb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 18:33:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/184daa70b7ad4556a52e68321cedc47f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f730e0ad4c346dfac32d9bd96c214ac : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d7b46dda08b4e80bb1a4ffb1e99546a : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5e6dbc920b439ebd765096c3aa24d4 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af1c747f3f4e4709a312e1e65b97b1e9 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5796bf6686b1444bb41e830f1d638420 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/174fa8b44ed84cdab2b3f02d1fe35fc2 : SUCCESS in 59m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29eb0829001b4d6aae89eb134f6f9995 : SUCCESS in 1h 19m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18b026ec7fe6438982bdbd5e8f51ab9e : SUCCESS in 1h 02m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/295d50a83d834a4c8971eff536a0adb0 : SUCCESS in 17m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a816c759e7e43abaa50fd9884c83474 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/990564e0ccf846ea9b5450c6c5ee56c2 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b08a6c1867a942d298e34b8e1a2181fc : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/224a84b7c4634ae2ae838548737f1de7 : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2bb50117fb474462a3daa11ab8691fc0 : SUCCESS in 1h 00m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b9ade76e58149f18fd0b9d04a622635 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/24a81c00deda409da84c35fdcf5095b5 : FAILURE in 2h 23m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2dbcb41c9ee248cebdd0468d30658107 : SUCCESS in 1h 33m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c3645ea20e44470bfed2ec64f073af9 : FAILURE in 2h 33m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c010327858bb4768bb31b29af8362b74 : SUCCESS in 2h 35m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b40310e27b5647058f48a1ba082ad96c : SUCCESS in 1h 16m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9252708f8ef84ab78667e14f92eb3c60 : SUCCESS in 1h 34m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/166c69aa597a469bb590bf3392210086 : SUCCESS in 59m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa016d93a4724afbabf11ce723860d2f : SUCCESS in 1h 33m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5f9999f8d01e4ee8867405cf878b3e2b : POST_FAILURE in 2h 29m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2de73bcec0b04735870ba003997ac76d : SUCCESS in 1h 37m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/098726ff1fa641eea6613f6c7f793d96 : SUCCESS in 55m 52s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/a7eaaf136ee94dee96fa62eef5e9b91e : SUCCESS in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de79b3463774d948f3b980bf59996389cfec64d5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-24 04:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/887851/1/fc : SUCCESS in 0h 42m 50s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/51/887851/1/iscsi : SUCCESS in 0h 52m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1e6ce12f95f568525faff78b8ec3bd0080a119d2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-08-17 15:24:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a768ced85a17d4fa2251e83885858f8c8665feb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-17 15:54:09.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f4da5beb5fd34457be8ca51f56deef56\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/94ce77190f8c4147862bdbe962039623 : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2596f93960cdf308ca1373c7b9c2772731049109","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 15:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f984c5272fe40c1ac35749cf4cc8eb8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68e7f228515b4c42ac70989b4d057831 : SUCCESS in 28m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9dd01cdd045641e4b009d43eeb6ca86b : SUCCESS in 21m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4110cf609d3a4763812d3bc3e0af7114 : SUCCESS in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2811885847a4ac8ba6dfd60de633d1673f4a912e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 18:34:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3acceb0cfb2b4380b0852a5f3abe39ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1189ea66e884beca66d3a3348a1c139 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/623b930739d94dceac3fa3bba1d9600f : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/564cceacf6cc47f8a83a45785db87cf1 : SUCCESS in 12m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f01296c38501479eb08c6a627a297939 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186022885a644719bb88b94dad083d6a : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/87473d7df0084620ba78e85bfb8abef4 : SUCCESS in 58m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07e56af5c87e48dda156172f52435e0f : SUCCESS in 1h 46m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c2fe7c606894e948c508f4597a24a75 : SUCCESS in 1h 06m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/78fa5893212d49da9579903fab9138d5 : SUCCESS in 15m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa640ba3909e403389c10ccc93895bb8 : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c440c4f82cf248af885cfa192fb897ae : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fcd6f17e61a040c3bdbadfa6fb9c4d60 : SUCCESS in 6m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/491a4583ebc84f0f8c19955202e40fd1 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb1b2badb1e54d2d8f3751cff8e59531 : FAILURE in 23m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6e68975047e348f5af34eb363b18d475 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b37ec4566e30407386cfa4bab35d38fe : SUCCESS in 1h 56m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbb81f53bea04ae3945c93fcf4f6cc3c : FAILURE in 2h 12m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec9be59de8c14d6f89e46f2991f68760 : SUCCESS in 1h 38m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d08e5c96077541b0ae4108679cf34bc0 : FAILURE in 21m 20s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/714bf0b741c047f8966fec88231461b1 : SUCCESS in 1h 20m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58f0b835aa2247efa9cee5d2f068f2c0 : SUCCESS in 1h 25m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/130ecb20567c45ab8e4adf53fe594e48 : SUCCESS in 41m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e2f2af30a3814d09b071a53aef916ffa : SUCCESS in 1h 26m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/05a5b0c73d424af09a44dba0292681c3 : FAILURE in 24m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4c06d82dee8409094e3869163c49e9c : SUCCESS in 1h 23m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f38753e945224fcb90876e1603abe4da : SUCCESS in 53m 56s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5fd5dc395eb84229ac647975a4943969 : SUCCESS in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a51cf0d64ca7596305e7c4aadb1c3dd4f06dd29","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-08-17 18:35:25.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"2dd985062c2ad64103006caac9208c6c1f061304","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-17 18:50:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_PowerStore_iSCSI/c428236/DellEMC_PowerStore_iSCSI/210 : FAILURE in 15m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_PowerStore_FC/338e261/DellEMC_PowerStore_FC/217 : FAILURE in 15m 59s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_PowerStore_NFS/6690015/DellEMC_PowerStore_NFS/219 : FAILURE in 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_PowerStore_NVMeTCP/f183af9/DellEMC_PowerStore_NVMeTCP/209 : FAILURE in 13m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"1ae71dcb4f3bcf2978f8695e2db825a54da79e43","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 19:44:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_Unity_FC/d385194/DellEMC_Unity_FC/603 : FAILURE in 28m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/887851/1/check/DellEMC_Unity_iSCSI/12a3496/DellEMC_Unity_iSCSI/604 : FAILURE in 14m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9b52bf9894940c64cb0c1aeb7855571322b87655","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-17 20:05:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/887851/1/upstream-check/cinder-tempest-plugin-ontap-fcp/b240095/ : SUCCESS in 1h 04m 16s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/887851/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/618626d/ : SUCCESS in 1h 29m 39s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/887851/1/upstream-check/cinder-tempest-plugin-ontap-nfs/9039340/ : SUCCESS in 1h 28m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/887851/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/07028c5/ : FAILURE in 1h 25m 03s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"be53392ac8583b99441727f0c89088c922527228","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-18 03:08:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/160dbf9da88744aeb75e21cb664dc96f : SUCCESS in 1h 43m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8c18bfb22f454ddfb60b48b34583a6d8 : SUCCESS in 1h 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cfca0e946467b8404c917a6dc857ac5317d7ea5f","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-08-18 03:13:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/887851-1/HPMSA-iSCSI/ : FAILURE in 1:17:43\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/887851-1/HPMSA-FC/ : FAILURE in 1:08:45","accounts_in_message":[],"_revision_number":1},{"id":"c8818e209d5c2efe1bd82d18e06771a69f13ed72","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-17 14:59:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4290ba4b28d8f88da5b6de111ad70f8b03c4a04f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-12-15 16:08:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"922bbdd126cf2895715d4d3a3a3446e78cbfc235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 16:09:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfe698f0f9d54cddaedeac918b81b053831aa631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 20:02:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1fe9ff02914c989ed9ea509aa7b7ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660535bc23264220b6f7ce235070e433 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c70924a8e8b473c890ef9e2815a5333 : SUCCESS in 12m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3274202ae3dc4c88aaa52fd09452a6b9 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5dc15f62c114924948ef4e03b7e1552 : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8696f60543394cf287fdfee97ecf343c : SUCCESS in 57m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/493a8cf44d8b44a39b79ba6063a74c3b : SUCCESS in 54m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/388b5e30195349eda4c7079e4f682ca0 : SUCCESS in 1h 14m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f111752cc1724928b42f23d6b83c0c92 : SUCCESS in 58m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/530c58f7f2b744fbad425a163481be50 : SUCCESS in 1h 11m 51s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fcc18682cc774597ac76fbb184f33102 : SUCCESS in 2h 06m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c851ea4f4b7441698df1d64876b517c : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f5943ef955cfc021daf48b3fb77cd01fa9e08d51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 20:02:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5dbc5f50e8a40a9ac17a9986b2c42cd7bcd0b5dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 20:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96785253034d4eb3a4faf268b19468b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eac9608d7d04ac7b723b8e899749078 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fa491a8d9699de73b251870e7185dbe308368f2","revisions":{"7fa491a8d9699de73b251870e7185dbe308368f2":{"kind":"REWORK","_number":1,"created":"2023-07-06 14:05:10.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/887851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/887851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/887851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/887851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/887851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/887851/1"}}},"commit":{"parents":[{"commit":"25455c476f2d918d550ddf6f209cdfc05a2cf420","subject":"Merge \"db: Remove the legacy \u0027migration_version\u0027 table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25455c476f2d918d550ddf6f209cdfc05a2cf420"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-07-06 14:03:33.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-07-06 14:03:33.000000000","tz":-240},"subject":"mypy: Add backup/rpcapi.py","message":"mypy: Add backup/rpcapi.py\n\nRelated-Bug: #2025277\nDepends-On: Iabfebacfea44916f89584ffd019d848e53302eaf\nChange-Id: I9068b6bd32309f1c9a32c1aee57f700a4a5865b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fa491a8d9699de73b251870e7185dbe308368f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fa491a8d9699de73b251870e7185dbe308368f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
