)]}'
{"id":"openstack%2Fcinder~888492","triplet_id":"openstack%2Fcinder~master~I236b8b6158a9e6813a047a0c9601ff3d9e209f39","project":"openstack/cinder","branch":"master","topic":"test_export_import_backup","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-07-14 21:11:02.000000000","reason":"Change was abandoned"},"35851":{"account":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"last_update":"2023-07-14 21:11:02.000000000","reason":"Change was abandoned"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-07-14 21:11:02.000000000","reason":"Change was abandoned"},"35836":{"account":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"last_update":"2023-07-14 21:11:02.000000000","reason":"Change was abandoned"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-07-14 11:05:22.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-07-14 21:11:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I236b8b6158a9e6813a047a0c9601ff3d9e209f39","subject":"Automate generation of backups api-ref samples","status":"ABANDONED","created":"2023-07-13 17:53:33.000000000","updated":"2023-07-14 21:11:02.000000000","total_comment_count":3,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"759f3d6a3ce15a44e8a39d0584893688f7208dae","_number":888492,"virtual_id_number":888492,"owner":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-13 20:01:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2023-07-14 11:12:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-07-14 11:12:01.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"reviewer_updates":[{"updated":"2023-07-13 17:58:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-13 17:59:10.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-07-13 17:59:10.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-07-13 17:59:10.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"state":"REVIEWER"},{"updated":"2023-07-13 17:59:10.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-07-13 17:59:10.000000000","updated_by":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-07-13 18:16:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-13 18:35:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-13 19:39:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-13 20:34:31.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-13 21:54:15.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-13 22:55:50.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-13 23:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-14 01:41:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-14 10:19:20.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"9088f3c8d9db43e9180bd9fd03d98f918301bb87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-13 17:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"791453063bce6508cef1ab53badaf7d2873ccdea","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-13 17:56:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"42e93c4dee44bf2690cb7e64442388cb468821ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 17:58:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/550f507832934a179d43c39c5d479568 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d615f6e3e7f6489e8c073b3bacc5f9a7 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d474659a1f8dd3cb1083e26e03cd44819abab027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:16:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c061f7f168bc4da58b5fb418490d1b4d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dd92a487a5249c4bd87d6e6ef944af7 : SUCCESS in 20m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7977217691ae4337b99f035e9a5757a8 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45022dcc73b84bf3abc26a52eed13d73 : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"926374aedaac346b75636965841d402758a84504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:35:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f49c88cc49334768a1202c1c393b250b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b3ae0692a5f4e99b0346fab3cbbe315 : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0f6b9ab2e74be1bcb2235ce9d1e6ab : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7af11f7086794f73bc6a71e33f538229 : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97b11e7747794a45a1cb83b6a7836fc9 : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06b66828dbe7446298f5ba5d06b62850 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93482c703f6e47aa88d805ee8b4de0a0 : SUCCESS in 10m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/303cc89883bc4f4882ea14b6845e375f : SUCCESS in 16m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8af5deb7276f4f17bdf11a6be6fab1cc : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11b4bcc9d36b4566befac78637117962 : FAILURE in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ed505257f764f57bb87c663ade39a23 : FAILURE in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9d2cdb0414214254afd059e308bc03b5 : SUCCESS in 40m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85ce54fec6ac4a80a9f2512db9aa11c0 : SUCCESS in 6m 27s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/ae81ac0b6e5e4634b13ce29e48836096 : SUCCESS in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee637f866fab777058c3109a052c5f37e6f1d473","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-13 19:08:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7668e05dec4b4c61dd99c2f280883ec63c71e87b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 19:13:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/73f6398f562f43dcb45efe13683edcdb : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0a11b493a2cf411eaf1e5549f3a2c6f7 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c9565543c19dd08db82ce8d6fa4a6d5df583a51a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 19:23:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0fd1a7e139441a9be262036e777705d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21491ed027b24014a87ad407276544b0 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85d8e1aeb997412a8251058ed791bc91 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/60a556bdbd794e359d9c65c2238f1742 : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b90e146f0bd82cd94b94d5b241a8c3b37ddcf964","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-13 19:39:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/888492 : FAILURE in 30m 01s","accounts_in_message":[],"_revision_number":2},{"id":"52d26f744220f0386a920d810c4b4ae218f8afde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 20:01:37.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3706b66b10d4f12be0120751be0df56\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/41fdb101455541378fdebd16d1a01772 : FAILURE in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943737ab26454b70b13e5304620dc301 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d97a4a058514ba7b983a1c6d2ab010d : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c34653e4dc34b4ca0c3d57ce7df7749 : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78035a72ac2e43aeb2f4838cf20b1fde : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55793c9c86c481c8dec204abd31fad0 : SUCCESS in 13m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd421f0a1bda410b8afe5fd2082fa51d : SUCCESS in 17m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6a6c1d3848074fe8b80432bcbd1f1b25 : SUCCESS in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a143db5ddcc24a43b864d272b19c67bd : FAILURE in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76e8b98ec7d64a2b9bac4ed2d4108524 : FAILURE in 7m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91dcf10165c44657b7b95e721bf910b1 : SUCCESS in 39m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78d5d3406cf844af88626152328c8cf0 : SUCCESS in 6m 17s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/89e97f26e0a54d92ad474b3fa093ea03 : SUCCESS in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"215a23f23df63484db4c6cabb11bce24e55cf27a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-13 20:34:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/7bcf52ed9fcb4ca88a04508cc956cbee\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6fc5471578f44b4ca0ff13fffb43355b : SUCCESS in 1h 14m 12s","accounts_in_message":[],"_revision_number":2},{"id":"77659c4c8dcabf3f7900557838b341863e34365c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-13 21:54:15.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-888492-1/ns5_nfs : SUCCESS in 3h 58m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-888492-1/ns5_iscsi : SUCCESS in 4h 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ce982e8039f6ae4957719c2997fb70d617975b3b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-13 22:55:50.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-888492-2/FC/ : UNSTABLE in 2m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-888492-2/iSCSI/ : FAILURE in 3h 47m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"bc738e57a4ab354060963e0853637e1c1e01c0fa","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-13 23:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/92/888492/2/fc : SUCCESS in 0h 39m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/92/888492/2/iscsi : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5d9e6336e55ded6ad9357d69f0a5bd3fc27b7347","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-13 23:09:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-888492-2/ns5_nfs : SUCCESS in 3h 53m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-888492-2/ns5_iscsi : SUCCESS in 3h 57m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"3ca8579935180527b812b5ada44c436a5a370752","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-14 01:41:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/888492/2/check/DellEMC_Unity_FC/7cce65f/DellEMC_Unity_FC/426 : SUCCESS in 1h 48m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/888492/2/check/DellEMC_Unity_iSCSI/d35ed1f/DellEMC_Unity_iSCSI/428 : FAILURE in 2h 08m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5abc388a9cd95637167ed1fc687d4d6a04485c02","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-14 10:19:20.000000000","message":"Patch Set 2:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1585/","accounts_in_message":[],"_revision_number":2},{"id":"c89832b7c09b5757212ba7c5520f531e01ac3ef8","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-14 11:05:22.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6840182b702a28f90714f9747484d23d945bd608","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-14 11:12:01.000000000","message":"Patch Set 2: Code-Review-2 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"759f3d6a3ce15a44e8a39d0584893688f7208dae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"date":"2023-07-14 21:11:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbbcb6ccf9002c8e0473c64b0d076e41b5631441","revisions":{"8614171104db9d3239f38ce9615164b2f2c4bcbe":{"kind":"REWORK","_number":1,"created":"2023-07-13 17:53:33.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/92/888492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/888492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/888492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/888492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/888492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/888492/1"}}},"commit":{"parents":[{"commit":"c33e138cd69a91ac5700025257d5815dd233e421","subject":"Merge \"Tests: Provide filter arg to VolumeAttachmentNotFound()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c33e138cd69a91ac5700025257d5815dd233e421"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Oyekola Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-07-13 17:52:58.000000000","tz":0},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) Export backup\n\nPartially-Implements: blueprint generating-sample-files\n\nChange-Id: I236b8b6158a9e6813a047a0c9601ff3d9e209f39\nSigned-off-by: Oyekola Toheeb \u003ctoheeb.olawale.to23@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8614171104db9d3239f38ce9615164b2f2c4bcbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8614171104db9d3239f38ce9615164b2f2c4bcbe"}]},"branch":"refs/heads/master"},"bbbcb6ccf9002c8e0473c64b0d076e41b5631441":{"kind":"REWORK","_number":2,"created":"2023-07-13 19:08:14.000000000","uploader":{"_account_id":35836,"name":"Toheeb Oyekola","display_name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","username":"Toheeb"},"ref":"refs/changes/92/888492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/888492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/888492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/888492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/888492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/888492/2"}}},"commit":{"parents":[{"commit":"c33e138cd69a91ac5700025257d5815dd233e421","subject":"Merge \"Tests: Provide filter arg to VolumeAttachmentNotFound()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c33e138cd69a91ac5700025257d5815dd233e421"}]}],"author":{"name":"Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-06-23 12:33:37.000000000","tz":60},"committer":{"name":"Oyekola Toheeb","email":"toheeb.olawale.to23@gmail.com","date":"2023-07-13 19:07:58.000000000","tz":0},"subject":"Automate generation of backups api-ref samples","message":"Automate generation of backups api-ref samples\n\nThis patch adds the following tests:\n\n1) Export backup\n\nPartially-Implements: blueprint generating-sample-files\n\nChange-Id: I236b8b6158a9e6813a047a0c9601ff3d9e209f39\nSigned-off-by: Oyekola Toheeb \u003ctoheeb.olawale.to23@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbbcb6ccf9002c8e0473c64b0d076e41b5631441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbbcb6ccf9002c8e0473c64b0d076e41b5631441"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
