)]}'
{"id":"openstack%2Fcinder~888905","triplet_id":"openstack%2Fcinder~stable%2F2023.1~Iabfebacfea44916f89584ffd019d848e53302eaf","project":"openstack/cinder","branch":"stable/2023.1","topic":"bug/2025277","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-07-19 17:10:12.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-07-19 17:10:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iabfebacfea44916f89584ffd019d848e53302eaf","subject":"Fix a regression in restoring to sparse volumes","status":"MERGED","created":"2023-07-19 03:04:43.000000000","updated":"2023-07-19 17:11:46.000000000","submitted":"2023-07-19 17:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888905-bug/2025277","meta_rev_id":"3afcbb370c0583d10a9b23b33060c69cf5d68547","_number":888905,"virtual_id_number":888905,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-19 17:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-07-19 12:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-19 14:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-19 14:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-07-19 12:51:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 03:18:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-19 03:27:01.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-19 03:31:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-19 04:24:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-19 04:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 05:00:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-19 06:37:55.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-19 12:51:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-07-19 14:43:23.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-07-19 14:51:32.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"6c09359975261625f7c49e1fa9c4161eeea517fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-07-19 03:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2c7db1c357a059cb6d9ab510150fc1dd94dabf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 03:18:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a42149822dc4554a76eead7a4691f2b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/489640243bd045eb9e9e2fd2caff0c22 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d311bc521f047f58bbcce1058d9d5e0 : NODE_FAILURE Node request 300-0021722296 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec5232f68ba44af034b5efeccacffb718a3ef5e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-19 03:27:01.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/50b59e7cf0d24a7b8d8142083576d895\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3fe9ec5e146744dd8b2d7b922cdb49da : POST_FAILURE in 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1b1cd6fde725acba261352a9461ba2b240cbca69","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-19 03:31:27.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1616/","accounts_in_message":[],"_revision_number":1},{"id":"ad3484204d622ca72858ab28b553b7b7e694b90a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-19 04:24:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/888905 : FAILURE in 1h 18m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2ff4b0e2ca19ec4333ba728d565df8bf4224da74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 04:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f6a56a376142aa833add934943e304\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c89a4e5759442b0b64014d0363596c5 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcd5be97247a45b4a10d2af4411408e0 : SUCCESS in 14m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af294a3b9b114bb9a3e00351fd432232 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331d8107be60451a8f891f3b551e7c6b : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/3233c237ca3f421195e295f53c8e5de7 : SUCCESS in 1h 04m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c1a76c9f1f5e45ef9e5f24190b63d63f : FAILURE in 32m 56s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e93224d098d4484cb080a8fa738332c1 : SUCCESS in 1h 37m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f49f88011bc145078cde4a4cd955d497 : SUCCESS in 1h 02m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1969929f281544a4b4ac63ebe013387a : SUCCESS in 7m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/09ac2387baff4301bd3a8f76b24ca666 : SUCCESS in 15m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/89204946b2ec41f3a7a950b24886c5b9 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ebf1b14e33e4dcaa25090d52cddc940 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/272139c7b607472299ae43803d39121e : SUCCESS in 7m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/22be0a5368584265ab19b7c2828997e7 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9f40579c6f44a67b1e414378da00070 : SUCCESS in 1h 05m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8316f360d8949ab989e072a4702421d : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/883d521ab56c47209daa463f159e43dd : SUCCESS in 1h 28m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/70db610061bb4a8cb1beb3fd85a86cef : FAILURE in 1h 20m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5747b66fa2c4146956dda457cb05a06 : SUCCESS in 1h 49m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ebb60e2798cf48b7b5b8e38f227097f4 : SUCCESS in 1h 23m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1a2c90de8d04593a94303d7b7070a77 : SUCCESS in 1h 24m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/20c8155d5d6946ff9498d8c4e1bb76a6 : SUCCESS in 32m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/490636a3e0504f9b802afb3f6865190c : SUCCESS in 1h 09m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60b71b334ed64c4d84c6cf8898a8eff8 : SUCCESS in 1h 22m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7dfa0068402145a6801df18157b6fc0d : SUCCESS in 40m 42s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d8edd17448d4bc1b6e0c4dfd678b388 : SUCCESS in 1h 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b03c228890774d462562b9b26b05bd5eaa108657","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-19 05:00:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/888905/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/828da2d/ : SUCCESS in 1h 53m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/05/888905/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/e2894fe/ : SUCCESS in 1h 42m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/05/888905/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/44bd41a/ : SUCCESS in 1h 49m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1c575f71d89fa1b5524f7302ba847c7a1e1a70f6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-19 06:37:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-888905-1/FC/ : SUCCESS in 2h 19m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-888905-1/iSCSI/ : SUCCESS in 3h 32m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"82b713de66a4a29d096a8cfc7e06acf6a3470d8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-19 12:51:15.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e16deb6717b7d567d2f53e5747ec9e9267b94f0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-19 14:36:58.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1619/","accounts_in_message":[],"_revision_number":1},{"id":"3e86fa3842dc9c7555f83a53aea231be741cd691","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-07-19 14:51:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13c5f8764401b05581d6b7488d63f7cabdc4c72e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 14:52:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed67244b39a25416432cf9b16b055d1dfc97cffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 17:10:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97ae2f1de4b4596972ea0e2e12e1b94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb70d26078240a8b536798410ae9a2e : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/390d64d652ba48659ea5cfeaf1ee0bf7 : SUCCESS in 10m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8411ed75b02243c09fdad175bfedc18c : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7eb8af1edd4602af60688c5899a954 : SUCCESS in 13m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/9782449c80db44048b5f6c90d6472f36 : SUCCESS in 1h 00m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86c5008574e443c19b74e713e73db2e0 : SUCCESS in 1h 44m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/042c0152b04e4ce9863a9949d9e3289a : SUCCESS in 1h 02m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f88d08f5027240e6bfbcae88d4bbbb99 : SUCCESS in 11m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e2f5443bd87f406e90bb2d028ee7d3b7 : SUCCESS in 1h 16m 54s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b090a3c9e9e4c67b2307e03ffbbf85a : SUCCESS in 1h 46m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f891dc0798e64eb2a340b7a0b0eecaa3 : SUCCESS in 42m 52s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99dbc538909944a5a67806b29187e652 : SUCCESS in 1h 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d536a3117ca4ef62c1e725ed59e4c412adb80520","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 17:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3afcbb370c0583d10a9b23b33060c69cf5d68547","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 17:11:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f3782dc6ce455cb2d2182cc7e2bd58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbccf7cc441404ea1c53725fbf9031d : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fa2ffb2cb0d4913b32399ad7305c889 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81af154d85a760e3f157ed5b4acd2624939d09b2","revisions":{"81af154d85a760e3f157ed5b4acd2624939d09b2":{"kind":"REWORK","_number":1,"created":"2023-07-19 03:04:43.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/05/888905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/888905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/888905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/888905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/888905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/888905/1"}}},"commit":{"parents":[{"commit":"f542dd07b07162f39325a83fae67cda166e36798","subject":"Fix glance metadata properties filtering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f542dd07b07162f39325a83fae67cda166e36798"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-07-01 06:30:55.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-07-19 03:03:10.000000000","tz":-300},"subject":"Fix a regression in restoring to sparse volumes","message":"Fix a regression in restoring to sparse volumes\n\nUnfortunately, the commit b75c29c7d8e0e6ac212b59f9ad8d140874e55251\ndid not update all the places where BackupAPI.restore_backup()\nwas called. One of them was in the flow manager.\n\nAlthough this regression being undetected in is disappointing,\nwe are not changing how unit tests are performed in any\nfundamental way for this patch. An outstanding patch using\nMyPy is aready in review, which would capture this case.\n\nCloses-bug: #2025277\nChange-Id: Iabfebacfea44916f89584ffd019d848e53302eaf\n(cherry picked from commit b75c29c7d8e0e6ac212b59f9ad8d140874e55251)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81af154d85a760e3f157ed5b4acd2624939d09b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81af154d85a760e3f157ed5b4acd2624939d09b2"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
