)]}'
{"id":"openstack%2Fcinder~889267","triplet_id":"openstack%2Fcinder~master~I52121496ecae6591e74965e5303616bf4118d26e","project":"openstack/cinder","branch":"master","topic":"volume_list","attention_set":{},"removed_from_attention_set":{"35851":{"account":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"last_update":"2023-08-08 15:53:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35851\u003e replied on the change","reason_account":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-08-06 23:15:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I52121496ecae6591e74965e5303616bf4118d26e","subject":"Automate generation of volume transfer api-ref samples","status":"NEW","created":"2023-07-24 10:57:03.000000000","updated":"2023-08-08 15:53:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2be756a9a08240edee0feb7eb5e65335dbc9c97","_number":889267,"virtual_id_number":889267,"owner":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-01 08:30:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-06 23:15:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-24 11:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 11:25:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2023-07-24 11:37:44.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 11:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 11:45:58.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 12:06:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 12:12:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-07-24 12:17:22.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-24 13:11:33.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 13:38:59.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-24 16:27:28.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 17:42:41.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-24 18:01:32.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 18:04:55.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-24 18:06:34.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-24 18:45:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-25 02:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-25 02:32:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 05:39:36.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-08-01 08:33:03.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-08-01 10:08:02.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-01 13:18:57.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"}],"messages":[{"id":"3380617d8b064af086f1c2ab2e67abb1e024ccb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-07-24 10:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"747b0e1542a78ebb545335310dee3bb18b32c1ed","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-07-24 10:58:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8acd9ac586786ec7c12c127a1f87f253c4f89e63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 11:08:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/683fb992319c4ca88237c8baf39ff7f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77c590281cac4ce7ac317b778f4c332a : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c64ce24ad9fb4638b3b1d1eb9410df82 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf33a056f31c4ffaa8f3c042eea8e045 : SUCCESS in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb780a912375ea21c2dc42690681736dae0c4993","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-24 11:25:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad147943dc1895d78e7a68f0c1b80d9d890951e3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-24 11:37:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/67/889267/1/ : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"660802478029ce82f2fb584f151d100f58da2d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 11:41:36.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9da5cd5942df44fe85465b2301f5ef1f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd06197caef648e39a8c048964189910 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85234b26fee64b3591a8f6d50d2747dc : FAILURE in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b116e63480e49548a5fcb2ae768377b : SUCCESS in 12m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab74c93a2e48466c9623485f9136f700 : SUCCESS in 12m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a733fc3de5344511ab62d74314467a89 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46634754bf2e407d83aed839a3d9642e : SUCCESS in 13m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5ef8e5241ff8441b9c2401643297bef9 : SUCCESS in 18m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d360b888e4f74a16903014eaa2c59262 : SUCCESS in 4m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52c168d25b604a3f96be306604a53737 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cd0834f6bf64c7b9b69573a67896f7a : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11652c6e4e81457a8c121f972390afa9 : SUCCESS in 43m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9bd1806623304259b10187be66c1c8eb : SUCCESS in 8m 12s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/e4426f5c62944c0ba1a486dc9725d282 : SUCCESS in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b139cb6c4c77d431f95f170ecbe22ce09d18646","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-24 11:45:58.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-67-889267-1/ns5_nfs : FAILURE in 48m 23s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-67-889267-1/ns5_iscsi : FAILURE in 48m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"016aaad29443a52e3a9066551292d25504fb2e85","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-24 12:06:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/889267 : FAILURE in 1h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7a9df5766c6b8025154929da220ab170428cacfd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-24 12:12:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/67/889267/1/check/iscsi-driver-alletra-6k/5bb28ac/5bb28ac.html : FAILURE in 1h 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"66d1c9af3e63f6ea3d1d15079787cfb6a18c3a47","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-24 12:17:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/5ea90fe5e6b2449e93c42044b03a5a4f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b134f1d4d88e4bdca00414238c532dac : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":1},{"id":"621aecf5fc7d75d6d4212af673bdd59f42127c51","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-07-24 13:11:33.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/67/889267/1/check/fujitsu-eternus-dx-iscsi/add5593 : SUCCESS in 2h 12m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/67/889267/1/check/fujitsu-eternus-dx-fc/cc89448 : SUCCESS in 2h 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8b4e33bc63ee8690d4978c6a5fd9e55f9b3c0ace","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-24 13:38:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-889267-1/FC/ : SUCCESS in 2h 41m 34s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-889267-1/iSCSI/ : SUCCESS in 2h 24m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"5d700cf053e19d06f889976f01c60c9c705ffb69","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-24 16:27:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c9c0f99f677e4b9c9ab6a05ec7ad3e5c : SUCCESS in 1h 17m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/528de88efe2d43be9da29bdab01915cf : RETRY_LIMIT in 1h 07m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b588e2ee4b1706724ea86e3078d5c3e78b15d1cb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-24 17:42:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1643/","accounts_in_message":[],"_revision_number":1},{"id":"76c089ee51210cb739b1dce7ba112dbbda788cf0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-24 17:51:15.000000000","message":"Patch Set 1:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/394/","accounts_in_message":[],"_revision_number":1},{"id":"081bdd8eb28cbd08eb85a10b2747fb519c6ff58a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-24 18:01:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/67/889267/1/check/DellEMC_PowerFlex/db7bce8/DellEMC_PowerFlex/312 : FAILURE in 1h 34m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"bf6541c4b3d37301f934ff3c54a9033fda878a41","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-24 18:04:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/67/889267/1/check/tyds_test/5d8e6b9 : SUCCESS in 4h 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6c2f316dae0f32e29881eb2c4feb036cab7ec12c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-24 18:06:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/1/20230725001041/NEC-V-FC : SUCCESS in 2h 46m 33s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/1/20230725001041/NEC-V-ISCSI : SUCCESS in 2h 55m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a7bb8d4954b5490ad6e15f9cfecb4adb2257d3bf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-24 18:45:53.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/889267/1/check/DellEMC_Unity_FC/235e4ad/DellEMC_Unity_FC/466 : SUCCESS in 1h 54m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/889267/1/check/DellEMC_Unity_iSCSI/bd35057/DellEMC_Unity_iSCSI/466 : SUCCESS in 1h 54m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5af3e3d1c211e407345cd12c5aa8abde0abb9254","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-25 02:00:09.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/889267/1/fc : SUCCESS in 0h 33m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/889267/1/iscsi : SUCCESS in 0h 45m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c105510bb5aa4773f30cf3fbebb9808cfc250bc1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-25 02:32:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/1/20230725073803/NEC-FC : SUCCESS in 3h 46m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/1/20230725073803/NEC-ISCSI : SUCCESS in 3h 54m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e34b64c9da40c8cd9ad5e21f2f9e09a1392676d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-25 13:28:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/401/","accounts_in_message":[],"_revision_number":1},{"id":"159a0331c06113b5db7917542b73c34044e7687b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-27 05:39:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7dfa0caadecadba77162c878f6aecbea92950664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-08-01 07:37:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2fab8d0eda9de30f01f51f47ae47ea70af9e4eb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 07:50:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33f756d6a1554e33b3ade8c97c5e7cf9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27603c338c9c4b9284f9ab0118596aad : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac9d9b7af2be4c29872fc9d653163a56 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ee87284d67a456ca175ef51e8aea18f : SUCCESS in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14cfa893ec7e7aba96fae90b4c02f3c7d94ad0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 08:30:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49aadd5af484f25a9440d8d5cdbe83e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59555938f4094b7c91b4bd2350aebd91 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d7b11c1ffca4b34a83fa484b1f74f20 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/067cd0cf3895478ca4fb5e0b1d18c788 : SUCCESS in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3136d0c4cdca4b4ca470ebf983715495 : SUCCESS in 12m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20a636b882ed40a3bb7f6516a09e2906 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb07f711b094ab0a37b165326806f19 : SUCCESS in 10m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4270d2cd80524d76a0cb6d8e73e70cd5 : SUCCESS in 15m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/56ae6452a4724688aaf8b4f5021bca35 : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b77114efbbbf4abd9d739a5ee7474bb1 : SUCCESS in 6m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7beef97944dc4060a35b291d36b64847 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/665153f416724c69a5c6e9febd12fd58 : SUCCESS in 48m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7eb8a598419341ee9903a4cc93bac695 : SUCCESS in 5m 50s (non-voting)\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/b8e238aadd4f47faba0af5035b03744a : SUCCESS in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80ac3b61668d50d656ae59720871cd63ce89ee08","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-08-01 08:33:03.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"35dbf6b5e021855ca7a0b8a0e9cd72f5aa18908f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-08-01 08:45:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/67/889267/2/ : FAILURE in 59m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5f9d724923a3899d6a4ae3d9844250e82763b2d8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-08-01 08:52:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/67/889267/2/check/iscsi-driver-alletra-6k/f3844de/f3844de.html : FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3145a7d5d790cdc634eb8217bc4a2b8cabdaa313","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-01 08:57:37.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a517a2bdbc6f44f5b221d0c2e2f53700\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5af0f6243e684cc5aaa0e6dba0f51951 : FAILURE in 1h 17m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4a0a8260ecd4bb45239ece2b194312230865b28e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 09:08:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/67/889267/2/check/DellEMC_PowerFlex/3008f0d/DellEMC_PowerFlex/340 : SUCCESS in 1h 31m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"01cb229b26751c54272ab522bd852e0aae32fe08","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-08-01 09:09:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/889267 : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ebe024279bf9dc2cc2b7501fe27c3ae7e072c5b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-01 09:14:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b2d88bf0abcf48e291834dfb7fbe8b6e : FAILURE in 1h 35m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/69cf786e6bc5478d9d06dd6132199f37 : SUCCESS in 1h 29m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8ad6cf87f313ca11fac3dae93d925f3b681937e9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-08-01 09:21:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/67/889267/2/check/fujitsu-eternus-dx-iscsi/5eff85b : FAILURE in 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/67/889267/2/check/fujitsu-eternus-dx-fc/f7325d4 : SUCCESS in 1h 43m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b90768cbfd70d979160181811eec7a6b7203d2d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-01 09:49:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/67/889267/2/check/DellEMC_Unity_FC/e254f26/DellEMC_Unity_FC/495 : FAILURE in 2h 08m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/67/889267/2/check/DellEMC_Unity_iSCSI/9500c73/DellEMC_Unity_iSCSI/495 : FAILURE in 2h 11m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"faadf7f4c2a2a8cf6c4806462dcd73a925efef36","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-01 10:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/889267/2/fc : SUCCESS in 0h 38m 23s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/67/889267/2/iscsi : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d2b871b644d5903a2a8a4c97bcaae8d874f22a29","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-01 10:08:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/889267/2/upstream-check/cinder-tempest-plugin-ontap-fcp/f907f44/ : SUCCESS in 1h 05m 26s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/889267/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/1c0c307/ : SUCCESS in 1h 31m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/889267/2/upstream-check/cinder-tempest-plugin-ontap-nfs/15f9ff3/ : SUCCESS in 1h 35m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/889267/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f1e4980/ : SUCCESS in 1h 27m 08s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"841dfb24233838c8ceb9600782cf3616ac3816b2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-01 10:14:07.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-889267-2/FC/ : SUCCESS in 2h 36m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-67-889267-2/iSCSI/ : SUCCESS in 2h 14m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"79682c2178e0ceb6f21e696e79bac909e9540683","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-01 10:29:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/2/20230801163717/NEC-V-ISCSI : SUCCESS in 2h 44m 22s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/2/20230801163717/NEC-V-FC : SUCCESS in 2h 52m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e3f78876505a4501cfcb19f8c825f1ce86c833d6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-01 10:39:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1668/","accounts_in_message":[],"_revision_number":2},{"id":"48fa5a4bf0f63e013381954a2c3377b291a44d58","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-01 11:30:11.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-67-889267-2/ns5_nfs : SUCCESS in 3h 46m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-67-889267-2/ns5_iscsi : SUCCESS in 3h 52m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d9f79adcd180837c798f67dd50ae993356f69244","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-08-01 13:18:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/889267-2/PVME-iSCSI/ : SUCCESS in 2:03:47\n- PVME-FC https://os-logs.seagate.com/dell-ci/889267-2/PVME-FC/ : FAILURE in 1:51:31","accounts_in_message":[],"_revision_number":2},{"id":"dbaae6bcae6ae577abc99909c60d55d96d7574ce","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 15:41:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/2/20230801163717/NEC-FC : SUCCESS in 3h 48m 52s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/67/889267/2/20230801163717/NEC-ISCSI : SUCCESS in 8h 04m 17s","accounts_in_message":[],"_revision_number":2},{"id":"346c8918e6627bbc2021ba78030a5a2c7f41991a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-06 23:15:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2be756a9a08240edee0feb7eb5e65335dbc9c97","author":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"date":"2023-08-08 15:53:07.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d489285102e4a1227518665860ee123700d6813","revisions":{"bdd73e26c8a4fc35cc6286fbfbcb271b273e0855":{"kind":"REWORK","_number":1,"created":"2023-07-24 10:57:03.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/67/889267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/889267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/889267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/889267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/889267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/889267/1"}}},"commit":{"parents":[{"commit":"a0268b0faa02612128699d75394a5a714f60a40c","subject":"Merge \"api-ref: Remove non-existent attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0268b0faa02612128699d75394a5a714f60a40c"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-07-24 10:41:44.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-07-24 10:45:07.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer list for microversions 3.55\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I52121496ecae6591e74965e5303616bf4118d26e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdd73e26c8a4fc35cc6286fbfbcb271b273e0855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdd73e26c8a4fc35cc6286fbfbcb271b273e0855"}]},"branch":"refs/heads/master"},"3d489285102e4a1227518665860ee123700d6813":{"kind":"REWORK","_number":2,"created":"2023-08-01 07:37:16.000000000","uploader":{"_account_id":35851,"name":"Desire Barine","display_name":"dessira","email":"desirebarine16@gmail.com","username":"dessira"},"ref":"refs/changes/67/889267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/889267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/889267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/889267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/889267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/889267/2"}}},"commit":{"parents":[{"commit":"a0268b0faa02612128699d75394a5a714f60a40c","subject":"Merge \"api-ref: Remove non-existent attribute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0268b0faa02612128699d75394a5a714f60a40c"}]}],"author":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-07-24 10:41:44.000000000","tz":0},"committer":{"name":"Desire Barine","email":"desirebarine16@gmail.com","date":"2023-08-01 07:36:03.000000000","tz":0},"subject":"Automate generation of volume transfer api-ref samples","message":"Automate generation of volume transfer api-ref samples\n\nThis patch adds the following tests:\n\n1) Volume transfer list for microversions 3.55\n\nPartially-Implements: blueprint generating-sample-files\nChange-Id: I52121496ecae6591e74965e5303616bf4118d26e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d489285102e4a1227518665860ee123700d6813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d489285102e4a1227518665860ee123700d6813"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
