)]}'
{"id":"openstack%2Fcinder~889826","triplet_id":"openstack%2Fcinder~master~Ib4ebe3cafb67a16aef4b961ccf30dade99fdec7f","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2023-09-11 07:15:28.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-09-11 07:15:28.000000000","reason":"Change was submitted"},"33301":{"account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"last_update":"2023-08-31 13:21:31.000000000","reason":"\u003cGERRIT_ACCOUNT_33301\u003e replied on the change","reason_account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-09-11 07:15:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4ebe3cafb67a16aef4b961ccf30dade99fdec7f","subject":"NetApp ONTAP: Fixed errors on failover-host operation with REST API","status":"MERGED","created":"2023-07-27 16:02:50.000000000","updated":"2023-09-11 11:25:39.000000000","submitted":"2023-09-11 07:15:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"889826","meta_rev_id":"2366ae7529c9d4edc9ff2413760b8d088e19131b","_number":889826,"virtual_id_number":889826,"owner":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 07:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"all":[{"value":1,"date":"2023-08-31 13:21:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-06 14:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2023-09-01 00:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-06 14:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2023-09-01 00:01:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-27 16:12:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 16:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 17:10:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-07-27 17:23:51.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-27 18:07:28.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-27 18:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-27 18:36:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 18:39:34.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 19:43:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 20:04:09.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-27 20:26:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 20:39:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-28 00:45:52.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-28 02:47:34.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-28 12:48:25.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-30 23:51:37.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"CC"},{"updated":"2023-08-30 14:24:11.000000000","updated_by":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"CC"},{"updated":"2023-08-31 08:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-08-31 10:14:28.000000000","updated_by":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2023-08-31 10:14:28.000000000","updated_by":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2023-08-31 13:29:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-31 13:41:09.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-31 15:31:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-08-31 15:49:27.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-08-31 20:06:37.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-01 00:01:29.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-09-06 13:12:09.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-09-06 14:48:37.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-09-11 07:09:02.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"20d64cc12d0b81d3f294bcd261fd887965c7abc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-27 16:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99a2f9f965f5d7f4d8a9e5439d95190e0d745709","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-27 16:12:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/830688a33fa6491c92f3216f49600e43 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3a064997025b4654a9536fe4ecdd0ece : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4b705584f0c7bfc3d8aba732a6bbb3b45f59296c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:14:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996e0df29b4f4509816398e1af30bcb3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0badb9b154cd4d8d95cb2beca247cdea : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c49b754c58a48a8bf9d8b07ef8c6c37 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5bc85a1a94bb49d48519872c46e8d444 : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"146d42be1187cf88c81c65ee033b6f5c847824c8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-27 17:10:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/26/889826/1/check/iscsi-driver-alletra-6k/56bea9b/56bea9b.html : FAILURE in 1h 05m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c38308b4b8bb590f4cd278749619f42680ee6747","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-27 17:23:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/2e5c92fd5a384691889518263f5e3043\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b80e36d25ac54bf78d50c5916c3db3ce : SUCCESS in 1h 15m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c3a7ed53a1b67fe60670a66d7865d732e513aeae","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-27 18:07:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy pure-devstack-cinder-tempest-iscsi-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/26/889826/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/e631119/ : SUCCESS in 2h 01m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy pure-devstack-cinder-tempest-nvme-tcp-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e76bb704fba7f7e8a77e970e4f6068b4068db932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 18:24:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9163f5ec531485585cebcac79a7930e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed3023c039c41b6abf3ca428e05d53d : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89059e123e274fc99ca53fa2332d7fce : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4635bf5ba7fa4e509f8f3b2ba1e39c7f : SUCCESS in 12m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f195086f4ff445a7abe1f57a56d19d75 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7397662296ad447aa778e16e07acc3be : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/73f12d781d1942ccabc6037c6564550f : SUCCESS in 46m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64f2f47185704a1c97322cbfd973ef15 : SUCCESS in 1h 32m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4963a56151da4ca986b4e021a064d16c : SUCCESS in 1h 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b6f2379742a41c3abbdc1358a82c52b : SUCCESS in 7m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e73a8f4e889b43d7bf08fa889c1f0c13 : SUCCESS in 16m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/60bc71416ad94df8a6268e17e6b8c696 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf671e5ae0fd4ad6a4a5041f1e21770f : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6d69d1b3405494ebba4d1d1b7296611 : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ed2db37349f4a78af256f630f67a411 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/388ee69bd8b947f6ae4931af6658e8ee : SUCCESS in 1h 04m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c17a1585fc874efe9a8fce8dde133a3e : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2b6ceb2187e424cb04a0fc57d9de143 : FAILURE in 2h 13m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/213655a9a9dc4434aac29438e203d076 : FAILURE in 2h 19m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3524793f27cf4ff991df786ac732366d : SUCCESS in 1h 58m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3b1f98703bff4a7faf2d82f6e63ed68a : SUCCESS in 1h 51m 04s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13e033556e764601a3e092d09789ff1d : SUCCESS in 1h 20m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cbf18cc6f7a44a1d977b0ae3c68cb19f : SUCCESS in 1h 38m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/34df30c4e0ff4331816a93082cefda71 : FAILURE in 58m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b89f1e799d964a978568231a38d73395 : SUCCESS in 1h 08m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e22c1bd39b8b48cdacf101b675020fc6 : POST_FAILURE in 2h 15m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb30398406bd43d791ef84596155cbaf : SUCCESS in 1h 26m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/834807b33b584f1f87cd2c88f6ab5c06 : SUCCESS in 42m 10s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/dc92b7b8ec4c4196a5013c1f0ae379d2 : SUCCESS in 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5650f55dcbd2419ae520bee5dd4df34a49494e29","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-27 18:36:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/889826 : FAILURE in 1h 30m 26s","accounts_in_message":[],"_revision_number":1},{"id":"daa7fdfc6578bda014864a047c7b27b1d6082859","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-27 18:39:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/26/889826/1/ : FAILURE in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e34049d032177e5aea4f60340ec2e4f3d535622b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-27 19:43:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerFlex/0893dac/DellEMC_PowerFlex/326 : SUCCESS in 1h 56m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"0ff7ed961de70013b73805e5ed4409bb15a34496","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-27 20:04:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-26-889826-1/FC/ : SUCCESS in 3h 45m 59s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-26-889826-1/iSCSI/ : SUCCESS in 2h 19m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f79e13d7604b1574d5191c7197106d645c2e1a11","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-27 20:26:12.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-889826-1/ns5_nfs : SUCCESS in 3h 59m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-889826-1/ns5_iscsi : SUCCESS in 4h 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea0ee222242de8ae027501a84e66edd2df3c5d5d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-27 20:39:40.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_Unity_FC/89edde6/DellEMC_Unity_FC/481 : SUCCESS in 1h 58m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_Unity_iSCSI/0bcdbce/DellEMC_Unity_iSCSI/481 : SUCCESS in 2h 17m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d94ae3f5a8f767181efd884760582e79d0ea09ce","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-28 00:45:52.000000000","message":"Patch Set 1:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/423/","accounts_in_message":[],"_revision_number":1},{"id":"4ded624b2496496482e9f8df14cb73461668f0cf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-28 01:08:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-889826-1/ns5_nfs : SUCCESS in 3h 47m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-889826-1/ns5_iscsi : SUCCESS in 3h 51m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9295506f222b4617dbe7f178338734b93a2ef79b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-28 02:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/889826/1/20230728085313/NEC-V-ISCSI : SUCCESS in 2h 41m 21s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/889826/1/20230728085313/NEC-V-FC : SUCCESS in 2h 54m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6875fc0cbbbe6459d468e19bb50a810c76e5dd4d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 12:48:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/889826/1/20230728174345/NEC-FC : SUCCESS in 3h 45m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/889826/1/20230728174345/NEC-ISCSI : SUCCESS in 4h 04m 37s","accounts_in_message":[],"_revision_number":1},{"id":"48b08e27849938816ada714009ffd005b2166679","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-07-30 23:51:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"536b4e9ddbfa7c1f9c09d27b14019f61dba3d600","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 00:18:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab3f6516df1c4fecbe9064f6aca249d2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b5bb9c17f7a4fad80e5e3b2b602986d : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdb91e716851433985bbae851e239149 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d37c7e3ce82d4ca697c0c2d024c1c671 : FAILURE in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7510afba9214daa74c6807afe59b7f0cc8bc2558","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-31 01:13:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3f0e482298444604a7d35a2945e1ab71 : SUCCESS in 1h 19m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2a65cacc2e8a4c9c81636268158e32a2 : RETRY_LIMIT in 1h 00m 32s","accounts_in_message":[],"_revision_number":1},{"id":"36e05b486c3910ead8b38c63f1c1c2839b48a4e8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-31 01:17:10.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3a3b9be10bdf4c398f1c343aeca26760\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2eef77f0396748a69ecba8685359848b : FAILURE in 1h 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"940204c12b00c56ca274c50b9a09a08dfa3864c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 01:50:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cef4cb331d264fb28b22e3f1f578096c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654edcb4ec73493daeba087d9e7ba613 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1273005f88144040b527c39ab7998e8a : SUCCESS in 9m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efab3b6f6da84594ad9efb5a1da303ec : SUCCESS in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54dd497364794b378181fe093b71c1eb : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e01672874004cf0b67d081371da0e45 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/4e868f119dcd40a49c3d11ce037f0afe : SUCCESS in 42m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cefbf4532ea344b7bd4f9cfc8e4bfda4 : SUCCESS in 1h 05m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ac71cb92e944455a3dc0bd018ccda01 : SUCCESS in 46m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e32576c6dae45b68a972e7f7c4e60e1 : SUCCESS in 8m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7509a1efae9245c999a524ab3d49320c : SUCCESS in 18m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/227a51b6d5794f8b8f1fa2ba5fc26f83 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ccbf732d1aa4e6aa13ffb388ccc5e0a : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10f3495e35634396942e5cf909d1ff63 : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2354224309f45cd8ca18b875743994e : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2bde08cc5ef5474faf4d867724864aac : SUCCESS in 34m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a414d94f03f4e07ac5a1a1819223691 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f4fd1168f025427b8b785769c3295397 : SUCCESS in 1h 51m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b9e598f0fa04f69966ea80b7b612c69 : FAILURE in 1h 47m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7616d74293b743f48a99579d640fe2b3 : FAILURE in 1h 39m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fc4c17f5987d41caa0acc18c09c3f78b : FAILURE in 48m 21s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d9de4d7b96a742d389d359052ed0516c : SUCCESS in 1h 07m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/020c85555fbe40999736294b1b448017 : FAILURE in 1h 44m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da6528a684c54b6db1bdc75d425ea684 : SUCCESS in 43m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e1c198bf5ea4ae6897f1b6d06d9405d : SUCCESS in 1h 18m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/470d9c7e27fa4441a6cb54d2ca187d83 : FAILURE in 24m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c876a8b80b594d2a9520e11bec2cc2b7 : FAILURE in 1h 09m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06ce488a4d0a4eb784cae75125f31828 : SUCCESS in 48m 29s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/0a956ce03298485a9b2ca7331d65d162 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc9f5e5b8e00e0f86e873d58e64fd2687c0e6c05","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-08-01 14:06:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6aa7a983d6b695ba06145630d6f7bc5814a611fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 14:28:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/268a47ecb7a848bbb7a67852b569c4c8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a47b94243fdd408e844e3d730fcb9e0b : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9bbe8978dd94e2fb0dbb350de6f4257 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91594507bc5a43db855fa91027d9eec6 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed070db7a2d510791e902d6bd2fb582346906cf7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-01 15:46:04.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/46b9db6826c6479a81a722074828e09a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e6c04d520efe4665a89e1f4d5635fb5b : FAILURE in 1h 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a3163712a722a3a3129602f8972097419a81980c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 16:19:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49ff5760185043f19ee25f99b1880d4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216f37c86aee4ea3b1713555bc186842 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/473f6f84689c4c3e934ca0fd896be9ad : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/238ea1bca1ee47e09e439c1831f0e95e : SUCCESS in 14m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83a44b4bca6e412582bcfdc2db924730 : SUCCESS in 16m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7ce59041644f639ccc3e711b05290b : SUCCESS in 14m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/231ee634b57a4a4a96313f0d65b4b81a : SUCCESS in 53m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5438b21731244e1aa97503dc39844ed4 : SUCCESS in 1h 08m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22da99bd46224c9e8f5b69f987bfecb3 : FAILURE in 46m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a16062476ab4dfaae2fa2ed7266ffe3 : SUCCESS in 8m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4491744d266f47999383a319655d85a3 : SUCCESS in 19m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b9426e176c2e42d7acbbb757c0108c0b : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df59ea6d35ec41a289081df479242731 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1445dea90e5c4884a401a7bf2853254e : SUCCESS in 5m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7623169871134110a36cedfa813c789f : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ea4bf9e7da841a6945c1f9b5c1addd8 : SUCCESS in 55m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50b8cf3818e44500b43b7e80812dc356 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02c4c6ce158e421880d111da9e46c9a6 : SUCCESS in 1h 59m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/05462d17009c44a0a8870bdcfca4f54e : SUCCESS in 1h 17m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7821fef402114f87b36f86afa5b26624 : SUCCESS in 2h 05m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9321456c8073418db8bb424698688942 : FAILURE in 43m 45s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8c77cc2b4a5d4c58b0445df5f0c436ad : SUCCESS in 49m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6ea13462055c422f9893dd7f2e39235d : FAILURE in 1h 45m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/638741ceee154172a35800d85497f718 : SUCCESS in 31m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/88c7e81cc290411bac56c03b2dfde7f6 : SUCCESS in 56m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3743cd9e444348458be9120db00c052d : FAILURE in 22m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ae080cfee124d5eb5d640801e437857 : SUCCESS in 1h 28m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b23141f55f8d406eb5967f7ea5fed1d0 : SUCCESS in 45m 29s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/fdc7da7b74b543279bae2fda9dc5ca18 : SUCCESS in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d865033ec723f1e9748791d7e3108e348c32052","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-01 17:22:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/55ae8c479a3f4a67addcdfe53b991b74 : SUCCESS in 1h 22m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/65b9c8fc185d4a6f8fd07e9f2b98c927 : RETRY_LIMIT in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0a750b176e52d3342c9dcdaf15c15bebc2ff13b0","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-08-30 14:24:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dda5a0c72a357486f7f0aab8e675b2b29e4a4202","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:50:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c64ace75b742b484c25d800b19e892\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/206d8db8ed384488a77ade1b46bce71b : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08a26e3df48f47f9a7c944da48f4ab27 : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ee52a260b1c4b82b6ff83f9a7cc7029 : SUCCESS in 23m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc3131b1ecdd41e887734c44985e4a1f : RETRY_LIMIT in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3851300040a3eecef086bb6ae412698f288241be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 17:08:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38f705af8a09490bb64d792ef0c18257\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aba20623d8534340aed500373f4d1b78 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f96806692d8b4494aa3f00e8a25120c3 : SUCCESS in 10m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/755aec61239a4189b39af452d5f9911f : SUCCESS in 11m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42ceef16ae324857bc10b6c9d080b982 : SUCCESS in 11m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38d1142b28314d1f8f56e23d4e968426 : RETRY_LIMIT in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc34db594cd4fa58c78fac5e3ec3cc3 : SUCCESS in 18m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/6f97364caf0a42de90f6ebd59193f4b6 : SUCCESS in 49m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63b882c3df9841ae8c83c00e764bf4bb : SUCCESS in 1h 31m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4837a9eeab74159889eeeb287091fcb : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95edb6228974c0b9168d2d701fd1333 : SUCCESS in 7m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34a4ba18e5744898962b06dd2d48ed1b : SUCCESS in 15m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3648a6e021e948a19f539a8952239371 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea7a0951479a4b3ea874b300f2fb7464 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0c04fc008654443b945dbac4df734eb : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b9923f7891143d7ab062b1bab6d370f : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2264c3b0e2c7462fb5956cad8ed3d72c : SUCCESS in 1h 07m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2037b2da4544f82be52b26078713a8d : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d66b9b0e3df4777ac5b3d77c5a78986 : SUCCESS in 1h 59m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/67292e206df04330b2f05d02071c9e65 : FAILURE in 2h 15m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/87d2b5bbfad74f60b522327e1ae563f5 : FAILURE in 2h 13m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cce08028e91f482bbd661da2ec5670f9 : SUCCESS in 2h 33m 37s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4ea7aa8f5e245598ec0afd7dc654ae5 : SUCCESS in 1h 23m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f5add9a3c6be412f8a639247dd7c08ae : SUCCESS in 1h 20m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4f2b243c2eba417d953696b75ccf6033 : SUCCESS in 38m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df4aa7e811f740359fa1c08216a28821 : SUCCESS in 1h 25m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8f4ff8edc7314a7db6c9b512a3189eda : POST_FAILURE in 2h 10m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9d2fa7153b824e1790616001eb6b2af8 : SUCCESS in 1h 27m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f971df40c84147029142a7a764095583 : SUCCESS in 1h 08m 58s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/9055bd5b803046778cf209e1ecfbc354 : SUCCESS in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58bae4d1083b8721adeeef99df54395a36874452","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-30 21:49:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cea01339789849549c731befd3c015ef : SUCCESS in 1h 09m 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d2dfde9ed9564409be18c75405b7c80c : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0096677c39bb1bbdcdadf7393d814cc2cf6eba2f","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-08-31 05:11:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec2a433acbf1c89ef450e403265f94d9a6cc0bcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 05:26:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/382106d897b34194b53d2bbe3aa5b584\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8d5f7296c46417f8524891549cf6d18 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2be58177340f4c918669b0bb0dd99efe : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73fa6e11dfbe4610b137f24ce8c7ba32 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/56ad3653f86c4bbfba31a90e07298cc9 : RETRY_LIMIT in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b88d7a728f7b91f7bcae8e187240478c564583c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-31 06:28:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f1b80f51b865404aa011d3079f74c8c8 : SUCCESS in 1h 14m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/142eebaa9b1f42038e93380dfd32c14d : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1c97865c7f4f8842e9dd0c1d72961d12d8f7413d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-31 06:36:42.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ade7948ca8af4961a1591bb5b31a94ef\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/19b717dfc56d4cc7a6c86431b79b9b63 : FAILURE in 1h 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e1e09158f8c0846e4e9e57f28b9c707cde5475a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 07:32:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e992683ad9e44258bfa933649589ef20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624bb0a2569c477d8edf16c8045cf078 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ff1ba56db7845018c92728a6feccf13 : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3a3545751a148fc875763064131617e : SUCCESS in 11m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16dd7294e87443ba90585236a1744609 : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8a9db2d93ef4d13bf96e6ab4bf3d68c : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e6b483922c42f19de62dfac31c0d01 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/3939ee5093064464a5172c9c36617a46 : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/172d59e5e0184517a2f2ba48af0aab71 : FAILURE in 1h 25m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39dcef68e6e141fea1d7bce566aea000 : SUCCESS in 46m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/509af578582e4da5b1380c7a7ae6bd4d : SUCCESS in 8m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c01f0e7b074b4e0599d27831af547a12 : SUCCESS in 15m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9bffe4516a234aa7b6d6290fd4b1bf45 : SUCCESS in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a30c6c5c6e9349799d343a2db419aefe : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf5633932452497b8568a6baf743c0db : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dae24dea009041298b9cad88232f1f87 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/121bb9900a6b4174ae68c96a195d2fd0 : SUCCESS in 51m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0103cc603e9244df992575bf4338e2ce : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17f8b12bb6bb4c66b6293f233874e3ac : FAILURE in 1h 39m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18850a80bea1462b9cd9adb4f0b1dc63 : FAILURE in 1h 30m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc511b3f27a3495193df0499bf247c9c : FAILURE in 2h 02m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6a359654f6404f65b90aa7805e08d65b : SUCCESS in 2h 16m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5cbabd4e153f4b9988e225bbd63e6b01 : SUCCESS in 55m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/42455547ae4f438eaec46121c6e596d9 : FAILURE in 1h 46m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/674c284a032c4457a17ea27dae305e84 : SUCCESS in 31m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b67774d0dc942b3b3771cb900e7d632 : SUCCESS in 1h 12m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6aaba088fb5e46d99d6b6655747dd851 : SUCCESS in 1h 44m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15e95dcc346d4b28a070719cec89abe9 : SUCCESS in 1h 29m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d0ebfa1aaad4c85a15e9df5947845e7 : SUCCESS in 1h 00m 37s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/7ea8cc8d03764e7284c6b470e056f9b8 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f666e967ca52beb45ba4dd7cbebfca242e0f01cb","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-31 08:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/889826/1/fc : SUCCESS in 0h 32m 55s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/889826/1/iscsi : SUCCESS in 0h 52m 15s","accounts_in_message":[],"_revision_number":1},{"id":"65b4533ae35a5f2c6e677b8dbdc987901d3c0a23","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-08-31 10:13:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56c703292c0388bc441ca1d4d809542e1b454676","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:42:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1a5c939c28c474c8059ed0a1991c3b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee544f3681c24478a1b0f494c206041b : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e785c5860fc54aafbe255dda600d9819 : SUCCESS in 26m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a564571f32194da199a988e39b4bc6fc : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3b3986a25fd4ff2993ccbeb8b054211 : RETRY_LIMIT in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"042daade0623ed08e4a0459dd8c8e30c296194e9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-31 11:24:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e3255961c27f41078e6323d9410391ba : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5324f40f117f47bfa8e1bba286d372e5 : RETRY_LIMIT in 1h 09m 03s","accounts_in_message":[],"_revision_number":1},{"id":"99e0579e6446454eeb18dac7aecf377b4a2aafaf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-31 11:36:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/00deb7a329b4405e8ff0a76b8854c9f0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1c8ebd8fd775456ea1f1d1b1ceb685d2 : FAILURE in 1h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6d77b3ef6d377ba19aa2fb6f04e8b88a00833987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2cc18f8c5554422ba79c487f448d7b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c606977e834424a823de168d2d9fa6 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d43c2dcd0ef444eb94a07045b941ba8e : SUCCESS in 13m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cfaf2eb269f43c5848891de2199a3c3 : SUCCESS in 11m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e08cac892cb44a57be816139d881802e : SUCCESS in 12m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/492c03b9e79f475e9508d404ebf246bf : RETRY_LIMIT in 2m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d3e1358bce46b1b62be70c3f6038a3 : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/084e92780fa149d6be1c79416230f851 : SUCCESS in 1h 04m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed71e6d05347472aaa9d3e00fc51767a : SUCCESS in 1h 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1b83056a48f4d919a2b612db827d685 : SUCCESS in 1h 08m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d8f4230f75c4683a2c7dc7f7d993bbf : SUCCESS in 8m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c5273b8ee681421d87eb5e3b128a0c12 : SUCCESS in 16m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6c38e27cd9d46d38474bb9240614689 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02ecf0b14c4140508144e88d77f709c7 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb5b9a4219c04776831be27c2f33dc0d : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fb181fa66d14a5fb4cb2229d6993471 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2429fbc45f1e4571b595b5785008bf87 : SUCCESS in 35m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/41c9cd1fc532448086ed0759eac0c279 : SUCCESS in 7m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15e7da89f0f84f7dacbbd7d1ba1aae57 : SUCCESS in 1h 53m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/440d585c8ef44a3480160fe6b4fe2d7c : SUCCESS in 1h 50m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/705c9be1d22d4a568a77d8c3e2b59a37 : SUCCESS in 2h 03m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0d8e402b27fc496b9240436db56ab8a5 : SUCCESS in 1h 52m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/223958ee781143cf8f6ff2f9656aea76 : SUCCESS in 1h 13m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ebf98ab2784a4e7993bc3185c7ccaa0e : FAILURE in 1h 43m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3692c0d1041b472db1ad5bebdaf86493 : SUCCESS in 53m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab7ac0d5079841a3b07fa4bbed47d04e : SUCCESS in 1h 16m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/96eb268999fb4328a63243bbfd4c2f47 : POST_FAILURE in 2h 04m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/656856bc8dcf4cba88e3e60e952933aa : SUCCESS in 1h 34m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/484a5072cdda4ac1ac4e5ac525867b9b : SUCCESS in 50m 48s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/54daf434e91e487db65d9a0719fd123f : FAILURE in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e5eef84dd264cf57e7b0f4a4d4cbf4e569d883f","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-08-31 13:21:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4b01dd9c80cb7056b36da0a65ab8cce16ba16b2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-31 13:29:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-fcp/8a829a1/ : SUCCESS in 1h 00m 07s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/33e10bc/ : SUCCESS in 46m 12s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-nfs/fc418e9/ : SUCCESS in 42m 53s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3c8e0b0/ : FAILURE in 45m 43s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/acfd2c1/ : SUCCESS in 47m 27s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/8604c4d/ : SUCCESS in 47m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"de95794e48727ec028882e99cd10b760e5042b90","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-31 13:41:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_iSCSI/526b5df/DellEMC_PowerStore_iSCSI/245 : SUCCESS in 1h 00m 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_FC/1b679b0/DellEMC_PowerStore_FC/254 : SUCCESS in 1h 16m 59s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_NFS/2f6d73f/DellEMC_PowerStore_NFS/254 : SUCCESS in 53m 20s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_NVMeTCP/bb072ca/DellEMC_PowerStore_NVMeTCP/244 : SUCCESS in 1h 12m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"c0f1baa043cced0efe1efb4391a11c6063200789","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-31 14:49:59.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerFlex/788f210/DellEMC_PowerFlex/488 : SUCCESS in 1h 35m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b745ec43f540162137f08789995baf317acf4bc0","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-08-31 15:31:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/889826-1/HPMSA-iSCSI/ : FAILURE in 1:18:11\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/889826-1/HPMSA-FC/ : FAILURE in 1:05:25","accounts_in_message":[],"_revision_number":1},{"id":"b1ddfbac9a6c22bf04c6cbb652b08f425051b172","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-31 15:48:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_Unity_FC/ceb8a9c/DellEMC_Unity_FC/702 : SUCCESS in 1h 42m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_Unity_iSCSI/31123ce/DellEMC_Unity_iSCSI/711 : SUCCESS in 2h 05m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"daa8cbdac47b6cdbc1d687fd849b619101d5cf3e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-08-31 15:49:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/889826-1/PVME-iSCSI/ : FAILURE in 1:28:18\n- PVME-FC https://os-logs.seagate.com/dell-ci/889826-1/PVME-FC/ : FAILURE in 1:14:40","accounts_in_message":[],"_revision_number":1},{"id":"9d5d99ad1879b92316931f4fa53dd3bb665bbf26","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-08-31 20:06:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerMAX_FC/81a695e/DellEMC_PowerMAX_FC/205 : SUCCESS in 7h 42m 26s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerMAX_iSCSI/43cd3c3/DellEMC_PowerMAX_iSCSI/221 : FAILURE in 7h 25m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"81545e37709aae7a146d701f8965ee8c4fe21ac1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-01 00:01:29.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3cb75174b3dffa6d24544b5d0c79b18f026bb65f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-09-06 14:48:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e05e5cedd879902365e2e79c3f0aafae4c54e96a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:49:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20782ba72ba2a32e6dfd28f8b3f16fc9b77cecdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 16:48:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b60811403ee14409bd0935e4b9af629c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dbfe75e49bb41409cc7b35c537f64c3 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65cc86f1d2d3404a88389b1f94dc37e4 : SUCCESS in 12m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89f068e748574820912fbdbdfbf36d41 : TIMED_OUT in 40m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64fd3c2f21b74df69af24cc799614bc9 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9caeb86363842c8b39b91f5692f2a5a : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/be9ed37eb984456885333545e4677230 : SUCCESS in 1h 21m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7b61e594f4f4427a2e85e8e691d2c93 : FAILURE in 1h 30m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cf991d9152242fb93c1764005ef7884 : SUCCESS in 1h 01m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/766c7a65281549ee8f6ef066b773e2b4 : SUCCESS in 7m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e61e0eec02bc4bb2bd94ad83075e086d : SUCCESS in 1h 18m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35cf32277df343448c333c312617ec94 : FAILURE in 1h 53m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b563747d795144f499d73f5d67c09c1a : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b5e4313c5c0cf33db6def4996722ed6639685246","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-09-06 18:12:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c91f3251b410079c858ce5256d01587e70ba4785","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-06 18:15:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3721665ae3d4c226562b508eb265c49c9c59f075","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-06 18:21:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c308175a740d48b39fea7b2c367b12d8 : FAILURE in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"997ef18aa7ec34b76ea4267cb710f826b8128fbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 18:35:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e9eefd55784bfc8ee0b6ad0212ffff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4571aca4b6e44f1f940a9f52d1b15d1a : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7cbbd603d4a4a569a50bb0528563b2e : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/854c0475cf1f4813b4ed8a64513dc253 : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4220b35af5a64b54ab40b8547e2f7ef5 : SUCCESS in 19m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01497f8e0b8afd12e829fcf8cf67383498b1cb53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 20:27:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b69974bf88349a3bc9d690f97ed791b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8487812e804539bb8182fa52a440a3 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c04739d11966455aba43d39ed646447e : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fa90bbdc85e4e16b48614ca1c0ca597 : SUCCESS in 10m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a73800de93244af8a9bf4d991d05fec : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38fa6972ac7841a4b11df3074501f1d2 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b5b45a4e254e7d8b062b8d9505d94a : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/e2476d3f085a4bb6bc0e08ece41697cc : SUCCESS in 1h 01m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3413f53b2f746569003e2d3745e5bc8 : FAILURE in 1h 16m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b43f44bf688c47bc9df9389c9924d2fe : SUCCESS in 41m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f7fcee76615430e9eddd976e7180ff8 : SUCCESS in 9m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c06a3541fe54b59b68589cd413a536c : SUCCESS in 13m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b12e4c5566bc4942a86694f5c8e65b0e : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d7520d647854c5c9d832032c4180207 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/33ea6e7b0e6c43b1b7c68ca7a1dd0661 : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e511d87009948ccbe81bc9ce2f86c07 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd7da37757484a238068e3d858e8fc75 : SUCCESS in 54m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64e0d6eb05834abcb6def1e0094d043b : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/07e303a0eec84ff8b51ca687cee20459 : FAILURE in 1h 31m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bbdd28fd63164a3398c18edffae4c04a : FAILURE in 1h 45m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0a25ec1484b426c99b4e15c1de158c5 : SUCCESS in 1h 14m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e60f27ce16b844e1aec47a53096fc9b9 : SUCCESS in 2h 09m 39s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/107822a4573d42c1a562e3659672820a : SUCCESS in 1h 11m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/277246707442461a911df369c79e78a8 : SUCCESS in 1h 22m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2f00f5c7a9f04cdfa6206dce384e48d7 : SUCCESS in 29m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/46641c66d07d45eb966ea088240a4898 : SUCCESS in 1h 23m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/437708726f5849218353176f4cac5c61 : SUCCESS in 1h 05m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd2b30e90e844b31b121d4d7f9f82209 : SUCCESS in 1h 12m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06cfa23cb85d4982b142181d65411aeb : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fccbb92908c29f722980726f744fd673b54687fb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-06 20:43:16.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7252ee4f86884be8bd44ae3d2d8f74c0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3213a10c48ea4532ad917859fda6a75d : FAILURE in 1h 21m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ffdd685b9c2ce742e6eae01ca1e907bff099777d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-07 13:48:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf90f03961d99a8d4054623df65ab08d300033b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-07 13:58:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6210a367d11f4bf8819270acb46f37e7 : FAILURE in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"707629d95bec92dab179faf1147591f085c95c08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:10:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff5489069f0a4bf5ba4b19204dd9c4e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ba8627501d24517be9dead465903a10 : SUCCESS in 20m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9fc768edd294b1583e71641710ffc05 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ed4642d5a85d4e34aac28fe940f5e3d6 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/70d423d823b54176ae97a1bffa850772 : SUCCESS in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d9a095325631b25c82b21880ffaa93e4cf6c856","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-07 14:31:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96c237708bbc9f510f9310a2d9281da526c22a0e","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-07 14:32:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c56a74f7a0571d58df2c3a25fd546737960f61e0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-07 14:41:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/56406b31754a44aaafe828204b511511 : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"db97f644fc7ac2e283885fc2c6b07dceb0c550f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:53:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b24e0a12a37444c880aad92edac72f14\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2b5742a03624954a194f5772ad9329b : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c12a649382a7454486b4d132c787805f : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8a147c46ff84633bf3d30fcec3629dc : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d6fe97cf3c646ebbcb9d6120de3a12c : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ece528c7fd7daa677de998f2c3920406e75448aa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-07 15:43:46.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d3d15afc764d4f6f95f3fe068649ab81\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8f7f38d452734a9ba40b02cdbdc5d897 : FAILURE in 1h 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b0f4c26feef17f58c6b0dbe9506afa99f75dead8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 16:09:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da855131254d468584343a70651df3ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591affe55902414c921716a9558e06cf : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba25092428a345f7bb4e70f0153dbe3f : SUCCESS in 12m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd53b88009134cf5a89d386aa75a97bc : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eac6d2760eae47f8aa15bfb3e9150437 : SUCCESS in 12m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a93bee5d59984493b1305b41378761d2 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16f3ae19af74116a164a2f03317d65f : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/15a7d2dbaf924822ad6c52e8e9e59dc0 : SUCCESS in 59m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86b15b909ed8424d96ebdd2b8c90d241 : SUCCESS in 1h 25m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74413d1a86454521aa90367e2c4ed9fb : SUCCESS in 52m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8410e82e049a41acbb87b0613557edc5 : SUCCESS in 9m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3a8f57970d554ca5903ea298b9fc4c0e : SUCCESS in 16m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fdfcdf4b1e2549bf9a307ad54df633b9 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d344ee8a2834bc98b72af547da4124f : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6723c14343d24278887b03d6d4b310d7 : SUCCESS in 6m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3fb84c4301a048dead4e2a480abeae93 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4f28194e5d644e6f995230451990c134 : SUCCESS in 35m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c6b8929a85cf471db19d459c4d296c13 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa7418ac1c0647dd8c83080ebbd690da : FAILURE in 2h 19m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/152b851dd49d456fa5c2ce8ac44a9659 : FAILURE in 2h 14m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/12ba0386aa6345c387db976e1ba9cdba : FAILURE in 2h 12m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/70de522c4fee4a468ca6d507b3e262a7 : SUCCESS in 2h 11m 18s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/97d05e706863460fa8016a506a93e775 : SUCCESS in 57m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/adc6965a42bb489297a4aabc95bf7cfa : SUCCESS in 1h 37m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fac19dde96b544398a9ae24b37b841d0 : SUCCESS in 47m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4e8cac1291114c6ab7c304ef46c129ab : SUCCESS in 1h 35m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1d9c78035eb242589bce1461e9969e33 : SUCCESS in 1h 23m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/928f265945e34f1a8f5204aaf944a571 : SUCCESS in 1h 38m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39676d7ac5af46fb9d24c60315438ebf : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"747ccb0ae6791b0402569bfdd706d9ee04bf1139","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-08 02:48:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d948691b669f1cc9590556d7156d1c0503b9f5c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-08 02:56:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7c1706fd482e4e98868e205efb059b2e : FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b9c61926a654d7868601e216d1143c1a29157ee8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 03:14:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0869df9f85e401b89179c01bf68216f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f29892fcda694de7bed73a4f4045e7fa : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/489344f8ba6943ac861796474f4cafe8 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b43a94b097644abae686cff66121bce : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11b20cad4530458eabc089d33ece6563 : SUCCESS in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19081b822a736a89be56b4aa63247d24c7247ebc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-08 04:08:19.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d5665ba086af484b8c0739ce82607db4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/10e30c55b83e4ca99fa8dd6234319f54 : FAILURE in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d8782b2d2565650b41bd654c6f47e4993d15c683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 05:23:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc984e577d91407490adfccbeee63d14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcdf2a5cd3744ec4a94a926a0f8d1386 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1601dfa3612401fb6f121b8207fc5a6 : SUCCESS in 13m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b76e4fe25642008302bcf9f89ad206 : SUCCESS in 13m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b7e4acd8e4f4bcfa9a8687ba6bedeec : SUCCESS in 12m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fecadeccc0844f5801ca3ae0208e084 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db363486383746eea8d3a563b706eea2 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/4299d917840f4618b8b2e959de8abb00 : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98a526bc3e9845c89af46bbafa3cab7d : SUCCESS in 1h 25m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1891cba67304a779ee67e90b43564d3 : SUCCESS in 53m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37c9e7f2598b4237ab367f956ea40324 : SUCCESS in 8m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8d41fa67cacb47a59b14ea09e11ddc91 : SUCCESS in 17m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc229e071736401fbc3496160ea47402 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15ed1e3da384415ab9b91a71eb137302 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62111678d223451fba8448e280568023 : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6263d4d5a0674a24bb0d805cacf997de : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e20e1c8f1e634bb5b52ab844e39643ca : SUCCESS in 51m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/399a7fad03f744238b095366b3f6c0a0 : SUCCESS in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/56f21fdec33b48709d6e161608f07c6f : FAILURE in 2h 01m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4ac53817a8c747b6b6934974f5c36226 : FAILURE in 1h 59m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c83e21b57fc4662a4189df8d71b4744 : SUCCESS in 1h 43m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6f5154e9ed234d52934fca30208f14f3 : SUCCESS in 2h 30m 14s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7578f2a2e3bf433084251d53a01017ad : SUCCESS in 53m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c16486680bed4d6d90b8f1bd0fa97f21 : SUCCESS in 1h 18m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a66de49c2e314941a3094c455dbfa429 : SUCCESS in 43m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4774c2e1497b439fb86cda2c41b68772 : SUCCESS in 1h 31m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2184db4a7b3648ba9b20534eb5c4c732 : SUCCESS in 1h 22m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d02d1a94be8b4e98b6b9d3fba406508d : SUCCESS in 1h 29m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d37b90d48ba45dea124f9660d558e78 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bf7876b54b488fa9a0d96434cf0b53632a12acda","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-09-08 11:45:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecb06659284ec6b6a18aa3ea345da1b44574373e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-08 11:53:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03cad23dd0362faeecc091c36ee988b514bd5af5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-08 11:54:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/135f47d000fe4bd593d7557bf5cdac3e : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"22f0086c01d28b72efcc3eb79b1068ac2d5a74e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 12:00:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87417eb3d982400a8e5d0317624e3eb9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfdbd07e4a8e4ed6bd21e9d58120c39f : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f46396a7dea48d2920315470022c239 : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea1787de46354cd9913ff2f87e061fee : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ced0277823c940d89951558f41d26034 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc306af7733d721c78a6f098e517442134134129","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-08 13:08:49.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/17d38a4540344a39adb41f32aa9c50ec\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0d1b289bb5a44c45ac52603b42cd8f9c : FAILURE in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1756dab4b8e56af84cfd535b52426bb3480fcfb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:22:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8383ed5b5d8943158ed7eaa0ced39094\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4307e1260e5945069711b86e1d6264c6 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55954a3dd0a14d7496b16f6b6ecf225c : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62319707ec334de3955044f8cc21c237 : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f48086be4ff14a948f85fc5972a635da : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c11b2326ec44429b6f0500873b02ed2 : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b478f6e3e83b40f8a23a76af067e72d5 : SUCCESS in 14m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/4e18f6d830fe4a2d97fac9a98fda6697 : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92352f44a86b4687b64930653a232385 : SUCCESS in 1h 03m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27f7cb808d3649c98a86513ee856aba1 : SUCCESS in 49m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da3653d4a21d481787b978bad3ce9fb0 : SUCCESS in 7m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/887e5bc8a5934ae988fd77f9f2cf17a7 : SUCCESS in 14m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ee67751163f94f1380bd0219f335c0cb : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58af7dc6bf0e430b9e489d56211bb90f : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a77b1bf33fb4e3fb97d89b6f60bd2d1 : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2737e84c0b87422ca854a499768b49fb : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0d6e46b056864bf496eabd0e2aa9ef69 : SUCCESS in 50m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bffcd0be10614d929c9fab4da79dac19 : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/179f93c91b34464d82c82ede2cc9eeae : FAILURE in 1h 57m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0223b5bc13914b0789caf755874de290 : FAILURE in 1h 58m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/149ab2500d984dda9d7ebed3984d90a3 : SUCCESS in 2h 04m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5a5735962ef94e0092310355d8a9b9b8 : SUCCESS in 2h 35m 20s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de45db51dc9b4be680b66b495b08133a : SUCCESS in 1h 09m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bff5c2d9fe3b494686c1e63471545e4b : SUCCESS in 1h 37m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a6539066915343d6956efcdb5656c115 : SUCCESS in 54m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/123c71f3b84e40c18fc4de0c1a807c99 : SUCCESS in 1h 15m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3e906dca883e4e38ab99fcff301cf607 : SUCCESS in 1h 11m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b7bf3a91f27849a0b22094b0f64759fc : SUCCESS in 1h 35m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e94e512788ad4ad7b48c21e86896218d : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ab92ac9b98ca4691d3ce83f8936c2cbe118ad0c4","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-08 15:05:37.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1eb2da378132dc4a74874036b69e2253dfcc1283","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-09-08 17:07:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fe8d8712e90e1cb8a28d74d40fd97ad17c92f0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-08 17:16:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2c008b5424514f7b8402d0800da7012f : FAILURE in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f60b147b758beb9e13194c5ce02d121a469da0ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 17:27:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/154f8974f3064065ad2250e125895390\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0f44db79a3c4bbcace9e4886dbcf6d2 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12d04ed6372d4baea462edd62d2d55fe : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3603586fdaf4c6081890a8c213c1b1b : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb37b8acbf1c4977998c25a0c4cf72a2 : SUCCESS in 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"262647849788d86942d1b115fb8c764000b742aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 19:18:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73b8b7ed71dc4eae9f9ce42bff2c3812\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4efb4e63fc942978023d8aff7997f0d : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e2c9729277149d1b03fb65b0ea5ac1c : SUCCESS in 11m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39791fd988074c7da0276b835fe04e59 : SUCCESS in 12m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c2d610467cf4d7ba351f0e1800fdcd0 : SUCCESS in 10m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56dcac80f5ed4fb68f21c17b28e5a2cb : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8a683968844ad8b21dc61d10ecd959 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/0e188d0c108641e6bb20376f7b2120c1 : SUCCESS in 48m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96c5ca2ba4284294897913c85ac20dbb : FAILURE in 1h 02m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a1c52050c9b404f848a2c6db513b161 : SUCCESS in 46m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d81d55d0fdf42849488383063516548 : SUCCESS in 7m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/748eb6a60f684ecdb3bacc40db0039b3 : SUCCESS in 14m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/62b4af07aeb544ed99a135abaf9ad8e1 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e9b869dd4c1404989a9c08586cfd4d1 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c96dc3f6e255466980a30f0effa930fc : SUCCESS in 5m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8dfa58e2e0ab49fa8f63f5861d4512d1 : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f2e9307f93244ce9847c6fb68b8b0a09 : SUCCESS in 38m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6459985a09624ed1bcf34ab20be40381 : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e1d4ebec7614f5bb5dded690dcfc1a9 : FAILURE in 1h 56m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cc9b8d1e621f4dabbccffa7831b890b6 : FAILURE in 1h 30m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c657203955b45748a0862eb594e1da2 : FAILURE in 2h 03m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a22ce45026d14b589eba88ceb535d2ea : SUCCESS in 1h 39m 49s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/284df0a7da294bc2b943aee94eb77f56 : SUCCESS in 1h 13m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48fbcdda035d4b04b19e03fa42472e42 : SUCCESS in 1h 19m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38c0c175c8c34fa1abaedfdcaf285864 : SUCCESS in 34m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99c7aee2447a4024bee4365f99aa9eaf : SUCCESS in 1h 23m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a63b3666a2d14183bce1d275af09e735 : SUCCESS in 1h 41m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/61a9804db1b643ed91429e13b93c428c : SUCCESS in 1h 11m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2fe6315043a40ff89ef34f3078a9f78 : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":1},{"id":"88a9d5c961cb21b895c0742eab51bc31573af6db","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-08 21:24:19.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/55be3d4a747045828b6ee8cdce9ea7d3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9ca6caaaf0e54886a6f6849c0ac5ae04 : FAILURE in 1h 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8fafd37182ca9d240ee68646b4761a4d655ebb03","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-09 06:42:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6947f94873987365a400a1cd242e51f01287bc0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-09 06:51:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3862b4a247524e6b99084305ed2dec31 : FAILURE in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6337e1ba1d6de8f6cc21316e7c3a9d9e5dba23fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 07:08:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ccd587c777f4c9cb3ce478601552040\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8a199474b6646659a70d501a78bdfdc : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9050d83181a74a2786532eaf68e1622c : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b14f84f6171440d981c1ae1fc2850645 : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1996772f1d5d4ae881082cec7df2f451 : SUCCESS in 20m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"929d3d8298d20f54f36105351fa322bde11ae655","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-09 08:06:04.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6796531527c749f5a8f5638ff0f5a5c4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fd330ff72b1343d08ca061e1d7645a63 : FAILURE in 1h 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"09e9ba030b2c81afebaad2d07aadaa56ccda6274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 08:30:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03caadb975fe4ce78d923d6cbd03d6de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c64ad86b11745789ad210b68f98f3d2 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2c1ac107f6b4d6480d97385e72e1cda : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eaed9afc94f84f19b80a71b7d68c18b2 : SUCCESS in 10m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/863695819f844688a9e5f6eb5c4af68a : SUCCESS in 9m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b369463d2b747bead3143fe981d22ec : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd166a761bc4fe2ba12a0e7bef5dabe : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/6a7812e090d4433f8d83e5a35d03fdde : SUCCESS in 41m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b604005f565c476d8d13921ecc1cbd82 : SUCCESS in 1h 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07d100ff51c84b9a88f1d14586769630 : SUCCESS in 48m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b77016f51fb74ee8a678e315f997410d : SUCCESS in 7m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c381f6be56c24ec9b7c13b418bb195ec : SUCCESS in 13m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be4ddc90b08547b9ab0e77e612016f1a : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81bbb0900203443cb5a7876fdb365d64 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/284f2424c9ed47338222191fb53b906a : SUCCESS in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58c7369a308d46a8b95d16ebfe702ac0 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f77f4895fee4fc0aaa677445319b2be : SUCCESS in 35m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93c1c7c0fbdc450cb04c54c5226d9417 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15ab2ea6de04422bbafa6100e5bb329c : SUCCESS in 1h 30m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7010aa00654c40228681ce5bdac42e28 : FAILURE in 1h 46m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c3144f7a7e04364a6166ab32926a2cd : FAILURE in 1h 19m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5c8499b7d442461d8f715c10da0f4e2b : SUCCESS in 1h 32m 35s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/54ad0fe0a54b48e48033599becf231c5 : SUCCESS in 49m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/99acecc519dc4ed989a2f78797f7bc96 : SUCCESS in 53m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/219880e817b04899b0eb298119ee7b47 : SUCCESS in 36m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f2c3c6de359848af9012fbcce171132f : SUCCESS in 57m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fa2a7ca3d51e402f911bcca70deb6005 : SUCCESS in 1h 06m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e57f8fd0c5004174ab3b05da77908d2c : SUCCESS in 1h 04m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c10c8a4df2e45499c42c5f702f86627 : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4b20bed8b77e7ab5532b5710649d7dd2fd512622","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-09 14:01:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c75e601db61eb253c5e8922cd8ecffebce423fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-09 14:10:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ac4164bd0dd34ca491a5b7d1ba4296b5 : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad5aa6928abbb98df5de880613d618070d5adf6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 14:27:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b02ee1e1ab04a308ff3e3b40572d9c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c418172662346c09ab4471113745ec0 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fa3469b8aaf42dba5283749f2afd553 : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48fa3023438143d6af818e4a2d42c1b4 : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3dec137ce1a04d6283b53ea5035bc5d0 : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a89d03fce436adb4aa0f5098027414f13269436b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-09 15:27:19.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e621bd87365842369f1edcc1c5e8bd78\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/daff9e12784c440a9e1d7f0e8c597ac4 : FAILURE in 1h 23m 06s","accounts_in_message":[],"_revision_number":1},{"id":"356d114cbbe520453d97bca27ef445ad31ef3bc3","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-10 16:23:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8947c3ae95983b05e8591275563e1026102ec32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-10 16:32:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee01deda42f54a1cb8fcf7bf15008bf0 : FAILURE in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"20bea480e62fbb1802acc54c6bca8bb7d2c362e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-10 16:47:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a96ecb39d3c4a7e8b94b1c48fce6ff9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6837c7e713e490384bcc19e20a9c32e : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e08e4861ae53442fa046614e44742a99 : SUCCESS in 20m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/388b644fe31945159edf1cc7bad654a7 : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71d8c8df82c44142b04333bbabc3cf05 : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1978d8426b658cccd4563f61d725da4f3b36bad4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-10 17:53:51.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/56c42f28c0954c15a8a66d6f08552445\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5b224e48b1ce4f66af6ef8715b86bed0 : FAILURE in 1h 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5ac65f4f603df134d572f2ce5adfdc99bdadefec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-10 18:10:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bbd8138412b40f688372534061084a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e066b6504a3441188989bc05a4d49d0 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d9e5548c554c4c86a869b8a2e1bf89 : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64bd06f0c6af4f77ae971384ec021933 : SUCCESS in 11m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d4e728475764c3e965521ba5f8073fc : SUCCESS in 11m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c59fe15a57e4a68bba9ed095797c248 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7bebfc93e1465db6a06fe1dbc7cb5c : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a22fa83b25e549d983356cf15ca11718 : SUCCESS in 1h 00m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b29f6779a1f41d08c2e9ec5c47b2c1a : SUCCESS in 1h 32m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3083c1aa4e264ea5b48a658a5880f788 : SUCCESS in 1h 03m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ccd37dcc9e0411bb62bf3300444065c : SUCCESS in 8m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0affe1a36c564578af1432c119c65731 : SUCCESS in 12m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d556ee18b14a4939b2e90d52a7581d50 : SUCCESS in 4m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f625ee503b444288832b4e2b39e07ddc : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f16d7cf144194dce8ad7aed982ad436c : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a506e895d86e49aca9dd5c635b8a6409 : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d8e663ad3a6946199887bc4a82ab590d : SUCCESS in 31m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0a49691322864e258de8c441d52e20b4 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/31eb6a5829124353894ff67e31582e1e : SUCCESS in 1h 34m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/735771109ca04d4999f9c22efe281e52 : FAILURE in 1h 33m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c7e2264eea4d454cb5f3b5211ae8389c : SUCCESS in 1h 17m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2b221dab5eab40989b6f34fd24dc290a : SUCCESS in 1h 21m 34s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1110e0d855f4cfebcf49131678afbc5 : SUCCESS in 50m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/515c7802e4fb49a39521646e33c1244c : SUCCESS in 1h 11m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c51e96825314f159a8915288441ece0 : SUCCESS in 27m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/903d7204c93d4623ba413339d35bb03d : SUCCESS in 1h 04m 57s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/54e697e89c31400eb3d3cacdb731d413 : POST_FAILURE in 1h 46m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0129239787154ec198007ada6d661b6b : FAILURE in 1h 19m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f67cd2829a74ab28757665d90489c07 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"48c4a2baf8962f787ac2034abe518b5c39fd42a0","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-11 03:11:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3254a294ee40a99558b926c6f03148b1865e67d6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-11 03:20:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8b5621d56b894cbdb918b1d9042bbbce : FAILURE in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a292ad5153833bd3e8ca2be67048d9df77ed1966","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-11 04:32:18.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2b0457480fff4c738e75673712817bc1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a1cd0f8855064cf3ae1a86d98bc57e7f : FAILURE in 1h 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5a6e15ff6ac37403652aa455ada23c545eea5463","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 04:47:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df4709fa0a074acda1f38a48603cfac6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b54a44c3ea84d19b50641df6c4e571c : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1f85124a6b8420298f2ac51a53fd80a : SUCCESS in 22m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b754b9c430b74936abc7b4c1caee574f : SUCCESS in 19m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/668a8a2c27ae4e4aac0555840f4a5948 : SUCCESS in 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a345cd4d91f54a442a712c8f09fc74b1bf98c225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:25:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4ebbfdccb7494189713acaea7f6368\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58727e5aa81749259f088413d430e79d : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a3b6f790909442ca718b0ee94db39fa : SUCCESS in 12m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/789263b2737741dfaa063fb28cf438e0 : SUCCESS in 12m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2061d11c5684ab3914b790d30c3cbfe : SUCCESS in 13m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1480b624f3f546aea7786fa9850222ac : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec28001130b41ca961b5befbbfa693d : SUCCESS in 14m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/cb5d368b815b45c596b69d38a58be372 : SUCCESS in 1h 01m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30c4f9da3bf642d19456a1e20962db12 : SUCCESS in 1h 40m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e19e06900fd740b7ab0984400aa33a20 : SUCCESS in 53m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea193d668ea942c787af66f136c0c153 : SUCCESS in 10m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b3f8ad56cb14f83a9fb0e01f3ccee42 : SUCCESS in 17m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9aa7eda8d6b94704b0c68960eb8c8580 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef5fd2235cd34cdab0419e74d60b1cf2 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/547a02ce9d3740eb8edf0acc994896a9 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/300c32ec49d24c19ba7d86db9a8b1859 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7c1ac5490211426195f742b934fe0862 : SUCCESS in 52m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f37f9f7831e4f8ebad7407b99580923 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5b160feeae484ed98c00c849d0ed425b : SUCCESS in 1h 43m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/127f7460e0364451bc19cceec6b0c08f : FAILURE in 1h 38m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76a6f156629a4a5399809a099c0266ad : SUCCESS in 2h 08m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2f6e12f97ed34c4698f1a300be14ff7f : SUCCESS in 2h 04m 26s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/acc35c9aed064215be184395b016cf5f : SUCCESS in 1h 09m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b9b90c13e874d358ade7841310b94d7 : SUCCESS in 1h 19m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0015566b6a2e4e0b96fe00c957725cf2 : SUCCESS in 52m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8cdb9da7b75943e3ba445b1d1d1b96d4 : SUCCESS in 1h 28m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ccff97229a11475c866bc0601bd8dca6 : SUCCESS in 1h 20m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00dfcdd9a61e45d4859af7743b181506 : SUCCESS in 1h 13m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb6a4d432c864c819962ae7977eb952b : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"69c084ce1ce7b0ecd01cb24434bf727a0bc7168b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-11 05:25:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_iSCSI/04c12dd/DellEMC_PowerStore_iSCSI/273 : FAILURE in 1s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_FC/0c53c0d/DellEMC_PowerStore_FC/282 : FAILURE in 1s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_NFS/9c78158/DellEMC_PowerStore_NFS/282 : FAILURE in 1s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerStore_NVMeTCP/e0851f2/DellEMC_PowerStore_NVMeTCP/272 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"462dc28869378bf0d2dc55adabcf4162afa83953","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-11 05:26:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_PowerFlex/0d06460/DellEMC_PowerFlex/567 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ec71ecf4583998853bf98fe66e55f2c2bfd6bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:26:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a93bcf72f99c6a76b7cf650c09d2808a01b24c7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-11 07:04:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-fcp/67d0e1d/ : SUCCESS in 1h 08m 20s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/e1f90d8/ : SUCCESS in 1h 27m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-nfs/8376439/ : SUCCESS in 44m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/0baad59/ : SUCCESS in 45m 55s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/a448910/ : SUCCESS in 47m 07s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/889826/1/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/fdffd3a/ : SUCCESS in 1h 36m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"001d67d42fd1bcc966af26b813703b65d56d700a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-09-11 07:09:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/889826-1/Lenovo-iSCSI/ : FAILURE in 35:12.82\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/889826-1/Lenovo-FC/ : FAILURE in 35:08.78","accounts_in_message":[],"_revision_number":1},{"id":"839029de86642ad4d24b48ffe27b93e9f502d87d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:15:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e985ddccd8c24f698cc3db7c4a467221\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfaabcc0c90445f8bda5455ce40c39a6 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f5a140ba324692addf3edd557b9ee4 : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de8d26e6772a47d09ea29b9596273f43 : SUCCESS in 11m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2e08831cf8a4887b5c2b21f32cee260 : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2fa6f2104924bd88551b2e28488973c : SUCCESS in 15m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/ee22892d97324102a91b5141be2bc882 : SUCCESS in 47m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/383553d391424e6f927ddced022b1dde : SUCCESS in 1h 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43388a545f7940d39d8d7a4fb9cbb17f : SUCCESS in 1h 00m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9bdc27851864f8c9fb1f5a63867a6e9 : SUCCESS in 8m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bf2a149220bc458bad897908f76ee450 : SUCCESS in 56m 43s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a886a69463fb425aaa6591b3d5f79cea : SUCCESS in 1h 48m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8821bfde7d5438e9ba2781dc1a0828c : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fd490ff7584f08fcb85bb0643e8f7e35e1ca2566","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:15:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ecc68e8966efa7e2a6c1097929f31be60151a94a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:16:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc30bc24af954323bd651a0df2012af8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3db553c1c94e688eef19aa3d2e5822 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c348c539e26043a09377245099ea3a79 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"7ac693e19988313aae98664a76869deb23f910d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-11 07:37:07.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_Unity_FC/ca03d11/DellEMC_Unity_FC/765 : SUCCESS in 1h 57m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/889826/1/check/DellEMC_Unity_iSCSI/bf1280d/DellEMC_Unity_iSCSI/774 : SUCCESS in 2h 11m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e8ee8e74000a19f358f84f03b826553df0b7f00","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-09-11 09:07:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/889826-1/HPMSA-iSCSI/ : SUCCESS in 1:44:08\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/889826-1/HPMSA-FC/ : SUCCESS in 1:23:45","accounts_in_message":[],"_revision_number":1},{"id":"2366ae7529c9d4edc9ff2413760b8d088e19131b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-09-11 11:25:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/889826-1/PVME-iSCSI/ : SUCCESS in 2:00:01\n- PVME-FC https://os-logs.seagate.com/dell-ci/889826-1/PVME-FC/ : SUCCESS in 1:42:21","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64b5543d925d9b176595ab3787e4964fd40e5e91","revisions":{"64b5543d925d9b176595ab3787e4964fd40e5e91":{"kind":"REWORK","_number":1,"created":"2023-07-27 16:02:50.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/26/889826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/889826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/889826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/889826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/889826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/889826/1"}}},"commit":{"parents":[{"commit":"0dc8d75229fec860c6fd82ecd772c242839e4e87","subject":"NetApp ONTAP: Fix create FlexVol pool replica","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0dc8d75229fec860c6fd82ecd772c242839e4e87"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-07-27 15:57:22.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-07-27 15:59:56.000000000","tz":0},"subject":"NetApp ONTAP: Fixed errors on failover-host operation with REST API","message":"NetApp ONTAP: Fixed errors on failover-host operation with REST API\n\nFixed the functions _set_snapmirror_state and break_snapmirror that\nwere causing a failure in the failover-host operation.\n\nChange-Id: Ib4ebe3cafb67a16aef4b961ccf30dade99fdec7f\nCloses-Bug: #2028857\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b5543d925d9b176595ab3787e4964fd40e5e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b5543d925d9b176595ab3787e4964fd40e5e91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
