)]}'
{"id":"openstack%2Fcinder~899548","triplet_id":"openstack%2Fcinder~master~Ib386bfe1474fa17be82fb507095589f3ac18c007","project":"openstack/cinder","branch":"master","topic":"hitachi_aix_degrade","attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2025-03-13 10:14:23.000000000","reason":"Someone else replied on the change"},"37298":{"account":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"last_update":"2025-03-13 10:14:23.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-11-20 06:23:27.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2025-01-13 19:28:26.000000000","reason":"removed on reply"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-11-20 06:23:27.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"22312":{"account":{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"},"last_update":"2023-12-13 13:16:09.000000000","reason":"\u003cGERRIT_ACCOUNT_22312\u003e replied on the change","reason_account":{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2024-11-20 06:23:27.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"Ib386bfe1474fa17be82fb507095589f3ac18c007","subject":"Hitachi: Restore to support ``AIX`` as host OS type","status":"NEW","created":"2023-10-30 05:08:26.000000000","updated":"2025-11-07 14:46:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce2cdb5fa7425259f24fd97139d297d5f6973204","_number":899548,"virtual_id_number":899548,"owner":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-13 12:12:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2025-03-13 02:07:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2025-11-07 14:46:16.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37269,"deleted":true},{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-30 05:10:19.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-30 05:43:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-10-30 06:18:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-30 06:39:57.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-10-30 06:40:11.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-30 07:05:57.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-30 08:03:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-30 09:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-10-30 09:59:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-30 13:58:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-30 14:00:12.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-11-02 05:17:08.000000000","updated_by":{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"},"reviewer":{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"},"state":"CC"},{"updated":"2023-11-29 06:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 08:49:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-29 11:40:05.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 07:57:20.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-12-20 08:44:35.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-12-22 00:44:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-18 08:41:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 14:11:08.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-02 03:30:15.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-03-05 00:13:25.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-03-05 01:57:55.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-07-23 08:16:08.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-23 08:17:00.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-09-12 04:59:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-09-30 05:23:19.000000000","updated_by":{"_account_id":37269,"deleted":true},"reviewer":{"_account_id":37269,"deleted":true},"state":"CC"},{"updated":"2024-10-04 16:05:15.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-10-05 20:21:28.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2024-11-18 05:43:03.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-11-20 02:59:31.000000000","updated_by":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"reviewer":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"state":"CC"},{"updated":"2024-11-20 06:24:19.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2025-11-07 14:16:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"572d7a7303730dfd110b5a4e27b57ecb6614b354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-10-30 05:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"745f691e2624e6326e7cb3e90e02b47a938d5ae2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-30 05:10:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/774c0f7c78c74b9d99faad55affdc39c : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"b55a08d50dfd3c9d3bdf50124ee27889a9f1f269","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-10-30 05:43:27.000000000","message":"Patch Set 1:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/855/","accounts_in_message":[],"_revision_number":1},{"id":"daf0ee94324c8a0c225ea9b48b8dbb67a4c0119b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-30 06:18:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/48/899548/1/check/DellEMC_PowerFlex/7551b43/DellEMC_PowerFlex/823 : SUCCESS in 1h 09m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e5e4d9e3cc4574e4bac467412e722f470bb302e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-10-30 06:39:57.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/631049b5000d4c79b535c676b00516a9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fcb64a370372487ea493925cd2549d33 : FAILURE in 1h 22m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fc47d557ee19e8ce28be106a789dcfda01977368","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-10-30 06:40:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/48/899548/1/ : FAILURE in 1h 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2c37ec08b903f6e563e08c49ec24d3da76264020","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-30 07:05:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/48/899548/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/7da0c99/ : SUCCESS in 1h 55m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/48/899548/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/e5d34ff/ : SUCCESS in 1h 44m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/48/899548/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/7331a1a/ : FAILURE in 1h 49m 52s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/48/899548/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/9ab4491/ : SUCCESS in 1h 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c6b55feba3276a85097f382cbb4a6c86bd2c935a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-30 08:03:34.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/1/check/DellEMC_Unity_FC/2969f39/DellEMC_Unity_FC/938 : SUCCESS in 2h 11m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/1/check/DellEMC_Unity_iSCSI/cc606ea/DellEMC_Unity_iSCSI/948 : FAILURE in 2h 54m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b30dc0784844870cf7a4f4540c15b66b4a04564","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-10-30 09:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/1/fc : SUCCESS in 0h 26m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/1/iscsi : SUCCESS in 0h 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bd56105b1a97f6f2ae1d5ac9d81930f9f7a12f28","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-30 09:59:23.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-899548-1/FC/ : FAILURE in 4h 26m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-899548-1/iSCSI/ : FAILURE in 4h 50m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc6d6601c85b848dbab84be808e3b2d5dd23747e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-10-30 13:58:04.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/1/20231030140827/NEC-FC : FAILURE in 8h 34m 44s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/1/20231030140827/NEC-ISCSI : FAILURE in 8h 49m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0898f16515902a30484e9416d61f8c1bfd114f67","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-10-30 14:00:12.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/1/20231030140827/NEC-V-ISCSI : FAILURE in 8h 00m 16s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/1/20231030140827/NEC-V-FC : FAILURE in 8h 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a5af3936cce1c3a915ab9f862d54dcca715b4151","author":{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"},"date":"2023-11-02 05:17:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb73c84905a671ac9224a54bc60cc9c0701060e4","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-11-09 10:30:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58a09b76c3ed3da208f68a09ad1f919324048e35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-11-29 06:18:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3f5b4d1f5f3174bc3b5de26938c6bc60e830a36e","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-11-29 06:20:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36d90715eb7cb8c9fa9bbe29c3e8549cc2e48270","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 06:39:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c14b03233144fb7a5c4e64ca44242d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dea0fc67486746e0b6c0e641ddfa0756 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d63338a4d6c4425a9c877ef96482acc : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39caed506b6149b4ae7bc577a9c8d661 : SUCCESS in 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"918d7336aaacfa7ece9ce692c08276f6fe93e065","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-29 08:32:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/81e60a16918d48e7be06a4817c8d7f07 : RETRY_LIMIT in 52s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/183eb10ddd10434a93f2dc4b39645b4c : RETRY_LIMIT in 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/716c938ce52343cb96f781dfc37f37cc : RETRY_LIMIT in 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/4f97ca60b6c440ab9167020654ca9d5f : RETRY_LIMIT in 53s","accounts_in_message":[],"_revision_number":2},{"id":"e3832b3d2d8518c92af804fd21a6443cf8ef694f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 08:49:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cda971a4455941a28cc987eda0a26576\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e37c542a6e4419abec4fb395fd72e9 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47bbd59adc4b4169aabb3f47fe186745 : SUCCESS in 9m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65d3ad7fb600497eac0675ec6454551f : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4af929adb6f4cc2b404a527b3683029 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/3d745f5797bf4c468e706acdf7d47ab1 : SUCCESS in 1h 16m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dc902bbde32f4f2fab617ec903ed1f30 : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c555ae4f601d4bea98c746f542ce39b3 : SUCCESS in 1h 30m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c83eba748084f01be64baeed32ea36f : SUCCESS in 1h 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502f63a0a1c24a4e9db6ebd297ef6f93 : SUCCESS in 7m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f685f024b6c14973a3a016fb7f055cd8 : SUCCESS in 18m 34s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/554c8b3bd84e4deeb09d3188e495c1eb : SUCCESS in 13m 23s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ecac94ec735544b282ce3b4292c578a4 : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2505501e2ba8458bb7ec37bbb73b73ad : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cb805a8faa24eed8894964d935d7ecd : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff118600e65340df80786944141ebfe4 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/20e6b46bb53245a8950b82b229f4bdc8 : SUCCESS in 1h 05m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/387c1601444541bebeb8b328d2f658e3 : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/164bfa13c7d4445b9900a8f0f6dd58ea : FAILURE in 2h 00m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1c83b5b4b374920a6d21faa164b1d96 : SUCCESS in 2h 08m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/584fe819a11c45b0980bba6d4c4ff250 : SUCCESS in 1h 21m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0147e3e483894d5c82198eb78006c151 : FAILURE in 2h 29m 52s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6b5014a9e0ed444ea06789364efdbffc : SUCCESS in 1h 19m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a63fab1e0de4f82a0b83b87f584235d : SUCCESS in 1h 45m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/963a8f2881454ffbb0be357987071bf5 : SUCCESS in 50m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6c998dffd7f344e7b2f415e11169d71d : SUCCESS in 1h 02m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0f861f3be2d5441db8914485be95844c : POST_FAILURE in 2h 03m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d213448a15e2435195d6d622758cb7e5 : SUCCESS in 1h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5132041ff404461af8a3699424e38b5 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":2},{"id":"461006db26f5e89b6440100592dea1b43852119a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-29 09:37:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/48/899548/2/ : FAILURE in 1h 21m 04s","accounts_in_message":[],"_revision_number":2},{"id":"8dafdcd73bd534cfc7996f9b3703e575e9c3c7c8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-29 10:17:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/c5a7d3c4f9af43d9aeaa4195f4ab0936\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e3b167abe39043f08cb22f561c3ae51b : SUCCESS in 1h 22m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b2a115e671e680ef2ff620093874439936a23c14","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-11-29 11:40:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/48/899548/2/check/fujitsu-eternus-dx-iscsi/2564c91 : SUCCESS in 1h 53m 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/48/899548/2/check/fujitsu-eternus-dx-fc/edba25b : FAILURE in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"56025dd5f68a33a3da84a870588cc281c46abed9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-29 14:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/2/fc : SUCCESS in 1h 23m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/2/iscsi : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":2},{"id":"82f0a23cbe4df724b16891b2bed2205b3a6b2a17","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-29 15:26:41.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_PowerFlex/055576d/DellEMC_PowerFlex/952 : SUCCESS in 1h 14m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ab6f93f3b88f6325bdb2c51f0154215aaa8f808","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-29 15:57:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/176/.","accounts_in_message":[],"_revision_number":2},{"id":"c60eedb68efc829ac2313247d2e11dcf8e6dcd6b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-29 16:22:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3edd9b02de1448029573d1810efad02d : SUCCESS in 1h 11m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/35533d883bad45348911e3f8589121fc : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a4375c82f9122a080217ad38f1d456127614614d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-29 20:42:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/172/.","accounts_in_message":[],"_revision_number":2},{"id":"6073382b7b8d83129585c96150b6befde5732207","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-30 06:41:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/2/20231130121110/NEC-V-FC : SUCCESS in 3h 23m 42s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/2/20231130121110/NEC-V-ISCSI : SUCCESS in 3h 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"66763c6bfeea0fe725719893fc4f48ce5c8726e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-30 13:08:22.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_Unity_FC/7c6564b/DellEMC_Unity_FC/1019 : SUCCESS in 2h 18m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_Unity_iSCSI/e1a01fa/DellEMC_Unity_iSCSI/1034 : SUCCESS in 2h 15m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"44b6142cd86b4445394de5cba2bf457ddb8b160c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-02 03:11:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/2/20231202034709/NEC-FC : SUCCESS in 3h 55m 21s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/2/20231202034709/NEC-ISCSI : SUCCESS in 8h 23m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0aa489d19053f93d6ea54bddb05f23ac8c40482f","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-12-13 01:41:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf66a891ff6d7dab452309526948fb79a4ce26a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 02:07:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9059593ea88497488ea6898865b3fe2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67b2c6e9599d48cdbbf1eef666654888 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/619417473c2b4d2cb15cb299a652ff77 : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d141b0523dff40da8ec7ea5fd056eaee : SUCCESS in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7ac81d280468b3d45effed87865eedbf3f8bebd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-13 03:18:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/44e5b31678de41319a47ed91ade2a524 : SUCCESS in 1h 13m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7e8aa30513fc41efb0a275bd7295585e : RETRY_LIMIT in 13m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e4dae8510fcd159716196250210a20b9a535104a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-13 03:45:19.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cb3e960d5d2e45ccae1711205e94c028\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9be0e99d2dab4b6cb1427ed32a7c4c34 : FAILURE in 1h 22m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c042a2d966e168f436c3208c841a7c96fb00b996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 04:12:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5164af92d28410194ecf5f9bd5cb449\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6539938a16fa4cc6aa644f95bfd5afbc : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8430743a1e6413888515e2f11e9139b : SUCCESS in 11m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c43516fd90d14a96b06f7830a760969d : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fdb0b05cbb407582cfbc686d5777fc : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/3918c802bd224b2f8be276f584c9becd : SUCCESS in 1h 07m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/45e9e71e300a45b69352286a6fdff1b3 : SUCCESS in 55m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/322f5dda96504c9fb4040506554e6694 : FAILURE in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/298b19531ffb43b6a5bbb42f02090755 : SUCCESS in 1h 03m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1207b64af4a4035bad2f62eecb6fd0c : SUCCESS in 8m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9f39576078c45e19edae4d27b8acfa8 : SUCCESS in 14m 14s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9ee1264710964b9aacb86d73c72356b4 : SUCCESS in 12m 54s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/49e85614d37e488cae61b81549c0a04a : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/094c329ee65f47c1bd1419dd69f6d278 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38dc8b003a0441b2b3f62ac31843e02a : SUCCESS in 5m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/768c06a4cd744c97a0874d3523d35b1f : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad1525539946414da8c63aa17e03c3c7 : SUCCESS in 1h 00m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eaac772c288e4dc49e9aaa9355d77623 : SUCCESS in 7m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9a793429a3440e2bfaa8d346853254a : SUCCESS in 1h 42m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11bd1f5812fd4cc0b630e2e0a2ad23c4 : SUCCESS in 2h 19m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6b0b64e0fc5b444c9b44db9ff76bc61e : SUCCESS in 2h 06m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dd0532de902347ea975d86e4010c0556 : FAILURE in 2h 23m 29s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c2b6d132c02f4da5aca1a79f8271b596 : SUCCESS in 55m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f159e95dfa0a45659fbbc00ac86b0232 : SUCCESS in 1h 12m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f449d83a3ea4489d91f1ae6f8e606a30 : SUCCESS in 42m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1d156acc074b4992bb6775c6ff78de70 : SUCCESS in 1h 14m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4595fb5b1dfb43cea9d0c84935b62e08 : SUCCESS in 1h 44m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0de22809de5e4540ab97fce78d69f476 : SUCCESS in 1h 41m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8be124a2d524dda92a657e088cc01d0 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":2},{"id":"453dfa0f8847bbefc91be5f18a7f648ecce23087","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-13 07:00:09.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/2/fc : SUCCESS in 0h 47m 47s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/2/iscsi : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bf957310f9489e59627af834a99eb70b38b9bfd2","author":{"_account_id":22312,"name":"Andrei","email":"andrei.perepiolkin@open-e.com","username":"andrei.perepiolkin"},"date":"2023-12-13 13:16:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f352395be2047c7a5c48c313bf00a3fc3483fe36","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-12-20 02:24:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d147f6efaed44b570fab0795f147ce3abde17e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 02:56:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/413c50282d2047328e2f29a815efe34a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb35bdd814294c15881f786d619766d4 : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5441c3677dd74779875cd7bcd413fb63 : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2d0ccc0bbea4515a582c3883c66fbf0 : SUCCESS in 25m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cda69a9c859a24185c70abbb2e87c8351ec6e0fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-20 03:39:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a554ed7fd0d6418fa905d4d5f6ecf2cd : SUCCESS in 1h 10m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c9a0543f61be45c4bdb84df7f46675d8 : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":2},{"id":"9ca283bda8f9819c2dfe0c10774281dba45ca530","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-20 03:55:50.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8acaf7cd705d46aaa733945c6836d4f0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2eea939fcb3c4d5d937687721c0f322c : FAILURE in 1h 27m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ea9348f5326e911b57e15fed12bfa86290df88a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 06:19:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08870d875029430c85cb80973d19f970\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c82256264b934fa18112cf797e7aab1c : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9038d9875bb645b694a74ea14170d29c : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33016f5baf5a41a9ab3e68b3a3fabdc3 : SUCCESS in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4ea6a5efca4dd9bc4086e0c57a7154 : SUCCESS in 14m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/53b9d9750510495baa077ff1d3626f4e : SUCCESS in 1h 10m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9a63cf1d3f04477bb301453ce601c12 : SUCCESS in 1h 06m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4c5ccb9d59e4210b12bae295baad0ac : SUCCESS in 1h 39m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7667cc761274a34a1353d9544ac2b0e : SUCCESS in 1h 02m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f632db41709944dfaa8208189602bef3 : SUCCESS in 11m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34e1d47f67144aa0a720555ff58502b5 : SUCCESS in 16m 17s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d8015b7d0edd47b5b3f7112f5da592c5 : SUCCESS in 12m 04s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/703321efadd147bbad89ac34ef5a6b32 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9db56c6c4b544974b4960fc310efb360 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6abde73365ea45b5850cd1ab0d1cd1b4 : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df6a80b3a2464640a59380ccc6cb48c8 : SUCCESS in 7m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/029d0bcd508444e2b8c761d83c9c6865 : SUCCESS in 47m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cd8eb9cbbb7435fae1baa941e4fa0fa : SUCCESS in 7m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3008bcc9ebdd4af898e8e0bdd90644ea : SUCCESS in 1h 41m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/043cc06e56a940b6b7c63108757ba9a3 : SUCCESS in 2h 05m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4ca3190035eb44bdb36a4bc68eb80462 : SUCCESS in 2h 43m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/74d9634881d947328285bee4204feb28 : SUCCESS in 1h 37m 17s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fb9c31584bd34e59bca119f70f7d1983 : SUCCESS in 1h 06m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61ee6a34ea494c6ba98a419569c566af : SUCCESS in 1h 41m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/787aafa02aac492b9c4ce5e8bf9f97c4 : SUCCESS in 34m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/16651d26f4d34515a2dc32881b031cfe : SUCCESS in 1h 21m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ab4529153046470db155dc3ff3930fd6 : SUCCESS in 1h 44m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/25ea376711494d25bfcb90489bedbbeb : SUCCESS in 1h 02m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0797dde9ee6c4a6daade31e41d4c7b0c : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":2},{"id":"cb40a93a2653f278c42ae88c27a5a5a4e3aad799","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-20 07:57:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_PowerStore_iSCSI/b63008c/DellEMC_PowerStore_iSCSI/478 : SUCCESS in 1h 09m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_PowerStore_FC/be26b54/DellEMC_PowerStore_FC/481 : SUCCESS in 1h 22m 50s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_PowerStore_NFS/a780ab7/DellEMC_PowerStore_NFS/487 : SUCCESS in 1h 12m 27s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_PowerStore_NVMeTCP/483693b/DellEMC_PowerStore_NVMeTCP/471 : SUCCESS in 1h 37m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"92dad2bf3f41bf9edb8281cd224049df3bcdef9d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-20 08:44:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/2/upstream-check/cinder-tempest-plugin-ontap-fcp/f9bf4e4/ : SUCCESS in 1h 15m 12s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/b0d203a/ : SUCCESS in 42m 11s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/2/upstream-check/cinder-tempest-plugin-ontap-nfs/b089d46/ : SUCCESS in 1h 59m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/53b4b05/ : SUCCESS in 42m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d6cab98e0d926262727d88332e3c6ede3efb545d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-20 12:31:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_Unity_FC/b708885/DellEMC_Unity_FC/1133 : FAILURE in 1h 15m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/2/check/DellEMC_Unity_iSCSI/7f05671/DellEMC_Unity_iSCSI/1145 : FAILURE in 2h 07m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4a8db5864e78dbac2e199ca3ff86d41bd2537d25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-12-22 00:44:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d85cd12c9e47d3d8189ca23e57dcd2e31f2b2ec7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-22 00:44:18.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16309/","accounts_in_message":[],"_revision_number":3},{"id":"1046aaa1c31c45a2830ca325dd1c279388747e58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-22 00:44:19.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16309/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4dfea7284252c3a3eff48743bc74348d602124c2","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-12-22 00:46:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"972ef65788ae9b8260c471a256341117bd00d4fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 00:56:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2ee4a25d514fbbab447991c61bfaad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4fda3240b28450aad33700f5fa60851 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b84fa5d4e7e041908415b91aee8e94c5 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd90e149469a40e486ef11928750045d : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbd245ff3ee48d17586f67814f32103847ff488a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-22 02:01:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/01a3fc2cc3c6472ca50e96191bada54f : SUCCESS in 1h 13m 55s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d33685d6c34b4807812963d25b71c9fc : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":3},{"id":"82826ac731c1d8ed810c1fb319a40e67a6371020","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-22 02:12:53.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5d3530c7c8504f9e96def9559a7847ba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dcd473cb00ca4425b09283116b9b43d9 : FAILURE in 1h 24m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3876dcb5f2419210972704cbdac6f8fc55e40c78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-22 02:18:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/311/.","accounts_in_message":[],"_revision_number":3},{"id":"4300d0c31f97763f3834d7755de6bd4c11f1f1f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-22 03:15:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/48/899548/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/4836b79/ : SUCCESS in 2h 22m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/48/899548/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/5109415/ : SUCCESS in 1h 58m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/48/899548/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/d5b5832/ : SUCCESS in 2h 24m 46s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/48/899548/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/9b7ec5e/ : FAILURE in 2h 29m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f6b227451b7bf6797678304633c2dd5385e17d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-22 03:23:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcf00b77f3041e2adf3ff294cd18e76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877e0ad7ee774645a7a00894af8d7e90 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d9fa48e0c8f44878711bf6da9e7243b : SUCCESS in 11m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a643d8fe6d354bf6ba6faf5bcd51fbea : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394e47a4cc174b44b67b719ee5f240e3 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/1f409eb91c264e458b3cc3d066387819 : SUCCESS in 1h 01m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/089d12761bf34769a1004417c92772cc : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f51bf7456724193be676b5b290ecb7f : SUCCESS in 1h 34m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ba33d2a80104f02a74b8e7e8fe54a2f : SUCCESS in 1h 05m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db08ea101d934005961fe2329f51a391 : SUCCESS in 7m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0501dc1d45ae47719bd52874eed2ce73 : SUCCESS in 14m 58s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/677f702d3fad4a13b4e144d184446705 : SUCCESS in 12m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d70df82e30d74846843d0b8953c46c06 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29e835e5874d488d9f95a08f9906f2e5 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/01dd67dd6aef4ff7bd28918b9f8f6e75 : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/209c424969d74db6946daaaa59769277 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d8cbb7d4cb17426189925f6491e2c940 : SUCCESS in 33m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a9d2a266cd0642d4b963551a73419bdb : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57ca81c2e702466384df7942e2e11ac5 : SUCCESS in 2h 01m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34711e783664415a8373833baf7103ec : FAILURE in 2h 32m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/284263b6853f4e4e9bef8f5511377ff0 : SUCCESS in 2h 11m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0b1d88b3ead14969985d7af77d95fe83 : SUCCESS in 1h 36m 07s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/64c57eb60d0541f1a23b9980ba14140b : SUCCESS in 1h 11m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e1058d210078482c9a8318b944008f3f : SUCCESS in 1h 11m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a5c7bb28b31454b85c13d65d749a020 : SUCCESS in 41m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3002f45e92704d22a8bdce593da75587 : SUCCESS in 1h 22m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bea0d5ed27b6438681e87757f199caf1 : SUCCESS in 1h 29m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b313c243ced446519ad42c6f48812697 : SUCCESS in 1h 45m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/708f837d0e4c444b95f9aab266a974d9 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b4ae4b23701563fd8739fb194ac2f93690cead45","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-22 04:45:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/3/upstream-check/cinder-tempest-plugin-ontap-fcp/97de7ce/ : SUCCESS in 1h 20m 13s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/0cdb540/ : SUCCESS in 1h 20m 27s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/3/upstream-check/cinder-tempest-plugin-ontap-nfs/d6969a3/ : SUCCESS in 46m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp cinder-tempest-plugin-ontap-nvme-tcp : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"20b00b46081083822ba2284aa3d8a28298194349","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-22 05:00:11.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/3/fc : SUCCESS in 0h 53m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/3/iscsi : SUCCESS in 0h 50m 33s","accounts_in_message":[],"_revision_number":3},{"id":"8586dc8974f6b48e736481c95e6d5492bf241b5c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-22 05:19:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/3/20231222103618/NEC-V-ISCSI : SUCCESS in 3h 08m 46s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/3/20231222103618/NEC-V-FC : SUCCESS in 3h 43m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d8faa1580e3b4da515605a8a106654194e97e86d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-22 05:32:44.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-899548-3/FC/ : FAILURE in 4h 47m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-899548-3/iSCSI/ : FAILURE in 2h 57m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"3a2dd824227b3c0ba792b708a1db36b509dc1a7a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-22 09:54:56.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/3/check/DellEMC_PowerFlex_3.6/09dbbbe/DellEMC_PowerFlex_3.6/1106 : SUCCESS in 1h 32m 59s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/48/899548/3/check/DellEMC_PowerFlex_4.5/6c3b1e8/DellEMC_PowerFlex_4.5/83 : FAILURE in 1h 58m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"60fe4de90d912727885498241bc945f6b9a6c18b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-22 11:30:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/3/check/DellEMC_Unity_FC/4166e30/DellEMC_Unity_FC/1153 : FAILURE in 1h 13m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/3/check/DellEMC_Unity_iSCSI/30f4820/DellEMC_Unity_iSCSI/1165 : FAILURE in 2h 18m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e0977a1b5c26654efc9979e46699340392adcbab","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-22 12:00:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/327/.","accounts_in_message":[],"_revision_number":3},{"id":"23bdf345035a3cc6dcaaaa53d072bf60a6ec3991","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-23 23:42:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/3/20231224044842/NEC-ISCSI : SUCCESS in 3h 46m 09s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/3/20231224044842/NEC-FC : SUCCESS in 3h 53m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a14439dcb5a5c7fe1ae50e32229a05f73615105d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-25 16:59:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/48/899548/3/ : FAILURE in 1h 31m 39s","accounts_in_message":[],"_revision_number":3},{"id":"da8499aafd014d63dbc918231ed9a3d84c25212e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-01-18 05:17:23.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"53d3c84d9aab32429830114397abc51a4b7bb0b9","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-01-18 05:17:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a76d5f894fb463d329fcf5594572e4c6b4d71787","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:32:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae306f368a3449282b9a8735013d6fc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1dafcc4e9ce4772b5d79cbafd057b25 : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fe37c7d2ba34656b56f75ac72153e7e : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ead312ccce44c118323392fed708656 : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d23fcf15da4d27c34823686a9ef98fdecb01663","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-18 06:03:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1d7c9479d4ea41ad8d17a6ce63a3d535 : RETRY_LIMIT in 6m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6dbf6fe5d85f42a5953efbe639983d53 : RETRY_LIMIT in 6m 36s","accounts_in_message":[],"_revision_number":4},{"id":"8de8f7f55898673731207077005c1b306ccef83e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 07:58:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5140d40c6c3f4b0bbd7fd08d82af6533\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce93e1634ed459eb6c53851abbdb1c2 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f0a27620c054bdf8332709fd1fb2267 : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6993c162025c4c9486e0c806995d6a9e : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425ad81f533f468990fdf794532227ab : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/f0d6201b36774672a1825b54f6f0d359 : SUCCESS in 57m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1d0d82615b62490a9c73bdedc7c4567e : SUCCESS in 45m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68745ca7cd3e453e884af0d0c006e0d3 : SUCCESS in 1h 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7909ac27386d4199b52dafce697381e6 : SUCCESS in 20m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e00942fd2428481889e0445c388711c2 : SUCCESS in 8m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b9989fe5b6948c58bcd0e667c45bfcb : SUCCESS in 16m 53s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dc020744b01c455ea7bd77e7f2d1c3eb : SUCCESS in 13m 23s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/697aa1d0119f4ba48ae4ab3d5fac628e : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dfa3c5a08b984239b59e3fc57ca0bba8 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0028f5deed444279a165f767aedb6637 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c3ced4a19b04e348a6e168d59bc836a : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0aea3679b8c149d795b8bcb2e2fbcd63 : SUCCESS in 54m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4be78fbf9ee847d7b84986a9a8f94571 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce80f6968ad44327aa4ae92a040dcfed : SUCCESS in 2h 02m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2d73c8c23dc2444e94b0b24a697d7d37 : FAILURE in 2h 28m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d87bbea6c6f94510b0d9f0b18f6cd341 : SUCCESS in 1h 43m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2d0229f4d56943f797a44d3f2d4fba8d : SUCCESS in 2h 03m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/374ecef2ea7d4cf4b4dab5279d835ec5 : SUCCESS in 1h 16m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b14a45c1cd0f413d95716554d708912e : SUCCESS in 1h 40m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d6d4018f78f94650b6dfad59d7be2745 : SUCCESS in 54m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5ca995984e574a24a4db5a9b0f843448 : SUCCESS in 1h 30m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e9c0db17050e4021bc8efe8c34256584 : POST_FAILURE in 2h 17m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3d9ece250734031966c9fbd8a913350 : SUCCESS in 1h 29m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85ff0df0e84b43e6ac4a25d184fbda5f : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":4},{"id":"465756adf830a4c88df714509da33c40613a22c1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-18 08:41:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/899548 : FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":4},{"id":"b4bdeab35cfb67701663259c1a6b7446e5eabd33","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-18 08:46:22.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c113cc7dc0c3430da32fa0787e205fe0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3eee3ea46e58488f97c80b6199b73555 : FAILURE in 1h 25m 15s","accounts_in_message":[],"_revision_number":4},{"id":"171e86321e537adbb826e9b0a0729ee6979d4f2c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-18 09:33:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/4/upstream-check/cinder-tempest-plugin-ontap-fcp/2f9ccba/ : SUCCESS in 1h 33m 04s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/ebbdc7d/ : SUCCESS in 44m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/4/upstream-check/cinder-tempest-plugin-ontap-nfs/5f2657d/ : SUCCESS in 43m 34s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6befe33/ : SUCCESS in 47m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"9099a243e8b5b80f92a282c3437ec1d67af8bd2f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-18 10:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/4/fc : SUCCESS in 0h 48m 18s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/4/iscsi : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":4},{"id":"019c44f52dc7e43dd9479bca85a13b61df718f8d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-18 14:22:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/4/check/DellEMC_PowerStore_iSCSI/8d8b5c4/DellEMC_PowerStore_iSCSI/955 : SUCCESS in 1h 02m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/48/899548/4/check/DellEMC_PowerStore_FC/dfaefeb/DellEMC_PowerStore_FC/1111 : SUCCESS in 1h 23m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/48/899548/4/check/DellEMC_PowerStore_NFS/e1fa980/DellEMC_PowerStore_NFS/641 : SUCCESS in 1h 03m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/4/check/DellEMC_PowerStore_NVMeTCP/bf0ceba/DellEMC_PowerStore_NVMeTCP/507 : SUCCESS in 1h 44m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"8dff7732ae422899b317f0941de2e5e619caedd9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-18 15:07:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/457/.","accounts_in_message":[],"_revision_number":4},{"id":"dfeda18216565646297a198ac7eb3a51f0f29c53","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-18 15:47:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/4/check/DellEMC_PowerFlex_3.6/af4c77f/DellEMC_PowerFlex_3.6/1241 : FAILURE in 1h 18m 00s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/48/899548/4/check/DellEMC_PowerFlex_4.5/677fea3/DellEMC_PowerFlex_4.5/212 : FAILURE in 1h 56m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"796a0ad5e30e20d06d19bdffe7f6a7a8125952b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-18 16:20:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/446/.","accounts_in_message":[],"_revision_number":4},{"id":"478c261bbf794470f912520c7d8dd6d19f09e6be","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-01-18 16:50:22.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/4/20240118180554/NEC-V-ISCSI : SUCCESS in 3h 11m 38s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/4/20240118180554/NEC-V-FC : FAILURE in 7h 44m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fad7688552f717cdae5dc39066fc6f874ed08a4a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-01-21 22:11:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/4/20240121223558/NEC-FC : SUCCESS in 3h 52m 48s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/4/20240121223558/NEC-ISCSI : SUCCESS in 8h 35m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b8198e79a7eeefca9ced50be38426229b187e41e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-02 05:21:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"14942800fc216ecf036cab745fb405d50aed1ccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 05:43:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e976a08081054737a1d73786da70f7c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15534b7378b844a685d5e752be3c82cf : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6874ddd4ada340939cbaf45522dba3c5 : SUCCESS in 15m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3849df159d64213adb76c4ddb794e1a : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"63e60db2dc1cbb9beb458db67085cc40eb63840c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 07:56:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02f26ccb65c542c8acb452ac308e0dcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a20b084dd24b45988c823a7c116c26 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e215aa0929304162910dd2b01e36cd09 : SUCCESS in 10m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c346ca27568a4ba4845bb68ffaa6daf1 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec1ef15b4ec449d8ecd054bcdc60f5d : SUCCESS in 11m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/6b5a219fce434e78a64f849290657045 : SUCCESS in 1h 06m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cc0c7d79f2224d2dacf84ff7f8729a72 : SUCCESS in 1h 13m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60798727adad455394caa4f0c78ff617 : SUCCESS in 1h 04m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45b7815c4c6e45258eb8a766e4ffbe9d : SUCCESS in 39m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/992b4eb5f39b43e79974f698692517f1 : SUCCESS in 8m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c73c6570e3544feeaef3265c7a5ad5d2 : SUCCESS in 14m 54s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/06d81bd61d2b4b0c89d1b5e2eeed6ddb : SUCCESS in 12m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/873a2deaf49045769a7507f9286400d3 : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03791a21a2c34d1ea7b2a1edf489f29a : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62a799d1cc1f49ce9e9c86f09da69d13 : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c787041aad74423875bd7b31a92f245 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6c517e8123b04a80a109b47c7805341c : SUCCESS in 33m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05b77a033b834807b3dce839980d2fb4 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/186b5c2541344e2e84faefb6d4348928 : FAILURE in 1h 33m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/692399f1070d4684a7825c8d884433cb : FAILURE in 2h 24m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d8f6ffc8cee4ddbb80b93970454a75d : SUCCESS in 1h 39m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e2b9ded5878f4e6280db9b0c8cb9105c : SUCCESS in 2h 10m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eb28c72961a54288aa668cdd5a4b3dde : SUCCESS in 1h 15m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fc80c89f6c75411a81d2371dde4f4821 : SUCCESS in 1h 07m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7be8eb62fa7649dc8d593469468a1d1b : SUCCESS in 50m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c2b51542da346adbaf690fbef4d8e90 : SUCCESS in 1h 03m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/556e1a2948d84ff28ed172e92689c7fc : SUCCESS in 1h 40m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a0f9cd838994578ad7a3b70661dfbe1 : SUCCESS in 1h 39m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9ec51d00b714dec9078866dd8bbca27 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":5},{"id":"00951d824e5a99f555f6c43d72252fb91f41cc3b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-02 08:23:45.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b1f4d3fee1784b71b6ad53b824dfd658\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1a6ac58da2b4498f9deed6bbd656772f : FAILURE in 1h 22m 09s","accounts_in_message":[],"_revision_number":5},{"id":"d7864c3360be6b7fa9ffbe1d308837642b33f83a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-02 11:40:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/899548 : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":5},{"id":"bdc950f436a6827f7177028dc6e44f861be2ad23","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-02 15:56:13.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-899548-5/FC/ : UNSTABLE in 25m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-899548-5/iSCSI/ : FAILURE in 6h 29m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"c434f30cd69cee4f2d78d80fce1d99e10be6c29a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-02 16:05:05.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/5/check/DellEMC_Unity_FC/158c377/DellEMC_Unity_FC/1206 : SUCCESS in 1h 11m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/5/check/DellEMC_Unity_iSCSI/dee78cd/DellEMC_Unity_iSCSI/1225 : FAILURE in 14m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"67404648ef072b42cc5946ae38bdcb66b2be0a9e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-02 20:38:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee19183f602447f59453297ee5a56517 : SUCCESS in 1h 12m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6a088b00cc0140598488d218ab4f7fa4 : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":5},{"id":"20ba918ace94f6e154bfcdaf96f3e95f0d074bc0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-02 21:54:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/572/.","accounts_in_message":[],"_revision_number":5},{"id":"725e5091643562ebcf7f7f56e05130a32bf9ecf2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-03 02:29:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/584/.","accounts_in_message":[],"_revision_number":5},{"id":"6e0882499abb8252f948e43ce9c1b084b4a519cb","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-03 03:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/5/fc : SUCCESS in 1h 01m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/5/iscsi : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":5},{"id":"2f954e58e79eea62348d7af9ca43ce20bc53e658","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-03 19:38:58.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/5/check/DellEMC_PowerFlex_3.6/80b469e/DellEMC_PowerFlex_3.6/1355 : SUCCESS in 1h 33m 43s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/48/899548/5/check/DellEMC_PowerFlex_4.5/9cec1c1/DellEMC_PowerFlex_4.5/323 : SUCCESS in 2h 04m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"812bdce62737cd0c4c2fecce6986690793074cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-07 02:16:26.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"63388f6a448186b05f0b8a9910490f9b3daee18e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 02:44:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4f3571dd8f49ae97690bdcbaf249e1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c5af84fe9e245b5831718dbcc2d4dbb : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c16dca70ab7942139e975f4049683d6e : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9ee3b2b078847acb6fc40a49dded43d : SUCCESS in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f47eedf2f4e16f84375d887de5617ab2b5d8baf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 05:33:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0be021aa245407b9b1cee5e601cce54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64953c3edf3a4fd99a18855abcd07abc : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06bd6603d90f45b3a802808cfc4a9284 : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/818cd60f8ec94824a32f7dc8f9744dcc : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cd7230d5e943dabe7715b6f946572d : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/40ed5506e1814c088f3847c7c1366ca2 : SUCCESS in 1h 06m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7248979583d74aa2b43f8d5cc58009af : SUCCESS in 46m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3257b0165718446f9ced07d9927c43b9 : FAILURE in 1h 07m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4475cd4a38124aeba82788c849baca21 : SUCCESS in 29m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d6a4e10250b48eb9e761d048ced9f3a : FAILURE in 3m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/79b977c47e824c46a85d7bc2221c76ab : SUCCESS in 16m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/36ff02f5cf534d5db34ab65f15331d76 : SUCCESS in 12m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2da481afec144ccd8227735da3a04930 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c993dfaa937847969fa47a4b74c5d9d5 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/260e4c5e6da144ab8d58e9c8d68459d3 : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cbe96e78945242e6a006e0461fe39d1e : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/39bff0fccef641b5ab9be7bd9c53c0a7 : SUCCESS in 47m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66d48304ddfc4128801af293a722af8d : SUCCESS in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c05fd7edc6ec41ca970c623402003e77 : SUCCESS in 2h 04m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8fc3e31fc9c94000b07259df87c90ad7 : SUCCESS in 2h 19m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c06c37349ccc410284781d61e30730f7 : SUCCESS in 1h 50m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/37bad7c953f64dd5b5631e8707b6f89d : FAILURE in 1h 07m 44s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e55dcd00d7a9441082c9149bd4cc10e1 : SUCCESS in 1h 19m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5fb3d3422771426296ede09f46164e5c : SUCCESS in 1h 30m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54e370b09e4e4460b357a2d935c8ad86 : SUCCESS in 36m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c8028911e2347a3a7f489cf5fe57e01 : FAILURE in 1h 16m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d1b07d438cc444d99e1fde2f1bb4ddea : FAILURE in 58m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bcc8742c3e6a43068792082d7e391cc4 : SUCCESS in 1h 27m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aad72a9efb4d4ded8378c40e18e0e25c : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":6},{"id":"79924032f99dbaa403798f3d161f63230640ac5c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-07 05:45:27.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cd8926595f384a768f101c0327d8f869\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1f4b0c1be2b1493bb9417f25204b8df3 : FAILURE in 1h 26m 19s","accounts_in_message":[],"_revision_number":6},{"id":"18d6bb26e6af97c6082e9ce524a6f787da313319","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-07 06:03:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/899548 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":6},{"id":"e66078ce50e152b022dd93a4e08ff22624c83e1c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-07 07:29:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a9ba883d120d46738eac6ab8cdf32d09 : SUCCESS in 1h 11m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c4c47b1f7df44d3e8df71a88d0213626 : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":6},{"id":"11a76870a13561480cb195c36402dc640c331810","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-07 08:44:59.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/6/check/DellEMC_Unity_FC/c77bbc6/DellEMC_Unity_FC/1259 : SUCCESS in 1h 26m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/6/check/DellEMC_Unity_iSCSI/8b04398/DellEMC_Unity_iSCSI/1291 : SUCCESS in 1h 23m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ff3cb34ef7f55655d4b4fc0f94b4b1813ceec93f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-07 09:36:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/613/.","accounts_in_message":[],"_revision_number":6},{"id":"c722a318c021646625ec0d85c97968c7a6f920a0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-07 11:15:10.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/6/check/DellEMC_PowerFlex_3.6/7af2077/DellEMC_PowerFlex_3.6/1397 : SUCCESS in 1h 38m 47s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/48/899548/6/check/DellEMC_PowerFlex_4.5/86a2c1a/DellEMC_PowerFlex_4.5/363 : SUCCESS in 1h 48m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"4d9a1ac1ae4b8d463746478cbd7b25c3e1a1a6f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-07 11:20:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/623/.","accounts_in_message":[],"_revision_number":6},{"id":"58a04cb97934182d3bfccc013e2214ba45f68ad3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-07 13:43:29.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/6/20240207151006/NEC-V-ISCSI : FAILURE in 7h 28m 02s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/6/20240207151006/NEC-V-FC : SUCCESS in 7h 33m 19s","accounts_in_message":[],"_revision_number":6},{"id":"59921fc0b304924211c7463ecd7a08ef74044cbe","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-07 18:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/6/fc : SUCCESS in 0h 45m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/6/iscsi : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":6},{"id":"c3cce21f73f1b9f33d287c19ec408a13c600e376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-27 00:41:49.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"622fc40f8e76384bf1985c05541d6982e8e71db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 01:01:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/901b1d01d68a4ee48379d56aebb04fbb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cd7081f37c548a6ab71592c65c564e1 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/768c0226f5d64d8680d1ee8b3027a32b : SUCCESS in 15m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42840c377246467a962061d28fc209d7 : SUCCESS in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e2c48017bf6e741e3f7125c716beb426914ba964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 03:11:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b862e6c20154c66a577433d74652cdb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c2bb2b5ff0a49e183b151d1a24c33c8 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f11f1c3b1f648ff97abad7bff3ae718 : SUCCESS in 10m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fffc723f3ac542c88e06dcf48c003e12 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c053fa8dba54d46800f6985d622af4d : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/5a839d53a24f4198a0b2ad91e647279d : SUCCESS in 1h 06m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2fcfcebdc8174a139a34efb371f825d2 : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98f9117252a54af59545edbcc3ab3f31 : SUCCESS in 1h 30m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/197816c74a074b78bf18eca519cdab6a : SUCCESS in 56m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27da1857834948e1acfeb2d181c106ba : SUCCESS in 9m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ce7bb302f904bb093a654d739ee74c0 : SUCCESS in 14m 09s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d82ade15a978471490e91f15a8a942fa : SUCCESS in 14m 47s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78aee0010eec44e68e636b5f2cbbaf8b : SUCCESS in 6m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d504607c26f4cc5bccb52181192303f : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/deafa38dc6aa4e74892795a95691fd23 : SUCCESS in 6m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a430a2fb4eb3417eb6bd06a2270ff5ef : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/47673423231c43529dcbc0b0ab3c9dfd : SUCCESS in 46m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/451e3bfa7d804212af19a6d082cf51f8 : SUCCESS in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8485ddd9f9324810bd6c24f1bbad69b1 : SUCCESS in 2h 21m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9954831891c8485eaf1aa8a4c501df06 : POST_FAILURE in 35m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00aa2dc6ce774abc9fed6dda5b4e49e2 : SUCCESS in 1h 36m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/938fd53aae2b4ad3aae639975cb99a2b : FAILURE in 2h 24m 18s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89c2a8bc080a478ba57067c328485f8c : SUCCESS in 1h 15m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fb663dd592d54a29b9698337f3653682 : SUCCESS in 1h 31m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5132abdb098437a8a9d5650041fe9da : SUCCESS in 54m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99abbbf3d28747bf87e8c2a0b6dbd377 : SUCCESS in 1h 35m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b6f1e31b6685469ba03f619bbc8890db : SUCCESS in 1h 09m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4acc99be23d349fca4f51a7195b799f7 : SUCCESS in 1h 39m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc46b9f373434210954f8b8741124b7e : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":7},{"id":"688ddb191cc71c6491d87c33f5178d67a9df8289","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-27 05:40:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/899548 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":7},{"id":"0d1ff54dda7baf1bcb4e6c0d4e6d35dfef0a7d04","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-27 07:28:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/7/upstream-check/cinder-tempest-plugin-ontap-fcp/b519d36/ : RETRY_LIMIT in 8m 25s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/ddb288b/ : SUCCESS in 2h 06m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/7/upstream-check/cinder-tempest-plugin-ontap-nfs/d316366/ : SUCCESS in 2h 08m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4950899/ : SUCCESS in 2h 08m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"dcd495c565a8620d49334fc808bab4950f6a2b5e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-27 08:31:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/48/899548/7/ : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":7},{"id":"4839726b168f6e3854a2e4eeb8ff17f07b7ac5b2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-27 12:38:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/805/.","accounts_in_message":[],"_revision_number":7},{"id":"a83c1a0c7b2463f53dbf795af514967af7434366","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-27 14:11:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/899548 : FAILURE in 57m 16s","accounts_in_message":[],"_revision_number":7},{"id":"50c57877f36614513217e5cffae75d72817995ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-27 15:04:38.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/7/check/DellEMC_PowerFlex_3.6/2749959/DellEMC_PowerFlex_3.6/1541 : FAILURE in 33s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/48/899548/7/check/DellEMC_PowerFlex_4.5/f3fd1fc/DellEMC_PowerFlex_4.5/505 : SUCCESS in 1h 33m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"4154847c1d93e073e6c80e1e03dda24c702a4fee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-27 15:12:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/899548 : FAILURE in 57m 21s","accounts_in_message":[],"_revision_number":7},{"id":"1f2ca36eca206a3b2844fca5c49eefeb3f587614","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-27 17:15:50.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/7/check/DellEMC_Unity_FC/9d1b77e/DellEMC_Unity_FC/1401 : SUCCESS in 1h 10m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/7/check/DellEMC_Unity_iSCSI/a319cad/DellEMC_Unity_iSCSI/1434 : SUCCESS in 1h 38m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"28b7afa92a4e45451adb19936ac2a6c67bfb80b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-27 17:39:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/17e918adc6f045f3a4c8eaa145540524 : SUCCESS in 1h 08m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8d3b3a8c2a5149be9be07804f8e54561 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":7},{"id":"926800077e4b003573939606c4f291e12d080012","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-27 19:16:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/7/20240227215457/NEC-V-ISCSI : SUCCESS in 3h 13m 08s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/7/20240227215457/NEC-V-FC : SUCCESS in 6h 20m 59s","accounts_in_message":[],"_revision_number":7},{"id":"2a4b01976186c085c29892e9daaaaed688da3611","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-28 00:00:07.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/7/fc : SUCCESS in 0h 47m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/7/iscsi : SUCCESS in 1h 15m 56s","accounts_in_message":[],"_revision_number":7},{"id":"7ddb5d6ccd3e2ed40de72b8ba9e62e9946308f7c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-28 18:52:24.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9879f9e0b1794946961827c6220d42e4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7f6aee54f87d46ff9212c663778e3f62 : FAILURE in 1h 31m 51s","accounts_in_message":[],"_revision_number":7},{"id":"c4fdc545f455b137c62a0d5358426fad5a2a4ed5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-03-02 03:30:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/899548-7/HPMSA-iSCSI/ : SUCCESS in 1:55:29\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/899548-7/HPMSA-FC/ : FAILURE in 5:32:38","accounts_in_message":[],"_revision_number":7},{"id":"64a674f277d345cdd252a968f4b1e554a2b06eb5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-04 18:13:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/7/20240304224436/NEC-ISCSI : SUCCESS in 4h 05m 19s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/7/20240304224436/NEC-FC : SUCCESS in 4h 28m 39s","accounts_in_message":[],"_revision_number":7},{"id":"d6241b8103389f9acdb8a360d8942735c0082a5e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-03-05 00:13:25.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5815d94b9cf1976854c11264787e709781aa1bc3","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-03-05 01:57:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b755294e5136fd4707db2b4e1b6a042d2639282f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-04-16 14:04:59.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"cdfbb7b4f01f562bd934183e60a9aed6a64b6b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-07-23 08:15:57.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"908a389f0c706dc1315c58be06b0dedaa2afde03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 08:16:08.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/119/","accounts_in_message":[],"_revision_number":8},{"id":"41c84256ab70ecb76d572cb9a466287578f1dc47","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 08:16:10.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/119/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"2e5ce14b011a2eb13eae83bcc666c7189700494a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 08:16:28.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 899548,8","accounts_in_message":[],"_revision_number":8},{"id":"70e7047ebb47b9e15d3ba4c07a18afc978ff0275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 08:16:30.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 899548,8","accounts_in_message":[],"_revision_number":8},{"id":"ee65875ddff78f599944a86f0917a6af4f931ea8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 08:16:38.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 899548,8","accounts_in_message":[],"_revision_number":8},{"id":"2bcc2b3cb54f4284c53fb7c24584aa7f9ed4fd3a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-23 08:16:57.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"7f34d4e6ca8723593ac01dfd911a2906f4971167","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 08:17:00.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"e181193bceb52019cb4e07717fe4f803d84e3510","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-23 08:17:00.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"73f6bb85aaeca2ca62fc02fb4a48899d5dc36b24","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-23 08:21:08.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 899548,8","accounts_in_message":[],"_revision_number":8},{"id":"5cbf52d05704de919c2c9ffdbebde7eddd36fc96","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-23 10:08:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/778/.","accounts_in_message":[],"_revision_number":8},{"id":"2cf02c066ef3dbd22271fa2409cab6eb5f2b2ed9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-23 14:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/8/fc : SUCCESS in 0h 45m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/8/iscsi : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":8},{"id":"09dab7a9174f389ca905e0faa4e2227c9af2c7ba","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-23 15:41:25.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/8/20240723203115/NEC-V-FC : SUCCESS in 3h 50m 04s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/8/20240723203115/NEC-V-ISCSI : SUCCESS in 4h 10m 07s","accounts_in_message":[],"_revision_number":8},{"id":"2be1f18133adb78b43b8266668df3f4d4fcbd54a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-24 04:24:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/8/20240724050827/NEC-FC : SUCCESS in 5h 46m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/48/899548/8/20240724050827/NEC-ISCSI : SUCCESS in 8h 16m 06s","accounts_in_message":[],"_revision_number":8},{"id":"d87e68e986bcf6dce9d791d4cea77d5177c70808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-23 06:36:56.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"43a66dfb32256e5b3932c80a0d36832e3c7f70ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-23 06:37:05.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/380/","accounts_in_message":[],"_revision_number":9},{"id":"136f237294922d3d875bc43ee709d5249f146ecb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-23 06:41:37.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"91ef1d957998eec30caabab6d54cb39f1986b6d5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-23 06:57:35.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 15m 00s","accounts_in_message":[],"_revision_number":9},{"id":"380324037dbbd3749d68077bef9e5716675deed6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-23 07:38:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/899548 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":9},{"id":"f738b01dbe69eb07df52d1847055ce56b64649bf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-23 08:05:53.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/976/.","accounts_in_message":[],"_revision_number":9},{"id":"7ab770063f8134d0237233b653811b7183851b23","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-23 08:07:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/391/.","accounts_in_message":[],"_revision_number":9},{"id":"4cdf1b84a60f2b5bebbd9ebc782b61cd14363586","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-23 08:53:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f1f99ed8ac384d64b1fd9f692d99f4bc\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0b2869549fe8474da48452b9b246a8f9 : SUCCESS in 38m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8412237d41074c5e9e0010203becec32 : SUCCESS in 42m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c3a94017ff642e695277d3b5f09d373 : FAILURE in 2h 16m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3cf95541534a4493b561519a239acd3e : SUCCESS in 39m 47s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"ed73b95b349338dd50420678c150ba3a12e157da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 09:19:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7f9f325060742099020fc3807e72978\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf0174bee7442ce97b44136ab807119 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1d6f70e24c243b0a76e9772182ba46c : SUCCESS in 14m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/738eb7f8580c47e4bddbc6de33d1bdff : SUCCESS in 12m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f736dfa6a5fc421fb149ec07df1c9988 : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cb8b9018f2438eb235cbd7bba9c446 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/96681ae8972a44e68f171ba47f6238ad : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bd4b8c51e3a4900805cb2c7d0f27a8d : SUCCESS in 1h 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/119d1738409a4868824a7c6bb10326d7 : SUCCESS in 57m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6323f1a851314ce4baf62d8cf52f4ec1 : SUCCESS in 20m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/877a5b5a1ba4447fb3a230b77bb1635c : SUCCESS in 18m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eff4f330508743799f5ab9843b44ad5f : SUCCESS in 4m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/981027ab10d9439691a12d85099775f0 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be7f52c8a81e4fdab8df3f59745a1771 : SUCCESS in 5m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3a03cb6f674e4cfa9f3942f21ae2dad0 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/20bafea9633a4c03802cc78967ca067a : FAILURE in 31m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60fe5532378a417785d61e946b5a864b : SUCCESS in 4m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4b8b6766aad43999805b299f5bb9f84 : SUCCESS in 1h 36m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/40b15ceb5f00418ca2bc8e66ca1e53ae : FAILURE in 2h 36m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4aaee05abead484ca94da522a5054e76 : SUCCESS in 2h 11m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6319439024fc482c98a2a658bc174b47 : SUCCESS in 1h 51m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/88fec93a2721428b81717b14bb1a05bc : SUCCESS in 28m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/806d8cece8d5475d9b22f63430e3b8f0 : SUCCESS in 1h 05m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f634fec7494488292a95d10036d3edc : SUCCESS in 1h 28m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1cf742343e5045db97cd41afe2905821 : SUCCESS in 34m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ccb6ef989ea4bf48c7eb8b2d848d9c1 : SUCCESS in 1h 30m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c201519caa854b5b9a1febc4e5b2f6bb : SUCCESS in 1h 42m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fc7792130bd440d81260fc151d71878 : FAILURE in 1h 42m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e33e3abf46b4f13b9e2289652c105f4 : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":9},{"id":"8945500d23e32e0d66df54aeb4d93e4daab7aeb3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-23 09:21:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1f74984eb9d74ca49c9a0d8d22406a97 : SUCCESS in 1h 18m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dba904706c1f4bd7baccc34638270ad1 : RETRY_LIMIT in 2h 42m 34s","accounts_in_message":[],"_revision_number":9},{"id":"3298c419da72618ed66292ddafef92e78fd7a42f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 12:48:12.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0d68b73a594041ac514ac69bce6dc9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6cafe09a24d4590b6ef694c19b90c45 : FAILURE in 33m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dbd3299d4908461d9d051fa6fe195d0c : FAILURE in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f45921c5b68252d74fe1e2a2fe74b315bc7e8f96","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-08-23 12:50:08.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a438575f06f7bebc72718a3ee593b8f705e0a6e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-23 17:18:27.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/9/20240823180722/NEC-V-ISCSI : SUCCESS in 3h 42m 31s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/9/20240823180722/NEC-V-FC : FAILURE in 8h 11m 01s","accounts_in_message":[],"_revision_number":9},{"id":"46da871d789fe7150e63fa839f39e73f5f687ae1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-08-23 18:00:08.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/9/fc : SUCCESS in 0h 52m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/9/iscsi : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":9},{"id":"aeb28b8eeb9495f93030abfec233337748f6b824","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-24 02:10:58.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_Unity_FC/7401bab/DellEMC_Unity_FC/2491 : FAILURE in 2h 31m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_Unity_iSCSI/c3a50c5/DellEMC_Unity_iSCSI/2495 : SUCCESS in 1h 52m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"2b72d48f3e322784447c52044ae38a46c639c91b","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-08-27 04:45:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a7f9a2a382186b59557e6b17d3c69ebc14296fc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 07:03:50.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd8e0f9887be48fb876effc386fefd20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b385ec98bb534d70a05933b1288f06a3 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3684b5d2a6d4edeaa17cef2bc45fc01 : SUCCESS in 11m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32ac3ad3176f44d8b54e21cfd407751b : SUCCESS in 11m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de6f64092def4c199741a806af7842c0 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dde4f8caa91418ab847e986679b289a : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/b392b00f62bc449daadb7247b8744f75 : SUCCESS in 1h 12m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b79eb99232d4e7ea7a18fc975aa665f : SUCCESS in 1h 40m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e24a2c041bb44485aaf6fba410b785f1 : SUCCESS in 45m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af4b315dc6cf4477b1449e45c9c98340 : SUCCESS in 13m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b5a97210350143a29aef69d0ab88ea2b : SUCCESS in 18m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b522026a55174b6c8dd80e5a5fb9068c : SUCCESS in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34f547622ade4e93b2043018b0c39769 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97ad3ced45ff48018daee6ae88085abc : SUCCESS in 4m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c07e33a6f4d04ea0aee19dba27957db5 : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/71c48432f0904f3aa83f2a1d6309cf6d : FAILURE in 40m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/436c004595aa444ca9f1184b70d9a346 : SUCCESS in 4m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b06ff5597b94935946659392a8d8a17 : SUCCESS in 1h 33m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/44263675b84349138f443bbb65c7bf9c : FAILURE in 2h 17m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5028a7c01eb94069b68df8b8fead27c6 : SUCCESS in 1h 36m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/916a61dcec1247e6adc58f314e938ea2 : SUCCESS in 1h 39m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/01a1dedd5cf541e689787abe8241ccbc : SUCCESS in 37m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9917cec39bc41b4a7465159add5d5c0 : SUCCESS in 1h 12m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cd971172ff7345b1b7a384adbc20b341 : SUCCESS in 1h 30m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b670c88bc334af18931530f7540efe2 : SUCCESS in 40m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c562542e82664750b67c57bb31e712aa : SUCCESS in 1h 38m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a11725bb8120481a8483c956032b4f7b : SUCCESS in 1h 49m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aea10bd6d34b4796ad7bcd5319b187cc : FAILURE in 1h 43m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b242925958244070a042c0fc07792730 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":9},{"id":"6911130787e461035f0d36bf2dffc0dc3821e949","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-27 07:12:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/baab39b3e4204717aaf5a86110662d34 : FAILURE in 1h 18m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/621abfc189e048d18d4cf95fa936d18d : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":9},{"id":"efc4c9f0caae7409fae2b47e3d4b1bed83d81cf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 08:46:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c35b0c1a3e2487e96354cb2d59af6c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37d081f5de984cd2bc1efa0778eaf23d : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61126cd305e9447ea81317d4c888738a : FAILURE in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7614a3052d854eb4f2284ebf56f6048718d7ded8","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-09-12 02:37:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a9ce9031d9ad8796a7edbb486cec79ee0e4870ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 03:37:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0795f67bb5ba454aada7682dd68cb276\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50b769b28a7a40d284cceac7ec637ac7 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f4238ead4ee144bfbd6283e80fa425ab : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"144c2503baabbd9cb598989416102df6590cfc89","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-12 03:56:17.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f13e5511902c4608bf4e68730e3f05bb : SUCCESS in 1h 16m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/23ffcf3b496a4ff9af91a439a79fe424 : FAILURE in 1h 13m 54s","accounts_in_message":[],"_revision_number":9},{"id":"ae4ad9ea3dc580d0a62bd2bf15f860a2c877495d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 04:53:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d5c34fd0e3402aa7a2e72dccdf1c4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0d6b9b6763409a85080760afdd6a66 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7eb734f1bb241bfb34ff14b7ae4be29 : SUCCESS in 13m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df289dd287f64f9b8e71f9ac70abb52c : SUCCESS in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/031b62778bf342cb954a8e6bce88844a : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40928c6595f842078c27deff32bddb7a : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/f102e4c2ed5d4280ac094f357c2bec62 : SUCCESS in 51m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8d6f1ff071ce49f08dbc2b5052d4140b : SUCCESS in 1h 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6834eaac864d43e8b3bd9b417f930fe7 : SUCCESS in 50m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19586624ef1146e9bca346138ea04f69 : SUCCESS in 19m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cfac84f0a34c4286a3c16c76b14bc36e : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7626d60f3454ad7ae275ae6e7feac6e : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3333919ae960466ca9c2658dc962cf3d : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ae8dd50ae234816b07eb50100aac2f1 : SUCCESS in 11m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5207d53c24904084bff32155d5493c66 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe1d9df8ac70468dbc7ac74ffa0ab6e0 : FAILURE in 36m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecd978424be2481fb1692c8f54d209b3 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d17d20f405844d828a06cb542cb1154d : SUCCESS in 1h 57m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c416f1c6dedb46bfa93ea53baca88295 : FAILURE in 2h 12m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b4fba75c567747ce9eb1d692511fe40b : SUCCESS in 2h 04m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fd6576d4741f483e891ee2a9dd55c53f : SUCCESS in 1h 51m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7927c67e99c94c8ab03ac4a209ea6b81 : SUCCESS in 26m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/94c22b90db514812924cef3cf62e6dd5 : SUCCESS in 54m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/76af8fad2e024bd5838f71ce1142c823 : SUCCESS in 1h 29m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/708daa5b05304ce5b5e4383a4cd641b3 : SUCCESS in 38m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf9a211dab714740b6f9d764e6837cb1 : SUCCESS in 1h 43m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0665c0d5513b4f658b6981ed66d17cea : SUCCESS in 1h 35m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91b6ee7d0aac41feb1dc4f85af383fbe : SUCCESS in 1h 40m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2ef87c512ec4849af8d75c59ac40eb5 : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":9},{"id":"65402687f82d1c4f5d81c4817ee6357ab82a9827","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-12 04:59:12.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_09_2024_04_53_55\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"90e7f171605fcd337fedd9cb1658008370a6f895","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-12 05:47:59.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/9c2cf67/ : SUCCESS in 53m 35s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"eafbc6a2bca53fb90a3d82ecf3b470e34cf91dc8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-12 08:23:51.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerStore_iSCSI/11e4ea9/DellEMC_PowerStore_iSCSI/1731 : SUCCESS in 1h 17m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerStore_FC/13e2f4a/DellEMC_PowerStore_FC/1929 : SUCCESS in 1h 15m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerStore_NFS/1598573/DellEMC_PowerStore_NFS/1424 : SUCCESS in 1h 10m 22s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerStore_NVMeTCP/12c027a/DellEMC_PowerStore_NVMeTCP/1284 : SUCCESS in 2h 06m 49s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerStore_NVMeFC/11fd771/DellEMC_PowerStore_NVMeFC/248 : SUCCESS in 1h 06m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"3660d474dfcf78f7d198496d9790b7ae641a0cf1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-12 13:34:10.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_Unity_FC/f2b2c36/DellEMC_Unity_FC/2592 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_Unity_iSCSI/fbe7d5f/DellEMC_Unity_iSCSI/2597 : SUCCESS in 2h 17m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"a8e1907df2b80c6decf50603704ee5651624e590","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-13 12:45:00.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerFlex_3.6/e15014f/DellEMC_PowerFlex_3.6/2084 : SUCCESS in 2h 09m 00s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerFlex_4.x/a7def5d/DellEMC_PowerFlex_4.x/967 : SUCCESS in 1h 38m 33s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/0811e70/DellEMC_PowerFlex_4.x_NVMeTCP/917 : FAILURE in 5h 54m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"349ece21593bb97c23718ef740979a49b43c7cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-09-30 05:19:32.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"05167896ea979bc7bf69b1a39fe13701d150d2e4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-09-30 05:21:36.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"971cdb5eaa8647e56f3c6604e230a08e8cca58e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":37269,"deleted":true},"date":"2024-09-30 05:23:19.000000000","message":"Patch Set 10:\n\nBuild Started http://10.225.179.242:8080/job/openstack-ci/35/","accounts_in_message":[],"_revision_number":10},{"id":"db9eafd45f3b000a59f2e2f5a1bc75aa6e396f85","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-30 05:29:54.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/649/","accounts_in_message":[],"_revision_number":10},{"id":"60968f3fdf85a5a0f06e738dade9a8cfb5a88472","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-09-30 05:35:20.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"abb0c2729cfc9fc20ee8706470b1d50b4ef351df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 05:45:37.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf0113bc34854cf69ccb81b67a052aba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd4be7b3c0224c8b83362f14abd215d9 : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c2986f090e724f5cbb766a26c2963a25 : FAILURE in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"38ae757a78e99dfa276df8ecfb42b93d31333229","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-09-30 07:24:50.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/899548 : FAILURE in 22m 40s","accounts_in_message":[],"_revision_number":10},{"id":"32da1929a19c1992ec6d55dbc653c393a02b3c8b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-09-30 07:32:09.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/899548 : FAILURE in 42m 39s","accounts_in_message":[],"_revision_number":10},{"id":"8379c8a8f22375fa152dd6926aeafe459ae25b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 07:42:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eccf0372153432e84b67cc59d1ac148\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7c191d14fd49758deb9fa7979b6a99 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3674d3708ba04bea971dd1ed20a7230e : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e5fda36462f464da49c58bd794e09c5 : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56c01cf92cf48748ccfc0edd633b67c : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/72e400068dd34b51ae81400caeb5ab38 : SUCCESS in 1h 07m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08a586e379e4447f8da12d773b356f30 : SUCCESS in 37m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c73a16f034b5462ead448b351eeaba83 : SUCCESS in 1h 06m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fab8ad099a3c4f0fabf62afac67b96ed : SUCCESS in 34m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/523617e8aa954e9bb98f99c79a6fc966 : SUCCESS in 14m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5827f09525a426aa7045cdf8980835b : SUCCESS in 14m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f71b9ad6a7ec4810b6fd23dc271f5012 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77b66048b7f64080935ed51fc35b6755 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4187876c4ea4afab8252bd59e79e612 : SUCCESS in 9m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c5468641fc34483aa6e4dc4e658564ce : SUCCESS in 5m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9fa0853e29294caca78da237415d9fa9 : FAILURE in 26m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb62a46be71841b89638849c90264d9f : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8545a1b0e3764406bce4ea6b093793d8 : SUCCESS in 1h 21m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e6363799ebb943b5a524879ccdc48781 : FAILURE in 2h 20m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3d8ccf5ae2894228b8d1226884b84ed5 : SUCCESS in 1h 34m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f35d1ccd93cc44a2848abaf8ddfc4d91 : SUCCESS in 1h 47m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d38f7592208d4404b6dfdadb6718b19f : SUCCESS in 21m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dabc38f919254d5e9bd226c3220e3bac : SUCCESS in 58m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e88668132ed84192aeb780c06b622e93 : SUCCESS in 1h 33m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c9c559bb6644ec2871a8a945e7e6db2 : SUCCESS in 44m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0b3247c5356c4db09f521ad302054369 : SUCCESS in 1h 02m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/910d1f771f78449b82db1195f1a69fe1 : SUCCESS in 1h 48m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d7d8d3dfc40445b82fe7414ed2c64c0 : SUCCESS in 1h 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/350433d52ee04c7fbd904d3468e6b08f : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":10},{"id":"055e1cb4e9ee3a53fe4ab29c28d8a1c926e08a73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-09-30 07:48:32.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_09_2024_07_42_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"5568058e3e2e0bace37ee5461531709f333b1531","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-30 07:57:12.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dee618611ac8440d895efa0a4d4f7ef2\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5962b523d97248f382ff24192ecb4c49 : SUCCESS in 48m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/12868ba148ff4632ae67312318c419af : FAILURE in 39m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3d48079ea45e44b0b96a9f0b1ffc82d8 : FAILURE in 1h 49m 53s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94a840e091b54526ab2de232dbd43fe1 : SUCCESS in 36m 00s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a6621b44b4e4b68bc1a2572966533f0 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a7805b28f9949b987d82a10da56f9c4 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22bfca1f0ff94cdea5d845e8985c405b : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/32b6be5ee2a14475b020f86a41e48b88 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"0d871f81773d6f00a59b52c9ba5c7b35c900747e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-30 07:58:28.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/4de5438/ : RETRY_LIMIT in 15m 09s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/48/899548/10/upstream-check/cinder-tempest-plugin-ontap-nfs/3ba93ac/ : RETRY_LIMIT in 9m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"4ef210428f28cbe0444e876370af21146446a465","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-09-30 08:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/10/fc : SUCCESS in 0h 42m 23s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/10/iscsi : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":10},{"id":"abf83e53f25f53e47170f7cd18105b9b67f26b8d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-09-30 09:37:29.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0eed54d09c544af7ad96708d9568b553 : FAILURE in 1h 17m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/33668c45b6164e96a582bd2f5127f8fa : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":10},{"id":"2b452d9ac9aaab6c29de36e3bbabbcb74a9ea805","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-30 13:51:38.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerStore_iSCSI/779cc7f/DellEMC_PowerStore_iSCSI/1776 : SUCCESS in 1h 24m 49s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerStore_FC/a76964f/DellEMC_PowerStore_FC/1974 : SUCCESS in 1h 12m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerStore_NFS/a2c9046/DellEMC_PowerStore_NFS/1469 : SUCCESS in 1h 18m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerStore_NVMeTCP/b5589cd/DellEMC_PowerStore_NVMeTCP/1328 : SUCCESS in 2h 14m 32s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerStore_NVMeFC/fe768d1/DellEMC_PowerStore_NVMeFC/293 : SUCCESS in 1h 17m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"afcadff2c6ae4ab4a9dcba5fa64e87c57122a3d1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-09-30 17:42:08.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/10/20240930181847/NEC-V-FC : SUCCESS in 3h 44m 39s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/10/20240930181847/NEC-V-ISCSI : FAILURE in 8h 23m 17s","accounts_in_message":[],"_revision_number":10},{"id":"b34505933d092d15efe5a0825db32472999fb1f8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-09-30 19:24:48.000000000","message":"Patch Set 9:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/899548-9/HPMSA-iSCSI/ : FAILURE in 3:27:03\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/899548-9/HPMSA-FC/ : FAILURE in 43:02.98","accounts_in_message":[],"_revision_number":9},{"id":"5379e9aea243a14f9e8a2b8d167c276fa7be5138","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-30 23:01:01.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerFlex_3.6/f15c9f2/DellEMC_PowerFlex_3.6/2162 : SUCCESS in 2h 17m 38s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerFlex_4.x/035d407/DellEMC_PowerFlex_4.x/1046 : SUCCESS in 1h 51m 50s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/b608894/DellEMC_PowerFlex_4.x_NVMeTCP/985 : FAILURE in 5h 53m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"5864711ed3e1b86844cb03128511c2a8b62f271b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-30 23:19:25.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_Unity_FC/cf1f817/DellEMC_Unity_FC/2669 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/10/check/DellEMC_Unity_iSCSI/29698f3/DellEMC_Unity_iSCSI/2677 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"cdbb4e44330815ca62ffae7d6a186a33fec72c14","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-01 22:40:21.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/10/20241001222438/NEC-ISCSI : FAILURE in 8h 55m 37s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/10/20241001222438/NEC-FC : FAILURE in 9h 15m 40s","accounts_in_message":[],"_revision_number":10},{"id":"0eeb538d215267b68d258336048d24043e7e13de","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-10-04 16:05:15.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"35b995c70dfb828a4d426fe25f5042e4c6c4b0ff","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-10-05 20:21:28.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/899548-9/Seagate-iSCSI/ : SUCCESS in 2:22:59\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/899548-9/Seagate-FC/ : FAILURE in 6:19:16","accounts_in_message":[],"_revision_number":9},{"id":"12e810150e6347b996bd59aea0f3559f1e093460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-11-13 13:34:05.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"f0cfcba9bdcd2a0c4f685b22a0ca35c32634aa57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-13 13:34:33.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/912/","accounts_in_message":[],"_revision_number":11},{"id":"632f317eff3f817222d17264cdb72600a22b9d90","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-13 13:35:25.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"62fc869a4862320c6e7d7929cf2381047a52e526","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 13:59:30.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ca45db4f38c4cf78f1d002aeffa70be\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8cf46db49574fcebde93782770ea79b : FAILURE in 20m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9084949e25d74203adc3d519f277462c : SUCCESS in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"58c1df800194619b4baa2364f03906eb8e4af6f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-13 15:11:59.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b1484b0be0c44a6c8e21a297eb400611\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ce387f2a68d5420ca4a59a627cb55c9f : SUCCESS in 43m 02s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6b235f3921c4860ad31ece8c69428ef : FAILURE in 48m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de2a94982ebe4c59b7d5fd5b5c4aef13 : FAILURE in 1h 34m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8ed95c10044749f08e2e35c4ea8a7109 : SUCCESS in 40m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e4eb99d3fd745ee8e0ff7225972d45b : RETRY_LIMIT in 45s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/108f919c2ae945058fc0d3e890151b26 : NODE_FAILURE Node request 200-0000014782 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d74ca1b1dc7340819e9b4eceafe72699 : RETRY_LIMIT in 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fdeb6aa8ace462bbacbca1c014b9683 : RETRY_LIMIT in 57s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not ma... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"c7b7ecbd2de5fc4e59bc0ce3f630da141119a283","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-13 15:13:53.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/148d467082fd4d58aedf089cedc0d92f : FAILURE in 1h 29m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1b57b12fc0ba4a7494bfdd840f6c040e : RETRY_LIMIT in 7m 02s","accounts_in_message":[],"_revision_number":11},{"id":"84422bfced8afc87cfb22ff5a057ba6beb2b3b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:23:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a393c3f6867e4afba7e3ed1fdab5cf33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba22e81375a4fd79c79c3845948deb0 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be05701925c54b378e88c6a9daad87d9 : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/655e10cfcc2c48a88305f234698ebdbe : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc7b802bd554f098bebbde506007abb : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a8b6b87a126548d58a24b3a8dbcace56 : SUCCESS in 59m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/745c6215ba6f4e5e905bb3368ded838f : SUCCESS in 35m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b32dd017364e47d89a138d4c423fd878 : SUCCESS in 1h 32m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78c6831c62224a83865dfface1cf446b : SUCCESS in 1h 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97687fed225d483aa689bd79ee79b45d : SUCCESS in 14m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2cb1fc7aa05c4cafac8693664cacbb5a : SUCCESS in 12m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9038a2c42c8b490393782e2942516119 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c23c6b8c6b44c42b45254ea97e23612 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a7502eeb5614b49bb95d95d0df16202 : SUCCESS in 6m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a3e3600566f4137802ab12872ae9a5a : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/979b7f8ba0ff44cc86ace6a9a1daaa8a : FAILURE in 20m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e8bcdc104ed49bcaf5cc4bd3246060c : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a36559eb9e264149b695737a77c095e4 : SUCCESS in 1h 49m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/345d746c1d3e4d3b9edebfd09c01f266 : SUCCESS in 2h 44m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3cca2f363eac480fba4b3b72f5f5bbde : SUCCESS in 1h 40m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8d7f659bdd3b4d3faf455fa11fe31c04 : SUCCESS in 2h 01m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e719a7522394f44bc1da99cce856dc2 : SUCCESS in 28m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2fcc52e256a544d0b12f2a33c2e69aba : SUCCESS in 47m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5bab2f389f57488fbfa50404b642a624 : SUCCESS in 1h 30m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75bb920c84754bcdbf69edb6b1bc6e60 : SUCCESS in 50m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6fb72df37ec94d828031b742fa05ee08 : SUCCESS in 1h 02m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/804b650d86a74ea6b76f483a362fbbd2 : SUCCESS in 1h 14m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b166231070247f6ad57ae58a5c474db : SUCCESS in 1h 40m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/252d1a509e814061b95ce0f4d29d240f : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":11},{"id":"3c598093cf8ec01501231935cdebc2e526093f27","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-13 16:40:25.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/58103068019b412d99e242fe92321f56 : RETRY_LIMIT in 5m 05s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"d3634af8ea7c7607dbb116f99ff0db09e7c23784","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-14 03:11:07.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/899548-11/Seagate-iSCSI/ : SUCCESS in 2:30:55\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/899548-11/Seagate-FC/ : SUCCESS in 2:18:55","accounts_in_message":[],"_revision_number":11},{"id":"ec1a5e1b8354ea993660dc0a088ff118e3df6bae","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-14 03:42:46.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/11/check/DellEMC_PowerFlex_3.6/de62616/DellEMC_PowerFlex_3.6/2423 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/48/899548/11/check/DellEMC_PowerFlex_4.x/b4b7494/DellEMC_PowerFlex_4.x/1271 : SUCCESS in 1h 55m 25s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/11/check/DellEMC_PowerFlex_4.x_NVMeTCP/853430f/DellEMC_PowerFlex_4.x_NVMeTCP/1165 : FAILURE in 5h 54m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"a178fdbdeb95fa8c3282ea7979185e52d41e4912","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-14 03:43:21.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/11/20241114041339/NEC-V-FC : FAILURE in 8h 22m 17s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/11/20241114041339/NEC-V-ISCSI : FAILURE in 8h 29m 37s","accounts_in_message":[],"_revision_number":11},{"id":"99cb12c8ada60124e6f8d67ae003aadc9568ce06","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-14 04:00:09.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/11/fc : SUCCESS in 0h 55m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/11/iscsi : SUCCESS in 0h 54m 00s","accounts_in_message":[],"_revision_number":11},{"id":"d55a09a9d3cf4187213f689de767ecd6157d554c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-14 06:59:25.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/11/check/DellEMC_Unity_FC/1ac2ec7/DellEMC_Unity_FC/2876 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/11/check/DellEMC_Unity_iSCSI/6b565b6/DellEMC_Unity_iSCSI/2844 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"78723335e4afed583578b2549a7a366bb44eadea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-14 18:54:18.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/856/.","accounts_in_message":[],"_revision_number":11},{"id":"0712e8fd5062d2af36ae3a6ce380888fc53b185c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-14 23:59:24.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1422/.","accounts_in_message":[],"_revision_number":11},{"id":"b024019aba8f2cf3959f2e5b15a5e534556d0726","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-15 01:32:56.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/11/20241115023941/NEC-FC : SUCCESS in 5h 47m 30s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/11/20241115023941/NEC-ISCSI : SUCCESS in 7h 53m 11s","accounts_in_message":[],"_revision_number":11},{"id":"3a8a8dc7a262fd5dce764b86dceddcfe937ed14e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-18 05:43:03.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"477895eef493b3b8bf15e8907e0f64fe0ac236ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"date":"2024-11-19 04:55:30.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"d01a0080815b81f5cefe6ed08c75a6a7db8b694e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 04:55:42.000000000","message":"Patch Set 12:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/941/","accounts_in_message":[],"_revision_number":12},{"id":"0651ce8eaec239110b77d51c6ceb7c49f8d72bfe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-19 04:56:36.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/12/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"96a9ccf7074b0ff2480b0fba053114f81ef571c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 05:28:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fe557641c34b45ad2d0f2d199d0805\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfc18737d15041fb9ca1ddae34905362 : FAILURE in 22m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cebb939debcd4330a550e967e35b8a32 : FAILURE in 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ba68e6bdd3514fff1a9b9bff104fa2bdeaa2bbbf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-19 05:34:02.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/899548 : FAILURE in 36m 39s","accounts_in_message":[],"_revision_number":12},{"id":"824e3a1d71d81f1a4570fca0bf81a996255fe389","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-19 05:51:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/882/.","accounts_in_message":[],"_revision_number":12},{"id":"8b51b6d46f5101583b31f9a889108e5725b6b645","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-19 06:14:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ef91289ba9d049a08807d200814eec51 : FAILURE in 1h 16m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2de7d02665f6424ea6eb37aaa96984be : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":12},{"id":"3d3c79b33a1c131220590853969ddbe6da1ce24f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-11-19 06:44:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1448/.","accounts_in_message":[],"_revision_number":12},{"id":"bdca84762a981300dea3c08dc3fcd0859b7a0891","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-19 06:48:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/40efaff3953d4094bd2ab75aeff86fbb\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5aeb40f5f92640ffa3e297a56d598b6b : SUCCESS in 51m 45s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a86db37e72e466d841b88d76a50721e : SUCCESS in 42m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f491fd8b18694e24a5245a0444d7bd4f : FAILURE in 1h 51m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a37068df9b642a18958dc34dacc68ff : SUCCESS in 42m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da4f68c7a2314a24b407cd8f8ccc6357 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bcf4ccdd9ccd4a62a589f9a302ee0246 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/050bb85e06ae46f59266f84f30704399 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9dfbce5b272141619a4f35c3bcccc5cb : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":12},{"id":"8d46f18713afdfc65df051763205626866bd95ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 06:58:08.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7c8cc0d5a49403eb98d7a54db4b4f78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84c7432602748c99cd4635e20be9845 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a2e2074bb34afdbbc8959a57c3ce30 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08ffcec916244b2bbb3dfd46756e285f : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54077e3582c34f3ab08e88de7bc28c4c : SUCCESS in 17m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/7b12c37f8a9d43cd89464773be4ac9cd : SUCCESS in 1h 04m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28e1d2d367b1465fa9af7155a32db3c2 : SUCCESS in 56m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a324ea11eccf48cda883e4105f97b0ff : SUCCESS in 1h 27m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c396dc21f3145e5b58b7f2966083dd2 : SUCCESS in 1h 04m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/002064209077429aaf8b29f3f2656167 : SUCCESS in 15m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab7809b6a9984cbe94344795e99ca4e4 : SUCCESS in 13m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f4d87de378e4b998be418a3c5af8c00 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/964964cdcbc248feb5f64809fe3d9e96 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81b5fe021f394a888c9181d05f7be51d : SUCCESS in 7m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9af1363c41954edc817c095b0a760d9e : SUCCESS in 8m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ce4fa16d3d04fbc805fb4f68de920ea : FAILURE in 18m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e347026285044e12b8f5db63f0e3c28b : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5304e26fbb3846d3ab4ad42ba2ab9565 : SUCCESS in 1h 39m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b7d932d82fa64a98872e55bcdd2958d7 : FAILURE in 2h 01m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0f890e2ee92341aa9c9e844f5f091d4f : SUCCESS in 1h 41m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e1ef23be1e6244e9bf9d7d2374d721c0 : SUCCESS in 1h 33m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/16b307d05c9a483781786609f5ec7849 : SUCCESS in 20m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d02725ae0b394278afb8d01e9e482e29 : SUCCESS in 1h 00m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb9fd6b616cc4af982e545bf2c0960a3 : FAILURE in 1h 27m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a80f311c728f405284d283320ec0a7e8 : SUCCESS in 44m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9ef27742589d42afa9561f4e50d04131 : SUCCESS in 1h 30m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a0c6341a10284c44ba421250379023a7 : SUCCESS in 1h 18m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/beafbe0d6116421fb732558feee1f17e : SUCCESS in 1h 52m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d85249687fc440aca03504182caaf140 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":12},{"id":"733764c19b1c25cac044f8abd6dddd8d34d60aec","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-19 07:15:08.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/364754f44a9948d4a517c75513677b86 : RETRY_LIMIT in 5m 09s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"e2593dada548a14ace73f2bd6a847af025192001","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-19 09:35:30.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/48/899548/12/check/DellEMC_PowerFlex_3.6/6ec993c/DellEMC_PowerFlex_3.6/2448 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/48/899548/12/check/DellEMC_PowerFlex_4.x/55e6892/DellEMC_PowerFlex_4.x/1296 : FAILURE in 1h 59m 41s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/48/899548/12/check/DellEMC_PowerFlex_4.x_NVMeTCP/3900353/DellEMC_PowerFlex_4.x_NVMeTCP/1188 : FAILURE in 4h 39m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"606f30a7e588251d1ac7106186dc0909ce8e079b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-19 10:00:09.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/12/fc : SUCCESS in 0h 50m 35s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/12/iscsi : SUCCESS in 0h 57m 57s","accounts_in_message":[],"_revision_number":12},{"id":"db0e39f770e1fc1166e83fe5cb2ec74f49aba04b","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-19 16:29:24.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/899548-12/Seagate-iSCSI/ : SUCCESS in 2:24:33\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/899548-12/Seagate-FC/ : SUCCESS in 2:17:34","accounts_in_message":[],"_revision_number":12},{"id":"b0f18d6051708786c6fb815df80113aa6434b479","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-11-19 19:43:43.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/48/899548/12/check/DellEMC_Unity_FC/b0b62c4/DellEMC_Unity_FC/2900 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/48/899548/12/check/DellEMC_Unity_iSCSI/ea03596/DellEMC_Unity_iSCSI/2866 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"10342ec74518d2ca8e50cf97cf980e593be72d46","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-19 19:44:12.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/12/20241119194821/NEC-V-ISCSI : SUCCESS in 4h 10m 24s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/12/20241119194821/NEC-V-FC : FAILURE in 8h 55m 46s","accounts_in_message":[],"_revision_number":12},{"id":"0b498cfa77f939f45812252432b9668575b4cee0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-20 00:54:46.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/12/20241120040559/NEC-ISCSI : SUCCESS in 5h 20m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/12/20241120040559/NEC-FC : SUCCESS in 5h 48m 44s","accounts_in_message":[],"_revision_number":12},{"id":"cafe6efac985bd00dcc8e23106c3098072a22740","author":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"date":"2024-11-20 02:59:31.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b9c6dde9cc5144d1a63a60b184517b12da1d6919","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-20 06:23:27.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ffdc8901baf92ece6950f9d2e194299a66aec64e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-11-20 06:24:19.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e1491ce9af01fde7c4358533e0d3afd5c02b2627","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 15:46:21.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_15_40_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"37d2d0e359335d951fd79f3d6d146db1aa801c2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 19:28:26.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_19_21_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"e703829983cb37083e7e1598f630758075eebf5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"real_author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2025-03-13 02:07:05.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_37298\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"}],"_revision_number":13},{"id":"bebd23f2183344b2e0e27849a9b918c4a415a39e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-13 02:07:13.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1473/","accounts_in_message":[],"_revision_number":13},{"id":"ca27c9dbab57b2f9db8481126ff7f30f041a2773","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-13 02:07:15.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/899548/13/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"3f493f626410a6ccfdebe33c41c54f8b909348ca","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-13 02:20:20.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/402/.","accounts_in_message":[],"_revision_number":13},{"id":"ca77a4ef15a5d5aa04c49cb8acd49408d627849c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 02:34:46.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39564e89b33545d9b13a0a173fb308b3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93e01d46d3af4f3d8930f1763d6ae6ba : SUCCESS in 21m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/08b98cf1bcee4c07bc14e2582bdd1522 : SUCCESS in 21m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5dc706ff2e603f571ed5c104f0729461f76a0de5","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-13 02:47:57.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/899548 : FAILURE in 39m 26s","accounts_in_message":[],"_revision_number":13},{"id":"9439d0d2c1d73fb4b47cae00fbf3a6e079f67364","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-13 03:36:27.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/edef9798726c46538b01917055b55bc9 : RETRY_LIMIT in 6m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8882beeffe6a4f10a4d77f5aa172bca6 : RETRY_LIMIT in 6m 24s","accounts_in_message":[],"_revision_number":13},{"id":"b2d9bdb1b628716d907ad54773fe7ef43c84f84c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-13 03:59:37.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f3b37e47327943ab97754f87cdb17253\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c6ec4df9a5f403996c724a1a244bd8e : SUCCESS in 41m 57s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6dce92f2203a4dda9a8d5b8191efef5b : NODE_FAILURE Node request 200-0000029916 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f376f28ab4a4ab19dbf4d4d6bc93741 : FAILURE in 1h 47m 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e09a91043388406cb73a28de25ffe935 : SUCCESS in 41m 06s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/321ab5d6f1f4480dbe48ae8b79cd70e3 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36da158175734573ba34bb9309020f5b : NODE_FAILURE Node request 200-0000029920 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0b6674a17c245fa9f0d77284f6486a9 : RETRY_LIMIT in 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/808e6768b6ea4d139a1fd45307f74c8f : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n           ... (truncated)","accounts_in_message":[],"_revision_number":13},{"id":"281e70399b8be06d959d2e689a660b35d87dfd77","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-13 04:14:06.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/899548 : FAILURE in 1h 05m 41s","accounts_in_message":[],"_revision_number":13},{"id":"725722894a2057b4205087cddd57d91b15fa0e40","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-13 05:13:40.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"e0b8fb584e7139f87db1de5e774c52ec36632454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 05:23:14.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81746e213d844a549d6702a3ff685261\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f13041a19ff43f1bbfb7437a953ddd6 : SUCCESS in 18m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41d23ab028384ae5bef121f875321f5e : SUCCESS in 21m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b963691ce294c329b70020c1441bb14 : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c8f595c1ca4817806a1e72c8716c1a : SUCCESS in 28m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/a9edf9edd6954a2ca2ff4cca8fe6049d : SUCCESS in 1h 28m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3b961953aa84c929204ee595a29f21b : SUCCESS in 1h 24m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc80ba9b9b804b119c09308941fb1fb3 : SUCCESS in 1h 34m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa70f8ac2f6f465d8db513363d96bfc9 : SUCCESS in 1h 01m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df7f8fb3d9de4587bb246fca406ef21b : SUCCESS in 16m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/71a5ae681fe4403dac59dfd4cd2495e6 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d00ca6d3acc4afe8b764b3e120dfdee : SUCCESS in 9m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/55928771a39846d3b0a743afb60bdea6 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f3cbdeab5de41999c60615f04186870 : SUCCESS in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d39dbde9c064d5a82f259357764a816 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6775d872bae44364b10cdfd6fa077da5 : FAILURE in 23m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e3710bed0f31419fb6a9208bfbc78550 : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/303e0d30dcae4fbe87df98812fde43ad : FAILURE in 2h 02m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/882fe2431dca4d65b0af8d63fef2b32a : SUCCESS in 2h 39m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/338144168dd642749798bdb856036e4b : SUCCESS in 1h 51m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/717f51f99d6e4f0ca71e4bcf5a0c4dc1 : SUCCESS in 1h 57m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0bb468c8b9884420832ef85eaf28220d : SUCCESS in 33m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/65aa71eb3a004a0d97f787569f839cde : SUCCESS in 1h 07m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6fec8efd03074d6fba0e8346f71405a3 : SUCCESS in 1h 21m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/16b4930cdd564d519a9981aae41b2fa5 : SUCCESS in 25m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc08c68919ab4e04b65d9b03c28109ca : POST_FAILURE in 1h 27m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5da7327fcccb424db756e1fbea5c54ab : SUCCESS in 1h 40m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08bfc50e21df4129b0d9350f70d471d2 : SUCCESS in 1h 46m 44s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74273947284240cc91c9099f6f05b340 : SUCCESS in 1h 59m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ad3ca86c64e4e05afd00e215b76c7f9 : SUCCESS in 29m 33s","accounts_in_message":[],"_revision_number":13},{"id":"8002fe5de46f7b73a707f9b29261679e54e53934","author":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"date":"2025-03-13 09:26:29.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1430e43fc129aeafe5a86aabb287bf3918c1e99c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 09:47:20.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ebea2dc5cf641e0b67cf641d6f2f7c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57fde40b4af04ee880ee554e715f01a9 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c22b4439e68941a086b8683d831654a4 : SUCCESS in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"feb8329a3a8763d35e2ac5c8b1b5bf15190bd9d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-13 09:56:01.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee9f7d9ddd4642519f64ae9a99732457 : RETRY_LIMIT in 6m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/587dd06f27174bf08add1477d7f4e608 : RETRY_LIMIT in 6m 09s","accounts_in_message":[],"_revision_number":13},{"id":"3afdf4cb4e52746c58b72bbe63658682ae07bc7b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-13 10:14:23.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/486/.","accounts_in_message":[],"_revision_number":13},{"id":"94d9f9c5dbdc7a8558cbcffc2cf7c1345372d877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 12:12:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd3ae6d0c5b41688488091d2d4affa7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c6ef9ef96d4a4f872dfe4b693d5843 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6576407866454bdeb2e85015de79723b : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5bc029436124becbb525bf73d93829f : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330cb210bdc84f41b718a0ac7e760a9d : SUCCESS in 5m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/c48dd365886d4a4797cd4d794ece2272 : SUCCESS in 1h 14m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d448ed6ff6854290845c854e3980c833 : SUCCESS in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b032798507347bda920547f3f047796 : SUCCESS in 1h 29m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d10ebb975314e9cb0ec378148cdaae3 : SUCCESS in 1h 05m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae87e2141774fec9a0221cdc4fb9626 : SUCCESS in 17m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d5eb13dc5e644febf6ae29372826f25 : SUCCESS in 17m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5ae47d6b3ad44209b2c1235de6d330e : SUCCESS in 2m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dd61a0da656146229a1b64e827fc3309 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba9841a01b1c4b88b2b343be58ed8554 : SUCCESS in 7m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fa21a30a81c44e67a96e1e2e2fe4d309 : SUCCESS in 5m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5fc989341dc7464886832a870684b949 : FAILURE in 43m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8c14c2c9d644617b957b6f499febf1f : SUCCESS in 4m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c1ce784ce884310a9e016d9c09d9a25 : SUCCESS in 2h 02m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5c40106fe3ca41ea9cd1d58589620630 : FAILURE in 1h 38m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b45e6de3ed334e549db973080c317bd6 : SUCCESS in 2h 39m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cb03dcde97c546b687b20f90fd037bac : SUCCESS in 2h 16m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d8204a7c562d49a480e474be9dc13c35 : SUCCESS in 42m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb6ce79b17ae48ac8a938f36ef4b8f25 : SUCCESS in 1h 01m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77af71fa0c0d4d5d98928f020f01867d : SUCCESS in 1h 39m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b3f904283be4d04b2ca71a9a4c1f84b : SUCCESS in 39m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0364cca9b2c4fcbac33bb7907268e50 : SUCCESS in 1h 23m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d8dc2b82da4d46dc97c5d0e616594b80 : FAILURE in 1h 03m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a119b7d8855445b6a45c2046501603d5 : SUCCESS in 49m 47s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcd03894f71640a28da321a0d199e5b2 : SUCCESS in 1h 45m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83488f1d81404ec0a93adf89f44751ce : SUCCESS in 47m 04s","accounts_in_message":[],"_revision_number":13},{"id":"c6a8be50fd1b2974018343837436962ed85d967e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-13 13:47:53.000000000","message":"Patch Set 13:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/13/20250313141514/NEC-V-ISCSI : FAILURE in 7h 59m 51s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/13/20250313141514/NEC-V-FC : FAILURE in 8h 32m 37s","accounts_in_message":[],"_revision_number":13},{"id":"006a2d9be609832623cf33c449f0006056e50e66","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-13 14:13:23.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"d158cb63f198f18ea5b241542713bb8246209fae","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2025-03-13 19:00:07.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/13/fc : SUCCESS in 0h 59m 25s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/48/899548/13/iscsi : SUCCESS in 1h 26m 32s","accounts_in_message":[],"_revision_number":13},{"id":"feb53fcddcc99df23f3465abae116ff3a0ea69b5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-16 21:34:07.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/13/20250316220649/NEC-FC : SUCCESS in 5h 56m 52s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/48/899548/13/20250316220649/NEC-ISCSI : SUCCESS in 8h 27m 14s","accounts_in_message":[],"_revision_number":13},{"id":"ebca724856f33cb081b30e71876721a804ec0588","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-07 14:16:01.000000000","message":"Patch Set 13: Review-Priority+1","accounts_in_message":[],"_revision_number":13},{"id":"ce2cdb5fa7425259f24fd97139d297d5f6973204","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-07 14:46:16.000000000","message":"Patch Set 13: -Review-Priority","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"4e990396750426ca473402fb446d699719a321a4","revisions":{"0b96786b310a0264950474dcfc1fa4d47be5ccf0":{"kind":"REWORK","_number":1,"created":"2023-10-30 05:08:26.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/1"}}},"commit":{"parents":[{"commit":"0ebd1bcd0f7dffc1f8f1397d6c4bdeac13de0a98","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ebd1bcd0f7dffc1f8f1397d6c4bdeac13de0a98"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-10-30 05:03:14.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-10-30 05:03:14.000000000","tz":540},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b96786b310a0264950474dcfc1fa4d47be5ccf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b96786b310a0264950474dcfc1fa4d47be5ccf0"}]},"branch":"refs/heads/master"},"dc4fbaeae31915bb025fc526881100460bf4fdc5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-29 06:18:18.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/2"}}},"commit":{"parents":[{"commit":"8834c376d36dc413df06333d2e7b0ddb2549f4a9","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8834c376d36dc413df06333d2e7b0ddb2549f4a9"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-10-30 05:03:14.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-11-29 06:18:18.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc4fbaeae31915bb025fc526881100460bf4fdc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc4fbaeae31915bb025fc526881100460bf4fdc5"}]},"branch":"refs/heads/master","description":"Rebase"},"74cfc1c8eeb2e3b870d5a67d949d5ec0e346c9b5":{"kind":"REWORK","_number":3,"created":"2023-12-22 00:44:07.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/3"}}},"commit":{"parents":[{"commit":"8834c376d36dc413df06333d2e7b0ddb2549f4a9","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8834c376d36dc413df06333d2e7b0ddb2549f4a9"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74cfc1c8eeb2e3b870d5a67d949d5ec0e346c9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74cfc1c8eeb2e3b870d5a67d949d5ec0e346c9b5"}]},"branch":"refs/heads/master"},"a263290ce204ddd5114c2bcc0abbfaa4f7becc6f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-18 05:17:23.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/4"}}},"commit":{"parents":[{"commit":"cd68d5f47d93f81ed68b45ab4eafcff8ce4ff77f","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd68d5f47d93f81ed68b45ab4eafcff8ce4ff77f"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-01-18 05:17:23.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a263290ce204ddd5114c2bcc0abbfaa4f7becc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a263290ce204ddd5114c2bcc0abbfaa4f7becc6f"}]},"branch":"refs/heads/master","description":"Rebase"},"c1ecd1d5d938f481620e27081a1fd7c1b86c9da2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-02 05:21:20.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/5"}}},"commit":{"parents":[{"commit":"59d1442a1245cdfe6cfcb25c052c3fa2c55a0148","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59d1442a1245cdfe6cfcb25c052c3fa2c55a0148"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-02 05:21:20.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1ecd1d5d938f481620e27081a1fd7c1b86c9da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1ecd1d5d938f481620e27081a1fd7c1b86c9da2"}]},"branch":"refs/heads/master","description":"Rebase"},"d4e3a1c1c4252ed4f8bf8d0d106f17677b39cf83":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-02-07 02:16:26.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/6"}}},"commit":{"parents":[{"commit":"3352bd66c60014c5bcfe083c7ffbc3b11555e21b","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3352bd66c60014c5bcfe083c7ffbc3b11555e21b"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-07 02:16:26.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4e3a1c1c4252ed4f8bf8d0d106f17677b39cf83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4e3a1c1c4252ed4f8bf8d0d106f17677b39cf83"}]},"branch":"refs/heads/master","description":"Rebase"},"ca774addb5e1d0be7347a2c7a2c5eb1f431705a2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-02-27 00:41:49.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/7"}}},"commit":{"parents":[{"commit":"dc305fe36f5f588d779ee1ffe444cdb5881086ed","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc305fe36f5f588d779ee1ffe444cdb5881086ed"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-27 00:41:49.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca774addb5e1d0be7347a2c7a2c5eb1f431705a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca774addb5e1d0be7347a2c7a2c5eb1f431705a2"}]},"branch":"refs/heads/master","description":"Rebase"},"72bcc156aa567303c320f72d3cd72fb8b9390ca6":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-07-23 08:15:57.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/8"}}},"commit":{"parents":[{"commit":"9473485f2354b983bf7786816cc373f630a15538","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9473485f2354b983bf7786816cc373f630a15538"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-12-22 00:40:43.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-07-23 08:15:57.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72bcc156aa567303c320f72d3cd72fb8b9390ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72bcc156aa567303c320f72d3cd72fb8b9390ca6"}]},"branch":"refs/heads/master","description":"Rebase"},"ba21021b5bc17abb07200ac1c81f98dc1131872d":{"kind":"REWORK","_number":9,"created":"2024-08-23 06:36:56.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/9"}}},"commit":{"parents":[{"commit":"688378f76d010bb2e3450f5f19a6c7a4e2054610","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/688378f76d010bb2e3450f5f19a6c7a4e2054610"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-23 06:26:23.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-23 06:26:23.000000000","tz":540},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba21021b5bc17abb07200ac1c81f98dc1131872d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba21021b5bc17abb07200ac1c81f98dc1131872d"}]},"branch":"refs/heads/master"},"e089146920d72ef2e16e85c9e0a360674849e252":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-09-30 05:19:32.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/10"}}},"commit":{"parents":[{"commit":"27a00ec6a8ed8fe53a4d7966ad029912ee185590","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27a00ec6a8ed8fe53a4d7966ad029912ee185590"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-23 06:26:23.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-09-30 05:19:32.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e089146920d72ef2e16e85c9e0a360674849e252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e089146920d72ef2e16e85c9e0a360674849e252"}]},"branch":"refs/heads/master","description":"Rebase"},"74efc78b6c85661c9412c80b293a81ae28d95efe":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-11-13 13:34:05.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/11"}}},"commit":{"parents":[{"commit":"06c3e1aca3d594e0ca798fb9e24775f89466704a","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06c3e1aca3d594e0ca798fb9e24775f89466704a"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-23 06:26:23.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-11-13 13:34:05.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74efc78b6c85661c9412c80b293a81ae28d95efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74efc78b6c85661c9412c80b293a81ae28d95efe"}]},"branch":"refs/heads/master","description":"Rebase"},"c426a33f159bb42d9a633b7898f3f8ee92497172":{"kind":"REWORK","_number":12,"created":"2024-11-19 04:55:30.000000000","uploader":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"ref":"refs/changes/48/899548/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/12"}}},"commit":{"parents":[{"commit":"06c3e1aca3d594e0ca798fb9e24775f89466704a","subject":"Hitachi and OEM: masking REST API token on log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06c3e1aca3d594e0ca798fb9e24775f89466704a"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-23 06:26:23.000000000","tz":540},"committer":{"name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","date":"2024-11-19 04:55:29.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c426a33f159bb42d9a633b7898f3f8ee92497172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c426a33f159bb42d9a633b7898f3f8ee92497172"}]},"branch":"refs/heads/master"},"4e990396750426ca473402fb446d699719a321a4":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-03-13 02:07:05.000000000","uploader":{"_account_id":37298,"name":"Shogo Takazawa","display_name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","username":"stakazawa"},"real_uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/48/899548/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/899548/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/899548/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/899548/13"}}},"commit":{"parents":[{"commit":"2f90a5d345cf30fe438ea8421ad4a479dd40ab85","subject":"Merge \"Hitachi and OEM: masking REST API token on log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f90a5d345cf30fe438ea8421ad4a479dd40ab85"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-08-23 06:26:23.000000000","tz":540},"committer":{"name":"Shogo Takazawa","email":"shogo.takazawa.rq@hitachi.com","date":"2025-03-13 02:07:05.000000000","tz":0},"subject":"Hitachi: Restore to support ``AIX`` as host OS type","message":"Hitachi: Restore to support ``AIX`` as host OS type\n\nHiatchi driver had supported \"AIX\" as host OS type\nby the following patch in Yoga:\nhttps://review.opendev.org/c/openstack/cinder/+/828061/\n\nBut the feature has incorrectly removed when merging the patch in Zed:\nhttps://review.opendev.org/c/openstack/cinder/+/828696\n\nThis patch restores the feature to work correctly.\n\nCloses-Bug: #2041784\nChange-Id: Ib386bfe1474fa17be82fb507095589f3ac18c007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e990396750426ca473402fb446d699719a321a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e990396750426ca473402fb446d699719a321a4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
