)]}'
{"id":"openstack%2Fcinder~899972","triplet_id":"openstack%2Fcinder~master~If2d4282b8eec07de3089901bbe90338a9a25bebf","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"36422":{"account":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"},"last_update":"2023-11-13 10:20:31.000000000","reason":"\u003cGERRIT_ACCOUNT_36422\u003e replied on the change","reason_account":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-11-16 17:14:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If2d4282b8eec07de3089901bbe90338a9a25bebf","subject":"chore: adding test test_get_nameserver_info_exception_handling","status":"ABANDONED","created":"2023-11-02 16:31:22.000000000","updated":"2023-11-16 17:14:08.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bbf8f6adbbbb2fabace5eadc7b5306f0bb463f60","_number":899972,"virtual_id_number":899972,"owner":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-02 16:47:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-11-09 15:19:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-02 16:32:40.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-02 16:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 16:51:46.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-11-02 16:52:11.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-11-02 16:54:59.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-11-02 16:58:33.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-11-02 17:50:01.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-02 17:52:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-02 17:56:38.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-11-02 18:25:35.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-02 18:28:05.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-11-02 18:34:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-11-02 18:44:56.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-11-02 21:06:35.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-02 21:37:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 01:00:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-11-09 15:19:33.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-11-10 01:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"45e557100d713fd77565b1df425603f9256c8eb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"},"date":"2023-11-02 16:31:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f6ab9675def02133415ea58cf941dc904ab281d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-02 16:32:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7630263ccb1e4566a3cdda6c5ef8678c : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8247c96416cd3f85a6c7236ba5afd2fa58593fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 16:47:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693041b1d75448de9cc062fc170ebc9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac77cfb0aed4b948516f852f11a514c : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a04b9edf536e403b8aac25b324b8c40e : SUCCESS in 14m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8a4e41fbc3f4afd8c1a98acfba6d545 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec836cf80343434682eae0d0f7e97b50 : SUCCESS in 10m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d40869abef0c49c4ad1e738687f973cf : SUCCESS in 14m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/43c54cf5ad344d9990acc28aabbfd9a7 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5fcffdca024bdeb2d5b5d654f10b0a1ff30f0fa0","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-02 16:51:46.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"434ad93bd94c62cdec7bde653878af205cacf888","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-02 16:52:11.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"0c9940f91cdcd2248990260efe76d46c1aafa4af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-11-02 16:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_11_2023_16_47_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe90e8c049f517464cfc3edbe4e0fc8458d46c42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 16:55:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cabaa5d3bfa4324870553d9be5b1a5c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed419375617247ebaff5b0f948919acb : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/588a1650dffa4f60831d96e2450649c4 : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f160dc55f306469bb36dfaa2de14980a : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adfac5b23a829aecf2720b7ed3387cc4dcd0d625","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-02 16:58:33.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"3987f1ea61bfd6bf5ea34eaafd7ffd02e2d31169","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 17:50:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerMAX_FC/574eaa0/DellEMC_PowerMAX_FC/256 : FAILURE in 1h 02m 15s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerMAX_iSCSI/0678ee3/DellEMC_PowerMAX_iSCSI/268 : FAILURE in 38m 25s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e08773c8d72d22430c185b38a20ba407212734e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 17:52:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerFlex/271f18a/DellEMC_PowerFlex/841 : FAILURE in 1h 20m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"4fa241584476c2b9dc0e5f600748d45422129aec","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-02 17:56:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/6883e6dee917454295404675de918340\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e2925c592b4249f583f033067a26c954 : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"919a5c29962cc22d602282b7eb3a5f379949f8de","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-02 18:25:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/72/899972/1/ : FAILURE in 1h 32m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cfad2311713fc9d42739abc37728ce5775c514a5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-02 18:28:05.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerStore_iSCSI/cca86d7/DellEMC_PowerStore_iSCSI/392 : SUCCESS in 1h 08m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerStore_FC/fd45c11/DellEMC_PowerStore_FC/393 : SUCCESS in 1h 29m 20s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerStore_NFS/7255e6b/DellEMC_PowerStore_NFS/400 : SUCCESS in 1h 03m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_PowerStore_NVMeTCP/ae0434f/DellEMC_PowerStore_NVMeTCP/383 : SUCCESS in 1h 40m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"86828323ff86c12f4e8b12a9d68a140309ce5055","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-02 18:34:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_Unity_FC/51a24af/DellEMC_Unity_FC/949 : SUCCESS in 2h 02m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/72/899972/1/check/DellEMC_Unity_iSCSI/b224e84/DellEMC_Unity_iSCSI/964 : SUCCESS in 2h 02m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe672a78395929094adac0c53c188b57cdb7f020","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-02 18:44:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/14/","accounts_in_message":[],"_revision_number":1},{"id":"639e59a084f8b777560e32858906c07102c72cdb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-02 21:06:35.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-72-899972-1/FC/ : FAILURE in 4h 34m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-72-899972-1/iSCSI/ : FAILURE in 3h 33m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"4c3bc9ade916354c5a9c517ae8e1525f3debfa14","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 21:37:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/72/899972/1/20231103023257/NEC-FC : SUCCESS in 3h 58m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/72/899972/1/20231103023257/NEC-ISCSI : SUCCESS in 4h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f5c89ef4ff5b61696ac4d9eae5933bd434ef3752","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-03 01:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/72/899972/1/20231103064901/NEC-V-ISCSI : SUCCESS in 3h 06m 09s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/72/899972/1/20231103064901/NEC-V-FC : SUCCESS in 3h 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"028ec6eafa4859c71f458f77e9d48e976032cb38","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-03 08:00:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/15/","accounts_in_message":[],"_revision_number":1},{"id":"bb43b0555b6ea8a1939749648c6c531349185510","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-09 15:19:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a1216b345e78f040f32875361c9ba28cf72e7c0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-10 01:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/72/899972/1/fc : SUCCESS in 0h 30m 51s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/72/899972/1/iscsi : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0394f5b3225579934ddfe9db1341b1d15331687d","author":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"},"date":"2023-11-13 10:20:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbf8f6adbbbb2fabace5eadc7b5306f0bb463f60","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"},"date":"2023-11-16 17:14:08.000000000","message":"Abandoned\n\nThe change was superseded by another change.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4db13fa4ae9cd887a738aaea843c34cdfe216f4b","revisions":{"4db13fa4ae9cd887a738aaea843c34cdfe216f4b":{"kind":"REWORK","_number":1,"created":"2023-11-02 16:31:22.000000000","uploader":{"_account_id":36422,"name":"PREMLATA","display_name":"premlata84277","email":"premlata84277@gmail.com","username":"premlata84277"},"ref":"refs/changes/72/899972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/899972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/899972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/899972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/899972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/899972/1"}}},"commit":{"parents":[{"commit":"9639bdbb9bd1922cb3f53d8dccbf2c7dd089df82","subject":"Add unit test for successful _run_ssh execution in Cisco driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9639bdbb9bd1922cb3f53d8dccbf2c7dd089df82"}]}],"author":{"name":"Premlata84277","email":"premlata84277@gmail.com","date":"2023-11-02 16:29:22.000000000","tz":0},"committer":{"name":"Premlata84277","email":"premlata84277@gmail.com","date":"2023-11-02 16:29:22.000000000","tz":0},"subject":"chore: adding test test_get_nameserver_info_exception_handling","message":"chore: adding test test_get_nameserver_info_exception_handling\n\nChange-Id: If2d4282b8eec07de3089901bbe90338a9a25bebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4db13fa4ae9cd887a738aaea843c34cdfe216f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4db13fa4ae9cd887a738aaea843c34cdfe216f4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
