)]}'
{"id":"openstack%2Fcinder~900037","triplet_id":"openstack%2Fcinder~master~I8abb69834778ba988a43c46f92827ede9004c67d","project":"openstack/cinder","branch":"master","topic":"bug/2034677","attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-11-04 03:33:52.000000000","reason":"\u003cGERRIT_ACCOUNT_35378\u003e replied on the change","reason_account":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"}},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-07-18 08:01:24.000000000","reason":"\u003cGERRIT_ACCOUNT_35378\u003e replied on the change","reason_account":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-11-03 16:56:16.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-18 08:01:24.000000000","reason":"\u003cGERRIT_ACCOUNT_35378\u003e replied on the change","reason_account":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"}}},"removed_from_attention_set":{"35378":{"account":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"last_update":"2024-07-18 08:01:24.000000000","reason":"\u003cGERRIT_ACCOUNT_35378\u003e replied on the change","reason_account":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"}}},"hashtags":[],"change_id":"I8abb69834778ba988a43c46f92827ede9004c67d","subject":"Fail to migrate, resize, evacuate VM contains volume state backing-up","status":"NEW","created":"2023-11-03 12:48:23.000000000","updated":"2024-07-18 08:01:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97cf9070c56fdb277e27157ecaab2cd7654231b7","_number":900037,"virtual_id_number":900037,"owner":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-22 19:08:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-05-22 17:02:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-22 17:02:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 12:49:53.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 13:14:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-03 13:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 14:06:09.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 14:42:46.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-11-03 14:43:11.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 14:49:09.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-11-03 15:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-03 16:49:31.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-11-03 19:03:41.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-03 19:03:59.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-11-03 19:26:23.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-11-03 19:31:04.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-11-03 20:56:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-03 21:44:25.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 23:08:19.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-11-03 23:37:29.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-11-03 23:51:04.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-11-04 01:32:22.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2023-11-04 04:50:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-06 07:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-01-08 10:50:03.000000000","updated_by":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"reviewer":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"state":"REVIEWER"},{"updated":"2024-01-08 10:50:29.000000000","updated_by":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"reviewer":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"state":"REVIEWER"},{"updated":"2024-05-22 17:49:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-18 08:01:24.000000000","updated_by":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-07-18 08:01:24.000000000","updated_by":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"abffdd4dc9bfa914ad7e12607f94beaafd8237b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-11-03 12:48:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2a9fc733bf5e85a63593f6840f2b4a773b6cbea","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-03 12:49:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/468081a939104e3ca3009704a80bff3f : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9ae4cbd65ff1a636ed1f87ab2a05e6d49c17e288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-11-03 12:51:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ad1c134d70795a08f907b86bde787ef2ddea6eb3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-03 12:53:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/165c3570288d4659b75617e0a701896e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"56383b878a64b212aa4bc38527f12f4488f65eaf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-03 13:14:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/ad2b3eaecbbe4a84ba0b41c4263f6955 : RETRY_LIMIT in 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/0a5b81076cea47989d2855cd06ff7762 : RETRY_LIMIT in 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/ad9b969fe0994536aa552b74ce06e886 : RETRY_LIMIT in 47s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/086659ce0fb44fe9a78cda85c9c4a1d9 : RETRY_LIMIT in 44s","accounts_in_message":[],"_revision_number":2},{"id":"4c9b9062de8152b770d4c4d14c4b886446cc9e9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 13:20:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc333df038e4a6fb746d8e364a5fa58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a56f2d02f63a4eb4879fd515255af7c7 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ab0de52fec948a8acf5c7230d07edef : SUCCESS in 25m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf419d51c7404accad56280c5f4c6a87 : SUCCESS in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"441f0df65c8fb0c5b3bbca15589ecd7527f748ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-03 14:06:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/37/900037/2/check/DellEMC_PowerFlex/44fc145/DellEMC_PowerFlex/845 : SUCCESS in 1h 08m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"328fac710e5e8e7ad036e1036ce4591fb33610f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-03 14:42:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/ef51b2e021ef47ae8aa33394b9e3baf5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fe535e7d1a2c426a87d8f32d97750a3d : SUCCESS in 1h 22m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b22a29dc8b0988b4cc171132b7e5bda02501f44f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-03 14:43:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/900037/2/ : FAILURE in 1h 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"695e0966a72a19e2b5022f6ba21a1963e6741eae","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-03 14:49:09.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/18/.","accounts_in_message":[],"_revision_number":2},{"id":"d4fd8244b229e380dbf53a245d907505a9cc0195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:05:21.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14d3437adedf446ea1fba3dc3f5fca49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ccf96bcd1ae4e35ace289933d542a52 : FAILURE in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac42697fc09e40a482dbcad3f0013879 : SUCCESS in 11m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d95fd2544a75447583ef87b6a7f0991d : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35921ecd5454888802f8b041adb80a3 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/c328f329b336487e85b9c3aa743722d5 : SUCCESS in 42m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/185ef0051a3444e7a9fa5cd61544f7e2 : SUCCESS in 40m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/febe7fab90fd418a8ee744c8f07fd80e : SUCCESS in 1h 23m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56bbfdc02bb546c1aedf808f03f1f0f1 : SUCCESS in 45m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ef898f8101b4ea59168eb7fe9796f18 : SUCCESS in 8m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28fe42ff55c1488490c0ee33ebe1af37 : SUCCESS in 12m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6e77eb67eb174f469904cd015c6e2f6c : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efaa304565204f098fca4ef663ee5ffd : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d64b4f1cd296408db3b90a7c31fb89ef : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/043f7870f7c9469bb1412163af127ca9 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91814c8f72114ce09163fba3585df8f9 : SUCCESS in 51m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3a16231ac2e4a1a919bb54eb1257b15 : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/62ef0a97712c456aa53aba1cd2ca873d : SUCCESS in 1h 52m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3cdc9219462a442b87846854cac06c86 : SUCCESS in 2h 11m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/036800cbe1d849c9a07d3c059a47e440 : SUCCESS in 2h 01m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4e0e73ee0770493abfdec6ad9269d33c : SUCCESS in 1h 38m 18s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/46a15fd2530349939ecbf3318ea5f43c : SUCCESS in 53m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f77bb47d087747558f42c4625b737ca6 : FAILURE in 1h 35m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/abb2df1c86d34a2f97d61c5f54127eb4 : FAILURE in 37m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8b776beb0f124ea185918059fb482f31 : SUCCESS in 1h 09m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e84050b8f7e74e4fadbc49cb6c8cb874 : SUCCESS in 1h 31m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f732b857401c4f348be4ed1b2fcf06a7 : SUCCESS in 1h 23m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22e502d6108645c58165d7ec283c5486 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":2},{"id":"7e32f1cdf64e4f93b38d0d2523b088eaa88a4d78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-03 15:07:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/18/.","accounts_in_message":[],"_revision_number":2},{"id":"31fb685aca74c482d880b33818baad749dc7afdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-11-03 15:49:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f9706a103382c623311ac098f2640b99dc0e76ff","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-03 15:50:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1656c2f739f4412283130e50073d8089 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"52522be9fcab4db91e05c1cf5e6351608c186719","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:27:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d7e1125267e4435a0945a8ff6123319\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f79c3de4dc45402c90425ef8d0633eb9 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f0653e1b0c7b40eb9193978f47cdbb57 : SUCCESS in 26m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8306751c29ad4b10a9d7ab771cc11b56 : SUCCESS in 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"56558949f5ac73053247fc147c1a89843735ddde","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-03 16:49:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72d98da34ea64743ac02cf3b4bb49be6a4cf88c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-11-03 16:56:16.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5796c805d044161fbde895e41f6b34363be0d1b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-03 16:57:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/94eebf691f054c52a849c73bdd6e7c73 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"816d148793264856ba92f94ceb4a82f9807b0ff9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 17:24:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24bb02834061403b91c6442185c3c8e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/669ce1346413430fbb0b410146bbd1da : SUCCESS in 25m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3337ca9a469242d5b9ec3fcac7ede92b : FAILURE in 23m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb55d27df6024441bfefd1a4f48964da : SUCCESS in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f70eecdb619e70f646eb1ba468683b894edfcb27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-03 18:07:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerFlex/093c60d/DellEMC_PowerFlex/847 : SUCCESS in 1h 10m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"d1392080443ba9a25f691fd91c37475178055a63","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-03 18:18:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/20/.","accounts_in_message":[],"_revision_number":4},{"id":"921a7232ac399c04535d51b8799189b01b747862","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-03 18:19:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/071eb7b3412c4626a8ea80865c638ad9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9179dfdce5d04ba88d27295d0c20b2b8 : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":4},{"id":"55d21c4bb7214b92fa54ef4d3a529671a0774558","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-03 18:32:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/37/900037/4/ : FAILURE in 1h 27m 37s","accounts_in_message":[],"_revision_number":4},{"id":"dcfbe95d2fc153f330610c5c8b1f49f8a6d18d94","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-03 18:58:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/37/900037/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/91e0412/ : FAILURE in 2h 01m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/37/900037/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/0fed535/ : SUCCESS in 1h 44m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/37/900037/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/9808902/ : SUCCESS in 1h 44m 24s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/37/900037/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/eae2565/ : SUCCESS in 1h 44m 24s","accounts_in_message":[],"_revision_number":4},{"id":"4cb73a72c55d0d34a25b94a93a2bc784a5c97e78","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-03 19:03:41.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-2/FC/ : UNSTABLE in 2m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-2/iSCSI/ : FAILURE in 5h 44m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"01158bcaffbc9ef8a0b0bb717839461cf01d6ccc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-03 19:03:59.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_Unity_FC/64ec336/DellEMC_Unity_FC/955 : SUCCESS in 2h 04m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_Unity_iSCSI/5c13cae/DellEMC_Unity_iSCSI/970 : SUCCESS in 2h 07m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b682e08afd90eaa1adacd18a7ef9127fd243069a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-03 19:05:20.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-1/FC/ : UNSTABLE in 2m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-1/iSCSI/ : FAILURE in 6h 16m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"5537b7fc92e7e6844e6d257dead60f18ab429e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 19:16:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bd23826171f445ea1b8e6da6a330464\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70507b11ef3470ab0ce3e93cd618ee9 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c07102992484ef983cdcae16802be43 : SUCCESS in 12m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2db69d89b0b4f4a9b07bbee8891f50d : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c22c4b966d4242a9e6403f1e8709bc : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c64d3f16708d4355a0675e96eed27785 : SUCCESS in 1h 01m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/49ef580e56cd4bc6825315659ee2ff11 : SUCCESS in 47m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16b94708f1b24140bab18a7c7cde6a97 : SUCCESS in 1h 20m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dbbde03762a45ecb7baa75c1c96e5d3 : SUCCESS in 1h 04m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7042ac99241e4741a493513216b20134 : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ade97f4936c64c7c9d35d6c71bdd7f58 : SUCCESS in 14m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ae38dff5849345ddaa9544cfbc90eeff : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/46713abfe0444e91a5dd5c8bb7427040 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48de4f01226840bdb8af5dfaba1c9681 : SUCCESS in 6m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/405e928161764ea5a46561f11a882321 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/753ed3fde3254a9299bc41094648c41b : SUCCESS in 39m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33c4cdb56f114fdca7c79c1913981bc8 : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16df408dd18043e6b61d6569d873aead : SUCCESS in 2h 13m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5e999d6c51a6443c955d1ac332ca7d3f : SUCCESS in 2h 09m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a65b8e47dff443e9ea976ff18f3b96b : SUCCESS in 1h 53m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b8d013a1a21f4bb3973b1c42a5a67caf : SUCCESS in 1h 50m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2817e8c46f1e464083b02437ee4ed9e6 : SUCCESS in 1h 02m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aa2fc02cb9da429d9b5a1e5ac2656ca2 : SUCCESS in 1h 25m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aa1dcee518f146f2a6ad4dc30912723e : SUCCESS in 45m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ee93d1f3da9f46df915e3b70ea331b01 : SUCCESS in 1h 11m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d3f22c4f07714be4a3377409841d8194 : POST_FAILURE in 1h 55m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc41983ece8549b5b22167580040d3f4 : SUCCESS in 1h 06m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66d31077d7c4488faec39a2dcc526769 : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":4},{"id":"5f582d360538771a14aae7b1b11ce7fa8f80457b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-03 19:26:23.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"7f046f75875f99adb3f5c8f19814c143cd5beddd","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-03 19:31:04.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"3eb6542a2721ae57f5fc9cb6a50c2f6ae882f986","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-03 19:46:35.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/20/.","accounts_in_message":[],"_revision_number":4},{"id":"42d087070c4df2b9b3be142ee6c09a6b68cadc0a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-03 20:56:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/37/900037/4/upstream-check/cinder-tempest-plugin-ontap-fcp/96719b1/ : SUCCESS in 1h 04m 40s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/37/900037/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/d640309/ : SUCCESS in 1h 32m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/37/900037/4/upstream-check/cinder-tempest-plugin-ontap-nfs/788b3f8/ : SUCCESS in 1h 38m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/37/900037/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3145913/ : SUCCESS in 1h 36m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"57f9175b989bafb749b822e3ee1dc0249f348763","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-03 21:31:17.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerFlex/2ccb552/DellEMC_PowerFlex/850 : SUCCESS in 1h 10m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ba2265ab114edabd8f5a01de6a3bb081a4b1577e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-11-03 21:44:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerMAX_FC/9512784/DellEMC_PowerMAX_FC/260 : FAILURE in 1h 08m 23s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerMAX_iSCSI/ca1f6b1/DellEMC_PowerMAX_iSCSI/272 : FAILURE in 46m 37s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"90c9e6e17c65ee6be3a6b166084ce8ae3983b2ee","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-03 23:06:35.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-4/FC/ : UNSTABLE in 3m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-4/iSCSI/ : FAILURE in 5h 54m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"ad2c936dcf6359dc17789126c89db6b82c28651d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-03 23:08:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerStore_iSCSI/b91838e/DellEMC_PowerStore_iSCSI/396 : SUCCESS in 57m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerStore_FC/24e2aa6/DellEMC_PowerStore_FC/397 : SUCCESS in 1h 30m 08s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerStore_NFS/74b0e4e/DellEMC_PowerStore_NFS/404 : SUCCESS in 57m 19s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_PowerStore_NVMeTCP/d887954/DellEMC_PowerStore_NVMeTCP/387 : SUCCESS in 1h 30m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"dfec6b70d671de5c0b530f022e8d64070d99c088","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-03 23:12:51.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-3/FC/ : FAILURE in 6h 31m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-37-900037-3/iSCSI/ : FAILURE in 5h 50m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"482af01d8575529a08ed1508ffedab32cbd8cd52","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-03 23:37:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900037-4/HPMSA-iSCSI/ : SUCCESS in 1:52:28\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900037-4/HPMSA-FC/ : FAILURE in 1:44:15","accounts_in_message":[],"_revision_number":4},{"id":"44999c3ffa5545f408ab81d46e4d9efb6782a9b2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-03 23:51:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/4/20231104025650/NEC-V-FC : SUCCESS in 3h 14m 48s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/4/20231104025650/NEC-V-ISCSI : SUCCESS in 5h 54m 10s","accounts_in_message":[],"_revision_number":4},{"id":"0ae66bedff564840612369745fea8d093fd01f44","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-04 01:26:07.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_Unity_FC/678a6c0/DellEMC_Unity_FC/958 : SUCCESS in 2h 06m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/37/900037/4/check/DellEMC_Unity_iSCSI/ccc7cd6/DellEMC_Unity_iSCSI/973 : SUCCESS in 2h 07m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"53283dc608fc51300a47b73f8c58d2109e5a90f8","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-04 01:32:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f272b0638a8896b33f608f69fb0ed2ba1a54ca35","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-11-04 03:23:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61d33b7b8c7fd5869c09320bc648a29d2f66dbb3","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-11-04 03:33:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ca235ee41a91f2322e6279eb0f6be1249ec8145","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-04 04:50:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/4/20231104051135/NEC-FC : SUCCESS in 3h 46m 18s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/4/20231104051135/NEC-ISCSI : SUCCESS in 8h 38m 40s","accounts_in_message":[],"_revision_number":4},{"id":"9b7531d0b6f14855690c3532fc5e3bf47d780047","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-06 07:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/37/900037/4/fc : SUCCESS in 0h 41m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/37/900037/4/iscsi : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":4},{"id":"45817b989e60a0b97481008eb20bd732eda90873","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2023-12-26 10:34:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5fde214a03cb0d728937afee79c1f73eecb333da","author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"date":"2024-01-08 10:50:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a4b34b438cb14999a260359e369296e0828b54c","author":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"date":"2024-01-08 10:50:29.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b4bf784de89fe0ebefeff8821bd20aef9f3439b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"real_author":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"date":"2024-05-22 17:02:11.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_35378\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"}],"_revision_number":5},{"id":"8a490784e59dc0ba938c9a693f81945650d15b04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 17:17:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432b2bf0d28b478886ee3697409acb88\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/394e60232d914bed96998ba4d91f19c0 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/49065c593ee34089b90f2ea3ce25a459 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4b4b22e533a40fcec1470956fd0d95ad6be37b9a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 17:29:27.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c58dab61d70443649ff726f359b46a06\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e8e796dc75084f50b6b0653ab56e028c : POST_FAILURE in 20m 04s","accounts_in_message":[],"_revision_number":5},{"id":"a11328ae78704a885581eaaa2af7df86f1a2ce47","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-05-22 17:49:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/900037 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":5},{"id":"283dd9e307bdbfeeb45648434d181888f506d49b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 19:08:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cffc5a6d986497eb24cfa32716baaab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fcf2242fa6249bc98a1570d26cddede : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e4d088a65fd437ebc3e2092785021b5 : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8be79b62b444f8f99fa6e9e9c1240a2 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/521a7f2d1433407caa6fd12a64fb306d : FAILURE in 16m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1118ec470dfc40a793f80c52c2b20caa : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7b6de9985c0f4d5793f81f2e908a4dfb : SUCCESS in 49m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c2b59f8bd854c00a8d8b551caefb460 : FAILURE in 1h 16m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2baa65462f584283acf38c102a56df11 : SUCCESS in 1h 01m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/826b62711799481e8401ac6bed9ebd59 : SUCCESS in 11m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28f396fc75b34459a2bf68bea22e0b43 : SUCCESS in 16m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/df946f9b2fbc4799acb4bc9830cd7582 : SUCCESS in 6m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fed0c0dc5ed64bb1bf28734450971327 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/04f5d53faf4b4565a9088bad14261873 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0ae9e0f975e8465186da9b72d81c3754 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ac2a871f6164d388795958083fd1fd5 : FAILURE in 46m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/123182eecada4d248d3e7fc9921ca913 : FAILURE in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/382ed4251c0d43698af2e5967f043152 : SUCCESS in 1h 47m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/55c47777eede4647912998bd7173e22e : SUCCESS in 2h 05m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/562ea1a59d5944d99c939639b2daca1a : SUCCESS in 1h 32m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/28819a42355b4548b709f4d581d18b93 : SUCCESS in 1h 23m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/df6582589b5e40c8b922f8b185fc3726 : SUCCESS in 39m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bf7666bb68804c1c91130b1ef54a8c6b : SUCCESS in 58m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/890175e749d7465f91dd52a9dd358000 : SUCCESS in 1h 27m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7f0a6c2a2ace436aabe21de1eb7781e1 : SUCCESS in 45m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5bd8babb6df443488c8e563955a986be : SUCCESS in 1h 12m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6ac6c6bc01a94c4d8c93c0c4848977f1 : SUCCESS in 1h 18m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc478013b53b455890ad5e141adc0005 : SUCCESS in 1h 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b69cf422d69749c6b14b635dc3d34d76 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":5},{"id":"ec0e5fdb5c4c9b4d4156f2a85ea8fb04b9bec334","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2024-05-23 06:06:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0afbbdee956afff374e1f11fa7209bbf3cf778b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 06:15:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967e7887962e4fbea4cf64591f2c6960\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74f2ad4f48e940139774faba61dff7f7 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e4a2f9b2c4db4af1abd7e892a204f7b9 : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5317be7d61a77bb78fdedb65fc0c8515ca271c06","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 06:29:22.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2834d18816084628b7ee653f23d1e6f7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2758a733cb4e49b98c35e705e074ecbc : POST_FAILURE in 20m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0b94bc9143c05d39937dbb2577ed2d4e737b36da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-23 08:01:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/471/.","accounts_in_message":[],"_revision_number":5},{"id":"0d6bde7454e00ff2a40574e1bf32aa31b99e5900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 08:18:35.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/324fe153fcfd4327982db47564619bba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1242e955f34b1d9acd28a93362b373 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69c84d8b0a55430880045d20a9ef3ca1 : SUCCESS in 9m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32e203e0d35f4037b40770e1a30d46e7 : SUCCESS in 9m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b1a97742238457fa56b42bd27f5db25 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9141abb18494e73b48a829aab494b82 : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/fe1772c6e42c4e71860c5f017e7d7a5a : SUCCESS in 49m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f5cfbcd5cbe435c8fbfdcdae2860bdc : FAILURE in 1h 16m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20292148f9c045c5a21b4ae6cfea3966 : SUCCESS in 51m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75d41e09a451467fa07b6c254b494711 : SUCCESS in 12m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4dc6b41ea844d3291b308b2e62ad732 : SUCCESS in 15m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2415648dc5f4dc9a3f2705876af63d6 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8757601857d84a97813e0a5a4db6f394 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f494ff5cccb24cbcb5ddb8511826f0a8 : SUCCESS in 5m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e61e19e773c3467497f00c8a2380ad2b : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c73c1e1fbc4142b5928296b858dc219c : FAILURE in 29m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd1013412b334b80a3c699991a8fcb91 : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ccadc4ed2a294a38abff977cf415b915 : SUCCESS in 1h 52m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/47b8added36f460b8d71feba30dbdc30 : FAILURE in 1h 39m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97adfe12ef304e5a8cea83943b145085 : SUCCESS in 1h 40m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1909d6ccf24b4eaeb9ccdc308f1913bf : SUCCESS in 1h 30m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1e8ed66d69ed4ce19cc03975e477b766 : SUCCESS in 37m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/446c1e82b8404a0b9f50d1f9397964bb : SUCCESS in 1h 10m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6ab3659478824e43a2ad99e5b2fdf127 : SUCCESS in 1h 17m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7d45134cf27c4ac9886dc4570427ccee : SUCCESS in 41m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/18049a7b28164bdcb91cfd4fc80cbe20 : SUCCESS in 1h 19m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bff3a29d6e814dc1b15b1d3b043243cd : SUCCESS in 1h 03m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ecf65d386982400288ce2b7037700ae8 : SUCCESS in 1h 41m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52f4073457444d69824197c3920f2268 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":5},{"id":"959247408e17b038a85370879ffa9a2c2f7d7dcb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-23 15:00:35.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/37/900037/5/check/DellEMC_PowerFlex_3.6/2ecded8/DellEMC_PowerFlex_3.6/1429 : FAILURE in 1h 57m 20s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/37/900037/5/check/DellEMC_PowerFlex_4.5/3962201/DellEMC_PowerFlex_4.5/343 : SUCCESS in 1h 46m 56s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/37/900037/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/54a3145/DellEMC_PowerFlex_4.x_NVMeTCP/284 : FAILURE in 2h 56m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"eef38f992de3028f69adc979986cdffc5e75d71d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-24 22:31:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/37/900037/5/check/DellEMC_Unity_FC/279ca24/DellEMC_Unity_FC/1951 : FAILURE in 1h 10m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/37/900037/5/check/DellEMC_Unity_iSCSI/5472326/DellEMC_Unity_iSCSI/1977 : FAILURE in 1h 45m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"7865a6748163f7b1b7cf5086d18ffed4424ac3d8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-05-31 01:14:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/5/20240531014044/NEC-ISCSI : SUCCESS in 5h 01m 58s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/5/20240531014044/NEC-FC : SUCCESS in 8h 33m 48s","accounts_in_message":[],"_revision_number":5},{"id":"62ee9be7d205e26dfdb46448c1bd5749475733b4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-06-03 15:23:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/5/20240603203957/NEC-V-ISCSI : SUCCESS in 3h 31m 15s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/37/900037/5/20240603203957/NEC-V-FC : SUCCESS in 3h 43m 51s","accounts_in_message":[],"_revision_number":5},{"id":"97cf9070c56fdb277e27157ecaab2cd7654231b7","author":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"date":"2024-07-18 08:01:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"355af5ee97e11e0227b219d5893932518e1fe81d","revisions":{"fdc7007c1ebb64ffeb759c9a76f20aacf7ad59df":{"kind":"REWORK","_number":1,"created":"2023-11-03 12:48:23.000000000","uploader":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"ref":"refs/changes/37/900037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/900037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/900037/1"}}},"commit":{"parents":[{"commit":"5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470","subject":"Merge \"Fix test_nvmet_driver tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470"}]}],"author":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 12:48:08.000000000","tz":420},"committer":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 12:48:08.000000000","tz":420},"subject":"Fail to migrate, resize, evacuate VM contains volume state backing-up","message":"Fail to migrate, resize, evacuate VM contains volume state backing-up\n\nCloses-Bug: #2038367\n\nBehavior: VM fail to perform normal operation such as live/cold-migrate, resize, evacucate\nwhen has any volume state backing-up.\n\nProposed Fix: The _attachment_reserve function from cinder api not allows volume with state backing-up,\nthe currenttly valid state is available \u0026 downloading. This proposed fix will add backing-up state to expected\narray cause of backup process will take action on the clone volume and not impact to the original volume\nattachment process.\n\nChange-Id: I8abb69834778ba988a43c46f92827ede9004c67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdc7007c1ebb64ffeb759c9a76f20aacf7ad59df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdc7007c1ebb64ffeb759c9a76f20aacf7ad59df"}]},"branch":"refs/heads/master"},"35e3d7a7fcdd81b2fbc9fe1096d6c94ba1b97b45":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-03 12:51:48.000000000","uploader":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"ref":"refs/changes/37/900037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/900037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/900037/2"}}},"commit":{"parents":[{"commit":"5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470","subject":"Merge \"Fix test_nvmet_driver tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470"}]}],"author":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 12:48:08.000000000","tz":420},"committer":{"name":"Sang Tran","email":"SangTQ8@fpt.com.vn","date":"2023-11-03 12:51:48.000000000","tz":0},"subject":"Fail to migrate, resize, evacuate VM contains volume state backing-up","message":"Fail to migrate, resize, evacuate VM contains volume state backing-up\n\nCloses-Bug: #2034677\n\nBehavior: VM fail to perform normal operation such as live/cold-migrate, resize, evacucate\nwhen has any volume state backing-up.\n\nProposed Fix: The _attachment_reserve function from cinder api not allows volume with state backing-up,\nthe currenttly valid state is available \u0026 downloading. This proposed fix will add backing-up state to expected\narray cause of backup process will take action on the clone volume and not impact to the original volume\nattachment process.\n\nChange-Id: I8abb69834778ba988a43c46f92827ede9004c67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/35e3d7a7fcdd81b2fbc9fe1096d6c94ba1b97b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/35e3d7a7fcdd81b2fbc9fe1096d6c94ba1b97b45"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b09f69daeb0e23810bccc1ea362a74007741ce25":{"kind":"REWORK","_number":3,"created":"2023-11-03 15:49:20.000000000","uploader":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"ref":"refs/changes/37/900037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/900037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/900037/3"}}},"commit":{"parents":[{"commit":"5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470","subject":"Merge \"Fix test_nvmet_driver tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470"}]}],"author":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 12:48:08.000000000","tz":420},"committer":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 15:47:32.000000000","tz":420},"subject":"Fail to migrate, resize, evacuate VM contains volume state backing-up","message":"Fail to migrate, resize, evacuate VM contains volume state backing-up\n\nCloses-Bug: #2038367\n\nBehavior: VM fail to perform normal operation such as live/cold-migrate, resize, evacucate\nwhen has any volume state backing-up.\n\nProposed Fix: The _attachment_reserve function from cinder api not allows volume with state backing-up,\nthe currenttly valid state is available \u0026 downloading. This proposed fix will add backing-up state to expected\narray cause of backup process will take action on the clone volume and not impact to the original volume\nattachment process.\n\nChange-Id: I8abb69834778ba988a43c46f92827ede9004c67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09f69daeb0e23810bccc1ea362a74007741ce25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09f69daeb0e23810bccc1ea362a74007741ce25"}]},"branch":"refs/heads/master"},"bf090390a1c16933c4630faa4e0408c65d28ccef":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-11-03 16:56:16.000000000","uploader":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"ref":"refs/changes/37/900037/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/900037/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/900037/4"}}},"commit":{"parents":[{"commit":"5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470","subject":"Merge \"Fix test_nvmet_driver tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ea10b2ecc41ad235cbdbbf94b4a1de98e30e470"}]}],"author":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 12:48:08.000000000","tz":420},"committer":{"name":"Sang Tran","email":"SangTQ8@fpt.com.vn","date":"2023-11-03 16:56:16.000000000","tz":0},"subject":"Fail to migrate, resize, evacuate VM contains volume state backing-up","message":"Fail to migrate, resize, evacuate VM contains volume state backing-up\n\nCloses-Bug: #2034677\n\nBehavior: VM fail to perform normal operation such as live/cold-migrate, resize, evacucate\nwhen has any volume state backing-up.\n\nProposed Fix: The _attachment_reserve function from cinder api not allows volume with state backing-up,\nthe currenttly valid state is available \u0026 downloading. This proposed fix will add backing-up state to expected\narray cause of backup process will take action on the clone volume and not impact to the original volume\nattachment process.\n\nChange-Id: I8abb69834778ba988a43c46f92827ede9004c67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf090390a1c16933c4630faa4e0408c65d28ccef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf090390a1c16933c4630faa4e0408c65d28ccef"}]},"branch":"refs/heads/master","description":"Edit commit message"},"355af5ee97e11e0227b219d5893932518e1fe81d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-22 17:02:11.000000000","uploader":{"_account_id":35378,"name":"Sang Tran","email":"SangTQ8@fpt.com","username":"sang.tran"},"real_uploader":{"_account_id":35916,"name":"Pham Le Gia Dai","display_name":"Dai","email":"daipham.3213@gmail.com","username":"daiplg","status":"Full Stack Dev"},"ref":"refs/changes/37/900037/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/900037/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/900037/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/900037/5"}}},"commit":{"parents":[{"commit":"ccff71a251a68c0bcbc1a38f6e8435cddacf33c5","subject":"Merge \"swift backup driver: Ignore 404 during object deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccff71a251a68c0bcbc1a38f6e8435cddacf33c5"}]}],"author":{"name":"sangtq8","email":"sangtq8@fpt.com.vn","date":"2023-11-03 12:48:08.000000000","tz":420},"committer":{"name":"Sang Tran","email":"SangTQ8@fpt.com","date":"2024-05-22 17:02:11.000000000","tz":0},"subject":"Fail to migrate, resize, evacuate VM contains volume state backing-up","message":"Fail to migrate, resize, evacuate VM contains volume state backing-up\n\nCloses-Bug: #2034677\n\nBehavior: VM fail to perform normal operation such as live/cold-migrate, resize, evacucate\nwhen has any volume state backing-up.\n\nProposed Fix: The _attachment_reserve function from cinder api not allows volume with state backing-up,\nthe currenttly valid state is available \u0026 downloading. This proposed fix will add backing-up state to expected\narray cause of backup process will take action on the clone volume and not impact to the original volume\nattachment process.\n\nChange-Id: I8abb69834778ba988a43c46f92827ede9004c67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/355af5ee97e11e0227b219d5893932518e1fe81d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/355af5ee97e11e0227b219d5893932518e1fe81d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
