)]}'
{"id":"openstack%2Fcinder~900118","triplet_id":"openstack%2Fcinder~master~Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c","project":"openstack/cinder","branch":"master","topic":"dell-powermax-restapi-101","attention_set":{},"removed_from_attention_set":{"36625":{"account":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"last_update":"2024-02-13 19:47:02.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-13 17:05:44.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-02-13 05:07:04.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-02-07 13:44:00.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-02-02 23:03:40.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2024-02-13 19:47:02.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-02-13 17:52:57.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2024-02-13 19:47:02.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-12-18 11:04:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c","subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","status":"MERGED","created":"2023-11-09 06:18:18.000000000","updated":"2024-02-15 10:06:32.000000000","submitted":"2024-02-13 19:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":65,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900118-dell-powermax-restapi-101","meta_rev_id":"8d64e4caafc2f362c2c38f96b7d25cdb15cd4db7","_number":900118,"virtual_id_number":900118,"owner":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-13 19:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":1,"date":"2024-02-13 05:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2024-02-13 17:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-02-07 13:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2024-02-13 17:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2024-02-13 17:52:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}],"CC":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 06:19:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-09 06:42:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-09 07:45:20.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-11-09 07:48:11.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2023-11-09 08:20:49.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-09 09:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 10:07:33.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-11-09 10:28:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-09 20:05:38.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-10 05:10:15.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-11-10 05:17:59.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-11-10 09:56:57.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-10 15:32:55.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-11-11 15:39:46.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-11-12 16:19:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-15 06:59:23.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2023-11-15 22:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-11-24 09:48:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-24 11:14:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-24 12:07:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-11-25 14:38:18.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-11-30 11:43:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-05 03:20:18.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-12-05 03:20:18.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-12-13 14:02:27.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-12-18 11:04:38.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2024-01-17 20:49:15.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-01-24 03:46:55.000000000","updated_by":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"reviewer":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"state":"CC"},{"updated":"2024-01-24 08:01:06.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-29 05:06:21.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"CC"},{"updated":"2024-01-29 05:06:21.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":36396,"name":"Bryan Neumann","display_name":"Bryan Neumann","email":"bryan.neumann@dell.com","username":"bryanneumann","status":"Dell"},"state":"CC"},{"updated":"2024-01-29 05:06:21.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"state":"CC"},{"updated":"2024-01-30 23:41:34.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-02-01 11:42:02.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-02 00:56:38.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-02-06 02:49:50.000000000","updated_by":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"reviewer":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"state":"CC"},{"updated":"2024-02-12 07:35:44.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-02-12 07:35:44.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2024-02-12 07:35:44.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2024-02-12 07:35:44.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-02-13 05:07:04.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-02-13 14:44:14.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"818731acc9de01f96c00ae14540896597596ba78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-09 06:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97e38f522f2ac0db82a8a7f5546fca0c24a98184","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-09 06:19:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/55dbaa66eaf64765b1345400e03c3bdd : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6d38422710e831daf740f32a13c48c7efb38cc78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 06:42:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6257dda9cdd740998cd2e5631b0add51\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f0acdf2b2ac4d418be68eba39c7a664 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f6e20c82c0f646ffae48765a847668cb : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a2c65e48506484eacc63c2dc553a484 : SUCCESS in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e71b8985ec1afa33d0f4dd5d4d67cbad151537d2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-09 07:45:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/bac4e7ae70c441dbbab8d0ff196a891d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7c4c09b61579466bb0b5f615f4ef0a6c : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8436ad5cda0bb2d9e2a36f2b7cda684db91409a1","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-09 07:48:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"584e61d4a85dc5645ce3923daea3e23e8e663712","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-09 08:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/900118/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/d6501bb/ : SUCCESS in 1h 44m 30s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/900118/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/b5b83a0/ : SUCCESS in 1h 48m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/900118/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/c6166ed/ : SUCCESS in 1h 39m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/900118/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/456e02f/ : SUCCESS in 2h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"54b61b41a510eb1d438e24255c94c8b937762214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 09:01:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b64b78c22d6a482e8b12242323456e05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe21fccc36c46a39031f7dea612c164 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dd9b71367bb407fbe8ca4d3bb6d8dd3 : SUCCESS in 8m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1135b28a4f848198a97d3c0e7020ed9 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b4631fbda048379f8f569c5273b04a : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/c146dafc9940408f912b07caa8de7794 : SUCCESS in 47m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0a33f96a3028403c8bbddcd48ac3f6b5 : SUCCESS in 49m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21e6cbef354d491fbc2d4222f8892c60 : SUCCESS in 1h 06m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1053e241e5984488add4e7f89ec86d66 : SUCCESS in 55m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/120f06d49b394919882df748f7d994a8 : SUCCESS in 13m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/60749276c2aa443f93705b0deaa4bce9 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc228d68fd4e415e98bb3576dd7c9c68 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d441b1a8f24f4987b963a65e2eee3d94 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d10c7905120848a0b82c569071616a67 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/848bc1bb397a4ee6a394c2e34b5267c7 : SUCCESS in 33m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d87ead8c2dee4b3c835fe3f7930835fd : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d0104f7011674d789b54e305dc2dabe3 : SUCCESS in 2h 02m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1bb5d8624e6c4e4dab21b7ed608cab03 : FAILURE in 2h 37m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fae6d57522b5415fa1cab412e2e28e03 : SUCCESS in 1h 42m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/74614938a0b849e7b6e0bf8fda1ba0bf : SUCCESS in 2h 25m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d3a36713dba440aabd23fdd7851ed43d : SUCCESS in 1h 23m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b6eac86cfb06468ea2bf724820461e25 : SUCCESS in 1h 26m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f66f8110d9364a1cb4d1370987acb09d : SUCCESS in 51m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e8e129b4cb8f4db2b2c38122b0cecf5d : SUCCESS in 1h 38m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/490054f2ded541b1a55818755c6b7e7d : SUCCESS in 1h 43m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e0f80fb657b04bfca19ffa52d17637d4 : SUCCESS in 1h 36m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7705ec32242a45d58fb9e7157952f09d : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"428f7530f1d9b06e941c91891c03501862151227","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-09 10:07:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/45/.","accounts_in_message":[],"_revision_number":1},{"id":"0695e423b5b7c60a01cbf77bc2fecce2295c8f85","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-09 10:28:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/1/upstream-check/cinder-tempest-plugin-ontap-fcp/fdeda90/ : SUCCESS in 1h 25m 44s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/38520b9/ : RETRY_LIMIT in 7m 18s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/1/upstream-check/cinder-tempest-plugin-ontap-nfs/9db8874/ : RETRY_LIMIT in 7m 16s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1fe79ab/ : RETRY_LIMIT in 7m 19s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d3348090c27ee8e14b9500b99832597a23919a1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-09 11:23:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/46/.","accounts_in_message":[],"_revision_number":1},{"id":"f52ac8daa13308d0f5c7dd3981c7e22d97f77974","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-09 20:05:38.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-1/FC/ : UNSTABLE in 55m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-1/iSCSI/ : FAILURE in 8h 14m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e71e8df591462cd82ea28c68c75b7ed8ad65ac64","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-10 05:10:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/900118-1/Lenovo-iSCSI/ : SUCCESS in 2:29:56\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/900118-1/Lenovo-FC/ : SUCCESS in 1:53:40","accounts_in_message":[],"_revision_number":1},{"id":"b7f663dc134a14a7ce73ffbe5ba26f2d9dc1d779","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-10 05:17:59.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"1d23840610664150ef576faa77aff29014074364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-10 09:54:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"719de7cfa0745a3a24e41435a9d88a9b73e3d764","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-10 09:55:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee9284282fc64418a835aba4582ded57 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c89a387ef259304c4869be9544adccc96be60f84","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-10 09:56:57.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"390b95ea57f9ad802d60b8edf4efbd2ad75351f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-10 10:15:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41dd0d9598293f6d1ab9ebfd5fbc870b6247aaca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-10 10:16:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a45b73407f2842a8b5eadc117de07ca5 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"a7723cfac14e8087fc4949779e5a58d99d14d38e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 10:37:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cebac72085c4d0893f0e74030b06a71\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d2a8dadb0904afcae2a4f0108a70615 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8672549903164867bc3a5d84e643d72e : SUCCESS in 20m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a6a704739be4eddb4e6c029fd668ab9 : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fef3d24a2a7a771cec7bb7313673a9a16169d718","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-10 11:37:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/786f4572779a45d099c77cecb5c6884b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d879f7153c6c4405a466ac2945d85bf9 : SUCCESS in 1h 19m 34s","accounts_in_message":[],"_revision_number":3},{"id":"492eb563523094da3043356c7288c5977bf3b300","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-10 12:03:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/900118/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/3f14c45/ : SUCCESS in 1h 47m 05s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/7c4a1242823a46099b55da16289be7fb : RETRY_LIMIT in 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/8ae0612a03a54b6da7b0b5619120c119 : RETRY_LIMIT in 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/0653258762004a25b99d2220e7ef5350 : RETRY_LIMIT in 41s","accounts_in_message":[],"_revision_number":3},{"id":"38a2f34955bf550223e18ed69efc2ffca7e2a2eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:36:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c19564c8e192415799ab17b7c0291e57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c60e3126eee461bb42ce95245c441c3 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8d69c771d464770a59c8b3a5019dd27 : SUCCESS in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dac56fe85828478d9e79783bb9a6def3 : SUCCESS in 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ad8c1d955449818f8b0424f14c1005 : SUCCESS in 15m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4067679f88f64239897ec4e1bf538a52 : SUCCESS in 1h 01m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6cdf9eea4de4418808c1e9e050e96c9 : SUCCESS in 59m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb5530495ca4401cbb3dc7981d87c133 : SUCCESS in 1h 16m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed43978f27ea498289f95a902f17a001 : SUCCESS in 56m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b0e181531944d859a25bf46e6bead1f : SUCCESS in 7m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0736ddca4ae746fbbbe0dfbd1de24064 : SUCCESS in 13m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/950a251b55ca4e909eb1e2b6a0faf10f : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9716ae71632c4060ba607c75b878f149 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3a1d77181c74a92bee3204637bc9d4d : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3de1cc66469149a5893c0d4abcfc1378 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/521466e1d3824b4bbf299e97e2bc3b2b : SUCCESS in 51m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/596a98bcadeb466a95a7aded10d6d4bf : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aad4a8c9f08945788288fd3a39a125b2 : SUCCESS in 2h 14m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b948daf6eb8a43019a54883f3b48315e : FAILURE in 2h 02m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6a77518bc9aa46d68ad3565fec560b82 : SUCCESS in 1h 52m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/baa399f12de44299ac5c3fc6e384a31d : SUCCESS in 2h 03m 12s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d03608b1d0243b6835f03d03b48796b : SUCCESS in 1h 24m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a03564f24d0460fb6022d719f1714bc : SUCCESS in 1h 31m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ae4e503a7eb48ec8b3c08e80fb6c886 : SUCCESS in 39m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1079a26c9e0c4238970e1ceb04db772e : SUCCESS in 1h 27m 50s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c7d125176de64ec5896f4df194fffba6 : SUCCESS in 1h 34m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d9c0f7f2acf4d3fbf21bad12b4ac3ee : SUCCESS in 1h 37m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/380bbe8e528e40a2baa1210739bf6958 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":3},{"id":"b3bfd740919a4538923f769e97f3c0af5563cd78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-10 12:51:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/50/.","accounts_in_message":[],"_revision_number":3},{"id":"c42ca6c14e7b76d23e4e440087de599dcc8f0e2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-10 12:52:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/49/.","accounts_in_message":[],"_revision_number":3},{"id":"7909fc0d710e84cc7dc017fe3676335cfceafa10","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-10 13:43:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/3/upstream-check/cinder-tempest-plugin-ontap-fcp/8b88ed3/ : SUCCESS in 1h 06m 20s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/3ffa714/ : RETRY_LIMIT in 7m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/3/upstream-check/cinder-tempest-plugin-ontap-nfs/03ccf30/ : RETRY_LIMIT in 8m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8bb108a/ : RETRY_LIMIT in 8m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"761dbded995fd145fcd0aec4c2427861d3ae84ee","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-10 15:32:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900118-1/HPMSA-iSCSI/ : SUCCESS in 1:52:57\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900118-1/HPMSA-FC/ : SUCCESS in 1:35:44","accounts_in_message":[],"_revision_number":1},{"id":"73212d04a04101fe902d21024cf0b432d8a97565","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-10 16:32:20.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-2/FC/ : UNSTABLE in 3m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-2/iSCSI/ : FAILURE in 6h 37m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"a729db9e5bfc7df81104f4260248fd4803be1c00","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-10 17:06:37.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-3/FC/ : UNSTABLE in 2m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-3/iSCSI/ : FAILURE in 6h 11m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"f14af5e01eda5f4ffad1105500389d59789f7b20","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-10 20:29:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/900118/3/ : FAILURE in 1h 37m 39s","accounts_in_message":[],"_revision_number":3},{"id":"8f828a871657aa9c25da614153ff72c24e0341d9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-10 20:55:14.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"67ead8d7982d257dcb58c657de36102001133728","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-11 01:26:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/900118-3/Lenovo-iSCSI/ : SUCCESS in 2:27:55\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/900118-3/Lenovo-FC/ : SUCCESS in 1:54:15","accounts_in_message":[],"_revision_number":3},{"id":"4f96cfed9659d7148dc53f4c753ae07ebf5a801a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-11 05:31:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900118-3/HPMSA-iSCSI/ : SUCCESS in 1:55:20\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900118-3/HPMSA-FC/ : SUCCESS in 1:35:27","accounts_in_message":[],"_revision_number":3},{"id":"436513bedc4d8ab84de290d8b59cc0478a7c09f5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-11 15:39:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/3/20231111211621/NEC-V-FC : SUCCESS in 3h 12m 40s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/3/20231111211621/NEC-V-ISCSI : SUCCESS in 3h 23m 19s","accounts_in_message":[],"_revision_number":3},{"id":"b05c61040dccd48e02504734b0bfcc0744d0455f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-12 16:19:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/3/20231112171013/NEC-FC : SUCCESS in 4h 12m 15s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/3/20231112171013/NEC-ISCSI : SUCCESS in 8h 09m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e4ac27247f45509fa9a4f65fbdf19f9dfca75482","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-15 06:59:23.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ccfd783e7c92437e7e448e955de5a519f98125dc","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-15 07:08:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4828fe85c30fe294f2707ee633e634b0e1ccec83","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-15 22:00:08.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/3/fc : SUCCESS in 0h 33m 34s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/3/iscsi : SUCCESS in 0h 52m 21s","accounts_in_message":[],"_revision_number":3},{"id":"76f5657b2502cf76c9ae575d8bb05739c5c3095c","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-16 03:45:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09058b73b073cf86d3c8ca67c61625243d8e60dc","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-22 08:00:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d2a55247f649d8e46520b8ba957954f0e6bf625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-24 09:48:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"92f9e0c01cb7150fdc79dd33a711b7dd7c6988cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-24 09:48:58.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16179/","accounts_in_message":[],"_revision_number":4},{"id":"f288327d8d357510c042d60e22fc2656f79f0b14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-24 09:49:00.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16179/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"3e7ae7d40c79c735f5f17dd0a3334acfd70a4ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 10:12:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8239ce05f58846359a3a4d775526d34b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/381a36873cc6496cafce633a26c01d03 : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a0ab86bb2b646908756d0ee171cc138 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02dabc4b5e9b48b8b2dfff2ff34c693b : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d53a5523f5b4d0461e4f467c1cd67f39564057b1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-24 11:14:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/19454acf16594bfc8c48d17ef0650969\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f1fdd4dcb678408b8fb6f577f2aa8b52 : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":4},{"id":"b5c2f6c060b2f43c237b2198e12ad99cf2f06b11","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-24 11:14:21.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_PowerFlex/4342b9f/DellEMC_PowerFlex/914 : SUCCESS in 1h 25m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"fd70abf5be1fc825b4b31e3915b26d03ef52c26f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-24 11:29:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/900118/4/ : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":4},{"id":"3d11376fa646c8b70bcccd68cf9ff63a864df88f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-24 11:42:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/dfa9cd3/ : SUCCESS in 1h 47m 29s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/5033aca/ : SUCCESS in 1h 43m 20s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/aed5f45/ : SUCCESS in 1h 49m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/91fce56/ : SUCCESS in 1h 52m 51s","accounts_in_message":[],"_revision_number":4},{"id":"93b80b93c1519614c5a471957ba6fedb00d8181a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-24 11:57:30.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-4/FC/ : FAILURE in 2h 8m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-4/iSCSI/ : FAILURE in 1h 27m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"fe77be819d650ff6af1b66be732ac41f3d9d0d4a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-24 12:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/4/fc : SUCCESS in 0h 29m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/4/iscsi : SUCCESS in 0h 43m 29s","accounts_in_message":[],"_revision_number":4},{"id":"f889ed530850c6de1fe5f5aa2b660c868960768b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-24 12:07:35.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_Unity_FC/fe1c327/DellEMC_Unity_FC/985 : SUCCESS in 2h 12m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_Unity_iSCSI/e10c901/DellEMC_Unity_iSCSI/1000 : SUCCESS in 2h 18m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"17162e864bbfe56c85cf436cfea6019ec1b06eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 12:13:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/818d136095c04ca4a0e47b6f8a948f49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47efd3b6522a4ccd8fefbf494c2f153d : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/930b194c233f40c780912f9df6ed3fb2 : SUCCESS in 10m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6e60f504d834bb8871b7143ee578cbd : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110d4a742f8f4c9fa2531eac39b09d5e : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5adad78407e5400693b1d2ed007429e0 : SUCCESS in 46m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/273417a110494b3fa7354730b9b81b56 : SUCCESS in 52m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63bc3328cf19401685b87d1120244e98 : SUCCESS in 1h 30m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f488e9b5f1a141a694ec2e472d609f64 : SUCCESS in 59m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cace8a6c0e34423a088c687f566dee4 : SUCCESS in 9m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3df8443111a8486986432444916943cc : SUCCESS in 17m 45s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/19f4cddffc56470fac066b6882c3f19c : SUCCESS in 13m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/020f3398fedc42edb40707fb01712fc4 : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e7ec7eaba18343149fe4c2fd1a3896ad : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42d79b39cc5346c29e77d0a7023b30a2 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9430e39c84c949409df634c395b1628a : SUCCESS in 7m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bfd5382e3d414d858776be6d36c8efb6 : SUCCESS in 53m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be4bace2b9854785a48be76acbf19e27 : SUCCESS in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f46c744881604ddd84bd060f6e51ce39 : FAILURE in 2h 13m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b964fbc0f504597849657abbba2b731 : SUCCESS in 2h 16m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc0f89a8a77c491cb2bcd4ed484e6b6a : SUCCESS in 2h 06m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9d8af571b94d473ab6d5906cb4e2ef97 : SUCCESS in 2h 17m 03s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c2ddd840796466c9a90823c52d9eecc : SUCCESS in 1h 18m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/feb9dd93a0c94b0092133d95e5b61dcf : FAILURE in 1h 58m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d1556aa224804c289b3187915eabbf5f : SUCCESS in 32m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2da06d26f3c345d88a445a2aba3f0f49 : SUCCESS in 1h 14m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ebe8b5d9a50d4eb4b0910ecade3b0a37 : SUCCESS in 1h 16m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/793e7dce2d024e9bafdef65a3e234561 : SUCCESS in 1h 14m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/185ec7cd4cf44a6680e3921cb8a72570 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f278e4362c7110fbadcacc519ea169e3f6221e26","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-24 12:56:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/4/20231124184849/NEC-V-ISCSI : SUCCESS in 2h 49m 39s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/4/20231124184849/NEC-V-FC : SUCCESS in 3h 07m 11s","accounts_in_message":[],"_revision_number":4},{"id":"19fa88d5b10c4d0a6816b7ccabd16a5b5a77dda1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-24 13:30:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d826ef4936bf4c87a411e3919b748ded : FAILURE in 1h 12m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0ef94c8315bc48429f266ae86555c321 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":4},{"id":"109c7279d9d33ac066441bc207e8496fa33f7c01","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-24 14:16:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6baded85fae1c69bc3378ea00412e23b86207a00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 14:47:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b219c1d5904bd6a6998ccad09e31ad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d8e54aacf3c4b77a6281a657eebbfc2 : FAILURE in 28m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c2a86590ba54239a82cd3dc5ff57da2 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1514130be3fe4e44ba6d94fd793958a7 : SUCCESS in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d13165567363222450c0a45809016577bd8c2a8f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-24 15:41:42.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d26799e1e3af4a38823e9f728754902e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5697763160534a1ab434cbe5408b7ff8 : FAILURE in 1h 22m 37s","accounts_in_message":[],"_revision_number":4},{"id":"82ca68039ee20da446b50e5af1a2e9ea99bc60d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-24 16:30:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/a838759/ : SUCCESS in 1h 49m 39s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/f56e8cc/ : FAILURE in 2h 12m 58s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/c0d265a/ : SUCCESS in 1h 45m 20s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/96ea185/ : SUCCESS in 1h 43m 19s","accounts_in_message":[],"_revision_number":4},{"id":"f729424b30d391b8a9ce78c15d5449260c7d50f8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-24 16:34:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b20b72ac6e444b73af9e4c5fcf699a36 : SUCCESS in 1h 09m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ab962ec5479146ecbbfc7e7167825847 : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":4},{"id":"f3096fb1e3f0daae861559baecd30e855b40c5f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 16:51:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38d57e3711724a2cb5676bd4cc11a6f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/285e21193a7541fc824060f818511dfe : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd74ceb49204b6099c2196e5d2836ac : SUCCESS in 10m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/babbb03727c84f32831823a78f28d244 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c6a954a2d24fd6908d42c4d24f6a03 : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a8fa5eba09b64ad5a316db229427819c : SUCCESS in 1h 14m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5609014533d742f2a0e8a49b1a1fc0b9 : SUCCESS in 42m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93aee28a6993441da958b6c3d86b56c1 : TIMED_OUT in 2h 05m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d32242b18c0b413e84ed9ba89da63c6f : SUCCESS in 50m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8a747b509ab49a5bf588de218444fc8 : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/32751d2700154153817630e018c16c1c : SUCCESS in 15m 56s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8966b7f592604f5987aed5bfcc096f17 : SUCCESS in 13m 49s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0d2dad7f78d44cbabc8babbb657979d : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8bba82cccaa44e1a648e1ae4182c25c : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6e0d8ec8a6d4309b12b170c5d75c2ca : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f83028dd884451abc8f8189985aaf63 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ca747f318fe4b3d8919adc62e086aa0 : SUCCESS in 53m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f44ed89cc084d20a85a9cabf3a5a9a4 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a878ef2d07ff40509e77b473fd44a2f8 : SUCCESS in 1h 51m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a01e4cbac3f4bd0afcafd79a877a5b6 : FAILURE in 2h 27m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9771324b9f2041c78ae96c93ea2386e6 : SUCCESS in 1h 57m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/72a6ad917c8b43de89471fa6449d9e6b : SUCCESS in 1h 43m 16s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/508f7a44bd4f463190ccb75452521ffc : SUCCESS in 1h 14m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b088d7351e540e883e20918d6933c10 : SUCCESS in 1h 45m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ea09f11b207e4e5b9864007b44edb5d4 : SUCCESS in 46m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f3759f48ecd54e3ea5268e2cc8a41ed1 : SUCCESS in 1h 44m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ff620aca9cdb424187106f6252aa527e : SUCCESS in 1h 50m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af940b6b86214a389eb929d4b9a16854 : SUCCESS in 1h 18m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/838bd6030a3344709895d42ba95c3dbf : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":4},{"id":"107c40d10e28383e00579519b02b406387af3d36","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-25 10:42:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a77a241fe3c629185c22a60cd9731f671968bd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 11:10:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e158233c9814549839e195d72f4b983\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/320a22d5aa6548028c1d197f93959d6d : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1a48bd079784e6f820682c40f836399 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1be2019899e549e38f0c3ca9efd2d7ae : SUCCESS in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb462294b0ea9e22df5132181cbcfcedde486aa1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-25 12:05:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/52fc7eb4454b42b096717b9239c8ec3f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b09d582705d949008307d7681803271d : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a558364f89018b2e673c10166ef3c5050c9e79ab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-25 12:52:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/355132f/ : SUCCESS in 2h 07m 03s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/0049d44/ : SUCCESS in 1h 45m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/9c044e9/ : SUCCESS in 1h 48m 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/900118/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/df2e2fa/ : SUCCESS in 1h 47m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b6a08040405b05baf0ad9be8a20de1b9d5b526d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-25 12:56:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2b6b318a97b54ab4b4bfa2784d3c9d22 : SUCCESS in 1h 11m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4815e5661bae4fa0989d1638bbf88292 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":4},{"id":"772be353ce63cd1f8d0bbc13220f779bba226ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-25 12:57:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa12db42a0bd410f90f4632319de70dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bad766669f94c22b1d90452375e69af : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d439ad15bb8849d1b24d09faafbdfa1e : SUCCESS in 9m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc986ffe19c944d5a1cb052b5a99667e : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a87cf75d0e4dda90161aefe5a0211c : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/8be5fa89fcbf4ed9aec3fe23c590902a : SUCCESS in 46m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9599d2e050624e4098c5fad994c3d0ac : SUCCESS in 40m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f50856e81cff4c748eba381eabdefdda : SUCCESS in 1h 01m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f18eb79004a84bafae38e015be5481aa : SUCCESS in 42m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af1068f21e2545938a16a845b1ac3944 : SUCCESS in 5m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/397fb1bb1aa24d149711671d81ad55c0 : SUCCESS in 13m 12s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cfcc71708fe54bb2988ee403a8c12a95 : SUCCESS in 11m 05s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6aa9d686dcd24d91b76bd6f6c04dc82b : SUCCESS in 4m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c40c97c3df7f4a3a8b119ce41a9c18d5 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aed7fb26c24a4c21a406e6593c95cd59 : SUCCESS in 6m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8eedf84c2f454bcf8222e4d2a3016fe1 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fcc0eaa26eb84ce59facbeea358f4d76 : SUCCESS in 56m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecd491cec45549b090aeecd3000e7a6a : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/46ad78ac80044b14b14e739b226ef825 : SUCCESS in 1h 54m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/380987a1a6cb48f38d4bb9b3cc0e8735 : FAILURE in 2h 08m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08377647837e47dcbee5a2279b4d8c2b : SUCCESS in 2h 00m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1b589f83e75740e1911feeb7f8182f0a : SUCCESS in 2h 08m 45s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/66e4e594ad184ddc86cc3007666a329a : SUCCESS in 1h 22m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/32484a328f61427eb5c0c0fdc6c91670 : SUCCESS in 1h 28m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f6224e2e75794a368b1400f84664dcab : SUCCESS in 46m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/48714b69c007432589bb52b076f852ce : SUCCESS in 1h 36m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5188f0adbcd440c8b2ce6e6f4a27f6ae : SUCCESS in 1h 36m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94d691c3528e4d56a15a1d0f6da1eb20 : SUCCESS in 1h 05m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3815083c3c904c0f803cb41550214ab8 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5c395dff24a371f4f5cbe767b1dbf65e8f8dc292","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-25 13:02:32.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"39fca15c69ee3a698603bffc8bab1fa9358d55cd","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-25 13:15:32.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"412511bc8eefa9b20e21ebc9b31df340159b0e2d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-25 13:18:53.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"b93c13940044fe443aced93da700f09bead64928","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-25 14:18:09.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_PowerFlex/cca75c9/DellEMC_PowerFlex/916 : FAILURE in 1h 20m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"d7c0661f3028b54413cd41c3e5e5753b6c332f5a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-25 14:37:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/4/upstream-check/cinder-tempest-plugin-ontap-fcp/e8c69ad/ : SUCCESS in 1h 08m 50s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/0e82118/ : SUCCESS in 43m 40s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/4/upstream-check/cinder-tempest-plugin-ontap-nfs/120fc6a/ : SUCCESS in 1h 25m 27s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/41dccc4/ : SUCCESS in 1h 22m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3aadfe6175827a7234a1a587a0ffec0fa59de20c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-25 14:38:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_PowerStore_iSCSI/46b08d6/DellEMC_PowerStore_iSCSI/416 : SUCCESS in 1h 05m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_PowerStore_FC/a388cdc/DellEMC_PowerStore_FC/419 : SUCCESS in 1h 39m 12s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_PowerStore_NFS/57ee7ef/DellEMC_PowerStore_NFS/425 : SUCCESS in 1h 00m 00s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_PowerStore_NVMeTCP/910c65b/DellEMC_PowerStore_NVMeTCP/408 : SUCCESS in 1h 40m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"2747ac1bfbe4ecc0ab5723d8bdacabbe0fd8adad","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-25 15:19:34.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_Unity_FC/b877acc/DellEMC_Unity_FC/987 : SUCCESS in 2h 14m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/4/check/DellEMC_Unity_iSCSI/e82dcd3/DellEMC_Unity_iSCSI/1002 : SUCCESS in 2h 21m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"cd41312c1fc5db6f1d1e3a08a6400b79a1445b24","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-25 20:15:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/4/20231126011256/NEC-FC : SUCCESS in 3h 58m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/4/20231126011256/NEC-ISCSI : SUCCESS in 4h 02m 42s","accounts_in_message":[],"_revision_number":4},{"id":"a3c09fbd3c103d2664867cc0e28c616f4ffe712d","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-27 10:56:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ab84f3956d3ce2b4e156883f04fdf60f4235e19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-30 10:45:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4322d32310be7beac99befa200966875e72150dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-30 10:45:49.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16216/","accounts_in_message":[],"_revision_number":5},{"id":"2db2da35d3aaa420d7062f4a4e1f23e0660ebbd2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-30 10:45:52.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16216/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"a0ec0e199b35a3d2fcdeb2932e774a6a698e9724","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-30 10:45:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6f893b163815c9ff2ce9f27de02d6003517d0c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-30 10:48:28.000000000","message":"Topic bp/dell-powermax-restapi-101 removed","accounts_in_message":[],"_revision_number":5},{"id":"60c190b3a6b788238834c05602e49d126eceb28a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-11-30 10:51:36.000000000","message":"Topic set to dell-powermax-restapi-101","accounts_in_message":[],"_revision_number":5},{"id":"65563987f1e491f8e6e103aa38e950a59e536048","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:19:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38ea52574e0c479786891be9b0e5cd23\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11b9e7c70b5f4fa499265e27dad84db2 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/038ef4821c0d4580bfb5abf77e2265a6 : SUCCESS in 27m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd1e811680004da692eeedb16f2169f9 : SUCCESS in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e1d445b6f20e65a8b4058e1690f9a34168df0b33","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-11-30 11:43:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"02e9410c081f9c485b08159b242a56ac8dd20129","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-30 12:14:13.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0a0e34f4f1a94704a502d01625b65ab5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/65416dc5bf214351b6995fb195e60337 : FAILURE in 1h 26m 22s","accounts_in_message":[],"_revision_number":5},{"id":"f707ba7b023952ec0c17dacbc3f79a2e38c7516e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-30 12:36:39.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-5/FC/ : UNSTABLE in 2m 46s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-5/iSCSI/ : FAILURE in 1h 24m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"75d53051e42777709b78ca52aed84520ccbef0a7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 12:40:19.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerFlex/4d6dc63/DellEMC_PowerFlex/967 : SUCCESS in 1h 10m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"a9b970ed4294092465092799c7bb088f8d05e051","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-30 12:42:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fe57b0ea67524d6d9793c3dd7525d533 : SUCCESS in 1h 19m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9acb6748a310420bbcf1b5c20c130a8c : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":5},{"id":"10d0c9fa51029373cf29a716cad9b2aa8909b468","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-30 12:46:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/900118/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/dd8dd7a/ : SUCCESS in 1h 47m 25s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/900118/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/586b40d/ : SUCCESS in 1h 58m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/900118/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/ed52b56/ : SUCCESS in 1h 56m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/900118/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/9610cc5/ : SUCCESS in 1h 47m 13s","accounts_in_message":[],"_revision_number":5},{"id":"1d885bfd16de6e6603309be4ea98813e9fde62ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-30 12:59:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/190/.","accounts_in_message":[],"_revision_number":5},{"id":"3689397a9956ed63cd2e36d123e0c3075f9d110e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-30 13:13:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/900118/5/ : FAILURE in 1h 06m 03s","accounts_in_message":[],"_revision_number":5},{"id":"7d14f18416e4ce005d2c4b68e5c78a58f6304b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 13:42:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a833b53a3b4f2186137366c7a84577\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d703e4512d14e82bb9c06f4ecbd0a27 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74ae6666a393412dbb4b703ccf832144 : SUCCESS in 13m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a0ab3472acf4f26aaa6d08dd6117809 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668c0e02ca3d4c51ad0efabde6cc6c04 : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/92d2b4530ef5454f8c6390f6136f13cd : SUCCESS in 57m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2f8166d8fe64408b1623793b0bb09ac : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c24b0601ee0e4058bc79bc3d6f3d4f7b : SUCCESS in 1h 16m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66305686cb1447a5a4f7043873036cd5 : SUCCESS in 1h 02m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d85b13ed119463b9a715183d503e86d : SUCCESS in 8m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1ce83d07e8424eedb3c9a521fff4fb65 : SUCCESS in 13m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/535a5c9be6e84c5f99f62c8c24189776 : SUCCESS in 13m 50s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b56774204ba347f4b95390a985e1834f : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f42722edc2be4292a93322cab6606e6a : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51a027d4a93e4d20b1c561c37b88434b : SUCCESS in 6m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4556ad5dfd784ad9af9573aee76f863b : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c79f7f8ea38f41e58bbf74eb52314d07 : SUCCESS in 51m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26645875d28b49b2bbf53b305ee327f3 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb86979b56cb44a28f81f0be04717169 : SUCCESS in 2h 14m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d255be4bc0c4077a4cbd01b3b5acb68 : SUCCESS in 2h 23m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2516a6022f3f49778cab4b211d928b84 : SUCCESS in 2h 19m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f83dc2c3cf834e449aa09ef1d7169665 : SUCCESS in 2h 50m 43s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ad6ebca5c474a208475e4dfa8ad5f26 : SUCCESS in 1h 23m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f4a93c18f0f4c88ab8bef570afa99e3 : FAILURE in 1h 21m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/29ab8f9d82e145ac8e033aac00312582 : SUCCESS in 29m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2930c9f7c4794aefaeca7d11ffea7dc9 : SUCCESS in 1h 19m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/51fe13378ab846bd9b173537cc4d60ae : POST_FAILURE in 2h 01m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6307e98fd0b4c9589d3dbe60f683068 : SUCCESS in 1h 30m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ca17bfe7ed34056ba899fc7f209c730 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":5},{"id":"427515da4cc1c6954443c6d62c14c16c31934e57","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 14:55:30.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerFlex/b9580ed/DellEMC_PowerFlex/969 : SUCCESS in 1h 05m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"f64ccfb925c786657ca8424ed19cf4aa30735796","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-30 15:21:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/184/.","accounts_in_message":[],"_revision_number":5},{"id":"de375793a5848af5cd99823937112496b721baed","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-30 15:31:37.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerStore_iSCSI/7a7e21b/DellEMC_PowerStore_iSCSI/437 : SUCCESS in 1h 47m 31s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerStore_FC/93c8f1f/DellEMC_PowerStore_FC/440 : SUCCESS in 1h 23m 42s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerStore_NFS/cdd55c6/DellEMC_PowerStore_NFS/446 : SUCCESS in 59m 44s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerStore_NVMeTCP/db42ae2/DellEMC_PowerStore_NVMeTCP/430 : SUCCESS in 1h 35m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"b3d1322ab04b7b76153efec6bbddd8a4371241ee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-30 15:41:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/5/upstream-check/cinder-tempest-plugin-ontap-fcp/ca003a9/ : FAILURE in 1h 57m 34s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/cff2080/ : SUCCESS in 45m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/5/upstream-check/cinder-tempest-plugin-ontap-nfs/f1d8799/ : SUCCESS in 41m 20s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/89a3e71/ : SUCCESS in 1h 15m 31s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"09d3adf834894b6e5d3aaa9e9643569e8e88afc3","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-12-01 10:28:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/900118-5/PVME-iSCSI/ : FAILURE in 44:48.08\n- PVME-FC https://os-logs.seagate.com/dell-ci/900118-5/PVME-FC/ : SUCCESS in 1:54:15","accounts_in_message":[],"_revision_number":5},{"id":"cfe1319936b338717c765e54f320c2e004e6ed46","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-12-01 13:45:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/900118-5/Lenovo-iSCSI/ : FAILURE in 45:08.34\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/900118-5/Lenovo-FC/ : SUCCESS in 1:56:29","accounts_in_message":[],"_revision_number":5},{"id":"4dc78e9c78a1c7938cfd2a11e3ab88fa1145dda5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-01 14:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/5/fc : SUCCESS in 1h 07m 38s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/5/iscsi : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":5},{"id":"0b79962e010d994a986426e2b1175769d5383b78","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-01 17:41:43.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_Unity_FC/61bc860/DellEMC_Unity_FC/1028 : SUCCESS in 2h 09m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_Unity_iSCSI/ba005f8/DellEMC_Unity_iSCSI/1043 : FAILURE in 3h 08m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e02aa1862315cfc31def92a7f0154d9e9942d197","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-12-01 17:45:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900118-5/HPMSA-iSCSI/ : SUCCESS in 1:50:17\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900118-5/HPMSA-FC/ : SUCCESS in 1:35:00","accounts_in_message":[],"_revision_number":5},{"id":"0f30a44aa5fa146bee6972c7d95e2cdb5ec0e2c2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-02 12:50:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/5/20231202183922/NEC-V-ISCSI : SUCCESS in 3h 03m 33s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/5/20231202183922/NEC-V-FC : SUCCESS in 3h 11m 03s","accounts_in_message":[],"_revision_number":5},{"id":"95ad546b7a96161f88a69a614606351b9ac00fe7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-03 17:00:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/5/20231203220519/NEC-FC : SUCCESS in 3h 48m 30s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/5/20231203220519/NEC-ISCSI : SUCCESS in 3h 54m 47s","accounts_in_message":[],"_revision_number":5},{"id":"00f1f65b473d4f88384b9bab3ee717f781180884","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-12-18 11:04:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"55ddb2f9153009e4ad5149a082f26cd7a61d5f98","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-12-18 11:07:19.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b14abab0ece8cd443484f326a2cb0c2b8623e3e2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-01-17 20:49:15.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9ebcc9552cac67ebc3690bb40450902cf136219d","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-01-18 12:43:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8fd6c6f7fb027626e697388574aa93a77fe6a4af","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-01-24 01:35:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"302a80f075b8009b7a3e36ffe8a076c0a5bcc8f1","author":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"date":"2024-01-24 03:46:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37298cc13c85fb3006c6a6952f0d2e9d6d01ed23","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-01-24 08:01:06.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerMAX_FC/249a80f/DellEMC_PowerMAX_FC/312 : SUCCESS in 4h 35m 12s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/5/check/DellEMC_PowerMAX_iSCSI/6da076d/DellEMC_PowerMAX_iSCSI/334 : SUCCESS in 5h 32m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"1a1ff331659f48ad895fd6850684033b831e3c99","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-01-24 08:15:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4beaf394ffa06598a7d27784b9119ec4fce381f7","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-01-24 21:34:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"62a06ce9a5a9a0f9cd17646c1c5368e641fb1780","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-01-29 08:02:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1a16a36eb9a4436a6fa4131ed287828c7ce752d","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-01-29 08:03:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1d2eb19c13230346c5fedd5a9ac556a7c4799d4a","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-01-29 11:02:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"897829a4ee5b048babea09702b4504d1b2d9905f","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-01-29 11:05:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fb77d75c99b7d4302ce68e7b15041af83c220f49","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-30 23:32:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"890460a211ee83f7ec77e37fc70c32476915fc9b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-01-30 23:41:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"857ca9663ce7e433b7484e1e4e90f3f1faa6813c","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-01-31 05:13:18.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bea9242acc27828d5caffc77323321c5dd0a67b5","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-01-31 06:05:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b8f12bbceffc2de1f50ba01edd037e389f24ddc2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-31 13:36:11.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7ba99802d554b813fc3f251ec7621a58ac1f3d45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 08:44:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"956491bb3be8021b93cc54de3c0603f4509dc728","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 08:44:48.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16439/","accounts_in_message":[],"_revision_number":6},{"id":"c34674ae7fc654362cd20d596ee713c484e06779","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 08:44:50.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16439/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"40fd76f48ffd145dd4057c6e322ff7d7458aa332","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 08:48:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"148d5685be505fb82c3e084da6fe2b653e82ef5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:56:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c33c90a37e48169358c9098acb5de8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8bcf2d18f0743048653ee8023c3bd98 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de3f6cec33e54688a74a2248f1813cae : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/997cd5e4af864555ab4a12c7fc2fc60f : SUCCESS in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2f93bcfb5ab603e1de176f2ea1b14bd789a8b77a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-01 09:05:57.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/6/check/DellEMC_Unity_FC/6c59de7/DellEMC_Unity_FC/1179 : FAILURE in 9m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/6/check/DellEMC_Unity_iSCSI/6ad8d4c/DellEMC_Unity_iSCSI/1195 : FAILURE in 15m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"b3f0d3e98c3b135a821b1776628a1eaf8dfd6687","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 09:12:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"493bc62b8c18b3fec6076ac3d0f401ec6a153bb4","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 09:13:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"386a329ce8abaf1826c5138b52f531a2f4f90944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:06:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8e9d9680c8143d6ba8d4a5f1ff2bdb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095cf9e0f2534d2bb260442ab4c40fd5 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60d367baa98a484bb1093e33072048fb : SUCCESS in 8m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4174656ade0c430f9019e27e6e49873c : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce3eb759d8e4f79879d700e821ba640 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/f6fba6de84014144ae5dde44e251b2b4 : SUCCESS in 1h 03m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6c502e1a98d64413bc67a5a0654ff7cc : SUCCESS in 1h 00m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a79cd9bd7c4e49cbad8ccfe559778808 : FAILURE in 52m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/302759356cab4fea924e052cb7878900 : SUCCESS in 36m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f2e6270cf1d4ab8a9d8e5b7d7da6aea : SUCCESS in 5m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/24542034fb8248149587014e3f2dfdf4 : SUCCESS in 16m 25s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/80a0dddabb824d35b390e83f01142f84 : SUCCESS in 13m 01s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf1ef00b2052496e961a5184f328ae24 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/822448a221dc469dbb54892d52d07f40 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6da19c9203a04174978a0e98905dfe60 : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f5e4142b2a14fe087bfd5ddcb4e8f54 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e4a29dfdb61e4aac96ff811f291598f5 : SUCCESS in 48m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd9fb31cc8524207b78ef0460d2af7ca : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/184d017d3f494a7f9265cded6164e771 : FAILURE in 2h 00m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa6fb702b0a649b6878096d8000f9c24 : SUCCESS in 2h 11m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0166b8ef467a43caab588f9465553b08 : SUCCESS in 1h 53m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f0b4126c228844e0a606e62fe5ad1a31 : SUCCESS in 2h 13m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a6b7fc7cd684afea023156e500bbf39 : SUCCESS in 1h 08m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2630bc8437464467b62366297955e490 : SUCCESS in 1h 29m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b90795f611b7402598487722eca7e4af : SUCCESS in 53m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6725474da1d1429da15af18e2402b014 : SUCCESS in 1h 21m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43c99e4f4e0b41f09b27aa09238c69a2 : SUCCESS in 1h 23m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f53263ca5034384a127805ce5dce4e1 : SUCCESS in 1h 49m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa6835dcdced42bcbbbfc8059aa8dac2 : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":6},{"id":"8d6c9e6ce4c721f93f222fd265e32e32f0024935","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-01 11:42:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/900118/6/check/fujitsu-eternus-dx-iscsi/2c0bcc3 : FAILURE in 2h 31m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/900118/6/check/fujitsu-eternus-dx-fc/57a2643 : FAILURE in 2h 56m 42s","accounts_in_message":[],"_revision_number":6},{"id":"28b8627a7b7bfda1c4369a949f3ea2350bf9038e","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 12:02:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c867958be56487f00841eb021d1bdd903c6f8308","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 12:11:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce04049eddfd4c76b8ba07212a5f1500\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42f6c83efa3e4494b7a1703fe1a10eef : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a871265ba0f9417aaedf1bbdfc70b8d5 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/086901fb517c478f922e5be7179b9e25 : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d7580fc4a69b4394a31dda66a9db631739402d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 14:23:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2bccf325ba446a08613b8dc680ad8fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad6cdc3c80140e097f5ca2609515835 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63cd8d57061b4e048c0037c24b4b74b3 : SUCCESS in 11m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9374290d3d054385982696f07558e938 : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9daa8b2b2bc745aca4e91379743f2fbd : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7f76ff5e0bbb42e18008dd174b3901d4 : SUCCESS in 1h 05m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/91dcbd46b724480e80cac8da1f0f5fcf : SUCCESS in 1h 05m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e882497bfec840108f0363aea47eff10 : SUCCESS in 1h 16m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3ac94caa73c47479a72647e1939d165 : SUCCESS in 26m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfe6d4d8d10246c59c6e6930cccf6b91 : SUCCESS in 7m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0049a8c9f8774286ab0b2301b19894df : SUCCESS in 19m 04s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/83f2e20a995746bcbbfeeac9af242791 : SUCCESS in 9m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/562502cffa1f48c593ba0c43fe2c8f91 : SUCCESS in 4m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/47a4ca2846a84a278caab414e083a760 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b706aee35fb949cbab3e6a9635e5ad18 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/841c20a125a043a3bbc8120cc21f8ccf : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a4862f508a04442bf7b944086d1632b : SUCCESS in 46m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f018000b89244ce9b27a837ac70ef090 : SUCCESS in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/641e5fa765ba4c7496af02174286719f : SUCCESS in 1h 15m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f6ab83817bf14c919d27b9bf1b138206 : FAILURE in 2h 14m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cfe27ed405a44d38bfb29e0098083bf9 : SUCCESS in 1h 54m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1ae0f78407f34f0d80da9582a6f8294a : SUCCESS in 2h 13m 03s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b85f1d73fbcc4655903a1717f69624a6 : SUCCESS in 1h 22m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61617cf84ac242abad9fe1ec79fffdd2 : SUCCESS in 1h 43m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fd0523da77314bbf9b84843c76b87fec : SUCCESS in 48m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/175e9d67cc19422fab2cb058016c200c : SUCCESS in 1h 28m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0b3b6bec173d401a9ae0ad82a09122ef : POST_FAILURE in 1h 51m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d693221a72eb47fe9ed898bdc2ebcea0 : SUCCESS in 2h 06m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f762d16aa7d147f295531c440ccad121 : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":6},{"id":"6b6363dd8c4a7835272665a05a0ae1ee04166216","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-01 15:00:37.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"764faa101bca4d8b96617480efa5cf9c00f32707","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-01 15:02:05.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77cf30a0479f2107c9e331a33ddb24497d54fe11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 16:10:06.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2799af39a6ef34788b214c1e3c9b1f61078a93c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 16:10:16.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16448/","accounts_in_message":[],"_revision_number":7},{"id":"dd1e4ea5e32ac534c7acd444e0005fe28ebf83ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 16:10:19.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16448/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"06996582aba1628628b4050dd70cf4630bf1f207","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-01 16:11:49.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3bc36e7370769bd4249b2bf8abb797bb3e6a120d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 16:22:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b4cc5de0ed44e9d98b57be05ff37d92\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15684817f78742309efad21c10384a3d : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72055695025a418d817064fe1c400088 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c0d023305db84658a4e3223321267892 : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e5dd4f4511eb3a0315ba9f1c4b04bf37f01da63a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 18:35:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88af8a09d7e3452799de1bbca8057fa7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b359761e6f6a40e1b280ff3e0427ad00 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b8d2d9050d421faf780b8c724479a2 : SUCCESS in 11m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5891a98c32dd4bd499e85c9639eb620c : FAILURE in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba7ccd4680d4337abe256b8af65c410 : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/f433f1dc7e6b4145b0bbb322f3b9ab92 : SUCCESS in 1h 12m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/16d9640cd59747e2883ad5b87b0a5e7e : SUCCESS in 1h 02m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7dd8015b00eb451ea45fd3f9154be331 : SUCCESS in 1h 29m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3862e19512e421080060db39b3488e5 : SUCCESS in 25m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04e05ea1910f4597a312de63bd5227b2 : SUCCESS in 5m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d66f035cfaf42d297a4624aaea631db : SUCCESS in 16m 14s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c7deae6973d34df5aa6c2dd18a38d09f : SUCCESS in 13m 41s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51a29c5dfcb14aafa70f67835bed2359 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/116e32732730409cbef9607d1dc97a7c : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2131b9b34274673b3e9f485c17ee754 : SUCCESS in 6m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfdc13e3fc8b4cf0967e45f89590f4f3 : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/34ff72baf2294861bdbb538c5129d5bd : SUCCESS in 53m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/986e371a3e5041d4ac15a39dd90e7be9 : SUCCESS in 4m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a1e7526505b4382b6a189edf7a8a570 : SUCCESS in 1h 43m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8cb872114e594d7fa02508127b14c042 : FAILURE in 2h 14m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc7bc429f371454aa71ebcbebb4905ab : SUCCESS in 1h 49m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1800c2f1c9ee4505bd071f8071696f94 : SUCCESS in 1h 46m 27s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/574b1582af59479fb744f339f10d1176 : SUCCESS in 1h 19m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d0093f1e5454f93a31821b3c86c26d8 : SUCCESS in 1h 30m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/636e4a0ff7154cafa0ac027bda97feeb : SUCCESS in 46m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b1600e4ea61d4864a8c43d8816035d0b : SUCCESS in 1h 24m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5a4210bde2534b7fb5f033dc54cf35ac : SUCCESS in 1h 48m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9baa167343547d6852e2f3e4e0212e9 : SUCCESS in 1h 40m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3161de9d7fc3493c8621fdb07814e465 : FAILURE in 1h 01m 36s","accounts_in_message":[],"_revision_number":7},{"id":"4c2318f77deb208196c2c7a15f7e148b0e66b061","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-01 21:04:38.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-6/FC/ : UNSTABLE in 3m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-6/iSCSI/ : FAILURE in 6h 26m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"9970ef9da5171fc9072d74b9a8f82b5369a993ed","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-01 21:49:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"82abbb5b64c4777fa2f5f07f113fe3ecb5439bad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-01 21:49:50.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"625423c2c10e1af4ed705879ccd1daaff4cf5916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 21:58:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb63b3ef317a430fb6d03ad32d40e3a1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ad6088772bf448981cdc89146b841f7 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dcf40a2d6d7341adb328dfc7b09e019a : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cda26aaf60b946fba013ed35f8ea5673 : SUCCESS in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e8ba121373cc35ed87d83de98da96532471d2854","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-01 22:00:10.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/7/fc : SUCCESS in 0h 57m 23s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/18/900118/7/iscsi : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":7},{"id":"94ffda1cbf56408ed150e861112a3da385e5c33b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-01 23:23:28.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/55f47926a2524234af0928f0e7adf3eb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f8d8b1c90c3a4dc9976f075cb4343fcc : FAILURE in 1h 25m 57s","accounts_in_message":[],"_revision_number":7},{"id":"d81c9a0695e185b2975744f2561b3fd807ca02f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 00:31:55.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c8a9f91d7e74baa9091fc8955294c0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f21e38f2840438491ca55b67927bb81 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603f2a7ec6f84e4f91452c28da4cdd99 : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/447c311f46d44eda89b4c6a10aa37dae : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481e0b883dbf4b4d84e894e301a32d2d : SUCCESS in 16m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/0a49e477ea6141e7bf48f8cb610e73b4 : SUCCESS in 1h 01m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4879cdf7505747dc8263222535d1d6a0 : SUCCESS in 46m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a74df66ca611450ba088574f344ffe54 : FAILURE in 1h 26m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abe01781425f4de7abcd21b55b1972a5 : SUCCESS in 24m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f56e17fbb084449897b62536c35f559 : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca8d9edd1aef463f9062626448840097 : SUCCESS in 15m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e359cd3aad3247aea178e15b12ed4984 : SUCCESS in 9m 17s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/957d93572b5f486aa57ea3a22772b056 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86fd3ac0d86946c09282919bb703d8ec : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d634ccee1cb14885b86b1c4732900e9a : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15b68406f2e14959bd5192ed8e3b3090 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/054a2ecce89b4822a48ca8c3dda99da1 : SUCCESS in 48m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b57f4c181bab4267b13aee0c82fc174b : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/636adc435ca347e0a11bc6354cf4a79c : SUCCESS in 1h 49m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/11b75c3e3b5446609b21d51f2a9dc1a7 : FAILURE in 2h 11m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5564652479ea49099ec492b5df9d275d : SUCCESS in 1h 42m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/23f0035928eb413f8045fae4405eeabd : SUCCESS in 2h 34m 02s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8ae37eb3a04d4a7c9b715a0712aa1c95 : SUCCESS in 1h 13m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbb6d00b477146fd9c8ff5657314f705 : FAILURE in 1h 30m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/870775031eba4ee9bd3bfdebd661648d : SUCCESS in 58m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5030cc24028f4a1b8720d407ab9b61f9 : SUCCESS in 1h 30m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/66aa6bff2255430c81fa7ad03cea8f5c : SUCCESS in 1h 50m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa88a6e90c8541469588d7df54cd42a4 : FAILURE in 1h 43m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d187ce7575d47ddbcf9e9c7c4cb57d7 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":7},{"id":"36eb2fc0df0be710f58a6d6b276655fa8d109d12","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-02 00:37:32.000000000","message":"Patch Set 7:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/900118/7/check/fujitsu-eternus-dx-iscsi/6607a40 : SUCCESS in 2h 32m 19s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/900118/7/check/fujitsu-eternus-dx-fc/6fef61d : SUCCESS in 2h 35m 27s","accounts_in_message":[],"_revision_number":7},{"id":"0e6e6ac5845fdabcfd162402ea487e989c067c89","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-02-02 00:56:38.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c68f9ba98a99ce78e7ce57109ffacb21d49ca130","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-02 01:30:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/900118 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":7},{"id":"b086a57e65243d58a9e954ec1e003a6636d56fb0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-02 03:27:33.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-7/FC/ : UNSTABLE in 51m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-7/iSCSI/ : FAILURE in 7h 12m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"d8ec67ed7265996f73b5aeb049a13de4d1f7b636","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-02 04:35:10.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_Unity_FC/e13f554/DellEMC_Unity_FC/1196 : SUCCESS in 1h 08m 56s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_Unity_iSCSI/c1f53b9/DellEMC_Unity_iSCSI/1213 : FAILURE in 15m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"259d403e9169c842608ea61667d8c38c7c2453d0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-02 04:40:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/14780e3a53264d68b05d6dea2ba9ceda : SUCCESS in 1h 14m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/20dab4598be7438e8f8f77bb4173796a : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":7},{"id":"1ee4a8e80d3b1dd4856e0132239b4b29126ea508","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-02 08:07:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"57c209de65046246a4cdf7480b3fbe6369c8f932","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-02 09:29:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9eb1954c53090f562007cfe3e624e5bc94b47781","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 09:50:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19ead777bbde477fab3190dd1988f88e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd7544dcc62f4bac81cbe4fa2c3a2d64 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0cbff08670642bf8f69b034fd738635 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66ed5c2dd652407e87744beb9a5bd14c : SUCCESS in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"487b142d0658abe95af2f6fd41c04cffcf310f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 12:04:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc5710ea4984319bbd253a25481e3c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a1d5c825b44292b351f803cd5d2f0b : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c4141c0bf274d5aaa2ba2512c7c8536 : SUCCESS in 10m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97533c5c64cc4e868cebf16bcf09704d : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e0c3fa89d641138d753f4ebda571c9 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b0bc241c481c41d5b05546e5cc3dbccc : SUCCESS in 50m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/88da2b67fa2d48aab8d1c01e14ce36dd : SUCCESS in 52m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21ecf0ed41ab4a469aef67d2025d8259 : SUCCESS in 1h 09m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a26df6804fe4c96b3799193235a2c04 : SUCCESS in 27m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dea00eb1aab43b1bc0dd36210cc4424 : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6544bef39bcb46cdac3d12727e5fdca6 : SUCCESS in 13m 00s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d48167cf3cd84c28b9f6f0fcb149704f : SUCCESS in 12m 48s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9fcc7582e3514c12bf96ad0f948a04bf : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4248eaa85c75465cb10260f16a0168b5 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4032c7864a5b422fa447e90e565effe0 : SUCCESS in 5m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af4361bbb7914293b37765743260badb : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebf83978268943cb912b45300a825ecb : SUCCESS in 53m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20c11ef663b048c3b7eec30d6b6a5e39 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f450dcb1ce2d488383e72ff2f85034cb : SUCCESS in 1h 55m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a1eb3ef5f87400983085db63a4a5b78 : FAILURE in 2h 20m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4046443a24904d5d9b67b5cb4e3d78a7 : SUCCESS in 1h 33m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5f5dbd2a28404b04b082d0b4f3a041b1 : SUCCESS in 1h 40m 38s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/674740a3240945a28187aa541f76fd96 : SUCCESS in 1h 05m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48cae46774e14d99920634fa2b7d521e : SUCCESS in 1h 33m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a6e343c1ec24cb085e9c66df5a63988 : SUCCESS in 56m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d21e6e886acd4c3a8e478a5a9bda2b45 : SUCCESS in 1h 21m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6607f043e502434fa76ebafc1d546cb8 : SUCCESS in 1h 22m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/99263193e52547918c612a0e136e65e0 : SUCCESS in 1h 26m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a0d46783a664944950eed455e4ffc82 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":7},{"id":"74fc8c2546799159b7e79bc5290f83529e86fef0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-02 13:10:53.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/7/upstream-check/cinder-tempest-plugin-ontap-fcp/8791034/ : SUCCESS in 1h 05m 39s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/499c9f4/ : SUCCESS in 41m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/7/upstream-check/cinder-tempest-plugin-ontap-nfs/2fb65e6/ : SUCCESS in 42m 57s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e557347/ : RETRY_LIMIT in 8m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"e146d7ec459aa157b4db9d5fb2536575ba1f0561","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-02 18:19:12.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b59b4646ef544316adb8f22928a0322e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d12d9029c4be4eff909b67d2e9d1acdc : FAILURE in 1h 21m 53s","accounts_in_message":[],"_revision_number":7},{"id":"5fc6cf1793ccc65a4314c63df15b4b0b5f73cf84","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-02 22:40:12.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_PowerFlex_3.6/142e026/DellEMC_PowerFlex_3.6/1345 : SUCCESS in 1h 28m 30s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_PowerFlex_4.5/ef3a2b9/DellEMC_PowerFlex_4.5/313 : SUCCESS in 1h 55m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"8e40d0a332477a9e59b25a4d95cc4fde15902584","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-02 22:52:40.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"72b2be9084374f142bdafbaf145301f872a72260","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-02-02 23:03:40.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"68f33897b920d82805929883ebce1f00114cd4d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-03 04:04:25.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_Unity_FC/11c3985/DellEMC_Unity_FC/1216 : SUCCESS in 1h 08m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_Unity_iSCSI/3fc4103/DellEMC_Unity_iSCSI/1238 : FAILURE in 14m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"1939aebe15cb3e4b1f312d05ed512b364b302734","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-03 14:50:16.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_PowerStore_iSCSI/21f5824/DellEMC_PowerStore_iSCSI/1021 : SUCCESS in 1h 04m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_PowerStore_FC/d2712cd/DellEMC_PowerStore_FC/1176 : SUCCESS in 1h 31m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_PowerStore_NFS/9496902/DellEMC_PowerStore_NFS/707 : SUCCESS in 1h 06m 32s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/900118/7/check/DellEMC_PowerStore_NVMeTCP/5b86ac1/DellEMC_PowerStore_NVMeTCP/573 : SUCCESS in 2h 14m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"6e0f098d876f70f105618cddabea1f3b19c83c49","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-04 00:05:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/94f2f8ca3b064c7ab6472e6129d204a1 : SUCCESS in 1h 09m 23s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/afa7e15138d84674b91b8724de7f4123 : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":7},{"id":"95357824d89166219032716f05a7c4038f180dc7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-04 08:18:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900118-6/HPMSA-iSCSI/ : SUCCESS in 2:03:19\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900118-6/HPMSA-FC/ : SUCCESS in 1:36:43","accounts_in_message":[],"_revision_number":6},{"id":"2f350419886824d9186a33e8496f177d2fde9992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-04 13:15:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e21823c0e0c34b22d4ad56d96bd18f92bf3bed3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-04 13:15:28.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16472/","accounts_in_message":[],"_revision_number":8},{"id":"97cb9181fadf2f0c787b5cb28b33c5c88bb61153","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-04 13:15:31.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16472/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"28f23ae67e95e9d70a112d05c85469348a19f7a8","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-04 13:22:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3566dea966b5809978de5fb27529591a20cec85c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 13:27:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1295a88bb3af47c4ad074b61a1cf2a9f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c61c3641afbf4b44963e0a2bfe2ecdf5 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0cc6c4af894a4afd9dd6f316bcb67f95 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e669ca2616e451bb338705374e49566 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ef652b94e0b90c1a09616ac7540e7b388ceb1f7e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-04 14:10:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/900118 : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":8},{"id":"4fe527ae4222a8f71cba2a8eb47076ff2f302aae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-04 14:40:34.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7aba532b19f24902a91297f068df2217\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bd672f0723534f61826bdb39b11b4552 : FAILURE in 1h 22m 26s","accounts_in_message":[],"_revision_number":8},{"id":"e17575e997a36b97685e570dd872a3827baf9973","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-04 14:41:13.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_Unity_FC/a038065/DellEMC_Unity_FC/1237 : SUCCESS in 1h 06m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_Unity_iSCSI/e8cd9aa/DellEMC_Unity_iSCSI/1266 : FAILURE in 1h 11m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d71a6c1e176355180f3567d760bd34808f2572cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-04 14:53:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/605/.","accounts_in_message":[],"_revision_number":8},{"id":"dbc3433058a05abfa1e0e4b517c12ac1149489ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-04 15:00:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/594/.","accounts_in_message":[],"_revision_number":8},{"id":"390b2aac8ffc16e6c05d5a5c9a91e270894f998b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 15:38:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd6956d24f5d4057aa8344f780813450\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59de727b415a4b8385baca233faf5c62 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf49febad38e41cfa492076c92c64883 : SUCCESS in 11m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c84d8f2be6d4343994c38d48c9b4e87 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35b42b46e7540ff92be477078210013 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b1218c8a54034ce08eb03342ab45a399 : SUCCESS in 46m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d7d998ef15bb4e67b050d6aa41a1aa32 : SUCCESS in 45m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/921f5326f6b54554920635cdb93ca781 : SUCCESS in 1h 21m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01510de45acd4932bfb3cb0b58d18775 : SUCCESS in 25m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a61abc9b39354d9a9364d9f8fde4dbb4 : SUCCESS in 7m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/31187313bad84d098949519bb0bf7344 : SUCCESS in 16m 31s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b09170725cfd4769aa8af41279537b7f : SUCCESS in 12m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/964b13dc1ecc415a90c50c098bfd09aa : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb2ef1577acf4d44924e6410b698eecb : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/deac716ea9134929ad4ef9e49fe92a4c : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/32f92be6dce84d83835bcd30101be96a : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dee23377c92648e892a6c51e71d5dcb5 : SUCCESS in 32m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e5c1e08d90e4b72ad071ffeafdfb5f4 : SUCCESS in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/64e322d913454a9b9541455f767bff3f : SUCCESS in 1h 40m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/15e98e9f54d84e69acbc4716cf4bb344 : FAILURE in 2h 17m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85e73de4504549aa9b1a2598ed2c8ea2 : SUCCESS in 1h 36m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4b4b453dc18049adafb395e759f24ab9 : SUCCESS in 1h 39m 50s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e697bb5f671b4ecea79461b317952927 : SUCCESS in 1h 08m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5235dd7425914273b808d7b02766583b : SUCCESS in 1h 14m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca92a5b7a4b6464ba9a2c490cf8d1392 : SUCCESS in 36m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/826bd0aeabb7473c837927100a7c14ed : SUCCESS in 1h 20m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ed9053943ec94832a641c6b7faa618c4 : SUCCESS in 1h 22m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b408c6eb078c4a39baa4bb6b7fa801fd : SUCCESS in 1h 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e54f589cbfc4895836f94a1894e294c : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":8},{"id":"9524391be57508ce0eb3f25d72839cd3e1afe9d0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-04 16:49:55.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_Unity_FC/50626d4/DellEMC_Unity_FC/1238 : SUCCESS in 1h 03m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_Unity_iSCSI/dd18af2/DellEMC_Unity_iSCSI/1267 : FAILURE in 1h 11m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"78e100edd40ed3279de76deabf11be3db07a416c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-04 16:51:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/8/upstream-check/cinder-tempest-plugin-ontap-fcp/2ffa97f/ : SUCCESS in 1h 12m 33s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/5c15a3f/ : SUCCESS in 40m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/8/upstream-check/cinder-tempest-plugin-ontap-nfs/f958151/ : SUCCESS in 41m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/900118/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f98f7eb/ : RETRY_LIMIT in 7m 12s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"71feaee2c5fe95c1db3434d12b78e6aab4632d82","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-04 17:08:29.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-8/FC/ : SUCCESS in 2h 48m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-900118-8/iSCSI/ : SUCCESS in 3h 52m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"64a57a0f5de36a7fdca8678cc8ef226098994000","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-05 03:30:43.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_iSCSI/7d8a22b/DellEMC_PowerStore_iSCSI/1036 : SUCCESS in 1h 09m 19s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_FC/a967aee/DellEMC_PowerStore_FC/1191 : SUCCESS in 1h 39m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_NFS/88f3ab9/DellEMC_PowerStore_NFS/722 : SUCCESS in 1h 01m 03s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_NVMeTCP/fbc4880/DellEMC_PowerStore_NVMeTCP/588 : FAILURE in 8m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"bd9f495fb7dfa451b5a3b1a9a7669b7898e48d03","author":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"date":"2024-02-05 13:20:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b9b8e9883b22b0fa03dfc982c80422e0b119a3b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-05 13:31:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/925c18829c504972bd12c70c599c5933 : SUCCESS in 1h 10m 15s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d6c7cef9e0974336a51c3d0bea7f9246 : FAILURE in 1h 15m 17s","accounts_in_message":[],"_revision_number":8},{"id":"22939ac9d0703e50822bb86005d5ddfebaf8dd7f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-05 16:09:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/900118-6/Lenovo-iSCSI/ : SUCCESS in 2:06:47\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/900118-6/Lenovo-FC/ : SUCCESS in 1:55:54","accounts_in_message":[],"_revision_number":6},{"id":"960d5d1a57455599c8cee93ac0cb98aab47f3226","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-05 16:52:41.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerFlex_3.6/5abc530/DellEMC_PowerFlex_3.6/1378 : SUCCESS in 1h 35m 10s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerFlex_4.5/c39a9ee/DellEMC_PowerFlex_4.5/346 : SUCCESS in 2h 02m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"136f93cec8657a863d724de8ffd73db1ae092d42","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-02-05 18:53:09.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerMAX_FC/8a8ea37/DellEMC_PowerMAX_FC/315 : SUCCESS in 4h 32m 10s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerMAX_iSCSI/137e2fa/DellEMC_PowerMAX_iSCSI/337 : SUCCESS in 5h 32m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"4397a44414bb7c88f8195936c9c1a853883ce712","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-05 20:43:38.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/900118-6/PVME-iSCSI/ : SUCCESS in 2:07:29\n- PVME-FC https://os-logs.seagate.com/dell-ci/900118-6/PVME-FC/ : SUCCESS in 1:55:00","accounts_in_message":[],"_revision_number":6},{"id":"83d486df94487ee5c2ebf319dbe0568901e3ca95","author":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"date":"2024-02-06 02:49:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f50d6574856649b2ca1c1526dc5e13cf1a503308","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-06 05:44:19.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_iSCSI/348ac37/DellEMC_PowerStore_iSCSI/1045 : SUCCESS in 1h 10m 37s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_FC/22a412b/DellEMC_PowerStore_FC/1200 : SUCCESS in 1h 25m 39s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_NFS/d2a38a8/DellEMC_PowerStore_NFS/731 : SUCCESS in 1h 00m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/900118/8/check/DellEMC_PowerStore_NVMeTCP/07457a1/DellEMC_PowerStore_NVMeTCP/600 : SUCCESS in 1h 39m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"3d1e04f615165520d54d2e53a85958971b2e0ab8","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-02-07 13:44:00.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7107f47488c1f1a723bfff0d13f90de501f44d00","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-08 04:05:27.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900118-7/HPMSA-iSCSI/ : SUCCESS in 2:01:54\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900118-7/HPMSA-FC/ : SUCCESS in 1:36:23","accounts_in_message":[],"_revision_number":7},{"id":"db2288ec5d616bbf567eadd1aee8f48e477a8ab9","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-08 05:19:18.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cf25a2236f7a9e4a23cef2455a70b939b6b254bd","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-10 02:10:10.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/900118-7/PVME-iSCSI/ : SUCCESS in 2:04:08\n- PVME-FC https://os-logs.seagate.com/dell-ci/900118-7/PVME-FC/ : SUCCESS in 1:59:29","accounts_in_message":[],"_revision_number":7},{"id":"5ece931a1b629e179978768df01c94cb7d6abddd","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-11 00:24:25.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/900118-8/HPMSA-iSCSI/ : SUCCESS in 2:01:17\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/900118-8/HPMSA-FC/ : SUCCESS in 1:37:11","accounts_in_message":[],"_revision_number":8},{"id":"bc9b50cc7ef09ec2d2ea130e4efe3c6f8f0025b4","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-11 00:33:49.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/900118-7/Lenovo-iSCSI/ : SUCCESS in 2:05:29\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/900118-7/Lenovo-FC/ : SUCCESS in 2:00:06","accounts_in_message":[],"_revision_number":7},{"id":"b96378d2eeff421745ee4db831ad20605d27a868","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-11 13:46:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/8/20240211192904/NEC-V-ISCSI : SUCCESS in 2h 52m 45s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/900118/8/20240211192904/NEC-V-FC : SUCCESS in 3h 17m 19s","accounts_in_message":[],"_revision_number":8},{"id":"b69dc9146086eaad6b67a0a9add11fcbdecb9a32","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-02-13 05:07:04.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5002c5ce47163047c9c82f5849b0da415493cd0d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-13 17:05:44.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"de6a8ed7662445abba74f97549e6f5c47ebc8f78","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-13 17:52:57.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"3458d21b89cfab0d7d8b613a942b6a5a863a4ebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 17:53:32.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b2509e91d875c7dfb682a1467fe1a3c1d7732172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:47:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71a3638291b04dd0b446ef31bb8031b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41c99639baa4a838ae44d06b2e63807 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5165181ab2d4373bba19d2b53977628 : SUCCESS in 9m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c413af606aea4638bd4f9add947a0973 : SUCCESS in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19cae055cdc44939be6777d9c7de98ad : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b76e2e0b99d54e2a81af86087b3e8824 : SUCCESS in 1h 00m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a1e313d60cf64af781d63bb55706e865 : SUCCESS in 1h 05m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/625eba5579fb4437afc7a7982cb17658 : SUCCESS in 1h 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67153e3a890547fbad8ab23d4dd2481b : SUCCESS in 31m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7700341b827b4f4cab799b5d2f70b04f : SUCCESS in 10m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7ac4a046608d49e7adb279a95d927ea2 : SUCCESS in 1h 13m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9a372b96f224dd486c817188e795256 : SUCCESS in 1h 52m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b1e34d8eaa34d8eb9301b9b0fb2e746 : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":8},{"id":"6a2c8914d2182dad667217002abdc86f823bf853","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"428441fca7dcc578f6551833fb1e9abbb08ec001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:48:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/547bd7648df645b7bf24f83535a56fc5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1148d8e1c794fe3a20022fe588f4be7 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b0b99c66ef14ae28f0520b3aabdbe3b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":8},{"id":"e89cc44c834faed46dcf8bc7397238e4b4eb4a2a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-14 19:51:57.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/900118-8/PVME-iSCSI/ : SUCCESS in 2:07:32\n- PVME-FC https://os-logs.seagate.com/dell-ci/900118-8/PVME-FC/ : SUCCESS in 2:07:26","accounts_in_message":[],"_revision_number":8},{"id":"8d64e4caafc2f362c2c38f96b7d25cdb15cd4db7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-15 10:06:32.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/900118-8/Lenovo-iSCSI/ : SUCCESS in 2:06:01\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/900118-8/Lenovo-FC/ : SUCCESS in 2:05:25","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5d725806ff274d60aaeef83903303e1939502343","revisions":{"4715a7a88bad7d365568a6d5b94679de24274eee":{"kind":"REWORK","_number":1,"created":"2023-11-09 06:18:18.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/1"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"subject":"Add support for Dell PowerMax Unisphere v10.1","message":"Add support for Dell PowerMax Unisphere v10.1\n\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4715a7a88bad7d365568a6d5b94679de24274eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4715a7a88bad7d365568a6d5b94679de24274eee"}]},"branch":"refs/heads/master"},"dcb348eadb56280f927bb304849c5acd4e5e5ebb":{"kind":"REWORK","_number":2,"created":"2023-11-10 09:54:18.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/2"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-10 09:53:00.000000000","tz":0},"subject":"Add support for Dell PowerMax Unisphere v10.1","message":"Add support for Dell PowerMax Unisphere v10.1\n\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcb348eadb56280f927bb304849c5acd4e5e5ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcb348eadb56280f927bb304849c5acd4e5e5ebb"}]},"branch":"refs/heads/master"},"f4b386a1d2a8962edfc07662fcab3a571fb5d44b":{"kind":"REWORK","_number":3,"created":"2023-11-10 10:15:00.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/3"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-10 10:14:46.000000000","tz":0},"subject":"Add support for Dell PowerMax Unisphere v10.1","message":"Add support for Dell PowerMax Unisphere v10.1\n\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4b386a1d2a8962edfc07662fcab3a571fb5d44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f4b386a1d2a8962edfc07662fcab3a571fb5d44b"}]},"branch":"refs/heads/master"},"1f1fe011387d3425407328606ddfd46337786094":{"kind":"REWORK","_number":4,"created":"2023-11-24 09:48:48.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/4"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-24 09:48:24.000000000","tz":0},"subject":"Add support for Dell PowerMax Unisphere v10.1","message":"Add support for Dell PowerMax Unisphere v10.1\n\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f1fe011387d3425407328606ddfd46337786094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f1fe011387d3425407328606ddfd46337786094"}]},"branch":"refs/heads/master"},"61123b1bc4664c9206f50ba65af9a5a09933bf1a":{"kind":"REWORK","_number":5,"created":"2023-11-30 10:45:40.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/5"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-30 10:45:22.000000000","tz":0},"subject":"Add support for Dell PowerMax Unisphere v10.1","message":"Add support for Dell PowerMax Unisphere v10.1\n\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61123b1bc4664c9206f50ba65af9a5a09933bf1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61123b1bc4664c9206f50ba65af9a5a09933bf1a"}]},"branch":"refs/heads/master"},"c04bc7460396fe54e759c0d410a38160b01d2cd1":{"kind":"REWORK","_number":6,"created":"2024-02-01 08:44:39.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/6"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-02-01 08:38:08.000000000","tz":0},"subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","message":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\n\nFixes an issue introduced by change Ie9561eeb30a54539cbd, which added\nsupport for PowerMax Unisphere 10.\n\nCloses-bug: #2051828\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c04bc7460396fe54e759c0d410a38160b01d2cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c04bc7460396fe54e759c0d410a38160b01d2cd1"}]},"branch":"refs/heads/master"},"c3fb9c7bc2f7e4c2b242cd060c91a674e90bb9c7":{"kind":"REWORK","_number":7,"created":"2024-02-01 16:10:06.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/7"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-02-01 16:08:30.000000000","tz":0},"subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","message":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\n\nFixes an issue introduced by change Ie9561eeb30a54539cbd, which added\nsupport for PowerMax Unisphere 10.\n\nCloses-bug: #2051828\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3fb9c7bc2f7e4c2b242cd060c91a674e90bb9c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3fb9c7bc2f7e4c2b242cd060c91a674e90bb9c7"}]},"branch":"refs/heads/master"},"5d725806ff274d60aaeef83903303e1939502343":{"kind":"REWORK","_number":8,"created":"2024-02-04 13:15:16.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/18/900118/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/900118/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/900118/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/900118/8"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-02-04 13:14:16.000000000","tz":0},"subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","message":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\n\nFixes an issue introduced by change Ie9561eeb30a54539cbd, which added\nsupport for PowerMax Unisphere 10.\n\nCloses-bug: #2051828\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d725806ff274d60aaeef83903303e1939502343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5d725806ff274d60aaeef83903303e1939502343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
