)]}'
{"id":"openstack%2Fcinder~902120","triplet_id":"openstack%2Fcinder~master~I3a02857f0f2af3e63d7b62101663ead30902c73f","project":"openstack/cinder","branch":"master","topic":"gcs-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-19 02:40:44.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-19 02:40:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a02857f0f2af3e63d7b62101663ead30902c73f","subject":"gcs: Remove unused fallback to oauth2client","status":"MERGED","created":"2023-11-29 01:17:38.000000000","updated":"2024-08-19 03:27:49.000000000","submitted":"2024-08-19 02:40:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"902120-gcs-cleanup","meta_rev_id":"e0d568edc74e54a30378c13cdae916ee09a21bea","_number":902120,"virtual_id_number":902120,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-19 02:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-15 14:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-04-17 14:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-15 14:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-08-15 14:54:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 01:17:46.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 01:47:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 03:15:31.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 03:19:03.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-29 03:26:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-29 03:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-29 03:44:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-11-29 03:46:48.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 03:51:45.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-11-29 04:41:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 05:14:54.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 15:16:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-11-30 12:36:33.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-30 17:43:23.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-11-30 17:57:19.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-11-30 18:25:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-30 22:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-12-01 14:46:49.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-12-01 18:00:47.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-12-01 18:19:52.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-02-02 06:26:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 03:42:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-20 08:22:01.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2024-04-17 14:38:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-08-15 14:28:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-15 14:40:33.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-08-15 14:42:37.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-08-19 01:07:37.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"3c5a2966f78b8f23edddb47f1ecb1cd2dfa7b786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 01:17:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11ee99f95f1b3c20a683c867982745be0d15273f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-29 01:17:46.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16198/","accounts_in_message":[],"_revision_number":1},{"id":"2753889e356cfdd275ad9f12bd375d52e135dd17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-29 01:17:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16198/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8fa2e702846fca9ccf78015a160d8d2d208f3bfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 01:47:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c108900602446f888c1405da5bd1db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d384ecdb734d45528ea8969f2ebe4dda : FAILURE in 28m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4032338b55b483ea40150f2017f8244 : FAILURE in 29m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f3f0fad0c1f3408eb2029f97ec248c9c : SUCCESS in 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7ebd27b0f68f9ad023d647c855dfef9f923e261","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-29 03:15:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_PowerFlex/b4da60b/DellEMC_PowerFlex/942 : SUCCESS in 1h 04m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e394a051b39bc1f1e90cab57fcc7e49db0c497d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-29 03:19:03.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-902120-1/FC/ : FAILURE in 2h 1m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-902120-1/iSCSI/ : FAILURE in 1h 23m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e63c2d845968589e7e45195c2159e7deb32cafeb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-29 03:26:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/20/902120/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/49e949f/ : SUCCESS in 2h 07m 20s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/9a790042f9844263bf14f2d47b6515d5 : RETRY_LIMIT in 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/5c65196daa3b4092b8862676536ce466 : RETRY_LIMIT in 46s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/20/902120/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/3a03c6a/ : SUCCESS in 2h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0e2e27c1aac029cdbfee6e38faad4baa7b82cae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 03:31:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9bdae8fdc2d4a54957bf894454d46d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d33b1600e1344d182e1918130938600 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fd158f7c98e428a94f4f1380735407e : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0484628d57c466f905d567f782b7c6c : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8699968b6ca45eea5b693aab1287605 : SUCCESS in 14m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9bda0784db1f4866a4dbc6482e705420 : SUCCESS in 3m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/5995801c80284d839e3c66ac4f05b96e : SUCCESS in 54m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8ec691fe5975464aba3481a9fe05f297 : SUCCESS in 53m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d5ada36f4e140ef9b125c3410b30a78 : SUCCESS in 1h 23m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8897f9e404ff4946b4f43bfe68b3e915 : SUCCESS in 57m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eff53a2b71334d279e92ec59e8a15862 : SUCCESS in 12m 45s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/661b2b58512b4c919ed0bf974ade7613 : SUCCESS in 14m 19s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7dd5d40242db4620a91a09e35d6dadfa : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de8e80d0317e48cbb524d82d2d07b812 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1968cc660c94f738e2a3e507b85d5c3 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f7f8029ad4df4169988698d12f12d532 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/958e0322bf0d461f81adaf3517194e20 : SUCCESS in 48m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b027bab8d1f04206b8e3d75b8f100360 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/360aab145a884c72b43019afdbed97cb : SUCCESS in 1h 54m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/035a0cdfe742449597ff1648cb98f084 : FAILURE in 2h 09m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0aa56ffb8a6b4ff4b517b06e0bdb056f : SUCCESS in 1h 44m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f6429a83da094a5ca14e0f605bcb9df0 : SUCCESS in 1h 54m 12s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa829b6d6845436fa0cc534d415bd7a0 : SUCCESS in 1h 04m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/96ec9f54f43d4c6ab03f47e98ccb901d : SUCCESS in 1h 32m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/154d856c7fef4d26aafcd9dcf6e3f274 : SUCCESS in 40m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1ce693eedba844f5acacba99049a2fda : SUCCESS in 1h 27m 20s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6fac7901efc548529ea16a40fb44dda1 : SUCCESS in 1h 48m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb6ec388d09f4d028bb686a2a3ca9ec4 : FAILURE in 1h 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42646ddf0b6c49b7aa277d3f9c08d67d : SUCCESS in 44m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bceb373cc0d52c6a720e61155a64f5660730a62d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 03:37:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69bf098ac722d9d6a4555ec367be801c0c4a3d58","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-29 03:44:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/164/.","accounts_in_message":[],"_revision_number":1},{"id":"182260b263a32d6b24203c2a273b09d92d7d41cb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-11-29 03:46:48.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/20/902120/1/check/fujitsu-eternus-dx-iscsi/6d556aa : SUCCESS in 2h 18m 20s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/20/902120/1/check/fujitsu-eternus-dx-fc/392e5ea : SUCCESS in 2h 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3e047f127c2826abe505d4bb87179bd37c464ded","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-29 03:51:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/66e6143b64b34141848a19b7a6fe444a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/555d99d20ff449e68ed31bcc214cf10a : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"22723817bcc1a21bba4a655db327ad3fc7bd7987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 04:12:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b9ea955b1f47fb9e29f86de67447ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf0b116be4274239956392fb99f3a7cb : SUCCESS in 31m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe555d7421d64b01bc782024963f91e6 : FAILURE in 30m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a49299915a94c1b861ec9d96b578ccc : SUCCESS in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98c623e9ab3536a31050130876164a5baf003b59","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-29 04:15:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/168/.","accounts_in_message":[],"_revision_number":1},{"id":"4fb19ed634801cac383f5b02b8e89910193cdc28","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-29 04:41:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f3de6dcc91ed446ebd5ba83625db9c93 : SUCCESS in 1h 10m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/88d6169b61f341158b3ac842fa62bde2 : RETRY_LIMIT in 2h 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7c3c61bda2222fc5a8048b7a39f64744a806a403","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-29 05:14:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/20/902120/1/ : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a83a745f4ffd24c29358badbea731b25338a6240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 06:08:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d964b5b88b714743b3a254b4eece15f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b60d5a1f04400991a00ed06d8640e2 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/697007d072db4978a1473e9f28c504e2 : SUCCESS in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1df718ac10314f7abfdd0c81081a17c6 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78598c223ee40ecb447ffe5785dbac3 : SUCCESS in 12m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6228ca866c104f6a8348b78074cd7084 : SUCCESS in 3m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/7160d97c217d4a4f8f152b26b4e8b3a1 : SUCCESS in 48m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d8e1fe969dc4be19d8a64de0a92ddad : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a94cb1cdd35448980c32012b4c5dcb3 : SUCCESS in 1h 34m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef2a74611d084b2a8558d6993d4a2a88 : SUCCESS in 1h 07m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95be767d3e7245eeb4a292bbdeb882c7 : SUCCESS in 17m 48s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2271f284256b4fc2a5a6c2c30b6d4c38 : SUCCESS in 14m 35s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ef7d7fd46a904dda8c8461157527bc2f : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccfc00b0e1524d5c9625d97fa2fe812f : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4099f35db19a4b458801b267bcd945a4 : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9188ded14f384079a73e2f776ec89ead : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/db11657a505c42baaed53f495419b530 : SUCCESS in 46m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cde63cc9d6b64a11b2ff40835f7361fb : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d55c59a337ec42659cc07fef0ef1c4f4 : SUCCESS in 2h 22m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d842cc23b48945d68d9c38967f9788a7 : FAILURE in 2h 22m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f80b0b7db5c24063aeda6a5de896c0f1 : FAILURE in 1h 43m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/604688d3c46b44cab9811a6211449187 : SUCCESS in 2h 01m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a901d0e2b19b4c17996a9ab95456fc96 : SUCCESS in 57m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/43f8e2c55a8046259be3e4decbad8cc8 : SUCCESS in 1h 50m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e0ce6e20309740c3aa28fd1ae61ec47e : SUCCESS in 38m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e4cc5a3157274a93bc76c0934549fa28 : SUCCESS in 1h 03m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c4135e4403e249729cdba6359eddb75c : FAILURE in 1h 15m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/944edd238d0d4c5794aa48d9f6086003 : SUCCESS in 1h 38m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1daba77793d44b6b1b2f9bfb911fe33 : SUCCESS in 45m 31s","accounts_in_message":[],"_revision_number":1},{"id":"65210433ebdf491511b93a99a923e40515948bfd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 10:09:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1793ffdeca79d4a92594b1189e90983674ad2630","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 10:48:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1546a6cbe994a4dabf46e1c5ae8fc2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e25966dd9db74edda39edb5f55b587cf : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d0fea1de1cf4b538768a6f91db1470c : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3c5fa5fb9c04045ae5439c77700cf0f : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75755e7b15bda36faec4f2d41d28ee4cf2e46cca","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-29 11:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/2b67542a91cf42fe9aadedc443fa5aea\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/629f725c0a7646da907585bfb1a38947 : SUCCESS in 1h 23m 08s","accounts_in_message":[],"_revision_number":1},{"id":"893c3c9223d97b47d1eac452b3d7f6229a5dc0e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 13:15:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97a96b02d8764a3690101036eee2465e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67f7c7b0dd443a0a15fd79596c9b209 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e901c5438efc4ef09e27000b816ddec0 : SUCCESS in 12m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95603e9e10f14cf4add5e4038b89154b : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff2092c774840d496bdbc38fbd8fd23 : SUCCESS in 14m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d49337a8e5664a40a9a1c74f4af0df30 : SUCCESS in 2m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/1a0108b1aa5c4950ab23f23c74fd7a80 : SUCCESS in 57m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b5bf275e97de4cfc9a8c4bf3eb35122e : SUCCESS in 49m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7a5d12d869e472f9829a9233991b393 : SUCCESS in 1h 29m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e69363d5b7984f65a47d7abb367aff43 : SUCCESS in 59m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e3c255d72f949d5bb15463d7b611567 : SUCCESS in 13m 50s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3d87944d2c204077b2cfe07a8a91e82b : SUCCESS in 11m 01s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/19cebb7a13104c5ba777f55a84d3c4ac : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/70d5e00ce4fe4c269cda0cbd7ab3dfbb : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c5b36091615e4d3682acc37f05f759a9 : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d04d72f28374c4a92608ce866edaeb4 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/85b5db7787264b70a07d25db322c925c : SUCCESS in 39m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02c7c44efcde4fe3910377191897c279 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa813093f93a4935b08f3d4d2c0ebee1 : SUCCESS in 1h 53m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a82f422a3864874b9b0294ae7c2da66 : FAILURE in 2h 08m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3214ed9c284649cdb8fe54acab44a965 : FAILURE in 2h 05m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/05aad9262c214047bbd9a50c13d56b57 : SUCCESS in 2h 30m 09s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee3c3aa23a6641bdbd5ba157a94f6754 : SUCCESS in 57m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a50c93704df145bab64631c9a729d7d6 : SUCCESS in 1h 33m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/502e11ef0fca4ed1a7c83205cdd1aeea : SUCCESS in 46m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf0753c0a53b40db8468fbd1caaca99a : SUCCESS in 1h 21m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/024b7f653f9141f19656aa9fd9055c05 : SUCCESS in 2h 04m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/136bebc4243740ca9022aba4971066d1 : SUCCESS in 1h 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bad024b04df4f9f9878e154efbda535 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4ebe27b3e3f2427cf8c7e9b986b1252193903c2c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-29 15:16:00.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_Unity_FC/5f9df51/DellEMC_Unity_FC/1010 : SUCCESS in 2h 12m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_Unity_iSCSI/dd21a97/DellEMC_Unity_iSCSI/1025 : SUCCESS in 2h 25m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"30c5ec7d06c1ce65cba031366963f937704e06af","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-29 18:36:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ab14e77fc41144d4bee94a73f5ddb42b : SUCCESS in 1h 09m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9838b48e0bbb4868af2529244b34b2d7 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5e18daff547405b43b39ef85a98c252be70a221c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 12:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/1/20231130131354/NEC-FC : SUCCESS in 3h 58m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/1/20231130131354/NEC-ISCSI : SUCCESS in 8h 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b5030172dd3e4dd0bf97f396d2c852539b929d0d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-30 14:07:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"182166437636fc59ef6155d21955704ae7e5b329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:17:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98daebec0acc4c5d8dc0e1a5b9220a69\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c15e4bf6946e43bd81aab627fb5971b0 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18879129f4b24162b8665207757ebaf3 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55aeb1d555dc4d69852b5e15847e790e : SUCCESS in 9m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4bc4df1ae2be926f10ac24b1179420779f69c43","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-30 15:24:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4592961a995442f4b3598b0d6d5d18d7 : SUCCESS in 1h 14m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/00b01cb6f1b54926b2799ffaaf10e17d : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4fd6b32bcb7b8c3c7dae3c9c7df83eabab58c8c1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-30 15:33:12.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c7c7693945954eac8f5bcb10a01a771b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/59159e0f2bdf4789be4221d5c9260dcc : FAILURE in 1h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"78e06f9e485bc3ceec6af7be99cd5d2d97d9df59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:22:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/980f76ed86ad41bb85bd2024b4573566\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8986e2c8a92c4fc1b9d9f4af313509a5 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acfb269ae27a4ef6a3b81b38320913b1 : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ae712123c3c406d9ae30b332a44ad2d : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff1d684a3064dc781d1ec838751da0b : SUCCESS in 14m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d769a2023284cea8691ead43e6491aa : SUCCESS in 2m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/10cc7c43da854df5b96dc4e4d580ce44 : SUCCESS in 55m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/09fcb3e0605a4b2ca5298a28d4b0ad13 : SUCCESS in 49m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2868d11c89b4e14aadc7f215cfbbf7f : SUCCESS in 1h 34m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b2ac399a61f4cd38b584e21fd973053 : SUCCESS in 1h 17m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0d36a011da1455d92ed76f25d770852 : SUCCESS in 15m 07s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/34c41a208832416390e13e620682c99e : SUCCESS in 13m 21s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/494bd55ea4644d2b9c1196405b1ff0e3 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5767f1daa0a140278b50156a6aaa2609 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/01c7e19ee2ce4086aa7f6300982ce5ce : SUCCESS in 6m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/72b04c0429ee496980ab8b731e4346d4 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8a806cd405e84e2dad9ca736e3d7576b : SUCCESS in 52m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe180a9450184ece87f48e1a5c7d3481 : SUCCESS in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48c1a1d5100b481d83982991bd8e6496 : SUCCESS in 2h 00m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1387750618324d3fb31d17f26abdc169 : SUCCESS in 2h 04m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/18dd4211e92541f2b3bcef1dede41d65 : SUCCESS in 1h 42m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3afe615ad1eb43ed829a8c7023b6f67f : SUCCESS in 2h 07m 46s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ee0d6e55bee4b428be52ec4046892bf : SUCCESS in 1h 26m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d7a93adab214fc8b78f28bb589380f9 : FAILURE in 2h 07m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f85cb4c472e049098ea69547e8982553 : SUCCESS in 37m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bcba41ab02b04cb98d7075392e5582e7 : SUCCESS in 1h 13m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e4cd730eb0c44301914df2c4466be503 : SUCCESS in 1h 24m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23f4293b6db34de18afe913032c73c2e : SUCCESS in 1h 25m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f027def6288b4bc4860f18db627ca6f9 : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6d9f017601fe1391645cf5dc75d4221ebda91ba5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 17:39:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_PowerFlex/e399fcc/DellEMC_PowerFlex/970 : FAILURE in 1h 16m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"300d4f8cff221319fbec22fd389e6a3003e3fd65","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-30 17:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/1/20231130230715/NEC-V-ISCSI : SUCCESS in 3h 06m 56s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/1/20231130230715/NEC-V-FC : SUCCESS in 3h 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a4a80e3fd614dbc160a1a6b859c540a93319b97f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-30 17:57:19.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_PowerStore_iSCSI/188f5b6/DellEMC_PowerStore_iSCSI/438 : SUCCESS in 1h 04m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_PowerStore_FC/9c49e5a/DellEMC_PowerStore_FC/441 : SUCCESS in 1h 30m 28s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_PowerStore_NFS/2c8c2eb/DellEMC_PowerStore_NFS/447 : SUCCESS in 1h 07m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_PowerStore_NVMeTCP/2911e0f/DellEMC_PowerStore_NVMeTCP/431 : SUCCESS in 1h 34m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"45cc1b182900c80cd9fe0b570183cb5831292d12","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-30 18:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/1/upstream-check/cinder-tempest-plugin-ontap-fcp/6a0011d/ : SUCCESS in 1h 23m 10s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/c53309e/ : SUCCESS in 44m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/1/upstream-check/cinder-tempest-plugin-ontap-nfs/21ff002/ : SUCCESS in 46m 33s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/0bbbf38/ : SUCCESS in 1h 46m 14s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ac86fb3a30f99d623b5c9e0708e0d7b28676bb9c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-30 22:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/902120/1/fc : SUCCESS in 0h 58m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/902120/1/iscsi : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c3143bf453208d15e66a81df4a98857a4572903e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-12-01 14:46:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/902120-1/PVME-iSCSI/ : FAILURE in 45:08.15\n- PVME-FC https://os-logs.seagate.com/dell-ci/902120-1/PVME-FC/ : SUCCESS in 1:55:02","accounts_in_message":[],"_revision_number":1},{"id":"2a3c081aa781ac9a1ceb5ec9c9e813a3ccca5f36","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-12-01 18:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/902120-1/Lenovo-iSCSI/ : FAILURE in 45:03.53\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/902120-1/Lenovo-FC/ : SUCCESS in 1:53:36","accounts_in_message":[],"_revision_number":1},{"id":"4bab3b904374b6308cf1610b52000a715a48d30e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-12-01 18:19:52.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"9c978f0375af10282dddf1043598e2fe4aed9ad1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-01 20:16:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_Unity_FC/c816a0a/DellEMC_Unity_FC/1029 : FAILURE in 2h 12m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/1/check/DellEMC_Unity_iSCSI/cbb0bb2/DellEMC_Unity_iSCSI/1044 : FAILURE in 2h 34m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b99af087c77f488ecbff90916b1016370b0ca37c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 02:39:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b248b33cb357628904dd525160181a01accfb1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 02:43:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"37ec02299bd799e765d407e5300ca4fc43e2e6e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 02:54:34.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"297ae0242e56aa75e8395d3f4aaa25d1ddf36f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 03:22:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4afbedbade4ccc833757a6cf979a69\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14b406f2ed9f4f618f1029160b13d694 : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc00480b5daf415fb4a345d35fb4b355 : SUCCESS in 23m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a4cd70766044748b6259be993380136 : SUCCESS in 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e5bd1e4f9f828298bdf7bd6b1f95b696e24ea8ba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-02 05:28:10.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/efbf0039cc6448ee827696c3767ac08b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5219ab23f7e5458992475f3a074ef270 : FAILURE in 1h 22m 18s","accounts_in_message":[],"_revision_number":4},{"id":"e9dc7cc06c121b7dfcac8aeca450a52aa5e0f3a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 06:25:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae97dbc571c4794a6d0d4b6f214ca2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f9b47282f941c98f2404a85235365b : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee751ebeed6d4beebac3c4386db00747 : SUCCESS in 11m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbd78205fc804b2ea56f290c64afe28a : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71702de30e494986a617918ec5fc7a72 : SUCCESS in 12m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/733fd9a4098248cea205e3a115bbf39a : SUCCESS in 2m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c9e534e4d11548fa806d14fafbbd87a3 : SUCCESS in 48m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/212a9c784b384a439066564ddd4cb77c : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/798aa1fa90c1450d81447353f434ca86 : SUCCESS in 1h 30m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17507813de5f4a06bfda439430fe40d8 : SUCCESS in 41m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ec67ab63568404f925eada5f8ab8c62 : SUCCESS in 15m 53s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be3d5391f2274891a76ff099d36bf10d : SUCCESS in 16m 56s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/44b98a9750424f27896026de5137a82e : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/120ee177936c41db93e205bed4fcb0ed : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a8db96ad8844b04a7057e2617af4f9a : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef9f4049dd64494a853b34d1fc555fb3 : SUCCESS in 7m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e5c091b6ddc48af94f38fb63759b553 : SUCCESS in 46m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bcd8d927df7f4ff6b6515854c8c176da : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a4ae90abacc64c25b34e0e6ba47d3d68 : SUCCESS in 1h 49m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/73c6977e2dec409e8ea5b79e1c2f39b7 : FAILURE in 2h 08m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc74d59e111b4d09a847bfa46c2346f6 : SUCCESS in 1h 38m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f67401809b5d4f098a828d82bd52763e : SUCCESS in 2h 09m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3578cf9c8f7e4b3dac39b58f8b582180 : SUCCESS in 1h 02m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/87580f70103742dcbdd2369b0a1aafb6 : SUCCESS in 1h 18m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3bb03c3972004a789995416fcb40f208 : SUCCESS in 35m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/69f80d94737b42b19f78c04427b5cc82 : SUCCESS in 1h 06m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/43c47d557d9846899eaf7168bf31a574 : SUCCESS in 1h 16m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97aac0e48b8043879d16ba35c76ee7cb : FAILURE in 1h 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d172f9001bf64627b83ffcab7ad725f1 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":4},{"id":"3c88c385b1156265631276ee9954c06b75aa660e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-02 06:26:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902120 : FAILURE in 34m 21s","accounts_in_message":[],"_revision_number":4},{"id":"826f59212e073f1fc8292fe467fc427c80046241","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-02 09:11:20.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-902120-2/FC/ : SUCCESS in 5h 14m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-902120-2/iSCSI/ : SUCCESS in 1h 54m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"365041ba9b95b0693b56907f93baf13b6f6387e8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-02 10:13:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_Unity_FC/cf769ec/DellEMC_Unity_FC/1201 : SUCCESS in 1h 08m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_Unity_iSCSI/646034a/DellEMC_Unity_iSCSI/1220 : FAILURE in 15m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d0050d0e8c3383806ff29fd80a857d8668921212","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-02 11:46:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/20/902120/4/check/fujitsu-eternus-dx-iscsi/3509db3 : FAILURE in 15s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/20/902120/4/check/fujitsu-eternus-dx-fc/15ea0e4 : FAILURE in 3h 52m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d0c1076f7351182c10baeaf23909e33b293424b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 12:50:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7640d768d68445a38f236bdf14a4dd401f91a4e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 13:16:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f5d146eed6e4aeca873abab6057b197\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d83ffcaaf094f3fac551133d3fde46f : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fb98ced3e7d41df8426164bb82260da : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/224c45f732884ab2a20bf66cdcd33cbd : FAILURE in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"09150001da7043183d745bce5c0e9ecce2267a58","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-02 14:45:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b9ddc2e29fe648348e8014a7040d9614 : SUCCESS in 1h 49m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f48d21d3de8a45dbac5ef23d88026f0f : SUCCESS in 1h 33m 30s","accounts_in_message":[],"_revision_number":4},{"id":"3a247dfa8fb9d38b6ed19d28e223267f01ec54c5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-02 14:56:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/567/.","accounts_in_message":[],"_revision_number":4},{"id":"980e398dda31aa83ec62ff9666554a0a539af87d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 15:26:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8332162cb74b5ebd59bee0f507a948\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de36d786353443e483c1621c76c28ba1 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b5dcd6e0cf8408abe4afa0164cf527b : SUCCESS in 11m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1da6c77d2a2a40c986eeaf2051985fd4 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308a05b8973b4531b8a9999074f3072c : SUCCESS in 11m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f189f192ae94321978fc1a3fee221ca : SUCCESS in 3m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b3ed730c5c3b45ea9602e400cb9bfd6f : SUCCESS in 1h 02m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/97d18a29a19d4809a5c9956e88255b10 : SUCCESS in 52m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/485a8e87830649249963342ff4405503 : SUCCESS in 1h 19m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e225f1f81f2a4bbcaf13e83617ce1c5c : SUCCESS in 28m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f12c75b98e1406197b357b679b5088d : FAILURE in 14m 19s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2411311c18c946dbbfad0c65ef8e32cc : SUCCESS in 12m 49s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7c8b787abe4424e86a091c0e796a3d9 : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9df83366d08a46bdb0e282052df979d1 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/860a6418ea2d45b1b5cf2a6da35daca4 : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b75003ea0764afb98eb55ae39d19977 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6857dbcf7d064776840e0b24d576180e : SUCCESS in 49m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd959b43f06f4c0490bf02082569027c : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/89272876491c4131ba0d0af482f3266d : SUCCESS in 1h 48m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec0d91428392463a917e7d120cf03f49 : FAILURE in 2h 14m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/938f08d444ac4980bd5c9ad20ef4b8e0 : SUCCESS in 1h 55m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e89ca3c99cd248828885c39e9cb9948a : SUCCESS in 2h 01m 45s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b22ca911fe14ff2843e9ca10397bbf8 : SUCCESS in 1h 19m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b952b5f79fc64396a56557b15160a66c : SUCCESS in 1h 32m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/acb6b21effb748a09ec627bf9a7f4ce8 : SUCCESS in 37m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/21d5a766ffb2401ea88987a7b344247b : SUCCESS in 1h 22m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ee4a4354d5ed4618bcd67cb9998d1ca3 : POST_FAILURE in 2h 29m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1d21507bab494c3d8ec17cba6639fc58 : SUCCESS in 1h 33m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e374f11d3be048af8cf4360b987b58d0 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":4},{"id":"e1727738aba90f2669e1f1db1bf215561fe0b771","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-02 15:45:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/4/upstream-check/cinder-tempest-plugin-ontap-fcp/7e5e01d/ : RETRY_LIMIT in 9m 11s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/f999aaf/ : RETRY_LIMIT in 7m 21s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/4/upstream-check/cinder-tempest-plugin-ontap-nfs/c4d8d0d/ : RETRY_LIMIT in 8m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/de12664/ : RETRY_LIMIT in 7m 35s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"76cf06d7cc12e25100c313c8dcafed76d95c187d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-02 19:55:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/579/.","accounts_in_message":[],"_revision_number":4},{"id":"61e7e77e2715a6f6ff6068c86a263215a3146246","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-02 20:08:31.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/03d950c2fefc4490bf3b1d32240d747e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b2275ce68ad0450993f00c98b7f1d6a7 : FAILURE in 1h 22m 45s","accounts_in_message":[],"_revision_number":4},{"id":"9fb2aca3bded1484ea0c82b097e4c2ff28c1aeed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-03 06:24:09.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_Unity_FC/bb30444/DellEMC_Unity_FC/1218 : SUCCESS in 1h 08m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_Unity_iSCSI/01abc35/DellEMC_Unity_iSCSI/1240 : FAILURE in 14m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"fc169aa160fe9a8484135daad418afbb9b4a4a2a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-03 09:13:37.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_PowerFlex_3.6/269f895/DellEMC_PowerFlex_3.6/1350 : SUCCESS in 1h 43m 13s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_PowerFlex_4.5/6d96062/DellEMC_PowerFlex_4.5/318 : FAILURE in 1h 58m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"4ffca434b29f9a548935a39c93b87409bec4582d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-03 19:19:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_PowerStore_iSCSI/9142b59/DellEMC_PowerStore_iSCSI/1023 : SUCCESS in 1h 06m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_PowerStore_FC/43b8019/DellEMC_PowerStore_FC/1178 : SUCCESS in 1h 27m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_PowerStore_NFS/0e250e6/DellEMC_PowerStore_NFS/709 : SUCCESS in 1h 00m 10s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/902120/4/check/DellEMC_PowerStore_NVMeTCP/1c34a06/DellEMC_PowerStore_NVMeTCP/575 : SUCCESS in 2h 17m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"29f6fc18229bd97b20d816f412f0bfc9459339d7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-08 13:05:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/902120-4/HPMSA-iSCSI/ : SUCCESS in 1:59:32\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/902120-4/HPMSA-FC/ : SUCCESS in 1:36:40","accounts_in_message":[],"_revision_number":4},{"id":"10c817974c5504fcfbcbb4538362ea76687e26c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-10 11:11:43.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c2e3d799e16a5a3169ac5c3bc7d5cc7b12189dcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 11:28:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd4cbeadcb724b5c92539a82e9f06212\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bba27f9248a4f4fb6318b582fa3dcea : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e5075a8cae640aea5523feb2e2bd7e2 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac1d710dab8d442db81d4a86998a33d3 : SUCCESS in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3ac396b757ddcfb006e7389a01016abf6b8bd25e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-10 12:53:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902120 : FAILURE in 42m 18s","accounts_in_message":[],"_revision_number":5},{"id":"79731a2d792890d1b1fbaa9adf49d039d8385ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 13:22:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e687d989146e48e58cb0bf6fea7f7685\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e1ba6f5da3241408d4ce562b04e5b63 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b4022cc0fd04660933311ad8339a905 : SUCCESS in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/528cab68ec6e4732b2e6b8d4a03ed562 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c10d5ed98b0496c82de32a24be29bad : SUCCESS in 12m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f11a3bbfb8f494392c590860ff01ba3 : SUCCESS in 2m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/9164dcadc28b4d40adeba7372bc2f85e : SUCCESS in 51m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6853d70c88d04a879a8f4bec054c8ceb : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2cb6f0159e7c401fadbdbd783c086698 : SUCCESS in 1h 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3107abead907466a873683bbf8e39c97 : SUCCESS in 34m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/561593391f1748c08390d11b782d9901 : SUCCESS in 16m 44s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b579644458cf4526bff4a7be2c92625d : SUCCESS in 13m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4848809d1f6548679f0268c79521e5ee : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/739f6ab42d0f423498604e6ffda7b665 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d957c44f6ef04bafbf57d62e0b71f875 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54a87310db47470896f5e2bb82ac67a4 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b5fafc3bdec34fb2a16b4c1006f91843 : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/222ce5a4c5ef4c4b8a2641eba7b0cf7e : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0cac0fc238864bb0a22352d3db149d07 : SUCCESS in 1h 38m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4ab7d34a31d0436996401fe672748301 : SUCCESS in 1h 44m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30bf0ca3024d4ba2a38fe3acff722ac8 : SUCCESS in 1h 30m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/38c201b703a649c0a1ed79af4b5342ed : FAILURE in 52m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90287f85aefd4bd499935a54a5e0c09e : SUCCESS in 1h 02m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6dfec6314ded490aa7d62b0738faea5e : FAILURE in 1h 15m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d9fc26b22e7348a4895b0aca33508afc : SUCCESS in 47m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/034b92227c674561a1bb6ee302d47849 : SUCCESS in 1h 21m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7e8306e3bb8a4d5ebe4983a823998d6d : FAILURE in 1h 02m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54172f329db14e6db90e1230a8991f9a : SUCCESS in 1h 16m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9416698192694935b9df1aa7881055b4 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":5},{"id":"7a348e001bcdff8044c4c182af33c38dd0bef869","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-10 13:43:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/47c919b5f6ef4b05886d9b183e07a78f : SUCCESS in 1h 16m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d0cd3b51e3fb4fbcb30ef940237902f9 : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":5},{"id":"18c44c7d66dcce3cebc9a6e34b414da42f8a0b3a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-10 13:44:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/5/upstream-check/cinder-tempest-plugin-ontap-fcp/5cb70d3/ : RETRY_LIMIT in 9m 31s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/c79200f/ : RETRY_LIMIT in 10m 04s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/5/upstream-check/cinder-tempest-plugin-ontap-nfs/0f24cbb/ : RETRY_LIMIT in 8m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f2180d1/ : RETRY_LIMIT in 10m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"204282fa9b01950df0a99a843870fd0d3b5b6424","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-10 14:07:37.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_Unity_FC/80b0720/DellEMC_Unity_FC/1292 : SUCCESS in 1h 13m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_Unity_iSCSI/098fb75/DellEMC_Unity_iSCSI/1322 : SUCCESS in 1h 30m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"203df50d773a102c0396d10198f6ff694fb0a194","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-10 14:09:53.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f525909d73bc45ef8d0cb9c07363b727\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/514f3bba0dc34805a7f973fb1131e4f9 : FAILURE in 1h 28m 24s","accounts_in_message":[],"_revision_number":5},{"id":"30e073a3381498587b802488c8ca6c41083c10ff","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-10 15:03:00.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_PowerStore_iSCSI/fe5b0df/DellEMC_PowerStore_iSCSI/1063 : SUCCESS in 1h 04m 43s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_PowerStore_FC/525b9c9/DellEMC_PowerStore_FC/1218 : SUCCESS in 1h 34m 52s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_PowerStore_NFS/62895b8/DellEMC_PowerStore_NFS/749 : SUCCESS in 1h 10m 08s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_PowerStore_NVMeTCP/12a4e50/DellEMC_PowerStore_NVMeTCP/618 : SUCCESS in 1h 40m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"d5c8c32c4b235d2b174cff275960c8c141e824ec","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-10 15:07:49.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_PowerFlex_3.6/e67ebec/DellEMC_PowerFlex_3.6/1426 : SUCCESS in 1h 36m 37s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/902120/5/check/DellEMC_PowerFlex_4.5/cd729d6/DellEMC_PowerFlex_4.5/392 : SUCCESS in 1h 51m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"838df771575d951539b6d155dea02c60438d9ad6","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-10 20:08:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/902120-4/PVME-iSCSI/ : SUCCESS in 2:05:08\n- PVME-FC https://os-logs.seagate.com/dell-ci/902120-4/PVME-FC/ : SUCCESS in 2:00:46","accounts_in_message":[],"_revision_number":4},{"id":"608d67733e71dc9a217e620bdaa0748fa483c95c","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-11 18:31:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/902120-4/Lenovo-iSCSI/ : SUCCESS in 2:04:38\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/902120-4/Lenovo-FC/ : SUCCESS in 2:02:59","accounts_in_message":[],"_revision_number":4},{"id":"892fbcc8eaab7ac00eb36dc3651e8129997f1ad0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-12 08:31:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/682/.","accounts_in_message":[],"_revision_number":5},{"id":"5f57cab1d2887ec2495e7b2c40348783a01ca163","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-12 12:15:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/688/.","accounts_in_message":[],"_revision_number":5},{"id":"fcabc0d519ea2fef35febf76238fa1e906a2f7fd","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-16 17:58:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/902120-5/HPMSA-iSCSI/ : SUCCESS in 1:57:54\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/902120-5/HPMSA-FC/ : SUCCESS in 1:37:43","accounts_in_message":[],"_revision_number":5},{"id":"45fc6cc0aa5886e610cd2d70d1df7cf3ce37d36a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 01:16:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"03f06124b45ad1516f6103a120ae3d9eb4849354","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 01:16:33.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16532/","accounts_in_message":[],"_revision_number":6},{"id":"adfc17905227b5c3f66acecfbf5aa27916e027fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 01:16:36.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16532/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"9a18e8f02f33a68c4839a57cebc1d0bb8e54ae97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:27:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/663da3fd832b47deb773a7274bad33cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/808cad34626040dbbc362e88bb6a6457 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9b69d62ff264dbf996b99fdc18d0d5c : FAILURE in 4m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb89b2a3f39544d8a8b748dcfbdc6a8e : FAILURE in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"877f0d3ea9a4167e55878fc0c25f0ec489a05287","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 02:03:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9b5b0315a34e46adbe0a19715e62e4e0 : RETRY_LIMIT in 7m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/95c6a69e35944910abbac5671dbe4141 : RETRY_LIMIT in 7m 14s","accounts_in_message":[],"_revision_number":6},{"id":"94c71aa075bd571d0fd89ead0f7573d38c622e47","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-20 02:49:45.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902120 : FAILURE in 55m 52s","accounts_in_message":[],"_revision_number":6},{"id":"7ab69da31b24251c02fb189806eb7281c59ae037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 03:06:58.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"a3418a74e9b50898030b465076e95e9cd5059a99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 03:07:05.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16534/","accounts_in_message":[],"_revision_number":7},{"id":"a11bc6d34a09f980b0b1156a33c70d315fa62bbd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 03:07:08.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16534/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"127ec0acfd4b470f484585f9bdbdf4d1530c449f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 03:29:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8843645ffb374affbaaed1c1707bc578\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea0825e8abd444688f4c37b7a1bb8975 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb156b8300314f12b22653c1f8cd34be : SUCCESS in 18m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/903074699c8e4b37938f840fad9a4661 : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cd8add8c5c4b341b67bb968035a5538d153f764d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-20 03:30:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902120 : FAILURE in 19m 11s","accounts_in_message":[],"_revision_number":7},{"id":"9ebacf6a9a07601884797313e085e1b4501e1541","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 03:40:33.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7093f111d2ba4596a03466acd5657012 : RETRY_LIMIT in 6m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/711c35dd736d49b6b743125d532e3301 : RETRY_LIMIT in 7m 59s","accounts_in_message":[],"_revision_number":7},{"id":"664002842e58eaaeb1e616c0bfb2921d95a3bf16","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-20 03:42:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902120 : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":7},{"id":"68999fa85fe18530fff28b8ad7a255164feee862","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-20 04:28:47.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_Unity_FC/3f827f5/DellEMC_Unity_FC/1363 : SUCCESS in 1h 17m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_Unity_iSCSI/4e089e4/DellEMC_Unity_iSCSI/1393 : SUCCESS in 1h 21m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"556929685678223869dae149165c38200f52f4bc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-20 04:33:24.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-902120-6/FC/ : SUCCESS in 2h 35m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-902120-6/iSCSI/ : SUCCESS in 2h 49m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"9ed96240a0ea3d53099a5828e300acdc9b953ce4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-20 05:10:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902120 : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":7},{"id":"88f2fd368fdc8bb12e4e9b701d0e926631bc6593","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-20 05:19:10.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_3.6/16430c8/DellEMC_PowerFlex_3.6/1494 : FAILURE in 9m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_4.5/fac9e69/DellEMC_PowerFlex_4.5/458 : FAILURE in 2h 09m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"00bdec009e2afc38ec6b0b6675de5751c12686bd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 05:30:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/739/.","accounts_in_message":[],"_revision_number":6},{"id":"b2d85fcd3443f8e90113edda947067a18c122021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 05:36:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bfd49c25db843a4bc50b86b7f4bedb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b995e508d5af4266b0c4ab90e029b72b : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2046e6f996c4c8699ffebfaee4afa48 : SUCCESS in 12m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a9f43ce3d5044e68b27321e5e3e5171 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6568bf9e17724adea95624baed7392fe : SUCCESS in 11m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e0640eada0a441bb9dbb694bfc1938d : SUCCESS in 2m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/b7b68b735b294686b758b7651427fdc0 : SUCCESS in 1h 09m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f297941be01435e8c68128cbfa593e4 : SUCCESS in 1h 04m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4040a6477bbb4e15b3086051e9298a96 : FAILURE in 1h 04m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c0f4bb17c1d4eb4834516a3e0b4842f : SUCCESS in 53m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9c03d995c856497c960c50dbfc56a2d0 : SUCCESS in 14m 59s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b93ec9ba473d4be684565651554534b9 : SUCCESS in 13m 03s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ebf6f5abb50441791a3a7029f5850d2 : SUCCESS in 6m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3031da71131f44fa80539d717ba361b4 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05a24f599d9f44ec94cf53e0798ddc94 : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f4d649283e34c80aa2d2c7e3dbeb494 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ff04c3337f5450688e2d254c2fd4ef9 : SUCCESS in 48m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d796dbf37331409ab803b416449fe65f : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a8990f4ecfd34956a4aa88a84414c5a7 : SUCCESS in 1h 56m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b23868d450a74a529c2c5e577ec7ebcc : SUCCESS in 2h 21m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f53bf4db66224fb88b40b2cb8e99c611 : SUCCESS in 1h 43m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dffbd3b456e7479287b7a4fb5543022d : SUCCESS in 1h 48m 23s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e261d7783e0f4cba8d0fd11ad8f4e69d : SUCCESS in 53m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/172055e4a2b243e0acf4be4b61119208 : FAILURE in 1h 27m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e648aeb35b1f4a7da881d073f67664e0 : SUCCESS in 42m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db7ba5b19d5d43019a731b073bd93baa : SUCCESS in 1h 21m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/08a351fa707e4977bf32091a3aa902a7 : SUCCESS in 1h 25m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62003f0b6a794461bb3e25ea46d69795 : SUCCESS in 1h 22m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4af6670026341c2b46fb024a5241cd9 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":7},{"id":"0c3cceb93d70cdd18f2c2d3961584efccf2c6755","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-20 06:20:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/20/902120/7/ : SUCCESS in 1h 32m 36s","accounts_in_message":[],"_revision_number":7},{"id":"320e611b50430bfe49ba8139ae7fb048efa3269b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 06:57:14.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/749/.","accounts_in_message":[],"_revision_number":7},{"id":"7c9d0cf2d431b7d0ad39c4dddd6e75507145d53c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 07:26:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/740/.","accounts_in_message":[],"_revision_number":7},{"id":"3e8ef0e42d096d95c9ebc18ed793dce3521a8303","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 07:43:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7f34bf1b013fd0da5de3ad6a7338aa4aac4df906","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-20 07:55:51.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5ed50bf6db604a31a1d2093b5befcf6e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ff3093c935cf435b81ebefb3be5d6013 : FAILURE in 1h 28m 00s","accounts_in_message":[],"_revision_number":7},{"id":"02b85adb360b7213688593fc656fed3770a5ec12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 08:06:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e90e27929143d198db63dda643d4f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86b2fa9df90b42909bcf795a5614fe2a : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/25492f9eb2f04bd2bc54d17469e6c766 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23091ee70a18495da5f7efb5f06d68c4 : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"22dc72cdde90a386675b8b1128d0b8aec02d7492","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 08:09:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c28692b3a43341c69c49d1819d998951 : RETRY_LIMIT in 6m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9f3236338467476992adc6a31b2269cc : RETRY_LIMIT in 8m 07s","accounts_in_message":[],"_revision_number":7},{"id":"0ab641982862fd600be44010585fd528f1ee0466","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2024-02-20 08:22:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b50d1756e971ab1d93cfdad3b5a3a0932b0f2eab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 09:36:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0904bf1b68ea464ab844e560177594ac : SUCCESS in 1h 12m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e30d3f9eac5e45ebad11471dab474a25 : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":7},{"id":"b2cf5d6306d94938edf416cd9b9fb0e2ec09fe7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:51:24.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2140e6a18d94af297bee9b325c13109\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e2583e4bbe46c38f7c74b4edefcf43 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e54d29ce19246288cb2ee35af1ffae9 : TIMED_OUT in 40m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53b471103be8486087d1e5ccbd18f5c4 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ad389c45324e1d8eb4abc761a5fb0d : SUCCESS in 12m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0ca9dd7746e74e22b922fe2e0b3b5162 : SUCCESS in 2m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/e329d249a0214b86827929cd7a38f5c0 : SUCCESS in 54m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c004eca8eea400f9b5cf212cb88f08f : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9671ab8b3c345f19634929286fc179a : SUCCESS in 1h 18m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d45b2503eb964a40ad22025ab90b7b2b : SUCCESS in 59m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3bc0131039664c08bc4c429bebff0929 : SUCCESS in 16m 34s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5469f4fd359f478d9e389573707dc95a : SUCCESS in 12m 36s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e506943c0d1495fae87df42e9341dd5 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/52b707f9839f4d34a5281e7b92485be8 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15fc962753864bdc8a8535d5107d6a78 : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99166158b99342e983e52dd90b89d79d : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c967484f7f7742e596084801d4e19e61 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0db4d19e8e3b49f3b209d626e385882a : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/157abe3f900a4e9596e14807374ce3c6 : SUCCESS in 1h 49m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/02397b53453540f3a390278d86401e97 : SUCCESS in 2h 21m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cbfcf0325d194b8f8b0322eacc1b8f12 : SUCCESS in 1h 38m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/52bcc44732d84569a2ff1c1724cdab55 : SUCCESS in 1h 52m 46s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f94a0bcb760046eb8ee71147fef3c38d : SUCCESS in 1h 17m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c56a3ddbe91438bb0730b91190ae460 : SUCCESS in 1h 27m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b57c916560a483facf76e4539fc62a9 : SUCCESS in 43m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57203fe1f7b6445286a27b2ac1788c99 : SUCCESS in 1h 14m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/50b7f25790314742bcb33500d5722077 : POST_FAILURE in 3h 00m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a3f8560be084d53a2e5ffea93b685e0 : SUCCESS in 1h 35m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5445af463bc54fb2b500b5fca65155f9 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":7},{"id":"4a3464fbd2742aac7b01a69f03b0d4884b16f7d4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 14:11:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2eb7d4e11ba766a479920663cc0eb8feaf2d4c12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 14:28:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c151e9a56df14465974fc605155c9111\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a464dbc174d441609aa28a1cac08192f : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e302fb961bc4104acb786bf18196b2a : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9548ff8f3b384287bb423b86b8c09c58 : SUCCESS in 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9439cd7d5c65cfb22d2d4c987125110f9fc41fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 16:29:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c968fc9b9b84fb1baa7b32eac00ac25\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6b47d107744ea099e5e8c25c99ef02 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c9abc9fc8c34ec7a69c631c7ca0f9fa : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc62bb12478442228572121766628579 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcef08f214f34d2ba1c395ab66731d5f : SUCCESS in 12m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e3ee8312fa314a68ab32f15a256ec1ec : SUCCESS in 3m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/6c9a1d3609ec43e19bfa4f8cea18a9d6 : SUCCESS in 55m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c80308b812145efbe0585336d9301f6 : SUCCESS in 1h 05m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6879f654e8d445b6a024a30fcd7b9e91 : SUCCESS in 1h 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cf99463240540479c48ba83dcb63033 : SUCCESS in 1h 01m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77a794e4dae740bf98909794fd12cb55 : SUCCESS in 13m 55s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3d01cd9d119840a89d2196ff11d512d8 : SUCCESS in 11m 33s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/615f7dfd97904e798b23be92b77638d0 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0f137c456d6436684ff3a513ee3a325 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ea82f2b12c042ecb942c8ffb805f26f : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2fa42d76035476eaa4f030d6120f314 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dfa315ad310047dcb073477f407c6b53 : SUCCESS in 45m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5162704c5062488f80a8551f769c4b33 : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b3e0f46f1684eeab613a62fe515d763 : SUCCESS in 1h 57m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d3c0b63ffe4649118588717ded761538 : SUCCESS in 2h 16m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2da85e768f5c4f6798dc3b9bffd9ba86 : SUCCESS in 1h 46m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9e355bc4bf0c4f9680ca52edeee20f3a : SUCCESS in 1h 32m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71b6d144e65646cd9570310ffee6a11a : SUCCESS in 1h 05m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d32340196d984530b020207800297c03 : SUCCESS in 1h 14m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/979d581aedaa42d9adea6952bf63d40d : SUCCESS in 36m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a358e208720e494c84015c303f08d6fb : SUCCESS in 1h 09m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0c74ad16bf944cf9903b2d933d85f0b5 : SUCCESS in 1h 32m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/10c3a0027afa46159294bb84476e1ba4 : SUCCESS in 1h 48m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba985e9a7e6a45dba593b51002490002 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":7},{"id":"21e38c4924b0963bf73da80d16abd3e0d9b5dffa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 16:37:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/494dedbd2c34494ebed2a84713551253 : SUCCESS in 1h 10m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f8b2dc34916a45aa89a1606be52de03c : SUCCESS in 1h 13m 05s","accounts_in_message":[],"_revision_number":7},{"id":"4cf4a36b7959928a72b74e75f46490ff1aced304","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-20 16:53:42.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/82346e5b7816485eb3986681857ee58b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f2b4970cf3234eb2ab16f711117943b1 : FAILURE in 1h 15m 15s","accounts_in_message":[],"_revision_number":7},{"id":"4dae6642e4abcd30aa8dd6a528e375740eb344f0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-20 17:17:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/7/upstream-check/cinder-tempest-plugin-ontap-fcp/22e6e01/ : RETRY_LIMIT in 8m 37s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/9153e6d/ : SUCCESS in 46m 16s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/50a1420/ : SUCCESS in 46m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"805fdb36fe68b6a9c0b91313c0147b6eba6d448b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-20 18:20:25.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_Unity_FC/3c89e09/DellEMC_Unity_FC/1371 : SUCCESS in 1h 11m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_Unity_iSCSI/2009028/DellEMC_Unity_iSCSI/1401 : SUCCESS in 1h 39m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"d78462e2a438c97a7927e6e7fd73feb54be0e0d1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-20 18:34:27.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_iSCSI/242c2b9/DellEMC_PowerStore_iSCSI/1100 : SUCCESS in 1h 10m 10s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_FC/c12fbb6/DellEMC_PowerStore_FC/1256 : SUCCESS in 1h 29m 37s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_NFS/71860ca/DellEMC_PowerStore_NFS/786 : SUCCESS in 1h 04m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_NVMeTCP/13f344c/DellEMC_PowerStore_NVMeTCP/655 : SUCCESS in 2h 04m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"64c9cac9d8bebfa50822b0c2c636f7899ba78129","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-20 20:49:09.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_3.6/bc6874d/DellEMC_PowerFlex_3.6/1501 : FAILURE in 2s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_4.5/6d2adde/DellEMC_PowerFlex_4.5/465 : FAILURE in 2h 12m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"c6e41ed25da0862bd12a20ecfb87a35a78f7616a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-22 07:47:46.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/7/20240222130519/NEC-V-ISCSI : SUCCESS in 3h 02m 53s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/7/20240222130519/NEC-V-FC : SUCCESS in 3h 42m 23s","accounts_in_message":[],"_revision_number":7},{"id":"98d2fee6f6da38fa50a4a5a445721ac7d39ad627","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-23 17:10:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/902120-5/PVME-iSCSI/ : FAILURE in 37:01.73\n- PVME-FC https://os-logs.seagate.com/dell-ci/902120-5/PVME-FC/ : SUCCESS in 2:07:26","accounts_in_message":[],"_revision_number":5},{"id":"672660f42e45aa2cb31b5b10ac31df3bd91f0a2a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-24 03:39:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/902120-5/Lenovo-iSCSI/ : FAILURE in 37:32.22\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/902120-5/Lenovo-FC/ : SUCCESS in 2:06:03","accounts_in_message":[],"_revision_number":5},{"id":"873600f5492bf7f2d622e42bbeebee7b96b61fb7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-24 20:20:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc650ea7e02f4c81bfc3244ab69ea37c : SUCCESS in 1h 13m 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0df1d687d63647e684e17e75e14c3dbf : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":7},{"id":"cf3a20c7209079cfec684c2827f38f6722dfcc57","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-25 03:51:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/47bb2fa9c81d45ea99ea190ed472b558 : SUCCESS in 1h 16m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1181c892d6e840f5b0b37ca31f988902 : SUCCESS in 1h 12m 16s","accounts_in_message":[],"_revision_number":6},{"id":"be93187daee8a49b0eeac18d73050bb07c3450e1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-25 20:34:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/902120-7/HPMSA-iSCSI/ : SUCCESS in 1:54:38\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/902120-7/HPMSA-FC/ : SUCCESS in 1:36:35","accounts_in_message":[],"_revision_number":7},{"id":"7655ad0b47505274c067927d511d2a08f22b8720","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-17 14:38:13.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"d4ec39dcec63985f0d2c381b8bf7ecc761f06125","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-17 18:00:08.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/902120/7/fc : SUCCESS in 0h 45m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/902120/7/iscsi : SUCCESS in 0h 58m 15s","accounts_in_message":[],"_revision_number":7},{"id":"b88d07fea2a920ea54173f8dca87322a55cdfa29","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-07 19:49:43.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/7/20240707191945/NEC-ISCSI : FAILURE in 8h 48m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/902120/7/20240707191945/NEC-FC : FAILURE in 9h 29m 54s","accounts_in_message":[],"_revision_number":7},{"id":"1f97a93ba03669bcbdab0cfbcbf3dcece99ee515","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-15 14:28:22.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"93bd61520294c8f2475dc86dc901ace0a208c3cd","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-08-15 14:42:37.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"fce871b38a9f29e1cca35d59f63f4d1d57b5d4b6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-15 14:54:50.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1a9f63d0190a3b16311b376f8de21ffa4cb5fe61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 14:55:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c49fddcbdd847b91af2d960188750e8c074b4f8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 16:40:26.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/902106 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"7d14656a7fc6ebbefba8094de674bfff6a4c8db3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-17 17:03:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"55396322308533ad8a0d3af9ef137a8597c199df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 17:24:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0424a5c738e4fe69ccc9ae5a3794f5c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acc433ee41ca465a987651291f2cf37a : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/206749be01874ca9a24c4b47f0bc785f : SUCCESS in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a08ee5ec2e45884958f04826393899db75fa0e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 18:59:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7a26bdc8c284069a947aa1efaf70265\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87868ab993904addb94b2db2ebb7df27 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34047c1e9be04f579a31b2e383fb0397 : SUCCESS in 11m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddac223c1b894d71bd79fbb5e710d675 : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fe9f99f75e24f45a8f4dda7f7499e37 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb707a548bc54c20987cdeffb4b02851 : SUCCESS in 10m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6fa99ac560aa485da42bba8ef9a0aa72 : SUCCESS in 3m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/bef79844b87247198c7eea315ffb2010 : SUCCESS in 43m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9edc859d85aa4290b948f39e287dcc96 : SUCCESS in 1h 21m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4a77d5b4e7a46c186537c9333b6c933 : SUCCESS in 42m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7bf059b81ba4734b958f95f3a7a92dc : SUCCESS in 15m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ae40b358563740e3b2da62052b2424d4 : SUCCESS in 8m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5903c5bad1d9409495c86344298c35dd : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6ea85d5a48447ae9e69563eb8e6b367 : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/677bc305029746789460accd9f223a75 : SUCCESS in 8m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/93803827c4094d5cb4d9c4b9ae3fa99e : FAILURE in 23m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac91b1df533946a9943a53a215ea162e : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/158ff8359cd24e559f1058cd5ad6ef55 : SUCCESS in 1h 20m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f7b6911355db49f7b1c3d6133f67c4f0 : FAILURE in 1h 45m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63fc0a92b38e476a8c49c1ea76b619c8 : SUCCESS in 1h 30m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/647ef931b5e54e7888d6a958a07aaae9 : SUCCESS in 1h 23m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/932012b999a544859db80d9b83a36e44 : SUCCESS in 29m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c0cc9aa779e4b3f8d8b8bdd030288d4 : SUCCESS in 53m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/885e23680a82427f93f2937e1610ed11 : SUCCESS in 1h 07m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ffe0b6ac9f57414a97bcb7087a6363df : SUCCESS in 37m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23acb3c2973a4f64897443d31295ed98 : SUCCESS in 1h 07m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/01497a0907db425bb61323af6d13ad51 : SUCCESS in 1h 23m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dacad278b4294ecfa84e113079e4390c : POST_FAILURE in 27m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20bdf0d3d7f740b686c3521f8e0202f1 : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":7},{"id":"f2daecc6a11d042f10bf95d99c511ba3513b4259","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-17 19:43:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/86d91d190bbf4cad8626823f4c534016 : SUCCESS in 1h 21m 43s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1f252f84efe44bf68e91da319c813d9a : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":7},{"id":"c26f42a894f4e56ab2c1cad418edfa1579d85eb9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-18 22:57:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a24cbad1fbfa0a61d812cea03a7d59f64a52d0e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 23:18:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cebbcce50c464807bba995576aa2a7a0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1e66ae2dcdd42f3810cadbc7ceecf04 : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8cfd060d3d2c48808ddad126f7dca17e : SUCCESS in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b0a3ce6517e747adc4983453a1b4642a76f8a09","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-19 00:18:46.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fd1f5461e9b74dd4a2370176677b3542 : FAILURE in 1h 19m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5d131c306dd84e9ab68f3d819fc3a429 : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":7},{"id":"f80b2075c5d3eab130f03899323f6a5b67943920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 01:01:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd781a023a84112b38b1f9497b6c412\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840c34838a6e4db1b823aff9dffa4a86 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfdfa9e81979452ebd12f4552af0949e : SUCCESS in 10m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f210c40caf245f495a837795212aa45 : SUCCESS in 9m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db680af9f51c4252a8da5946fee2680f : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89db87595c564316b8277c8825788270 : SUCCESS in 10m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bfca5e38c094180a615a5a0870c0292 : SUCCESS in 3m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/ba1e4661cf154b77ae63db09a901e2c1 : SUCCESS in 1h 03m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19d6cf6297294384b4b505ed644fdcad : SUCCESS in 1h 07m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed90fa560dc345ba8ce22398e64d7420 : SUCCESS in 1h 01m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/090efa93af6d44fa9814f0645ee37326 : SUCCESS in 13m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1901661c8c884f22bff9bbb293329097 : SUCCESS in 4m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6bd7806c97f04f8c995462f3690bd7bf : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6021af728ef4b59b75faf616948f661 : SUCCESS in 6m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/45c09a80c50146cf855c5044b0e314f9 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aa41ec35a0e647fe8aaa6035bc04aa44 : FAILURE in 22m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f9b7a256389416c998400ffb4b5436f : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa40c0524f7c421099b2d76463bc9adc : SUCCESS in 2h 03m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d414b30a63f4c99bc56849ebccaa055 : FAILURE in 1h 40m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a987be7696884136b59d8b429ab2465c : SUCCESS in 1h 46m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4d6f10188eb841e796c464cd68c2ce69 : SUCCESS in 1h 57m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/40ac804785d6405ea3c90fd99c13f93c : SUCCESS in 39m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d448c2a8105417999976f9c6672a1d2 : SUCCESS in 1h 10m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8262bf19003249d496c22a804e60841f : SUCCESS in 1h 19m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/602a349000c3429a8184906723c366cb : SUCCESS in 34m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4d5fb492d0ef4c4f89763f589590a0e5 : SUCCESS in 1h 05m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b43b5cc0f7a647e38bb631b36186e93a : SUCCESS in 1h 53m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6db435c15654c01b28df4bc8f22171b : SUCCESS in 1h 45m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4eb96e796f2c4938b35c2452855e919b : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":7},{"id":"fa361fc4cc417a78c9323f6ff5a1c8724ac5887e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 01:02:21.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b20e0138e63cee8aab374b68da384bb77b7a796d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-19 01:07:37.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_08_2024_01_01_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"1e6fc51b7bde8070a34b16cf74a8f754afda8fff","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-19 01:22:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/54c844d/ : RETRY_LIMIT in 3m 09s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/902120/7/upstream-check/cinder-tempest-plugin-ontap-nfs/4a309b7/ : RETRY_LIMIT in 3m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"ef8aba7ea4de8e1ec6ae9dfdd9b5c29b46c0825c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-19 01:30:50.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_Unity_FC/fe195a6/DellEMC_Unity_FC/2436 : FAILURE in 21m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_Unity_iSCSI/e5bd058/DellEMC_Unity_iSCSI/2439 : FAILURE in 28m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"ba8e2f2cffb754ba3b7a4d984d7551b1e20b9f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 02:40:44.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/568aa9b293e14253b026c7972e9fc772\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eefabc29166486fa3a577a0580294dd : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fb36c6fe6444e3b93af4c80bb249faf : SUCCESS in 11m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b58a7276f2e41f794a6b6bae9b9f91c : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9120450fadae45cf89ef6cac04dfd86f : SUCCESS in 11m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a34faae4cf64a23b30a0c56e9337200 : SUCCESS in 2m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/bbbd603cd7de4f51ae3b7087796a43f4 : SUCCESS in 43m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88372883c24f49c7b2f1e108703f553f : SUCCESS in 1h 31m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61c35d2ed87d427cb3ce1af6c071ac06 : SUCCESS in 44m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/03a2d1e74dba4a3faeac6f8e9d8506ee : SUCCESS in 1h 13m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0502b2f9bcfa482b8661469a8d87f78c : SUCCESS in 1h 25m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/612ba0025e11492785714b636ff3893d : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":7},{"id":"0033da92c099311f659cd357da86e85b92ddb26c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 02:40:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"e74d3d6e44fc1b1a5fc3cc94f2f78f0258665962","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 02:42:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6adcbcee08dc4f3a84bdd81e4417c836\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28941cf09d1548109b42b37cabac8e8d : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":7},{"id":"8919a9e84a48d0301bfb6d151087970c943ebb9c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-19 03:00:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_iSCSI/cb0eaeb/DellEMC_PowerStore_iSCSI/1609 : SUCCESS in 1h 58m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_FC/d735b7e/DellEMC_PowerStore_FC/1805 : SUCCESS in 1h 06m 11s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_NFS/4545ba1/DellEMC_PowerStore_NFS/1301 : FAILURE in 1h 17m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_NVMeTCP/660eae1/DellEMC_PowerStore_NVMeTCP/1161 : SUCCESS in 1h 57m 01s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerStore_NVMeFC/299a4b5/DellEMC_PowerStore_NVMeFC/125 : SUCCESS in 1h 15m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"e0d568edc74e54a30378c13cdae916ee09a21bea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-19 03:27:49.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_3.6/8be74b1/DellEMC_PowerFlex_3.6/1915 : SUCCESS in 2h 25m 36s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_4.x/6a8ef58/DellEMC_PowerFlex_4.x/796 : SUCCESS in 1h 53m 05s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/20/902120/7/check/DellEMC_PowerFlex_4.x_NVMeTCP/3c43454/DellEMC_PowerFlex_4.x_NVMeTCP/767 : FAILURE in 15m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d71530c2a1dc893c54efda47813100940025dc93","revisions":{"c1a15e9352432d84de382412fb1d51b63351fa8e":{"kind":"REWORK","_number":1,"created":"2023-11-29 01:17:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/1"}}},"commit":{"parents":[{"commit":"84ac8df6313fbbdd0f2ac52d4cb5a7880d5ebeab","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84ac8df6313fbbdd0f2ac52d4cb5a7880d5ebeab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:17:21.000000000","tz":540},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 so\nalways exists.\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1a15e9352432d84de382412fb1d51b63351fa8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1a15e9352432d84de382412fb1d51b63351fa8e"}]},"branch":"refs/heads/master"},"378d756ab371128f6b407c52a016f438c97aad4d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-02 02:39:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/2"}}},"commit":{"parents":[{"commit":"a8f33b5e9e3479544801d4db644b2a21eeb5dbee","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8f33b5e9e3479544801d4db644b2a21eeb5dbee"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 02:39:13.000000000","tz":0},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 so\nalways exists.\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/378d756ab371128f6b407c52a016f438c97aad4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/378d756ab371128f6b407c52a016f438c97aad4d"}]},"branch":"refs/heads/master","description":"Rebase"},"1388e1d900cd6f726be1f6a8c5cb7ed2f7b7be79":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-02 02:43:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/3"}}},"commit":{"parents":[{"commit":"961d48a23ae1dc5bcb569e2aa309ac5060c139db","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/961d48a23ae1dc5bcb569e2aa309ac5060c139db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 02:43:48.000000000","tz":0},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 so\nalways exists.\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1388e1d900cd6f726be1f6a8c5cb7ed2f7b7be79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1388e1d900cd6f726be1f6a8c5cb7ed2f7b7be79"}]},"branch":"refs/heads/master","description":"Rebase"},"dd49a4008c930c4ac6cb79b63e5cbcfda6df7e4e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-02-02 02:54:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/4"}}},"commit":{"parents":[{"commit":"961d48a23ae1dc5bcb569e2aa309ac5060c139db","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/961d48a23ae1dc5bcb569e2aa309ac5060c139db"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 02:54:34.000000000","tz":0},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 [1] so\nwe can assume it\u0027s always available.\n\n[1] https://github.com/googleapis/google-api-python-client/blob/v1.11.0/setup.py#L43\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd49a4008c930c4ac6cb79b63e5cbcfda6df7e4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd49a4008c930c4ac6cb79b63e5cbcfda6df7e4e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7e2da64a1c1970b200a7068b85e1c00af6b1bad1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-10 11:11:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/5"}}},"commit":{"parents":[{"commit":"6702a620b5c6703c8ad7e6ef0f6693ba9db008a9","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6702a620b5c6703c8ad7e6ef0f6693ba9db008a9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-10 11:11:03.000000000","tz":540},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 [1] so\nwe can assume it\u0027s always available.\n\n[1] https://github.com/googleapis/google-api-python-client/blob/v1.11.0/setup.py#L43\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e2da64a1c1970b200a7068b85e1c00af6b1bad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e2da64a1c1970b200a7068b85e1c00af6b1bad1"}]},"branch":"refs/heads/master"},"41a174a463537e349b7d0b287a68e8e82db617cd":{"kind":"REWORK","_number":6,"created":"2024-02-20 01:16:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/6"}}},"commit":{"parents":[{"commit":"8fb2b3550c54d495d4314fcd4f9d683b3103cc6f","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fb2b3550c54d495d4314fcd4f9d683b3103cc6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 01:16:07.000000000","tz":540},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 [1] so\nwe can assume it\u0027s always available.\n\n[1] https://github.com/googleapis/google-api-python-client/blob/v1.11.0/setup.py#L43\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41a174a463537e349b7d0b287a68e8e82db617cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41a174a463537e349b7d0b287a68e8e82db617cd"}]},"branch":"refs/heads/master"},"d71530c2a1dc893c54efda47813100940025dc93":{"kind":"REWORK","_number":7,"created":"2024-02-20 03:06:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/902120/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/902120/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/902120/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/902120/7"}}},"commit":{"parents":[{"commit":"8fb2b3550c54d495d4314fcd4f9d683b3103cc6f","subject":"gcs: Remove logic for google-api-python-client \u003c 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fb2b3550c54d495d4314fcd4f9d683b3103cc6f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:13:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 03:06:57.000000000","tz":0},"subject":"gcs: Remove unused fallback to oauth2client","message":"gcs: Remove unused fallback to oauth2client\n\nThe oauth2client library is prepared as a fallback in case\nthe google-auth-httplib2 library is absent, but this library is now\ninstalled as a requirement of google-api-python-client\u003e\u003d1.11.0 [1] so\nwe can assume it\u0027s always available.\n\n[1] https://github.com/googleapis/google-api-python-client/blob/v1.11.0/setup.py#L43\n\nChange-Id: I3a02857f0f2af3e63d7b62101663ead30902c73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d71530c2a1dc893c54efda47813100940025dc93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d71530c2a1dc893c54efda47813100940025dc93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
