)]}'
{"id":"openstack%2Fcinder~902122","triplet_id":"openstack%2Fcinder~master~Iaaedcdaf53d327be35bb3b9f586a4188204ee23d","project":"openstack/cinder","branch":"master","topic":"bug/2058601","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-26 17:12:33.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2026-02-14 03:43:07.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-26 17:12:33.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-08-26 17:12:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaaedcdaf53d327be35bb3b9f586a4188204ee23d","subject":"Make google-api-python-client optional","status":"MERGED","created":"2023-11-29 01:22:54.000000000","updated":"2026-08-26 17:14:51.000000000","submitted":"2026-08-26 17:12:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902122-bug/2058601","meta_rev_id":"a94db1f7cf846a1f02205a9bc6fa45feb5acf8ef","_number":902122,"virtual_id_number":902122,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 17:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-08-22 20:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2026-08-26 15:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-08-26 15:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-29 01:23:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 01:35:38.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 02:10:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 03:05:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-29 03:48:42.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-29 03:49:05.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 03:59:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-29 04:20:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 04:24:26.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-11-29 05:08:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-11-29 06:03:48.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-29 17:44:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-11-30 17:13:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-30 18:38:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-30 19:30:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-11-30 21:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-11-30 21:12:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-12-01 17:53:18.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-12-01 18:42:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-12-01 21:23:28.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-02-02 07:14:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 05:50:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-03-21 06:57:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-04-09 14:22:41.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-04-09 14:43:47.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-05-17 14:47:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-22 02:40:11.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-08-22 05:15:40.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-08-25 04:53:34.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-28 17:14:53.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-09-17 19:54:34.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-20 16:35:24.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-07 10:14:35.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-02-07 12:14:43.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-13 05:33:14.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-14 03:43:07.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REMOVED"},{"updated":"2026-02-15 11:19:35.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-07-05 15:42:23.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-19 01:21:48.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-19 01:21:59.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-22 20:16:41.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-08-26 15:30:44.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"c0c5ecb93ed0b09b1f10b7fe07cef259539e2e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 01:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ead76b0d6ab23a725e5fceb5edf230fff99c181","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-29 01:23:06.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16199/","accounts_in_message":[],"_revision_number":1},{"id":"024e8424ccac3847c66acba67a3b87661ecdb69e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-29 01:23:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16199/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2caf802de2969dd2e3b47aa9f5bace92880089df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 01:34:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b7fe0725966b49906731f1f722e2bf1b9ca88f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-29 01:34:46.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16200/","accounts_in_message":[],"_revision_number":2},{"id":"22ef98bf306bd55ec563dc978b083a4cb53f7abd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-29 01:34:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16200/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a845d15c60dcc4cae8fd5a682b2b98e32dd5ffed","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-11-29 01:35:38.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"00d839d4dd6c65bbecda92f9b1811d98b88ece21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 02:10:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9b39681d31440ea2ae2391361caaf5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1beca932fec84e64a44b7f84a935fcf9 : FAILURE in 30m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/446704b2448d4f35b1cadbec9a150f99 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ee0fcb654884ea6803d8cc8a3073fbe : FAILURE in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d26aa1e0eb38cd64bdf6c95c32673c0798019aab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-29 03:05:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/d1417acccca24237b20ad904416f43cd : RETRY_LIMIT in 45s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/cff5c6e43c2242c5941dfa7d3e0bf862 : RETRY_LIMIT in 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/7e4bc2f088a94734a26c552a91f55551 : RETRY_LIMIT in 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/9bf2b2e7f26647d089a29b5f8406fa4d : RETRY_LIMIT in 47s","accounts_in_message":[],"_revision_number":2},{"id":"b131f195fb9b174509fe0931748ec110f6fca68b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-29 03:48:42.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-1/FC/ : FAILURE in 2h 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-1/iSCSI/ : FAILURE in 1h 33m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"20f3afa09004364cb3a28eb1ba3866b38c46e9ca","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-29 03:49:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/22/902122/2/ : FAILURE in 1h 07m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6c8a1bdcdadc8f6e7486b488fde55cb8ce99ed80","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-29 03:54:17.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-2/FC/ : UNSTABLE in 2m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-2/iSCSI/ : FAILURE in 1h 36m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"3e77852d9765c1aab7ae04cadd441a4ea8543542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 03:59:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c1db78be90a49bca40e7d4e0ae7d33b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97fbcec4319b47928f1385e53aecca2d : SUCCESS in 2m 01s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/daa989e5ca2e4ccc92389c3787188b7d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a14340b67f043f18d4bb27e5cd1abe2 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1520658770554534ad5e11ff31dd2492 : SUCCESS in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc8706c42e284ea3ab9ac9720d35e866 : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4ac43f680c4efabf1e58ca110a579b : SUCCESS in 14m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b23cbe3be4a42e0aef10c4e3bc5cb9e : SUCCESS in 2m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6b1565b79ca8426cb9c1a59d94715028 : SUCCESS in 1h 02m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/81bb618564034136a2eb012a7466245c : SUCCESS in 1h 03m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27ee62d8e6e44e46b95f8ba3fce5fbc8 : SUCCESS in 1h 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e023430bd6f94e8695d89c5302003ad7 : SUCCESS in 1h 31m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac09a63e0b96461cbd459d830a7fd7e1 : SUCCESS in 8m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ab195062d2d64b76977c7cc76e2c8629 : SUCCESS in 13m 54s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/12caa4c7630c48d88c01dde3505be726 : SUCCESS in 12m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76f553a4cf2948aea9988fc5e5987525 : SUCCESS in 4m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/641e66fa10284418a830443e997261a7 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/edf83235ba50428a8aa026eef3d48a5c : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7cf137b1154489187baabc53ea2a705 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e1ea4e2245ee465487f9d5ad67f3c96a : SUCCESS in 41m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23455e5e8b3943b2b48175a73c608cc5 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a36e560fd1de44db81784aa85b0d298c : SUCCESS in 1h 44m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/12f66a33ca5746fe9f18a7cb8b1575a1 : SUCCESS in 2h 12m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2da1e73e75514325940f57bb5aff9830 : FAILURE in 1h 37m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/74a881350c9a41d9b05451b21308fe8a : SUCCESS in 2h 04m 04s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/013995ba03804792bed36dd280e3f3f0 : FAILURE in 56m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b311decb66a547ed8876ed2f095b16bb : SUCCESS in 1h 33m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d8360e52d09742c48117f47caab3925f : SUCCESS in 36m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f859c976d39242b781ba686a3cca9b79 : SUCCESS in 1h 31m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e48d2228fa55404cb58f035be607dd48 : SUCCESS in 1h 41m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/23ab6f402c9045a0a70a1547eef10c3e : SUCCESS in 1h 36m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab2ce4fa2c9b4c5d8840961cc1c308db : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":2},{"id":"2f8e44df24e6b079694694961a0bcd7ecbf79f17","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-29 04:20:29.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_PowerFlex/fc0c358/DellEMC_PowerFlex/943 : SUCCESS in 1h 05m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0cecb64b8f7e5a702b35aff95621f5b8184a187a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-29 04:24:26.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8b9eb2fc1d0240469b9dd478144919f1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b906bd3120e2495487e3a723bd2ccb38 : FAILURE in 1h 26m 09s","accounts_in_message":[],"_revision_number":2},{"id":"25147e10c9ae11aa9d6a2afed1f8b378294cd565","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-29 05:08:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/165/.","accounts_in_message":[],"_revision_number":2},{"id":"6b6c6ea603fa5106a932bfd046bfbfc0f4f4b5c4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-29 05:38:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/169/.","accounts_in_message":[],"_revision_number":2},{"id":"7ab19fade6e2b546e4a1b9a65d5aaeb744748f7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-29 06:03:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cf366e2923224dc99610c97dd088ca00 : SUCCESS in 1h 08m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8cdca5acf82048678a44afa016368431 : RETRY_LIMIT in 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"53c92fe810459fb50ea038aec64dc55cfb16af23","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-29 17:44:43.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_Unity_FC/d27fdc2/DellEMC_Unity_FC/1011 : SUCCESS in 2h 11m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_Unity_iSCSI/698c9f0/DellEMC_Unity_iSCSI/1026 : SUCCESS in 2h 28m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"31e2dd0bfc0ccee788cfd448dd9205bb65430f87","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-30 14:07:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a284b6873143f3fde418f099f4886a4ae3714c0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9877c05fa83543c3a7664bd4bcb3601e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a776f65f6d294eccb23c6bb41a402075 : FAILURE in 28m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0290976edcf42bab3edede92771a43a : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eadf110fa4a942dda40a7685cdf5da8a : SUCCESS in 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"892a1a65afe537624c240c816be9f155fa45a57f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-30 15:32:47.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c7b8bac3595e45799e38f436010a4c68\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/376d66f612c34a0e897ec8ea5568e887 : FAILURE in 1h 19m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6d4c903275ca67d0f1bc46b21526f2b3d8e09a76","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-30 16:29:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/12272a15f4c64e33adc0ea3fcf3b0589 : SUCCESS in 1h 07m 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/643a72717ef94548b6e34bdbbc5f8caf : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e62a45c8e4fc47b6c122f4c62ae813cb4243a90a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:39:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f8b1e2354224946bb2e28975984c980\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1013d522a2b04bfc8586f80cb9de63dd : SUCCESS in 2m 50s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8166a109c3a94615963fa11d54b1bb78 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e812774959847ffb4a7f1029b7b78c5 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4155cf229bf4d0280775499bb3b9a1f : SUCCESS in 10m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6fb82345fd4455cb1bd392705a6aa82 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b0bfdbd001445eb546a78aa21c85b0 : SUCCESS in 14m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b74435c559b4d1f992076455945c4b3 : SUCCESS in 2m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2f6a46afe62f4f02a4cfd53c3428d6d2 : SUCCESS in 1h 07m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ba217ab0f59949289db6874f3d0084da : SUCCESS in 59m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b15a2d48716a4fc2afaec64cee21323c : SUCCESS in 1h 28m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d0d3a4bbf6249fe95ba80d769fe58a2 : SUCCESS in 1h 05m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5768d0295b24b1492a14b344a2e6a29 : SUCCESS in 9m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1aa8937ed2c6440aaa4ba37af775e9bf : SUCCESS in 17m 55s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/09b10326f46947e095838f95de79a1f9 : SUCCESS in 12m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/486e61f88960433c8ac5807bfc5aa292 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e1591c2d35948a492cc5481da7a6f93 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/665ec5c9a84c4f909031060638302aca : SUCCESS in 6m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e668eee61cc149cf9dcd84bad13d0809 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c44906d1ba78459cbde5c49eaaa9ab5c : SUCCESS in 39m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93226d250f6c4aa18422c413350157d7 : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/27a9de00a4b843ad9d312137947acc20 : SUCCESS in 2h 03m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3f9a3a8bd66447778179e5c181c3565e : SUCCESS in 2h 23m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/49defbc12dba46f4966f8d25376f48f0 : SUCCESS in 1h 40m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/886b8b6f22714431b45d96b0c056e2e7 : SUCCESS in 2h 13m 13s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49d227d7718a4244a7d3d02d8838f91d : SUCCESS in 1h 20m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e53771e39fe343278d4a537e9bbcc9ae : SUCCESS in 1h 26m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/59be3cb077c5421fa1ff7be3df8b6853 : SUCCESS in 37m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4e079b60589f4f668366d11857763ba8 : SUCCESS in 1h 16m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b0f28a964bb04544b5a1cfa1a3081619 : SUCCESS in 1h 43m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd0965ace203408e93bedb0f7af2df4f : SUCCESS in 2h 22m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf5ce56366ce404aae8d381a5c45c3bf : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"7973f3cfdd812a10f4aaa2c961b307f6f9e9ac59","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 17:13:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/2/20231130213636/NEC-ISCSI : SUCCESS in 3h 49m 25s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/2/20231130213636/NEC-FC : SUCCESS in 4h 37m 12s","accounts_in_message":[],"_revision_number":2},{"id":"543f5886e06bb25e98e0fbc941966840a7bf3d0d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-30 18:38:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/2/upstream-check/cinder-tempest-plugin-ontap-fcp/c09a4d0/ : SUCCESS in 1h 28m 07s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/144e087/ : SUCCESS in 1h 44m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/2/upstream-check/cinder-tempest-plugin-ontap-nfs/8739891/ : SUCCESS in 1h 45m 55s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/651ad9c/ : SUCCESS in 1h 43m 25s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"90292bf971bf7c246cc0f918b8890cbe37aa689c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 18:48:55.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_PowerFlex/b7292f0/DellEMC_PowerFlex/971 : SUCCESS in 1h 09m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6e854a3f9a039bb0959f7d5c471301563702bfdd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-30 19:30:03.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_PowerStore_iSCSI/77c3f2f/DellEMC_PowerStore_iSCSI/439 : SUCCESS in 1h 15m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_PowerStore_FC/91d6eae/DellEMC_PowerStore_FC/442 : SUCCESS in 1h 36m 29s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_PowerStore_NFS/f716d8d/DellEMC_PowerStore_NFS/448 : SUCCESS in 1h 02m 57s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_PowerStore_NVMeTCP/b4199df/DellEMC_PowerStore_NVMeTCP/432 : SUCCESS in 1h 25m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d07b0632f996df6bf0f840e19d6a16335d19fea2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-30 21:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/902122/2/fc : SUCCESS in 0h 48m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/902122/2/iscsi : SUCCESS in 0h 57m 43s","accounts_in_message":[],"_revision_number":2},{"id":"03baa68302d3b7304103055ccc7e452709eb9c2f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-30 21:12:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/2/20231201024326/NEC-V-ISCSI : SUCCESS in 2h 55m 33s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/2/20231201024326/NEC-V-FC : SUCCESS in 3h 28m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7b67bcfa4c2712208705011246e4a79ad0adf821","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-12-01 17:53:18.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"4e2bb3f8153324d4845af98613217515aa997178","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-12-01 18:42:24.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"0d4636da5aa54f0f6de3de2b96b854d6d0ba4587","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-12-01 21:23:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/902122-2/Lenovo-iSCSI/ : FAILURE in 45:17.98\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/902122-2/Lenovo-FC/ : SUCCESS in 1:53:24","accounts_in_message":[],"_revision_number":2},{"id":"981f54a38766342b087d405c5d67749e3ec8c4ea","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-01 22:36:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_Unity_FC/966f608/DellEMC_Unity_FC/1030 : SUCCESS in 2h 03m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/2/check/DellEMC_Unity_iSCSI/3161db9/DellEMC_Unity_iSCSI/1045 : FAILURE in 2h 20m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f44f0151ad562cf65e632e4a81e4124298e2ebe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 02:39:20.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"aca008e26042d216cb9c25326f07d38e05f384fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 02:44:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"60798e7298faae738df30ec0247e35c4c51be514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 02:54:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2f6a939dcaf30c9c85b45ec6adcf331e5f15a7e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-02 02:54:49.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16454/","accounts_in_message":[],"_revision_number":5},{"id":"86399ca827225f108712255a9e258eaf7410a5ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-02 02:54:51.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16454/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"1a7d6e9ad4be2c622a4bb3069c436527f4ed1c2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 03:20:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3c0ad83bf864d14a36b2067b96cf4e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/516d3a0dd72e481ab8b47da42638f5d1 : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d008cd4a77004035ab64a69b1650672d : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dba122a57f2549c49ae0863931abf4c8 : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"60f609c24c4aa29bdb05a922b5799f0453ed2587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 06:37:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a74aa7766be448108d3ed15707b16def\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7db182c426f14da5949fd0a95e343eaf : SUCCESS in 2m 25s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5b43aeef52f545308c0d276ec5674270 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ac59c27abf481cae22dec0477e1185 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90685efd514f4a56bfc544857bbdf907 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9b8bf94f55444558bea94cf9677a7fb : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c24f6c8fa940b48b11103790095d10 : SUCCESS in 13m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e80b4e8deeda4e589b9030b4bbf04a7b : SUCCESS in 2m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/54a0272e344a41dc8de095e99e5c48a2 : SUCCESS in 59m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ee7884c61e874a409514f15aa126137b : SUCCESS in 1h 03m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d7380acb20645299e8d99f80537aa0b : SUCCESS in 1h 30m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2da2aba43f4e456f92c79c70e1662bba : SUCCESS in 40m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8aebacb562d41488138dd732d483e45 : SUCCESS in 7m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f62179beebe64fc284b7b2327cdde909 : SUCCESS in 19m 01s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/036b335cfc2a411eb06c7c5d33c063d3 : SUCCESS in 13m 57s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f12d511dc11b4231b16edff58466e41c : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3030418f62a04b79ab36579fac10756d : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30e820c4b8934f65bc4fda39973a28cd : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51f10a05c38f48f8bcc2a35872de2dee : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/635640ab1ebe46afb2ba7dc620309aaf : SUCCESS in 47m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/94a3b87c54124489bf8c15c9b83a9283 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/62f267c1d9aa4a6ea55f54e84bfe95b6 : FAILURE in 1h 50m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d823e64385548b6997c133e08ec222c : SUCCESS in 2h 00m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/473968ded5414ed29b2a17529085beef : SUCCESS in 1h 47m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ae713788e6ba4f33a57a78a15ba7ec7b : SUCCESS in 1h 39m 14s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da5b922ac2a74f97a66437a4ab1b3f55 : SUCCESS in 1h 18m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d1269bacdbb04beea01f772374ee84ca : FAILURE in 1h 15m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8eef0a52ee4140b2b2b5aad016aad3c9 : SUCCESS in 30m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/656e5d478bfc4b1b8811f6872e7bcdd2 : SUCCESS in 1h 07m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/29352a0c9c9a4287ad8c01b6f5c43107 : SUCCESS in 1h 28m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da7463ad2a7c4f2d9b11ac4345df471c : FAILURE in 1h 33m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/106cfb5a591b47bcb4022744a0fca391 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":5},{"id":"c83f7458408581421da7787fe7ad9af7974e517f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-02 06:54:03.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3b1007f4c2644af79ae23d5bf1231dc2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/08b262f53956454b87300288b202d0d3 : FAILURE in 1h 24m 32s","accounts_in_message":[],"_revision_number":5},{"id":"c84935bdd31a1ae43b161b3bd54dfeb3f10eec8f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-02 07:14:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 43m 53s","accounts_in_message":[],"_revision_number":5},{"id":"97d33cf5edabcca5b8dd29655a92fcd0233dafc2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-02 09:19:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/22/902122/5/check/fujitsu-eternus-dx-iscsi/6fe2a48 : SUCCESS in 2h 16m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/22/902122/5/check/fujitsu-eternus-dx-fc/d19fa8a : FAILURE in 3h 52m 24s","accounts_in_message":[],"_revision_number":5},{"id":"19dac0442ece04489d2c24c94a60d87b06532e91","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-02 10:00:29.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-3/FC/ : UNSTABLE in 2m 33s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-3/iSCSI/ : FAILURE in 5h 33m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"acaf206727443563b5f3aa3f962200ee4c4da942","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-02 11:02:02.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-4/FC/ : UNSTABLE in 2m 23s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-4/iSCSI/ : FAILURE in 5h 4m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"cffc58eb173b23614c54310a70980567cf0574e2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-02 11:18:33.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_Unity_FC/496f274/DellEMC_Unity_FC/1202 : SUCCESS in 1h 05m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_Unity_iSCSI/9b790c8/DellEMC_Unity_iSCSI/1221 : FAILURE in 15m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b265b00d06578514c1dff789b9041f041e9b5364","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 12:51:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"60c2cd74d88206ccd739e94bf710df62e7372fc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 13:19:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed259f1454ba4906abdec8d993556e4d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aca97b9bf027480caac272c6934f5a43 : FAILURE in 26m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cbef658c8e548a8a0a7efece3a57968 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5c4d92056a3415084e84f67bac4c465 : SUCCESS in 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0e61d0fd9cba54c611b3b83a2dd997f035d780a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 15:14:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca78a7e803a74dc8a686ee086c53a48a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7288ec82cc274568a82744023aec734a : SUCCESS in 2m 32s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eebda97679b148fa89ab54df6892801d : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b403aa1e91f464ea21a6fb9731b8c12 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dbb2021c2dc45fcabe6279006d7fdd0 : SUCCESS in 12m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97f3776850034a86a1ed9f8b14e5deae : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1af13fc2f645bb947eb4f16baaca5e : SUCCESS in 13m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c44e72853d5e4089a4cd9a55e336d169 : SUCCESS in 4m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/529d2705d53b4d9fbd8700d83ce3526a : SUCCESS in 1h 05m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/21b5c2b2465a4c249c5e6cba9b66f809 : SUCCESS in 52m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cffedb8dc17434dbf58952bc0f387f2 : SUCCESS in 1h 21m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bed87b8f4aaf468c979b091ca5e31929 : SUCCESS in 35m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/511956e88b1a4c0597e378cbbeb9405b : SUCCESS in 8m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7391f7f1430648e0a8db9d2f2529b448 : SUCCESS in 12m 33s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3aaf7479919b4eec926cf2f4a431a2f7 : SUCCESS in 12m 04s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a1e6590b7754129b5cd864f67ce0c07 : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5cd20efd1b344bb840e472a383699f1 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0cd6f97d56b141868eeb3bcde7779eb8 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1979433b55fc43e883eb345b4daea17e : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/214359cff6414affbd2406b93037eadf : SUCCESS in 50m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c709cb8072744119bca5ec469e6d0fbb : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87f9fa54b9ec4d349f1bd35e213b9bf3 : SUCCESS in 1h 52m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b0b618bb1f784f3b892674a5ff1f509c : SUCCESS in 2h 13m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69298c34f1ae40c5af9529e02311e586 : SUCCESS in 1h 43m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dd78fc43caa24aa5a56488105a31d54c : SUCCESS in 1h 49m 43s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8123015790714432a6c5df0a88a5b87f : SUCCESS in 1h 09m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/014b304b3603465998cae260aefdf57e : SUCCESS in 1h 28m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da7222ecec31413db32a410a79f051fd : SUCCESS in 46m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/681fc3b6094f434c93421f60265765ae : SUCCESS in 1h 13m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9b4ec531136d4212b2c50bc74cee9057 : SUCCESS in 1h 33m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4146887925b04d108da8f6e3867c8fe7 : SUCCESS in 1h 24m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eea1a0b1bc15415287351ebcd303b260 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":5},{"id":"ea5795b735f5d953e142ad665dc134b2a32da2b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-02 15:53:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54b4e67f8f98417b82e6c8ae045b8a6a : SUCCESS in 1h 13m 58s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/26d4ae2370954bc78c942310cb44018f : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":5},{"id":"cc1972643fe817bcafdd6c1ce257336745646ad5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-02 16:30:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/5/upstream-check/cinder-tempest-plugin-ontap-fcp/003d693/ : SUCCESS in 1h 14m 48s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/2132350/ : SUCCESS in 45m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/5/upstream-check/cinder-tempest-plugin-ontap-nfs/b585349/ : SUCCESS in 46m 08s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4398377/ : RETRY_LIMIT in 7m 49s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"99032aff230d8097d392016cf6f8ffb58e14c01e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-02 17:53:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/568/.","accounts_in_message":[],"_revision_number":5},{"id":"7b4ff79762295dc0b6f3bc984b9b787d1e3815ef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-02 18:19:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/580/.","accounts_in_message":[],"_revision_number":5},{"id":"2d6b33c4af5cadff79abcc3d4c9f1d64a15658c3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-02 21:38:11.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e94d5d089b76413391f1f949afe15d8f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/01013968c0034f8b8048f9a4d5e5178c : FAILURE in 1h 27m 17s","accounts_in_message":[],"_revision_number":5},{"id":"8f55759704d3b4c39e1be1b00ada1c2ac0df09f7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-03 05:15:32.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_Unity_FC/b20c108/DellEMC_Unity_FC/1217 : SUCCESS in 1h 11m 05s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_Unity_iSCSI/45e3ce2/DellEMC_Unity_iSCSI/1239 : FAILURE in 14m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"5d14173f1ffd75173bdec5d9f97c835b38d0f82b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-03 11:21:07.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_PowerFlex_3.6/ed54b21/DellEMC_PowerFlex_3.6/1351 : SUCCESS in 1h 51m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_PowerFlex_4.5/44f31d5/DellEMC_PowerFlex_4.5/319 : FAILURE in 2h 07m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"216d7d4bf66e02318f70b0a253a9084c1cfa43f5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-03 17:01:51.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_PowerStore_iSCSI/450bd70/DellEMC_PowerStore_iSCSI/1022 : SUCCESS in 1h 07m 08s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_PowerStore_FC/520c5c1/DellEMC_PowerStore_FC/1177 : SUCCESS in 1h 37m 56s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_PowerStore_NFS/511bd02/DellEMC_PowerStore_NFS/708 : SUCCESS in 1h 04m 39s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/902122/5/check/DellEMC_PowerStore_NVMeTCP/3aef529/DellEMC_PowerStore_NVMeTCP/574 : SUCCESS in 2h 11m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"65e939a5b27a7a224a3d7e4a540d68890cd33c98","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-08 08:20:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/902122-5/HPMSA-iSCSI/ : SUCCESS in 1:59:56\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/902122-5/HPMSA-FC/ : SUCCESS in 1:36:23","accounts_in_message":[],"_revision_number":5},{"id":"74200e3b74bd3ebeab720262b39768705f5bf92d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-10 11:10:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/902122-5/PVME-iSCSI/ : SUCCESS in 2:06:59\n- PVME-FC https://os-logs.seagate.com/dell-ci/902122-5/PVME-FC/ : SUCCESS in 2:00:08","accounts_in_message":[],"_revision_number":5},{"id":"2811da279a3fb85e07611506db04dde3475933cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-10 11:11:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"271e058a0caec08290e2838c51336b1275018ea6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-10 11:11:52.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16502/","accounts_in_message":[],"_revision_number":6},{"id":"808d39935b471b78c07ffd580064afc71edd64c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-10 11:11:54.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16502/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"64db225a4a2e5b0d88c82ab7e708f15e2d46d2d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 11:27:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5e37502e364cadaae691ddec5396b1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/202c5915dc644605b3f3505cecea4512 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/805247aa567b4a2d8215b17df791b6f8 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e106e95bb24407dbb41cd3f705f0231 : SUCCESS in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"273a11d8d0b8d2a226664d2c775c33b8698d25b7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-10 12:50:37.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d43876184c064cd49898258fe5799d5f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/822caff6655a4aee92c8af93c3788a58 : FAILURE in 1h 29m 05s","accounts_in_message":[],"_revision_number":6},{"id":"c718db0cdd22fbfbebff6490cd08ee6d84a55328","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-10 13:09:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 12m 15s","accounts_in_message":[],"_revision_number":6},{"id":"1a81f70c3e459651a9b7c1bf2bb6c7e03cd20aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-10 13:31:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bacafc0a6d194c159ac560ab89fed6ee\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3b08aca6d6f043d29212e981efaba7b7 : SUCCESS in 2m 12s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/13dad61731464b1dbcceacff42480ad4 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7aa95c5eeec4af494405a452f44a323 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f432dec512b94f3186ff91746a55405f : SUCCESS in 10m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74fc09f352af40a5b67c21523fe3e900 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae2c81e085c41529e4c8590ec19f669 : SUCCESS in 12m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/446063b05af9421eb52aca9de581d3be : SUCCESS in 3m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/8a9de649ad0243a8b0125b6d0742d4fc : SUCCESS in 52m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e61a32a8aaf9409dba39185c85e379a8 : SUCCESS in 47m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abd29e0e700a41788dfa337da71832e3 : SUCCESS in 1h 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df5d94f7ce5b46f7aed4bf93ca57835e : SUCCESS in 36m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26283d3de99e49a0bae490dc1091a7c9 : SUCCESS in 10m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/16d4a12c80554695b681e8ff8cdf7418 : SUCCESS in 13m 52s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cee9b6b31d034a30aac5119aea754fa3 : SUCCESS in 11m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5a01322ccc44025965cca853aa65484 : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/853388113b3d4eadbb464c4eca99247d : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95b1b7af50474c3f97337dd10e3f1905 : SUCCESS in 6m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e37c90b32e1b4175b3254d6c335e03e0 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/593fbf2e254b4218ac11e7dc2bafb6f7 : SUCCESS in 46m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2c823aa348a4759b6eeabc3bc54133f : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/804265c736e3408eaeb5521ef154ca9e : FAILURE in 1h 34m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd5d39df55e24c9bbac7c5950ca64d4b : FAILURE in 2h 08m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9c7ee78afd1b458f83422f022261f288 : SUCCESS in 1h 37m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/db368db799a2416a9d2bf736dd900b6a : FAILURE in 54m 27s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/07f900367d9545c7b745a5fbea3edd12 : SUCCESS in 59m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fa12623a85654e1d8881be0e8d1ca530 : SUCCESS in 1h 31m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6156887b18de4b9991a5ca873db765fd : SUCCESS in 46m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5f5ef414b038463e8fa5d6198e506b9a : SUCCESS in 1h 36m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c650b6d52a514b7284df8497858c8627 : FAILURE in 1h 01m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d8dcf95bec24aebbf4d0c6a75a28a62 : SUCCESS in 1h 39m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d9af963f9f9497e8ad445d2cae3aae4 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":6},{"id":"e6a0b34b53f74a7001e7ba0741533d087e7d24c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-10 14:55:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1acc53700d1040449ccafb6c5c4c9468 : SUCCESS in 1h 10m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/87c813c7ef634ebe955be2b62cfd8159 : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":6},{"id":"950664a59205357eee617e91b12ad961d30e6134","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-10 15:34:42.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/6/check/DellEMC_Unity_FC/d9b8654/DellEMC_Unity_FC/1293 : SUCCESS in 1h 10m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/6/check/DellEMC_Unity_iSCSI/5d998d3/DellEMC_Unity_iSCSI/1323 : SUCCESS in 1h 27m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"0042cd94e66a748870ee6877f67a15d7c6dc7dfb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-10 16:58:22.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/902122/6/check/DellEMC_PowerFlex_3.6/d870ed6/DellEMC_PowerFlex_3.6/1427 : SUCCESS in 1h 42m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/902122/6/check/DellEMC_PowerFlex_4.5/968c3a1/DellEMC_PowerFlex_4.5/393 : SUCCESS in 1h 50m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"63da36120c9847d7862515962834ff01c44eb101","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-11 09:32:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/902122-5/Lenovo-iSCSI/ : SUCCESS in 2:03:56\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/902122-5/Lenovo-FC/ : SUCCESS in 2:01:21","accounts_in_message":[],"_revision_number":5},{"id":"5a8ecc689de6b33501e7ca2acfc3c472063d64e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-12 08:35:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/686/.","accounts_in_message":[],"_revision_number":6},{"id":"68ecd1c7117f6f7961f869895b1a30de1b72891c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-12 11:57:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/684/.","accounts_in_message":[],"_revision_number":6},{"id":"b87b01ebb4d89daee40a9d74aedc097d915f9c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 01:17:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"612a8ca1eaccd04aa258671f13d38e00f9fb4b0f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 01:17:48.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16533/","accounts_in_message":[],"_revision_number":7},{"id":"2170a970aec277a1675c662a8c0d4e2c7bea115c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 01:17:51.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16533/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"5564cfe6ecb72369c586f93563adbf0b0b36ff7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 01:28:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/999bbb804cb344b185546a283063d64f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68f6bd3977d44600a765e49447d219e4 : FAILURE in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0de76aff5ea4b83a5f82575d0965a04 : FAILURE in 8m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37c24f149de04dedb7ab8f12ba17060a : FAILURE in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"53505788639fbec19b426095f56c45c1844a3fd8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 02:03:57.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5f2fb41e9c3242b0805551d908ce2dee : RETRY_LIMIT in 7m 10s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/31b8279dc1564a419634d47df6362231 : RETRY_LIMIT in 7m 30s","accounts_in_message":[],"_revision_number":7},{"id":"62f3ef82ae3979bffb5ecdfcfa9be8ec42bed97f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-20 02:32:36.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ae993e0430334b479df39f140f55a851\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8ea27bc5b0ee44ab8bd885f434774627 : FAILURE in 1h 10m 29s","accounts_in_message":[],"_revision_number":7},{"id":"9ff7ea1be71c3c1b6699236f7919d5c51d10adbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 03:07:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"deeca3fbf4de12d7afb1214a04dca80d9e69172b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 03:28:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74648f7a30ce4317ae2d49fa355bada8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0abe5876ce1b4f7f8f57f53558034a4d : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea1b0da9daf44a3681af87ad5cc65a04 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79a32ea2ecde4bd8bbf3d6b3bfdd6224 : SUCCESS in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3ae29b45944847c1534f953f37918d687958d117","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-20 03:41:08.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e2ef3306b91a4bc68ded6a8231d6b938 : RETRY_LIMIT in 6m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b96e9e55c8a74bd9bb3650c8a1f11e87 : RETRY_LIMIT in 7m 26s","accounts_in_message":[],"_revision_number":8},{"id":"377b55bc9a0d13bd46ae657e6d38fc5aa3c9501f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-20 04:06:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 33m 00s","accounts_in_message":[],"_revision_number":8},{"id":"f5b40b2928e8824b5bf39257e12c1a9dc2b7d957","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-20 04:40:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/22/902122/8/ : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":8},{"id":"95972a3c43c385f042826776371b505b7fcedaa0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 04:58:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/748/.","accounts_in_message":[],"_revision_number":7},{"id":"0a405f733e26db0d2819aedf5f05c35a03661505","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-20 05:00:13.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/902122/7/fc : SUCCESS in 0h 36m 59s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/902122/7/iscsi : SUCCESS in 0h 49m 41s","accounts_in_message":[],"_revision_number":7},{"id":"d611e3cc50afd442d53013d990b2ce71376b90aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 05:31:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9303f6211a24c899282c43c28b0bf0c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/37ae9eb5776b4ffeb5c7873b74d63be3 : SUCCESS in 2m 01s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd33762209d244d3b0e36bced25b028a : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbfe51855b04b26848a99f161d37dad : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be01cc4a424546fab5fdcfbb71c4ed42 : SUCCESS in 10m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85c8b8a75ef6417ca425314728980146 : SUCCESS in 14m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc075a67bb7e44dcab4063e9274a3990 : SUCCESS in 14m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c69173c10a0410cb4087b992aaee332 : SUCCESS in 2m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e0ad15cdaf7a4cceb613ae0691274dcf : SUCCESS in 51m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/25e550f6342740a89d2661a9b7a8213e : SUCCESS in 1h 09m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1bf015c980694cf683ea2496809e2199 : SUCCESS in 1h 23m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1499a257023b42fcb2b3d11877db0723 : SUCCESS in 48m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bccc97b8d8744c4a398920097784a82 : SUCCESS in 12m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68f60c159b04453a91494fad33d5fdb3 : SUCCESS in 17m 47s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bbd3481a81ac493a808a434698b6e835 : SUCCESS in 15m 11s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78beaa507f104683882337e1671df850 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/526024b5662b446e8f535a99bb1d59c1 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/842a7f86ea9d407dafa5cbaadeea9955 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e571dbe7508841299eedba7a81c09e00 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/547c97beabf3483db3adfbdf6ad7462f : SUCCESS in 45m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40ff69fdcfb0432f9624fe9ecf879024 : SUCCESS in 5m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d9cf29d8c75f48c3bbf052f5695ee3d7 : SUCCESS in 2h 10m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c7a4803c6844dc1895199171b71a2f5 : SUCCESS in 2h 12m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36f1087093314b46b33bda72cbfccbcd : SUCCESS in 1h 54m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/99e64dfd44624748b34cf75699857556 : SUCCESS in 2h 06m 01s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7b0ff7107d764a25a298349f7d83b3d3 : SUCCESS in 1h 11m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/30741d12bcc1403b863a56b8cf5c0130 : SUCCESS in 1h 15m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c6b5cde3febe4b7a9dd4d69834340614 : SUCCESS in 29m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0305d9861aec4dffb28c1b1dcb960239 : SUCCESS in 1h 24m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a7adf6345861407397033b3ebe43c8f2 : SUCCESS in 1h 17m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a114ebac627540c893071dc757d7b194 : SUCCESS in 1h 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac15c4965e644c07b1a7ebc10b3b2f8b : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":8},{"id":"0ac7a3bcdb5eeca84ce0c620bc65ea4992e6699e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-20 05:36:51.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e30da45bfdcd4fb392e30f6f0a03ea3f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/888a86251b174e15bc9a808d6dd9a816 : FAILURE in 1h 29m 25s","accounts_in_message":[],"_revision_number":8},{"id":"67d0fccff5d14aad2117058d1effe00b2b230e21","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-20 05:50:24.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 35m 48s","accounts_in_message":[],"_revision_number":8},{"id":"dd25e3f5db15f8dd2e473af4ddab358e3636e085","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-20 05:58:03.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_Unity_FC/92d100e/DellEMC_Unity_FC/1364 : SUCCESS in 1h 10m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_Unity_iSCSI/81e65f4/DellEMC_Unity_iSCSI/1394 : SUCCESS in 1h 29m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"e20eeca328a3aaf27b15064b20f128e2ff22b9d6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-20 06:59:11.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 1h 04m 42s","accounts_in_message":[],"_revision_number":8},{"id":"583b1f0bd8aebe53685a138ffc988c7d9ad97e5a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-20 07:15:24.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_PowerStore_iSCSI/2e59502/DellEMC_PowerStore_iSCSI/1097 : SUCCESS in 1h 13m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_PowerStore_FC/1ce6864/DellEMC_PowerStore_FC/1253 : SUCCESS in 1h 33m 24s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_PowerStore_NFS/15437d9/DellEMC_PowerStore_NFS/783 : SUCCESS in 1h 00m 07s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_PowerStore_NVMeTCP/371a38a/DellEMC_PowerStore_NVMeTCP/652 : SUCCESS in 1h 43m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"f246255c5ec587183bda81d491dbc00bd6cc852b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-20 07:26:40.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_PowerFlex_3.6/381dddb/DellEMC_PowerFlex_3.6/1495 : FAILURE in 9m 48s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/902122/8/check/DellEMC_PowerFlex_4.5/00ccb1c/DellEMC_PowerFlex_4.5/459 : SUCCESS in 2h 07m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"4bd31336b625896bcabb440ad7c6862a92a61504","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-20 07:44:25.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/8/upstream-check/cinder-tempest-plugin-ontap-fcp/da0c5bb/ : POST_FAILURE in 59m 07s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/e94469c/ : SUCCESS in 44m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/8/upstream-check/cinder-tempest-plugin-ontap-nfs/f93c407/ : SUCCESS in 2h 11m 39s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/afef15d/ : SUCCESS in 2h 08m 49s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"3ecf052bcba12cd7d86607720b223a211d4130dc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 08:39:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/750/.","accounts_in_message":[],"_revision_number":8},{"id":"480bc014d092717b4d1786d8d187a3a89e3ed2e5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-20 09:25:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/741/.","accounts_in_message":[],"_revision_number":8},{"id":"7fcd68805f0bb91723ecbab1d9a6a1ecce198566","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-22 11:02:27.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/8/20240222164749/NEC-V-FC : SUCCESS in 3h 02m 56s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/8/20240222164749/NEC-V-ISCSI : SUCCESS in 3h 14m 34s","accounts_in_message":[],"_revision_number":8},{"id":"921be5543f5d554dae013f04291f04234c48d32a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-23 12:05:15.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/cinder for 902122,7","accounts_in_message":[],"_revision_number":7},{"id":"79735629a1ee025f2b1e5ab3bf7f319ba3a09b2c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-25 02:37:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1b6ea5248413463eac561532d11cbab1 : SUCCESS in 1h 15m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/146b8dfe890a4f9f8cb4f26945163729 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":8},{"id":"16211231296227ea6b95129071bc9ffc6d238748","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-25 05:20:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/902122-8/HPMSA-iSCSI/ : SUCCESS in 1:55:12\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/902122-8/HPMSA-FC/ : SUCCESS in 1:36:50","accounts_in_message":[],"_revision_number":8},{"id":"9cdd6edb6b70ab51824ab436081c43e19453440a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-21 04:45:27.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"78df2695f95e49c5c9b1d73e6d0d25a0ed8398d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-21 04:45:31.000000000","message":"Topic gcs-cleanup removed","accounts_in_message":[],"_revision_number":9},{"id":"f29f182112dbe99956bf2ae63b133c92f5eae8df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-21 04:45:36.000000000","message":"Topic set to bug/2058601","accounts_in_message":[],"_revision_number":9},{"id":"0aaaf9e154571159466a7af94430d4e03bc4a1a3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-21 04:53:50.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c00e1d0cdc984462a44f5818c3e5746d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/407d32aecca54661ba044961153d6bb9 : NODE_FAILURE Node request 200-0000002856 failed in 0s","accounts_in_message":[],"_revision_number":9},{"id":"ddb23ddf92d96e8b5283a2ccc967d1846c5eb797","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 04:56:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07d31a8a8ad444c3993f7fb0608ce845\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa0774334f4b44a4aecc4db8079c0ed8 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d90bbfec73f24634ae9c6153dacf937e : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0802a0ea40be4ce4899fb8986a56ff8a : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7bd900088cfeb03c3c1fc6e6d4b61251e17926ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-21 05:34:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 47m 08s","accounts_in_message":[],"_revision_number":9},{"id":"21a6ffaecfdc2ac6299217524a02cd924e13878c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-21 05:36:42.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 49m 05s","accounts_in_message":[],"_revision_number":9},{"id":"76755ab25a9a1eca97aae843d097fba4a1ae2ccd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-21 06:03:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/108/.","accounts_in_message":[],"_revision_number":9},{"id":"ee77f5af0b820e8a11942aadd1a6a12013ecc0fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-21 06:06:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7acbf0f5cea14c4594f8f2ae11b89316 : SUCCESS in 1h 18m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7ba4862c4c664ad88dc6289fd456a842 : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":9},{"id":"6aedfbc7411586517dd09aa5293157f6a131fc64","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-21 06:29:43.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 49m 07s","accounts_in_message":[],"_revision_number":9},{"id":"ce9ac4604729135706ebaa10fe59cba4cd2e2cd5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 06:39:05.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerFlex_3.6/29e0727/DellEMC_PowerFlex_3.6/1683 : SUCCESS in 1h 48m 57s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerFlex_4.5/d28ae19/DellEMC_PowerFlex_4.5/620 : SUCCESS in 1h 53m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"5941379c09a1d255fcfea4d9d8be5c10e8a0ef70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 06:51:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e605246346bf43d496a4ac4a3c36d5b6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a8467aaf82343af8ee88e098534b324 : SUCCESS in 2m 03s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59bbde2aba3b4575b15c07d00a5e1df4 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f33ca801c24c12b8b30a2d1218b138 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f73be58602404c478ba0c08343a64ffe : SUCCESS in 10m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0b93ccab37b44a1838703269ac7eb5c : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac486f05d2d4d3e9068c5b2c67eb0d8 : SUCCESS in 11m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b92b68ade01474e999bd631c2730850 : SUCCESS in 2m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a1a45f29488549e19d25211cf42c6711 : SUCCESS in 50m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ed10e1860e184d35b23830276d2eb406 : SUCCESS in 47m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cfd20c1ee42408f9a91d36d1bea3c3f : SUCCESS in 1h 32m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d4c79b19b8c45f4a47f47e817d02374 : SUCCESS in 52m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79b410fbb10e42078fda4dded6653475 : SUCCESS in 12m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca1fb25ed4904a9094a074827a9f32e9 : SUCCESS in 12m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/da6688983cb34d45a5466f7ed1ecbd5c : SUCCESS in 13m 06s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a1389e553eb48d8929ab893108f8564 : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7de325a61e3a4052a51226e8f3c8fbf8 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a8fe83753894551bb2fdc720350d4b1 : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efada1f2baa641e9a83838d98529d51f : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5dda9a180f94ff7b1366da35605f611 : SUCCESS in 37m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7bd1d0e39fcb4b02ac1d393ef7f0ae9f : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/311f10b66b584700b057df6f0702ef36 : SUCCESS in 1h 57m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb9e9ee115a64335845cc2f9146c2ce3 : SUCCESS in 1h 53m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/87970b9efd454f88990dcc7ce746f926 : SUCCESS in 1h 36m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b2d3f7931b2b49f3b9eb90be29a0d590 : SUCCESS in 1h 33m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d772518424ef4dc7a2ab0f893191af1e : SUCCESS in 24m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c5b24e50e7f484097e947f0fc69f31e : SUCCESS in 53m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29a0b0066ce34780bcb546d1afa31be8 : FAILURE in 1h 33m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b6f90a4b13f14b048a672b614842ea4e : SUCCESS in 29m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c20d7d301bf349a4ae98aefaf64c2897 : SUCCESS in 1h 25m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a143cec020104fb4ac6c3193bb95da80 : SUCCESS in 1h 11m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe74be613369412d8a07afdc47f9c3a3 : SUCCESS in 1h 38m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c015a9dc1d84438a4e0974aac42f3a7 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":9},{"id":"2f7b23e45a6c6b9c1617b29d70d4434023e3ed27","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-21 06:57:47.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_03_2024_06_51_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"cd48fb7861356253c18163c67f3e4c74fc27e745","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-21 07:12:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/9/upstream-check/cinder-tempest-plugin-ontap-fcp/37abc5b/ : RETRY_LIMIT in 4m 03s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/37c207b/ : RETRY_LIMIT in 4m 11s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/9/upstream-check/cinder-tempest-plugin-ontap-nfs/867b1b0/ : RETRY_LIMIT in 3m 59s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9c59323/ : RETRY_LIMIT in 4m 18s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"aa8abb004d5d0dd4be8a35791fbf69a76de19b86","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-21 08:44:28.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerStore_iSCSI/7e26c8b/DellEMC_PowerStore_iSCSI/1196 : SUCCESS in 1h 08m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerStore_FC/ac50b7d/DellEMC_PowerStore_FC/1366 : SUCCESS in 1h 40m 52s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerStore_NFS/70311f7/DellEMC_PowerStore_NFS/884 : SUCCESS in 1h 05m 43s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerStore_NVMeTCP/be77997/DellEMC_PowerStore_NVMeTCP/746 : SUCCESS in 1h 52m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"25ea264d85a6a7bbe2b4fe5deec69514a03489cb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-21 10:39:22.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_Unity_FC/1640454/DellEMC_Unity_FC/1563 : FAILURE in 1h 36m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_Unity_iSCSI/01ad9a5/DellEMC_Unity_iSCSI/1595 : SUCCESS in 1h 29m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"463d475429e95d90d022ac1f2dc1794ae46beeaa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 12:06:55.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerFlex_3.6/27a53aa/DellEMC_PowerFlex_3.6/1686 : SUCCESS in 1h 55m 26s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/22/902122/9/check/DellEMC_PowerFlex_4.5/f3700eb/DellEMC_PowerFlex_4.5/623 : SUCCESS in 1h 44m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"8d91e86c06ae2db9abba7b26b0a408344f301994","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-21 12:25:54.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-9/FC/ : FAILURE in 7h 40m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-902122-9/iSCSI/ : SUCCESS in 2h 2m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"ef5d3939e13c7ba5955996c6c170719c8b69c2c0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-21 18:00:30.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/902122/9/fc : SUCCESS in 0h 51m 50s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/22/902122/9/iscsi : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":9},{"id":"56bed96076f43e1f3aceff3062a7c5510018f3bf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-31 11:34:27.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/9/20240331100712/NEC-V-FC : SUCCESS in 3h 17m 28s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/9/20240331100712/NEC-V-ISCSI : FAILURE in 10h 27m 12s","accounts_in_message":[],"_revision_number":9},{"id":"403aa30667ad1e1978a70b8a81178bf276c4959a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-31 21:32:21.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/9/20240331215528/NEC-ISCSI : SUCCESS in 4h 32m 36s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/22/902122/9/20240331215528/NEC-FC : SUCCESS in 8h 36m 50s","accounts_in_message":[],"_revision_number":9},{"id":"abebda48fb0dc54284a8f633f7dec8062da6d99c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-09 14:22:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"45c60cf19ea9cbff4814fc425ee4af2e7aeb9a22","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-10 03:06:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"41f1aed7fe5a231482abb52ba46031714663abea","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-17 12:49:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/902122-9/Lenovo-iSCSI/ : SUCCESS in 2:08:35\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/902122-9/Lenovo-FC/ : FAILURE in 3:34:50","accounts_in_message":[],"_revision_number":9},{"id":"a770319387380e32e63a69c7c7181549b0bb10d7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-17 14:47:45.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"17927ec96b718af5381f919f180e438deda21fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-22 02:40:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"d914ac94dc43e7bf73f841bdbcec6f831e1cb9d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 02:40:11.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/362/","accounts_in_message":[],"_revision_number":10},{"id":"e9817365db287a1bbb7517bc81110745776be3e2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-08-22 02:44:20.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/362/, iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/10/iscsi.log and FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"52b00614f307c48503a5f86057fb99c82342bd67","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-08-22 03:13:43.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : ERROR Failed to update project github.hpe.com/hp-openstack-blr-repos/Devstack-gate_new in 14m 59s","accounts_in_message":[],"_revision_number":10},{"id":"773ea28b4a1a75e8ea091416c6b99e525919754b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 04:14:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0fd87ec9b12483cb40eeb4d7841dab5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b573f9ebdb344268bed0987bfbcfcb8 : FAILURE in 20m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e3a269fb03e403984dc5dd9e19476b2 : FAILURE in 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6ccff3fb7708597543ae26ea929f8816093d7e34","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 04:20:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/373/.","accounts_in_message":[],"_revision_number":10},{"id":"9210f287207a2291f364e01a115d8c09d156f87b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-22 04:21:09.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c499a4e88cd34daaa938022de8267354\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca6e594605524a1d86c8ab3aa68b64b7 : SUCCESS in 34m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/def3309139114452b841de597770d2d1 : FAILURE in 40m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b6157798f084e909a14638b3ff6deff : FAILURE in 1h 34m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6ca0f9167b5e4f16bcfa80884edcf76f : SUCCESS in 42m 36s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#67\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#81\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#16\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#38\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#157\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#369\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/de... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"5f1b27297d7cab6f28305a43a1658bdc5aa0c008","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-08-22 04:25:29.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":10},{"id":"86c5b34bd5c2d1ffbdd2d2b594633a8c652edb66","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-22 05:07:20.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/26c7507c12df42f6b9a77de245794c88 : FAILURE in 1h 10m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c9eaadb41ffa40d2811f73d6bf632ff1 : RETRY_LIMIT in 14m 01s","accounts_in_message":[],"_revision_number":10},{"id":"98926a7adca641d1dd6d4a6f1fd43e45903e5f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 05:14:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4837161646e94f258caaf25dd0c0d1ab\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/301cab0da5dc4868970292cb6f8f0c32 : SUCCESS in 2m 22s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8169fa209a254367a164144d9b398fcf : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f72d80ead6d4f14ba397518ce705547 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd603766a20f4d2a976b7362ec31a010 : SUCCESS in 11m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fa3eac5835e4daa8739c2aeabc5205a : SUCCESS in 10m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cac460f9bc24e39aa866313e4ac667c : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e86583aac14bd8a26920c1d9e9aab3 : SUCCESS in 12m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/13521d69681749b19eb5f06340afa857 : SUCCESS in 3m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/484ee5f49504466abd690be0b7855dc2 : SUCCESS in 57m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9337a234422d40619be7be63492caf64 : SUCCESS in 1h 26m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c666cb55c0ed4c038297a7a74bea7885 : SUCCESS in 55m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9da299664d845f7899d6506605980a4 : RETRY_LIMIT in 4m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/834c8db4a6ec4613982b6b25b65267b5 : SUCCESS in 20m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/96747d159829416aac70a702d10b60ab : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e206f2df86034b29ad384c02654e8fa1 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be4fc16fd8474fd08384d92d23a9af9a : SUCCESS in 6m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0611e1c5065647c2b0478861bdd74743 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e0ce804ddb4f42168b8022daaa9c4f31 : FAILURE in 39m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/762caeb726654a189d6c213e7ba2f2f9 : SUCCESS in 7m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/46f01d70a34044b0984f3a42d6071242 : SUCCESS in 2h 00m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2016df2493574d9185d488b5d0f52bf7 : FAILURE in 2h 26m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4cec932b8fdd408faf0547e524bce0eb : SUCCESS in 1h 43m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/46bf62e63b78486cb0aa6659f8095c00 : SUCCESS in 1h 55m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e54d2b79f3b944ae8c889282139ae6a5 : SUCCESS in 29m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/44c85806d80e468d9e158252e7da7e0d : SUCCESS in 1h 13m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ddbd20ff7195455a813835a44584277d : SUCCESS in 1h 25m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9aa020a2ea2b497a97c03e0043b87824 : SUCCESS in 34m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/620a22fd700e48b284f5908b43dfe793 : SUCCESS in 1h 42m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a46e8759560a4017b1a5eb4f96376eba : SUCCESS in 2h 00m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8bc99d0bcd2b46f78e9a375a12690aac : SUCCESS in 1h 38m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/532cae973ead4045ba8dde6f65a24695 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":10},{"id":"429e5e8a79d7d404213dfdfd8ef7f0e339b079d1","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-08-22 05:15:40.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"905a25ae482c584e27b5b32cfc2793ff16cfce03","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 06:46:56.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/961/.","accounts_in_message":[],"_revision_number":10},{"id":"7f86225387b021f7ab2339ec767898164e368e8f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-22 11:21:30.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_Unity_FC/a73abb2/DellEMC_Unity_FC/2475 : SUCCESS in 1h 46m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_Unity_iSCSI/3924be7/DellEMC_Unity_iSCSI/2479 : SUCCESS in 1h 58m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"104c03fdafc6f9c6ff9358359c458663232476fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-22 14:27:53.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"224aaedd094a3eab2d80ec06746ec1d324c04f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 16:44:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde3b68bd25a4b6ca9bc47bbe2a8216c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/06bce4ee9561475291c402b678d8fd93 : SUCCESS in 3m 04s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cbd96df5c82948c8b3ae9cfb6f77f93e : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f76daf1b277452599267adfe8830115 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/751c0b08764e4d6bb7c1335563817be7 : SUCCESS in 12m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ec1d8101ed34e9a8641b0a675fdc811 : SUCCESS in 11m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d68918a565a147e486c93991fece9140 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f270a16d0a94fbe859848696f8deb49 : SUCCESS in 12m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cfdb87f9a2244e598eca30fb12c121bb : SUCCESS in 2m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/24ed6de48e9c4ccab2dea7381b7a2d21 : SUCCESS in 1h 04m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee5f8de333534f2b9a05ac5fa97ca2db : SUCCESS in 1h 34m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9816ccbd5f044ddc8c7cbeacf3219afb : SUCCESS in 1h 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0148c56370f94a589ccfc9f4682d757c : SUCCESS in 18m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e6bada7eda1f465ebae8a3c833011f2a : SUCCESS in 18m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b4b734549ac42168544f109d0b2da26 : SUCCESS in 6m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/468d3d4544f64ebc8845dffcaa276c2d : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5ba0c81113214db6b3a0b40704b8fb6d : SUCCESS in 5m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/702e6674f6d44fe08b6da9a7249d031b : SUCCESS in 9m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1d48a313b6c45f3b772fa212c56d85f : FAILURE in 27m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1355958268ea4ef48d62d0b355cd9b44 : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/25467a32fa7c4c259a87002ce30b6a1d : SUCCESS in 1h 55m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/543142fb4c05479283747aa57bd28c15 : FAILURE in 2h 13m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1c1dde56c384d97acea5d5015181e19 : SUCCESS in 1h 43m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f1448e4ea15b4c14b795ab16cc4e462b : SUCCESS in 2h 14m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ea30c5f7af6342e7be398ba746fd7261 : SUCCESS in 47m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e61bb5465f804ed4850fb0edd9bdb445 : SUCCESS in 1h 06m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91341b190e0840da8b9b9fc0c25f8675 : SUCCESS in 1h 25m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c476bc4c3f2f461a9c87b4dfac98b52e : SUCCESS in 1h 05m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/202d7f17776048ed8c84148d2897f4fa : SUCCESS in 1h 33m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/53756afb740c4f3786036d64eb4ce857 : SUCCESS in 1h 29m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53b6794656f4444eae9bc89024e4c079 : SUCCESS in 1h 39m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eec03e561a08462ebdb57a263a0f9852 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":10},{"id":"a76e8e2a1d134a8f1769457380b131ada337906f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-22 16:50:26.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2024_16_44_52\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"0696ee8526969aedce30d4cdea7af5cc0f410b34","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-22 17:00:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/157adde/ : RETRY_LIMIT in 3m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/10/upstream-check/cinder-tempest-plugin-ontap-nfs/6fb2912/ : RETRY_LIMIT in 3m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"2dece005bc622d3108b255b776b9cf8bb3654616","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-22 18:36:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a539f82882944eaca0596d3aefe156f9 : FAILURE in 1h 18m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d7c439214f074ef592b6fd1de656b17b : SUCCESS in 1h 13m 30s","accounts_in_message":[],"_revision_number":10},{"id":"8ddb55c0dced75e9de4627fc796d929bcb6b14bb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-22 22:20:24.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_iSCSI/b9c2ce1/DellEMC_PowerStore_iSCSI/1632 : SUCCESS in 1h 41m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_FC/a9b2aaf/DellEMC_PowerStore_FC/1828 : SUCCESS in 1h 16m 45s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_NFS/8c319c6/DellEMC_PowerStore_NFS/1325 : SUCCESS in 1h 16m 11s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_NVMeTCP/a3b3a0d/DellEMC_PowerStore_NVMeTCP/1184 : SUCCESS in 1h 54m 03s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_NVMeFC/c4613ec/DellEMC_PowerStore_NVMeFC/149 : SUCCESS in 1h 07m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"db0389396b4b3cf5da730fbdd3d6de4e72199b8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 00:10:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ae7b253a8842d7b6a365430bf00775\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57aa14bae0e6466a8bfdc4684845f1ba : FAILURE in 29m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb46963900ee4d6dafbc843675d27e96 : SUCCESS in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e24fa3bd191265f5f3095afb3a5589058efd0573","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-23 00:15:56.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/e93f058/ : RETRY_LIMIT in 4m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/902122/10/upstream-check/cinder-tempest-plugin-ontap-nfs/46d05de/ : RETRY_LIMIT in 4m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"4199c7bc1ecfb3814c7b079455ba13a326d5f396","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-23 12:59:56.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_iSCSI/de47e74/DellEMC_PowerStore_iSCSI/1637 : SUCCESS in 1h 50m 53s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_FC/4fcf07c/DellEMC_PowerStore_FC/1833 : SUCCESS in 1h 11m 44s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_NFS/d1c298a/DellEMC_PowerStore_NFS/1330 : SUCCESS in 1h 07m 43s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_NVMeTCP/a478baa/DellEMC_PowerStore_NVMeTCP/1189 : SUCCESS in 2h 23m 52s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerStore_NVMeFC/37e8566/DellEMC_PowerStore_NVMeFC/154 : SUCCESS in 1h 16m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"c88f17f261d7a444ff9f68b3e7bf00cf87aa120c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-23 13:20:24.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_Unity_FC/83b5f7a/DellEMC_Unity_FC/2485 : SUCCESS in 2h 18m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_Unity_iSCSI/9b6d819/DellEMC_Unity_iSCSI/2489 : SUCCESS in 2h 25m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"72cf79ec6715b3b5d11e53a7ede69582f43f50ff","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-24 09:08:01.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/10/20240824141003/NEC-V-FC : SUCCESS in 3h 44m 45s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/10/20240824141003/NEC-V-ISCSI : SUCCESS in 3h 57m 55s","accounts_in_message":[],"_revision_number":10},{"id":"ebd75245734e0e507651a4af99d758959b5c45ea","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-08-25 04:53:34.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerMAX_FC/57f6a04/DellEMC_PowerMAX_FC/437 : ABORTED in 10h 30m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/22/902122/10/check/DellEMC_PowerMAX_iSCSI/ab1a7ab/DellEMC_PowerMAX_iSCSI/459 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":10},{"id":"9b7af9117c76c40f3ddebd9536d0b8c7c1f0c026","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-01 19:03:20.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/10/20241201221351/NEC-ISCSI : SUCCESS in 5h 23m 51s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/10/20241201221351/NEC-FC : SUCCESS in 5h 49m 26s","accounts_in_message":[],"_revision_number":10},{"id":"a2c7e0851f2f7ad0451fb00984a9f01055b56d0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-27 14:18:19.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"14072bbdcef7d648d1ffb6ec678ebe5a18714cb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-27 14:18:34.000000000","message":"Patch Set 11:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1375/","accounts_in_message":[],"_revision_number":11},{"id":"1092ae49c9aabbb46e4bda0078c283a55dac98d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-02-27 14:18:35.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"ad3c111912b34754af11a8806792a9c90e14d54c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 15:08:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c18216309e6404db204695c44dd24b2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f2b5a3e0d3543d187fcf6fdb3ed3234 : SUCCESS in 19m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b66ef9d7a8354c0cac2bd2f5e128d5f7 : SUCCESS in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bd19a11ea0547a9378d2827564d364a4a22d1451","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-02-27 15:49:58.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 38m 11s","accounts_in_message":[],"_revision_number":11},{"id":"020fe3da153ce7dd1fe00966e26ab0ca5c4df53e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-27 16:01:50.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/014bf1bb0da74ed5a253d281070ac331\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f646ed60042741ae8cd2102baa765b3f : SUCCESS in 44m 29s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52630fb97507452b8c9b62f4f93bc1f3 : SUCCESS in 40m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/734768292fb444d59ca9d9752c862d69 : FAILURE in 1h 39m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6bc4abfcfff749d78be80311859880e0 : SUCCESS in 42m 58s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40d74ad7339f45b19ee0a96c80c47f87 : RETRY_LIMIT in 34s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13eeadfc58684000b0196a842852f9d7 : NODE_FAILURE Node request 200-0000026920 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/763a9439ec214d99927bd12acefefbdc : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b99e3ec8db3540fea5fde5307964c0dd : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devsta... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"eb4f384f985bddece5f3495936ecb96ea8e755b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-02-27 16:13:57.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 38m 03s","accounts_in_message":[],"_revision_number":11},{"id":"581403f5ceca06ccff5ae1cd3e6e88e918660ba7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-27 16:53:37.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/97ef08dc7ca541a6a6e11e70fd9527ad : SUCCESS in 1h 13m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9b4ec7d450994eb896bd6e5c36bb107a : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":11},{"id":"6b594a6efc3bf72ce5e150287090ccca8b7bcc68","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-02-27 17:03:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/404/.","accounts_in_message":[],"_revision_number":11},{"id":"f05b97b28a9c02ce22d1db94fcf9e4b8570cf8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 21:40:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81ac10028743424689495de6db6a1440\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e573217f1b44d798f7ea86a44c811c2 : SUCCESS in 5m 41s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/161ee8170a8b4a8c9e5434a924c15fc4 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4360929d77c64a63ac11a329c0e19e6b : SUCCESS in 20m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1fa704ba0a04915aa37abcee9dc21ed : SUCCESS in 12m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78d73c5834a94975a3c6f0be66461693 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c52cc8f8e474ab7acd2fe99c7546dc6 : SUCCESS in 14m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e41745f5e26742bf8de01da39920ae58 : SUCCESS in 3m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/36a9e4f878554f1f973dd8dd48e8eea0 : SUCCESS in 55m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9785c07cff4e4e8cb2c5a7f51a137d37 : SUCCESS in 47m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d910c9e811446d599b8ac076a51dbf5 : SUCCESS in 1h 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a2b9d45434d4a60b4f23fd62587d8fd : SUCCESS in 1h 53m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/305a8354db6541669d2e096749f9693c : SUCCESS in 33m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c26e53fb74242a7b9440f5b990faa8d : SUCCESS in 30m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc9fdd5c4a3f4d32989245ecc7af759a : SUCCESS in 7m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/225285a57f4044f5aff36619b1d4159e : SUCCESS in 24m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c855675101c49ba96c73aa62a0e67e9 : SUCCESS in 26m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/165b77cba12a41e88e690ba429822f75 : SUCCESS in 29m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0da2fb4661f343b79edcec0def4d10f0 : FAILURE in 23m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae175897e5044666b5cbea19dad485fa : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa06f532b46a41ac9ee5c32e9995ce95 : SUCCESS in 3h 03m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3751e81fdd3b4a7b94464ccd2578d0ec : FAILURE in 3h 06m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a7df74c5c7549ebb088e822e83c34bc : SUCCESS in 1h 49m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1f177ab2080d4e35a07acae021070143 : SUCCESS in 2h 24m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8d1069ada47f41f897c157d84a8700b5 : SUCCESS in 34m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8704a09861064c6d8c2f9b894a0d90de : SUCCESS in 1h 11m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5257f955b3254404b99e957cd3dfb760 : SUCCESS in 2h 15m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fbf400ad592b4b0091521dd5d8ebfb96 : SUCCESS in 1h 35m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab69968f857543f6aa6d6ce336e05645 : POST_FAILURE in 1h 48m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5369c7bb75a8486c880b7482931e66dc : SUCCESS in 1h 50m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d90acb3a6c847f4a2621c715af48124 : SUCCESS in 1h 33m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7300678a3c29431da63b20787610e2c7 : SUCCESS in 1h 22m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7274ee8fe8d34e26a598ca2f86f02d6a : SUCCESS in 47m 54s","accounts_in_message":[],"_revision_number":11},{"id":"30cb524ef93e03faff8b0bc743e2280cbfaab740","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-02-28 11:42:01.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/11/20250228104526/NEC-V-ISCSI : FAILURE in 8h 28m 25s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/11/20250228104526/NEC-V-FC : FAILURE in 9h 56m 31s","accounts_in_message":[],"_revision_number":11},{"id":"efaba04218fb10803a9f2d221211adaca244d702","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-28 13:59:18.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6d4a770fe49a3508a80026c338a22bca3bdb125a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 14:22:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09cfde11967645ec9085233744bec1e2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b7fb7a9033c416e95fab868bed4a495 : FAILURE in 19m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3191444876804dc8bfbfe57018f23620 : SUCCESS in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a513cfd3b8efebfa467828f1eb8426905b23d078","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-02-28 15:39:46.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ec89ca5d5cb04c749fbf8f1afe32be65\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddbb770b0a314786947f24adfc77d62a : SUCCESS in 41m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b8306ac239644007ab6536c877d4fede : SUCCESS in 47m 12s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/755697867aa94cf8979f593b551c410f : FAILURE in 1h 35m 50s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1e85709941a46d89937552a55478066 : SUCCESS in 41m 59s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f099d1ff5b994973a811da953474a848 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7006ed0cc0c448a3baa3d75fbe1642e8 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2489384b2575493380d96f5f85a3e98b : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/057f62cb773b4497b77660294d9d5f07 : RETRY_LIMIT in 28s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstac... (truncated)","accounts_in_message":[],"_revision_number":11},{"id":"0421e0af107680e19cf13f1ad1cf8350b88f5332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 17:12:39.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdaad91663aa44279b944b77517b2950\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec81f5a5e97e4f9c9e4f41538ed4466b : SUCCESS in 2m 40s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d69cb80428842de8650a1573b4eb571 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b109e95a1887415182cb8ec7ded159df : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01574543867142a1bbf93c80e5e08603 : SUCCESS in 11m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd9babb608dd4ab584dda66090a346e7 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f832acedf69d4ed1841189968c5b8bbc : SUCCESS in 14m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4cb08f98baf41139561364e1dc8aef5 : SUCCESS in 3m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/5433d80004164c93b597989f22e5b29f : SUCCESS in 58m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbcbade88f2941e2b49378679b8147db : SUCCESS in 1h 05m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01df9d5235e6430180f467ef30181849 : SUCCESS in 1h 34m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a5b86b7cf344d63bd75d88a64537365 : SUCCESS in 57m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18cc09df814048449f59e80ff9cb8b40 : SUCCESS in 19m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0afa0d3fa9ff46df9638c245ca01ae94 : SUCCESS in 28m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5633c69929224904ad1f28078b2a3c60 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5346e792c4d04a92abd0c93a0a782973 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cedcd362b5724e72a89dd2ccca560e4e : SUCCESS in 6m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/95abdf45c5ac4a5ab2e5bff927076973 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3bbcfa9c11864cbe8ed4e2cb033e04ea : FAILURE in 46m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d153a81dcd94d5ba1fbc4bfe2563314 : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d9c16fb1663c495bb16f8817610a3eff : SUCCESS in 2h 25m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d5802a74cbc348f59f7a38d054cee664 : SUCCESS in 3h 05m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b01c1faa34b4480aaa8ad55692f841dc : SUCCESS in 2h 08m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bf01d93e1fc540e2ba7c835eec075117 : SUCCESS in 2h 13m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b6493f2affbd46d2978c031c0cb323d8 : SUCCESS in 53m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2aa6bb46a34543ed85e8192d5220dcee : SUCCESS in 1h 42m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8e06b66a9014914af57b044f39e3e0b : SUCCESS in 1h 47m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bd927c191fc74a4392d8b334fc20bac8 : SUCCESS in 42m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9ee3b084767a46b3b47905a288dc90cb : SUCCESS in 1h 42m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0a360897c8e84e68855d203d65c0cdeb : SUCCESS in 1h 55m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76aae12d3b7743b8910eb1a25b5a103e : SUCCESS in 1h 40m 25s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09955cb5ebd3498e8d2e23a7eb71bba3 : SUCCESS in 1h 32m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4124e628cfe4216b7d95a2641c88d05 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":11},{"id":"2c1de2f8ec02d20e50817cbd36be547642f0ba69","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-02-28 17:14:02.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"f11edb33e5572974ef6b06b06f3fe37e35e8e574","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-02-28 17:14:10.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"12343ec2896b547a13a773b9741de36308017a10","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-02-28 17:14:53.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"76914a9aa610cb8ed146376b2a58ed532e8650c4","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-02-28 17:15:02.000000000","message":"Patch Set 11:\n\nBuild failed.","accounts_in_message":[],"_revision_number":11},{"id":"0a973884de7fd244ff8bed7bdc07e5cd25d97a89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-28 17:18:07.000000000","message":"Patch Set 11:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/28_02_2025_17_12_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"6368dd44ce6bb3cf608f9bd13736329958ff6fa6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-02-28 17:20:07.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/3b095a961bc64238a47a818a1368dd82 : RETRY_LIMIT in 5m 10s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/3ce8a28247bd4c41966dfa48e53d1e3d : RETRY_LIMIT in 4m 52s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"2db3d8d8bca8077c104ff62523fc3572ad3e632a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-02-28 18:03:57.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/388203b313c640b3948be9d75972a8e5 : SUCCESS in 1h 10m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8e90f799115f491b8053e773bc32e383 : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":11},{"id":"f29bb7fa7b845d2a2c16654ddd4add7fb191f36c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-01 14:44:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/11/20250301152018/NEC-FC : SUCCESS in 5h 57m 34s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/11/20250301152018/NEC-ISCSI : SUCCESS in 8h 23m 39s","accounts_in_message":[],"_revision_number":11},{"id":"6e48431c8cdabb53738ba8795efc8bb67e2f706e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-11 13:52:26.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/902122 : FAILURE in 33m 09s","accounts_in_message":[],"_revision_number":11},{"id":"7be304984294907dca82ed51ca2fd861e675a907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-17 16:48:45.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"53835e5f0086c3493d707ed4bbb13d76778b075d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-17 16:51:44.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"0670647971e5fa8c93d0c92bd3b81a8f6a79790c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-17 17:07:34.000000000","message":"Patch Set 12:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/610/","accounts_in_message":[],"_revision_number":12},{"id":"eda7b8071091d256d2c01be0832386fe70430637","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-17 17:27:39.000000000","message":"Patch Set 12:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/12/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/12/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":12},{"id":"3fa7df3445f3bf0df2fdb9c33b779f50ea24adfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 18:49:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c361889d36842d1ad4892de164841a6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eaa3ea304a9c4dabacf17f1e0c125256 : FAILURE in 28m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17f9397ee1334b298a4a1bae1920380b : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d05e119ebe4747b5abe1fb1315fa315f : FAILURE in 40m 19s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f4e2ae802c7869ba079e6537e59e5448336bfd1f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-17 19:01:32.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 36m 28s","accounts_in_message":[],"_revision_number":12},{"id":"b0b7b2dc388f534eb448874a2ac798b5f9a888f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-17 19:50:38.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":12},{"id":"043f2a1149699cef43b765df1c18a69907004e60","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-17 19:54:34.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":12},{"id":"30e3b15cb9f1f355f9088839bcbbddace0491b85","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-17 20:11:53.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1822/.","accounts_in_message":[],"_revision_number":12},{"id":"1cf33c785cc446475130920397322bb2c327172e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-17 21:03:55.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1956/.","accounts_in_message":[],"_revision_number":12},{"id":"c108678332af8b9ca184c34214c7ef7731ae805c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-17 21:26:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dd1687badb5d44ae9a9a8c0f3c93341e : SUCCESS in 51m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0170a11b21464611a243154821b45c62 : FAILURE in 1h 02m 36s","accounts_in_message":[],"_revision_number":12},{"id":"ecfad3c7577198015b0b3c450a243306fca4da69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 22:47:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ddc00de5be148cea29ab6d888ab2add\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/00d96306300d4b12960a61dee25323e4 : SUCCESS in 2m 36s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6a946972370248f4b53109863fc1f618 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a2ed64d3ed2434380e18e1f17ccd644 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d8fbf4a34a640cd820c5af19dc40f52 : SUCCESS in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6837621e8b9d495abaf488b175c8d2fc : SUCCESS in 12m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7df33c2693346f7ae1f4f302d547b3d : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d192bf03204e2486cd5a03c025f0d8 : SUCCESS in 11m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0c1b34b5eda94ad7b680c1d7de62907c : SUCCESS in 3m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/6b77699b823e4bc6aca0e86da8eb9871 : SUCCESS in 58m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b0f464d150a44da901829ff57ce0c98 : SUCCESS in 1h 08m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30ebc990c7aa4de9b347c450e241d7d6 : SUCCESS in 1h 21m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdb1ef25dc0146b294fcc97cebe7630f : SUCCESS in 1h 01m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8937991baf574e13b7c0391324c2c912 : SUCCESS in 15m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/062b823a126942e1b2a154f65fa05360 : FAILURE in 19m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/534083ddf1554f85aac51bb8f8dab5f5 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15ddb02ca2fc495e9b0b7ae4fb4fc722 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20583faf01b841d8836187ce3233a6b2 : SUCCESS in 5m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ff64f1efee694cc08dc87f7e7ee240ab : SUCCESS in 9m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce5b2e29864c4a12a17d42e7b421d0f6 : FAILURE in 36m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/959b127c4e264147b01f6c55bf17cf6d : SUCCESS in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a875641650c34f75807065ee28f62834 : SUCCESS in 1h 40m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3733dc5e64244d5dacef4ccf19cb7b1c : FAILURE in 2h 32m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f0e961a34574043a2d8c2e4ab4de609 : SUCCESS in 2h 05m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ad80ba2e571b43f9ae1f43138e20bc1c : FAILURE in 13m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4e49c1d624894ef884d95d406f4eb1f8 : SUCCESS in 37m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5cab4ae19317429aad7b10b5fbaaa5a1 : SUCCESS in 1h 00m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9ffd6269c11e4e36adf26b0d1be42876 : SUCCESS in 1h 40m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ccc95ef35d2046c48c0f99ded5c3c51c : SUCCESS in 39m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2614565cbbda40ad87fe53b829a07edf : SUCCESS in 1h 37m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8e9e80078e5745bbaf09dbc06617f19f : FAILURE in 20m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de2e8eb3f36541f6be972012b5e994c6 : SUCCESS in 1h 39m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70e7a5ace4f04a8b9acae7e3bb816b89 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":12},{"id":"7d691422a26a51efd89bc4e43b8c7b0fa319f944","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-18 01:04:27.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b30271170132072718230da2386af718750ebf3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-18 01:07:01.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"68c13b5020623c4c25a185cbe745d3b0b92e6f4d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-09-18 01:10:15.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"e9e9b33a36312415b8e12fc5e30dc5357a2ed77b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-18 01:26:20.000000000","message":"Patch Set 13:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/616/","accounts_in_message":[],"_revision_number":13},{"id":"e49dadd70629ecdbfd6790ce0f25cccce67512f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 01:37:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98a1bdce2a6c42288f20ef8f7e47eff8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cefd466d2961485e9a9bdd9d0c94d402 : SUCCESS in 25m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a925fedaa7646e09b96246630ab44f3 : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5afa2e04e466401295af425ca9538c36 : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8405ad9e8281307da8eea85ea71aa24966e5dadf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-09-18 01:46:24.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/13/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/902122/13/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"032844666856d793053bfa54c98a03e88ea698ea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-18 02:48:46.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/902122 : FAILURE in 37m 57s","accounts_in_message":[],"_revision_number":13},{"id":"8155d2a8ee471b6fc2eddc80be0411439b6a430f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-18 02:59:37.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":13},{"id":"0ef3ef955ade109263f11c2b97b98e183bae3656","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-18 02:59:47.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":13},{"id":"cb654625e39112f13ce2fc5fb72a203748db5603","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-18 03:07:41.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ca9ee0b2978147d1a22f6a5c60ce4cee : SUCCESS in 45m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7a398357146c421e8f9d9478c289d13c : FAILURE in 55m 11s","accounts_in_message":[],"_revision_number":13},{"id":"7a939d3ec66f913e325b7ea34b3368f7257d6f7a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-18 03:32:57.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1827/.","accounts_in_message":[],"_revision_number":13},{"id":"7906a07336b7face6d82f2cef34804c347415fe3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-09-18 03:35:58.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1961/.","accounts_in_message":[],"_revision_number":13},{"id":"3bc30595872d95ef2868634a8b8e59a6c862b2d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 03:54:32.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc141c0b4ebc45ef8de3d211e3ba82e4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6fd2ee9ebfd24805bdce6e4e956beb9e : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3341eb71584cdfb866add26b58fa75 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a626699010ea4018802fec17d47eaafa : SUCCESS in 11m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b81d704d8b534444b05febc42942dc53 : SUCCESS in 11m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6786df2a29a4d33a31cb043c6d561ae : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285635b689984fc0a3014428779445ad : SUCCESS in 11m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c813600813ab467987c648bb583a0b99 : SUCCESS in 3m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/6fc1b1a7aa674590b1765983f3044dcb : SUCCESS in 31m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/660bc07c57ec4d5c981d2b04d03ab12f : SUCCESS in 31m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13536e02b44f4fd4a413143ed1eb8db2 : SUCCESS in 1h 35m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c520f386fa7e4921bf100d824fa2d8d9 : SUCCESS in 1h 04m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a0b04f13e7b42138a4526ce400946df : SUCCESS in 15m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9118e57fac43401a8ba165bddc2f7f63 : SUCCESS in 23m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0888c23c9393455aaead24dab92e71a4 : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97c4ab1c83284311aef770a94ce06ed7 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a117e405eb544158ab9c76ec88db8fd9 : SUCCESS in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d91d15a965f44359934f5d3bb96f162 : SUCCESS in 9m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d44f57a718724ac6b1cf9fc6ce071de0 : FAILURE in 30m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c8cc4d44a944b6b84c7be0acd764c5e : SUCCESS in 7m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b20af13bbe248d9a0ffc281993ed4b7 : SUCCESS in 1h 48m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57ec56441a9742c097e6198523b476f7 : FAILURE in 2h 39m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9abc2b5d1555413991f0641b026ac6e0 : SUCCESS in 1h 51m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/00b3d86daaa94dce89d1e4785a788af5 : FAILURE in 19m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ac2586402a314476bf2199261909c004 : SUCCESS in 30m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00a721af19f241e382aa2cb3dcadda30 : FAILURE in 26m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cd4902533ad14b008ad6839fa57aec0b : SUCCESS in 1h 34m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e3ccd5e3f0c048eebace8c2db79054c4 : SUCCESS in 41m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/482a13405ec848e1bceffb5979d61adc : SUCCESS in 1h 32m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7589043b74f6444e8f9d608efbdf5b92 : FAILURE in 10m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b80178c076e9402ea8fa4907db2cd5a5 : SUCCESS in 1h 45m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e5262ac460f4a50bc04a6e66ef20a32 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":13},{"id":"e226b2256f7a439a72e29a628579a4d52d7fcc0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-20 11:24:41.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"cabd8cc5fc36629b3d0b95efc4bd3ca6283454d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 11:42:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55fdb64549774fd5bd99788351e0dba3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e5df2244ef94fb68c3fc341331c723a : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c0ae18a9adf4f2caf12b15af6907ce4 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f604181462e0443485ec9420dffb00a0 : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"1aaa74c12790e58864bfcb02635397bda1f30786","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-20 12:23:49.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":14},{"id":"3def4cfb3bb5df1adb0a9142ae674bd7c6c8a09a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-20 13:15:44.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9552573fefa54ee3b2a1ed8fd3b17af9 : SUCCESS in 43m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5a7a2bd9bc8043638ded3db94f867512 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":14},{"id":"5f33651e7b332a047adcb39098cf66dda2e6f554","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-20 13:25:21.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 36m 43s","accounts_in_message":[],"_revision_number":14},{"id":"b0a6278aac898fdfdfb8a5e56816223a18d6b020","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-20 13:27:57.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2770/.","accounts_in_message":[],"_revision_number":14},{"id":"9becca6d8ed8623a6d382aad90c0d41025d960c4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-20 13:32:38.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2941/.","accounts_in_message":[],"_revision_number":14},{"id":"c2a0128f0f320d5184836c4bf9619a5614055bcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 14:00:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb06daea2c424c90a6e1c7eba7b45b32\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18fb32f63ff743b38b618ebd65044d1f : SUCCESS in 1m 36s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/003fbf3284d04b74aa9c9a913a17319e : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfa61a6ff3645deb09ec655fc43689e : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb7936253afa49e8b39a877e38254661 : SUCCESS in 7m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b39cbe055afb4974a2e5f13751ed4b8f : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89421c3492e43c29294e50985232f65 : SUCCESS in 13m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0eabdd169a24947be7f2e35bfde19c7 : SUCCESS in 3m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/5a699a655da9400a8489d9f9c277b9aa : SUCCESS in 54m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f42244c07a1d467a841f1aeb66b25edc : SUCCESS in 43m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a26c72c9af04095a7d2c44cffee4aed : SUCCESS in 1h 38m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/676a763109014c239be2013a61184407 : SUCCESS in 1h 00m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57840ba2c1c149d6a3bfa5eb3101c037 : SUCCESS in 15m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/253a7ed080ec404282312760f58e8354 : SUCCESS in 19m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/227bf368c8e145d7bbebc15d70fd088b : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94df43a5a4fe4950a87391f30bf50ab5 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05f071411ed740f4b2c0d08114a22a51 : SUCCESS in 5m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dcbb73094c1d408b8d87a393a1c9582d : SUCCESS in 8m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8392ccb39e7842b6b15abbb295d82766 : FAILURE in 39m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/76a5bc3aa8424d518f17f74a9213b870 : SUCCESS in 4m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9536f8126171498b966fabbebf4d0480 : SUCCESS in 1h 04m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/52fef05b084849c1a78010477edd60d9 : FAILURE in 2h 29m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9088c552e13d409cac7e609cfec7721f : SUCCESS in 1h 48m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/068ead588d4d457c87bdb8523c34b630 : SUCCESS in 1h 54m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a90be4e351414702be1c414f8a15b6c9 : SUCCESS in 17m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88df87a9d0a94c61868907ac4a11cf7e : SUCCESS in 1h 09m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c9971add09234ff5a09a4f4d4cea0148 : SUCCESS in 1h 32m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/63b384ead57240e18ece82a8d148801f : SUCCESS in 38m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e79556b8e6d4e49b050aa50ca25d641 : SUCCESS in 1h 13m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ace7c546550e4419a35368204f3a2659 : SUCCESS in 1h 08m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a628ac109284aa6818c3b6c2e824566 : SUCCESS in 1h 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30ebcfc02a4c42629c36dd91cad28a79 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":14},{"id":"83bd1b47d052a02c7278f9f386fd416eb3ae44da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-20 14:07:38.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_12_2025_14_00_54\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"0e45d2c0e1e8259ec8ffe2a89c711b29c70b709b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-20 16:35:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-902122-14-fc : SUCCESS in 55m 10s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-902122-14-iscsi : SUCCESS in 1h 10m 32s","accounts_in_message":[],"_revision_number":14},{"id":"d42c8a39384cdeacfcd6bfcbb48482d6cdb14047","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-07 08:57:03.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"22f631aab02c44c5d977d5b8a1e37ecc1c1d716e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 09:19:07.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e1ac7ed4b54aafb5207afa3030cab2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de785fd77d65451193dc04c4b7a3a096 : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42eb11dbc2b94a60962733931c941ecc : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/095dc8539bb14d5f8cd608ae9b394559 : RETRY_LIMIT in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2ad5d9d22b901e5e820d93371ed2cca7f1999f98","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-07 10:10:21.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":15},{"id":"344ea02882dd5a90a1f551084e9a84130d59b3dd","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-07 10:14:35.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4ab8847c1c9f4a6cbc1f67823d3326fa : FAILURE in 32m 56s","accounts_in_message":[],"_revision_number":15},{"id":"e45423a56e2f18a91fbcc07a219c21b795225f96","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-07 10:17:29.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 35m 09s","accounts_in_message":[],"_revision_number":15},{"id":"b7cd8fc047b67922a5115688aa3c0b3cbdcf8ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 11:02:03.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43d0ea8ddf504b83b17561f89235c446\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad7c844f795e44639aa204ce683f7ff9 : SUCCESS in 2m 40s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5981378009504c5081640c94150a0cf5 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c740b47d7ad43df99dadb5384a0830b : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94db2dee1e8e4339ab948abb22011e53 : SUCCESS in 7m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23abb1ce9b6d4605944c6082f2bfd8dd : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e7f8914a8b4b9296760daddc46af85 : SUCCESS in 14m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1e659af12e814a66aebc4646fcbe85a1 : SUCCESS in 2m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/76343095d59c424a834c6953a6962bf5 : SUCCESS in 1h 05m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3dfd8ba90f9040caa12c78adbb91d189 : SUCCESS in 39m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ae0c6c004fa456d9db4b899311e89be : SUCCESS in 1h 32m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e37a4533b66145129810b5c0420b95db : FAILURE in 40m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a517aef710148438377ca663eff01b5 : SUCCESS in 15m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/937aacae94b44f6cb4f0131dc32c7530 : SUCCESS in 22m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0ba6b956c2f04fa0a538103e8fa5fdbd : SUCCESS in 4m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f82a21cf13bf45f1a9acd7af93b4fe20 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44af57dd6f924677a15c1974c961e8e9 : SUCCESS in 5m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a5bc506edaf04577ad1c8febaabb6041 : SUCCESS in 5m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c3a67a6ee4284422a1b991052cd7be71 : FAILURE in 28m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11b7c71dd556479abb63ba420c1d2190 : SUCCESS in 4m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2aa9046d7d8547c29f4c112e3df38bb6 : SUCCESS in 1h 22m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3e72b2bd0790426eb01e3ced66818b73 : FAILURE in 1h 48m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aaa7a2cb4fb643d5978c1ae1142264f6 : SUCCESS in 1h 41m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/168d4c2dd0304864bd45f6d4d9841788 : SUCCESS in 1h 46m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/90af63b7c0b84a9db4d95dfd1fdfeebb : SUCCESS in 18m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d0931ea8208402dacc6c1ebcb17f1e0 : SUCCESS in 35m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/51323baca4f04167b961688eb2158a02 : SUCCESS in 1h 39m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7ed6d71deaaa44b7bffdfb61d6221a85 : SUCCESS in 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0971630bd06040c9aae5ac283b2fa16a : SUCCESS in 1h 32m 14s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e002c09c7e5b4da98bf091e2163275e9 : SUCCESS in 1h 54m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e39817a5c1cf492895612acb05387fa1 : SUCCESS in 1h 40m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8a991c3212a44da94b172b5f973a5c0 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":15},{"id":"0761c4bf4ef052199b4c8cc41b99df5fb3876442","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-07 12:14:43.000000000","message":"Patch Set 15:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_570/index.html : FAILURE in 99m","accounts_in_message":[],"_revision_number":15},{"id":"2e9c69c6dc6372a10e541252d2e87e319a7d6efb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-07 13:16:45.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fbe842edcbfa4f599c10581779546ff3 : FAILURE in 56m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4bcfda8e430f4b6ea3ecc05dadb0358a : FAILURE in 1h 12m 18s","accounts_in_message":[],"_revision_number":15},{"id":"16f5a599da4ea4d64252ae8854fc2614e5a09e9b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:33:14.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ac7026a760da4d2e87ed391093ed6c09 : NODE_FAILURE Node request 200-0000623374 failed in 0s","accounts_in_message":[],"_revision_number":15},{"id":"bce374fafacab9c98af5107f0f6ee374055a770c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-13 13:36:32.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"afbf03347c38730ca7b2481efa451a47c54fdbb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 13:55:02.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2709a0d570d24f7fbd4b0ea92c54bc4a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/105d3cbd442b4b1ea80708a341002055 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46fc7e86f93e4f9aad7c0c98a9be75b5 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cffa9f551afd4ce5bdfcd014c0ffe367 : RETRY_LIMIT in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"13da034487ac0c9e5616f3a8ec9efba1880041a4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-13 14:12:24.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":16},{"id":"12dacae3941fc149f762eb05b8ce66c159c4f738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:00:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9d631a303df4f06a6438519419a8880\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/004e11df19f94019897faa0574276e03 : SUCCESS in 3m 07s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/03cb351560df478cbdcbd989ad9a0f8a : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae79bbb591f4468b0ce2bcfd3c8d1ec : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdf8b30ff0c94a10b3cdf9427ed4fc0d : SUCCESS in 12m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3c681ef1b4442a1ac1b7d9f67fa16c1 : FAILURE in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f6bfa4ef9f4da28fc23fb59cca0ce0 : FAILURE in 7m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c5d8c18e04f4fe3a1ab8ea8e30ecc64 : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/fd97ec4cb8564c8c96df514fc8fce3a4 : SUCCESS in 1h 01m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/976ea30e15ee4cd2bd4de7c631cf63ab : SUCCESS in 1h 00m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75c8ca3b1af14e05bd128a8ac6cfbe47 : SUCCESS in 1h 22m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9624a31b7850436e83eb1eed1241a49d : SUCCESS in 50m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d02b9b7eac94fc0b537a23963da15d2 : SUCCESS in 21m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f295f8f52c004f3fbfdde59dada914a5 : FAILURE in 6m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1e9e343f6c834454ae23f6ee904421c0 : SUCCESS in 7m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/057bea8578a74ab1b2193d0c09cf843d : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dfbc8cb5d8ca43b88c1e984db4c77ba3 : SUCCESS in 5m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e4b5dc7789ca4365b8aa843e1a59717a : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/556d97051b314422a49c3017a248871a : FAILURE in 42m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a02b74e86f934c47bae20a87740e6452 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/398aefc5637d45f4be9cce90d5621dd4 : SUCCESS in 1h 45m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/76efbb9ecbd0488d804af0884bca76af : FAILURE in 2h 18m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13ab60fb85d24ed1b672c9a4f05fb40c : SUCCESS in 1h 58m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e4f2ef29c6a645fababb9f2ad8c02936 : SUCCESS in 1h 41m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6b1651f17d1d4c778889c20f35e58cf8 : SUCCESS in 37m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/43b6eb2323fe4eb1a4e5c6b2ea662d8c : SUCCESS in 35m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d78c3afbfe8a4cda9a30ae333452c1e0 : SUCCESS in 1h 42m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e3d19caf5ba545719c3bf6f2eb63d4f1 : SUCCESS in 24m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c84ec26a03344f7b9f33849390d354af : SUCCESS in 1h 04m 23s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/c5a070c546a948c7aba4461f8db3f9c3 : SUCCESS in 1h 24m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6cb6222d7a54fe4a4914a04191b01a4 : SUCCESS in 1h 57m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/391d6f3b25984f389ad53efd862b1d37 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":16},{"id":"9aa602dc4854bc7e0842a82b29729eebbbf94b11","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 19:03:49.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/b76ea7b08ad342d1b45322993becaeab : FAILURE in 23m 09s","accounts_in_message":[],"_revision_number":16},{"id":"b3581c3458a81eaac819ef4da15fa89c6aa9a6ec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-14 00:12:42.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 36m 18s","accounts_in_message":[],"_revision_number":16},{"id":"d0e6a60aef17b13d2cad700c957510fdc4967a26","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-14 00:36:52.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/461ad0ea08f243f0825536938b508ad5 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":16},{"id":"eea9f6bbc5f376765cbbd65172339b6a6e23f82b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2026-02-14 03:43:07.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_597\u003e.","accounts_in_message":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"_revision_number":16},{"id":"f26810f04dd8a1bf33272311ac90cbfe2a2dd71c","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-14 22:58:50.000000000","message":"Patch Set 16:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_658/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":16},{"id":"c4aaf6a53629fc1cd768cdb38442ff0be0d94ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-15 10:05:37.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"f7cac1f711130e36aa0338daef0bd14cf782212b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 10:22:46.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f5745d57854edbb5d001feaf81c226\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04cf5e008af441cc89bfbae1da9b7fa6 : FAILURE in 6m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10905457c57c42d08a5e5c2da0c1f837 : FAILURE in 8m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4ff7d8ff7c34421ad1b02318e7b1b39 : RETRY_LIMIT in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"52f70f9f3772e0e83f9f36f0347b1a07a440bc9c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-15 10:40:02.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":17},{"id":"9c4cfd694f1340ce75f04bde871a78e8c065317a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-15 11:16:17.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ea121f8ced0144c6b389a186d2772519 : FAILURE in 34m 06s","accounts_in_message":[],"_revision_number":17},{"id":"33852a47c10563a417be4921836ee47b15104e31","tag":"autogenerated:zuul:check","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-15 11:19:35.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hpe-3par-fc-driver build/4ec54440c3134c6d8ac6f64fd3afd963 : FAILURE in 30m 31s","accounts_in_message":[],"_revision_number":17},{"id":"e6f90422d6e5c03aa31246ec88afb4f3577aff0c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-15 11:21:49.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 36m 03s","accounts_in_message":[],"_revision_number":17},{"id":"3850e0bcd26d994f28ca414d2a10a903f827412b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-15 11:31:07.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3c45d64360944c83ba196df65b566146 : FAILURE in 35m 02s","accounts_in_message":[],"_revision_number":17},{"id":"c3e4e774c4f902890de0ddbeb9bac907fa1345ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-15 12:42:31.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e3c32d5896642339772c98943c5dfa1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/760d4fbd3f2e437bbbcc7f0032279425 : SUCCESS in 3m 03s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/280d0ec187e046b4a3bf65c3b4276bae : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/307e9aebc8ab48fea06f93aa697bf201 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d97ac430f7f4b859df1d754560f1ce1 : FAILURE in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96f49d026eaf44f4834dc7280954c6a3 : FAILURE in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22b4480376548b28e9f39f71eaeee0c : FAILURE in 6m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3bc1db6b7c5345e4856ec0c5626c5938 : SUCCESS in 2m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/b3deaa4f48b74397961e87a67406709b : SUCCESS in 58m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d4128e5c1414bccb3a38c1a8c720d00 : SUCCESS in 59m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c63a75e5e96483384ea145d6e6bfe14 : SUCCESS in 1h 19m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d7712de31b34aff98ba75063500513b : SUCCESS in 1h 06m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a882628d769943168291aff3f1b00c5e : SUCCESS in 9m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e17dd0ddd79b43ab8c106264eab1c380 : FAILURE in 5m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/144227ff65484c30a6a090edfc6cbdc5 : SUCCESS in 6m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95fdc963d95046eda67324661a71f1df : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8d72e9987484a0fb14fdbbd25971ffa : SUCCESS in 6m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6a0fbd97eece4fe0bf4d3188f4cb04e2 : SUCCESS in 3m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac4d42335604449daaf8743c36adbef8 : FAILURE in 26m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecb7bcd3c272452dacfcd1ab6b028ae9 : SUCCESS in 7m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2e203810d2ca4fed9946999a99dad4cc : SUCCESS in 1h 20m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/482e443df9874881a47f02db3a7c22a6 : FAILURE in 2h 04m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4ec39990acf9488bb234e56fff01563b : SUCCESS in 1h 49m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0e4dc779832f4bcba4d82275bea4c26c : SUCCESS in 2h 15m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c9c0b661e6f143ea8e66f66869899f43 : SUCCESS in 19m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5315b72c6ff84511ac9465064214c8e9 : SUCCESS in 52m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/702f2bd18e214044bf0d0c1f59a69498 : SUCCESS in 1h 42m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/602a223d0d9a472b9d277b0834b6f450 : SUCCESS in 21m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4de0e8b7decd46f189d5713e2e4e00cd : SUCCESS in 1h 48m 32s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/5e45c8d356c24fada1d13eb4be666879 : SUCCESS in 1h 07m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c7cc2dfa463456fb21457fb70627186 : SUCCESS in 1h 43m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0657815ea9c842aab6a74836f2bcdff1 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":17},{"id":"25d60eaa6ae573483ca4649c309fe2bce3b92744","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-15 16:08:33.000000000","message":"Patch Set 17:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_674/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":17},{"id":"9beb60bf02c9e948f7f48a6a45f488243f5e1088","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-16 06:18:35.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/de4fe7af05224e47bf8a0a1ff205f330 : FAILURE in 1h 12m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a3a36eb3986d4976a34afc0993c715b6 : FAILURE in 23m 02s","accounts_in_message":[],"_revision_number":17},{"id":"311a20cdc284c4bfdafcc8f4cfc2534ee146b9f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-20 04:06:00.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"3db8026a3ba39bceae1fb7eafb9947e682871663","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 04:15:59.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/efba236bf00d462f8c59b1ac2cdd3f48 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f56a0fc13d7041199c72977215866073 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":18},{"id":"59b1ed87133bbcb15db617d47815afa081b379de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-20 04:43:43.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":18},{"id":"4ed6596aea385d1518530dc3c447a2e7fffda80d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-20 04:50:49.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d32765d1a86a46259a6c70494c14540b : FAILURE in 35m 37s","accounts_in_message":[],"_revision_number":18},{"id":"71659afad8d6cc959d034e71c46883d78f17ecdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 05:03:30.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f050e0be96bc42c6a0b8c4f7c2469798\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7610186ae38a4cfb9453fa16a828dabd : FAILURE in 6m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c3bbb8dc1e854a27a9ee834c8c5f77b4 : FAILURE in 7m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aec36b3ee2ac437fa4a631a1e99af721 : RETRY_LIMIT in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3549d46eae091e59874dcd6fd6c70e4b7c82928d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-20 05:16:42.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-fc-driver build/bf599bdd106d44228648cd150870457e : FAILURE in 28m 33s","accounts_in_message":[],"_revision_number":18},{"id":"1e6852435b911713176c6911c29cd0b5d23a881a","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-20 05:26:33.000000000","message":"Patch Set 18:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/902122 : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"81c3c4521d884f48dd013047afec10cda61f43e0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-20 05:36:10.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/dc754cdc49714e7c8e3fcb332763bb07 : FAILURE in 34m 33s","accounts_in_message":[],"_revision_number":18},{"id":"eaab4ca045d20f2e4fbf9e843a3c178f579699ba","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-20 05:38:20.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 37m 13s","accounts_in_message":[],"_revision_number":18},{"id":"0a90523a9e22b13eaa91414ff2c8242fbbb23f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 06:39:09.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2245fc2efcee483cac85d3894aa8aceb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3dad869d7a5e419284f1f6d54bc07c9a : SUCCESS in 3m 10s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1b1541c963b3483a9e4d14c0a0b61db1 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c66d3e251540a78caa310959c3162a : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0faff3aa2aa344f4aede5cea170b66ab : FAILURE in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbd0dec4a83a48b8a32dc04f04095306 : FAILURE in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97eed248dc4d49fdbd1e112e92d8c32b : FAILURE in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8393346657c4349927e627ef544d150 : SUCCESS in 13m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7dd0c246846483da3fd5aef8d2735e8 : SUCCESS in 2m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/a1779956331f434c87355642abe75630 : SUCCESS in 1h 00m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da88f527d2d5408d8b98c0ebe66c33d7 : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07640463911a4c26aa1c73afb26d6ae2 : SUCCESS in 1h 30m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/820b3778316a45c990c3a941fdeac444 : SUCCESS in 1h 02m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/409bc105a6264c85b726b029d6f07a3d : SUCCESS in 16m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4b04a08949a453787a2137b0d03ea9d : FAILURE in 5m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9df1d1cdb82c47848e531d025c291470 : SUCCESS in 6m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f3947ef22b8444882d2f900b97fb4b5 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f87683a22a2b49b5b55228fc68c1c790 : SUCCESS in 3m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/155fb592898c4078b698dc05031d0139 : SUCCESS in 3m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8709d34b83c14f07a89d6aaf92d6b723 : FAILURE in 40m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10e123de5a284c8ba255d664feb146d5 : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b738145158d4d6b9bd9e6d97fbdcc8e : SUCCESS in 1h 43m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8fb1913426614f698da8dd4ccad0f915 : FAILURE in 2h 29m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8a55e44986a945c4ad0645b12aab36c6 : SUCCESS in 2h 24m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9447dd5d92fc475c9853ad9c81a11de9 : SUCCESS in 1h 51m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/273300dbd4a349ad9903c588def8721e : SUCCESS in 34m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a72d16f458f416595e633fc9104b438 : SUCCESS in 1h 07m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0968acf3f17e43c0b83fa49a2bd0ca66 : SUCCESS in 1h 38m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6a83de30a257427fa499a3c57c206dd2 : SUCCESS in 40m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e9b3c432edad4e97a5a51e50e7f67aba : SUCCESS in 1h 32m 36s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4581154c733a4f4e875b5c87fe4cbf66 : SUCCESS in 1h 54m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7842c10d639d49aeab85793c4db06616 : SUCCESS in 47m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c94f2b2b00e248329a6a5d8282ec4c16 : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":18},{"id":"8baaa902606f729374f86fe7ee50b3c51b8f8a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-20 07:01:19.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"10037f7b789c2a5ee307dfc95163319edc581dab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 07:10:50.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e94fdcbe546d4a628aeab5879b02bcfe : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/735876e9eb654500b88b2ecd3c156510 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":19},{"id":"7b538238df69831d2392c233be7042fb41b49cb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 07:19:49.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dc083c672c1452aa75e2ea2b9933ba2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cca17c361de24b7caa7072005a80a0b3 : POST_FAILURE in 16m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14a27efa9f224aaaafe55d52985dea61 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af3913752c574a24a3bf3aff84808f8d : RETRY_LIMIT in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"e431b4adee3ef27b8a3d8f53e8eb6a0c17c2cb78","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-20 07:38:57.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":19},{"id":"5d180278239a7490c4ca349bab05c8bc208157ca","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-20 07:52:08.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- hpe-fc-driver build/7febf879d7fd49db827548b5101def79 : FAILURE in 26m 52s","accounts_in_message":[],"_revision_number":19},{"id":"0c294905936052a6c79fa696c2e015deb3aff140","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-20 07:57:11.000000000","message":"Patch Set 19:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/902122 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"332165b6ccd476177ca58f4414e15d52ea16764d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-20 08:23:02.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/902122 : FAILURE in 37m 32s","accounts_in_message":[],"_revision_number":19},{"id":"300eab0209eb29ec4aee54929faf8b61e3757d7b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-20 08:24:33.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a636cbe69a524ab8a19d3be7415efefe : FAILURE in 34m 10s","accounts_in_message":[],"_revision_number":19},{"id":"f76747c1c8459f2445f40deeff109074224a5625","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-20 08:52:18.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/fb883b1681704c88901cec22ff30102e : RETRY_LIMIT in 14s","accounts_in_message":[],"_revision_number":19},{"id":"073164834ea263ac94d953c578a7339bb418217b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 10:16:20.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/052b97274cbc4c6aa3d06e6cb4d8a827\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fa51c2d04f24806b12ab92321f7343a : SUCCESS in 2m 56s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cec14eeca3534bed9ba2e18be6dbfd13 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebab8bf27c2c4ef29d9b96cf23fa0352 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03005614ebfa4cd5b6fed64907694041 : SUCCESS in 12m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de9d3836e81a4d4ea82d4d5db3eb3918 : SUCCESS in 12m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d388df493c084a45bc4ad64921d6f7a1 : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943c87f7f30640a294c03c33938d3eb3 : SUCCESS in 13m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2d86548333e494f9d42397ca93a8acf : SUCCESS in 2m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/0fbe9c7b416b4552a6cd000874675c8a : SUCCESS in 57m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/095c521c74604afbac19766f60a7ec6e : SUCCESS in 1h 05m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0640c2d89a1a47bda20be992385136ad : SUCCESS in 1h 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb2f9400de73437a9b87ac5a7507dd50 : SUCCESS in 1h 15m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/494aa16b2ff54e2584dd16a8c3ea31dc : SUCCESS in 15m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/faadfd59d0a640aab9cd024b1088d31e : FAILURE in 18m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e2f179c17a24f4a8c13cee5ae730cfb : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb43f403e9cb4b719b6024d7daaa168e : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5655bbdac0a44b48ab5e10ad2bafe87d : SUCCESS in 5m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/94e15bd78cd1413abca9abf3059aaa0a : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f22904a56cde424d81c0814dbc955319 : FAILURE in 40m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4220243c022e4f91a77f2c79d0f0a9a5 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/45767c06efda488c97b9f4f1b2dc7849 : SUCCESS in 1h 57m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f679a9facf454c99af7834a68645e9a0 : FAILURE in 2h 15m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1acd4c7003a41ffb68547ef6515a70b : SUCCESS in 1h 49m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3324619a719046b19923f8fba2d845aa : SUCCESS in 2h 04m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0971fb03a9e04d1f82d8474c1981969e : SUCCESS in 33m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aaf31592c2c8498bb80787732abd1082 : SUCCESS in 1h 06m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/68d7776fc6194ec2b09146a54f3051cb : SUCCESS in 1h 46m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/62e0c84989f144ab8e55645e189ce691 : SUCCESS in 42m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e81edaa90250430a98abe6e1f5eb81da : SUCCESS in 2h 04m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/17019f4aab134b61be1e0a59c169baed : TIMED_OUT in 3h 09m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6410b0a2d0ce4db7b30817c470eec6cd : SUCCESS in 1h 20m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3110524d680c47c58a5769138f5d3689 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":19},{"id":"b3fc3d98a6e0d43e57f0f8ee79be86d64fc86d7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-20 10:26:31.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_02_2026_10_19_24\nLeave a comment with","accounts_in_message":[],"_revision_number":19},{"id":"4f7c8302a1c66c5ee59979d3066d70c7e72b7d0e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-20 23:37:13.000000000","message":"Patch Set 18:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_780/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":18},{"id":"ccbaba21e71b916767220f8c8797a1eaaf60cc09","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-21 07:21:52.000000000","message":"Patch Set 19:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_785/index.html : SUCCESS in 87m","accounts_in_message":[],"_revision_number":19},{"id":"cddae2986bedcff0eb324d2361f416e2bf1f10a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-05 15:36:04.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"102497384586babbe1fac9f166050bbe5b57e4c5","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-05 15:42:23.000000000","message":"Patch Set 20:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/776 : FAILURE in 0h 1m 34s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/902122","accounts_in_message":[],"_revision_number":20},{"id":"b426e74af4bbb60d17a1c161d6b5a9cb8419de4a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-05 15:42:46.000000000","message":"Patch Set 20:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/504 : FAILURE in 0h 1m 46s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/902122","accounts_in_message":[],"_revision_number":20},{"id":"4e660b437e47674359678d761aa66ab72030a3ee","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-05 15:42:54.000000000","message":"Patch Set 20:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/419 : FAILURE in 0h 1m 48s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/902122","accounts_in_message":[],"_revision_number":20},{"id":"51752429b0e08ff167596f04be2c7a9cd0f18150","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-05 16:31:02.000000000","message":"Patch Set 20:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b29460ed18dc4ed2923fded355283b66/","accounts_in_message":[],"_revision_number":20},{"id":"80a5a47175a19ce5e36e0e3a35d3b376f15aad70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 16:38:51.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d56d7c9cc304fb1b6bfa7d6ee795b76\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae80af5c94ab4e55a522b5135e2abd95 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/653cdb32392e412189f5d592085da125 : SUCCESS in 16m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4f2daf29ebf4380b60b5911bbdc7da3 : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"468f6f861af0d83058b59bc393655cde6fe7047f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-05 16:44:32.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":20},{"id":"6cac0714763250759f4ab8cfa2b2ceeb690266ef","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-05 16:59:24.000000000","message":"Patch Set 20:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b29460ed18dc4ed2923fded355283b66\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/57e525b29c0947bd8b7ad66e9b7c21ea : FAILURE in 49m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b13887ba6b9b412abb501e994cf91016 : SUCCESS in 44m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/37a1555b2f364f8c8b0e6035adf1e386 : SUCCESS in 37m 14s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e25678e0ea64ba2b3fceee24d618d71 : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":20},{"id":"14323ae565227c96b4fe03163df360c55dd26c0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-05 17:34:46.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9714306876c740d09a74c105df2d33a3 : SUCCESS in 54m 53s\n- cinder-storpool-tempest-iscsi-multipath build/fdaae4bca171492eaa142a834da102f7 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":20},{"id":"51995bbaeadfc6d6f16c007b98d8e7dac4821920","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-05 17:48:53.000000000","message":"Patch Set 20:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4696/.","accounts_in_message":[],"_revision_number":20},{"id":"0df6f7a5b25e8252cdca38c8e78817cc0e58359a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-05 18:22:36.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bc4b6c0bb6a4c53acd290a7479752c9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a7b9f46472346a28a475ac04cc0a56e : SUCCESS in 3m 19s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/32744fd40c0b47b6a00edadaaae571a7 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348c14a897ac45a68e0cd742f6722ba1 : SUCCESS in 6m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc73e6f535194d3ab20702dc5238230a : SUCCESS in 10m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/786c969be96d4bee85d0f7470709dbba : SUCCESS in 10m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/42a0194df5ba42b983d1f341739ec5ea : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2de870a5efa4e098aa12d46b2fa1ebc : SUCCESS in 15m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15dfcc9c92e3468b8f8f831560fc0036 : SUCCESS in 3m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/e841ca86789c4b2ba95fb55d29d1fcee : SUCCESS in 1h 02m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b95138a36968486c8952407db2aecbdf : SUCCESS in 1h 02m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6f2977d15b847b980551ca10e42464e : SUCCESS in 1h 45m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d45481b02f714382bba9e6702336bcc5 : SUCCESS in 51m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d22b2a8d45ed4e938a7ed87d455844ba : SUCCESS in 6m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f724c138bc15473fbff89196adb7100e : SUCCESS in 21m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e5588bc867e44694967b9e5c46df1e42 : SUCCESS in 7m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1786e79e1b7490585646ba887df9916 : SUCCESS in 5m 57s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/58cce2e15cdf4f7d8ce1bf99df7194e4 : FAILURE in 20m 14s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d90e63eeb314d7b839ba9af9691710f : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1af4b51492624f689e2feda8b49feefb : SUCCESS in 5m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9118858da6b547b1bf3a42dc47525ac3 : SUCCESS in 8m 43s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9a91baaaa26b4214ad3765d6c1f5ca28 : SUCCESS in 33m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/576a1f4da9ed4139950f1631fe9f5d95 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51bf69983b64422b973e160367285463 : SUCCESS in 1h 15m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9c6c690d85ae44b8b4c512e044531e82 : FAILURE in 2h 36m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73ab2c73aa524f7cbb95bd331bc00811 : SUCCESS in 1h 57m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/4bac90e4a5b74d958850a0e6f54c8686 : FAILURE in 2h 17m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d5e0cad9ccad45e195a4eeb95f00b960 : SUCCESS in 1h 51m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cd65127b46174548bad397f868f541c0 : SUCCESS in 38m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/458cd00118f8488ba857d38f1b8ec429 : SUCCESS in 44m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4cdf170890cb479e847d6305ddfb13b6 : SUCCESS in 1h 49m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cd2448cd22c0478fbd5bdde4817baff8 : SUCCESS in 50m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e2fa7135c86442fa1409636acfd6149 : SUCCESS in 1h 47m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68c2d194a8a84214bff16befe21fea26 : SUCCESS in 44m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"ba88d8fdeb637789c77f839787b96834999c9a77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-05 18:42:59.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_07_2026_18_35_16\nLeave a comment with","accounts_in_message":[],"_revision_number":20},{"id":"c03d3392d860f0fc873ace5df18365608e04e738","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-06 04:51:57.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-902122-20-fc : FAILURE in 11m 5s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-902122-20-iscsi : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":20},{"id":"1d96d7174dc4554a1891b8c5bf7424751e3a192f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-07-12 15:11:02.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/20/20260712185550/NEC-ISCSI : SUCCESS in 4h 50m 52s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/22/902122/20/20260712185550/NEC-FC : SUCCESS in 5h 15m 09s","accounts_in_message":[],"_revision_number":20},{"id":"7ec62bba856983cd45c7093b62776266246d239e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-19 01:16:14.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"742dfdeacac12c632e4774f7c0d9896f39c06c36","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-19 01:21:48.000000000","message":"Patch Set 21:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/981 : FAILURE in 0h 4m 31s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/902122","accounts_in_message":[],"_revision_number":21},{"id":"b1b71f384c55a51ba8d7966ae23a84d7ac748c56","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-19 01:21:59.000000000","message":"Patch Set 21:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/887 : FAILURE in 0h 4m 44s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/902122","accounts_in_message":[],"_revision_number":21},{"id":"50b76b1b3f71c03534b2cb42448fdc3a77d73e1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 01:55:47.000000000","message":"Patch Set 21:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8d5077a45d624f579459a37a75f79182/","accounts_in_message":[],"_revision_number":21},{"id":"9f60e81ea942dc5e50da53b3d15926abc86909dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 01:58:11.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e916886c7f43f48ca879d6206e9dab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2543f16e98c4492a254cd02bec260f5 : SUCCESS in 17m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7de56300ba547a293d5b3b817c32ff6 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3eacabb5a69249efadbce118d33dbf29 : SUCCESS in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f4171d6d67b3ec2281bf200beaf76b1e65329cbd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 02:12:19.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iaaedcdaf53d327be35bb3b9f586a4188204ee23d .","accounts_in_message":[],"_revision_number":21},{"id":"391f7a3d0f6fde8868d29cf81b072129745f97dd","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-19 02:30:34.000000000","message":"Patch Set 21:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1250 : FAILURE in 0h 3m 31s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/902122","accounts_in_message":[],"_revision_number":21},{"id":"ba43dab4e25e9ea521ee2f1743886e55f541156a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 02:53:58.000000000","message":"Patch Set 21:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8d5077a45d624f579459a37a75f79182\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f324356168344888209c4fc6c8ad0bc : SUCCESS in 38m 25s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ff38c2e1a5747beb762ee04b0a6a8a3 : SUCCESS in 37m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33a28e281c5c4553a3b8618262a94b4b : SUCCESS in 40m 15s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17559d89a56c4f58beb8d245904e5468 : SUCCESS in 35m 23s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3dacd57ea4f54cff98292f2791a9f672 : FAILURE in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"e8e5f53a74cc137b8fc652be1f09fe64d93219d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-19 03:27:07.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/956b7a368b6e4816964e0afcc368de76 : SUCCESS in 1h 05m 13s\n- cinder-storpool-tempest-iscsi-multipath build/260d603a1f9b4ff78874b7b96744821b : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":21},{"id":"a672cfd7d7ae43b457c78de629a089b26e295029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 03:38:10.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0cbc5e0416b4616b4b56ccf0991ce9f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a42b08802bea41d18d27ecda508801d8 : SUCCESS in 2m 19s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/95dc8393930d485d85920398a12f17f5 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d34282cbd044cd922814262394ef3b : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b144f0183654c0090f432786e1487a2 : FAILURE in 10m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e422a723c14a4f3f8498f05080ef09e6 : SUCCESS in 9m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d19bebac620a490180fe5869cf52ef3d : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd32f6a0e50d4c0b8ec10bb11507b236 : SUCCESS in 9m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2e34ed8bd5b496c874bc6c3be22db6b : SUCCESS in 3m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/ab4570c5e5d24c23ba6d2c9e70df8b0a : SUCCESS in 1h 10m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f4f3e500773455eaa9554fc1cc63121 : SUCCESS in 49m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51df8521fe32479cb694a3699f281ae7 : SUCCESS in 1h 06m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baf2a083381c4dd3b4c48a3fc5c0907c : SUCCESS in 43m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db529d5913e64ef7a591592aa9fe9156 : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f9b564d5ca9547e698a5956c4c26441a : TIMED_OUT in 41m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8b7c87b8aaaf499786290e0b11ff12c7 : SUCCESS in 6m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cea297afb88149208c32ab4961ec4c7f : SUCCESS in 7m 31s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/c2482b594f774ece93f467100e3f21bf : FAILURE in 10m 41s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b456d4fdf32a4248bdcd8a5787b3b7d5 : SUCCESS in 8m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4d4a47ca9368437ea9b7764cef2039a6 : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9017f2ee94a4423f8e95f15760df4d0c : SUCCESS in 9m 16s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/559416de59174df7bac4122b546c46ac : SUCCESS in 34m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dff9fb5072bc492da29c6216b311b905 : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3fd565bce1ed41eeaaaf5157d3fd84cc : SUCCESS in 57m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed454d370a384e4391ac5d06e40253f1 : SUCCESS in 2h 00m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d08302ea556440c6a953b8cbf57095f2 : FAILURE in 1h 51m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/fc3f642478074f79825a4f8eea25721b : FAILURE in 2h 15m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b4e8447ac2da401c835e6dd32539c6c2 : SUCCESS in 1h 49m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/81b3cab3fe314bb6aa6bb16387b95c50 : SUCCESS in 25m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e0410db4d24b4520b8a633bccb318325 : SUCCESS in 1h 14m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/28319a3a96984ac1aac26ff89eafd0c9 : SUCCESS in 1h 36m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e8eacb633a594c69962deb4ad742ae94 : SUCCESS in 30m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7bfe86fb07b74ae8a088eb6fb2154e98 : SUCCESS in 59m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b5153d8fc6e4fcc8d06ea359d8036d5 : SUCCESS in 27m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"65724beb797a4edad21f07ec39d66240840ea002","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-19 08:12:04.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"fcb9ec8fd2ad4911ab183d64a45bb42dcbb1a103","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 08:34:21.000000000","message":"Patch Set 21:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/fdbc72bf5e6041d897110fd923e66fd7/","accounts_in_message":[],"_revision_number":21},{"id":"abb2e4c02d92bb1b96258c8185c34fb062f79826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 08:38:24.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092903c451804c3db993cb97d8755b97\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d4c82e4e9f1425391f480549d09cb3c : FAILURE in 6m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cca083375a5f4246b1b8781514827d99 : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9e439abdd9984a129b950567384597d3 : SUCCESS in 24m 31s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"8472281160f9ff25e8d1585c7da951b7f5913e48","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 08:56:59.000000000","message":"Patch Set 21:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fdbc72bf5e6041d897110fd923e66fd7\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/49c1e95f6c504044b94a6a7bc64a6a89 : SUCCESS in 38m 37s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/27e167a9009a4966adc264f82785df0a : SUCCESS in 34m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c0f4a2625aa14b308fa94ef519a4fede : SUCCESS in 44m 19s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7170a5577f9241238f85dfca3c690699 : SUCCESS in 29m 49s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/44e36fe44d9844468ba5810122f424aa : FAILURE in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"eee28912b087189275743d53195d846a2b35ea6c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-19 10:08:32.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest build/90374a855bd34bf19639e12b9c2e3b85 : RETRY_LIMIT in 25m 36s\n- cinder-storpool-tempest-iscsi-multipath build/acf067fa0c174715b597f81d0e4947a4 : SUCCESS in 1h 47m 35s","accounts_in_message":[],"_revision_number":21},{"id":"45c60e1638a2da110898aa1364d8892df7eebb7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 10:31:35.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042caabeeaf44dccadc6026243686874\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/713707b73f90437494d053267dedd46b : SUCCESS in 2m 44s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7f058016be784abcbade091866f87f34 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bfc717a562402184f7886161c541f3 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/824a6c84074440e1a275392ef67aa80d : SUCCESS in 9m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e488641079e54b4a9a6d466f7c3f6348 : SUCCESS in 6m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0b0971c21fb44169de5c2c457bdcf15 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d7c6255a7742d881daffab052662e4 : SUCCESS in 7m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/508a159a0e82435d83b7c2154aefdce1 : SUCCESS in 3m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/4158dbcd446e4c878655add14ce26d98 : SUCCESS in 43m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93a887d6d0544635bc7558d7c58ed9d3 : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afc20934f3b64f53aef207a972bfb774 : SUCCESS in 1h 01m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c136201fdc894aac8ea9a6e7e22615fa : SUCCESS in 38m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/673e6faa4f3e4946a6619a288623f798 : SUCCESS in 9m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c2d4c47886db43a48e761638bf2f2287 : FAILURE in 24m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c7be56f4482f460292bf8958f13ed593 : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de35ef3740f64eafa2aaa12a42ab5f56 : SUCCESS in 6m 50s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/893faf0f020c4b6cafd1adadbde38e9b : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3cadda43837e47df986027bbe6431841 : SUCCESS in 5m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6d362740ccae454eba94b05dd2b42009 : SUCCESS in 5m 28s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a3a1af99204f4120b1cc608fcd455c50 : SUCCESS in 7m 44s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/32873dbac898471fb2ebd27f6ec3cdb7 : SUCCESS in 23m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0c26db3f8d5421e978e8e4e245e89c9 : SUCCESS in 3m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4a5ca55ea2d4d1da06ed07a0d93b57a : SUCCESS in 1h 29m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6ee7e298107d45af8414defd026fc772 : FAILURE in 2h 02m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b8cf75f4d1a64332bb3d259b441204f4 : SUCCESS in 1h 45m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/adca0f598c7c4983be11e64cf8dbef7b : FAILURE in 2h 12m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8ebe1562194242708ddcd4102829aa4d : SUCCESS in 1h 54m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8dc23e1c18994691a60a9d89a74c19cf : SUCCESS in 31m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f704f9183fde4147a4301d067f9f8011 : SUCCESS in 1h 12m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3156ca1c30bd4ba99f5b1d39fd60e3ad : SUCCESS in 1h 34m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1535e29fc9fb40e8b1441860dbd63a3e : SUCCESS in 37m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2fcc1749d51c4e9c8c334b0ca69e42ff : SUCCESS in 1h 02m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/217251ff6d2745c5bb4f6614a938d951 : SUCCESS in 52m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"020f0003446b349122071e43ca90347bbd30ae9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-19 10:42:07.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_08_2026_10_35_42\nLeave a comment with","accounts_in_message":[],"_revision_number":21},{"id":"edbf8460f77fb012ddf6e2b36f8632649c26e440","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-21 08:58:24.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-902122-21-fc : FAILURE in 13m 24s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-902122-21-iscsi : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":21},{"id":"649330371454e4bf5e171f848167242a397735c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 09:35:14.000000000","message":"Patch Set 21:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5195/.","accounts_in_message":[],"_revision_number":21},{"id":"516db3edfd60c8103070d2af69685dac5594fd06","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 13:04:34.000000000","message":"Patch Set 21:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5019/.","accounts_in_message":[],"_revision_number":21},{"id":"84df34ba5c4bd1f8ae6f60d0511fe9f1f669a736","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-22 20:16:41.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"50e5ac8d07fc15f579c41ac4074fb7687fcbc140","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-26 15:30:44.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"c0aecd3be5935b685127d14ce3aeff0c0a6de2ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 15:31:27.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"a0f19580ccf864e3a0df6cb06ef1915999fded47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:12:31.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c39f81b877d24979ada96d33d18d095b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/891192269591448d947f5036ea43810e : SUCCESS in 4m 37s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c641a5048f074575b2c777f317151871 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec9de4beb244dcb8ffb7fe184ca4523 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ff9a6a4e09a41adb54c95a262f55046 : SUCCESS in 11m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d5d6d61fb3e49d086d846a4b0f48af9 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087d943949c44af6afb41192836c0084 : SUCCESS in 17m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c45b210fc3647518ecdf501c5b900ae : SUCCESS in 3m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/c6f435826325437b886116de2d85d34d : SUCCESS in 1h 08m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c2288e997cc4ba89850a2cb1886be80 : SUCCESS in 50m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fac819036ec04c1495d4738a892544d8 : SUCCESS in 1h 35m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73b7a009a7e5436aadd145d55a2a85f5 : SUCCESS in 1h 00m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54a89d70a8c84502aa3ae60f18ff529e : SUCCESS in 9m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4f34f273b0f642fa9d57a0e9e5f42c01 : SUCCESS in 1h 14m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b24250a2bebf45e385c83cf2af4016eb : SUCCESS in 1h 29m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b85634fa5094e37a7e1a08341070b12 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":21},{"id":"8ac3214098ce554437554d86a19591e2b1a61bd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:12:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"a94db1f7cf846a1f02205a9bc6fa45feb5acf8ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:14:51.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d7fd9d904a5454c8b87982e6c6b5a64\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d96cd2f2cf784432993ba46e13e90868 : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26035800bdd84d6eb0729cf875c8c8f3 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e2205605fb64052bf12d6433319741e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"5e4530f77f43c57ad201f3a042c64488df7178cc","revisions":{"f5f5889b3c3cd93c52cb744433eaa7185d748cef":{"kind":"REWORK","_number":1,"created":"2023-11-29 01:22:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/1"}}},"commit":{"parents":[{"commit":"c1a15e9352432d84de382412fb1d51b63351fa8e","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1a15e9352432d84de382412fb1d51b63351fa8e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5f5889b3c3cd93c52cb744433eaa7185d748cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5f5889b3c3cd93c52cb744433eaa7185d748cef"}]},"branch":"refs/heads/master"},"2428d34c35e75d4ffa6aa0bb73738475bf31fc06":{"kind":"REWORK","_number":2,"created":"2023-11-29 01:34:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/2"}}},"commit":{"parents":[{"commit":"c1a15e9352432d84de382412fb1d51b63351fa8e","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1a15e9352432d84de382412fb1d51b63351fa8e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:34:25.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2428d34c35e75d4ffa6aa0bb73738475bf31fc06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2428d34c35e75d4ffa6aa0bb73738475bf31fc06"}]},"branch":"refs/heads/master"},"700962ea9cd5bbfd8880202a4271e15848cca181":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-02 02:39:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/3"}}},"commit":{"parents":[{"commit":"378d756ab371128f6b407c52a016f438c97aad4d","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/378d756ab371128f6b407c52a016f438c97aad4d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 02:39:20.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/700962ea9cd5bbfd8880202a4271e15848cca181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/700962ea9cd5bbfd8880202a4271e15848cca181"}]},"branch":"refs/heads/master","description":"Rebase"},"10928d6257ccae493af8b8bb3fa9f28630d05dd0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-02 02:44:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/4"}}},"commit":{"parents":[{"commit":"1388e1d900cd6f726be1f6a8c5cb7ed2f7b7be79","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1388e1d900cd6f726be1f6a8c5cb7ed2f7b7be79"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 02:44:11.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10928d6257ccae493af8b8bb3fa9f28630d05dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/10928d6257ccae493af8b8bb3fa9f28630d05dd0"}]},"branch":"refs/heads/master","description":"Rebase"},"320015b0eab2963e464bda44e776eecebf63a75f":{"kind":"NO_CHANGE","_number":5,"created":"2024-02-02 02:54:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/5"}}},"commit":{"parents":[{"commit":"dd49a4008c930c4ac6cb79b63e5cbcfda6df7e4e","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd49a4008c930c4ac6cb79b63e5cbcfda6df7e4e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 02:54:42.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/320015b0eab2963e464bda44e776eecebf63a75f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/320015b0eab2963e464bda44e776eecebf63a75f"}]},"branch":"refs/heads/master","description":"Rebase"},"d516d722c854b656421f45c38fdc9060c612d2db":{"kind":"REWORK","_number":6,"created":"2024-02-10 11:11:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/6"}}},"commit":{"parents":[{"commit":"7e2da64a1c1970b200a7068b85e1c00af6b1bad1","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e2da64a1c1970b200a7068b85e1c00af6b1bad1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-10 11:11:22.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d516d722c854b656421f45c38fdc9060c612d2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d516d722c854b656421f45c38fdc9060c612d2db"}]},"branch":"refs/heads/master"},"d53b997b1f8bb97a374f94db13568dc5d48c1889":{"kind":"REWORK","_number":7,"created":"2024-02-20 01:17:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/7"}}},"commit":{"parents":[{"commit":"41a174a463537e349b7d0b287a68e8e82db617cd","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41a174a463537e349b7d0b287a68e8e82db617cd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 01:17:21.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d53b997b1f8bb97a374f94db13568dc5d48c1889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d53b997b1f8bb97a374f94db13568dc5d48c1889"}]},"branch":"refs/heads/master"},"046b965d86c23d86b257d663090e7de3460b4f14":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-02-20 03:07:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/8"}}},"commit":{"parents":[{"commit":"d71530c2a1dc893c54efda47813100940025dc93","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d71530c2a1dc893c54efda47813100940025dc93"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-20 03:07:12.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/046b965d86c23d86b257d663090e7de3460b4f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/046b965d86c23d86b257d663090e7de3460b4f14"}]},"branch":"refs/heads/master","description":"Rebase"},"3bb434024d635a897bf33ce9b97209d0ee2bf4a9":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-03-21 04:45:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/9"}}},"commit":{"parents":[{"commit":"d71530c2a1dc893c54efda47813100940025dc93","subject":"gcs: Remove unused fallback to oauth2client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d71530c2a1dc893c54efda47813100940025dc93"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-21 04:45:27.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bb434024d635a897bf33ce9b97209d0ee2bf4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3bb434024d635a897bf33ce9b97209d0ee2bf4a9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e037d3d863ae6228a180183c41eb86737c5647cb":{"kind":"REWORK","_number":10,"created":"2024-08-22 02:40:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/10"}}},"commit":{"parents":[{"commit":"fca6b5174081c5eca684689a0f0ad16e4a214b92","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fca6b5174081c5eca684689a0f0ad16e4a214b92"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-22 02:39:45.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e037d3d863ae6228a180183c41eb86737c5647cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e037d3d863ae6228a180183c41eb86737c5647cb"}]},"branch":"refs/heads/master"},"0c66064318e0a16d571e8a89283d007f12ab23a0":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-02-27 14:18:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/11"}}},"commit":{"parents":[{"commit":"fa553ab73344756cbc29d788150c8ea9a96721b9","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa553ab73344756cbc29d788150c8ea9a96721b9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-27 14:18:00.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c66064318e0a16d571e8a89283d007f12ab23a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c66064318e0a16d571e8a89283d007f12ab23a0"}]},"branch":"refs/heads/master"},"7fa7e31920bac6145f8ed4ec6b770e67965634ad":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2025-09-17 16:48:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/12"}}},"commit":{"parents":[{"commit":"900a0dbbec31c619cf5397dc68876bd53f343c80","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/900a0dbbec31c619cf5397dc68876bd53f343c80"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-17 16:48:23.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fa7e31920bac6145f8ed4ec6b770e67965634ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7fa7e31920bac6145f8ed4ec6b770e67965634ad"}]},"branch":"refs/heads/master"},"45225cc8bfebdfbfd31c8a7ec6cacb0133ee80ff":{"kind":"REWORK","_number":13,"created":"2025-09-18 01:07:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/13"}}},"commit":{"parents":[{"commit":"8d59d3de8e4e44bac7ab219052bfe85dc43f76b6","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d59d3de8e4e44bac7ab219052bfe85dc43f76b6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-18 01:05:40.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45225cc8bfebdfbfd31c8a7ec6cacb0133ee80ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45225cc8bfebdfbfd31c8a7ec6cacb0133ee80ff"}]},"branch":"refs/heads/master"},"1ccea2b32f91cd69a30f335a1a9f15c0ec5e8a7c":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-12-20 11:24:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/14"}}},"commit":{"parents":[{"commit":"c0e34704e544c18407765b8f4d8f5a0793a8a1c6","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0e34704e544c18407765b8f4d8f5a0793a8a1c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 11:24:30.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ccea2b32f91cd69a30f335a1a9f15c0ec5e8a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ccea2b32f91cd69a30f335a1a9f15c0ec5e8a7c"}]},"branch":"refs/heads/master"},"c1eab6f9b6dbe9bf5de4134975dd7933f55d4f2c":{"kind":"REWORK","_number":15,"created":"2026-02-07 08:57:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/15"}}},"commit":{"parents":[{"commit":"f6ebb45b5913dbc89c628b9058be0d466bc84254","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6ebb45b5913dbc89c628b9058be0d466bc84254"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-07 08:56:51.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1eab6f9b6dbe9bf5de4134975dd7933f55d4f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1eab6f9b6dbe9bf5de4134975dd7933f55d4f2c"}]},"branch":"refs/heads/master"},"282a5a1dd2f15a379bf0eb2254308f0c69bd6b4b":{"kind":"NO_CHANGE","_number":16,"created":"2026-02-13 13:36:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/16"}}},"commit":{"parents":[{"commit":"a960db0d0d12506d9b1513bc5848aa6dd71c531a","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a960db0d0d12506d9b1513bc5848aa6dd71c531a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-13 13:36:32.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/282a5a1dd2f15a379bf0eb2254308f0c69bd6b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/282a5a1dd2f15a379bf0eb2254308f0c69bd6b4b"}]},"branch":"refs/heads/master","description":"Rebase"},"16b9d08c385c590046859e97e5a29653f588805a":{"kind":"REWORK","_number":17,"created":"2026-02-15 10:05:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/17"}}},"commit":{"parents":[{"commit":"994f7e8a6594632635275c7e96d1d6d24d07f347","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/994f7e8a6594632635275c7e96d1d6d24d07f347"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-15 10:04:40.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16b9d08c385c590046859e97e5a29653f588805a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16b9d08c385c590046859e97e5a29653f588805a"}]},"branch":"refs/heads/master"},"8e0cede6744e76b0bc6c11371dcb0ed6626fc04d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-02-20 04:06:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/18"}}},"commit":{"parents":[{"commit":"d81f17e66eb07d82d658ea02ff9d8b70642e9c55","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d81f17e66eb07d82d658ea02ff9d8b70642e9c55"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-20 04:06:00.000000000","tz":0},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e0cede6744e76b0bc6c11371dcb0ed6626fc04d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e0cede6744e76b0bc6c11371dcb0ed6626fc04d"}]},"branch":"refs/heads/master","description":"Rebase"},"e1a2e347c8da81c6a75d50132c7af65586d78be3":{"kind":"REWORK","_number":19,"created":"2026-02-20 07:01:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/19"}}},"commit":{"parents":[{"commit":"d81f17e66eb07d82d658ea02ff9d8b70642e9c55","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d81f17e66eb07d82d658ea02ff9d8b70642e9c55"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-20 07:01:07.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1a2e347c8da81c6a75d50132c7af65586d78be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1a2e347c8da81c6a75d50132c7af65586d78be3"}]},"branch":"refs/heads/master"},"c19ae9392730b947c0a256b471a35fb2c306fe72":{"kind":"REWORK","_number":20,"created":"2026-07-05 15:36:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/20"}}},"commit":{"parents":[{"commit":"c355183f7b773350ebf97fea241db117e808e77f","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c355183f7b773350ebf97fea241db117e808e77f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-05 15:35:50.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c19ae9392730b947c0a256b471a35fb2c306fe72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c19ae9392730b947c0a256b471a35fb2c306fe72"}]},"branch":"refs/heads/master"},"5e4530f77f43c57ad201f3a042c64488df7178cc":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-08-19 01:16:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/902122/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/902122/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/902122/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/902122/21"}}},"commit":{"parents":[{"commit":"1a39a765fbe0c37ba917b7ca19c7ac826dc8b17d","subject":"Move boto3 to driver requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a39a765fbe0c37ba917b7ca19c7ac826dc8b17d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-29 01:18:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-19 01:16:03.000000000","tz":540},"subject":"Make google-api-python-client optional","message":"Make google-api-python-client optional\n\nThis library is required only when gcs backup driver is used. This\nchange moves out this library from requirements to extra driver\nrequirements so that users skip installing this library and its\ndependencies when not needed.\n\nPartial-Bug: #2058601\nChange-Id: Iaaedcdaf53d327be35bb3b9f586a4188204ee23d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e4530f77f43c57ad201f3a042c64488df7178cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e4530f77f43c57ad201f3a042c64488df7178cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
