)]}'
{"id":"openstack%2Fcinder~904143","triplet_id":"openstack%2Fcinder~master~Ia24938bb0db79f091854467df82dcd0981765db9","project":"openstack/cinder","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-12-21 17:04:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia24938bb0db79f091854467df82dcd0981765db9","subject":"mypy: Remove errant annotation","status":"MERGED","created":"2023-12-20 17:45:31.000000000","updated":"2023-12-25 13:54:15.000000000","submitted":"2023-12-21 17:04:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904143-mypy","meta_rev_id":"c32d1034d24e236feec2ec88061bca2d4bb42aef","_number":904143,"virtual_id_number":904143,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-21 17:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-12-20 22:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-12-21 14:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-12-21 14:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-12-20 22:54:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-20 17:45:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 17:47:06.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 18:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 18:36:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 19:11:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-12-20 19:33:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-20 19:44:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-12-20 20:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-20 20:08:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-12-20 20:19:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-12-20 21:35:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-12-20 21:43:43.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-12-20 22:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-12-20 22:54:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-12-21 00:33:39.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-12-21 09:44:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-12-21 14:43:49.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-12-21 17:33:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-22 10:05:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-25 13:54:15.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f50afaa0ca50592d0bbb7d5c0117132d9771e3f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-12-20 17:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6f9d1be225a6ff5756fc77a567a3874df70a3ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-20 17:45:42.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16295/","accounts_in_message":[],"_revision_number":1},{"id":"60e2f08501401a6c16691ffcb8ae5811c6727481","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-20 17:45:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16295/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"335235c69724471239b66f7a703fe4f76d4dda7b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-12-20 17:47:06.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-43-904143-1/ns5_nfs : FAILURE in 1m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-43-904143-1/ns5_iscsi : FAILURE in 1m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ffba45a74597c0c4e5f669a8d5030adef92e037b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 18:18:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ed37a6e5594fce876f9a1744a65ea8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92683064c743428694f5503d5a2b84e5 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f81b6b5ce3f9476eb0a340468d8c7572 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79164d4441c947fb808af532f98ed69f : SUCCESS in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec7952910ae40bd611346e4ff1d80e8ae70a99f3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-12-20 18:36:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/904143 : FAILURE in 50m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b948a4cf21969ee789fca88efa426fd2b5c8ea7f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-20 19:11:27.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0abf22ff6fa04f19946b3b74e1ccf0cb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1b4d9a2963ee4533a079439c7112f643 : FAILURE in 1h 22m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5a645d032d6bfc96c75ae1d769cf75f672acbdec","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-20 19:33:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerFlex_3.6/deefdd2/DellEMC_PowerFlex_3.6/1087 : SUCCESS in 1h 12m 23s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerFlex_4.5/31e6221/DellEMC_PowerFlex_4.5/64 : SUCCESS in 1h 47m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e06e9b3424746e9fdb57324014f661a4c79c764","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-20 19:44:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_Unity_FC/e4d174f/DellEMC_Unity_FC/1134 : FAILURE in 1h 15m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_Unity_iSCSI/fb0ad6b/DellEMC_Unity_iSCSI/1146 : FAILURE in 1h 58m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0a5259e717d38929f4d6789b97c9de9b4c64b210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 20:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ec6dc4b0cd24b9ca22cd9c307ae24d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2766350cc64eee9e2ff7c6a3eb769c : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88f5d0761dea4d2186323ad086b8858e : SUCCESS in 13m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e932492efff490d9087aa3967fc8476 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ce0b7c7d104c4094f8da3d45d2a58b : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d8cd263aa6b0498d93b97be15818f592 : SUCCESS in 1h 05m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/37bc6535ba1641ec8a502e22a5bf0670 : SUCCESS in 50m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00fb14c9e55b45feaab2f8c6f3c60fa8 : SUCCESS in 1h 38m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72fb930da59a41e08846369b4909408b : SUCCESS in 45m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/09330e011967400f8ebce14745452d8f : SUCCESS in 18m 48s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e6b6003b16174040bd2f1b7ab32def32 : SUCCESS in 14m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/de658f5b0af44dc792a3710a2f2ac2ce : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/931837a903c0467982d582650db311e3 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11db521b5dae445ab3115fc8017cd7df : SUCCESS in 6m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15791ed2bf864b7295903ef05b7ccafd : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/436c0b8ab26f4a208ce49958fca103e0 : SUCCESS in 33m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ae32d763dc6419287ae853f863848a0 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e179d42209149d691105d771d137476 : SUCCESS in 2h 11m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0deb8f378e0644428f1b9d73f43882cb : FAILURE in 1h 42m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46a92a4e5b9744ad9b2704597b93c90b : SUCCESS in 1h 25m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/97e01744fbfb4b29a9c1d1242ccc1e5c : SUCCESS in 2h 02m 02s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3e1ba23b3cf745d589d0551fa8a77be9 : SUCCESS in 1h 18m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a0acb0a21024f50b94dd4c7b871577e : FAILURE in 1h 06m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a7de4aa21dc34f49a6f131a44d090378 : SUCCESS in 53m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/353cac0013d2450ca07ca7d856055641 : SUCCESS in 1h 24m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0ed5fc091dec478da9f53831ded3430e : SUCCESS in 1h 46m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0608fcffbc684359ac6f2c7abe32e479 : SUCCESS in 1h 32m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cd6588c55c24d0997584cf48f2255ac : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"187d5cb946a293921746d55906a2354a3c0f36bb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-20 20:08:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/43/904143/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2f5894e/ : SUCCESS in 1h 46m 26s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/43/904143/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/919a86b/ : POST_FAILURE in 2h 21m 05s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/43/904143/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/f023e9a/ : SUCCESS in 2h 00m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/43/904143/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/b7ca2cf/ : SUCCESS in 1h 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1534b5db47c5308f1cfd099ef5c5dfe091822ffd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-20 20:19:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-904143-1/FC/ : SUCCESS in 2h 33m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-904143-1/iSCSI/ : SUCCESS in 1h 49m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b78cb409ee93bd60af56e236b8129db388e2dea","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-20 21:35:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/904143/1/upstream-check/cinder-tempest-plugin-ontap-fcp/d9b8e0a/ : FAILURE in 1h 32m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/904143/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/e73dffa/ : SUCCESS in 50m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/904143/1/upstream-check/cinder-tempest-plugin-ontap-nfs/111e999/ : SUCCESS in 46m 56s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/904143/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e5625b2/ : SUCCESS in 44m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"89fa48487f9c32b4a42918c72299fa6691875d97","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-20 21:43:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerStore_iSCSI/7805688/DellEMC_PowerStore_iSCSI/479 : SUCCESS in 1h 15m 50s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerStore_FC/076d6a1/DellEMC_PowerStore_FC/482 : SUCCESS in 1h 30m 51s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerStore_NFS/04ce359/DellEMC_PowerStore_NFS/488 : SUCCESS in 1h 08m 03s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerStore_NVMeTCP/88f40f4/DellEMC_PowerStore_NVMeTCP/472 : SUCCESS in 1h 40m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"465d2b0baba0a47fff8836113be8120b078121b2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-20 21:46:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerFlex_3.6/1085d19/DellEMC_PowerFlex_3.6/1088 : SUCCESS in 1h 11m 18s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_PowerFlex_4.5/8c7fa92/DellEMC_PowerFlex_4.5/65 : FAILURE in 1h 42m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"cf57eeac370936707765256f5fdd11a51ec43b5d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-20 22:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/904143/1/fc : SUCCESS in 0h 31m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/904143/1/iscsi : SUCCESS in 0h 48m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cdf79aa2bb3f74ebc7fe9b5ecf64d8a557463520","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-20 22:08:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_Unity_FC/77ba21f/DellEMC_Unity_FC/1135 : FAILURE in 1h 14m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/43/904143/1/check/DellEMC_Unity_iSCSI/3e41320/DellEMC_Unity_iSCSI/1147 : FAILURE in 2h 05m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"58dfe2dbc45bee8c243d53f76b278f9923bcb88e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-12-20 22:54:00.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1485b8f2ddacc0cab8b977505da1bfe6eccdbc49","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-21 00:33:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/43/904143/1/20231221060322/NEC-V-ISCSI : SUCCESS in 3h 09m 59s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/43/904143/1/20231221060322/NEC-V-FC : SUCCESS in 3h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"86e603a3f6b3e4a3a903cf2f6f6e59b83a7a4341","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-21 09:44:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/300/.","accounts_in_message":[],"_revision_number":1},{"id":"5768911df79fc8b4c076bc74ea3e1c0f7b81b192","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-12-21 10:41:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6116b29534b452333c827c6c9557ade83c0b2f74","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-12-21 14:43:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8aa674a3a47a6c3fe1f04cef65c6ab5b08fa4951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 15:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b9c29c53210cb9a5cb2bcd83e9ea53911e46981","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-21 16:28:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/320/.","accounts_in_message":[],"_revision_number":1},{"id":"ec4108d088711ea721be98c4654ccc517ad3eeb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:04:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f6db326ba948e394880fa8c8361602\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15341f007dd5494e9f810f224b8ce671 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d46f49a9cf9d4012838a56672783dbd4 : SUCCESS in 9m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bc1ba152b314180b0dd5c184611d425 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c3b045ef514baeb01cb47f89def802 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/a4a6f868aa594152a303e452ae5ac671 : SUCCESS in 58m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3e29a10ff100430eb28f187fe42bb0f4 : SUCCESS in 56m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff8657809a554d238b45a1fe3f669277 : SUCCESS in 1h 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c37836660a6d41ecbc0c12b933bc9ba8 : SUCCESS in 44m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a5f8bab7744344e4b4f9a04284c3c63f : SUCCESS in 1h 29m 25s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3513a89e09f466eb307c0a614490eeb : SUCCESS in 1h 51m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d1bf7e12e1245d0b4ed56631b811396 : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bea58afb2d4e069e3e0a479e37c401f51283dc2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:04:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3a89a2301e9b68ec0c08fb06c168aff700f683f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65302b838c004208bed7acd7cc249a4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c62cdc08fc47c999681d3e96741024 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4f2cbe66a3d59274c571d3306a09c865c49faee1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-21 17:33:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/43/904143/1/20231221181720/NEC-FC : SUCCESS in 3h 46m 11s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/43/904143/1/20231221181720/NEC-ISCSI : SUCCESS in 8h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"99b3fdb1039dfdeac28e551a7a74c96e5f0e6236","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-22 10:05:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a1618620d1474ecda03a7c5ae9c493f2 : SUCCESS in 1h 10m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9bf288db03fd42718ba52f0fd7f08d9a : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c32d1034d24e236feec2ec88061bca2d4bb42aef","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-25 13:54:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/43/904143/1/ : FAILURE in 1h 19m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e162a95d6520c0555265c0b4aaf44c3e85f53011","revisions":{"e162a95d6520c0555265c0b4aaf44c3e85f53011":{"kind":"REWORK","_number":1,"created":"2023-12-20 17:45:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/43/904143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/43/904143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/43/904143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/43/904143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/43/904143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/43/904143/1"}}},"commit":{"parents":[{"commit":"e9f94157895516c1c75052f6c982d9cb4aaeaa2b","subject":"Merge \"RBD: tpool.Proxy client object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9f94157895516c1c75052f6c982d9cb4aaeaa2b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 17:41:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-12-20 17:41:18.000000000","tz":0},"subject":"mypy: Remove errant annotation","message":"mypy: Remove errant annotation\n\nThe type can inferred and what we have specified is too broad, resulting\nin the following error on recent mypy versions:\n\n  error: Never not callable  [misc]\n\nChange-Id: Ia24938bb0db79f091854467df82dcd0981765db9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e162a95d6520c0555265c0b4aaf44c3e85f53011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e162a95d6520c0555265c0b4aaf44c3e85f53011"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
