)]}'
{"id":"openstack%2Fcinder~904453","triplet_id":"openstack%2Fcinder~master~I5f1123d24e450595d69fc1bd320432eefd61c325","project":"openstack/cinder","branch":"master","topic":"bug-2047693","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-01-08 06:55:29.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I5f1123d24e450595d69fc1bd320432eefd61c325","subject":"WIP: Add test for cinder-manage quota sync","status":"NEW","created":"2023-12-29 16:42:29.000000000","updated":"2024-01-08 06:55:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de32e4703ebc05476afbc257b0ec9db3d0853148","_number":904453,"virtual_id_number":904453,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-29 19:35:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-29 16:42:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-29 17:02:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-29 18:13:54.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-12-29 18:54:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-12-29 19:01:37.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-29 19:11:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-12-29 21:05:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-30 00:47:37.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-01-08 00:29:12.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-08 03:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"9183e1da244fd978e2272e33b625cf521769e4a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-29 16:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52f9f927f9397d501e8c94ea38a8c21947132aba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-29 16:42:38.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16322/","accounts_in_message":[],"_revision_number":1},{"id":"e669e993526c3b4e361453c03c800cdda422275d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-29 16:42:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16322/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"46eecfa38891c7a502e9a5f7a87f1faef590eea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 17:02:45.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57abb92dae1c4e84b5d700b5ef1653d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8480612cecab4cdd92862c5279f58cf8 : FAILURE in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81663b1ab7dc4a498162c01eb307bdac : FAILURE in 12m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a3e6307039945be8593ebea39205f4b : FAILURE in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d31268835284a1fa504de07cb363411 : SUCCESS in 13m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/87f17942d8c6456abe8337ec81ae39bd : FAILURE in 13m 52s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/73419e23acc340b8be89725ee51d4f06 : FAILURE in 12m 59s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a8b3df904f3845ae926f194b20c7572f : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a10d729c9adebfefbacac83893b820ef34c23fb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 17:03:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e386694407d441a6b7cf0a668e4b1045\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d1ccbca2e054df38aa777d43e1674f6 : FAILURE in 15m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fac0cc6948544326a4239022b6ee766a : FAILURE in 9m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f7e65f9141047fdb35687dc4aab7be6 : FAILURE in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"975b205aafe95aaef70f2d86fc329c99b4ea6ecf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-29 18:13:54.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/28aa07472ce14d12a1801f984ff31e18\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b13083dcc446433481fe66207565d2dd : FAILURE in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4bf3e017fd69a64939e4af7bad8e0e5aac426bf1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-29 18:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/343/.","accounts_in_message":[],"_revision_number":1},{"id":"882c871f7f0c129ea0d29e084041f51eb2a0f5f2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-29 19:01:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b83d1dc60ab64d538495d4980ec6e276 : SUCCESS in 1h 10m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1785cf70fa974c44a5e433b3ac528754 : SUCCESS in 1h 07m 06s","accounts_in_message":[],"_revision_number":1},{"id":"26e175835bf2b9677c5e8deb1f4c99a0b9c144f9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-29 19:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/328/.","accounts_in_message":[],"_revision_number":1},{"id":"619b0a8118c57af760bf20bf0bd94a79babb0005","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-29 19:11:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/53/904453/1/check/DellEMC_Unity_FC/37b8027/DellEMC_Unity_FC/1178 : FAILURE in 1h 01m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/53/904453/1/check/DellEMC_Unity_iSCSI/ad0bad7/DellEMC_Unity_iSCSI/1194 : SUCCESS in 1h 21m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ac381b0cb7113229ba86c8288332a32266b192f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-12-29 19:16:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ed063d9529cc581a214d5f07f302723114292f1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-29 19:16:16.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16323/","accounts_in_message":[],"_revision_number":2},{"id":"ad3f6b5f8382f6fa01d9971247871a1f2bf62eab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-29 19:16:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16323/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0a1133c9999fadf69870da8dceaa954ce65ba399","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 19:28:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4dd7e1a98034baf88a6e11342fd56dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9198826936504adebe9bc6471554d23e : FAILURE in 8m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/adb260b57bc6409b8428c2cc060a62eb : FAILURE in 8m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b30c7b8263e48d9a45404d977c8f0cd : FAILURE in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b13062645b1a71791f517337cf9d14d80ec75315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 19:35:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d3351f7a23d40698b1f0ff252da1ea9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d8f300603940c3ab608ce028c4f89b : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf2c7ee7a3a4873b3d346dadbc7a957 : FAILURE in 10m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/054e2e4bd0d2422cbbeffa0b7c58cbb9 : FAILURE in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee939901babd42e09cfc4cd31de810b9 : SUCCESS in 12m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eccc8260a52c4422954f3a208dc36e0d : FAILURE in 13m 49s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6bab570bd1bf4ebda45a07ae3c7f736a : FAILURE in 12m 54s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/351334a51e074d61b09e75bcf9119b5b : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d7cf4f8f87e678f6480c635cc082e62b79f475f7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-29 20:32:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/344/.","accounts_in_message":[],"_revision_number":2},{"id":"dc4ce844469b15ab5326f4dcc323b48758cea6a7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-29 20:32:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/99ea295c652b45d6aa0ec70c23426cc6 : SUCCESS in 1h 14m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3217d62d8b504a86b22039eb2db35f0b : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":2},{"id":"050ed44122bf37fa4978349404f1c744c5def80f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-29 20:41:57.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/53/904453/2/check/DellEMC_Unity_FC/c2fe891/DellEMC_Unity_FC/1179 : FAILURE in 1h 02m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/53/904453/2/check/DellEMC_Unity_iSCSI/a877f80/DellEMC_Unity_iSCSI/1195 : SUCCESS in 1h 25m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8aa7bfe395e673efbd2d5cf358d38abec16b73bf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-29 20:45:18.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/eac2f6810a7948aab9c4160c371ed10d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/af61a5fc54594f7f8b022da98a1d3f29 : FAILURE in 1h 24m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4c055ab8192c377e961a425205c19881a459217d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-29 21:05:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/53/904453/2/check/DellEMC_PowerFlex_3.6/2966970/DellEMC_PowerFlex_3.6/1132 : SUCCESS in 1h 12m 24s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/53/904453/2/check/DellEMC_PowerFlex_4.5/bbd2eea/DellEMC_PowerFlex_4.5/108 : SUCCESS in 1h 48m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c3cf755cf631cd9bb47bc16c12a039c89b3916da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-29 22:02:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/329/.","accounts_in_message":[],"_revision_number":2},{"id":"0fac93230b84aad5ed3be4810c686d84c666169c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-30 00:47:37.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-53-904453-1/FC/ : FAILURE in 7h 47m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-53-904453-1/iSCSI/ : SUCCESS in 2h 49m 12s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"246e2b9d182591f10d4eef9d607e438e0e8e6c4f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-01-08 00:29:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/53/904453/2/ : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d291021cb18649af9da7c1dc7f154fb47bdd40ef","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-08 03:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/53/904453/2/fc : SUCCESS in 0h 43m 29s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/53/904453/2/iscsi : SUCCESS in 0h 43m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ab7acab5592a182759ab15a06f69ae521a9ceb63","revisions":{"c1cc16c7892b06c5253379d94420fd9ea08d67cd":{"kind":"REWORK","_number":1,"created":"2023-12-29 16:42:29.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/53/904453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/904453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/904453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/904453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/904453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/904453/1"}}},"commit":{"parents":[{"commit":"4e511e5b4b6ad44bf35600ae70e1646496d66602","subject":"Fix error in cinder-manage quota sync cmd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e511e5b4b6ad44bf35600ae70e1646496d66602"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-12-29 16:40:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-12-29 16:40:36.000000000","tz":60},"subject":"WIP: Add test for cinder-manage quota sync","message":"WIP: Add test for cinder-manage quota sync\n\nTODO: Add proper mocking and a fake volume in a project.\n\nReleated-Bug: 2047693\nChange-Id: I5f1123d24e450595d69fc1bd320432eefd61c325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1cc16c7892b06c5253379d94420fd9ea08d67cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1cc16c7892b06c5253379d94420fd9ea08d67cd"}]},"branch":"refs/heads/master"},"ab7acab5592a182759ab15a06f69ae521a9ceb63":{"kind":"REWORK","_number":2,"created":"2023-12-29 19:16:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/53/904453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/904453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/904453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/904453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/904453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/904453/2"}}},"commit":{"parents":[{"commit":"4e511e5b4b6ad44bf35600ae70e1646496d66602","subject":"Fix error in cinder-manage quota sync cmd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e511e5b4b6ad44bf35600ae70e1646496d66602"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-12-29 16:40:36.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-12-29 19:11:31.000000000","tz":60},"subject":"WIP: Add test for cinder-manage quota sync","message":"WIP: Add test for cinder-manage quota sync\n\nTODO: Add proper mocking and a fake volume in a project.\n\nReleated-Bug: 2047693\nChange-Id: I5f1123d24e450595d69fc1bd320432eefd61c325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab7acab5592a182759ab15a06f69ae521a9ceb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ab7acab5592a182759ab15a06f69ae521a9ceb63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
