)]}'
{"id":"openstack%2Fcinder~906140","triplet_id":"openstack%2Fcinder~stable%2Fxena~I80923f9968de31738c554b77c0942a6182f0e67c","project":"openstack/cinder","branch":"stable/xena","topic":"bug/1982405-stable/xena","attention_set":{},"removed_from_attention_set":{"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-03-06 11:35:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I80923f9968de31738c554b77c0942a6182f0e67c","subject":"Fix Infinidat driver generic volume migration","status":"ABANDONED","created":"2024-01-22 15:56:17.000000000","updated":"2024-03-06 11:35:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":889240,"cherry_pick_of_patch_set":2,"meta_rev_id":"c7d1900357d314abac48e90b782dedeeb9cf7238","_number":906140,"virtual_id_number":906140,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-22 17:34:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 16:29:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-01-22 16:33:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-22 16:52:19.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-22 17:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 17:41:07.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-03 11:00:36.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 11:35:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"81610d77b633f53f2b144cafe6214de6d43f5266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-22 15:56:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"aa581f39ab9bdaf662e99e2b211e768c49dc208b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-22 16:29:16.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/faedb12dd52346449ef29996f8638bf9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7535503d06ce4138877f59eb2042f470 : FAILURE in 24m 51s","accounts_in_message":[],"_revision_number":1},{"id":"35972250725ccde530ff1fadd2c991e982c0afbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:33:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c27f8878692f49f387e44b540ccd3caf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71a5741cf89d4c24b739491e6079a1ca : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56226281fae140a58d7aedb123d2edfc : FAILURE in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71fb362cd729781930b56df7c0da426d17b79966","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-22 16:52:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/906140 : FAILURE in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b5c093f33da7c8328cc854067d985f91c405af2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:34:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82b5b5bcba97417f96c8dc3b95584f0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243ee95c595d470ca6ce6efcd89bc52d : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08d1a558ba044d478b0bcac332a76ab9 : SUCCESS in 10m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1cf3976da4f4fc990a4e99c391e649f : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ec8f150ef2a4cb1b683ec182e417990 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ac1fbf63264ffda71559ad019c6cdb : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/10e07409500b4aaaacec746397de3935 : SUCCESS in 1h 01m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b06e38201d14f29833952102c3495b5 : SUCCESS in 1h 18m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9578b292d50f4a6fa3315028e590cd4b : SUCCESS in 49m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffd1df823a8d416d8e8b506ccea7b7df : SUCCESS in 8m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/81a22c4c7b614aa09bdfd770a77cc96f : SUCCESS in 12m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/010412b04b384defbbff8ea90f3a8b06 : SUCCESS in 6m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7a6aa2185e745088c4cb4f865af5757 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ddc9d57695894b7699043767e7d82c61 : SUCCESS in 6m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61829f6ac9db4b76ba175bd7ee7062d2 : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bbd3684f9ac94b028be67b97161189bc : SUCCESS in 33m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a335958a6bc7493690c477dad9158038 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6201bf5fc3c45beabebbb44e407738d : SUCCESS in 1h 26m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5f315d9e45874afe82cc0b66d4d4e449 : SUCCESS in 1h 20m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e12045d7e4064d609bd14af78ae399c7 : SUCCESS in 1h 20m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a8ba913bc7249419bb554c76a3d02cc : SUCCESS in 1h 18m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b9be897e9564febbf6de6f6c7800fe4 : SUCCESS in 1h 14m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ae6ea51781b3402596f4ba35642d27e4 : SUCCESS in 48m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2300e21d59254ff09221e97b6dcb49dd : SUCCESS in 58m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57d4be2bbcee4efdb24cbad546f46918 : SUCCESS in 1h 21m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/333be1be99af4656920ccddb5cc39188 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b09b6f9812355632d6d6286c233e5ec45e3536e3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-22 17:41:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-906140-1/FC/ : SUCCESS in 1h 44m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-906140-1/iSCSI/ : SUCCESS in 1h 24m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"cad04a4f0cf6b063c91334608a56fa090ac452ee","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-03 11:00:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/906140/1/20240203160712/NEC-ISCSI : SUCCESS in 3h 52m 12s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/906140/1/20240203160712/NEC-FC : SUCCESS in 3h 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c7d1900357d314abac48e90b782dedeeb9cf7238","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 11:35:39.000000000","message":"Abandoned\n\nstable/xena branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/xena if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c337a1e6cfc3ae92939bfce34bf78bc22f5521e8","revisions":{"c337a1e6cfc3ae92939bfce34bf78bc22f5521e8":{"kind":"REWORK","_number":1,"created":"2024-01-22 15:56:17.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/40/906140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/906140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/906140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/906140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/906140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/906140/1"}}},"commit":{"parents":[{"commit":"77a135866b90009ba376b2f7dd80f81a993f09f5","subject":"Fix Infinidat driver to use TLS/SSL communication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77a135866b90009ba376b2f7dd80f81a993f09f5"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-07-06 13:12:47.000000000","tz":240},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2024-01-22 15:56:17.000000000","tz":0},"subject":"Fix Infinidat driver generic volume migration","message":"Fix Infinidat driver generic volume migration\n\n- Use volume name_id to resolve Infinidat volume name.\n- Add update_migrated_volume function to fix generic volume\n  migration between two pools within the same storage cluster.\n\nCloses-bug: #1982405\nChange-Id: I80923f9968de31738c554b77c0942a6182f0e67c\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\n(cherry picked from commit 9a29d57a61cfb845eabc2cc7f72480b48b9a68e9)\n(cherry picked from commit 40ee7d0739fd65258207709511842b5eddbac4b7)\n(cherry picked from commit 9ccf691cb6a72b6e23c3095bf12909498c430465)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c337a1e6cfc3ae92939bfce34bf78bc22f5521e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c337a1e6cfc3ae92939bfce34bf78bc22f5521e8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
