)]}'
{"id":"openstack%2Fcinder~906281","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I80923f9968de31738c554b77c0942a6182f0e67c","project":"openstack/cinder","branch":"stable/wallaby","topic":"bug/1982405-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-03-06 11:29:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I80923f9968de31738c554b77c0942a6182f0e67c","subject":"Fix Infinidat driver generic volume migration","status":"ABANDONED","created":"2024-01-22 16:00:26.000000000","updated":"2024-03-06 11:29:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":906140,"cherry_pick_of_patch_set":1,"meta_rev_id":"814022cf2b93c9d1e78a7ec0210f1a587aad2850","_number":906281,"virtual_id_number":906281,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-23 14:26:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 16:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-22 16:33:33.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-01-22 17:11:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-22 19:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 14:48:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-06 11:29:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"bceb5054e54da073193c23cc81641461038f7d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-22 16:00:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"335feaffccc121f173f9984f2997324cc7d90cf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9646d883f9504a90a4b3f4893791d9b1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0ce8fb6893a499ebac88a2103b164ad : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe7c8e55d0d4432c8c18c7cdbd337ea3 : SUCCESS in 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e207051d6e3d0088127f8f41a68dfd005e81c0e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-22 16:33:33.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8773bcbeabac4eb6ae5425b1d6acb12a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9724f6d1ac744f398e800889b67e6e3f : FAILURE in 23m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b794f61ac32d14c311177a9a10502bbebd079b98","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-22 17:11:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/906281 : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"720b9a063d8c1e2fc8d4039a03c3004db4b03558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 19:00:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5fd362b1b04492ca0e5e632bcb79ce9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8460ed20cd452bb9501eb06d3dd3bc : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21839c8d18f049ec9fc991794d82de6b : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5c864074bc41438d4a0f3aac7e880c : FAILURE in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c211bc0d95dd4c819f8d76ffe671db28 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cedf3fd315fd48e7a74daecf73c217ba : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33d06bdea709424e83903148aca19cfe : SUCCESS in 8m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/db5ff132ef91468e854ef1457118234c : SUCCESS in 12m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/64ef5ed1bd1d459fb6692bb1a4de46bc : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/27188c09784f4be589bcbdfede17f698 : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c91f14fc4714894b122b929169c2a49 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d97f017fa681409792bdefafd60abe45 : FAILURE in 46m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/826de9c8f9ca48be967b007fd9f3a768 : SUCCESS in 5m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eab4fa4b55d441d39ac371cb3c0e76e9 : FAILURE in 52m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/62d8a731aa08425f8e84dd49b76a5014 : FAILURE in 1h 17m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df315771d36f49d6a435efb70c99fbd4 : SUCCESS in 1h 55m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a1a536c1561e4de0883369729ab76033 : SUCCESS in 1h 14m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2624decac80b41a0a056f17831b9f526 : SUCCESS in 1h 14m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9bc6ad3198b0433796235c10a8c65fdb : SUCCESS in 2h 34m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0988012dd07745db819ea74494d67cf2 : SUCCESS in 1h 11m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9df8416abeaa4733a7f8d842ad13e52f : SUCCESS in 47m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bd3737063e04613afbae0fb525070e7 : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":1},{"id":"99943ab5921dfcf3c64a16b7e1947cbec7520bfa","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-23 11:49:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fa78a82ed0d327e48711f1b1cbbabc8af3fda9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 12:04:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb06f08de2f40689f2c75db20785182\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ffbff1606344dd3a263ab34b5de9e33 : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09c57c200b184aaa885eae084101cf77 : SUCCESS in 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04d0ac48ef46a6423f9cea120448e1b13aba6211","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-23 12:18:58.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/389c822cf6c74dd18cb6b8dbf96b6df5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6434aee446fb45e287ca2e53c3729967 : FAILURE in 24m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f0ab416925c7c993ca3ddea7276d2b6ab0afece9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 14:26:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c1a3c01812479ca924d91c15fac2ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed9e48e84d747ec86388262964b938d : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d389734c2b4e1183de7486d550a99a : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bc335d1764f4815a3e4ea2668d67e41 : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7aa32864e95947788c10c6510dd22ebf : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355d9e1597fd4a9b90596fde79f9cec9 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7efae294c144e43a4d98f0ecd67f4c9 : SUCCESS in 6m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3d31139cc8934ac58836ca6bd9f6f4c8 : SUCCESS in 12m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f2fe8f1e8e64cdfb0fd4807016e8c21 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34f3e03e77254c039dd94eb5e9b37a50 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88e779cc420c4742af1e433b757e99a1 : SUCCESS in 7m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/49a1fd993bd54c8f9e8321b99ea2644d : FAILURE in 36m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68409c69897f4aecab1d46e67a17a6b1 : SUCCESS in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e5547496be245f79ccf18da1b5b2494 : SUCCESS in 1h 10m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a0f8ac9eb138477895431cd3cc73c635 : FAILURE in 1h 15m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f05849d9a934f20b42ab12334fa6064 : SUCCESS in 1h 15m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38c47f97a6084f1e9e316e5b889ed1eb : SUCCESS in 1h 03m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab875b08ac8e41f0a63192a625955861 : SUCCESS in 1h 47m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78c1c532181b46d3870a6f0d0137143a : SUCCESS in 2h 30m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b30d95007aa4f89b8102eb3102c031e : SUCCESS in 59m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2462f9275261477db7b4cd9c2889a407 : SUCCESS in 58m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8154b5aab5ed4687a1498e19c7a6103d : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7b3c55c9807debda1201a19787ff5ee3638a13d5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-23 14:48:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/906281/1/upstream-check/cinder-tempest-plugin-ontap-fcp/c64e729/ : FAILURE in 11m 34s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/906281/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/fb658e8/ : FAILURE in 9m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/906281/1/upstream-check/cinder-tempest-plugin-ontap-nfs/247e51a/ : FAILURE in 11m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"814022cf2b93c9d1e78a7ec0210f1a587aad2850","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 11:29:41.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27f3d542c4aadf3db3affe4762cd6b07923f1415","revisions":{"27f3d542c4aadf3db3affe4762cd6b07923f1415":{"kind":"REWORK","_number":1,"created":"2024-01-22 16:00:26.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/81/906281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/906281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/906281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/906281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/906281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/906281/1"}}},"commit":{"parents":[{"commit":"8e6d782d0a645561e55a8506419ca73fa18ebf1d","subject":"Fix Infinidat driver to use TLS/SSL communication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e6d782d0a645561e55a8506419ca73fa18ebf1d"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-07-06 13:12:47.000000000","tz":240},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2024-01-22 16:00:26.000000000","tz":0},"subject":"Fix Infinidat driver generic volume migration","message":"Fix Infinidat driver generic volume migration\n\n- Use volume name_id to resolve Infinidat volume name.\n- Add update_migrated_volume function to fix generic volume\n  migration between two pools within the same storage cluster.\n\nCloses-bug: #1982405\nChange-Id: I80923f9968de31738c554b77c0942a6182f0e67c\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\n(cherry picked from commit 9a29d57a61cfb845eabc2cc7f72480b48b9a68e9)\n(cherry picked from commit 40ee7d0739fd65258207709511842b5eddbac4b7)\n(cherry picked from commit 9ccf691cb6a72b6e23c3095bf12909498c430465)\n(cherry picked from commit c337a1e6cfc3ae92939bfce34bf78bc22f5521e8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27f3d542c4aadf3db3affe4762cd6b07923f1415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27f3d542c4aadf3db3affe4762cd6b07923f1415"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
