)]}'
{"id":"openstack%2Fcinder~906292","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~Ibda3a09a181160b8ee9129795429a7f1795e907d","project":"openstack/cinder","branch":"stable/wallaby","topic":"bug/1982350-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-03-06 11:29:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibda3a09a181160b8ee9129795429a7f1795e907d","subject":"Fix Infinidat driver multi-attach feature","status":"ABANDONED","created":"2024-01-23 10:45:05.000000000","updated":"2024-03-06 11:29:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":906284,"cherry_pick_of_patch_set":1,"meta_rev_id":"c03d58ac8896cdd045ba97228d1467faed481154","_number":906292,"virtual_id_number":906292,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-23 12:57:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 11:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 11:13:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-01-23 12:55:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-01-23 12:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 13:47:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-06 11:29:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"4fceb960a35146325dca2a8c389c3b2871062195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-23 10:45:05.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"076700bc689ba88105d955d7a756743143d4f8c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:08:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dc83d4ec05e412a9699b85805a4722f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a29d6539c654a1f917202d8b34b7c0e : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24041fbdba514780b1c13bac665b1c33 : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fce780a89f74d8afa99fd612477f0783aebde798","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-23 11:13:28.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/20c652c5e435456880821bb7b647e999\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/634017c364b543fdb285ea928f165d71 : FAILURE in 23m 57s","accounts_in_message":[],"_revision_number":1},{"id":"81928479a1f455a632eeda7ebd0aca21ea8785eb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-23 12:55:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-906292-1/FC/ : SUCCESS in 2h 9m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-906292-1/iSCSI/ : SUCCESS in 2h 4m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"eb8afe1974f42f0a950d79488b3ead18ed744173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 12:57:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b7886dfbb254bfa91b79a76a4b7a54e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e6e4b782274a439bf803c8670493f0 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f53af78a1b7544d4bf1511cc6c2418fe : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3469086dd45a46b6b9669e172c0eb5fd : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c265575d339f48e2b8cc3e6b3c6b4044 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca25a2eac895405cb902483a17063be1 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc354f18567746f1924601a3b174765b : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad3a09df410c4b7689d49b99a4e5287f : SUCCESS in 12m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58a8f375887f495d98b3df2f4f50f091 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/869cdab4c2a04b6ab02f680ade51db98 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9f71e9807f74ba78cc37dd6f1194182 : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/afe9bd0e5ebc4751b0f7bde303ae3cf9 : FAILURE in 26m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88ed63ef73b2456e96d705dd416688a2 : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/884ce6bad0a94ddcaba385e9146bbe10 : FAILURE in 55m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/35d45e8d4f9343a9b5e5b851f72d1cda : SUCCESS in 1h 26m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fec5bbf144e54a9ab932aff2f80913f5 : SUCCESS in 1h 22m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/26fdc1e458a74e3eaf1c7b9e263e62d4 : SUCCESS in 1h 16m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c8107d3fc0f64b9495f4e22ad861169b : SUCCESS in 1h 14m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29150a29b8a9456ba39c27e5e05c55d9 : SUCCESS in 2h 08m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8bf8ede2ff7742efab20d3818de4f661 : SUCCESS in 1h 03m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e614c21caad4d78826a4d4e6c38fb2d : SUCCESS in 1h 01m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4691b0046be4cc4b3a0aa06f51bfd13 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"214c59c464d1ad5bf3ae8a9643e590ba4417e239","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-23 13:47:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/906292 : TIMED_OUT in 3h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c03d58ac8896cdd045ba97228d1467faed481154","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 11:29:56.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8288d3b9154e33aaca44b0833413ed80911ad05","revisions":{"a8288d3b9154e33aaca44b0833413ed80911ad05":{"kind":"REWORK","_number":1,"created":"2024-01-23 10:45:05.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/92/906292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/906292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/906292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/906292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/906292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/906292/1"}}},"commit":{"parents":[{"commit":"27f3d542c4aadf3db3affe4762cd6b07923f1415","subject":"Fix Infinidat driver generic volume migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27f3d542c4aadf3db3affe4762cd6b07923f1415"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-07-21 10:57:05.000000000","tz":0},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2024-01-23 10:45:05.000000000","tz":0},"subject":"Fix Infinidat driver multi-attach feature","message":"Fix Infinidat driver multi-attach feature\n\nAdded a check if there are multiple attachments to the volume\nfrom the same connector and terminate connection only for the\nlast attachment from the corresponding host.\n\nCloses-bug: #1982350\nChange-Id: Ibda3a09a181160b8ee9129795429a7f1795e907d\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\n(cherry picked from commit d8ca81e7c911add4f99f888d7315130ac3b60e22)\n(cherry picked from commit 8e5330c029b538252e8743f7dd2276f423a8ec1b)\n(cherry picked from commit ea95cbabd9348fcbac79b9e5d1d4de2ab0f68c77)\n(cherry picked from commit c26244d47a9816fdd88468741547229f2202b2dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8288d3b9154e33aaca44b0833413ed80911ad05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8288d3b9154e33aaca44b0833413ed80911ad05"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
