)]}'
{"id":"openstack%2Fcinder~906393","triplet_id":"openstack%2Fcinder~master~I590eb954203ccd2fc564117addfcf5d2dac5bd74","project":"openstack/cinder","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-03-15 15:42:32.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-03-15 17:36:58.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-03-15 17:36:58.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-03-15 14:45:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I590eb954203ccd2fc564117addfcf5d2dac5bd74","subject":"Run protection tests during gate check","status":"MERGED","created":"2024-01-23 18:06:07.000000000","updated":"2024-03-15 23:00:07.000000000","submitted":"2024-03-15 17:36:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906393-secure-rbac","meta_rev_id":"bf0d6095f0118562cb62b3c53b644fd704a64528","_number":906393,"virtual_id_number":906393,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 17:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-03-15 15:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-03-15 15:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-15 14:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-03-15 15:42:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 18:06:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 18:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 18:34:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 19:32:12.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-01-23 19:38:54.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-01-23 19:44:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 20:24:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-23 20:52:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 20:53:24.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-01-23 21:02:16.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-01-23 22:01:38.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-01-23 22:51:33.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-01-24 01:21:00.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-01-24 01:28:16.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2024-01-24 02:28:16.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-04 09:34:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-05 16:41:45.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2024-02-05 16:54:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-16 15:00:47.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-02-17 05:18:23.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-17 07:20:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-17 07:23:32.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-17 07:24:21.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-20 01:16:42.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-15 15:07:44.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-03-15 23:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"5acac511f0f685e8ce236dd86aa365dcb0e904ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-01-23 18:06:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d11779c421fb83c4ceb9b50100ff38811c12d812","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-23 18:06:17.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16396/","accounts_in_message":[],"_revision_number":1},{"id":"920cdca529bbbe5b7e8cf3aefbda37f341b4e2f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-23 18:06:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16396/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5fee19df9f85acf4a2c6ddb845bea9d3e61fea8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 18:33:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a9c956a86b64a608520dc1fd49299c0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab415e1eb6fb448096fbe9b52b0f87dc : SUCCESS in 22m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a6f453d732cf47fbbb6d47f56fd3e0a3 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d12860907624c5fb79d7619b79b7bb9 : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c35dddb4a6607d9601250f3f30aebf0c63faf8bf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-23 18:34:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/906393 : FAILURE in 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"241d476cad3e49806b1ee7fe8ba1d3064e81308c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-23 19:32:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/480/.","accounts_in_message":[],"_revision_number":1},{"id":"b8f33c264f2e240ef7fbc4ebb2d44bdc79c6ea18","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-23 19:38:54.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3f61f19d89014cf1ac7ad65f1b0270aa\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f5ed0e60e56b41e6a7b1ba14b7867062 : FAILURE in 1h 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e84832ccb2d0674d385f74de7c7c3ef60431ba46","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-23 19:41:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/478/.","accounts_in_message":[],"_revision_number":1},{"id":"512ec3b0c69d260129427ca63f7936d4a03cfdc7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-23 19:44:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bb554619e27c4c99a8c9492163473b82 : FAILURE in 1h 36m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/47ab1b48101146ef91a5cb8b84930ae2 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":1},{"id":"18602d506f3ee7fb2109b77705aeb6f510965698","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-23 20:24:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerFlex_3.6/cb63451/DellEMC_PowerFlex_3.6/1273 : SUCCESS in 1h 14m 43s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerFlex_4.5/eacb5b8/DellEMC_PowerFlex_4.5/243 : FAILURE in 2h 17m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9161bb55e9858cd8aee1f7ddaa39005cc052e458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 20:52:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef3d414e0184cbbbd5dbd81f68eaf83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152cbb44c57b42298873809b101f3a6d : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be2f8427aa394ad2bb7dc4d8c52fea76 : SUCCESS in 11m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/add1758befb143538e81f24f7f0cf192 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08e84a735b24466b6223fe4a6cd1ce6 : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/87a974e05f0246fcbbb330c23f51947d : SUCCESS in 1h 05m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b1823d99b3742a58452a4eeb306f321 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e9ce61d0a8245cd9dd11db8ad5657a9 : SUCCESS in 1h 33m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04d4a738d633489fb1fa5a185fedf236 : SUCCESS in 42m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f2e325c8c42443e097e9f0be40a69821 : SUCCESS in 16m 20s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c2bcd10fc6aa4b92954069edb5a0e891 : SUCCESS in 12m 24s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c44fa118f0c7430ab5d3ac696d1a06b6 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01218802f9544728b5c4d6b2135ac6c3 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/84b744c7bbf442f0ad3b47baaffc274a : SUCCESS in 7m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b434d6a915e407dac52300f73603aca : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/026a2524f505430691b0b717cef27548 : SUCCESS in 39m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1299354ab52145099c4dc8261f74f243 : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b52c3b895574c499cb875d6af9a6e7e : SUCCESS in 2h 07m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d4b5fe28f5a741c2b8d539a6f592d1b2 : FAILURE in 2h 11m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5cc69f0d42c4931a00d99796f55a62b : SUCCESS in 1h 41m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/693d99a16fbe4a53a9a03f5112010a4e : SUCCESS in 2h 39m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c2b95337f684f51a975ef8fbb9c1226 : SUCCESS in 46m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd4d68e5b698456fbdefa85ac6f200a0 : SUCCESS in 1h 14m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/42db53e2de55437dac3a8fa18be21b79 : SUCCESS in 1h 27m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/df82f914ea1749a7a3d814eb5a9d20bd : SUCCESS in 39m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd271d23f6024f5e8ec4f40b2d79d3b7 : SUCCESS in 1h 29m 28s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0f53d888777c486aaa3d06fc6cd1f4cf : SUCCESS in 1h 21m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/742ed9c8e6534f64a457335e580f28f4 : SUCCESS in 1h 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d84d9773cbc45e6ba184051f64b65dd : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9ca30d41798cc9a9e2b5557c3a79e0198168e080","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-23 20:53:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-906393-1/FC/ : SUCCESS in 2h 46m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-906393-1/iSCSI/ : SUCCESS in 2h 1m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"de2b25eddc16f7c9f51573451de72df5d9774a95","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-01-23 21:02:16.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"006e3b50d08dc4e6923ac72f942fe75cd54da7a2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-23 22:01:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/1/upstream-check/cinder-tempest-plugin-ontap-fcp/f2f2df3/ : SUCCESS in 1h 07m 55s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/71f00e1/ : SUCCESS in 43m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/1/upstream-check/cinder-tempest-plugin-ontap-nfs/40759a0/ : SUCCESS in 43m 18s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9acad97/ : SUCCESS in 43m 27s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f3160de0cc781bf4a2fef9b101bf8372d402a590","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-23 22:51:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerStore_iSCSI/2d4f529/DellEMC_PowerStore_iSCSI/972 : SUCCESS in 1h 07m 05s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerStore_FC/ad8b46b/DellEMC_PowerStore_FC/1128 : SUCCESS in 1h 27m 03s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerStore_NFS/d3ccdf4/DellEMC_PowerStore_NFS/658 : SUCCESS in 57m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerStore_NVMeTCP/bd964b3/DellEMC_PowerStore_NVMeTCP/524 : FAILURE in 1h 58m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"e371702eb95e6cab137c705f628d6e14d1cf4ea9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-23 23:01:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerFlex_3.6/f7ceff9/DellEMC_PowerFlex_3.6/1274 : SUCCESS in 1h 09m 58s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerFlex_4.5/cdec809/DellEMC_PowerFlex_4.5/244 : FAILURE in 2h 08m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1218f112fd22a402dd3e246a132ac80d1c23405d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-01-24 01:21:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/906393-1/HPMSA-iSCSI/ : SUCCESS in 2:06:42\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/906393-1/HPMSA-FC/ : SUCCESS in 1:37:25","accounts_in_message":[],"_revision_number":1},{"id":"cebd222bac809d74783c9de5a2ec15bb85fcce3d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-01-24 01:28:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/906393-1/PVME-iSCSI/ : SUCCESS in 2:05:15\n- PVME-FC https://os-logs.seagate.com/dell-ci/906393-1/PVME-FC/ : SUCCESS in 1:55:14","accounts_in_message":[],"_revision_number":1},{"id":"eed40b2bba4b895479ac4819f25ad88a2fc1da90","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-01-24 02:28:16.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerMAX_FC/e08b1b8/DellEMC_PowerMAX_FC/311 : SUCCESS in 4h 28m 29s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/93/906393/1/check/DellEMC_PowerMAX_iSCSI/561bd58/DellEMC_PowerMAX_iSCSI/333 : SUCCESS in 5h 35m 02s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"cac330800d7596f32f09d757be62667d636ef67e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-04 09:34:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/906393/1/20240204143912/NEC-FC : SUCCESS in 3h 54m 16s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/906393/1/20240204143912/NEC-ISCSI : SUCCESS in 3h 55m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2fa629b3118a8358cf3f67d4bd04323e34cb1682","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-02-05 16:41:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f57b0481c9ca83cda20d0d5114be0cea1977d89a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-02-05 16:42:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2fb89e7c850778f7960e87a0a2e59e00079dfd9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-05 16:54:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c105ab3bf658eedb7a59891fa4eef6627dce5ce","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-16 15:00:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f6e85e0626364f06c73bf17d834fcb7a5d67cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-02-16 20:05:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d047c323660cda6b9e346076687b3626e4bf7f06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 20:05:13.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16525/","accounts_in_message":[],"_revision_number":2},{"id":"f514a6480d59febfb9c1ba21ce61201cdb04e9de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 20:05:14.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16525/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1595b590269bc9f5f74b5e4cff233a93f4afe041","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2024-02-16 20:05:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24083a91644017fe79bbd4f3d5f0f0fc10f31651","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 20:18:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dadc695b27f4609bf87512e13178862\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ef36400322b4c1eb9128a50cc9045d2 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1903d828972248e5872df25c04076a2e : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be0774e10d6f43208eec6d22c66abeec : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d79b5611000a831acb0537c55bbc3e95c44e5ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-16 20:29:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b2213f066687417ebeaa29e7d4be3dda : RETRY_LIMIT in 6m 22s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1766523897c14746a13aa236a3a1869d : RETRY_LIMIT in 8m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4a1c5efeb8a290d3ff503f7cf7500a979a99f9ad","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 22:07:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7365d28bb971cd375b8ffc1f2dde0ff0fba0b439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 22:17:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4fec4a1ef5417d8f7e19fc7727eb92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58be1e0a1d4f43a9b8bc77c4e6313b6b : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f64ab6468d4447d39b3417f6fe005b8d : SUCCESS in 11m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbec5c5424534c069b505fb3b796e5b4 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ec5ac0f1b248449c14019ba3fea15a : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a164e66c8d86467483aa177650301380 : SUCCESS in 56m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/597cbfdd89944e569070650f67d164e9 : SUCCESS in 59m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b60b54085904eeba6b1605254df2fd1 : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/574ef57870284a7a9a431410b05a701b : SUCCESS in 1h 03m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b7b3602a7204ca6bf6e4dcd2bf24e23 : SUCCESS in 20m 23s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9225f555f90649c780e3c8606c84729a : SUCCESS in 13m 58s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c388f7a255a44499284b773bb945da6 : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4ab2ff865fc4b15847e259e9c7ae36e : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ab2f7f8f4e247bc8da0569eddd02f69 : SUCCESS in 5m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/747360fb06ff44238e35f7b52546f241 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89c37f42f71347008c8a02ef85609631 : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97b517bc492f4b0481b0fd32432bedaf : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e30a48c7034f4f7ab190272c1cb89ca5 : SUCCESS in 1h 46m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbfd168f246546c69e9a610f6406e8c3 : SUCCESS in 2h 05m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a06cbe23b8ca49c989f5fe6e027d386c : SUCCESS in 1h 33m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1a91f1f4720c4d0eaa75a83f01e2ed0a : SUCCESS in 2h 07m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5d4e037cd69940019839ec29268382e2 : SUCCESS in 41m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea19ab3634734ece8a9ca004d487a5e2 : SUCCESS in 1h 21m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1eca656c1e6f4d5184e525bb704d0cec : SUCCESS in 1h 17m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7536eb407c17491598835e99c2584d7e : SUCCESS in 38m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/28a4b72f1d23475d8c0f49b73adee7dc : SUCCESS in 1h 22m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c16ab15824c34e378e17b5eaa5a75cca : POST_FAILURE in 1h 56m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b9080366dc446c88c9d16c129f3d585 : SUCCESS in 1h 22m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51e37b08748c408094a7add0dd2597be : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":2},{"id":"aa8c78394f4960754355623b9e1e13580afec19b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-17 00:14:39.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-906393-2/FC/ : SUCCESS in 2h 40m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-906393-2/iSCSI/ : SUCCESS in 2h 3m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"64091cdc748743505c1b65d47cb894476e176284","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-17 00:20:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/2/upstream-check/cinder-tempest-plugin-ontap-fcp/f78007c/ : SUCCESS in 1h 15m 31s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/b1df2ea/ : SUCCESS in 1h 29m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/2/upstream-check/cinder-tempest-plugin-ontap-nfs/69151b7/ : SUCCESS in 43m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/906393/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c6906e2/ : SUCCESS in 1h 45m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"725d0ce718e9c7f8a29ef890934329b8266ffc36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-17 02:21:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/729/.","accounts_in_message":[],"_revision_number":2},{"id":"a25bbd70d0fae29fe4f535eed786f63c3be321bc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-17 02:33:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_PowerStore_iSCSI/0df6c9a/DellEMC_PowerStore_iSCSI/1089 : SUCCESS in 1h 09m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_PowerStore_FC/1429472/DellEMC_PowerStore_FC/1244 : SUCCESS in 1h 37m 42s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_PowerStore_NFS/3091170/DellEMC_PowerStore_NFS/775 : SUCCESS in 1h 05m 04s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_PowerStore_NVMeTCP/486cd28/DellEMC_PowerStore_NVMeTCP/644 : SUCCESS in 1h 37m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"797ca619bd361af69e2cc673e87d05ee762c739c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-17 04:20:08.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b137b3c5c0704a19af99f597b433e0db\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c7b5cbb9df5a4762bb8c1aeaa12554cf : FAILURE in 1h 28m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c0020344fbbdca504b58d1704d493d8ec8407850","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-17 05:18:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_Unity_FC/386786f/DellEMC_Unity_FC/1342 : SUCCESS in 1h 12m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_Unity_iSCSI/f8ebd53/DellEMC_Unity_iSCSI/1372 : SUCCESS in 1h 44m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cb9faa662c62e0f0085dc880b6d0cf8e3d068dea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-17 07:16:21.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/724/.","accounts_in_message":[],"_revision_number":2},{"id":"63401cff594dfe1779d1d73e13870c50d60d79ea","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-17 07:20:36.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/93/906393/2/check/fujitsu-eternus-dx-iscsi/18657f6 : SUCCESS in 2h 37m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/93/906393/2/check/fujitsu-eternus-dx-fc/4664a17 : SUCCESS in 2h 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3f2ba73bf2c96aa7a29af77d7d33f2d5f924bd09","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-17 07:23:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/93/906393/2/ : FAILURE in 1h 43m 18s","accounts_in_message":[],"_revision_number":2},{"id":"9939404843c86d90768a46a0895ed2a3cf45b9ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-17 07:24:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/906393 : FAILURE in 22m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f97eadfa2c5ff6e16a6aa575ee5411b75cc16643","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-17 07:59:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/906393 : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e7808fbde0c7aa189173964d5872e34b6488be8d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-17 08:52:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_PowerFlex_3.6/e14d1b1/DellEMC_PowerFlex_3.6/1475 : SUCCESS in 1h 51m 37s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/93/906393/2/check/DellEMC_PowerFlex_4.5/a9e4d5f/DellEMC_PowerFlex_4.5/441 : SUCCESS in 2h 06m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"601494296fe97a6c6c57ac98e9f9df9289bcaacb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-20 01:16:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/906393/2/20240220065730/NEC-V-ISCSI : SUCCESS in 2h 56m 34s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/906393/2/20240220065730/NEC-V-FC : SUCCESS in 3h 19m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c22203dab62db20967fb64b760d3b13efa1cbe09","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-20 21:00:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/906393/2/20240221014113/NEC-ISCSI : SUCCESS in 4h 08m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/906393/2/20240221014113/NEC-FC : SUCCESS in 4h 19m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f7a3706d4ed56754fb0b9b4cde5ade8596dbf8a6","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-22 17:06:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/906393-2/HPMSA-iSCSI/ : SUCCESS in 1:55:01\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/906393-2/HPMSA-FC/ : FAILURE in 1:47:12","accounts_in_message":[],"_revision_number":2},{"id":"9f4cac730d816cc0de6c5ec08c02226c1050d165","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-25 20:36:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/87cb5784f1e54f4db2c6c61cedeb6f1b : SUCCESS in 1h 11m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ab336db829c34a249ab2b5dad70ccb56 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6fd335dc124397cefb73b55e67082bef555cba22","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-29 06:17:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/906393-2/PVME-iSCSI/ : FAILURE in 37:48.40\n- PVME-FC https://os-logs.seagate.com/dell-ci/906393-2/PVME-FC/ : SUCCESS in 2:10:56","accounts_in_message":[],"_revision_number":2},{"id":"a65c6c7e0b8575d909dec7b23fd458d1bf49079a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-29 13:44:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/906393-2/Lenovo-iSCSI/ : FAILURE in 37:20.90\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/906393-2/Lenovo-FC/ : SUCCESS in 2:07:52","accounts_in_message":[],"_revision_number":2},{"id":"cfa58d356130316fd77a12f62987688db13851e6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-15 14:45:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24574e91b9a247d57ea8cab915828fcfb25d49d8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-03-15 15:07:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7943681c97f8d3419498414e711e651912bb8be6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-15 15:42:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a60a2962e8f51d6fc563da1d0938cfe577388f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:43:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cffeb7406a75802c355f43f1d015b05a14f43864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 17:36:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65c43f296f341bc889cb0c0171de773\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817ae0ae8a494c95a88b97c5d801eb1b : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55a6d2a0228f412999b9d86861d5caae : SUCCESS in 10m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc076642f2154433ab75cff618788b25 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c5bffb003c4920b93b4afdd6e06d01 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/6b31f871f9b540ad91da4ba7be43291d : SUCCESS in 52m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0d97b8359a77452ea29be2c9fb5f80e8 : SUCCESS in 41m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bee0abe6f25246e7ae7d9e14ffa8968c : SUCCESS in 1h 24m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10b07b3a8f6a484294139e5834a646fd : SUCCESS in 46m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b7194098600a4fbe8b33a4dc8fe38e84 : SUCCESS in 1h 07m 14s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ef9320677d1e43f8b62ccfddea4f2139 : SUCCESS in 1h 48m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04d22e207ab14e1fa1bef0b71e291ba5 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":2},{"id":"88197d4de987ba0f189c0a29105a0bd603a61759","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 17:36:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4c01e3c0f839672ee21de3f46e57358cd7eabafa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 17:39:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3abb65796154ee19603c3d3307bcbd3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8b997d234547dbae4e615f72ad67b1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2},{"id":"bf0d6095f0118562cb62b3c53b644fd704a64528","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-15 23:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/906393/2/fc : SUCCESS in 0h 35m 20s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/906393/2/iscsi : SUCCESS in 0h 49m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfdf54ac7b61b335ea15ab84181adc037a220390","revisions":{"055f1c6718a48d16729f63e4b6892c1d0582ec3d":{"kind":"REWORK","_number":1,"created":"2024-01-23 18:06:07.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/93/906393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/906393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/906393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/906393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/906393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/906393/1"}}},"commit":{"parents":[{"commit":"b8cd101fe41066431b3ccf715780e9cc8b3c5d1f","subject":"Merge \"[docs] quotas apply to image-volume-cache owner\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8cd101fe41066431b3ccf715780e9cc8b3c5d1f"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-01-23 17:57:59.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-01-23 18:06:01.000000000","tz":-300},"subject":"Run protection tests during gate check","message":"Run protection tests during gate check\n\nThis patch adds the cinder-tempest-plugin-protection-functional test to\nthe gate check.  The test itself is defined in cinder-tempest-plugin.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/906153\nChange-Id: I590eb954203ccd2fc564117addfcf5d2dac5bd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/055f1c6718a48d16729f63e4b6892c1d0582ec3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/055f1c6718a48d16729f63e4b6892c1d0582ec3d"}]},"branch":"refs/heads/master"},"dfdf54ac7b61b335ea15ab84181adc037a220390":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-02-16 20:05:04.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/93/906393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/906393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/906393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/906393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/906393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/906393/2"}}},"commit":{"parents":[{"commit":"0d82e610ab1bfdf3d071b6384b1163a84b11cd80","subject":"Merge \"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d82e610ab1bfdf3d071b6384b1163a84b11cd80"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-01-23 17:57:59.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2024-02-16 19:57:53.000000000","tz":-300},"subject":"Run protection tests during gate check","message":"Run protection tests during gate check\n\nThis patch adds the cinder-tempest-plugin-protection-functional test to\nthe gate check.  The test itself is defined in cinder-tempest-plugin.\n\nThis test enables Consistent and Secure Default RBAC (SRBAC) [1] on both\nKeystone and Cinder to ensure that integration with Keystone and Cinder\nfunctionality is working correctly when both services are enforcing\nscope in RBAC.\n\nCurrently no other tests in the Cinder gate have this option enabled.\n\nMoving forward, this job will help ensure that new changes to Cinder\nwill work in environments where SRBAC is enabled.  This job will also\nhelp as we continue to implement the next phases of the SRBAC community\ngoal.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html\n\nDepends-On: https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/906153\nChange-Id: I590eb954203ccd2fc564117addfcf5d2dac5bd74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfdf54ac7b61b335ea15ab84181adc037a220390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfdf54ac7b61b335ea15ab84181adc037a220390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
