)]}'
{"id":"openstack%2Fcinder~907139","triplet_id":"openstack%2Fcinder~master~I6ae8aba231c35b86790fb63db97ed6c103bd5996","project":"openstack/cinder","branch":"master","topic":"bug/2052409","attention_set":{},"removed_from_attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2024-02-06 08:11:44.000000000","reason":"Change was marked work in progress"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-06 02:01:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6ae8aba231c35b86790fb63db97ed6c103bd5996","subject":"Followup: HPE XP: Fix copyright in test scripts","status":"ABANDONED","created":"2024-02-05 06:38:54.000000000","updated":"2024-02-21 07:13:14.000000000","total_comment_count":3,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"dcd0a9f39d8db5521ff835065499ca2797496d20","_number":907139,"virtual_id_number":907139,"owner":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-06 02:20:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2024-02-06 02:02:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{"CC":[{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}]},"reviewer_updates":[{"updated":"2024-02-05 06:39:05.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-05 06:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 07:24:37.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-05 08:01:16.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-05 08:08:02.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-05 08:08:13.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-05 08:11:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-02-05 09:04:13.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-05 10:32:21.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-05 11:26:54.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-05 18:15:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-05 22:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-06 01:08:31.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-11 18:30:04.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-02-12 06:44:24.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-15 14:55:10.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2024-02-16 13:46:32.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"b6477c672e8b50cd7053a5474d8591eff4f81411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-05 06:38:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae7c052becba19b266fc1bcf4489e1da1cc122ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 06:39:05.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16477/","accounts_in_message":[],"_revision_number":1},{"id":"4cdebed7a7666b907bfc85044464cad63c36acdc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 06:39:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16477/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ccf596474a8b9db9820fda1548fa4bca0cebd353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 06:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1a59f2f63cb47cbb67d2103d34a99ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b70d91996045609922f465dcd18da8 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13d3641e70624addb0e7df609affe8bd : SUCCESS in 8m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbfa5337e9df47edbf8733970304f578 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ea1e4038a543c3ac123b403bdd435b : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e18e5a77cde649d4b66b8004c053d3ae : SUCCESS in 6m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1075b9566e1f4788a30569f37c9d4242 : SUCCESS in 12m 35s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/33aad0df06de4877a8eb055ea19b49fe : SUCCESS in 9m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b686deaaa3aa41a9ac4d2b3801473a94 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"08fed20a9e15633ceb55cc12d052b8206a62b330","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 06:57:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21d92b4f6ce4dceae6149fe1fb092e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92026e38e8c14c7189bc57ba1fa782e9 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f41a563025764335b326bf40e5866d29 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61d3bdaa525c47958f84f0a993a09e30 : SUCCESS in 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fa19acdd236f3a9a13f1f1590000b5518486c08","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-05 07:24:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907139 : FAILURE in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c6dfbc18751759708cde93c93665ca34e6d8106f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-05 08:01:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/599/.","accounts_in_message":[],"_revision_number":1},{"id":"9e8e4c775263dfe8bf302e4e5d21ce1ead0298a2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-05 08:08:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/1/upstream-check/cinder-tempest-plugin-ontap-fcp/1bf17c7/ : SUCCESS in 1h 13m 47s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/21f4cc6/ : SUCCESS in 45m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/1/upstream-check/cinder-tempest-plugin-ontap-nfs/1b35eba/ : RETRY_LIMIT in 8m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/01a0888/ : RETRY_LIMIT in 7m 54s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"18cb42816c7b98fea30c9b4ad433acb7cc9eb622","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-05 08:08:13.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e2338303f17844669ecebc43cf3eb37c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/72fb2f8bf9b549ee92fa3f96beb31ca4 : FAILURE in 1h 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"598f035b42d6be48f5e586135a4cbba245b80c05","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 08:11:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4722b1022af1df0514ffd78987b5e4800dcb4dcb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-05 08:31:03.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/610/.","accounts_in_message":[],"_revision_number":1},{"id":"71ee68def7c40a5dc676a1ba39c4c8d9ffe9e7fd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-05 09:04:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_Unity_FC/47160bd/DellEMC_Unity_FC/1245 : SUCCESS in 1h 14m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_Unity_iSCSI/3773c9c/DellEMC_Unity_iSCSI/1275 : FAILURE in 10m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"75bebc60d627a893f8417130de418f421c876c54","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-05 10:32:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-39-907139-1/FC/ : SUCCESS in 3h 49m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-39-907139-1/iSCSI/ : SUCCESS in 3h 1m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"527489fa0d141d699e467b22ebaac0637f87f21a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-05 11:26:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_PowerStore_iSCSI/dca7428/DellEMC_PowerStore_iSCSI/1041 : SUCCESS in 1h 15m 13s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_PowerStore_FC/1733c1d/DellEMC_PowerStore_FC/1196 : SUCCESS in 1h 28m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_PowerStore_NFS/859b6e5/DellEMC_PowerStore_NFS/727 : SUCCESS in 1h 06m 51s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_PowerStore_NVMeTCP/6d5a97c/DellEMC_PowerStore_NVMeTCP/595 : FAILURE in 8m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd8c5a7b82008a907bcd2c5846e7ba4021a29e47","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-05 18:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/62c6c4d37e6a473a854508e0ded993f4 : SUCCESS in 1h 12m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fb510c6d3e7545f78b4a317dba83f805 : SUCCESS in 1h 08m 31s","accounts_in_message":[],"_revision_number":1},{"id":"98ad02dd9ee9a54aef62c7d50fe64d142b47b2a2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-05 22:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/39/907139/1/fc : SUCCESS in 0h 56m 53s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/39/907139/1/iscsi : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a4d2bd09094229187150dffcd32de22106c98175","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-06 01:08:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_PowerFlex_3.6/3550c93/DellEMC_PowerFlex_3.6/1384 : SUCCESS in 1h 40m 16s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/39/907139/1/check/DellEMC_PowerFlex_4.5/e41775f/DellEMC_PowerFlex_4.5/350 : SUCCESS in 2h 03m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"792244cec43321669693dad27fa8052e96698419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-06 01:55:58.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0a64205938b44008517026d5187bb28dc081631e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 02:01:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b0140bffe70ae252c540af402bf057b5a31cc7e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 02:02:02.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"52deb5834658afb406f3357e5c186954da04505e","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-06 02:05:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81a0bd132fa17a956de65a7d3bbc1c9383b25d70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 02:17:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e29732fb219a47189cb0c9024049273b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fce380d4a6ab45a0ac3c4a98e5e62f81 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35f6d6bf99314b53afcf9ee0569f0e68 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/671c65b52a8c4d8fbd9218f9b886a5d4 : SUCCESS in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d15eb105faf580a66a08aa747b9c77349364ce55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 02:20:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272c2eabaed24a73b09165777bad6dbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf37dfbfc124942b4f8733aa24a4df4 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b81e6589abe448dbfdc542fc85b842f : SUCCESS in 10m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08f95e05a03949b091047a6cca02e876 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ebdfcf755b4ea2b094e1f54b2ab6d6 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c61dd328bdce431e904ab6bafc182e59 : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8dae89a5a8a542159f6007073f2cbc91 : SUCCESS in 14m 32s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/61202da38fc345c99798d756876e6a93 : SUCCESS in 15m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7cee7c3aa45473fbe89c41f6981b9aa : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0e552d974debed94665fbbad8f5ac42c74b0ad14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-06 02:41:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907139 : FAILURE in 42m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f3153dde898352a1c4e0dac60b7274175a0aafc2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-06 03:25:41.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d16b7418192b4f2e8782c3170035dbf2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dd3abaf926814e7497b47604aecc851d : FAILURE in 1h 24m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7b7056c29ca47f6ad30458b3151d8bf59b1549cf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-06 03:32:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_Unity_FC/77af2b9/DellEMC_Unity_FC/1248 : SUCCESS in 1h 12m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_Unity_iSCSI/8a465f2/DellEMC_Unity_iSCSI/1280 : SUCCESS in 1h 35m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"75cfe7d52a386e5bc2c515a042b02b724ae7aa45","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-06 03:42:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/603/.","accounts_in_message":[],"_revision_number":2},{"id":"a25605e4339c1aa7c7d9c45cc50376b359ce919a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-06 03:53:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/614/.","accounts_in_message":[],"_revision_number":2},{"id":"b2fda577a72d8c05db8cf2faa58a813fc8c55252","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-06 03:56:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/2/upstream-check/cinder-tempest-plugin-ontap-fcp/5da2702/ : SUCCESS in 1h 02m 57s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/eaef51c/ : SUCCESS in 1h 18m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/2/upstream-check/cinder-tempest-plugin-ontap-nfs/a1d3104/ : SUCCESS in 44m 43s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/907139/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/596a2e7/ : RETRY_LIMIT in 7m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"19e10c3e0ee790c6a289f67e1535edacf33ca24f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-06 04:04:37.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_PowerStore_iSCSI/959679d/DellEMC_PowerStore_iSCSI/1044 : SUCCESS in 1h 03m 36s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_PowerStore_FC/141e706/DellEMC_PowerStore_FC/1199 : SUCCESS in 1h 27m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_PowerStore_NFS/22dc56f/DellEMC_PowerStore_NFS/730 : SUCCESS in 56m 10s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_PowerStore_NVMeTCP/08be752/DellEMC_PowerStore_NVMeTCP/599 : SUCCESS in 1h 43m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"941d5506a2abad0e8d26ffefff67e7589d9cd5a0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-06 04:26:15.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-39-907139-2/FC/ : SUCCESS in 2h 29m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-39-907139-2/iSCSI/ : SUCCESS in 2h 29m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"2fc34d1594e7289fcd9de0c801ca08f439e516ba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-06 05:10:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3dfb4e57d9874a5db1e339e80d2e0d18 : SUCCESS in 1h 18m 03s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d1301d8a222e4ef6992be2684db9e7bf : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fd866d5aef1888fef5005a2dac06b4d236cc2c2c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-06 08:11:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"48db289aeae930bf53d6f06425c01ec3cb15341d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-06 11:58:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_PowerFlex_3.6/384881e/DellEMC_PowerFlex_3.6/1387 : SUCCESS in 1h 31m 01s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/39/907139/2/check/DellEMC_PowerFlex_4.5/bab8e84/DellEMC_PowerFlex_4.5/353 : SUCCESS in 2h 12m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"3827bf61f6ddfc9251254cec52ea3cc73dd2b3a7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-06 16:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/39/907139/2/fc : SUCCESS in 1h 09m 40s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/39/907139/2/iscsi : SUCCESS in 0h 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3e2792cbe04eec1547b1b176a404fa382059df45","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-11 18:30:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907139-1/HPMSA-iSCSI/ : SUCCESS in 2:05:52\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907139-1/HPMSA-FC/ : SUCCESS in 1:37:22","accounts_in_message":[],"_revision_number":1},{"id":"4ca6acd6f2dd49214a863aa37681f7342cd028f2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-12 06:44:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/39/907139/2/20240212091618/NEC-V-FC : SUCCESS in 3h 45m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/39/907139/2/20240212091618/NEC-V-ISCSI : SUCCESS in 6h 28m 02s","accounts_in_message":[],"_revision_number":2},{"id":"82d1536c3b6b44482cc97a24f25b8d7d0da571bc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-12 07:55:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907139-2/HPMSA-iSCSI/ : SUCCESS in 2:05:49\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907139-2/HPMSA-FC/ : SUCCESS in 1:36:25","accounts_in_message":[],"_revision_number":2},{"id":"111cca990677586eef35be3f87101549bc9214b3","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-15 14:55:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907139-1/PVME-iSCSI/ : SUCCESS in 2:05:38\n- PVME-FC https://os-logs.seagate.com/dell-ci/907139-1/PVME-FC/ : SUCCESS in 2:05:23","accounts_in_message":[],"_revision_number":1},{"id":"203a485c2f78c955829fb330115fa34e678b4451","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-16 08:55:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907139-2/PVME-iSCSI/ : SUCCESS in 2:07:22\n- PVME-FC https://os-logs.seagate.com/dell-ci/907139-2/PVME-FC/ : SUCCESS in 2:02:19","accounts_in_message":[],"_revision_number":2},{"id":"8e476363669fc85341187cefc94c871a32a3f6e5","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-16 13:46:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907139-1/Lenovo-iSCSI/ : SUCCESS in 2:05:49\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907139-1/Lenovo-FC/ : SUCCESS in 2:09:30","accounts_in_message":[],"_revision_number":1},{"id":"69364bdc69bde5a85eea69ae99da6b8c8384a95f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-17 08:33:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907139-2/Lenovo-iSCSI/ : SUCCESS in 2:04:00\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907139-2/Lenovo-FC/ : SUCCESS in 2:05:23","accounts_in_message":[],"_revision_number":2},{"id":"dcd0a9f39d8db5521ff835065499ca2797496d20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2024-02-21 07:13:14.000000000","message":"Abandoned\n\nThe code in this patch have been implied in the active patch \"https://review.opendev.org/c/openstack/cinder/+/892608\".","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb5fe7adb81ceeabdc48313cd2804e1a71fe4ace","revisions":{"743248a073c01f486904dbd72d8db2c87e3b1527":{"kind":"REWORK","_number":1,"created":"2024-02-05 06:38:54.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/39/907139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/39/907139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/39/907139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/39/907139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/39/907139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/39/907139/1"}}},"commit":{"parents":[{"commit":"cc269291808340f913941e2bbb57c6ed3b38aa76","subject":"Hitachi: Volume Replication for VSP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc269291808340f913941e2bbb57c6ed3b38aa76"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-05 06:16:22.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-05 06:16:22.000000000","tz":540},"subject":"HPE XP: Fix copyright in test scripts","message":"HPE XP: Fix copyright in test scripts\n\nthis patch fixes copyright descriptions in the test scripts\nfor HPE XP driver.\n\nCloses-Bug: #2052409\nChange-Id: I6ae8aba231c35b86790fb63db97ed6c103bd5996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/743248a073c01f486904dbd72d8db2c87e3b1527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/743248a073c01f486904dbd72d8db2c87e3b1527"}]},"branch":"refs/heads/master"},"bb5fe7adb81ceeabdc48313cd2804e1a71fe4ace":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-06 01:55:58.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/39/907139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/39/907139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/39/907139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/39/907139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/39/907139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/39/907139/2"}}},"commit":{"parents":[{"commit":"cc269291808340f913941e2bbb57c6ed3b38aa76","subject":"Hitachi: Volume Replication for VSP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc269291808340f913941e2bbb57c6ed3b38aa76"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-05 06:16:22.000000000","tz":540},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2024-02-06 01:55:58.000000000","tz":0},"subject":"Followup: HPE XP: Fix copyright in test scripts","message":"Followup: HPE XP: Fix copyright in test scripts\n\nthis patch is a follow-up for depends-on patch, and fixes\ncopyright descriptions in the test scripts for HPE XP driver.\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/892608\nChange-Id: I6ae8aba231c35b86790fb63db97ed6c103bd5996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb5fe7adb81ceeabdc48313cd2804e1a71fe4ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb5fe7adb81ceeabdc48313cd2804e1a71fe4ace"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
