)]}'
{"id":"openstack%2Fcinder~907258","triplet_id":"openstack%2Fcinder~stable%2Fyoga~I5e7fb425c92cdf8c16d5a86a58ca1a52421543d7","project":"openstack/cinder","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2024-02-06 10:11:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5e7fb425c92cdf8c16d5a86a58ca1a52421543d7","subject":"3PAR: Error out if vol cannot be converted to base","status":"ABANDONED","created":"2024-01-31 00:54:36.000000000","updated":"2024-02-06 10:11:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"33a815fb5004b8a510c689d7cd865f85f34712d4","_number":907258,"virtual_id_number":907258,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-31 02:45:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 00:54:36.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-01-31 01:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-31 02:19:02.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-01-31 02:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-31 03:34:49.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-04 23:24:07.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-06 10:11:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"24510f52dfeb621f66177355963712627b04a065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2024-01-31 00:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1dd3fc8a8b46dccc66a6c8124af63995c948fb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 01:05:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cabdd1038324507a441b4db6cab65da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/167eb2b27e944f4cb6de952697a670b7 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81c2c267fb754d33aafbc8041fadd747 : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c95c88533dfc4877fd68df22d06099c10585d1db","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-31 02:19:02.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/f61f5f6d86874f7bb4e95dcef4a23f58\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c207a7f163f641ac9f687457a909b052 : FAILURE in 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"824710cf0824adf06030ae51562ba52990b596ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 02:45:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8e9fba6e2b4c3cb91e19d005fff9a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f865eb9211c941199cececc099dc9335 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d6683192aa4daba39d8900fd7ee57f : SUCCESS in 14m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d77bc85710d84b258c67bf9f25677a57 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db72364b04d84bf481b8f9b4885facc3 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/49279a6fecb1463496a7b5b6e00e7625 : SUCCESS in 56m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aacad3b4ba304d1b967bc8441f11061f : SUCCESS in 1h 19m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71061efecb2c45c2b1d0f7d6ebac7947 : SUCCESS in 39m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f049c857742443496fe92efa4dffe7b : SUCCESS in 8m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/03315caaa1e145b3b12ae7c1c064c8ea : SUCCESS in 12m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d353e93a17144aeb1ccbfc34373f466 : FAILURE in 5m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f073cd3742d24ac192e5b412812552bd : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c04dfb51ebb47489a7c8fe3f03820c3 : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a048c150c1834cd3ad39c1db9502ca60 : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/06ffe06dee3b457798784b215fe05281 : SUCCESS in 49m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/faf54400bdb14e91aca5845663ed99e8 : FAILURE in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c451f11b6fa949ef85a421becbe52b1c : SUCCESS in 1h 19m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/448d8f38cb7947c8885e725de2fad3c7 : SUCCESS in 1h 20m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a524bba66aa44148dadc710ee11a8f4 : SUCCESS in 1h 49m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/653c8124c43e48d48f81f0a902a226b1 : SUCCESS in 1h 08m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/73b96ede4df4490987c75d4020db660e : SUCCESS in 1h 12m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/000472c1da38418298abedb99c737962 : SUCCESS in 46m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/908a091936d0423595ce140f284d4caf : SUCCESS in 1h 16m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/30620b5e0ea94f2b997cc64760697c66 : SUCCESS in 1h 12m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1034be3518348a290e916773a6c1eff : SUCCESS in 49m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4392ae25a7a245f36b9132e961d0544f4fc9fdd9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-31 03:34:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907258 : FAILURE in 43m 13s","accounts_in_message":[],"_revision_number":1},{"id":"908dab04a611d6334e3a550d9d41e61325bd3e84","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-04 23:24:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/58/907258/1/20240205050115/NEC-V-ISCSI : SUCCESS in 3h 06m 32s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/58/907258/1/20240205050115/NEC-V-FC : SUCCESS in 3h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"33a815fb5004b8a510c689d7cd865f85f34712d4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-06 10:11:22.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97d5541b8982923ffa994ef308fe608f2e373f7b","revisions":{"97d5541b8982923ffa994ef308fe608f2e373f7b":{"kind":"REWORK","_number":1,"created":"2024-01-31 00:54:36.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/58/907258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/907258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/907258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/907258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/907258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/907258/1"}}},"commit":{"parents":[{"commit":"063609c84fb65975daf1809eac1ace075c0ffc09","subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/063609c84fb65975daf1809eac1ace075c0ffc09"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-10-26 09:55:43.000000000","tz":0},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2024-01-31 00:53:53.000000000","tz":540},"subject":"3PAR: Error out if vol cannot be converted to base","message":"3PAR: Error out if vol cannot be converted to base\n\nConsider volume and snapshots as below:\n\nv1\n|\n`-- s1\n    |\n    `-- v2\n        |\n        `-- s2\n\nUser initiated deletion of snapshot s1.\nIt failed with some vague message.\n\nInitially, it was suspected that ...\nWhile copying volume v2 (sometimes an intermediate step to break\nvolume dependency), we send a request to clone the volume v2 to new\nbase volume; and the exception [1] isn\u0027t handled properly.\n\n[1] Conflict (HTTP 409) 32 - volume has a child\n\nHowever, on further investigation it was found that ...\nafter a new volume v2 (omv-\u003cid\u003e) is created and\nwhen we try to delete old volume v2 (osv-\u003cid\u003e),\nat this point the exception [1] is thrown as error.\n\nThis is now handled gracefully. Appropriate error is thrown\nif the volume (v2) has snapshot (s2).\n\nCo-Authored-By: raghavendrat \u003craghavendra-uddhav.tilay@hpe.com\u003e\nCloses-Bug: #1994521\nChange-Id: I5e7fb425c92cdf8c16d5a86a58ca1a52421543d7\n(cherry picked from commit dfd8f99743a29220ca3face5fdf00a1a6071cf48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97d5541b8982923ffa994ef308fe608f2e373f7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97d5541b8982923ffa994ef308fe608f2e373f7b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
