)]}'
{"id":"openstack%2Fcinder~907389","triplet_id":"openstack%2Fcinder~master~Ib089c5ddf00cac443a3ca8e16a86497e1b1139ba","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-20 13:01:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib089c5ddf00cac443a3ca8e16a86497e1b1139ba","subject":"Remove six from Synology driver","status":"MERGED","created":"2024-02-01 07:27:54.000000000","updated":"2024-05-22 05:42:04.000000000","submitted":"2024-05-20 13:01:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907389-remove-six","meta_rev_id":"d617ac64c8e1e880463096f29c32e16c3c628fb6","_number":907389,"virtual_id_number":907389,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 13:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-04 03:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-05-17 16:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-05-17 16:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 07:28:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 07:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 08:30:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-01 09:51:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 10:28:20.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-01 14:18:59.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-01 14:31:27.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 14:59:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 16:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-01 23:08:40.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-02 00:56:23.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-02 01:03:58.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-02 22:32:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-03 14:35:11.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-02-04 03:08:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-02-04 03:08:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-04 08:26:48.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-02-04 17:45:21.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2024-02-10 16:08:58.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-14 01:26:42.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-17 16:28:00.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"fd5f5c026ba53ed3704244dd988eaea2c0f15fa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-01 07:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc831dad7866a0fbaced5b1e442ef94184f21ba7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 07:28:01.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16434/","accounts_in_message":[],"_revision_number":1},{"id":"a49458db48226ca5f3bee94e3606442291743bd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 07:28:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16434/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"44bc70c097757332d639dff26f978a22a0c3b655","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 07:45:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0235df855c56494eb3dfae612ae5ebe4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a48b18c54958484b9f3edd8429006bef : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e605dedf98f449f8f8abbc729eed82e : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7e6d2a570f5d4e4c8cf8c01693f4b846 : SUCCESS in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36691d0bd8f9d4a0d30ae20da7436057bad695aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-01 08:30:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_Unity_FC/8255760/DellEMC_Unity_FC/1174 : ABORTED in 6s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_Unity_iSCSI/006bbc5/DellEMC_Unity_iSCSI/1190 : ABORTED in 2m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c32827af95782ad5ecf0718fa56c8ef1ebe6731d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 09:51:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b874575983e24f3a956ed59282121607\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29f549f91f14180b5d73167c617f1e6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b3d0fb215ae4bb9baeae563943dba36 : SUCCESS in 10m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f10c1f4e78954ad1b353916fa5050cb0 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57920dcfa6f94fed9e7a423f423709d4 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/011ccb406e4d4acbad48787ab3cdbb66 : SUCCESS in 1h 07m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/410247a573674e17bc7c66608721ad51 : SUCCESS in 1h 05m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea46d9480d29493593e9aa1203115eda : SUCCESS in 58m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b48881cff32f48218141a5d358622c19 : SUCCESS in 21m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1a7c4b2667e468089ac2996c89475c0 : SUCCESS in 16m 35s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/23499caeea0d4f93b95821e1956b5abf : SUCCESS in 13m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3f2a5d1719474b289f048937d52fb551 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d50ef0ba2e194fa5b99d77b03bcc4045 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95d84b8d4f364b3281a29b125464104a : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5d0edc8314fd4e04b470ca0a68af7c3b : SUCCESS in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13c4c8982ac84b9690dae2a6ecbf15b0 : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecaf5a50dcde469cbc0f0847acd1808f : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb242daede33448e8e8a47629a61752a : SUCCESS in 2h 08m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5fb9074d84cd479eb00636339865b547 : FAILURE in 1h 51m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9a84ab180ce43678d5210a30cb4a7ff : SUCCESS in 1h 59m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/987968009001454dae07afcd6fe5bc25 : SUCCESS in 2h 16m 30s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a21c67f32fde460d8d6203a26623c90b : SUCCESS in 1h 09m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/549a3e3bcb18486997a8793a423d3bc7 : SUCCESS in 57m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1ed2ea24f2f74774982b57fdfb305fc3 : SUCCESS in 48m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/028ed2ad3657400f849c0b2715f32b0a : SUCCESS in 1h 20m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2dfdc7b9f1214f09b776ad95179d1e93 : SUCCESS in 1h 24m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f6c99adcb8940a4a3cddde2411189c6 : SUCCESS in 1h 25m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65b2d71210704c6e8ed51906e62260c4 : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ec8f35c060f24dd64c7ddff7bcfbfbd9c91871fe","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-01 10:28:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/1/upstream-check/cinder-tempest-plugin-ontap-fcp/8bcfdf4/ : RETRY_LIMIT in 10m 29s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/9bbc57f/ : RETRY_LIMIT in 9m 11s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/1/upstream-check/cinder-tempest-plugin-ontap-nfs/cf85e4e/ : RETRY_LIMIT in 9m 05s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e7274cf/ : RETRY_LIMIT in 7m 36s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea739ca2ad0116e6b7d5f227b65e680aa9e2b3e9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-01 14:18:59.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/95ccabeaa8ab499fbd78ec22d636aa92\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8657c3937af74cfaa51836e3665293ea : FAILURE in 1h 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"65987e5631af6b481ab3e588d30725a4c5a51700","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-01 14:31:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cd381b2ead6e458a8bcb081b2a37fe97 : FAILURE in 52m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/15866c3fb2db40069c9330d7b9bbf0c8 : FAILURE in 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2c74bac8e7858fd4eb50a7b6faa406417e9b47d6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-01 14:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907389 : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bed73e0d9337ac0198473107939064293b8ad709","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-01 15:13:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_Unity_FC/e2bac53/DellEMC_Unity_FC/1184 : SUCCESS in 1h 05m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_Unity_iSCSI/a43056b/DellEMC_Unity_iSCSI/1200 : FAILURE in 14m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"957768545badf182ca5e02087327ec4aaf2137a1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-01 16:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/907389/1/fc : SUCCESS in 0h 59m 35s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/907389/1/iscsi : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1e4d351cf7e99c5b4eaf85cd610d5e508a91e7b2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-01 23:08:40.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_PowerStore_iSCSI/80d42d4/DellEMC_PowerStore_iSCSI/1000 : SUCCESS in 1h 10m 43s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_PowerStore_FC/16042bc/DellEMC_PowerStore_FC/1156 : SUCCESS in 1h 27m 56s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_PowerStore_NFS/bdeec69/DellEMC_PowerStore_NFS/686 : SUCCESS in 1h 00m 00s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_PowerStore_NVMeTCP/386aaa8/DellEMC_PowerStore_NVMeTCP/553 : SUCCESS in 1h 48m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"4c587f1d966ee7c33b87934af11fdaf202842597","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-02 00:56:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_PowerFlex_3.6/24205f0/DellEMC_PowerFlex_3.6/1333 : SUCCESS in 1h 34m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/89/907389/1/check/DellEMC_PowerFlex_4.5/305b889/DellEMC_PowerFlex_4.5/301 : FAILURE in 2h 21m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"60ee36900bd1769c672528c6f2e1303829977b40","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-02 01:03:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/530/.","accounts_in_message":[],"_revision_number":1},{"id":"e2f7f5787d2cc6af37caab743449cc315f0b3332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-02 15:50:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"49d71016360e528c30845b166e5203ab8ae2b834","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-02 15:50:24.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16463/","accounts_in_message":[],"_revision_number":2},{"id":"8076087e5f280523c89c93aaca6061ed2df71d12","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-02 15:50:27.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16463/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"206e36dff989f22e86cbde04a4851eab1235429c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 16:12:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1bc3a7f90a64b35ba8dbc64c6d22a8a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec1fa17054d74118aa519c19eb174dc9 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08427a08340e4376aa0bb32f4efdedd8 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bed5219d872248779a969f8ad83abeb8 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca05ab75f122a4cac73c4fa6f4776f7d756781e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 18:33:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f8dbf0cd5f74d8487c03d8462041970\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215111226ce341d8bd7e4c1fef702d9c : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ab4265313049eb8d3518a217a1096c : SUCCESS in 10m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb7aee9f3d694018b0b196a13d431400 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6806a1de6843a1ad40894ed8d98d11 : SUCCESS in 13m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/01b7464de3ce420783d634bb0aab11fc : SUCCESS in 49m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1b83817bb8ee4bff86a9768a512f8550 : SUCCESS in 49m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e37b260cba284f638e9de08bfd81ffa5 : SUCCESS in 1h 23m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32737bec2b5a4cb09b456915c9cdc5ce : SUCCESS in 27m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/876ae05cd2984a719b530d148c4225d2 : FAILURE in 13m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/04fcd789e95748878aa2bc51e358cfaa : SUCCESS in 13m 25s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2a316d7ccde049ceb32bc1525ef0cf09 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab8ae648e4f74c7eb115b6e9b6070f52 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f50bccf5e5db42b28132632c699510a4 : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93a367f683c746918c99f95f6764ad88 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5932e932379647f387a9188c11a30f56 : SUCCESS in 34m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b4b3c09263345b39e8bb91de4eba917 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3392de7e7be0408784dfcf193c412b72 : FAILURE in 1h 54m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1260d349af8a46d98567692eac1ebb50 : FAILURE in 2h 35m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fddd9997404448497901f21ef9a6b30 : SUCCESS in 1h 38m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4a438725efd842b395fc919f974836aa : SUCCESS in 2h 08m 27s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d930a6dc267f4c2aa6eaa9b56398452b : SUCCESS in 1h 20m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c581fbdef8464a128cc33f860018288e : SUCCESS in 1h 33m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69ca4db34bd744b78c1577d309289c31 : SUCCESS in 47m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6ba54c2f4ed5455397ab3c3dcdb03dd1 : SUCCESS in 1h 02m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b290f55750c94841af7dc580292daf28 : SUCCESS in 1h 37m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/337d508dde5048b59f3dd93928331ebb : SUCCESS in 1h 20m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42de11598bb844728c7c0db474200ecd : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":2},{"id":"91c713144ce37a65eca21cd61a5979fb3f36698f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-02 22:32:29.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-907389-2/FC/ : SUCCESS in 2h 41m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-907389-2/iSCSI/ : SUCCESS in 4h 34m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"f1c75d10a524d5c0d8976194dacbe65b1a728a20","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-03 00:21:09.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8b51056138554ef394e3ff0042b36234\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9a7f9112b95e42cc90ddb9a574829dd5 : FAILURE in 1h 24m 44s","accounts_in_message":[],"_revision_number":2},{"id":"53565f49432b5a5c79b64f078a41ad084a0ed985","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-03 00:24:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907389 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":2},{"id":"623071b2225d6861852bd8ad6ecb4867f16db7be","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-03 11:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/907389/2/fc : SUCCESS in 0h 33m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/907389/2/iscsi : SUCCESS in 0h 44m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6aa964b5e97f0ebe1983506a0b652c3de7282ec7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-03 12:09:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_FC/af4942f/DellEMC_Unity_FC/1223 : SUCCESS in 1h 04m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_iSCSI/83c6dea/DellEMC_Unity_iSCSI/1245 : FAILURE in 15m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1caaaa925db9b55412c704232c7a6ee601a4c24b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-03 14:35:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907389-1/HPMSA-iSCSI/ : SUCCESS in 2:02:36\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907389-1/HPMSA-FC/ : SUCCESS in 1:36:00","accounts_in_message":[],"_revision_number":1},{"id":"57df825cbe40100115a234be6218ec136278be4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-03 16:09:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/585/.","accounts_in_message":[],"_revision_number":2},{"id":"d7ca10952df87bf0eb0c8804b90d068d532a1a29","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-04 00:36:51.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/596/.","accounts_in_message":[],"_revision_number":2},{"id":"9888e5f767aa72baaf57b2b2a6248c9086eecb3e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-04 03:08:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a3dd327b52fb53291d49636b9a6e60a170c0b90","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-04 03:08:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"586fc7c1df754e28b93a01d066eddc36ade795f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 03:33:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f107209720fc474987701c8dc1e99692\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2476ab1a6ab40e0aebe663528e85c6f : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4e609d504b44510ac515ce5110052a9 : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbefbf4d0fa641629ade0ae81de0737f : SUCCESS in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90c3c98aae7255c688e8892353c9be326ab1c042","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-04 04:35:18.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/25591a5d5c944e4fbdfc2a242b24510b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/26b9ec9fe334489e921666963d29cea1 : FAILURE in 1h 23m 30s","accounts_in_message":[],"_revision_number":2},{"id":"6b7767a16cfd348b926f97e89a559a94ea42dea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 05:39:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fea6833e8a5474099f1dc18fe73a0e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd42ed77d164b0ba70c5bd1a18b9188 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae74da6bc80645aabb44dfa985ee1772 : SUCCESS in 12m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d103e0be073242e59fc3afe3cf002a17 : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df51c7ab905d42819efe5282e6e22e47 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/ff6bf1d4dcdc409f89147203716f236c : SUCCESS in 1h 01m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1b946f821dba406b9b0f219abac1809d : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a938b01fca364021bf141d072baf05dc : SUCCESS in 1h 34m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4e09123e5c544959d4521ff1009e9aa : SUCCESS in 36m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17af10e502eb4a25930c5278fff30265 : SUCCESS in 16m 21s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b5ae3e9fcfeb4a82bb23b26202c26d11 : SUCCESS in 12m 47s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7d0b52c7ff747eeacc3df1bc80430b8 : SUCCESS in 6m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d23b945e2124ff7b5580528053c2135 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22f1d5fe1d12423ab3591cb4fd09c673 : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d618436da38d46b487b537515ff83809 : SUCCESS in 8m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89a14ae3e60b4a2e89267a932874da71 : SUCCESS in 36m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32d76eebc0014bc4b4562c7e3cea39e4 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14f47d1acf8d4d2cb657596ecd669d2f : SUCCESS in 1h 59m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c47f86d9a2ec4e1d91791080ec42a84c : SUCCESS in 2h 14m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7256f602585743ed9fb910865b4562a6 : SUCCESS in 1h 38m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4c505f1302c04dfcbce19f79cf3cbc85 : SUCCESS in 1h 50m 33s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e22ed0762414160aa31749bfe93c772 : SUCCESS in 1h 00m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a3c04d41b32f49dea56377b7ab967dd0 : SUCCESS in 1h 20m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/35ba510ab12b48d1ab53d91305d573ab : SUCCESS in 56m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7d7f9c51d6c3471782aefd844c7e3470 : SUCCESS in 1h 33m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5e863b858eac460884e732e3f7df23a2 : SUCCESS in 1h 33m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e3ba0d9dda04c3b80c748ebe2249fcd : SUCCESS in 1h 31m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bbcfb1060c84f72b24634ca18812c9d : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b13965651442ecdd39e8cca558026a73b502adcc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-04 06:45:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_FC/13ab4d2/DellEMC_Unity_FC/1231 : SUCCESS in 1h 05m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_iSCSI/39bd737/DellEMC_Unity_iSCSI/1256 : FAILURE in 15m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fcb179e355284a2c47e1e7924318478882958c1c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-04 07:12:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-fcp/81dcb37/ : SUCCESS in 1h 10m 41s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/64c35a1/ : SUCCESS in 43m 03s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-nfs/8eae1cb/ : SUCCESS in 1h 32m 46s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4b7588a/ : RETRY_LIMIT in 8m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"29868a64a166353910893fea6ea729d443ca84aa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-04 07:16:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ef2f64ff9aef484a855fe34844ba99ef : SUCCESS in 1h 16m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b5c9c4eb93fb4c62a68b2291bd551840 : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ee85230a41ffdd353b5c3548825dad7fc1f7eb96","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-04 08:26:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907389-1/Lenovo-iSCSI/ : SUCCESS in 2:05:42\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907389-1/Lenovo-FC/ : SUCCESS in 1:55:18","accounts_in_message":[],"_revision_number":1},{"id":"53875223bfdbf53bc11da11b1dacf46350e4fc21","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-04 17:45:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907389-1/PVME-iSCSI/ : SUCCESS in 2:06:09\n- PVME-FC https://os-logs.seagate.com/dell-ci/907389-1/PVME-FC/ : SUCCESS in 1:56:22","accounts_in_message":[],"_revision_number":1},{"id":"b8dae22a2149195764e6756dd2825bcd507df37c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-04 20:50:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerFlex_3.6/8fdceea/DellEMC_PowerFlex_3.6/1368 : SUCCESS in 1h 44m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerFlex_4.5/fd45b97/DellEMC_PowerFlex_4.5/336 : SUCCESS in 1h 57m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5243f2b5ff53f72996d21c0ad601bccbae95770e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-04 22:14:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_iSCSI/3a7106f/DellEMC_PowerStore_iSCSI/1032 : SUCCESS in 1h 06m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_FC/dd6421e/DellEMC_PowerStore_FC/1187 : SUCCESS in 1h 33m 04s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_NFS/cdd3ae5/DellEMC_PowerStore_NFS/718 : SUCCESS in 1h 08m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_NVMeTCP/8e75dc3/DellEMC_PowerStore_NVMeTCP/584 : ABORTED in 3h 01m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"e3fbb8d803e62506f90a9c499c4747c54b4c795d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-10 06:27:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907389-2/HPMSA-iSCSI/ : SUCCESS in 2:04:53\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907389-2/HPMSA-FC/ : SUCCESS in 1:36:43","accounts_in_message":[],"_revision_number":2},{"id":"7770a50f5dbf9b7b2ffe4468aaad5330917fe969","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-10 16:08:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/907389/2/20240210183607/NEC-V-FC : SUCCESS in 3h 48m 58s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/907389/2/20240210183607/NEC-V-ISCSI : SUCCESS in 6h 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c1a800fb1d1001f827ba446fcdb91ba1e7ec20da","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-13 20:24:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907389-2/PVME-iSCSI/ : SUCCESS in 2:03:20\n- PVME-FC https://os-logs.seagate.com/dell-ci/907389-2/PVME-FC/ : SUCCESS in 2:04:57","accounts_in_message":[],"_revision_number":2},{"id":"ca31e58f529cd6fdde237e4fd26dd5f42b4c5387","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-14 01:26:42.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/907389/2/20240214013656/NEC-FC : SUCCESS in 4h 05m 10s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/907389/2/20240214013656/NEC-ISCSI : FAILURE in 8h 49m 42s","accounts_in_message":[],"_revision_number":2},{"id":"93636dc044c62bbe0e2a10b5e7424383e685a9b8","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-14 14:59:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907389-2/Lenovo-iSCSI/ : SUCCESS in 2:03:55\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907389-2/Lenovo-FC/ : SUCCESS in 2:03:58","accounts_in_message":[],"_revision_number":2},{"id":"4b4b394e8f6c3f3f6d71dbd18e61f1633d82fce6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-17 16:28:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a9611cdcaafe16241fe968fd90a1d9d2636fcf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 16:29:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cde3a76d59e7c7c3afe2b912594fbcfb90eb3889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 20:35:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccc7ef0d5d084ebdb10d994aeba009af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c37570224c4ab4a08e04618ce530a9 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4fc0f77c3254bbe90ac9c5219467686 : SUCCESS in 9m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/326f2edad4914ff89af4e4bdcde9f2f8 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a835bc1d0104e4ba9678e04c8ef31d6 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/f5d7ced5502545c381830f7d701840b8 : SUCCESS in 47m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/711e6106063f40bb894f6fe1e5885131 : SUCCESS in 1h 22m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8583642581f74d13810289de585b4b61 : SUCCESS in 1h 02m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/20f4424d561e40e0a4c9f8c87cf483ca : SUCCESS in 57m 02s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea44cba6b5a54512b9d8af80af0a1661 : FAILURE in 2h 03m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3612a8c91e24423b9215f796e20e1a4 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e8f8db5777bc8d218acba50a350365482c3aa72f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-18 02:09:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c4768cedf3d97fe57528cd61e9af0e2f13eb654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 02:22:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dad6d613a54c4cd3bce6218a5d86e7a0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/384c4123409b42ecb0072439e075cc65 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75a6b912a6204332abedfd5c1747be65 : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f67a20e2bef8e02163c0cc8946316a7ad88276a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 02:31:41.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c94fe69c2b3d425fb16794892e1c1e03\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2ab489b8a75a4fff8d23c13080747bc6 : POST_FAILURE in 19m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f8ec3e48acc38b743fab1574839c9934ad069231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 04:27:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cdb423dec07408e8e306c5bd08397ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2459cd6828d43edb561d510eff6c04c : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d7c1fe9094c4a9a854fd2b50802826d : SUCCESS in 11m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/710dc59b4a3c41db8f708fb3e300268f : SUCCESS in 11m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a56fde7a75d4654b1e55ee1392bd6a5 : FAILURE in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61048e5074841ad81e29e742c8612eb : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/fd5417173fe54200b5866be86219728b : SUCCESS in 1h 13m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8696cf634be4b439f4ecf9a7a8c24b1 : SUCCESS in 1h 07m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96acb9f2183d48068353cd7c88de1163 : SUCCESS in 1h 11m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/06c10088708940279ded1ed1960b35ef : FAILURE in 14m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd916d2131ed4faa86d01986f3f263cd : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29e08af481b24aaaa7fac5a87a648423 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fdfb95996b594a00ade8381c21e89e05 : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/36a8c4a11dc1400099167d1cefdb1073 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/246eee0d19e14526b8fcd86ed44588e6 : FAILURE in 41m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e4a5b740fad424ba547346feb395030 : FAILURE in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4355b48c52944763830f100b740a746b : SUCCESS in 1h 54m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f7c3d8bd575c470aa059f0a41ebd2a2c : FAILURE in 1h 41m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30ff150e86304ddb8850e68c948bf78f : FAILURE in 1h 54m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a0f16f3562e647479213731ee784546b : SUCCESS in 1h 53m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5fa3bcad15484d62a44c568cb4b3e4ea : SUCCESS in 43m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4d3cf013bf5f4425912169c40b485e67 : SUCCESS in 1h 16m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/59636c9b8d334f729b54c6cc26bb145f : SUCCESS in 1h 36m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/45b7c17ebc834c11b2048a9ae026bd9c : SUCCESS in 52m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b4fb7223e7954e4cbdeb4af6c638ad12 : SUCCESS in 1h 26m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/892d9338c004405ab82b09caca7e01a7 : SUCCESS in 1h 41m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/440da764f194432fbc105f798a321131 : SUCCESS in 1h 36m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94d2fe9f5fb34c56bb1ba6047961c6be : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":2},{"id":"b50ac6a940f786f70044f7f20c7bec13c6365e64","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-18 05:09:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64ecaa4bff5f3f3a29d095a5ec860c1ac39dbeb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 05:26:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80754f975cc475aa6f58db361571a00\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd5b6fbab09f40efa29ecc4bd1d31b8a : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/587904e57a4c456ca99c4e8f85d486bd : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb7a816b734511378bb715b178e4a89240ee1a03","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 05:33:40.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5666ee79d4f34596959e3fd074c0d1b4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8a81f2a874e8491aaa27274cff0ec6b4 : POST_FAILURE in 21m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1469c3351847be9622484b42829679e92f869b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 07:34:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/401c753e0d9548649bf6e495e783791f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a334d103abf4c2db8b4ae79f3627d7a : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff68a1df9b8d4e9a80471cf471552234 : SUCCESS in 8m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58507557971f48c8adb4c7c39ea233dc : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70b47099ad884b139689d9f47fa30f6d : FAILURE in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a77b341c1644c49d0e91710de73ff3 : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/75b3094719674ba1a41096ab9bdb04be : SUCCESS in 48m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5066867040f44f59bd14679946881fc2 : SUCCESS in 1h 27m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3980e45f1be04fa48f6f078aa278ce5a : SUCCESS in 42m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6fb550c512a94c0da1c802c39057b155 : SUCCESS in 17m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/73d766826fde4f128955bcd2295e3830 : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09661caf363c43d58f8c324fccb1660d : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/81417964dd1c44bf894447a3ac463ec7 : SUCCESS in 6m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6045dac82a0140099c68d2521cd422d9 : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/954ec684d86d47a18481022b8b88f4a3 : FAILURE in 43m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a18b1dcd3514afd9f885c53876c6af4 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cef7f468a0ac4300ba39de2003aaaf5f : SUCCESS in 1h 55m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5df4581d654f4f21b2d8e39379810fe5 : SUCCESS in 2h 06m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/129291c2b16241028a29823f6ed40baa : SUCCESS in 1h 49m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0d08f46921064035b4de7bdcc1b35182 : SUCCESS in 1h 50m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/200e0fc2fa8940868a96c66598a0f498 : SUCCESS in 38m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91fce2ebe2f14a64a59e4d7de428f432 : SUCCESS in 1h 27m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6d16db54c554632846ea5fdd363edf6 : FAILURE in 1h 46m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/609d0b5429224c46a77680fef3f953b9 : SUCCESS in 45m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e34b22d7369a408388e07e52dcb6c86d : SUCCESS in 1h 29m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/87090f3a83954043a317b2e6a70d14af : SUCCESS in 1h 09m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7799ea6ab12c4000b3e272d2e8e8c82a : SUCCESS in 1h 12m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d26f112edb441d5a1e769ce61a3534c : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"48742136c31998c168f20eb8e0cb34a5295bd9d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 07:35:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec75cc6359fee35d1a85f9cd8e8ea902aedc7570","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-18 08:23:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-fcp/bcb2443/ : RETRY_LIMIT in 3m 53s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/19658d8/ : RETRY_LIMIT in 4m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-nfs/2bde6ff/ : RETRY_LIMIT in 3m 26s\n- cinder-tempest-plugin-ontap-nvme-tcp cinder-tempest-plugin-ontap-nvme-tcp : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f571770762994fe1fc483c8c52cc14ac41187eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 10:21:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d809c8d97b674fb391587243c0af93ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8670fe30fc664274888e97b7fce40539 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d19318ab88b4768b692b7ee05896444 : SUCCESS in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc76acd6f8c84b5bac91fbcf3fa51f41 : SUCCESS in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c0fcdf0e50473d9101687d332d50be : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/75a4265c52b54d10acb627290f663944 : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95948026d92248f49b2624a305da8361 : SUCCESS in 1h 07m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdbdb21a42ec44c7b6d445a1c8f6b0b4 : SUCCESS in 44m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab94a80262914212b1cf720d065f7ddf : POST_FAILURE in 1h 10m 31s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5f46bfa9e834632b67a7cd8f584a58a : SUCCESS in 1h 57m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b150ef7857b4b6a9f13d2dee3b28079 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":2},{"id":"14b5f6f2989001138c2585ca7742f084593a5e63","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-18 13:57:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_FC/dacfdfd/DellEMC_Unity_FC/1888 : FAILURE in 1h 38m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_iSCSI/7848213/DellEMC_Unity_iSCSI/1916 : FAILURE in 2h 08m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7ae5447ac374cbbad8b6e40c0dde33221e754e2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-18 14:34:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerFlex_3.6/238e602/DellEMC_PowerFlex_3.6/1387 : SUCCESS in 1h 45m 41s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerFlex_4.5/8ae8684/DellEMC_PowerFlex_4.5/306 : SUCCESS in 2h 14m 25s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerFlex_4.x_NVMeTCP/92bf9d5/DellEMC_PowerFlex_4.x_NVMeTCP/250 : FAILURE in 2h 55m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f974472aae560a645f4ab2c3dacf64eb4e0547c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-18 15:11:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_iSCSI/0608ca5/DellEMC_PowerStore_iSCSI/1330 : FAILURE in 2h 52m 53s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_FC/e029299/DellEMC_PowerStore_FC/1513 : SUCCESS in 59m 57s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_NFS/40e7ab6/DellEMC_PowerStore_NFS/1024 : FAILURE in 1h 10m 47s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_NVMeTCP/1301a86/DellEMC_PowerStore_NVMeTCP/878 : SUCCESS in 1h 55m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"6dbe7530cd063962ee9c78540a8d7fb820032c91","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-18 16:18:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"571993766c0003b078aed722e16298c1a240decf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 16:29:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60f4df1d82224e588154aa26345f2889\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b90d9cdc11d4e26baaffefa8d4c1d20 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7914386e137449bd8478d66529952ddb : SUCCESS in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"519e7b7f028fb20e3ce44d5e7cf22bdd16f66f02","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 17:02:41.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/577ceacf0e0c46bfb5e25bcbe9e661b3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/815e9206f3cf41bba5409131cd1a89a6 : POST_FAILURE in 21m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ce5e927596b21b5825d57a442108b41158b1d537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 18:43:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f61a2a8d62844f9864928cbde1129d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80952abf68854db3bc1d3b2d76d484f7 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/827df4820e0d45a18b701ff2f0d1ef54 : SUCCESS in 11m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/303995910abd49148937b715635ecc89 : SUCCESS in 8m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8994ecf16fc470a9ce31e0f14e67803 : FAILURE in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5037d6506e3f4bb6be073c3e5db85dea : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/42db9989dae34c95b496cc3fc556fc72 : SUCCESS in 49m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b22b7da3a87b48878183569dd583aa3c : SUCCESS in 1h 09m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96abda0c08f54cf0b61a7127220ec0b2 : SUCCESS in 46m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cdfa3f8e7ab74de0812b2efd62c040ca : SUCCESS in 12m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bfd05adf43524b72afbf9ac5cbc8be38 : SUCCESS in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc5d9df473674a59872114ec3b30ee88 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e141ee3caf1410198b0a529954393b2 : SUCCESS in 5m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ca8bc8c6d4044062b4bacb8ba84036a0 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3251dbe61b0e47de9bd94e19d6962ba2 : FAILURE in 32m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8baf4ba8c8d4fd8b6a9a050bcdd043f : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/014f8770545c45af92b13418fe892828 : SUCCESS in 2h 03m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee3ea5d4d31f44e1b0500729f5e557b8 : FAILURE in 2h 16m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9efd6883bdc043a7b6e902eda850ee6c : FAILURE in 1h 59m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/229e8f05ca594b74b2e48dd34703d583 : SUCCESS in 1h 22m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1c269a63843a4570a6014d301814b120 : SUCCESS in 40m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/883f424123584c5789d99de69a2c4a4c : SUCCESS in 50m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38ccae21f1d24ac8b243e079d367155b : SUCCESS in 1h 37m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/823141a51d5f4a769c08e8934a88d691 : SUCCESS in 35m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/841aa341dc544e3f8c19edf4305b1e0f : SUCCESS in 1h 23m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/628ec3c67e1044e3bd453b80c1b35061 : SUCCESS in 1h 22m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6d4898e402c44ca840bbac8f9652f7e : FAILURE in 1h 45m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32f0c94b8d0944298d67ab68f1210651 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8cfac738e6c7bc9e0761b590ac39284e38a4fc42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-19 09:32:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c18d4ddb651547b1e5e56c62d7f3e011242c9965","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 09:52:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/565cb703bd62436989855cc443e267d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acf33c71f72a41fa8763dd6142a714a5 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d4df5f9c862848d88f337695d98c0c78 : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd6ed8a999e8b37f5e1b61a4d72ebf59399a97cc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-19 10:16:57.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9cbba64f9fc6428ca4fef387f1cab261\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/194676270b4e47fbacc1a10b82df78ac : POST_FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4d5ecba14f5227058ab9a3e5507335cca07db872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 11:23:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89442c33d3334fc38f61626d5d99cee9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aabc66dd19c4d84bec84835e71e32d2 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a44ae4239514800a9d95665330dbfe8 : SUCCESS in 9m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfcbd33949624b3ba44317ff6abdba20 : SUCCESS in 10m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d6ba8703cde4a9396635c7b7267b384 : FAILURE in 11m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141411380605405fb7f96d1aece3ca7a : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6daa6ea9527742cbb6cf982608813715 : SUCCESS in 40m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4bcc047919b46a18214aaa39bd34c07 : SUCCESS in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77da748816c34df6a0ca4744d1180c55 : SUCCESS in 1h 08m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1af32981e6f4b1eb049e96e57a00cef : SUCCESS in 17m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/764dba151d424deab33a72b7f686ec92 : SUCCESS in 4m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3fb77174c33c42419da02a87f368443a : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df6f27b67a604127b1224adc755f6d30 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e1b45547db0145ae9587f8c1f2b93b16 : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9f328174e84d4825b9d4b1218d0a8852 : FAILURE in 34m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/66514ae26bc64d5397e3c2894aa92930 : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ddc17900085d4df49d166f963a4a8508 : FAILURE in 1h 37m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b0b4149f14064e9bb44fb2c813a2cc7e : FAILURE in 1h 42m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4859c27178544bd2bbe4c0c1d197c204 : SUCCESS in 1h 43m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/45ef37a0e4264a47a7376991387b2749 : SUCCESS in 1h 38m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bae667de5e8d43c88da1c007042b87b8 : SUCCESS in 38m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c658c9b9fa064f008266b0130c14ed33 : SUCCESS in 50m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f75d7a7d6c944b54a58b10bee1e954d3 : SUCCESS in 1h 06m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f91c9a24f9c8490cb7ea5b74b7c2199e : SUCCESS in 37m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3c58e257c2034c6b90944fa0ee96ce13 : SUCCESS in 1h 30m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/60dd85d9bcf348fcbbe4cd1fd8abcf3b : SUCCESS in 1h 29m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dff60b5344a74eeb9172061051103f64 : SUCCESS in 1h 36m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1352be1e80454234bf789fc3ebcf7f89 : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2853a66c5d7f2ed98d288780e2aaf73ae3b23ddf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-19 13:42:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a636751e2bbddf890481f18b78baa13d5afd745","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 14:02:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0af00551adda4daeb9f13bd43ecf886a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/769a45e7501747ca8f3b42da8f72b5eb : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d8fb36f576d41c69b750de6c47afc32 : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"39261c57ff510c0215c51318088af344313b1c9b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-19 14:30:26.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9a0fad4055b7402fb7992a10f209909d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a91172d1b4cd40afa0f162299c941613 : POST_FAILURE in 20m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a86050075e95ba6efb6f0035e2f52afe94f8d9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 16:11:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63378b81709a4b60a36c57afbc21cb4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42146e52e35f4b6bab87489b58f82727 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/674061942e7d42018cef2980e076ec0d : SUCCESS in 10m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb4d586791374ab6881435e1d6f0bf6c : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/300e0101c33e4ad5968c57498f15a23e : FAILURE in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa0dcd33b4b4d94bd1d7f436e7ba72b : SUCCESS in 15m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/57b103c2b3694cc181db1eff8bf46f08 : SUCCESS in 49m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ce6a9f5eca6484a8d0347e7e4e65cae : FAILURE in 1h 03m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04aef935350c4051b6fcfa96bb4b9a64 : SUCCESS in 54m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/096142d01dcf49d89f2898ab9b9ece75 : SUCCESS in 15m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/715eca4149384f3ea20c105797760b42 : SUCCESS in 6m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58b32745986442628e429ee2d02084f9 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/382d2f12482b410181e599bca8933646 : SUCCESS in 10m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c4bb245f2e3d4ba8878861966b90150f : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba4e9814f70943cd9007aec240def769 : FAILURE in 31m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8857e9f7cf1844949da3c3db69c3bbf0 : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d8646f9e93b84432bbfa24fcba6454e1 : SUCCESS in 1h 24m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fe65d3bb463b46b39bfef8b142c53dd5 : FAILURE in 1h 40m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0376dca5cea84420a3c54bf8b3dccc74 : SUCCESS in 1h 43m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3b797a502a4f45c889ea8e2659c538e8 : FAILURE in 1h 59m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04667e4508194747a0d8583a8b3370fd : SUCCESS in 39m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/86def3ffc445444dbb62dfc81912fedb : SUCCESS in 49m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e49108f243804043a96f15a9ecb74f87 : FAILURE in 1h 32m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e7992de5e89d4bbfa3d0429c3bd61e6e : SUCCESS in 42m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f43c0f6347de4289a29bce9af278068d : SUCCESS in 1h 21m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f593bd853bd04e79a961c0c8b11cd2ec : SUCCESS in 1h 30m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/98aa1c5fe8da459db98e9ba2365ab5c3 : SUCCESS in 1h 44m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aab4da9e36b44cd9183a28333267c38 : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e2068a27d59ceb4c4bac0f6dc87d5a1ce900ea72","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 01:06:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55d4a5718dcdb038f5e789ddff4a63c5e19addef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 01:25:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2834a46d2d814400a41bd5c1785a39d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f54f4070544e4f9194bf87b9cb8d9f27 : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2194610d73d846129bec2352556e6321 : SUCCESS in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5386eb618922e7457283250728907f2f61ef0114","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 01:48:10.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8b08f77839184136a8bec404f6cf85d1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/294bb08b3e43469fbabe3c9633329214 : POST_FAILURE in 20m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6eef9f1431b0f20b2b40a78dbea65faa3b0bc3ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 03:04:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1575f9a94d164600be39a626af25c4a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f71b62f2a74102a98bfac1e817feec : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38a2d9c5c3144dacb4af88f45545e961 : SUCCESS in 11m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60adf84162ae41a0a6e88654f68a72a5 : SUCCESS in 11m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/318613dcb4784942a2efb543c64b8224 : FAILURE in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd24e20fa9484bbd90bb0ea279f50a7c : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/bcd79186f5244077a0fa1d1fee29819c : SUCCESS in 45m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d394f5d53aa14095bd40ca6385a54b10 : SUCCESS in 1h 31m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71a736beea2949f9943fd13c51fa6a27 : SUCCESS in 1h 09m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e012eeea9fd84d819e7fdf8cd577a90e : SUCCESS in 13m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b8655ad1f7b042baa3995678b0006d3e : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bac7046120b8484383992c51b468c8ab : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4a80e72e642465f9755f6e4b431f550 : SUCCESS in 5m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c25acde6985b4506a2223f17f56a817e : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/547243507fdb4e7db6cabf06b2b36e0d : FAILURE in 28m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9d3aea06ea4487685ef68b1c80aba25 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d801e6de822c4dd38372bdb45efe070e : FAILURE in 33m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c97b61409af645988eada0a1a4f66088 : FAILURE in 1h 44m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fef294b9b45240cda03d92392db2fe26 : SUCCESS in 1h 47m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f4c4267d38df4b94b3f2a6b53985c3dd : SUCCESS in 1h 47m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7c3ffcbab4e84a3fb3dc1a1ef1b18279 : SUCCESS in 38m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/44730c9cb9a0403cbd395484d27455b4 : SUCCESS in 47m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/acdb5f8063854bf4a2bd85db23e98358 : SUCCESS in 1h 27m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/425891a2ee92486bb8f63f3ec065fc3c : SUCCESS in 37m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5d8d5aceec7347dea896d34b738a9379 : SUCCESS in 1h 15m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/abf0168546de4231840ff06aed9b7527 : SUCCESS in 1h 17m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8def8780c14414d8b2f7db15d18225d : SUCCESS in 1h 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e658a3cf297401e9bd11e011779cfef : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a5094344dc4d6cf6cb1d4e8f9386bb36be968169","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 03:08:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0be77032b50d505948d435ef136ace2094713a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 03:30:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dc9064e6768486cb96d90c98ebf0f4e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/feba132d48a743138d73dbcbbffc8510 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e02a99986e314540b752c86dfb57dfc2 : SUCCESS in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6837ed74abe12354e6ac4f2f399be45794afc09","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 03:37:48.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/40d91fdd534c432c8e568a55859a32bd\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2eebed1d686748e79c0f3908415dd3fd : POST_FAILURE in 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d01cb9747ec6e2dda2b1522f2786d61635d26894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 05:44:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac8561dd333f4977b747f287499264dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d62a781cad49c0bf2dfc82d9c03b2c : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e0dc304c9ab4468b6e17fce0cc2fa40 : SUCCESS in 11m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b95669078543424688bcccb51254fe54 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c6b38476e314a9398f1773eaa6e39cc : FAILURE in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3d72515a574e34848b3265b770da29 : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/81df7a82af0f4d458b7dded54aa68f5f : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af4bd5f44ffe4cfb9803249f87b28f95 : SUCCESS in 1h 37m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba5d9fd9c03e4b7892c9682bb35ba58a : SUCCESS in 50m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ef89e52d36db44b1bbb28618a65ea4f8 : SUCCESS in 15m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b97acd0af734bbdb695973ec96c39a2 : SUCCESS in 4m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a9a6d6a888e4313921ddace98031fcc : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4693bd869d4c4229a6ddaddda69c9f13 : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/687d7d5e10ff4309b4f52b710de50ebe : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e359d494b081442f9d2fc874a17d7a11 : FAILURE in 44m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c7ae6bea350d41edbb90a17a0d2e2817 : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3866a70ef808414389020e4a46b0144d : FAILURE in 1h 41m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/244bafa31ab14554b10f1dee137c54eb : SUCCESS in 1h 45m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c74e6aead404298bd54071dea424a52 : FAILURE in 2h 24m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c528761151074be6aa554157485f178a : SUCCESS in 1h 56m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db138900ef6e4d3da3f72c775f62c0ef : SUCCESS in 39m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/708696210c534a6ba0e2b78e2a6c9d14 : SUCCESS in 1h 09m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d83ffdf0f59048248ba3c7f0592f7544 : SUCCESS in 1h 06m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6dc41a9803164c2786d5819a194b0d75 : SUCCESS in 37m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b94cf78358bd4095a501d2ab815dd46b : SUCCESS in 1h 47m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fafa012800964ba1a81c06b5e6deba4f : SUCCESS in 1h 12m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2a8f353af9f46ed86ef6744ad082b68 : FAILURE in 1h 56m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c79e3d81abd446de9abcb3229eceeeda : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e6b8617ebaf7c0bd425ea6eb0cfaf13aaa748f12","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 06:07:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23aed3fe5f364d06682d07a7ac99013721a4880a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 06:17:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef3b8ea5fc94800810fe20282815c5b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7c0ce3a2f9d4632a8f54c4fb30a6d7d : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c794d40965a481491ca9b734e18d89f : SUCCESS in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11b705ccc972f81f9cfc68444d68f05aba6aa0b8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 06:32:16.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/15a2e2650c8b48e6abdbca8202c7dbe1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9d0c3e3e56844e4ab6401109794ef7e3 : POST_FAILURE in 20m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7e3097a8d7fc1e7f7e9ab9e36e260701e7427dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:29:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdf64c6f1e54487899c7d5f151176d26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d051e6c8959459da8be55b8dc10ff18 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b890136e9e44cce8e12254565c391b4 : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25d5f7c395b740708094e83bb6395047 : SUCCESS in 9m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28fc3f0752f544ab9d15e6f9294cc23c : FAILURE in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c36740cba3d4670b29963d8b35a0a51 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/50dc9d62cd9b4d5497f1e9c8927de3a8 : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4605b4ee5dea457388a3fe30552f2a5f : SUCCESS in 1h 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/033b2e6def024dccb105936ff77633b9 : SUCCESS in 1h 21m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cad364ac40da420d901fc766209ab6f0 : FAILURE in 13m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/54973854cfbb46edb03a4e898e45c803 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30cf9fbbd842499695d016f6379b34c1 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c031e394a88b4cc68e61d065f299a8d2 : SUCCESS in 8m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/08d828f0a2fd488a93ee8580f373862d : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8426248aa0b44ffc98bbb7da77f8c3dd : FAILURE in 41m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ccbfd384ac64f6d8a90b8751857731c : FAILURE in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ec3b44fde434ae6ac7b095a07ff4745 : FAILURE in 1h 50m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48578e3d6c60443fbc51fa12f661cb6d : SUCCESS in 1h 42m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7b3debe59d154e9aae4d7e2af9c2c39c : SUCCESS in 1h 48m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/874a61432ad24879ad601a8173794b5c : SUCCESS in 1h 48m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1f3a337d87b6404e8061b133dfd78b9f : SUCCESS in 30m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/112c4f398dc445608bbd9eda1c7ca2db : SUCCESS in 53m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3dce938c56944b1ca49d6bb862123bb1 : FAILURE in 1h 35m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3ad38bf2a40b46cb90255923f6182966 : SUCCESS in 1h 02m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a3a4e92fbb8c4454b66b0c08cddd3e5a : SUCCESS in 1h 20m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/73355bb625d34e86910c235a4224704a : SUCCESS in 1h 13m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68e621dab30041c5b43029c3aad5f9f8 : SUCCESS in 1h 34m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a003f724591245289a165fd960f1abf2 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":2},{"id":"12f0abf281554c683ef1e321ed1b3186d8cd41df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 08:53:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49cda23439b078e79b5f4df8559d906bea163034","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 09:04:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8836c8323494272930a3b33105a4200\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdac5beb844e4feda8ae17bee20b9e5a : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/85b16b9b5e8b4b278d37ba8ecfc79354 : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c476b46d2e8370daeb5a3a7c5c6baf8e9c70e47","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 09:38:53.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/651191a713634d0099fb7460d4662489\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/574d8215c2ac4f5e9f674b733ae1cbad : POST_FAILURE in 20m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d98910295f06467a1f1da1dbebb054cd371e7c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:10:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe5e9df166c4e4aaeeb0d2dbc65a1cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b63d9698084b44b64f3946bb17b69a : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60657a7540fe4f7a8aa93614a10e9201 : SUCCESS in 9m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04ccdbe8d1164de0abe4919d4bd247e4 : SUCCESS in 11m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a1a519aaf6c45e9a8077cc9f8f0b92e : FAILURE in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9bd6dc23c5429986cc848c670cbb64 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/850c41f5f042458b83b252c5dde17f84 : SUCCESS in 1h 29m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22a803917cc14c259504e7abd99b96d1 : SUCCESS in 1h 33m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a01b0720a7824c4b81736b8b8d4914cd : SUCCESS in 1h 04m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da887040a0234465a7ade3c20b73af1d : SUCCESS in 15m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21b34b2f0d4f41ffb0a8897efe9328c0 : SUCCESS in 7m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/40f990d106724eeb8eac1ab11d7fc840 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2382d844139423badbaf28446337f70 : SUCCESS in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3880c6d0cc46493e9bb2e60175ba4f4e : SUCCESS in 5m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a439f60d8dcd4d2693ed566b46067dcf : FAILURE in 48m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eeccc21329394ce8a533b6f9d23c0598 : FAILURE in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58da79ce626b4b3ea9e230d5373fee88 : SUCCESS in 1h 42m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c51165c06446452a8ec85c14b432a92a : FAILURE in 2h 11m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07c868f1950343c99900ed0ca4ff49e1 : SUCCESS in 1h 24m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7f0a6e924b604f3d9f1212c7b6e79f4b : SUCCESS in 1h 46m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f49e6883b976466baf50bf64dd3493ba : SUCCESS in 41m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8fe7c8a0ad344a93829220c213711e84 : SUCCESS in 1h 14m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ba586d1ce7cc4579983c424734ae4d38 : SUCCESS in 1h 17m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86e2a595d6f24197993db1dab36b2ade : SUCCESS in 41m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2e490959e6b54dc9ba304c1aa90fd821 : SUCCESS in 1h 10m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/20047df7b0674e4da705d2d4a77c1593 : SUCCESS in 1h 24m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf7cd30ef3c1482793ce6dbcfe188dae : SUCCESS in 1h 32m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/961784653d5c4b7680f44aa295b956e0 : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3f79d4548fa6db228f9fa9aca2d10cfa8fb9db25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:11:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64c3beaa0109ae9383c789b881c8af6487fbe9a8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 11:47:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-fcp/b5f28ec/ : RETRY_LIMIT in 4m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/71f1ef0/ : RETRY_LIMIT in 4m 40s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-nfs/ba4b2f7/ : RETRY_LIMIT in 4m 46s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/907389/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3caa423/ : RETRY_LIMIT in 3m 29s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a1efaf88788a5d70d0c406ebb13e1fbb3f295e48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:01:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f03d31dddf04b1ea33cd272f9b2cafd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb5e1dd0a8c47cda5f8b9f2df53ae4c : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6195fb7f47ac46b1b029894dc1de95b9 : SUCCESS in 13m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dad2e33cdb284c55aa744add44c5f4ec : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080c819cd2084be7b2ca747e73fa114a : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/913f07d2672d4beab11ade3746aa9410 : SUCCESS in 1h 05m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88fb997beef1465a9a856eee30a4b48b : SUCCESS in 1h 23m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88e6ea14243f42ce8d637148fe10b9f4 : SUCCESS in 44m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0c41fe72dc04c468aa593ff7e76e12d : SUCCESS in 1h 11m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a19e9bf660a647739a3a92f9d41be668 : SUCCESS in 1h 38m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fea9c252b5e4e2a8349ab5cdd2977e7 : SUCCESS in 1h 29m 14s","accounts_in_message":[],"_revision_number":2},{"id":"704b65a68539245b7ef669686cfb808d52ab275e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:01:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"214db94e1f7e004ef0f8e1d09a02c901a1099cdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 13:03:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8def230e6be4169b9d1e678fd8d442e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a6239c923a46599ab30fc595390b6e : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a4bfc89c725d59e26cd2594ace94fa27f22dc167","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-20 21:06:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_iSCSI/5804687/DellEMC_PowerStore_iSCSI/1345 : SUCCESS in 1h 18m 09s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_FC/614e8ae/DellEMC_PowerStore_FC/1528 : SUCCESS in 55m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_NFS/ebc436f/DellEMC_PowerStore_NFS/1039 : SUCCESS in 1h 06m 30s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_PowerStore_NVMeTCP/3c33d09/DellEMC_PowerStore_NVMeTCP/893 : SUCCESS in 1h 42m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d617ac64c8e1e880463096f29c32e16c3c628fb6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-22 05:42:04.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_FC/1c856b6/DellEMC_Unity_FC/1919 : FAILURE in 1h 42m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/89/907389/2/check/DellEMC_Unity_iSCSI/ae9b75b/DellEMC_Unity_iSCSI/1947 : FAILURE in 2h 16m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d5798094ff0c2c12976206c742bb16810ecca6f","revisions":{"a28532cb1b6031eeb94ffdac2a7d5839329ad2c7":{"kind":"REWORK","_number":1,"created":"2024-02-01 07:27:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/907389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/907389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/907389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/907389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/907389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/907389/1"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:27:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:27:37.000000000","tz":540},"subject":"Remove six from Synology driver","message":"Remove six from Synology driver\n\nChange-Id: Ib089c5ddf00cac443a3ca8e16a86497e1b1139ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a28532cb1b6031eeb94ffdac2a7d5839329ad2c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a28532cb1b6031eeb94ffdac2a7d5839329ad2c7"}]},"branch":"refs/heads/master"},"0d5798094ff0c2c12976206c742bb16810ecca6f":{"kind":"REWORK","_number":2,"created":"2024-02-02 15:50:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/907389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/907389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/907389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/907389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/907389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/907389/2"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:27:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-02 15:49:28.000000000","tz":540},"subject":"Remove six from Synology driver","message":"Remove six from Synology driver\n\nChange-Id: Ib089c5ddf00cac443a3ca8e16a86497e1b1139ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d5798094ff0c2c12976206c742bb16810ecca6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d5798094ff0c2c12976206c742bb16810ecca6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
