)]}'
{"id":"openstack%2Fcinder~907391","triplet_id":"openstack%2Fcinder~master~I372aa42b868ebf356948737d714a3e1bea3185ba","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-05 18:48:08.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-23 19:48:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I372aa42b868ebf356948737d714a3e1bea3185ba","subject":"Remove six from netapp drivers","status":"MERGED","created":"2024-02-01 07:50:08.000000000","updated":"2024-05-24 08:14:36.000000000","submitted":"2024-05-23 19:48:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907391-remove-six","meta_rev_id":"0e741f333cf37ccee350ce91da740224ff5d0fe0","_number":907391,"virtual_id_number":907391,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 19:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-05 18:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-05-17 16:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-05-17 16:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 07:50:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 08:16:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 08:30:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-01 10:17:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 12:17:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-01 15:42:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-01 16:23:55.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 16:53:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 19:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-02 03:02:28.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-02 04:42:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-03 23:25:51.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-02-04 20:57:28.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-04 22:19:01.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-02-05 03:57:56.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-05 05:14:53.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-05 07:09:49.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2024-02-11 17:32:14.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-15 02:42:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-17 16:32:01.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"3d184869ad81b878fb130c379a37edd03b40bbce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-01 07:50:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2113cd50de8b4a433c27e480fa214b358195bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 07:50:17.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16436/","accounts_in_message":[],"_revision_number":1},{"id":"7dfa7b5d06e2e4051c8014f09fbb2fa37340a04e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 07:50:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16436/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3b2093b5f2ee26a3fceb3964970f7c7936bb7325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-01 07:57:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96f428d3ef931c85a789ee3afc3e2348eb6b57e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 07:57:32.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16437/","accounts_in_message":[],"_revision_number":2},{"id":"633dcf92e65ff55408103e67574c5def8908b134","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 07:57:34.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16437/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"5744e15a81410b97ad933726c4bf8e35c80dc27b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:16:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53846720cd25416dbb1fc56c7d1947a1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d41645f7c1c74de285341a54933e38d2 : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07fc92a8cb4944c1a789b0f6d4134026 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1ff07035dd447ad9cb9cba6ecc23815 : SUCCESS in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d760cdc151b75fb830f9558586ae16e98d0eb5de","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-01 08:30:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_Unity_FC/d0bca3b/DellEMC_Unity_FC/1176 : ABORTED in 5s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_Unity_iSCSI/6d39a56/DellEMC_Unity_iSCSI/1192 : ABORTED in 4s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"56ca81c952b6c0356cbe8ffa70c49285bf7c0b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:17:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3246a7cfd6944989ff92fcd13206701\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de84e8dd4661472fa8d25d7fc37e6b9a : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ce79edbb2774713bda1cf0bdcba8855 : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d8a618e1baa4203b41cd26bd21fb2a4 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3945e1ca179e40b09a40d314fd45ce03 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/4d86b8de72544d6a85da0b9b28eb7342 : SUCCESS in 41m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6ad27a91fabe4a558c784bfc29587153 : SUCCESS in 1h 00m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/149e4430f1d742fe95759f7f15529904 : SUCCESS in 1h 27m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b60a2bb24d5a4360b7590578baa74000 : SUCCESS in 34m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d882373e505a4ff18892b2dd4ae73dcf : SUCCESS in 16m 32s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/119ce7b72b4b49aabe6d2c8a48c9d139 : SUCCESS in 13m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5c0593d77824d31ab38e67abe361410 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27f9826a108249a4a61c5bd1b30f86ea : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94d3c510d24e47a08e46ddf2bf188be6 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/56b131561ffa4b8cb88b5cf44612b126 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/102c2c2ccc00442e91165d04b7240da3 : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e864ff9cb7604361913d82b99c1d24d4 : SUCCESS in 8m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0b82211aa7904c598e60daf84a4d3c54 : SUCCESS in 1h 50m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3dc21f26af724cd494666db7aad721a8 : SUCCESS in 2h 13m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/026232fe923e4df08f86652fd2925fd4 : SUCCESS in 1h 58m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3bab0df255ce4048b458277b5980900f : SUCCESS in 1h 39m 21s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98117595aefb4b20a9161a0f2b077f7b : SUCCESS in 1h 08m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b49e93fe8103472e930fd25ba562beb0 : SUCCESS in 1h 20m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5426827035774d3499aa3d46e4a00c5b : SUCCESS in 41m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ceb672e1fc4a449498333e0460082f33 : SUCCESS in 1h 24m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e004c7ebb1fb4503969393e0f5e65c9b : SUCCESS in 1h 46m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b777bffbb9c347a88a2c42d13ab3cfd0 : SUCCESS in 1h 42m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fcd191151344e32b3bf97eebc991502 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4b4506cd6c688ec6334b99f5208d35fac70a74a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-01 12:17:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/2/upstream-check/cinder-tempest-plugin-ontap-fcp/d193112/ : RETRY_LIMIT in 8m 57s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/e533397/ : SUCCESS in 1h 36m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/2/upstream-check/cinder-tempest-plugin-ontap-nfs/b7c7c3e/ : SUCCESS in 44m 29s\n- cinder-tempest-plugin-ontap-nvme-tcp cinder-tempest-plugin-ontap-nvme-tcp : NODE_FAILURE in 0s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/2/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/0771169/ : RETRY_LIMIT in 8m 57s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/2/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/b20d578/ : RETRY_LIMIT in 8m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"400ca9f923c8c5f24236c19781783d300dca7340","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-01 15:42:29.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fa12986e5e8a4a249a63771414f1b9bb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3c1d2afe2d6e4e30a0e34f738922eda1 : FAILURE in 1h 26m 38s","accounts_in_message":[],"_revision_number":2},{"id":"3bb758e859915e0fde880c475e9e7e0ed58ffc95","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-01 16:23:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9c127b2a60354115bd35ea66384360e1 : FAILURE in 54m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2d66b9db400141e299b4f9fcc081d54d : FAILURE in 51m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a3dfa2e529869c9e3eaba9be038c4f26cd1f9141","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-01 16:53:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907391 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":2},{"id":"422959075469a5971e357f1671e317098c42fd28","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-01 17:22:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_Unity_FC/76098d6/DellEMC_Unity_FC/1186 : SUCCESS in 1h 04m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_Unity_iSCSI/be4869b/DellEMC_Unity_iSCSI/1202 : FAILURE in 15m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"acbce1a9379a337e8e9c5e8fa44658476aa3ddef","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-01 19:00:06.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/907391/2/fc : SUCCESS in 0h 57m 54s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/907391/2/iscsi : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":2},{"id":"74e3875d300db719a90814de4203a41c79f05720","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-02 03:02:28.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_PowerStore_iSCSI/e2c817d/DellEMC_PowerStore_iSCSI/1002 : SUCCESS in 1h 06m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_PowerStore_FC/cc1a5b6/DellEMC_PowerStore_FC/1158 : SUCCESS in 1h 31m 07s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_PowerStore_NFS/a64bbc0/DellEMC_PowerStore_NFS/688 : SUCCESS in 1h 04m 33s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_PowerStore_NVMeTCP/43c2548/DellEMC_PowerStore_NVMeTCP/555 : SUCCESS in 1h 53m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"4a918c364b3f79e1f9f7601a7479416ed6aa80f4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-02 04:42:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_PowerFlex_3.6/b7b273a/DellEMC_PowerFlex_3.6/1335 : SUCCESS in 1h 31m 02s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/2/check/DellEMC_PowerFlex_4.5/5ecfc6b/DellEMC_PowerFlex_4.5/303 : FAILURE in 2h 03m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"398b55e490bc46aeb59998ac1326c78b6081e1db","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-03 20:37:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/db4ba23761ef48179df90c1050e2291f : SUCCESS in 1h 11m 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0a39cb7935b54acf8e7c955b22f12717 : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":2},{"id":"bc38ed6aa83fbadd7f9678c6f1c80aa07e581e06","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-03 23:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907391-2/HPMSA-iSCSI/ : SUCCESS in 2:00:01\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907391-2/HPMSA-FC/ : SUCCESS in 1:38:19","accounts_in_message":[],"_revision_number":2},{"id":"05938048a2bf70c35df024bb55d3e74c549bed53","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-04 20:57:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9d4db1e03ad4c823c512be3e78828018adcfa2b9","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-04 22:19:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907391-2/Lenovo-iSCSI/ : SUCCESS in 2:04:50\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907391-2/Lenovo-FC/ : SUCCESS in 1:54:05","accounts_in_message":[],"_revision_number":2},{"id":"49d80be2fe71d0a6c953f7ac54a413532863dbbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 01:23:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9d72ca6e1f274972da08c0f6e248c972969c7679","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 01:23:39.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16473/","accounts_in_message":[],"_revision_number":3},{"id":"7b1cdd130006aeef0d4b7aa8dcaa1a1536c7f403","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 01:23:41.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16473/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"bd84b587defe5e65862d2a1d3719122c67a92d56","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 01:24:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"882d479cea906500517b6532a11c205f1f4ebb32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 01:31:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"323f485a439b310f262e79a149da127a4525639e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 01:31:23.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16474/","accounts_in_message":[],"_revision_number":4},{"id":"6d5e0c266a2d760a3d0fe09629bf132462f7f2dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 01:31:26.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16474/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"5458fdac4e548f0c089e09d2efd011f4f04897dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 01:48:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94eb60b1ad91444da0985b832f6e3e92\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4656dba791546d2b823615f58976ecf : SUCCESS in 16m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6388a467db9949cb8aeacd095a92b725 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96577d81d0fa43d594fc732624bee4c1 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6881e584ebbb51ae277a1657de71d357ddb91e47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 02:16:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3b514cb789b8688f4b894151080afe32f818fdf8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 02:16:14.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16475/","accounts_in_message":[],"_revision_number":5},{"id":"f5291556235cf497d8a7e28d4c5eae2e487e2b3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-05 02:16:16.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16475/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"0a451203c8ad8729dd62a3252a015c8196cc3fc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 02:30:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e219a4416ba54a1394e31aa315c8c146\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/981be8837dfd42da90e965c99916fbc9 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0cbf582eeceb4546a12c4842ed2cbee8 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dbd11b4a9da042dca37c223c789cf70d : SUCCESS in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3fd53e982ac93a2a955da8e0dd45ca994071ffcc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-05 03:13:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907391 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":5},{"id":"4f97c34b4dc9afa33ba452e2c6d875e89e4ec061","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-05 03:36:52.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_FC/6c6c606/DellEMC_Unity_FC/1241 : SUCCESS in 1h 20m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_iSCSI/83770f8/DellEMC_Unity_iSCSI/1270 : ABORTED in 4m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"929407683445fc40209369de1f406bdf2abc5a7f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-05 03:46:52.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/73a8568ab12e4c9b8b81d9192209f9d7\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1662f221668b43128e94272a4ad27d27 : FAILURE in 1h 27m 37s","accounts_in_message":[],"_revision_number":5},{"id":"9baeccd24edab264214d552b668d2f85629165e4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-05 03:57:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/597/.","accounts_in_message":[],"_revision_number":5},{"id":"30a8060fb44571ff0cd6a560b1242a940c0599c5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-05 04:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/907391/4/fc : SUCCESS in 0h 30m 52s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/907391/4/iscsi : SUCCESS in 0h 46m 15s","accounts_in_message":[],"_revision_number":4},{"id":"15e1035cd3c9adad4c327a5a770ae672a7ba2821","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-05 04:39:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/608/.","accounts_in_message":[],"_revision_number":5},{"id":"f6a40f4be0cfb0c4bf07de79e4c8c906405ae1b7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-05 05:14:53.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-91-907391-3/FC/ : SUCCESS in 2h 35m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-91-907391-3/iSCSI/ : SUCCESS in 3h 51m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"eefa99384b84b19ff17a7965fdd6b0952a178569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 05:26:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4484d8854d7b4eddb8a425a457be8b99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec012e9564849d1bd1e7bd8feb03786 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c77117072064b19a37f80326d7ff2f1 : SUCCESS in 12m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc43311852a2453aabcb0818941cd396 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6f097a4ce7424db0fc66db8ad9e3c6 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/b457202ad1374ebd966cd834a4346511 : SUCCESS in 53m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c4ecb76aca454c70ac5f09fbe1888ed4 : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b775cc83c284770ae34bb63138b5d3b : SUCCESS in 1h 23m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/442cfda47f3448758b84f3630ebf75f6 : SUCCESS in 24m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5cb617d86663457d8ec067416f2d509f : SUCCESS in 14m 02s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/10eeb86179e944fa8a7bb3f4e3887d34 : SUCCESS in 15m 23s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/588efcb5c33043e7bba3d7b87efe228a : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06b38b2de88b44058f15136f02a6536a : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/acb5e20ed9c34e4091ab71b362ef96c6 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c38baafc49a472cb2adf3388243892c : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/530b9cfa975e46d38916c8c2101136f3 : SUCCESS in 36m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b175d4ea55146d9820fed24c55ae9f2 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a4b985a0e31b4a6097e1fd19c7fc1125 : SUCCESS in 1h 53m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9a7c48ce7b84da7866d75752ec4c231 : SUCCESS in 2h 10m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d78a9d21400e422ca20e451d35c74b25 : SUCCESS in 1h 19m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f8d102b930634302896afccdf19d4534 : SUCCESS in 1h 32m 44s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e38732f436384ea19253abd6b534ae82 : SUCCESS in 1h 12m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/84af5147dc80442caa05ac532e1d1b18 : SUCCESS in 1h 24m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69b5686245b14255a76c7161a6a507e6 : SUCCESS in 46m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4283374b857449599b072416c88a1c0 : SUCCESS in 1h 32m 32s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fe0657acaf524eec884599fe5b8a8b6e : SUCCESS in 1h 17m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1dda99ae0264436e98877044b557e817 : SUCCESS in 1h 28m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef7260114cdb4b76bae292172bdd667c : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":5},{"id":"5668b25c7767251a82aace5f554956a38eccc353","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-05 06:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/907391/5/fc : SUCCESS in 0h 40m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/91/907391/5/iscsi : SUCCESS in 0h 45m 30s","accounts_in_message":[],"_revision_number":5},{"id":"e225854264bdb912ce4eed54a89ad8b18ad63abe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-05 06:39:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_FC/0a48977/DellEMC_Unity_FC/1243 : SUCCESS in 1h 13m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_iSCSI/00846fc/DellEMC_Unity_iSCSI/1273 : FAILURE in 1h 12m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"976d180b5be2d32eb6e975156d62a8d9d1fa7bde","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-05 06:58:33.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/9de1786/DellEMC_PowerStore_iSCSI/1038 : FAILURE in 20m 22s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/0f1c72f/DellEMC_PowerStore_FC/1193 : SUCCESS in 1h 25m 12s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/cee53ba/DellEMC_PowerStore_NFS/724 : SUCCESS in 1h 05m 52s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/1973788/DellEMC_PowerStore_NVMeTCP/592 : FAILURE in 7m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"06105a1d3e8663a3f7241a4baf7322fc1964c39b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-05 07:09:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907391-2/PVME-iSCSI/ : SUCCESS in 2:03:10\n- PVME-FC https://os-logs.seagate.com/dell-ci/907391-2/PVME-FC/ : SUCCESS in 1:57:15","accounts_in_message":[],"_revision_number":2},{"id":"293a208ac9f15d5cc7fa3f3d4f64b1de43f10000","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-05 07:22:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/f562020/ : SUCCESS in 1h 12m 38s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/21bbc46/ : RETRY_LIMIT in 8m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/0179694/ : RETRY_LIMIT in 9m 05s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/dddaee3/ : RETRY_LIMIT in 9m 37s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/a478c75/ : SUCCESS in 1h 53m 46s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/222256a/ : SUCCESS in 1h 50m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"350073aab32d1efe34c7844d01d902f91fe714ff","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-05 14:45:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/593e0be9b0ac4ddaaf6640f59f5d7bb1 : SUCCESS in 1h 14m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7131507d0c6e4ce68492decb9bf5917e : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":5},{"id":"69c5004d9c76c435b42012fd2498112a1249f7b0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-05 18:42:01.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_3.6/5b425f5/DellEMC_PowerFlex_3.6/1381 : SUCCESS in 1h 34m 14s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.5/163b07a/DellEMC_PowerFlex_4.5/347 : SUCCESS in 1h 49m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"71787e7f3d52875ebec1ed4ebc12c035082f7ea1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-05 18:48:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fb52c5385256ccb419bc688aa21b179e8497c28b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-11 09:26:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907391-5/HPMSA-iSCSI/ : SUCCESS in 2:02:04\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907391-5/HPMSA-FC/ : SUCCESS in 1:37:22","accounts_in_message":[],"_revision_number":5},{"id":"2dd0875a4a539ce0a37ee56578d6aae439588ea3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-11 17:32:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/91/907391/5/20240211224630/NEC-V-ISCSI : SUCCESS in 3h 24m 36s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/91/907391/5/20240211224630/NEC-V-FC : SUCCESS in 3h 45m 41s","accounts_in_message":[],"_revision_number":5},{"id":"bea157cafb3b4f1b08bcc042b259314550c80c56","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-15 01:09:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907391-5/PVME-iSCSI/ : SUCCESS in 2:08:25\n- PVME-FC https://os-logs.seagate.com/dell-ci/907391-5/PVME-FC/ : SUCCESS in 2:04:32","accounts_in_message":[],"_revision_number":5},{"id":"05a27548d888380074b16e054c869deb689d93e9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-15 02:42:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/91/907391/5/20240215073111/NEC-ISCSI : SUCCESS in 4h 01m 49s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/91/907391/5/20240215073111/NEC-FC : SUCCESS in 4h 11m 02s","accounts_in_message":[],"_revision_number":5},{"id":"7d5c2da437d65e9babb2d8a5cc22cfe4f1765ad8","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-15 23:59:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907391-5/Lenovo-iSCSI/ : SUCCESS in 2:15:11\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907391-5/Lenovo-FC/ : SUCCESS in 2:05:55","accounts_in_message":[],"_revision_number":5},{"id":"3df74b69492522974f7c77318782d49d85a67538","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-17 16:32:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5acf42bdfe7c5dacf30e8e8c0d079364d73047c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 16:32:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ceef0c4932fa225fd574cdaf2c20318fc5ce720e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 22:35:37.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b309e22a14e544c19f289a237a295961\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8998922347442258cb01eca01bf2f8e : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b233615df50546ac98c34bd3a96fc01b : SUCCESS in 11m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33ef8d0d0da04502a7a1f136184cf014 : POST_FAILURE in 30m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672bb4ef22d443d591fff4fe552fa3c6 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/5c14021bad304b5e86f8800f54a2c6cc : SUCCESS in 1h 01m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7c3de0e2c2d41818cb0d3e0ea16a800 : SUCCESS in 1h 30m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0af24cbf9d042abb6ac10282454420f : SUCCESS in 55m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d5ed5fbb0a564d929a39d7f3a2882c84 : SUCCESS in 1h 05m 49s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32122fd49ade425bb4ff36b3d00f41e4 : SUCCESS in 1h 51m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ff270d63148464fbcba584d88455e0c : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":5},{"id":"143d73cf27938a8a35e029eacd3b29c991c4bb5f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-18 02:09:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"96db00d2179ff418c7dba872fed018c0c230ba50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 02:30:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7009460b20446e87d62f9ece619d8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70574676dff643c9b616d6abb14ef0f8 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/27484de4665a4033a281d1de7c63bf92 : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6cdd6d4771f9a11f4baccd58cd0cfb7a845aafea","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 02:34:15.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1b8995e8344b4fc4b275466b381b1db8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/472e7df5c0ab4120ab61d03f9321c7fa : POST_FAILURE in 20m 54s","accounts_in_message":[],"_revision_number":5},{"id":"323167dbc0ead4582e18b68555dc059407cd2f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 04:43:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cfeece803f84c61bf166fa73e4f0532\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ef3c9f42d34bf688e09ec5e5f78041 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60edeb44a6f64e798c963416fa765420 : SUCCESS in 12m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c2d1289408f4f32a76845a7ca8bb1b1 : SUCCESS in 11m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/486576f5a38a4afdb9b8e7340cb5b0bf : FAILURE in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6714a198a4074a45891f68252e27c961 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/1641657006a14533a0d59b7b88dcea0d : SUCCESS in 45m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0f9b25cc2a144cd8b996d4222d42fb2 : FAILURE in 1h 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78af7ffc32884116915419a6dd00eee3 : SUCCESS in 1h 06m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2f5e017eafc943108580fc5c126dced1 : SUCCESS in 15m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/597b11aa482b4c2e8bdb8e74d0f21aae : SUCCESS in 7m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e6f708a68234ad8988f23b9bdd5dbbc : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d3395077fe9484ba9993839a1f8da92 : SUCCESS in 9m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7a5a63268c184533a08c127c77b12da4 : FAILURE in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ee4cf1de5bb4506b11f480facde4887 : FAILURE in 49m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/379fd89bdfdb496994cce6b4c3438d3e : FAILURE in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/336906281d4b4114a188826fc8e878ad : SUCCESS in 1h 50m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/698db2e430fe4cc6ae3963fabe0342f8 : SUCCESS in 2h 21m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52dd980a8e624bccba48ae9226a0dd6a : SUCCESS in 1h 45m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3f7d8f4420d14c4db555275a17cf70bd : SUCCESS in 2h 01m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d53024fcc84f4d408b20cedeab9e7866 : SUCCESS in 39m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c06bb16d13414cd3888e89b6839e9ff1 : SUCCESS in 1h 12m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1851051bbf344d4bc4baa1c4077b61c : SUCCESS in 1h 29m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f4230e6b4d6f4dde9952583e748d12fe : SUCCESS in 45m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e5a35c902de747d181d84ecfcb31b770 : SUCCESS in 1h 22m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/fa5eebc8a2e6425eb85b32439bc677c1 : SUCCESS in 1h 26m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/509cec53a1cd47378548e0cf3a3e2a66 : SUCCESS in 1h 31m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b601b3d49c04a339531b7bdc89f5cca : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":5},{"id":"492fb0aba93c0a5ba6f3dfb29b6ae0e1d3b395e3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-18 05:09:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c236c09a8fff19c792e289b4afc5a2ad896015b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 05:26:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66be6ab79eaa4f6ca7c1988f12ef4dc1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87e81f2b444a4e4c8440ddc9ac09ae2d : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/65f09b7fa4ff456599c73c6eafda2c3c : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9dd6eedd12954f8943a14f0af295ea6f8a3a341e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 05:56:09.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/86de887f832a471dacb7562f2ab1637e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ca3bd43951b242ee850178ddddd27d10 : POST_FAILURE in 20m 58s","accounts_in_message":[],"_revision_number":5},{"id":"bf55133347fd5b52f485dc6eb834076738e3f297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 07:47:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da85a46ff03f4684821d64af59899569\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3145a5b8f05453ca57c49e348938f09 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56ffbcccc3db4346b99d69dde23e316e : SUCCESS in 10m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5ae9cbd32d3445c82e5758ac5416173 : SUCCESS in 11m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a60a39ab1e624fd1ba1a245ae82bb5c1 : FAILURE in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48dfc29e09441879529ef0943a87e2f : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/4162e403f7294fd5acb86fe4265c735f : SUCCESS in 53m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b95358d660bf425ebd907278e822b80a : SUCCESS in 1h 29m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/684c013c8a884feb89e70b10f2c7a4f0 : SUCCESS in 57m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/02b4e885c36f47fe94ecad0dcf544188 : SUCCESS in 12m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8b64ebb11f34b3e9474b6a035f2f39b : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b82bcc5723b24546aa58f6d94137d2d8 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc7c10ae4e8147e78089220956556c61 : SUCCESS in 6m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a8f8ee5d9339402a96ec1c2b1b9aef50 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82abd5a85c9e427ba16c1101674ae510 : FAILURE in 38m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4ecb5cc3943473ab1ca3717b72d8b6e : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f3bc5660edc455c8a37b68a080d00fa : SUCCESS in 1h 40m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f332fbc367ea4e0aabfea637578655cd : SUCCESS in 2h 31m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f98543f98d743e3a30865264492f570 : SUCCESS in 1h 20m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/25a6f3cad0494110a9f6e2ad2de08968 : SUCCESS in 2h 04m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6eeb0393258e4af8972ed81d15c0c260 : SUCCESS in 35m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23e5e2da11c44863b379849708513e60 : SUCCESS in 54m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/161774708b9040caabeef2745d8fee19 : SUCCESS in 1h 48m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5defb229ed764334a37ca9e8bf37ed4b : SUCCESS in 53m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/036ac72f47af43fe8c400abf20e55eb5 : SUCCESS in 1h 25m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ae7eeec902454ad7ad3411049362a57b : SUCCESS in 1h 29m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1078c05460b141c9926405cdb8326005 : SUCCESS in 1h 47m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f216999fb52f43209e7dd1d4e1fc812a : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":5},{"id":"523e49b2b79a0637118e13d8d9905dbf7e7f1487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 07:48:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c0562cc6fc4ec67d706ada62cb4c230bf1b38c8a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-18 08:03:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/9ac0978/ : RETRY_LIMIT in 3m 48s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/786012c/ : RETRY_LIMIT in 4m 18s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/e152c92/ : RETRY_LIMIT in 3m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/dfbb8e0/ : RETRY_LIMIT in 3m 39s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/a20e38f/ : RETRY_LIMIT in 4m 14s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/7c4b451/ : RETRY_LIMIT in 4m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"38ebcdf7967ac49f98819d0fa63f95624fdbecf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 11:03:56.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b62b0f6ce247218a2fdd012cf38b8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58f15e3c1464ed8ba3e25fa557f535a : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa505b39c94d4816b43c46f8c4217b9f : SUCCESS in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a435399d1724cbfae70eed2352c9bf7 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b75b921abf4eef84206cfb14f6420d : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/372b18fbc2174205a4002fa76c0dd1e0 : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7c5174d48214e55837725061b0f6f78 : SUCCESS in 1h 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c609fa8a10494dd7b2fd069a66305f96 : SUCCESS in 1h 01m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/51f74deaa7ad4f598197e6b3cc0159b9 : SUCCESS in 1h 11m 19s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6804fcebea9b4802adf860aaa03a99c0 : FAILURE in 1h 37m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/613c94423005432bbd15ebd0222eec55 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":5},{"id":"29e6b69e21e0054c2e5eac1b6e1690a0830e401c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-18 16:13:31.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_FC/b1c7533/DellEMC_Unity_FC/1889 : FAILURE in 1h 42m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_iSCSI/28ea425/DellEMC_Unity_iSCSI/1917 : FAILURE in 2h 15m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"003530423ec84989448c7ec01fe06618b5ecda0e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-18 16:19:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"01a624fac21b8e36d072dcf8608472b8592c984e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 16:38:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285cd30c36b94070aa699cab0297eb43\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5f187572647456cbbeb48aa7c13e641 : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cadfb8c676d84c5cb3c6df4a19df9df7 : SUCCESS in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b95a45adeabea9770e9730262973f3a300b2e54e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-18 17:28:20.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ce09889ccf5f4b189a0434bdfd9a188c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/57668589bcaa46a48e8f8d04e72d3515 : POST_FAILURE in 24m 19s","accounts_in_message":[],"_revision_number":5},{"id":"1733bbb2092ec31577c45251a92fa978fe3140f5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-18 17:29:16.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_3.6/92ee77f/DellEMC_PowerFlex_3.6/1388 : SUCCESS in 1h 40m 14s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.5/a3c963e/DellEMC_PowerFlex_4.5/307 : SUCCESS in 2h 00m 15s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/1ee7e8b/DellEMC_PowerFlex_4.x_NVMeTCP/251 : FAILURE in 2h 55m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"83b5be586a5129e67743af1dc8bc891cded9e70b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-18 18:05:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/6ccca48/DellEMC_PowerStore_iSCSI/1331 : FAILURE in 2h 53m 26s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/58adedb/DellEMC_PowerStore_FC/1514 : SUCCESS in 56m 59s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/98b2be8/DellEMC_PowerStore_NFS/1025 : FAILURE in 1h 12m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/3679387/DellEMC_PowerStore_NVMeTCP/879 : SUCCESS in 1h 46m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"5d9e556e1d8c1de9a1c103e1547a22f9a7bba9e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 19:07:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0d8febcb25b461a831551d4cc67adaf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6240b854edd48febb65c835950ee912 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf2893fef91341ddbe8ceb2471b4b3d6 : SUCCESS in 10m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a4eb26b78ea44fc9c1d387c2afe8fe9 : SUCCESS in 11m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e238d63abf4e49268eef48a5f685d19a : FAILURE in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d97507f01e834e3c9263fc4f59bf45bc : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/06daff3a9dfa4873a40eeb98f0cb4463 : SUCCESS in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b088a319007430486230f8ff0a6cb62 : FAILURE in 1h 06m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4d79dba5d2342e0b4004313da8a6890 : SUCCESS in 50m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba09a0cbdae047028bbbece63ffca599 : SUCCESS in 13m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/66b7cd8897364145acb48c368a7fb0fe : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/616d46301f654730a77b2356d3cb6835 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/624824141faf409897dd0c8e11389844 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2a9c94bad9fc4aa392e61b244ab3c484 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76b91026f2e649868d1f80745bbb8160 : FAILURE in 30m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5627d17b63e54577a94040e16902c32f : FAILURE in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0fb19bb61bc430e8f4ebfa651dd991f : SUCCESS in 1h 36m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b9b737b267754f7b8b92b0231b73d968 : FAILURE in 2h 11m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fe8cfd2524744f62baa47b26835072fd : SUCCESS in 1h 49m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c1d3200e32924186b4ee6d2171e97635 : SUCCESS in 1h 54m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/333941389013450f9600d6f3d557e773 : SUCCESS in 27m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9a99b81d64fc4447bbb990ed04f82518 : SUCCESS in 1h 14m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d3682f697ea46be9e05ae7320967f73 : FAILURE in 1h 37m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a583958b9774c86829d38d2a9bc6cb5 : SUCCESS in 46m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eeb4eb368b194235b0608bd657791ae8 : SUCCESS in 1h 14m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/95c9001f919140a68f904bca34ffb2b1 : SUCCESS in 1h 13m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0249cd96058548aea84f58d22fa42879 : SUCCESS in 1h 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3556fbfd70c94057bf401d171d3bd988 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":5},{"id":"9865e6a8aa1d847e70bd2ec46e9ee84088a869ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-19 09:32:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"287197f9b2f67cb8233ffcf4284368cd038e5cab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 09:48:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/319b3abe0e6240b3b39834c2649bfa38\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5e5d77ad2bd44e298735400881c221c : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a86d4733ca524a5e8fcc37c9b5ce2b92 : SUCCESS in 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bd4f73d425fa012b23b4447155b22c1a46dcf6c3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-19 09:55:45.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/bf5efe408fd34a64b2991d1bd6a645d8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cdacf0cd82d340d8b77db8370d8e5134 : POST_FAILURE in 20m 18s","accounts_in_message":[],"_revision_number":5},{"id":"89bdf6d65448b5ed36d18e5e12faed1255e88e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 11:46:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7e02fb01b5742da836466a7d448b669\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88136e9148ec446393644858c0a115b2 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae6cb9db81cd4f6c823d036a5170f986 : SUCCESS in 12m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a05248cf5bd4160952ce78e2a8ea117 : FAILURE in 21m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba8093257cb5463182684bf12b2da369 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a49932942b846b69103ee09dee9bd0a : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/fd8892ccd2b74ddbb1bd49628b28746c : SUCCESS in 45m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ce5e7bfa01246d9b4f826dcd3b719ed : SUCCESS in 1h 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e94db85a6e1a4fa1916f50a21292d1c2 : SUCCESS in 1h 04m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca0f11784d964b1b81af01fb3d1767c0 : SUCCESS in 18m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/41064a267e8a409395f6d615929b04f0 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6d4054141714e9888c3188941889511 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/319f3798fd604ed7b8307d43d0142cd6 : SUCCESS in 5m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/63300c774a0049659d07f0db7bbcba0b : SUCCESS in 5m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c2ada715fcca46fc9d3cdf0e54b53ec0 : FAILURE in 44m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2702ffee8fb4344a3227b581f87e60d : FAILURE in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e5dbcf0504e346bfb5e50dc50a7eb641 : POST_FAILURE in 54m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3659d45ab3474d8aab28caad0c7bf970 : FAILURE in 2h 08m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f2847a5d4f4455da0789e3859f1f021 : SUCCESS in 1h 34m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/12d146bc9b4e40739c45cfd11a416a42 : SUCCESS in 1h 30m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dc795a4357e540faa6693459e4d3a380 : SUCCESS in 26m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ea2edaa47a84e3ba2060caed7feda6a : SUCCESS in 1h 06m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ed9a8a3c05040ba9256f8f455d09d13 : SUCCESS in 1h 36m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5659d30b49c94d3dab9cf1b1d42cb038 : FAILURE in 44m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/651140f7934f4ec4833d627ec90bf3d9 : SUCCESS in 1h 22m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/31f58869280d45109f411ee9292dd609 : SUCCESS in 1h 37m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d523a268cc54ad6bd26685f43c36459 : FAILURE in 1h 36m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e94434b08c8741938027d459047ece47 : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":5},{"id":"72591d7d39c957c2a9b2bbf73cc45b7ab30385fe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-19 13:39:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5865ae490d19a9dbd8f9e5428748c3095ea05a7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 13:52:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6726651d01c4145924026be8536bda6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4c4be6bd01f4ff7895b968734fc7df3 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be9983ff8ad0424db062a172647e06ee : FAILURE in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9f0b1a07f4d90c1a064168e767deb63c0a392512","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-19 14:08:27.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a36f956a0918414983f969e68977915f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e139e428311f480c87af99d0420b076e : POST_FAILURE in 20m 51s","accounts_in_message":[],"_revision_number":5},{"id":"15cb8512d99245d075bb6415aa1c730e9a9e130a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 15:54:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81b54622ebd74f2faa8616282c0cc3a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c406fc0866b4687a1c32a3ae32748de : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c68186234c3749e698d459d8222410a3 : SUCCESS in 12m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16b5f0e5454d44a7a0bc2ad8fd23628b : SUCCESS in 12m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ae2b9387a30475b9f3c89bb81d7cc38 : FAILURE in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9618959dfd814e1bbf7b6f6cac5b5229 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/3caf8e89562f424abacddaf8fba1edb2 : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5c740e516c844b2892779239dd522fc : SUCCESS in 1h 37m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c0d34be0ca4a70a2d4f3e689e4c9ae : SUCCESS in 42m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d76e1529be754d618d3323bed79e5f1c : SUCCESS in 12m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2cf8c6f97fe481fbf94bcb65ca38734 : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eff66582b8f743579cb46dc44b63c758 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/591ac306c52e4acf8d3dc1b991fbcd24 : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/948802479f1a4e40a44f472dc37f4c6c : SUCCESS in 5m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce81414e20084290ad9561c4b248ae82 : FAILURE in 44m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4bd41d301c2f4ab8acbef2182463d9a5 : FAILURE in 4m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c974a17e7db40e8841b3fdbf140659d : SUCCESS in 1h 42m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ebed52586aa4accaf2af22c7d525abd : FAILURE in 2h 01m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59aa82f3070649c2a9158d1e8b3f7751 : SUCCESS in 1h 32m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1e71b086e8994d3b9150962d6a02375b : SUCCESS in 1h 09m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b6f52d25479f4b15833d2c2dd8a7aec4 : SUCCESS in 39m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9075017153cf49afbf3653a55cb40226 : SUCCESS in 46m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b9a22988ef0743b4969ab2e61aca27e3 : SUCCESS in 1h 53m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5fdfbb29a5b244e0b3711f0025c63464 : SUCCESS in 48m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/879ead5f3a4642088dc1180b37b9e3f0 : SUCCESS in 1h 10m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4839b82f36de44c4abd9f0b8a240891f : SUCCESS in 1h 04m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28f35ece20b44fdf95dc47c32a099f56 : SUCCESS in 1h 19m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29234646f5204926aa74f61d6b17a104 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":5},{"id":"3bb3c741ce33e480e4cddfd6bd999409ad05f119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 15:55:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4cf2abad8fc766f1eed12f33266e6c8b508fb482","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-19 16:31:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/1871865/ : RETRY_LIMIT in 4m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/54bcabd/ : RETRY_LIMIT in 3m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/8ace141/ : RETRY_LIMIT in 3m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3f48789/ : RETRY_LIMIT in 3m 26s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/d087504/ : RETRY_LIMIT in 3m 58s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/c8149ef/ : RETRY_LIMIT in 3m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"65ef063ba12d43babf170eafe6ff12f0e754bd85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-19 17:58:38.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/056c01957fcc4027b51a101e72e706cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbc229d7b094c4aa97b8d2db0cfc3b2 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74bb38ff292d4af7b5529ea1f6f7d4f9 : SUCCESS in 9m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d10dfe8a00674eeabd19f18a4cf74e9e : SUCCESS in 14m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12837fbf0b84fb6a781b635141a2645 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/edac1fa8d0c44f0ca98ab1a0fd9d2ab8 : SUCCESS in 1h 09m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ab4b4f5a31d470d915ac5b80dbb0c6d : FAILURE in 1h 35m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eaca4e22b6814b7ba9c5a40962c759a7 : SUCCESS in 1h 03m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/85f4beeed9f94a748c58fb1750de2bcb : SUCCESS in 46m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5b76d4a1dafa40d59a200afc5c4c44db : SUCCESS in 1h 52m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0bb246586754673a03bdf1bfc09b7e7 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":5},{"id":"e6fb52f66e9dcd14e1872c177df191b1e76f1580","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-19 18:01:46.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_FC/f1b4745/DellEMC_Unity_FC/1894 : FAILURE in 1h 45m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_iSCSI/e51b020/DellEMC_Unity_iSCSI/1922 : FAILURE in 2h 06m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a3813c01aaa5488186243d4447657e0e2fd19423","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-19 18:49:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_3.6/79fc7e0/DellEMC_PowerFlex_3.6/1393 : SUCCESS in 1h 48m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.5/c9d75ed/DellEMC_PowerFlex_4.5/312 : SUCCESS in 1h 58m 17s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/783ad76/DellEMC_PowerFlex_4.x_NVMeTCP/256 : FAILURE in 2h 55m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"2df26da4cf672ecdd1d6fe0cc522b80f73fc57cf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-19 18:54:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/b2df092/DellEMC_PowerStore_iSCSI/1336 : ABORTED in 3h 00m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/de4797b/DellEMC_PowerStore_FC/1519 : ABORTED in 3h 00m 04s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/f463bd0/DellEMC_PowerStore_NFS/1030 : ABORTED in 3h 00m 04s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/7f1fe8b/DellEMC_PowerStore_NVMeTCP/884 : ABORTED in 3h 00m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"e54bdbacb77b1319dd2849d9461fd02a1435adfd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 01:03:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"21ac8fb4f19e97cb2405aab106b85a7b60cc8776","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 01:14:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd49d9997ae4c5b836a5e2225027a19\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd5af81283224fe4ba8ed43c20a44dc5 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae0bdb7f726d4db6a7dc1633ea6b22bf : SUCCESS in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9779aab6086a03757da5e7e2e8760192ecf029b8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 01:26:09.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a9409e6df8674b0abdab84f6643f3c3d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7796d91091cc4c03a5c0c373a3de3971 : POST_FAILURE in 19m 29s","accounts_in_message":[],"_revision_number":5},{"id":"9951b63973f60eca463c4b64a3bcc8cf6612d3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 03:20:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0813a5224b5f4db69c4a74a01e977e65\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4afc15176403463fa98c050ad81ed16c : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a523521f933d4fc3ab8fdeb5a3a28751 : SUCCESS in 9m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56ed27af1e1a4ee8a8b1a54b4883dcec : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a8d6570652247f4b91047b33d36eb1a : FAILURE in 9m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db72fa18f79446f486051949d33a8169 : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/46ce1b28246f44a89c2bda523a2deadf : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f51935168de4649902015ca1a8496d1 : SUCCESS in 1h 37m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb0c90619bfa492199e0c7f20f339107 : SUCCESS in 45m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fcb1cf9feb2d4c1c9dd81738c30d9bff : SUCCESS in 12m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5298c859e1324335909528c0c901f8e8 : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2bc16afece5a47e1aa68775de6f33878 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ff6a4364dff4866b2f9032ff65c15d7 : SUCCESS in 5m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/413aacc4f00a43de914756afdfceac82 : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1a6d4c641854bc3ae77c434501fae30 : FAILURE in 46m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e223b7b8ed7489c889d8f6e8d53ed7c : FAILURE in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91cbee0cd3574d70b3d57c6e2e23823e : SUCCESS in 1h 33m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28cace2872164841b1911c30fc8fe230 : FAILURE in 2h 06m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8fc86048c43d499b8dffdd5fd50a42d4 : SUCCESS in 1h 30m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b7fb948335df43aa8320e86b58f4f87a : SUCCESS in 2h 07m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ca161c83e6324add9584d562e82d8da3 : SUCCESS in 23m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e66269735da04f7dbb4bf8fcca345989 : SUCCESS in 1h 13m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61b3df58c4564e54a9e5ad2ca83c671b : FAILURE in 1h 45m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/49b18046ebcb493a8d16135647c30bfd : SUCCESS in 47m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/157db216388c4f0daa1d06300238aee8 : SUCCESS in 1h 06m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/087e825b2c4949b884538d5e3176e923 : SUCCESS in 1h 01m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e66b753137c54aa1b8728071858bceea : SUCCESS in 1h 53m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87a9fe66e4fb43f7a05de4640506a036 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3f6368984d11adc75e926c942bf76cb56ec14d85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 03:21:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cdd6055c2461efafb73f459b9002909455652c0f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-20 03:42:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/3cfef13/ : RETRY_LIMIT in 3m 42s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/4c218eb/ : RETRY_LIMIT in 3m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/e0ccb38/ : RETRY_LIMIT in 3m 44s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ea7a8a0/ : RETRY_LIMIT in 3m 36s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/00ffb3c/ : RETRY_LIMIT in 4m 05s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/c28d999/ : RETRY_LIMIT in 3m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e14d987c86256dbb90e1b77834e20bd9fe77dbb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-20 05:23:44.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_FC/2571cb5/DellEMC_Unity_FC/1896 : FAILURE in 2h 02m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_iSCSI/31a36ff/DellEMC_Unity_iSCSI/1924 : SUCCESS in 1h 41m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b92d7e99c1136146b2e311bdefde5c69a9649106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 05:45:06.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6aabf4e260c41c5aedef38366d8211d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc0393703914251aac1dd20e237b98f : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1496e19ece9a440d8b9d68a2bb2fd837 : SUCCESS in 12m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dffb0f33e44c4809a2c6f84f4dbd2932 : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c20eeb11fa14b3188780de7c8f1d54a : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/bafcea872d084864a135a7371052a1f9 : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b680b7ccee9045bab59ad7c927c41c77 : FAILURE in 1h 26m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0f2a751e02d4c6c8edd0cd72fa0f165 : SUCCESS in 51m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d51abe7809954daeaec25241f6be67db : SUCCESS in 1h 08m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ad8585f7ee948078868eb122e9dd625 : SUCCESS in 2h 05m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bbe7f496a13417d959500a6d4df3d08 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":5},{"id":"ee6edf26f8dbda7f537468bb13878ef3f91c3d07","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 06:06:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38e5b0a8e77bcec29696310cfab75827a1666905","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-20 06:11:02.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_3.6/b663cf1/DellEMC_PowerFlex_3.6/1395 : SUCCESS in 1h 42m 41s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.5/ace49ed/DellEMC_PowerFlex_4.5/313 : SUCCESS in 1h 50m 39s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/4b77dd5/DellEMC_PowerFlex_4.x_NVMeTCP/258 : FAILURE in 2h 49m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"b84ea192f4fc8db72e2f391a89703a67f7db508c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 06:24:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/392f10d65dd94e31916b5885a52b272e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec533f518e1941b8b25380c4f0d2a75a : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bca19bbea07243f09bda632dc215926a : SUCCESS in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6823f410f7157f27828bba2267d0f7ef4ba17986","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 06:53:29.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3ee3e0f12f754eecbe4a1aa1de5155b4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/151183f282324d58acaece00e23a9ccc : POST_FAILURE in 19m 53s","accounts_in_message":[],"_revision_number":5},{"id":"8e7a68712b3ef9cfa48913e7a89d9b0232ea0e08","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-20 06:59:26.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/1df12ab/DellEMC_PowerStore_iSCSI/1338 : SUCCESS in 1h 18m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/74ac0f1/DellEMC_PowerStore_FC/1521 : SUCCESS in 1h 03m 28s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/96c76fc/DellEMC_PowerStore_NFS/1032 : SUCCESS in 1h 03m 13s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/0d0f866/DellEMC_PowerStore_NVMeTCP/886 : SUCCESS in 1h 42m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"cd94e22108d17b77be2d451d6b4cf5fc143a701b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:24:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/835a8cdbac9f42b5900fedfe92fa3db9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f62014e935842fd92e1607b6b298225 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21de9bc552b4717aef1c9ba030ff57f : SUCCESS in 9m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2af9c313a42b4b93aad71a6f24c96b35 : FAILURE in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cfc0782660d042a49e6d1ceb18e5e9a0 : FAILURE in 11m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d9995ea5a54991bafca37fabeb9239 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/865d4a7079444fb4828fc82cf93779cf : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d1395a63ebe495ab7f7b1f83d8b286e : SUCCESS in 1h 23m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cd6bec9ca3547f28740c60c111c5ac9 : SUCCESS in 47m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9517763211cd48f9b0cb7a5524320c3f : SUCCESS in 12m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f1ab92d3b54c4fe5b94634a4b972b7a2 : SUCCESS in 6m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/070a628dd1d54871986043472c1fa6b1 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e5bac26e14484ec8ab9436d220d7cd80 : SUCCESS in 5m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0d45a8658e064df99c273a0af1b15a28 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cbd6322330514ec4ad1f1e8b4d33f084 : FAILURE in 31m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c10245fd32f244e0bbbf72106eaff5d8 : FAILURE in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bece136965154fd1962aba76c14ed838 : SUCCESS in 1h 52m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e22aaf6bfa774a3386a37401d6aeb9c1 : FAILURE in 2h 05m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/697313ff46b340848e322b4c2af66ba0 : SUCCESS in 1h 49m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8760d18483bd49ceaa37669958f7d6a7 : SUCCESS in 1h 49m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0c104f5edfbb4dda9bef13239f566a0d : SUCCESS in 24m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4997592fdf0d438a9ca4b475f2133404 : SUCCESS in 1h 06m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9e799d53ac94446b687756a45a86bb8 : SUCCESS in 1h 14m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5506b16717404b0281f965dc691396c8 : SUCCESS in 50m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/154f57b87bb644309253351157fd3a8e : SUCCESS in 1h 30m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4408e2dbcb894f54a80c21b4dfb4cb8b : SUCCESS in 1h 02m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ce07563a7e8441190580ffb817e94e2 : SUCCESS in 1h 15m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09b06467f89c432597c004df677b78ac : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":5},{"id":"f1e58d5e0868d38fe31158aa63fdf4bfca71386e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 08:50:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"915080c3d996e244ff7076fb1505e89fd078aeba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 09:16:13.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a2f368a5a1874b5aa3c15288cf96e080\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dcd3531059264c9ea6d3144e14956de4 : POST_FAILURE in 22m 34s","accounts_in_message":[],"_revision_number":5},{"id":"3dd1089378ddb906c922c64eb17e5c0c6984367d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 09:21:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/487cd5f67d8840e8a080213effa30c68\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d52d19f9922436dad4578978fb40f40 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c0fca5789684648a930457a0ae2affb : SUCCESS in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8ad127f4bca512c6ae0f70ead170b2f0215bd9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:05:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82742fb491fc4bbda3811bafcb1c0175\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa92e59099ba4c728e7b3dbb2405ba0e : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/311f5f115ab946669248e18f8dc920f4 : SUCCESS in 11m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cb5364155a946e98c0539dbe28ff5cc : SUCCESS in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/259e26d089fa4747bf3f60556d63c526 : FAILURE in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9897a4cac37440b6bf0d8eb957f067b3 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/94e5d504d3f74aecaed2e7349d431989 : SUCCESS in 59m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dacf8070c9d44cbea924f226e949d518 : SUCCESS in 1h 36m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f509258119ac4cf081a0ebec71652967 : SUCCESS in 1h 03m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/85eef0a1ea104f8595c59ff2c32b887a : SUCCESS in 16m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e774a3fdaf9c4c0b902ec49a8f51354f : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0164689b8f68412898d558112bfb23a1 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ff62edeb4ae4824ba8934fe85f27da2 : SUCCESS in 5m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c4b951805cbe4df6982ca990e61a2ae3 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb21baa1413f489da9c10cb3dafe2855 : FAILURE in 40m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a53e883fc8624baeba3cf8fb4ca8a1aa : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd65d6e68cd94f10b3c2773f44d7846a : SUCCESS in 1h 43m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/723f922f74694b3ab489b2ba67447638 : SUCCESS in 2h 11m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2ca4d9fd5fd437a92d913cd97041bff : FAILURE in 1h 45m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8024ddb40fd64715af8107a24c306aba : SUCCESS in 1h 54m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f78578ac69d942d3ab9472f5fabf63aa : SUCCESS in 38m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24822068f10f4ae99c573f68bb6942ca : SUCCESS in 1h 12m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b0341db3c6ff460a9c690972eb263ce5 : SUCCESS in 1h 31m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8a8c68f9e2ce493594dc9adaec17acd4 : FAILURE in 38m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89d2ab8a347e4ccaa4112427f28964e3 : SUCCESS in 1h 25m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/13b0ece99f0f45e2a22bfbe2f8afcf0d : SUCCESS in 1h 32m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b431e18f53aa4a78b835acb74e92a68e : SUCCESS in 1h 44m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c4069d9768c46cfb524d29bba781978 : SUCCESS in 49m 43s","accounts_in_message":[],"_revision_number":5},{"id":"74eb54901f9d1db0b9a6d85c8be8aa81c4efca8e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 13:57:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d29e3b54151926feea356b1bbf0976bb55c3b8e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 14:23:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c1f8e368204e729094a231246c9ed5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16000988d6dd491eb44a0ff454099539 : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9f5c95a6adf4853889f895f50c3867b : FAILURE in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"905b4643d3a22b68a38b557242a2cc534ac2230a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-20 16:41:37.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4427dde898a24061a26353f3c8cfb78d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/dc13c9db1cbe461d8ab51e53d4474951 : POST_FAILURE in 12m 17s","accounts_in_message":[],"_revision_number":5},{"id":"225e814bc36b3c639a9358a2797dd51fba341387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:49:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb01b235c03c466ba71ea5e764994f20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6309cca8547480a88bf69a7442d32fe : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26a7f6edb04d4d39a23d4382a0cf83ab : SUCCESS in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29b5777d541547ac866eb4f8540ca170 : SUCCESS in 10m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa274032b32c4580a25e50a739ce2609 : FAILURE in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c441e66532054aed9383c45a61073ea9 : SUCCESS in 13m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/c2e4b887568144a3b6607cff8c2161ab : SUCCESS in 57m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e27e8013e774e22afa1dc70a750e68b : FAILURE in 1h 18m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e1e420c0624c1daa15b32f43a850cb : SUCCESS in 1h 03m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/23d0a85f941e45e890115370e0e5a510 : SUCCESS in 16m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4448b4f289014470b35ac04a4c35dbe1 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b7a712e3f3c04a4894f2b966db060208 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4e703e3964d43d5968dd80b054ded13 : SUCCESS in 5m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6fb0c2b605264a0485323d2baf43e078 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7d8c1ea8a03b493ca29bca948af6a34a : FAILURE in 43m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/133a1cafef8c476eb4a99fc1ce9e9314 : FAILURE in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e95a8665b2f4b8dae78d3d3599d38b6 : SUCCESS in 2h 03m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/09b203af9ee74da9b867349083207b49 : SUCCESS in 2h 09m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3be781c9913b41fca7663a4ff414c330 : SUCCESS in 1h 40m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/50bd059047dc4fb1b351f4367786c91c : FAILURE in 2h 46m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d3378d832a23482c95e17c05cb3f9cc9 : SUCCESS in 27m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f1d012c20e3a49a09836b365cdf33290 : SUCCESS in 1h 05m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ff95a157a1984ba89e285c50c0081772 : SUCCESS in 1h 42m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18c13b6ff6e74c3fb754c9ac4c9104be : SUCCESS in 43m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/baa1c77545b94cfc89148da5f494c1c8 : SUCCESS in 1h 11m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cce8b1a7431145b2878374160a8d010a : SUCCESS in 1h 24m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/797f93345ed14c50b0c6eace7298881d : FAILURE in 1h 24m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3f12e20de4540eeba5110644d86083e : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":5},{"id":"cc3f081c667d0045412ecf90c3849c2fda7bc92d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-21 01:07:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d49d71b96691ee1f172bb1cba4d8db016424d774","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 01:19:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02eb03b486794ebfaed1bc0f17123871\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c4d890b408844c792c2a6a79e0c5f87 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35fd60daf7ef49bbaa43c2e4529d6f0e : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6073e7e7ee3217026404e50b875c54747a464119","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 01:31:53.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/584de36d2a2a4e089c4692f39aba4791\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/97f553a1468a43258ffd78083c31552a : POST_FAILURE in 17m 38s","accounts_in_message":[],"_revision_number":5},{"id":"428eb5bff8908b6d7a2bf61c0f5c1d517e245c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 03:30:17.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef006113fcee4a9c84bfbaf6e327290b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6254a71f8a4c3fa8a4450e57e744b8 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08f5725be0f64071b18c0f909fe8ccc2 : SUCCESS in 8m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be00df27889c4a4489e9dc2a18d93305 : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/266b42d1e0f74457be313842c210e9e0 : FAILURE in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a86898f00584715b3923c236b375c60 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b8354f5604724694afd5d3f5bfe5323b : SUCCESS in 50m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b47b0cb8eb1f40418e69c2d21bf3b2eb : SUCCESS in 1h 26m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8793b4e0dd3466fb274d5bf76d0b5ea : SUCCESS in 42m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ec03d3b749b4a85885e3a973cca8754 : SUCCESS in 16m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8c4165a790a0441c897fbae56e4b65f6 : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c266aae3b1e4fd38ea039a6c643c9de : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4849780fcb3f474183b268a823afcbf3 : SUCCESS in 5m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f5d9282cc44943ac9045892d85d6c9aa : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f3d54c8d31884702bb191d231965dde2 : FAILURE in 30m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ffbf24be91243afbdd73f4cdeba3df1 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/635617920bb34700a1cc013872fae823 : FAILURE in 1h 42m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2cbd149a83a04c9388daebd910605329 : SUCCESS in 1h 35m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e12cb215c174984abb19a9617410be7 : SUCCESS in 2h 14m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/952bbb8a1f5e4ed3bbc30075f7f78323 : SUCCESS in 1h 59m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/617f05e07f5544ebadda93072be6d15b : SUCCESS in 24m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5bf7d03197e4d639757278ce57a9ed1 : SUCCESS in 1h 16m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c89a1b320c6e4491865d247009787295 : SUCCESS in 1h 34m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f1b3b6e849f4ac6b3e41e750cc65458 : SUCCESS in 37m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/028370a6ea124283961235b02648ef0f : SUCCESS in 1h 06m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c78302fd52d4b92a18040bd4bd40d7b : SUCCESS in 1h 07m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40bd6ecda19c47b9b95744870fe82f09 : FAILURE in 1h 48m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/097ffc5e162e44b181ae8a0da6071c15 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":5},{"id":"bbc4eb0d946111c18f52f2bf2fdc579c6239e197","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-21 08:01:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4e889bf30c49f6f5470780a9f7042e900779b0dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:24:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38495a8ac7a04cb99660ad9ffbbec1ab\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5cdf8dc1c244834ba2bcc6ed4998a36 : SUCCESS in 17m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2846343e75da48568d6a639f933eee28 : SUCCESS in 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4aeb225e3cd3d7cc6866087dc97e51db4fa50cd7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 09:46:48.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a8c0acba6349405992a0878e596352ed\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b7d28d7e91d94da5a061941222511a1d : POST_FAILURE in 31m 31s","accounts_in_message":[],"_revision_number":5},{"id":"6c6065c985a06f728c47aaec2229adad04dc549f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 10:35:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fba1142f0f72465890a31a1a51f1aac4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ed1e43ab0e48209b8da4ce3a0603c6 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30564edf027a482c9a0b82b1af4fdb1d : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a34ac3ce9664d069ae5423f8766444c : SUCCESS in 11m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/566441efcd804eb19ba40f69a5f0eea4 : FAILURE in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fda90e1dca0b4f1b80e3ef119329df01 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/668897ee0e5348c88c03a89b7f4c8580 : SUCCESS in 1h 10m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec81c90bfbb84d5bb834e7f66f7f6086 : SUCCESS in 1h 27m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77f24feba37b44c9a97b2a64547af82c : SUCCESS in 1h 00m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d2b9ccac497d4b9fa84e84396b25c269 : SUCCESS in 16m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6cc9776a62e84827b6c753c0b2c1d4ef : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb37cec940844c79abfc3e4fbcc4affa : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/849cfa30dae0411eb55f4da6cb78c296 : SUCCESS in 6m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c12c618c929a4de6a2e5cabc6d8ff7f6 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0bd6c882037f43d3a31d6e900bc68f2a : FAILURE in 46m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dffa36b78c79459a896666cebb685abd : FAILURE in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/488f7155b0354c5d9b6d98485e465420 : SUCCESS in 2h 15m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5225830d0a7483a9848bbad85400479 : FAILURE in 2h 05m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78c57db13c5d454dab9efcc8fbafce40 : SUCCESS in 1h 43m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c721d9ebaae74407b8c85a3f86cd0424 : SUCCESS in 1h 43m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/68b28d4bd08f4c7c82e6094bc0f44ecb : SUCCESS in 35m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c3edd733a1548c1ad227222d4bc6cdb : SUCCESS in 1h 03m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/99ee25a690604144a629230ff65dd42c : SUCCESS in 1h 17m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e87ca4ab4a0345179eff08911870e569 : SUCCESS in 44m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9cdf51603548435fa3e63e47c65ded1d : SUCCESS in 1h 10m 24s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3c275310c54c4caf97d7b316adc76540 : SUCCESS in 1h 26m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/687d79bc88fe4230b0e3787262e0a734 : SUCCESS in 1h 24m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c0753c175ef490aab2a18dd27d36005 : SUCCESS in 46m 44s","accounts_in_message":[],"_revision_number":5},{"id":"95a3130179efe13dce1539a2786660e595c14567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 10:35:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"54eb76ca03751b982cbf5cfad2124a394acbb0ae","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-21 10:54:59.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/7493d1a/ : RETRY_LIMIT in 4m 32s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/a210968/ : RETRY_LIMIT in 4m 00s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/0726a82/ : RETRY_LIMIT in 4m 40s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7bef185/ : RETRY_LIMIT in 3m 20s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/3a0b63e/ : RETRY_LIMIT in 4m 30s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/00f542c/ : RETRY_LIMIT in 3m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"7291be3035e3f18118616407c11de30d622c39ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 12:17:19.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5adf69a8a98451e8e219e3f739dec80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc005feb6ca6421a9c78313b19cab7d8 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e07cef383d8e4ff59766d5cc3b7837d4 : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df7089bb95c446bcaf09d412903ce1a9 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a72135e7e0b45edab807e13d1733bfb : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c10c2df2cd394e469e2ca2f53b2be4ee : SUCCESS in 56m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b09d758ad1674b3d99b7493c71bca340 : SUCCESS in 1h 26m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/384d4cfbf075433884ead8363214b657 : SUCCESS in 49m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/764cd51fc8e44907a54bf009a8cffccc : SUCCESS in 53m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b7501577dfb546a3b5f2d2b6cb66c6b2 : FAILURE in 1h 40m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b0048f12e664c59a5869258d5899f2e : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":5},{"id":"b0648fa4ae3b137f5a32934a4e7dbe5b1417a620","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-21 15:27:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"819d27e0a72f9e1ae09a6a05ba7f316b8fd5ca78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 15:45:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fd679e0663b43d6a1dfc20aca77188b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1ba82b47e4d4685ab79fc982c9bed0d : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/03f16c4761fe4c7f81983afbe74fa3bf : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bb390954ba2fdf115a9a6a2cbcd91fd135ba53a7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-21 15:57:36.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/8dbd5f8b33104cb4a3a7ee09644ba348\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0781b568e6754fbdb1bc25c877c4f670 : POST_FAILURE in 20m 57s","accounts_in_message":[],"_revision_number":5},{"id":"db517c940778e9b5f59927d0c8b43e703d08fc80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 17:51:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e77a823ba8248f494e603f9a8a9b0d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9752bd20120c466da1d147a0079549e0 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5d3f105fb0944138209285bde565475 : SUCCESS in 11m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f6b1b7f965f4aa2a106f09c8afe39c7 : POST_FAILURE in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e2b602ab4d045b5848f4c4475fa7030 : FAILURE in 17m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e5822a67b9487f9d8b9f2a4a213eac : SUCCESS in 14m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/c5a50243b72e454d952e04f0d51891c1 : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/422c296d6b504cc69f158f3a32b4bd35 : SUCCESS in 1h 42m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03887b5843a041b5896bfe296627946b : SUCCESS in 50m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/168070662aff40beb6842bbfb1b13fea : SUCCESS in 15m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da412fef620645a39e58640a2bc26c6e : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e08372831b434cb0ae4fe68c66ea254a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4e8086127b249afb3536e4caa286928 : SUCCESS in 5m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0e361b13bbdc4ae9a7c4fe28e67fba9f : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99cb651eb90d4a7cb286f11892457bd7 : FAILURE in 42m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20af4bb6634c48c68baf9c1d16c74032 : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c0e0251f1634c73870537a910589564 : SUCCESS in 1h 56m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b2c9b758b67c4e45bc13af1435b1ff95 : FAILURE in 1h 52m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/076d730c3f6244f6bc41d6bc144657d8 : SUCCESS in 1h 50m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/993f24861e184bddb8b23f0fe2f87cb8 : SUCCESS in 2h 22m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/25d3590ad0424fbda18e40657c4252bc : SUCCESS in 45m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad2c91c3a34948aba4b00475331a7d04 : SUCCESS in 1h 11m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92ede146968f476dac20c0594b085a80 : SUCCESS in 1h 40m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1cc6918c660342ffaf06416e81c6695c : SUCCESS in 42m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/73ff676529b44ca186071d28da2952e7 : SUCCESS in 1h 31m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7e3c0f45d0324bbd929152ef6f70d8cf : POST_FAILURE in 1h 52m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/103ec559968a4e7ca212186ff8df5fb5 : SUCCESS in 1h 47m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8798b80c1b64fb49dbf94070d693f93 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":5},{"id":"2bbaecc637167b40cbe7879ccbe2c30804c01096","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-22 01:02:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"351f38560b3195f9532a364d41457f2c2740699f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 01:16:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae72540f7688486094582a1c05f2800a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1af1ccd031d14d18a184527d86c11ffa : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8980c37d550142c2937c0652198b300c : FAILURE in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d4e43c4f055947cd1cb2bb152a11f4f3dc7b8fd9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 01:29:40.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/10505d4e09cf456a8c53d0ba6ff55e88\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/38d11c5ca2d34f26892df460b030be05 : POST_FAILURE in 23m 34s","accounts_in_message":[],"_revision_number":5},{"id":"9505bfcfc0e120aa3a795d6cad6db0f054f768dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 03:15:33.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa2993e2aad4488cac96b221276861bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749056a395274cbd80156785a2b402ba : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1720d31a5ce340af9a96258139f94c21 : SUCCESS in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd2b450e502646369e4580302969f572 : SUCCESS in 10m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4a6584cdc934f539df0b50446ad70ab : FAILURE in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0af8e23d7a949f4b14f1688c224fc19 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/16e86b353189418aaddb33da96763471 : SUCCESS in 47m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9c1fbf272b445598e949c238ad442ae : FAILURE in 1h 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/962cc7c397e94aa898c627c06906bbc9 : SUCCESS in 52m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce5028e905284d05a6ab5a940fa5bea8 : SUCCESS in 15m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/03c9cee3edab4ecb86158da45c1d9a18 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c454cc600b184bdb9d2a0ac09689a8e5 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46917d5d783d49838cb8c06a8bb233b3 : SUCCESS in 6m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/270ce451a91f40208924c1df1568b792 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd9fd16608e9449a84b06f6641df38ac : FAILURE in 45m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b401716b0c344f0995144218e3523ef : FAILURE in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ee151f4976140a3be38b0089ec3e607 : SUCCESS in 1h 47m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ba206b0a4e6b41fb93372af7c66d719e : FAILURE in 1h 39m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/24d373e33af346cab6af4e1a414c50c8 : FAILURE in 1h 27m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ac4c74b943b241a7a4faaca20f3c27d1 : SUCCESS in 1h 50m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/215ec7d884ec42f1a47b4ecf58512c51 : SUCCESS in 25m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/446321f1063540bdadac72060de70631 : SUCCESS in 1h 09m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9c4f9204b79c454c8402e7a3fc7b1799 : SUCCESS in 1h 28m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/372fbc5e337a41379d95a3bba07bd05d : SUCCESS in 47m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bedf96f478b745a3a899837526288478 : SUCCESS in 1h 33m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/58daa942baa2436cb428c07302448f45 : SUCCESS in 1h 23m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e66db341a664948a96dda4011b5a97a : FAILURE in 1h 32m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69fae181e4df48bc873147d3788809b5 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2abe463495b1a61a836f30257521ed2e26942ffd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-22 03:51:39.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_3.6/d4698f1/DellEMC_PowerFlex_3.6/1415 : SUCCESS in 1h 47m 50s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.5/b192832/DellEMC_PowerFlex_4.5/331 : SUCCESS in 1h 45m 00s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/b41d00c/DellEMC_PowerFlex_4.x_NVMeTCP/272 : FAILURE in 2h 55m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"941c954de8647269e8b9feec46b9325e41a5e211","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-22 04:33:31.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/046c110/DellEMC_PowerStore_iSCSI/1364 : SUCCESS in 1h 23m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/b6c3a18/DellEMC_PowerStore_FC/1547 : SUCCESS in 59m 43s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/de45a37/DellEMC_PowerStore_NFS/1058 : SUCCESS in 1h 12m 02s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/a8d72da/DellEMC_PowerStore_NVMeTCP/912 : SUCCESS in 1h 43m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"059cdb8278d938be4cec0e8d47a86aedf4aa38c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-22 09:16:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"35862fa0de82feae9d9b928055605f4ff51689e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 09:28:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/586db9f4878847a598bd335cae190e8d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93e5c13ab931470ab07173376cde775c : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6881e8ab9324672b886a4b308c519d0 : SUCCESS in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2fdb6832eebcd470c8687ec45f942514fe6c88a7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 10:13:04.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c28fa113018f42a6beee5d3dd68fde21\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0c88fade664f4c759210d4e37e426cb9 : POST_FAILURE in 20m 54s","accounts_in_message":[],"_revision_number":5},{"id":"fb15aa95ac3c7858fcab70b639fbe88f81014b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 11:42:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/177bfcaf262740f3a6de4c6a7f6907ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042d5f8513f846b297ea34e6d3c69a47 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6684b751f7b64e168533a4af43c2f9fb : SUCCESS in 8m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fe491348caf499baf3c724cf0c58d90 : SUCCESS in 12m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16d0cf0b5d0048bea2b23bf153f81688 : FAILURE in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79649d430b64d219fbf699ea962bb19 : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/bf916fa54676464a9ea2e613c3400d95 : SUCCESS in 1h 09m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7b5e143868f4c95a3b8aaf8e3d8b91b : SUCCESS in 1h 23m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8e2cbfa761e4ab69ba61fd8ebca8f8a : SUCCESS in 58m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d104665fa3744655a86338faac583311 : SUCCESS in 14m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e9364e2f5bb43b79c28f0d8a47f006e : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93e35057f44f44e397248bf8f21add31 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/801c1f6f681b4b0ab76bc3fbfafe32f3 : SUCCESS in 6m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e554122addcb4252bfdce367b860b9db : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb5eae554c874369b23bce1913767c91 : FAILURE in 31m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4ae8a14abe34676acdba0f4a4cc6b02 : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11cc2f8af6984206a5423c1e0a7414c1 : SUCCESS in 1h 49m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e48dda03ac647479e7eae88ec0eaf27 : FAILURE in 2h 11m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c8388c21c82e4f31818c4d6dc7e4ce65 : FAILURE in 1h 46m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f7711d99226540f7a73211eda24df164 : SUCCESS in 1h 36m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/95e6fcf1b40c4249897b3b5e3d34dc0e : SUCCESS in 36m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d15b5f4bf1a745b5a69733a2c39f72cd : SUCCESS in 1h 16m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a7289425386140a196ee56adf4784c87 : SUCCESS in 1h 27m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bde60a754a7c446cb4f8018e33df4ba9 : SUCCESS in 45m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5b768f6c54f44b26a5dfbd8842d2d365 : SUCCESS in 1h 21m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bc6c7f4948af451f8356162f190017fb : SUCCESS in 1h 00m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90435736561a4757a0e73eb2a7d03d7a : SUCCESS in 1h 21m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4db7b94e4636410c9f9841cb1ba8bb8c : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":5},{"id":"066f52a2edf5567a56386b83431d5fd501e5a9d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-22 14:38:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43773673c9dbc6cdea592c31f10f4478cb4d629b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:59:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e22aafd2625447498f3516cf00f3308d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e11c3f239e2b460387b7b6844730267b : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97d043924b8d40139c5f795bcfe0df84 : SUCCESS in 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ec806b98adc754fe07ed4a3e4627f7dcd871ddd7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-22 15:01:50.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/87c3dac4da914edb9b14c3dc6504e8b3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/20ce0dbfc7214e1d81429e8687f2eff9 : POST_FAILURE in 20m 09s","accounts_in_message":[],"_revision_number":5},{"id":"34ded190caed57c89039fb6b0b1f9e518b609208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 17:19:57.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a954dc90ad2347618f1efbe85967c68b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0045450371c843f6925180eaddebea84 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b1ccc81014a4af8be9b8fa47f5e58ee : SUCCESS in 11m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf03489abb3a44d593381d14c2999b07 : SUCCESS in 10m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea849717e5544bdeaf5cda8dd8a69702 : FAILURE in 22m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50e0082b2a4449daef1044c018c976f : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/51dae23dbb7c433bbb8720c6f97e4691 : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/690738f9023d41c3b48b379fe2d379e4 : SUCCESS in 1h 33m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/912caaee78304414a9252b3717114a2c : SUCCESS in 1h 02m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ebc3bea5d8eb483992c62463567421bb : SUCCESS in 17m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f133417aa224f5199d53b09f50f53df : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1966a6b23f1445deb3d268e58342adce : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4047413906c249949fcf5881328ace61 : SUCCESS in 7m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1b170d614f21406eaac40c945766ad0e : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/572acf98338a40ba8e16baa074afe71a : FAILURE in 31m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff06563e00104a619a8fc9936d5e6f39 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa4eadb5927c4f7bb489952058dd946d : SUCCESS in 1h 59m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e2769c4f461c4b07b058c2febdb8b95e : FAILURE in 2h 39m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/500695132fff48448b03477f5df5a61f : FAILURE in 1h 47m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/96dccba375f44c2aacceaed6012732af : SUCCESS in 2h 12m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/07bb26c3e1144c1691626e4872ef15c0 : SUCCESS in 25m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/41c24cfabeba4202835d59561ab8901b : SUCCESS in 1h 07m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ddff03df5d5a4b8dad554e365483bb7a : SUCCESS in 1h 20m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7e30d1f271be4a1eb355bbb18436c89a : SUCCESS in 45m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/500c028ebb3b4a6dafd1f5ffd23ae599 : SUCCESS in 1h 26m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6f6e3852eec54afbbf8deec4aa20ec00 : SUCCESS in 1h 30m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b31cd12f8c24efd819bab317ad6f964 : SUCCESS in 1h 48m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60572454d62d4ce287fa7069f8b5edc9 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":5},{"id":"6ad5ac6590e925fb63cfbc7ce7f6eea4c464ab1b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-23 01:05:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"89d489b170984dd7addfa3a70a7212520f4d7322","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 01:17:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae410beecc1c4683b80619c1034347e7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d21c807c8db042968f37638f43c531c9 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c799f1b861b48b9bc1799b9aa9e8dad : SUCCESS in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ea2ef8326d9ae73a1e60ab203fd3fb282a16c476","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 01:19:21.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5fa9bd2a0d9245ca81b8d7f7143a01af\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a8b5d7c179cf417190bf482ca7b530a1 : POST_FAILURE in 10m 58s","accounts_in_message":[],"_revision_number":5},{"id":"5658c3364967d44f90b54e6f25509e797940414a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-23 02:16:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec18a6096d1aeb37fd67d7902e8bfe4f5b3e669e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 02:32:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3c700009f5f4e1bab31b9bb8f987bce\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/936665e10da443bb840d978575a307bb : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b97287974bf4ee9a7a887d1b5b43ff4 : SUCCESS in 13m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c01ca95abc5ac29f944ba971df4502040e735508","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 02:45:05.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c98906bc5b704d87aab26f0d935387b6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bd3d967ac0ce472a8108d9a4c502b931 : POST_FAILURE in 21m 29s","accounts_in_message":[],"_revision_number":5},{"id":"44ccd49110ceffb98ed1cb88e66284cd2d4655e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 02:54:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a76540694774793a9b4e71629000789\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b347d30c4e248eb8666eb3e6044fcbb : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f886c13dd2a549d8af426618c2807d8b : SUCCESS in 11m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c97548eb0ca249de98f616b26b04c9ba : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c957d55a205d40bb9e80462be4822212 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd34f930a45349c285783bab0eaae583 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/9385795d25254e728c944b3a5757258b : SUCCESS in 1h 01m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/603a849721b54b0ba0588a5bbb50fdf5 : SUCCESS in 1h 23m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd6305f4ee9a4431892ed30a409280c3 : SUCCESS in 49m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8d6aee8a4e64d7f8531ebc75863eaff : SUCCESS in 15m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/db14ba60831f4864a759b50eda4dbbbc : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d3366e806de6491fb1c4d7eee442675f : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/afbd44227830486693df36a078cecffb : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bbe9e416042c4507b2ca718b3fa53567 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13ad7088527a407e9f1b0592f4ccd88d : FAILURE in 28m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b877c4fe6d9945379265525250032330 : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1c77a86bf4964463ab84ed6dc70a28e0 : SUCCESS in 1h 36m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d882e499bb164f8a85c0bb27ff549d96 : SUCCESS in 1h 40m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80109a8a7c0244da9b32f4b6734987de : SUCCESS in 1h 26m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f7d46c21728b4afe94c63199f98f8dee : SUCCESS in 1h 33m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/84d73588b1884d4c89d921ab79ea3eeb : SUCCESS in 25m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c42cd6cb1a68411c950540a2265dc10b : SUCCESS in 51m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b7fc59a991b47e4b9dede0fe9912dab : SUCCESS in 1h 39m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a6357da7d897495181c4e418d1c893dd : SUCCESS in 31m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8c41f66b8f5645b79a73ac24b3b7b5b1 : SUCCESS in 1h 25m 37s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6ee17458ca0241089acc2e607dce15be : SUCCESS in 1h 31m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9910642bfe534446af320938a303a0d4 : SUCCESS in 1h 42m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0a627594cc1452593868f4aaffc8c87 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":5},{"id":"75a62d000e466f38f36175a48001ec4ad55f6ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 02:54:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a03148512c4a4ed933e19fadd1ffdb2b629166fe","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-23 03:14:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/de59b8c/ : RETRY_LIMIT in 4m 09s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/a6f0006/ : RETRY_LIMIT in 4m 11s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/9a00c37/ : RETRY_LIMIT in 4m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4c718a6/ : RETRY_LIMIT in 3m 24s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/dcd1e9e/ : RETRY_LIMIT in 3m 48s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/80149eb/ : RETRY_LIMIT in 3m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"230d9cf6a8e46d11a92dfeb8371fbeffcbf45749","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-23 04:35:50.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/52edb25/DellEMC_PowerStore_iSCSI/1374 : SUCCESS in 1h 17m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/a455e34/DellEMC_PowerStore_FC/1557 : SUCCESS in 53m 28s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/93af3ff/DellEMC_PowerStore_NFS/1068 : SUCCESS in 1h 04m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/6bb39e3/DellEMC_PowerStore_NVMeTCP/922 : SUCCESS in 1h 41m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"3ac47c0e594d783d9055b07a0cc28c93082db478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 04:57:53.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51738ebea3f5460a81a60759d654e9de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28d431cf57d4799b90f93b286cab553 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0bc89b00304b8eaa16d502432de785 : SUCCESS in 11m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/867961b5eb2449158a64ed6585bfbc86 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe90a50b5904a2aae6dc7e89ef577b2 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/1e71144e193e4e94ab004a069db7f1ff : SUCCESS in 1h 00m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1444b9d1ddd4e96b33d52d1394fec5b : FAILURE in 1h 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b1a197c0c2443e3b7c0e2563848d55a : SUCCESS in 44m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c90f7c283ea94c62bb8505d6dad1343e : SUCCESS in 1h 09m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4afd33e7c3a74a429901f51b50a7bfa3 : SUCCESS in 1h 58m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8747152a52b54e6688eb714834055d95 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":5},{"id":"e99bdac2567181d7e64e46bb77c475ae52d0e681","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-23 14:21:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bd1bc8e3a946135323d9e567be4589acf791e3ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 14:37:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67657b950d0347c6b466a5b9301c82a3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd44a9ef91154fd7a7a1f421d9219a4f : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44a003fc318c427cb749a3e616946950 : SUCCESS in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bf868394497ddd3b2796b9692f6bf6ad853a7d47","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 15:03:53.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/75e188017f2045fb9d35149e226a4e82\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/47af73f2205d49d0b7a2903773a6340f : POST_FAILURE in 20m 00s","accounts_in_message":[],"_revision_number":5},{"id":"ef5598330c71f28361aa0891f7e1d80a7b6e79a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 16:43:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b448c461bd7e4c3bbd08c945f0f9a62d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/286dd740554841bbb2620247932ee58a : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d6ab8de1c814750a33e9ee47aea7cef : SUCCESS in 11m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2cb59a1f833947afb91bffb8613fe89e : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3fce954b7fe4a2d82e6cfe3d53b17a2 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17eec21526d842af97658972b8a02986 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/738690108d8941468d4fb3b24119069d : SUCCESS in 53m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/481e37c3ce4d450596c0088b492f452e : SUCCESS in 1h 21m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20269f0f153d45aebfa45a81877f3713 : SUCCESS in 47m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a7c0f0f6beb544cfba596cbc21a59c0a : SUCCESS in 15m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d142d0d3a304f718ac20234fe3041d0 : SUCCESS in 7m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/68e3d4c86cff4fba9dfaad890625f2a8 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aaf5135e30b84e1989e4a16b8e720ec6 : SUCCESS in 6m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/270dbf2858894cb2ac559c0695fb8261 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/525690153c2c41a5a069b0a845057155 : FAILURE in 48m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db8d46205a474197924b281c3989f90d : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f927e437190f475286d0049f8c53adbc : SUCCESS in 1h 42m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b7e2ffcb87f34c4d90c3f21a60b1ed8f : SUCCESS in 1h 58m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9984362661e4de197a36084dc4afbfd : SUCCESS in 1h 46m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e8d06098f4874c85baef9316175713f1 : SUCCESS in 1h 19m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b3302390be084772bef72cfa81bee6a9 : SUCCESS in 33m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8fe814705ade4309a2f2270fec1f7040 : SUCCESS in 1h 17m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0adf14d486344a7a935f62b0e860ffa : SUCCESS in 1h 47m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e4b4a615cb154d6db9ef3996e0e38934 : SUCCESS in 44m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d91d326056f8491382231e5fc104819b : SUCCESS in 1h 15m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e193ee5de3cf4fbf94c66fbcd0fcc55b : POST_FAILURE in 2h 09m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fba780ff80943649d5d94370d438807 : SUCCESS in 1h 33m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/270525ee1cd545ea998d1d0cf951bfec : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":5},{"id":"48ad0e1fd7de44ae523639ad69b2cba122220bdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 16:45:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e0771fb879939aff58affa5a92dc1f54710af8bd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-23 17:25:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-fcp/dbb9099/ : RETRY_LIMIT in 4m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/68ad002/ : RETRY_LIMIT in 4m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs/4464c5d/ : RETRY_LIMIT in 4m 14s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3baa9d7/ : RETRY_LIMIT in 3m 30s (non-voting)\n- cinder-tempest-plugin-ontap-nfs-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-nfs-rest/fc1c118/ : RETRY_LIMIT in 3m 28s\n- cinder-tempest-plugin-ontap-iscsi-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/91/907391/5/upstream-check/cinder-tempest-plugin-ontap-iscsi-rest/282c51c/ : RETRY_LIMIT in 4m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"434a1ece022f772647b610260a87f13a472a439b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-23 17:56:20.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_3.6/a16fe03/DellEMC_PowerFlex_3.6/1430 : SUCCESS in 1h 58m 25s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.5/ae6d678/DellEMC_PowerFlex_4.5/344 : FAILURE in 2h 05m 08s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerFlex_4.x_NVMeTCP/253e13d/DellEMC_PowerFlex_4.x_NVMeTCP/285 : FAILURE in 2h 55m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"094c6e127489ccf8d3a414287813ad20b220c272","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-23 18:37:35.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_iSCSI/750761a/DellEMC_PowerStore_iSCSI/1377 : SUCCESS in 1h 10m 15s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_FC/cb416a6/DellEMC_PowerStore_FC/1560 : SUCCESS in 55m 50s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NFS/06ffde4/DellEMC_PowerStore_NFS/1071 : SUCCESS in 1h 10m 27s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_PowerStore_NVMeTCP/c8b5e9f/DellEMC_PowerStore_NVMeTCP/925 : SUCCESS in 1h 53m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"a6031a1cac9891752922b816d8373d5300c4186d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:48:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e67b70296b24fe3ba1d2afc434ed316\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e441d194f6cc48a7b1c319449bddb97a : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5e514f2da3a479183c778d85736d35e : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2ec3f7275424598a1222a15bbaabfc7 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491467ea19c342acb321ab01cf46cce0 : SUCCESS in 13m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/137695459c384ef88742bf3869a85ee7 : SUCCESS in 1h 02m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b603e2c2898d41c68be5c8d888028bcb : SUCCESS in 1h 24m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ca734fdb3464558a59d50cb8a36f397 : SUCCESS in 1h 02m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a045acbe5b134d3e9204dac6cc0bd2cb : SUCCESS in 1h 15m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52a364838fb94e40a1debbe940251987 : SUCCESS in 1h 42m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68384845feff4c069c1c025bf16e3c22 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":5},{"id":"4f0ced5c381c834d5e8d96f936d129c5945afe23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:48:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"aef77be409f0021d7687e5e6b65245e7560527cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:50:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/251a8a383efa48e1b59975e4dd84dcaf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3246745bf7ef413b821ac0e4f8d70fc8 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5},{"id":"0e741f333cf37ccee350ce91da740224ff5d0fe0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-24 08:14:36.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_FC/25615ba/DellEMC_Unity_FC/1942 : FAILURE in 1h 49m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/907391/5/check/DellEMC_Unity_iSCSI/386271b/DellEMC_Unity_iSCSI/1970 : FAILURE in 2h 30m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"155bfcf98df7a72addfeb0cbd4f92c8aed44a770","revisions":{"4a3e3cea99bd2a7630a4776720d7f30db9e1082b":{"kind":"REWORK","_number":1,"created":"2024-02-01 07:50:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/907391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/907391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/907391/1"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:43:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:49:26.000000000","tz":540},"subject":"Remove six from netapp drivers (1/2)","message":"Remove six from netapp drivers (1/2)\n\nThis is the first step to remove six from netapp drivers and removes\nall the usages in implementations. The remaining usages in test codes\nwill be replaced by the subsequent change.\n\nChange-Id: I372aa42b868ebf356948737d714a3e1bea3185ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a3e3cea99bd2a7630a4776720d7f30db9e1082b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a3e3cea99bd2a7630a4776720d7f30db9e1082b"}]},"branch":"refs/heads/master"},"81c9214c12754c6f251bfb7367c9013eabef07fe":{"kind":"REWORK","_number":2,"created":"2024-02-01 07:57:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/907391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/907391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/907391/2"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:43:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:55:42.000000000","tz":540},"subject":"Remove six from netapp drivers","message":"Remove six from netapp drivers\n\nChange-Id: I372aa42b868ebf356948737d714a3e1bea3185ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c9214c12754c6f251bfb7367c9013eabef07fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c9214c12754c6f251bfb7367c9013eabef07fe"}]},"branch":"refs/heads/master"},"163083249505f0cdc89624463698a819ab7c39b4":{"kind":"REWORK","_number":3,"created":"2024-02-05 01:23:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/907391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/907391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/907391/3"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:43:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 01:23:14.000000000","tz":540},"subject":"Remove six from netapp drivers","message":"Remove six from netapp drivers\n\nChange-Id: I372aa42b868ebf356948737d714a3e1bea3185ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/163083249505f0cdc89624463698a819ab7c39b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/163083249505f0cdc89624463698a819ab7c39b4"}]},"branch":"refs/heads/master"},"1088866d7e38a2a1c8619cbc792ebe833eb0d051":{"kind":"REWORK","_number":4,"created":"2024-02-05 01:31:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/907391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/907391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/907391/4"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:43:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 01:30:58.000000000","tz":540},"subject":"Remove six from netapp drivers","message":"Remove six from netapp drivers\n\nChange-Id: I372aa42b868ebf356948737d714a3e1bea3185ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1088866d7e38a2a1c8619cbc792ebe833eb0d051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1088866d7e38a2a1c8619cbc792ebe833eb0d051"}]},"branch":"refs/heads/master"},"155bfcf98df7a72addfeb0cbd4f92c8aed44a770":{"kind":"REWORK","_number":5,"created":"2024-02-05 02:16:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/907391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/907391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/907391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/907391/5"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 07:43:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-05 02:15:48.000000000","tz":540},"subject":"Remove six from netapp drivers","message":"Remove six from netapp drivers\n\nChange-Id: I372aa42b868ebf356948737d714a3e1bea3185ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/155bfcf98df7a72addfeb0cbd4f92c8aed44a770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/155bfcf98df7a72addfeb0cbd4f92c8aed44a770"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
