)]}'
{"id":"openstack%2Fcinder~907420","triplet_id":"openstack%2Fcinder~master~I3290fd411da511f276d4f592820e9a88df958472","project":"openstack/cinder","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-05 18:50:57.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-05 23:10:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3290fd411da511f276d4f592820e9a88df958472","subject":"Remove six from Nexenta drivers","status":"MERGED","created":"2024-02-01 12:17:40.000000000","updated":"2024-03-05 23:12:17.000000000","submitted":"2024-03-05 23:10:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907420-remove-six","meta_rev_id":"36bce551cccf09f0d0429c9e0128b8f8d1efaa89","_number":907420,"virtual_id_number":907420,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-05 23:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-05 18:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-02-12 19:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-03-05 21:07:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 12:17:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 12:29:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 14:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 15:26:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-01 19:58:51.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 20:29:47.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-01 22:06:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-01 23:52:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-02 00:11:49.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-02 10:19:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-02 15:30:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-03 04:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-03 21:08:57.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-04 09:06:48.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-04 09:16:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-04 21:45:46.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2024-02-06 02:13:23.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2024-02-06 06:51:18.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-02-11 06:57:18.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-12 19:39:39.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-02-14 13:54:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c8a9915e68673269036c8105bed4a0f8a0ca81bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-01 12:17:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da6d0467781ceca323a95f74c79ad7faf0acf4ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 12:17:51.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16445/","accounts_in_message":[],"_revision_number":1},{"id":"dfcd8510171bbfd6128c41ab3a47fea375825be8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-01 12:17:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16445/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"34b58cc8bab2f37bea891351e337b96e5eebaf62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 12:29:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b0fef6bdaa048bca2928db6afe9a59d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/448fab878aac490a8120ac9f37e7f33f : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3cf833ef3c134d03b1a3a4463b614e82 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c0fec2de425241a2b2837b71049dbbd1 : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b194e2f611824493de70c43652ed99c7e9b6cb34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 14:39:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f587fa6053c04c97bf90120b80babe27\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c887cae56145458a6be3ac17839cb1 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae616bde34a74fe1b72755e36b3fd644 : SUCCESS in 13m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8666dd3934404f9c83bf2baa48361475 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b33e5e378b486486c729a6b409044d : SUCCESS in 15m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/bfe74f9b294c47b6aaadfd75242f5280 : SUCCESS in 1h 03m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6510160bec5b4dbbb964bdca2f2fba0e : SUCCESS in 45m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba8244497487441bbbd016943872b944 : SUCCESS in 1h 26m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10f09017ab76485d8cab2d771bf154b3 : SUCCESS in 21m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/367d10f96d3144d6ab592b056a29b60a : SUCCESS in 16m 23s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ce3452fa139a40908c37b6e47ec83235 : SUCCESS in 9m 20s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0e6570c57d8a4681baa6001f3071f619 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5555d52eef8424787aa69bc7f2c63f3 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc9353b71729499f94a9ee5f6b9c54f4 : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1e20904a17b4e1d956405241400fd74 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1592bbe197ab44dfa032636629f401f4 : SUCCESS in 44m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f27cfa41bd942dca003a659a1956bd7 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ae4162f0222452c8f8409855b216040 : SUCCESS in 1h 43m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7700210cd7af4a8c8d60fa66cdf2984a : FAILURE in 2h 16m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00a73a7575ac4638865af0b1001f7e17 : SUCCESS in 1h 50m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/750f3f21f4b54c8cab2dc7517311976d : FAILURE in 1h 30m 27s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63c1f77d8ba747909207429c7a369b0e : SUCCESS in 1h 16m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/870fe1564d114d25b3c725e9cf67f90c : SUCCESS in 1h 06m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dad3806a035345ef9a19190c7034fd46 : SUCCESS in 43m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef84e144c04340b0a0ddb0af6e73cc99 : SUCCESS in 1h 23m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3056acdc732148c988a2db229421e6d4 : SUCCESS in 1h 21m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/347095bbf5bf42ac9be7c4f74217fdd5 : SUCCESS in 1h 15m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd3e7463694146cba759f74daa72eb5f : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":1},{"id":"71891acd2ac6db376baade296882517faa61e729","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-01 15:26:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/1/upstream-check/cinder-tempest-plugin-ontap-fcp/78cf7ea/ : RETRY_LIMIT in 11m 24s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/11cac88/ : RETRY_LIMIT in 9m 18s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/1/upstream-check/cinder-tempest-plugin-ontap-nfs/dbc2d73/ : RETRY_LIMIT in 9m 34s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/81fe02c/ : RETRY_LIMIT in 8m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b66577be12e9350119e77b2fa52ec2d7e1817111","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-01 19:58:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/20/907420/1/check/fujitsu-eternus-dx-iscsi/c3c06e7 : FAILURE in 2h 20m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/20/907420/1/check/fujitsu-eternus-dx-fc/422c028 : FAILURE in 3h 52m 17s","accounts_in_message":[],"_revision_number":1},{"id":"79c0472d9cdeec9054976a28e1ec580c1691b5cd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-01 20:29:47.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/bcaacd92ddac49dbb1a5ddf253533243\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f6ce7f08e1c64e3a8bd1d3dc2675d12b : FAILURE in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f903d9b35893f8ededf73bb785a667af1dc84e14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-01 22:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907420 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a598ff421d99865db8732a1cd2e92c84adc347c5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-01 23:52:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_Unity_FC/ca1cbe1/DellEMC_Unity_FC/1192 : SUCCESS in 1h 04m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_Unity_iSCSI/be84708/DellEMC_Unity_iSCSI/1208 : FAILURE in 14m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9219dea6f93ece1fbf7bb41f6f0db30a75c52acc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-02 00:11:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9f541714e32c432ab722884e11808d64 : FAILURE in 1h 11m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a499923a06334c308b8327fc4412a07d : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"00f893c6549b5ff86f867df75559724f22a14a6a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-02 10:19:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_PowerStore_iSCSI/9b6a081/DellEMC_PowerStore_iSCSI/1006 : SUCCESS in 1h 15m 28s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_PowerStore_FC/a0ca1e8/DellEMC_PowerStore_FC/1162 : SUCCESS in 1h 28m 00s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_PowerStore_NFS/0f69f74/DellEMC_PowerStore_NFS/692 : SUCCESS in 1h 03m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_PowerStore_NVMeTCP/0473d15/DellEMC_PowerStore_NVMeTCP/559 : SUCCESS in 1h 54m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"cc582dfddcdc5d2300116f19913a34efb46b05b0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-02 15:30:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_PowerFlex_3.6/2f690a0/DellEMC_PowerFlex_3.6/1341 : SUCCESS in 1h 34m 11s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/907420/1/check/DellEMC_PowerFlex_4.5/4f2262d/DellEMC_PowerFlex_4.5/309 : SUCCESS in 1h 45m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"359510754e1971c89df1356ae8be102cf5272700","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-03 04:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/907420/1/fc : SUCCESS in 0h 53m 18s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/20/907420/1/iscsi : SUCCESS in 0h 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"675a5768b40926636f9db53e7774094b4d772724","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-03 21:08:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dd26a7b5369f885e368ab456b2b85663014a6e8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 06:01:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"aebf510e44d9dd1b8fd16c0c1b48d2dba1bb81c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-04 06:02:05.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16467/","accounts_in_message":[],"_revision_number":2},{"id":"0d270ea4d84fd166372aadd7442cdb83b3583391","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 06:02:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b528b13b6317b2cef1063db17f3557972a47469","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-04 06:02:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16467/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a8d730f843e02480ad48caae841a66261410cd72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 06:14:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8341648644c946eabf1163557481959b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b402c6829da74bea99a5e964761a2117 : FAILURE in 8m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c49c13e912c148dcaf832482a7758403 : FAILURE in 7m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c053083169554b8a973109118cca64e4 : FAILURE in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b598f0ff63dba2bb56fa767cea986a82058c234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 06:33:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"13042e7f7dec39bccb4c599df3b99372fc6db427","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-04 06:33:15.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16470/","accounts_in_message":[],"_revision_number":3},{"id":"a62c18cf211de3e9f37e8a8f17b1a3887d3d9362","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-04 06:33:18.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16470/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"73e8d04321b30e9e6a295d9f775d23a9cb5aa23e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 06:52:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a4da86aeea4b628819bc8aafe541da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/100cc4732de84d078614deb146e89174 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2dd71d5f10a43aa9a005259f375aa03 : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73e19220d1b442bb9c922e7447afd278 : SUCCESS in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"235cd6120875dec0a4884743954ed33743d01ef6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-04 07:45:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/907420 : FAILURE in 30m 21s","accounts_in_message":[],"_revision_number":3},{"id":"9bae963a173aa5c9e5e17959c7dd1505ddef4de1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 08:34:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd6dbb894bd8466f898c20abf8c57fb6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a0bc2fb96842989f2552906f59e67e : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a740f5cd825e4c069b0f898ae13adf29 : TIMED_OUT in 40m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b96091d439a042d3995d254529772665 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/021b0cc798504bf2b2f277ab45799c1d : SUCCESS in 12m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/990a0f0e62ba4156b807e499b7122c1a : SUCCESS in 1h 03m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4421a97e703a45bab624cf3a6c28d2ef : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7f6d9e8415742af8c0cbbc7c377598a : SUCCESS in 1h 28m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1af3ed0767924f568afcfbe2e4085e4b : SUCCESS in 30m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/363590a8824c4e22acfe8fef4b7ef620 : SUCCESS in 15m 42s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5ef11223ab724ca3a63e91287dac05a7 : SUCCESS in 12m 46s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/06afc07b342744aba4f33e75d3e72ae0 : SUCCESS in 6m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a1801efdd7c4b799cc180642f49dd97 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7764d9159f2b4ecb8a1cd05b82f8e784 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d6e1a1c81204ed68fe9438d704e7d03 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/368c2b17c2f1445191196cf2a6c03009 : SUCCESS in 46m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/94afa824efb14ccf8001fbbb1ef346a1 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd5a6f348dad4cd796d6a8bf681cf028 : FAILURE in 1h 35m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1bbbf86f03764410abedf938b073f7f5 : SUCCESS in 1h 48m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c434350e5fc94afd844dfc4e3a6d2faf : SUCCESS in 1h 44m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9aaa1b12728b473bba40ac72e35860c0 : SUCCESS in 1h 55m 48s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9af030c8398942caa98f7e70d2cf1c0e : SUCCESS in 1h 04m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/210a8fcf93fe4234ae7bc95d060fead6 : SUCCESS in 1h 29m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/21d3c5bd866a46f7a73a36aa1a66f984 : SUCCESS in 29m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b9242be3a39248b7bc13c4ecc522b065 : SUCCESS in 1h 07m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/57e8cf3a48394d0f973afcea5b619f83 : SUCCESS in 1h 23m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/205c4d8baf0b42059ac6f5d46d8d29b2 : SUCCESS in 1h 27m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/966c52846dc94384bc3c3ba6c225bdf5 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":3},{"id":"cbe8be29aa9651ca75bafdcdacbbf452a83641ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 08:41:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9564ff6979905387cd3b6f7c7406c255ed96672","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-04 09:00:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_Unity_FC/f491448/DellEMC_Unity_FC/1233 : SUCCESS in 1h 10m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_Unity_iSCSI/b193f18/DellEMC_Unity_iSCSI/1260 : ABORTED in 13m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"28f4d434c8762a5de922ccf1a71bbe9fb9424c4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 09:02:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a4841c09a34497b2303cf5c241a4b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/893efa2969374e4c85328fc99ae89e1a : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43a83ddd233a411e961f463520f8de19 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57a87b86ebd448cab0b75140d6031f4a : SUCCESS in 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ebea9b9c8f78605216182a440c601f68ea826106","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-04 09:06:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-907420-2/FC/ : SUCCESS in 2h 52m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-907420-2/iSCSI/ : SUCCESS in 3h 4m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"abeec95ecd21257a817a364be48a9a2d500bc901","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-04 09:14:53.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/cb35534500f043fa8e0cbcaa2069dcb8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c5b9292676f440ca8df3ca4672a7d8e0 : FAILURE in 1h 16m 10s","accounts_in_message":[],"_revision_number":3},{"id":"fa53276f83c487547a8838f55e0e87f231b34a3a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-04 09:16:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/592/.","accounts_in_message":[],"_revision_number":3},{"id":"07ebb360a4cd109c89ba7c9af14e2fcf4d460b51","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-04 10:29:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/603/.","accounts_in_message":[],"_revision_number":3},{"id":"ed16431e270d71de53d1cc374fc95016642c03aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 11:02:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3445c3ccd8c04f76b2d7bf5368073053\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeec8acf34a7408ea0139ad2c64f1952 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d498aa6435ba471c96ae1827ef97dd1a : SUCCESS in 9m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/100c2ac4a88e45afab22570a9abeb9dc : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2417ac0d84214850981c2d03f22a95f9 : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/71c59de7c5de4722835f3083dd02b6b2 : SUCCESS in 1h 03m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/da76dcb59a944ba69342c17c548f44ff : SUCCESS in 45m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f21ed5fecc74065aa7ce67ef7a7bddc : SUCCESS in 1h 24m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/126a3f9145b945439b883daa26e180fe : SUCCESS in 36m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/312c639f1aab41a3973a9ff095c3aeaa : SUCCESS in 11m 57s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e9b913451b848e78f38913f7c45f6b9 : SUCCESS in 12m 51s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fba95aba434c4ecbbbbf2aed9bb5b3e5 : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1c06e8122b5349ad83dcad9d64196f90 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/68a450cd6ac848efa20b8dcf46fe47fe : SUCCESS in 6m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fd9cfd8343e4dbf93bde64e424f7abe : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/26517f378990479486370e7653dfc1e9 : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/57aa608b804e4fceaf78f9419e5ad5a2 : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5b804008b236486d9eed78c465a63464 : SUCCESS in 1h 25m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/52b4e06ba08a48a696fc4f42fa7dc516 : SUCCESS in 2h 13m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2867b51feb294f33b4a6d67209c808ef : SUCCESS in 1h 18m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0b3c62b95cdc4548bdb63c7fc80c1fca : SUCCESS in 2h 15m 40s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b7c926a290014ea8b348b46f201c7890 : SUCCESS in 1h 02m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c7fb2a4e476f40caac2048556a90b176 : SUCCESS in 1h 09m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe6577d4341b4862bdd933a07f788b67 : SUCCESS in 55m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aece5d6950ef4af4aa47c02f88af2e08 : SUCCESS in 1h 03m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/040a861f0300456ba5f06af7e779cd73 : SUCCESS in 1h 13m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f36cab94904f4c04b1c12fa1fbf20083 : SUCCESS in 1h 06m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a871eb5eca4499bac6c6c8ac62342dd : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c2accc773d6807c59df5825f18cb9d1ea8d3cd28","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-04 12:13:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/3/upstream-check/cinder-tempest-plugin-ontap-fcp/0faf191/ : SUCCESS in 1h 10m 23s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/b8894c5/ : SUCCESS in 43m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/3/upstream-check/cinder-tempest-plugin-ontap-nfs/51cb81b/ : SUCCESS in 43m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/907420/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/943da1b/ : RETRY_LIMIT in 8m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f813e4383f82cb43a5e4e893a3995d88b4f57fc8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-04 12:16:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_Unity_FC/f25428a/DellEMC_Unity_FC/1235 : SUCCESS in 1h 01m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_Unity_iSCSI/79fbbbe/DellEMC_Unity_iSCSI/1264 : FAILURE in 1h 14m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2b648238c9c6cc56574f6c37edd07449c687bec8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-04 21:45:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907420-1/HPMSA-iSCSI/ : SUCCESS in 2:02:25\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907420-1/HPMSA-FC/ : SUCCESS in 1:36:33","accounts_in_message":[],"_revision_number":1},{"id":"819c77494ebe8d64fc2e914deb27306f66889362","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-05 03:38:55.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_PowerStore_iSCSI/f5d85f9/DellEMC_PowerStore_iSCSI/1034 : SUCCESS in 1h 13m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_PowerStore_FC/bf75fdd/DellEMC_PowerStore_FC/1189 : SUCCESS in 1h 33m 16s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_PowerStore_NFS/1b5c485/DellEMC_PowerStore_NFS/720 : SUCCESS in 1h 07m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_PowerStore_NVMeTCP/249c808/DellEMC_PowerStore_NVMeTCP/589 : FAILURE in 8m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"badd62b5114f9b4f26814cd757adcdcd7dd5844e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-05 09:54:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7859d98896d1452eaf6637dbccce6bb2 : SUCCESS in 1h 13m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b079702d80e449dba5118f2f8be468d9 : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":3},{"id":"be1600a8249d473b7c9f015e28a2b388f69224d8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-05 12:47:32.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_PowerFlex_3.6/c4bfd29/DellEMC_PowerFlex_3.6/1376 : SUCCESS in 1h 42m 12s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/20/907420/3/check/DellEMC_PowerFlex_4.5/233fc90/DellEMC_PowerFlex_4.5/344 : SUCCESS in 1h 47m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"4395fef06cf9ee729c75cece2497865c4bb2cd54","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-05 18:50:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0b3507012dd7fb7a4e74aaa601c7b97cc4af851","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-06 02:13:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907420-1/PVME-iSCSI/ : SUCCESS in 2:07:30\n- PVME-FC https://os-logs.seagate.com/dell-ci/907420-1/PVME-FC/ : SUCCESS in 1:56:15","accounts_in_message":[],"_revision_number":1},{"id":"d7eb0d9f8d8ac967873b0d8232682369426b7300","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-06 06:51:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/907420-1/Lenovo-iSCSI/ : SUCCESS in 2:03:49\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/907420-1/Lenovo-FC/ : SUCCESS in 1:55:55","accounts_in_message":[],"_revision_number":1},{"id":"6191dc3bb4c3be2bbe3061a92b5beac3639e176c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-02-10 15:26:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/907420-3/HPMSA-iSCSI/ : SUCCESS in 2:03:53\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/907420-3/HPMSA-FC/ : SUCCESS in 1:37:15","accounts_in_message":[],"_revision_number":3},{"id":"3716eeff5b0353068a1d6aa24f0754f36e921b63","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-11 06:57:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/907420/3/20240211122126/NEC-V-ISCSI : SUCCESS in 3h 21m 07s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/907420/3/20240211122126/NEC-V-FC : SUCCESS in 3h 35m 48s","accounts_in_message":[],"_revision_number":3},{"id":"8591e77f6604f216ffa2da06150f1b8750ea3f90","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-12 19:39:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cb9fec782af943df307947f246be45d23742e5a1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2024-02-14 06:03:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/907420-3/PVME-iSCSI/ : SUCCESS in 2:08:33\n- PVME-FC https://os-logs.seagate.com/dell-ci/907420-3/PVME-FC/ : SUCCESS in 2:04:27","accounts_in_message":[],"_revision_number":3},{"id":"167cb873079337684faad26e8103ad298891ef7e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-14 13:54:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/907420/3/20240214185335/NEC-FC : SUCCESS in 3h 56m 37s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/907420/3/20240214185335/NEC-ISCSI : SUCCESS in 4h 01m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2bddebf9be3fe48fee935af0a31290555dc0ff85","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-02-14 20:18:50.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"8d67f7b206e6c9adea88a00ebaff6575b81a4c7d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-05 21:07:29.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fbaf5a5cbad1da6ce8d70b8b47f52b4067649e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 21:08:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d94ceb8c14310b975357d2e9c3f80d0e8f9375d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:10:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d57002a23a476da6efba944ec56966\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af423d3db8a04caea3c3330d409207e9 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de43c1e004874311b46868624c3af647 : SUCCESS in 10m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba97d1b6b85046d4bb329b50c834098c : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2613dc8749042e492fb94310ad419ae : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/d3573eb785ab4b329e7cdd2798a4545d : SUCCESS in 50m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/245d99213a9c4d90b370cda933ef8804 : SUCCESS in 45m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2bec26e6614c4f349de68377bd6f7c61 : SUCCESS in 1h 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c0ce98a4bf4fceb654fbfc47db50f2 : SUCCESS in 44m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99ddd22f0a14409db4d29f27f771fa57 : SUCCESS in 55m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd2e0d5dcfc449c9a71d91965f65cf71 : SUCCESS in 2h 02m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bc90c4adc8b482aab72c24a82111123 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":3},{"id":"1c1ba649a584cd45251d06bcd84b1ff2892d61cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:10:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"36bce551cccf09f0d0429c9e0128b8f8d1efaa89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:12:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/611343e8da174d3b9b9a88b10bfc5067\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22990b0cbf84848a219f2694fe16501 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"337adf9a128e2897c73d2b91c2961649e38a95ca","revisions":{"8b3b9404a7ee91681bd00434d17c081cc030c6ae":{"kind":"REWORK","_number":1,"created":"2024-02-01 12:17:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/907420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/907420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/907420/1"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 12:17:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 12:17:22.000000000","tz":540},"subject":"Remove six from Nexenta drivers","message":"Remove six from Nexenta drivers\n\nChange-Id: I3290fd411da511f276d4f592820e9a88df958472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b3b9404a7ee91681bd00434d17c081cc030c6ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b3b9404a7ee91681bd00434d17c081cc030c6ae"}]},"branch":"refs/heads/master"},"b2342246b798611fa922eab34a0f5de9adf78e6d":{"kind":"REWORK","_number":2,"created":"2024-02-04 06:01:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/907420/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/907420/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/907420/2"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 12:17:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:01:41.000000000","tz":540},"subject":"Remove six from Nexenta drivers","message":"Remove six from Nexenta drivers\n\nChange-Id: I3290fd411da511f276d4f592820e9a88df958472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2342246b798611fa922eab34a0f5de9adf78e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2342246b798611fa922eab34a0f5de9adf78e6d"}]},"branch":"refs/heads/master"},"337adf9a128e2897c73d2b91c2961649e38a95ca":{"kind":"REWORK","_number":3,"created":"2024-02-04 06:33:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/907420/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/907420/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/907420/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/907420/3"}}},"commit":{"parents":[{"commit":"2a18843ab928ad104ac9b535f28fe1eea5c1a5c7","subject":"Merge \"Fujitsu Driver: Update extend volume functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a18843ab928ad104ac9b535f28fe1eea5c1a5c7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-01 12:17:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:32:54.000000000","tz":540},"subject":"Remove six from Nexenta drivers","message":"Remove six from Nexenta drivers\n\nChange-Id: I3290fd411da511f276d4f592820e9a88df958472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/337adf9a128e2897c73d2b91c2961649e38a95ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/337adf9a128e2897c73d2b91c2961649e38a95ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
