)]}'
{"id":"openstack%2Fcinder~908938","triplet_id":"openstack%2Fcinder~stable%2F2023.2~Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c","project":"openstack/cinder","branch":"stable/2023.2","topic":"dell-powermax-restapi-101-stable/2023.2","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-04-15 12:55:26.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2024-05-24 01:37:38.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-05-10 06:24:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-05-10 06:24:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2024-05-24 01:37:38.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-05-10 06:24:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c","subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","status":"MERGED","created":"2024-02-14 20:27:34.000000000","updated":"2024-05-24 01:38:51.000000000","submitted":"2024-05-24 01:37:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908938-dell-powermax-restapi-101-stable/2023.2","cherry_pick_of_change":900118,"cherry_pick_of_patch_set":8,"meta_rev_id":"f45883bcdc1f71dd84942bf9178ca56164787705","_number":908938,"virtual_id_number":908938,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-24 01:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-04-15 12:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-05-10 06:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2024-04-02 16:04:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-22 14:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-22 14:26:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}],"CC":[{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 20:27:34.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"CC"},{"updated":"2024-02-14 20:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 20:56:18.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-14 21:11:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 21:23:06.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-14 22:03:26.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-14 22:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 22:32:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-14 23:35:22.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 15:03:58.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-02-21 21:18:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-03-13 05:26:45.000000000","updated_by":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"reviewer":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"state":"CC"},{"updated":"2024-04-02 15:40:58.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-04-16 03:52:31.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-05-10 06:24:39.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-05-23 16:56:17.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"}],"messages":[{"id":"ec8d9736dd9af980ff6b7e889bfb76b0831f7864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-02-14 20:27:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"68d355a8ab6f80f7b4ecefb263bfdfbe65101383","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 20:40:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e6fa69f18cb4d2aa5026d89a1b001ed\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6e1ed8ae5b2411ea9f2e6ee8f2d60ba : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0e559b974434adeb6a645d923c7d13e : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68412d23264f4355984d03866bb53b15 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4df4e1613f3b46a0a74a019cad1cb401 : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f30276d0d88f6f31a6ea015cc61a1c4df1577bf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-14 20:56:18.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4a2094ad76c041feb377c44514e84ec3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3808ec8be0c8412ebc190ef6d0bb6cf6 : POST_FAILURE in 25m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e293782cd4fdc220b721e381eaa0fe2e2116a29e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-14 21:11:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/908938 : FAILURE in 42m 22s","accounts_in_message":[],"_revision_number":1},{"id":"51bf7afa9adc065071ac3c5866c5bfab4494baeb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-14 21:23:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/logs/908938 : FAILURE in 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"25adf4a8421898e1fa798294c9de65c330e2b01e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-14 22:03:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/710/.","accounts_in_message":[],"_revision_number":1},{"id":"c5f582810495a067824a0c742258fc07010a94a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 22:06:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f011006548584952baa92a34fe129e0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5797a2674abf4341a6d5f60903c1551d : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba9402e6400546008111b783ae416db0 : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06d7a6fd968545a6b09f12686b0c9091 : SUCCESS in 9m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7046684733644f8b9c07a971032a295e : SUCCESS in 9m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a5b7d789ef24ee0b11f3c01bcef9f4b : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba780eec4de401cab09cf2df381b016 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/99d501f43434401fa1dcc1f7267bd2d6 : SUCCESS in 46m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ad98623ff2640a08868929a7704be2a : SUCCESS in 1h 04m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8585ffa3f4bb4085adf8201b9e6b6270 : SUCCESS in 36m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c7af0bf65e948a383b092c8ad03e799 : SUCCESS in 10m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cbda145b405d416aa6bf3fcc4d8157d8 : SUCCESS in 15m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fba303b908d845afa55fdb22f8dd6571 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d602488e539445582fae343cc41517e : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92182de7dd144eeea058d865e5f9e50f : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e980a691b78d49d5bbdd6a026e46b94f : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/26e1cdd12d204db4a160684c24ddc103 : SUCCESS in 38m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6bde903ba9b24c2d903a1247413b8d53 : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d3dbfcaef1447db9eb4bff11eb76d91 : SUCCESS in 1h 22m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8cae78466c8a462688eba6361dfb060b : FAILURE in 1h 37m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4c0916bd86843d7b3280621d74a1c66 : SUCCESS in 1h 32m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/06e30ccaf8cf4e3aa9cc7452eaaec43b : SUCCESS in 1h 37m 24s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d707a64e07443cca7b4baad5e796e5f : SUCCESS in 52m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/297df07abb5c424c85f3ca187770e946 : SUCCESS in 1h 11m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/98c1e8a50bd14a3da0b8ba19903d2eac : SUCCESS in 44m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/13b49cf6018148abb9c5eeef81348d4a : SUCCESS in 1h 19m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7ec63531f665438cb0c1550d06d9f714 : SUCCESS in 1h 11m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abce8b4ac08f4d8291fb8aee3f7a9321 : SUCCESS in 1h 11m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b316808cb6247f1b4c7adb82874d22b : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"86fc5969d431fc40c04dafff251b3d1ddd2bc35a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-14 22:32:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/1/upstream-check/cinder-tempest-plugin-ontap-fcp/fa87cc0/ : RETRY_LIMIT in 8m 50s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/0919a79/ : RETRY_LIMIT in 9m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/1/upstream-check/cinder-tempest-plugin-ontap-nfs/fe44a0e/ : RETRY_LIMIT in 8m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"29179111a348699bf90d82a27ee9d81a4d3124c7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-14 23:35:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/38/908938/1/ : FAILURE in 1h 07m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c549c7bbdbbdfd6d85133ebaf02bb36d28a4dcf2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-15 11:13:14.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/712/.","accounts_in_message":[],"_revision_number":1},{"id":"3fe8a78c46624292487445c51fadda2895d32894","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-02-21 15:03:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d9d02f6e851c57f171845ad55acad1a7e08ff3e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-21 21:18:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5978213243925d31e6754280e68aec6d50b0469e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-21 21:19:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b1169cec20492170331b3332e5bd9ba790bab45","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-22 06:23:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9adbb0121ac7f707bf54b52af16ac38b44f4cf92","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-22 13:00:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd76fe477b5892e29411994b3aa261a9203a3fe9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-22 14:24:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d6120ada2cd392f23778bb31dd47553603222146","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-22 15:02:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2d1ee277b918c56fb37186091e83c356496f5f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-22 15:31:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"441d2202c0c386da6b94eaa7642a3aec5fbeae27","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-02-22 15:34:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6e9a02d86a975fc3fd06833b6e75112f1d487348","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:52:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0328eef958c74216b799901f29723383\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2a97ffefedb48a9bcbbe07d1a844fe3 : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91f5b3f9548c492f86e4c79657539786 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e959c2ecf6ea49dab5093a8f2befb5a2 : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d5f5cfe3172c4d168f93b2296fb4e545 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf344d63179f7a5186b592e65df8e8f33617a90f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-22 16:17:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/908938 : FAILURE in 44m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5ba14095cabbae93551952e881ff3ff43baded8d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-22 16:45:46.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b9523eb917bd40cfbd6413c39587e4fb\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0827e98bd638485da01e71d612567c76 : POST_FAILURE in 25m 28s","accounts_in_message":[],"_revision_number":2},{"id":"dd76c4e9c524954c9f7771b0a5776e313e338346","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-22 17:35:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/38/908938/2/ : FAILURE in 1h 19m 10s","accounts_in_message":[],"_revision_number":2},{"id":"c3c18cf38dce03b715afa4c3833aebf5fcf5a080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:43:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd6a892ccdfa4dc6826ded668f798866\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a0745e6dd04f309a52709a31547c54 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb0c0e70f2e84c8db2acd47405fc4404 : SUCCESS in 11m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae13978e13bb40b2a3c8fadcf68460b8 : SUCCESS in 14m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bd34b8009c14287825a0ec5a0f8f25b : SUCCESS in 12m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d02221f1877f45df8722d6eb5c40ef84 : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0ca4be0fb54e07ae062a703c2973a8 : SUCCESS in 14m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/ddc4eeadd5d94111ad575c67b774e40d : SUCCESS in 1h 05m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95ad2104bd57438a9eeb6dfe0685f544 : SUCCESS in 1h 34m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/324912dc10f9451eb84804b33350ef81 : SUCCESS in 1h 03m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1aa01ba27594418fade8b28eb0f5c31e : SUCCESS in 11m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68ce31937d9c49e99315449dfee263ca : SUCCESS in 19m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c62a3726ca04c208c7ed65967f4c5c9 : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/544e2f9833e34aec9a3d95dccfe61f93 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e849fed014cd45368b1ae1e850189d97 : SUCCESS in 5m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a994d405002421387cb9a5421a1d2ab : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/57ca8697af3a4124af3bb2d191612ee0 : SUCCESS in 37m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fd79c08f777248c09c5db7bf670d9fcd : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e6bad3df4fdb4c6aa2118818a7f5df42 : SUCCESS in 2h 04m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b04f3402b02944acbfa315ae9691e934 : SUCCESS in 2h 00m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/93f8ff5fcee84d6a810f22703aea8884 : SUCCESS in 1h 44m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5a9fc113984849f69c68f09ed6f6430e : SUCCESS in 1h 31m 20s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba1f44caa67d403dbbe0f8293a8a48b1 : SUCCESS in 1h 08m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e554dd1b4d564fda8cb9a3ce91ef494c : SUCCESS in 1h 20m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/097580fe64954aedaa18674788acb6ab : SUCCESS in 40m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d5b1836e4ad34aaf90a31e607c72b7ad : SUCCESS in 1h 26m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bf36e944b47f4d2eb535d9bcf257ae69 : SUCCESS in 1h 14m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03c7ac13ac9b4e499f53c5b2b9068a54 : SUCCESS in 1h 25m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdc02fe5d2e9475297894f452f9ce3c4 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3a39157648aefacb5428b9eb427f6c2ed166361a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-22 19:07:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-fcp/d3a9943/ : RETRY_LIMIT in 8m 26s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/6184d5a/ : SUCCESS in 39m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-nfs/c82c30a/ : SUCCESS in 1h 23m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"521a28646bcc21eba855b2a80a852605ee69e05f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 19:59:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/908938 : FAILURE in 45m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d8744efe787b36ccf78c2bdd0b8bf6d3fccfadf5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-22 20:51:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/908938 : FAILURE in 47m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b3c28ed6d298e710fd85852b46d3c941a53eb596","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-22 22:52:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/761/.","accounts_in_message":[],"_revision_number":2},{"id":"4d7fa98b34c90d3a695f70ebc8d414d658a5cb5a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-23 06:44:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/774/.","accounts_in_message":[],"_revision_number":2},{"id":"5b212310071ed4299a805a93ea150712fe61c11b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-03-13 04:07:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9563b9417690b0b364a5292e38026f83bc845099","author":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"date":"2024-03-13 05:26:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e9cc866385fa6029fa8ca9ef33ca30274f37382","author":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"date":"2024-03-13 05:30:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bee49f97d6b9dec893015a1095b8d71b59fa636c","author":{"_account_id":17450,"name":"Chongchong Lei","email":"Nathaniel.Lei@dell.com","username":"Chongchong","status":"Dell EMC"},"date":"2024-03-13 05:32:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1cf66d83f7f7317493103e3f44b4947a74d50d3a","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-04-02 07:13:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57e7d35ccff0d7547e38baf33c7675de11b2fccf","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2024-04-02 16:04:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"88512d86b621c3d900355c20ca0af88236bb8f7b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-04-15 12:55:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d39ae92b9a8b1c074dddf11c5ecb54ab02c06eaa","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-05-10 06:24:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9820399b526ea5c9105282ad77d65073b70e0d2c","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-05-22 14:26:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94153d5959fdfc22027f7aadcd5717e81ff9bb61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:27:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4471f83df13c4c3a1f218ef72edc3cf73fd95376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 17:24:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71d89d8c1ba1481398c3f21dd660e58f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cfdbabe8d64844a70cf01e82761db1 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e103478343f04032b7bffe00e603b5f2 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4a0050bd9a249db80269e6fd351f4d2 : SUCCESS in 11m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0420642fb3014e538f2cb637cf8120d7 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebda3a83c9b240eeae84dac5e8fa55d7 : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/078cb33a2c3942c297975310adbb9e82 : SUCCESS in 44m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45232ba5769b40c7a7e46411e33952eb : FAILURE in 1h 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f50052956544111b55846ed61401bb9 : SUCCESS in 59m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bdab036135045f09e6bc0622a0e7f10 : SUCCESS in 12m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7deb0bcf5c3d4358b6faa1627c985204 : SUCCESS in 1h 21m 47s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fad2d411720b418ab56e6debb1523c2d : SUCCESS in 1h 52m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49897428c4f041e59d701fb612ce99c3 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":2},{"id":"c0ddfbc66172169a8531464728ab251ee5a08b00","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-05-23 10:50:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4932afd8608a239098e263c01c2a55d2e344e45","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 11:15:10.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/68bca043aa28499ca8695868a5a1bd56\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7413aedc7ce14ce592dfeb2b2fd581aa : POST_FAILURE in 20m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c6031591836109afd26ee88e69ec18e617a5e0df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 11:15:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ecd4601a3934835afff20f0248fed09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dab711c7d634e53b834c34ce2687b8c : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13e954ed7a3144b1bf7d7e68f852642b : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edf48deac5b04b28a7f8a073c072ea53 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40d48d3b94df48afa7a3c961521a121f : SUCCESS in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02f1f9bb5a802e3e459762ff56195779ec13f421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 12:46:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d5d99d7e0884f4b90afea7deea6b3d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0669f98ec0244e46a7fc73fabbbc90d4 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3300a43f7c064b44adc4adb717b57a27 : SUCCESS in 11m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cddbe0dace1414c8fc745f23996340e : SUCCESS in 8m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4520e6d94b8c42d6b844af69e0bd4c55 : SUCCESS in 12m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fe229445db442558cb09e80375e9f25 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/106eecb4aebe4f0bbbb62a44099bc5dd : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/6518ce3c50a740cf94c0add6d09f0965 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4d5e4645ba74319aea86a12d4896558 : SUCCESS in 1h 25m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2b605cc796b4a2aa1a5aa53df7823be : SUCCESS in 44m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c06fb2d36384527a3753602eb73df08 : SUCCESS in 20m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8df233dbd9434bce81107ea97a1bf002 : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8b06f7dbf2f1470581dba3291407d859 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0969a7bae106419893874122818ea01a : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/134963b9e07f42538997fce84d4cc631 : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ec1078b7d0a4335805f9d2c73b11408 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d30cb6b5c904feabff1badb332805a9 : SUCCESS in 40m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86d974cae6ed45e093c5bde54a2c3903 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4bbaa42fbc00405ebfb1a4484e5bbc2b : SUCCESS in 1h 19m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/54a7b301244f47d58f4df3f148b604a0 : SUCCESS in 1h 35m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1cd684f4114047bab5cf7d3301666b23 : SUCCESS in 1h 24m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2772f391b8ce43658abfd9809018df8d : SUCCESS in 1h 50m 00s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6499a1bd6fec414986ac5c5bdca0a246 : SUCCESS in 1h 11m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/69aa02c4544d4d2599250e32174c9d29 : SUCCESS in 1h 04m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7fee00ee07fc4af3b965084a688360d8 : SUCCESS in 28m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/64677800925c4096a8bee0466e9f18ec : SUCCESS in 59m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d7cf4cdbb2c74a128518ac567d7db54b : SUCCESS in 1h 22m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6f8b93ddf074417893099bfd018134a : SUCCESS in 1h 12m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77d2b9fc118f414395a66afef6efe16c : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":2},{"id":"061cb8e78eeea7ae41749becf44820a9aaed9eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 12:46:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb89c9dfdb225dc655913e9b97285a6f289cf899","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-23 13:01:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-fcp/73a9be0/ : RETRY_LIMIT in 3m 32s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/680661a/ : RETRY_LIMIT in 3m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-nfs/0bf7055/ : RETRY_LIMIT in 3m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"bb9e974cd4355be70b504f501b99f6cabc25b819","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 14:52:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a298cda5ea544365ae7e55ab4f5bd190\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e35be8afa74f2a8942559051871799 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/028340925b1f4d61b76eb4e17c00af71 : SUCCESS in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67742249b384459ebfb67c19f31fbe21 : SUCCESS in 10m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b28826f2113427984c18a7efa2b8b31 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea4ff7c6e6e24ca1bb777b989cf05621 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/24d590cfe67e4ec6aebdd57e3957fbad : SUCCESS in 50m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bf10fc2e6284fc78235d547ac0025db : SUCCESS in 1h 30m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd19317db6cf42cebdd15df007a45fc3 : SUCCESS in 1h 01m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99edcc1a219d4701a3bba4387cf356fd : SUCCESS in 10m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d2741836df541cc84dff2956577f97f : SUCCESS in 1h 01m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d14a9a17693a46ff843a1bb03e96a01d : SUCCESS in 2h 02m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a64001ae80d459ba01c68ebdd318e5c : FAILURE in 47m 53s","accounts_in_message":[],"_revision_number":2},{"id":"17a9dde47f6070d81e30e27beea731e3abd761fb","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-05-23 16:56:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af613d66bfba5e894d134e5a49bf234a6b6a4fa8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:17:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/856450474798474eae80942944f1bd33\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a881a638d304f9ba217eb8204b9f4b5 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1efc287e90547a19a25085fa4f7feab : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9eb61738b4f4b1d8949a2e0ce6d3b56 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67166a83d1f541e3a8fe828fdc210231 : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ecad76e445a8aa064a3b88179ca420e96fe0331","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 17:24:05.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/844d84b2e4894c4a8d98313b615f22e1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5b74193bcdff40da9dbc1046821d841e : POST_FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7793a0cad7c633569dbbc340d5302f205abcc62c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71875c2c31484477bee04db6f13cffcb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301d783515664e01b4eeac4746ddcad1 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cddc244f51c0420e82584e7c00196806 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84874cc5963445278e44d172ea76460c : SUCCESS in 13m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1beaa440bb7a409a9232e40448a32dcc : SUCCESS in 11m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85a14431c7f74faa9d516f1e8dce24b6 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3745b9f0b6c4ed8adcd2f98c100f7ea : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/32becf88924648988479b3fd2f8785e4 : SUCCESS in 1h 08m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/34557a3a62f1481f94ce08e5f856ba19 : SUCCESS in 1h 23m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2077d721347f4eee8adb41302e32a512 : SUCCESS in 1h 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9b6e0760b8041ac9c92667fd2805380 : SUCCESS in 13m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08c9a68f72774337b8de79ead4ff8868 : SUCCESS in 17m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1bf4987ddcb43d89384871cf67812f2 : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bccf8a8183a14a368570e9b188d9f11a : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/046d8d83a56c482b86dcd6c36126de08 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da7b579f96ca451781e5ae0e5334cf84 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9330634350f0422ab41898f0ad3d3d9b : SUCCESS in 38m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3199590814a84767888226f970ed2cd5 : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2f457ffae3ca46d9abf7daa5e7d7d7be : SUCCESS in 1h 35m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3f8ab6a4ca6443d68a8898e193b1efe4 : FAILURE in 2h 12m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9231c4708935407d8b710a7783766bf6 : SUCCESS in 1h 46m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c8245e7e028046a9909c72d840de4f88 : SUCCESS in 1h 48m 16s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/657a25409a5e44f8ac54670b96cebd63 : SUCCESS in 1h 01m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d6318f147fd646e78b2abd1762709984 : SUCCESS in 1h 25m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e056277357384d6896f364f154cd754c : SUCCESS in 32m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c755bce18ac4422a8a97ad10c00d5860 : SUCCESS in 1h 18m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f323b370a924485db44da918490220bb : SUCCESS in 1h 05m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59b3113fff8843f8bde094bffb5adbc5 : FAILURE in 1h 35m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6272817f8b7a47a28235bb6bc20da975 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d2283240a7c5881f33c660b465facfd0b16f29bf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-23 21:48:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f098371602059a91464234c0e1c6644b194b236","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 22:02:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ffa72f090b471eaf4a21b21d73d13a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03cb14ae393a495b840805ce5caa2c43 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a19f5740e494edf914f54f5d1781685 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e1e9576949ff46188796bd4b13647d67 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c035cf0cf454f88b81e5f1355a011ee : SUCCESS in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44753ec26e114995dcea68ed90396e5ffa049543","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-23 22:14:17.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/50a1af9a8c6f4dbaa1cb3e62c9c06a9a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/935846df11d84055a3a44a7745760f85 : POST_FAILURE in 21m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c100ca4c159d22381fcdda6537d9b9045c593e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 23:50:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f15ce00b5b49b3819cc315162dd1c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666eab52f9ac46feb71b4b1099f9fc28 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b39670bb2cb0494e8fae468a0e358fe0 : SUCCESS in 13m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45fd525a72514ae3b9cbc9202b7e2c76 : SUCCESS in 11m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c8904255c4b4d62a52527582104afad : SUCCESS in 12m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c62cbf04860f400289f853aeb80276e9 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0817428f0dbf4aba8111c5f26c7fde52 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/864c1895b6434266967a4cb700db6d25 : SUCCESS in 52m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17d3695618224992b369a484f7386d37 : SUCCESS in 1h 21m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5eb5b6a443ae4bc48e5d8a58f68497f4 : SUCCESS in 1h 03m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6981d75d29b74f08babac6416f46daa5 : SUCCESS in 13m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d478a26cd304a17b80968d41720fd6a : SUCCESS in 13m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/315bda42ee8e45aca2395710baf524c5 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/721276f79d424de39a1902f7296dbbb8 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e94749ed8a93414ca433d960eff894aa : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d01dc5c1e4294cd9a2d06a354c6b3f1a : SUCCESS in 5m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/439f82595e874e9cb3f73316b061bd06 : SUCCESS in 43m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d98872935d0e4b6dbd5e435dceabac63 : SUCCESS in 4m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/07a354a90cd844ecb00bf7592af6afa9 : SUCCESS in 1h 46m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80f45a55185e45518d1faffe7eb6bbe3 : SUCCESS in 1h 56m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fea8344a0bde4eee9356a1b3c7c8248a : SUCCESS in 1h 43m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c58b2d3832b64eb4bef90d67a3df04bf : SUCCESS in 1h 38m 42s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2e3e30d99df41e092aa63d5dcde579c : SUCCESS in 47m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f897008dd5b45adb8d3e265e58a240c : SUCCESS in 1h 30m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/beb27c85b84741db9f260981cd81715c : SUCCESS in 49m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/745381142d4f46f9ab57f6e340897623 : SUCCESS in 1h 14m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0f13343372e045ea8c863852b03e67c6 : SUCCESS in 59m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb4d05821d744c8ea58438acb2eda985 : SUCCESS in 1h 26m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dea919e6964d42499f54fc3cb211e1fd : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":2},{"id":"53cf027938bc2af9693926e49033aefcd07d5a4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 23:51:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3dce6f8232b4948397a98be21d2a17a2c47d7a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-24 00:26:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-fcp/cfda9c5/ : RETRY_LIMIT in 3m 46s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/e7c459c/ : RETRY_LIMIT in 3m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/908938/2/upstream-check/cinder-tempest-plugin-ontap-nfs/bb0f452/ : RETRY_LIMIT in 3m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b5f4d5aa6455c053e47d852c7987efa809c20a2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 01:37:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2488d795b32947778277517c74c3c1e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e702c7454d4803b1f1a799159a4aa5 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d64866a23864ddab1d30760d9e20d78 : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cb07c83d42248d6b2a063f6372a39cf : SUCCESS in 12m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a81445a82f10463a95f1f7917e32e813 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b1e8c3f4c64450935599919d61bdb2 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/60bc149069d14e388e0c132ceab1f045 : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ffa9d857c16499181aa90de1d8760f1 : SUCCESS in 1h 33m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68f220e5e66a4a59af7229152882ab15 : SUCCESS in 42m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4439495439264c739d95c67ea515f454 : SUCCESS in 14m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f45c04ed1de64ed5944c850e8f6b934d : SUCCESS in 45m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eba4e3b6b646477085f46eb958b41d30 : SUCCESS in 1h 15m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07dec615ea8f4ef08bfdde6b6dbd942a : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":2},{"id":"037cb3b56f82d68f6b0da459424bc353f3db8618","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 01:37:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f45883bcdc1f71dd84942bf9178ca56164787705","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 01:38:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f5248953ca41f39a90bd559cab7e8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/193afec801f94fa986ccf8eae2c52c9d : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d2652a8d2f64aebb7d30ad0c8b92970 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68acf0b8161e394fbd44546eaf90dbd23efce684","revisions":{"bd190fd8357e019d7027176be03fe570215c6f7f":{"kind":"REWORK","_number":1,"created":"2024-02-14 20:27:34.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/38/908938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/908938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/908938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/908938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/908938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/908938/1"}}},"commit":{"parents":[{"commit":"34338cf26f9d070ad8106fceccd657f613c4d5a5","subject":"Merge \"Skip sparse copy during volume reimage\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34338cf26f9d070ad8106fceccd657f613c4d5a5"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2024-02-14 20:27:34.000000000","tz":0},"subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","message":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\n\nFixes an issue introduced by change Ie9561eeb30a54539cbd, which added\nsupport for PowerMax Unisphere 10.\n\nCloses-bug: #2051828\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n(cherry picked from commit 5d725806ff274d60aaeef83903303e1939502343)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd190fd8357e019d7027176be03fe570215c6f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd190fd8357e019d7027176be03fe570215c6f7f"}]},"branch":"refs/heads/stable/2023.2"},"68acf0b8161e394fbd44546eaf90dbd23efce684":{"kind":"REWORK","_number":2,"created":"2024-02-22 15:31:21.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/38/908938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/908938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/908938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/908938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/908938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/908938/2"}}},"commit":{"parents":[{"commit":"34338cf26f9d070ad8106fceccd657f613c4d5a5","subject":"Merge \"Skip sparse copy during volume reimage\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34338cf26f9d070ad8106fceccd657f613c4d5a5"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-11-09 06:16:01.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-02-22 15:30:12.000000000","tz":0},"subject":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)","message":"Recognize Dell PowerMax Unisphere 10.x (x\u003e0)\n\nFixes an issue introduced by change Ie9561eeb30a54539cbd, which added\nsupport for PowerMax Unisphere 10.\n\nCloses-bug: #2051828\nChange-Id: Ib66688f319d6aed5d7339f4eabdd7ca9ce7e0f6c\n(cherry picked from commit 5d725806ff274d60aaeef83903303e1939502343)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68acf0b8161e394fbd44546eaf90dbd23efce684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68acf0b8161e394fbd44546eaf90dbd23efce684"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
