)]}'
{"id":"openstack%2Fcinder~909000","triplet_id":"openstack%2Fcinder~master~Ia241027e0fb31c00e80d088f11ab50c375407519","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-04-19 14:48:43.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-04-24 16:56:24.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-04-19 15:08:57.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ia241027e0fb31c00e80d088f11ab50c375407519","subject":"Mark nexenta_encryption option as deprecated","status":"MERGED","created":"2024-02-14 15:30:49.000000000","updated":"2024-05-10 05:36:30.000000000","submitted":"2024-04-24 16:56:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909000","meta_rev_id":"3cf03a0080c6f8af87a8eff3d5c3d71830c5c5c1","_number":909000,"virtual_id_number":909000,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-24 16:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-04-19 14:48:43.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-19 15:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-19 15:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 15:31:00.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 15:51:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 16:00:38.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 16:27:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 17:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 17:53:58.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-14 17:58:31.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-14 18:52:05.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 18:52:57.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-14 19:31:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-14 22:19:20.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-14 23:47:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-15 03:41:41.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-17 07:05:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-17 08:26:46.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-03-01 19:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-03-13 18:25:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-13 18:31:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-14 05:15:30.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 14:20:36.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-03-15 14:37:04.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-03-25 05:35:06.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2024-03-26 17:57:54.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-03-27 05:31:41.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"}],"messages":[{"id":"2a9588e813936e41f1812da71e2a0aa5dd5ef51f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-14 15:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52a73e55cfb9b58508f71fe5e1c0cce40514ab67","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-14 15:31:00.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16520/","accounts_in_message":[],"_revision_number":1},{"id":"99282f39ed5e6113857e671d7192658cabbe3db9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-14 15:31:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16520/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"418a9e0453c61667b9f4ab881f5c709b983cc9b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 15:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c0bb89b883b4d7ebeb701bcf12f2efe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4280556e52c846bfbade9ba153c9dbc5 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6db50f81bea1407cb13bac237d1eea26 : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40056082545e44429249a533406c76ce : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f6734c66b68c4b366271c8cf1b224cab6d56072","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-14 16:00:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7c8e1d1c64ac4775885d91885911f1c9 : RETRY_LIMIT in 6m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/56382685d676402eb2d60471522c258a : RETRY_LIMIT in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"be088e8d234ee23f8264fb595f64c3f961838afa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-14 16:27:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909000 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"38d8b1e3fc540c3114467e824d335080194c3a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 17:32:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cce8df1a192468181fe9bc94ea3dcdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439746e2477242a1a42ef931bc7d911a : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99bea6df2cbf4ca38b85f85cef344062 : SUCCESS in 10m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c637cede36d648ab8666dd3d3c43dfd2 : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e65ed0b27243a7b66913459eb4290b : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/082d50f0df574cf39d6a5899b9070488 : SUCCESS in 59m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e874ee597ddb4c1d91026cecb1d69f07 : SUCCESS in 52m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ec2cdca47824a9c8600ff3438ba975a : SUCCESS in 1h 14m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4aa21cb98d04835af655aea4be4e981 : SUCCESS in 29m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1e5aa60863984268bd87d9d021005cea : SUCCESS in 16m 20s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1ae96473279248698f227e394a70b6a7 : SUCCESS in 12m 55s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d3e7bb60b7a64e4d986462256e3748ad : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/481cf7370f7f4c4698f9ad151b1b8ef8 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a948041bad943a5bea1fb0aaca91714 : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b8d617995e8e450cb765e8d3a1197953 : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc000a6e5d634c3aab6c646c62962f27 : SUCCESS in 38m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a61e9e8d7dbe4f7a992b6be24289e1a4 : SUCCESS in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0d1efa611a140fea45e89b6c43c5cba : SUCCESS in 1h 34m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/60b0642f39e84b9a9dc508b1cbdd66ce : SUCCESS in 1h 59m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9535840df3cf4dd585fac787d1b4c5c8 : SUCCESS in 1h 30m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9a394a11ead446238d57226bb54c9f88 : SUCCESS in 1h 37m 14s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bdc458d01e9b4303af640d085bc974ec : FAILURE in 1h 09m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b5ed1cfcc0e446a9b44aec3335e5a234 : SUCCESS in 1h 23m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3821995e0064a6baf0c389faf599897 : SUCCESS in 43m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8e28dd3aa2a345e5b1ed70b037ff7be5 : SUCCESS in 1h 42m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/36a3bb8629cc4a4baa4224c0b7c7a734 : SUCCESS in 1h 27m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa6f1872f6024fbaac29832406342dd3 : SUCCESS in 1h 22m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9af04957d5745a9b2af03b80bb04c71 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4a0bae5c27adb2ab9bb6e82e488668f4b491cba7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-14 17:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/708/.","accounts_in_message":[],"_revision_number":1},{"id":"ef4481db96c7760b1b2d60c64d4a9dd401e284bf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-14 17:58:31.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d2b49be126c5454781475d4e4d57a45a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/fcd711e8d9f14603b4527c5dce6b610b : FAILURE in 1h 31m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9051ad1d7f7309dff3269762817fbaf7b507858b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-14 18:52:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/909000/1/ : SUCCESS in 1h 25m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d7422202d7a1b205a70c7491142a5e56b07281b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-14 18:52:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/logs/909000 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3afa8b89d8a5558cca72555baa71cd74f0bdc38c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-14 19:31:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-909000-1/FC/ : SUCCESS in 2h 52m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-909000-1/iSCSI/ : SUCCESS in 3h 18m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e69c9681dd91161b3ed6b63d73f278048236e9e1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-02-14 22:19:20.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/909000/1/check/fujitsu-eternus-dx-iscsi/c451830 : SUCCESS in 2h 31m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/909000/1/check/fujitsu-eternus-dx-fc/0c63892 : SUCCESS in 2h 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"98f4ed2f143c220f2c80deb8c070d03e0e2098de","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-14 23:47:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/909000/1/check/DellEMC_Unity_FC/a5a57ca/DellEMC_Unity_FC/1324 : SUCCESS in 1h 15m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/1/check/DellEMC_Unity_iSCSI/0c445e6/DellEMC_Unity_iSCSI/1354 : SUCCESS in 1h 35m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4f5bac224bd6b037d32dd9088063b218f5348e75","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-15 03:41:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/00/909000/1/check/DellEMC_PowerFlex_3.6/2e4d8c1/DellEMC_PowerFlex_3.6/1457 : SUCCESS in 1h 38m 25s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/00/909000/1/check/DellEMC_PowerFlex_4.5/fa5f1a0/DellEMC_PowerFlex_4.5/423 : SUCCESS in 1h 52m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"69507f8452904149e988e9e6d0c06c8489fbbde0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-15 08:39:48.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/710/.","accounts_in_message":[],"_revision_number":1},{"id":"5079aa07f55ed0d613b1fd7f3264cb0c9c03f877","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-17 07:05:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/1/20240217114716/NEC-FC : SUCCESS in 3h 58m 08s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/1/20240217114716/NEC-ISCSI : SUCCESS in 4h 18m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ca6ea6e80d2fb036455341b7846cbc1a279a2fda","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-17 08:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/1/20240217135614/NEC-V-ISCSI : SUCCESS in 3h 00m 35s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/1/20240217135614/NEC-V-FC : SUCCESS in 3h 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0c2b124a483d6fd7b710f277b33277bac94640a7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-26 14:04:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4ada13b2329c4130b9ef77d4843d7a11 : SUCCESS in 1h 22m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fe34bdd7a0fd4826b45af7bf1b7b1a01 : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"447e9c70b921a968be9b06be3df489e55d3119ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-01 16:12:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"37211987a15039df955eba3abe3103968c8106d9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-01 16:31:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909000 : FAILURE in 16m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4ede3f5034fa1fcfa07cf4d82bcdddae53525db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 16:38:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b7df9f575ac42c0aa3b7d21088f6157\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f6e2b64667d4936afd501c37ec6e077 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a67c54ca7c743e1b2f48d0ec1e683eb : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/868eb86c56f542b49483f5bdb4bb95cd : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b3d5ea61c7f4eefbd352323ccb5be8adbc883f2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-01 16:40:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909000 : FAILURE in 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"7452a8f4e8c927b1106c8f70610bcfd4078119f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-01 16:52:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909000 : FAILURE in 16m 53s","accounts_in_message":[],"_revision_number":2},{"id":"7aa705e1a6a16f882d4428acf115d04357f6308f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-01 17:28:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/03271ccab31f4f6abc040d3e0091192b : SUCCESS in 1h 13m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/118899af4a9843c7b9e3bd3e7c339a56 : FAILURE in 1h 11m 54s","accounts_in_message":[],"_revision_number":2},{"id":"13281b3c6ea682f9ac577b95bb725cc418f38000","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-01 17:45:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/909000/2/check/DellEMC_Unity_FC/7cae2b8/DellEMC_Unity_FC/1440 : SUCCESS in 1h 12m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/2/check/DellEMC_Unity_iSCSI/c06c60e/DellEMC_Unity_iSCSI/1472 : SUCCESS in 1h 32m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"98cac172ba383f674508781271272147803eda1a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-01 18:19:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/00/909000/2/check/DellEMC_PowerFlex_3.6/c279f47/DellEMC_PowerFlex_3.6/1583 : SUCCESS in 1h 29m 57s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/00/909000/2/check/DellEMC_PowerFlex_4.5/90fefe4/DellEMC_PowerFlex_4.5/536 : FAILURE in 2h 07m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"18d1dae10fca220e4cad5886c2426ca82357c21e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 18:32:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7a94c5385d54304a0b208eb9dfb83fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512528295d2f41f283dc5ef1a4dbc0ea : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18c07feeecc94bbaa7abbb672826929f : SUCCESS in 12m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea888156362e4bd089cc8275cdc44c2f : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a3ed97ec704190a6c5dbdea932a4d2 : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/cfa2378bb1c343b5babd2b96273cb25f : SUCCESS in 1h 05m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4af2828a9cfd4ef397125df29f147ced : SUCCESS in 1h 06m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21d0453324854fbd9b4a3cb07d19a37d : SUCCESS in 1h 30m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fa64204dd3a440bbe8ac3c57c04d240 : SUCCESS in 1h 00m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b56bfca7e5e84137b29810c1ad4f9932 : SUCCESS in 19m 33s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bf3b210f47e14a418a9b07b88024b764 : FAILURE in 13m 10s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b9977ded9ce478293c3ad67edb8e69a : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/66e8604206c34212a506e7f0fa9b4e54 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8a4be64e64e34b8b95dec670c7adbf47 : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9283464d58444dc2b8b13170da214413 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25fcd0fdcdba488cb31491a9487930e2 : SUCCESS in 32m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d446626a7724047828ac63b23b1b579 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f167500d0e074183a9fc1624ce9ad4f0 : SUCCESS in 1h 47m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a60340e34a7e4b9bbc263a8e54c664b9 : SUCCESS in 2h 01m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8088b25829cd410a83a73ddb59d6892f : SUCCESS in 2h 03m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b9d7db748c344d9d828f377031d6594c : SUCCESS in 2h 07m 39s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ebbc3b1755824dd99b482be7bfe0c256 : SUCCESS in 1h 08m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bf932537b7534e299dc7ef0ab728ca3e : SUCCESS in 1h 53m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a73e566d4e0c4f46bd2765f780ebd44f : SUCCESS in 44m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c0fcdead6d9845678fefa51f120ec51b : SUCCESS in 1h 47m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b70839809b9944ebb591e59ec83f08c6 : SUCCESS in 1h 27m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/505c17bbd4264813b9a89c66caa71265 : SUCCESS in 1h 31m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3949a65da42d45dda319e8917b3580d6 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7f187fbdf7d220038ee580a01972554a90b757ec","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-01 19:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/909000/2/fc : SUCCESS in 0h 51m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/909000/2/iscsi : SUCCESS in 0h 53m 41s","accounts_in_message":[],"_revision_number":2},{"id":"db0a12af83adada31adbbe43cddbbe39817e9586","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-01 19:16:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/909000/2/check/fujitsu-eternus-dx-iscsi/b17b2d5 : FAILURE in 2h 46m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/909000/2/check/fujitsu-eternus-dx-fc/eafa055 : FAILURE in 3h 03m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bce8d22bf5f4b5a417435400ddd58447685c5cba","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-02 10:31:27.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-909000-2/FC/ : SUCCESS in 3h 50m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-909000-2/iSCSI/ : FAILURE in 18h 18m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"1356793d4f30f5ced7d8b9065b83e8f9b5af74ea","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-02 16:20:54.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/84ffb787e91e4a2584e21c70aaf6f319\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2e26176cc0f546829b8d11774d66220e : NODE_FAILURE Node request 200-0000002638 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"0ee329780dce4c6cfc1e0eaa9bd0341234f51b95","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-03 16:57:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/2/20240303212922/NEC-ISCSI : SUCCESS in 4h 02m 33s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/2/20240303212922/NEC-FC : SUCCESS in 4h 28m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a8c949413c9debc5ddca389898b40c5e2901ea08","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-05 15:37:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/909000/2/ : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ec0c738b35bebc7f20819b71e03967cc05f970b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-13 14:08:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f08e23da4be3be780e12193f0d3dc9133ba9b992","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-13 14:14:18.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/430f70bc68f148899200ed1d6abefdf3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e71e8408504647c19769899475f5a84b : NODE_FAILURE Node request 200-0000002753 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"dd7674a59432897e8650e15b1042e98583928345","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:32:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3713ba2c810547b2ac70c5fed9435d07\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48256f620b9544e19202ac842740aa77 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24f3dff42266495ea0aafb9afb5d3440 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0de183a88e304c7b8ea8ae17497cb4ef : SUCCESS in 22m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a50eea1192ecdcf30fc7eba72c624a4bd57bb014","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-13 15:05:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909000 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":3},{"id":"72b8e7ae590903ae66c36a6222c07ad897d54a29","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-13 16:16:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909000 : FAILURE in 36m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ea2d17b4b2c79b2ed6bf9781c388b2ba17986b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 16:36:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69eb5cac0ca84093a4f35ed15a7fbb7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b5a9bae125428ebbf57c942b685adc : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fec084e96fe4f97b02d9fddd54b56cb : SUCCESS in 10m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ab1f8767097427d9077a01cf8fdd8fc : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acd09a4e4c24f83bb58e0826f6712ce : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/e97bf51ddab84b9da84863111e5250d9 : SUCCESS in 1h 13m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e0bac55bcaf3422da021f83b65a4e233 : SUCCESS in 1h 01m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f2324e72aee4a8cbdd287f87fe05c06 : SUCCESS in 1h 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6407d9452f954c92bc8658b9c335d98c : SUCCESS in 1h 07m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95cfe857428d45428efd8d7c66c5664b : SUCCESS in 17m 14s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f26aa81d56fa4c6d8c0fe3c27ea75d8d : SUCCESS in 11m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0871c7a348b46e5920d23c729a4c9d5 : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc0581509a8b410f9bd934f88a1f2bc7 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a86d396350ed400abe3cc30a74aac3c9 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1037775969d7493ca0e486e516281fbc : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4090686eacf2467f94e3fea7192f7565 : SUCCESS in 43m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/789af0d794574619a6a74342aac8974f : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b9d3360a840470e8fa150cd8c58c0e5 : SUCCESS in 1h 56m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6b18ffadc2884e15944759af7dcf79a5 : FAILURE in 2h 16m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eead229f1b7f434db4782f075b33ebad : SUCCESS in 1h 35m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/acaeec0ebefe41a1bc805354fce4574c : FAILURE in 2h 11m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4e675a8f32314e1686868517c79d85c4 : SUCCESS in 1h 11m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6eafe69e3b62483ca654c82eb1104b9a : SUCCESS in 1h 41m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/510e461ed5e74dba9a9fe46ffaf71bd9 : SUCCESS in 45m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95b14eb208d44527875f5425ba920a35 : SUCCESS in 1h 25m 51s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0d67243860944920b7b24df9b8283401 : SUCCESS in 1h 32m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc3fe91f145f4d1e8c2cf7b2c5df1252 : SUCCESS in 1h 45m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/899a25030f654aa2a4e1881ac4b1bfca : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":3},{"id":"25fec6f5a2d3c6b67cb2e014774e921e5bff804d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-03-13 16:39:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/909000/3/ : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":3},{"id":"3b489291a80e68fcbe337c18d3eb4571b1244df5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-13 16:41:31.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_Unity_FC/47d13be/DellEMC_Unity_FC/1491 : SUCCESS in 1h 20m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_Unity_iSCSI/37c4b02/DellEMC_Unity_iSCSI/1525 : SUCCESS in 1h 29m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a2a53dbdc1bc224227a1b1ccc8cb5f59faa35c13","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-13 17:22:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909000 : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":3},{"id":"deb7f0be191c203fc915b4a4a728f9a9a7c0e149","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 17:52:49.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerFlex_3.6/a4252d8/DellEMC_PowerFlex_3.6/1634 : SUCCESS in 1h 37m 31s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerFlex_4.5/813deb0/DellEMC_PowerFlex_4.5/573 : FAILURE in 1h 54m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"4b3d4e7bc7b9e4a1a9ec4a5c0bff32c0bb0d67b3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-13 18:25:03.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerStore_iSCSI/fb2e741/DellEMC_PowerStore_iSCSI/1152 : SUCCESS in 1h 10m 38s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerStore_FC/f5f949e/DellEMC_PowerStore_FC/1322 : SUCCESS in 1h 43m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerStore_NFS/f1b0b9b/DellEMC_PowerStore_NFS/840 : SUCCESS in 1h 03m 34s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerStore_NVMeTCP/4fbbc7d/DellEMC_PowerStore_NVMeTCP/702 : SUCCESS in 1h 48m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"9198aed55ec0e89f9accf6279379f49fae627d28","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-13 18:31:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/3/upstream-check/cinder-tempest-plugin-ontap-fcp/d127c0c/ : SUCCESS in 1h 04m 33s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/d5f0d71/ : SUCCESS in 1h 23m 26s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/3/upstream-check/cinder-tempest-plugin-ontap-nfs/49bddc5/ : SUCCESS in 43m 17s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3d97ef2/ : SUCCESS in 1h 22m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"00f18cea4a12de35d3611953c271b9e66b949be8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 18:50:00.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-909000-3/FC/ : SUCCESS in 2h 25m 18s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-909000-3/iSCSI/ : SUCCESS in 2h 19m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"28f1fdbfa2013ab23439f8e692d48903e1c287f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-03-13 19:04:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/44/.","accounts_in_message":[],"_revision_number":3},{"id":"2973094b0e6a7d7ab99da87b375110294a9c0418","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-13 19:41:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/909000/3/check/fujitsu-eternus-dx-iscsi/63a5d85 : FAILURE in 3h 51m 49s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/909000/3/check/fujitsu-eternus-dx-fc/016c872 : FAILURE in 3h 53m 43s","accounts_in_message":[],"_revision_number":3},{"id":"0a2876eb13db80f8f7f5c238df2b950872aa76fd","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-14 05:15:30.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerMAX_FC/12e7bb5/DellEMC_PowerMAX_FC/338 : ABORTED in 10h 30m 05s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/3/check/DellEMC_PowerMAX_iSCSI/9576bf5/DellEMC_PowerMAX_iSCSI/358 : SUCCESS in 6h 14m 51s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"cafcb988c62dd3cbd42ec8a9731eb33c206cd594","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-15 03:05:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/3/20240315013037/NEC-V-FC : SUCCESS in 3h 35m 42s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/3/20240315013037/NEC-V-ISCSI : SUCCESS in 10h 34m 33s","accounts_in_message":[],"_revision_number":3},{"id":"f13c7c29ed924462194166c9669385b3a22b0966","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-03-15 14:20:36.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a27a7b4947595ac113e2bae9061e8f8a5948935","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-03-15 14:37:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cbec1881cfa43c6bcd01e05e616b892b8aa52b41","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-17 04:34:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/3/20240317052044/NEC-FC : SUCCESS in 4h 44m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/3/20240317052044/NEC-ISCSI : SUCCESS in 8h 13m 31s","accounts_in_message":[],"_revision_number":3},{"id":"fb3efa73cc6dbc4e21e0fe1ba78a6e452083e126","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-03-25 05:35:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/909000-3/Lenovo-iSCSI/ : FAILURE in 5:37:53\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/909000-3/Lenovo-FC/ : FAILURE in 3:02:55","accounts_in_message":[],"_revision_number":3},{"id":"b04db318eb1075864ea173f6bef25c38426a9816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-26 14:34:54.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f7f48a42d2202724b76b429035b607841951d650","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-26 14:35:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7605da206beb227794230648698b54751541b702","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-26 14:40:47.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7e17e8f7003b417fb2d5e236c7061d72\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6a968d70189040fca921bde3ba326b07 : NODE_FAILURE Node request 200-0000002884 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"036d344da82d50be0d18c4bf62ba6bd0ff88dcda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:48:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721f5d49b913415a9c6365529194fa30\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aaa2e034bfa346bf8bac26edb8e374fb : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b313ccdb03854861b44a61d0ff8c628f : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"22032d20c0bc3e26202c49cdba5fb3440c319263","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-26 14:53:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909000 : FAILURE in 17m 16s","accounts_in_message":[],"_revision_number":4},{"id":"8c3f28a408804dc766e7ae86a0eafdd0ef3730f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-26 15:45:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909000 : FAILURE in 1h 09m 29s","accounts_in_message":[],"_revision_number":4},{"id":"bb62e0ebbe4c04408478c9747e377bfaad54962a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-26 15:47:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/05d0812abd454770a940fdc2185a5121 : SUCCESS in 1h 11m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f7892805be8c407c8fd1a39e69860dc4 : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":4},{"id":"1361eac55b345d27d2a0571e8388f11eb2437145","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-26 16:19:41.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_Unity_FC/7678640/DellEMC_Unity_FC/1584 : SUCCESS in 1h 34m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_Unity_iSCSI/7f77d12/DellEMC_Unity_iSCSI/1616 : SUCCESS in 1h 44m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"26dc6c54978abce3fc7f342e321982654c795c75","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-26 17:07:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909000 : FAILURE in 1h 17m 27s","accounts_in_message":[],"_revision_number":4},{"id":"5c28d67a12a9fe7ae9620cea59408a2fa3787803","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-03-26 17:10:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/909000/4/check/fujitsu-eternus-dx-iscsi/e17a058 : FAILURE in 2h 34m 49s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/909000/4/check/fujitsu-eternus-dx-fc/2e189a4 : FAILURE in 2h 22m 19s","accounts_in_message":[],"_revision_number":4},{"id":"20cd78180772278aa1c899159e61a219e954f6b3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-26 17:27:09.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_3.6/6ba7f2a/DellEMC_PowerFlex_3.6/1094 : SUCCESS in 1h 50m 38s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_4.5/4ec1388/DellEMC_PowerFlex_4.5/72 : SUCCESS in 1h 58m 21s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/c64f4c5/DellEMC_PowerFlex_4.x_NVMeTCP/8 : FAILURE in 2h 51m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b062d7e91e2fc6007a4767a1220742f2368ca461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 17:51:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/593ca1df9c864fc5b48a024e59ae3a1d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac0970b181f0499b8009d899d2e3b199 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff176b9654cd4a8e91199a271aad029c : SUCCESS in 10m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4d7d09e7d624334af09213059c1e2f5 : SUCCESS in 10m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2d217193deb402c82458b298180dfbf : FAILURE in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e5de39e91646289e8dc7f6f167a1a2 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/07eeac3be31947a48cc69a2cd7205b9b : SUCCESS in 56m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c0603ec2143c48b89645ff83675975d3 : SUCCESS in 1h 04m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0f5ea49eca94c02953ae82c0676bae1 : SUCCESS in 1h 25m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/601dc2862364470790223b8b156e0767 : SUCCESS in 1h 02m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f192fc43fe5742e48b5982038eed2f32 : SUCCESS in 13m 31s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/46cffb2c06d44209ae4fa1110838e64a : SUCCESS in 13m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11f62d5488a0458c9881f4e0410768c8 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17eb1bb572d74433b173387673fcdb47 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca5c5007fa924ba49a4ce2cfa1f81e4c : SUCCESS in 6m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/331567ba094846a1aa5759deb18cf874 : SUCCESS in 5m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/104e5a06202d4c18b20a5c9873b9a0ce : SUCCESS in 42m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/99a304e2b7714d2e84725873557a74a9 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0faef423ce384927a5714082695ce2ef : SUCCESS in 2h 07m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3e15c0b33de849fda88d5b3197113648 : SUCCESS in 1h 50m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1ef7257cb1d42049038b66562f264ba : SUCCESS in 1h 52m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/55872e5cd61946fead198726279fbc89 : TIMED_OUT in 3h 06m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bbef5e0e7e8245289a44bba0fcadac59 : SUCCESS in 51m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c34d7955599f41dd8da5212dce291728 : SUCCESS in 1h 07m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/157f29ee832d4d9d973ce121e0ea410a : SUCCESS in 1h 18m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3d7ace4361624d2898c5e82982e696e2 : SUCCESS in 45m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4efa1f0b21174ef0afe0eb0d09613179 : SUCCESS in 1h 20m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/37dd4071e0ca410ab07e280df90adc03 : SUCCESS in 1h 34m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c83e9a89e7fe4fc39b4a18faf1ec6b6f : SUCCESS in 2h 37m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7510d4c1bb9401485fb2971320854a3 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":4},{"id":"5a6fea7e64c2a2cae2974d24a3626d2badca28a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-03-26 17:57:54.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_03_2024_17_51_43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"2cf6a456de2a33b8ed4811bc566c65bde7e8cbcc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-26 18:33:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-fcp/1f29ab8/ : RETRY_LIMIT in 3m 35s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/793bda8/ : RETRY_LIMIT in 3m 25s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-nfs/a9afe8c/ : RETRY_LIMIT in 3m 30s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/860f701/ : RETRY_LIMIT in 3m 32s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3c70b1e1bc666b91bf03409f0c99ec7e50b7a019","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-26 19:29:52.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_Unity_FC/07478b2/DellEMC_Unity_FC/1585 : SUCCESS in 1h 29m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_Unity_iSCSI/40c3b15/DellEMC_Unity_iSCSI/1617 : SUCCESS in 1h 37m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"4723c44b9c2c6f3269fbbd54bb69cdf60bce97e6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-26 19:33:03.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_iSCSI/8d4d5b8/DellEMC_PowerStore_iSCSI/1208 : SUCCESS in 1h 16m 16s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_FC/7b5155b/DellEMC_PowerStore_FC/1378 : SUCCESS in 1h 37m 22s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_NFS/6e6abc4/DellEMC_PowerStore_NFS/896 : SUCCESS in 1h 03m 55s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_NVMeTCP/7ae89b2/DellEMC_PowerStore_NVMeTCP/759 : SUCCESS in 1h 41m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"ca0f19c6e538d64b122398be75422fcd1db3d1e0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-26 20:43:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_3.6/b074ed1/DellEMC_PowerFlex_3.6/1095 : SUCCESS in 1h 54m 05s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_4.5/ddbb76d/DellEMC_PowerFlex_4.5/73 : FAILURE in 2h 05m 22s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/e44fd5b/DellEMC_PowerFlex_4.x_NVMeTCP/9 : FAILURE in 2h 52m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5c75caf899962e716b52e7c7f1b50f842b418965","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2024-03-26 23:01:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerMAX_FC/89cd17a/DellEMC_PowerMAX_FC/365 : SUCCESS in 4h 37m 03s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerMAX_iSCSI/9f7191c/DellEMC_PowerMAX_iSCSI/386 : SUCCESS in 5h 09m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"b71ce1ec456c3fa46309f6c6d387ba7144a6ebeb","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-03-27 05:31:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/909000-4/HPMSA-iSCSI/ : SUCCESS in 1:53:28\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/909000-4/HPMSA-FC/ : FAILURE in 5:33:06","accounts_in_message":[],"_revision_number":4},{"id":"da1f8b8e7a558d172e9a2d6f180d611a34c0a984","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-03-27 09:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/909000/4/fc : SUCCESS in 0h 38m 09s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/909000/4/iscsi : SUCCESS in 0h 55m 46s","accounts_in_message":[],"_revision_number":4},{"id":"609a64cecde606cd8de5744ea04f2762a6b9e975","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 10:35:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/4/20240327145640/NEC-ISCSI : SUCCESS in 4h 11m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/4/20240327145640/NEC-FC : SUCCESS in 4h 38m 53s","accounts_in_message":[],"_revision_number":4},{"id":"df3f8b3b6aca2df46f4198e50493c793c57ebf79","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-27 10:37:03.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/4/20240327084617/NEC-V-ISCSI : SUCCESS in 2h 54m 16s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/909000/4/20240327084617/NEC-V-FC : FAILURE in 10h 50m 43s","accounts_in_message":[],"_revision_number":4},{"id":"cdc976a6ad03cb8ade28e1e101623b4a1dd2d862","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-04-19 14:48:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"daafdb279ec4daccfa61116282ec6c6f4e9458c1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-04-19 15:08:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3fd9b736409d45a30148ed0f6264c7270c56ad33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 15:09:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f1edfd23ee650ab88fce6f1ed24f5c46c5297435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:32:33.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a1fd1ca25d444689119f342f619fc6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7676fdd1f02f4e2ba29a7c66f703803e : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0813a03c55f489f8adcc05e5abb2e18 : FAILURE in 26m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e634357b7541480fb58df94183bae532 : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2217c62aee49f6a05ac84b942045fa : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e11bf04f6fd94ca0851926814b30ad4f : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7041193413f04a17b042c488c6cf8ed3 : SUCCESS in 1h 19m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/994b0aae7d794b3b8b8ebf3b0771532c : SUCCESS in 1h 05m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29a78342c83a4aa0a60875b38d7e10ba : SUCCESS in 1h 04m 55s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63f998c54fdc4858949ac3d12d36f9b0 : SUCCESS in 1h 50m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/982dc2a9fbae42c5aa0323bfeeb86641 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":4},{"id":"466907641010f37c38f29b8f4c07d562c3ed3d9e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-23 19:15:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c3b6b9a1cb4bee4c5c9fa46cd406f9964fa8b829","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-23 19:16:09.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"d07ed13d06ffa26d1e5edc6240936e22dbcb90f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 19:34:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a674885c6d043dbb86e31821c6eec88\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f07c1c4454a474c9a930560da545a78 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa5d8bffd34e4f2da3376a6f7caeaaa6 : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1ee544907e12dd35797f671611db3b92b537da71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 22:06:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cc9b52920494d1da7cc691d6edc97ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cefa862301a4eaca1446d315d84f741 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/177134ecfc7a4b2ebf607ad25c42103b : SUCCESS in 8m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e46eef3e37b4002a2f0faf85fa18f95 : SUCCESS in 9m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dac8844f69fd409484c29694729ebad2 : FAILURE in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20af6c0ee0b4abeac65df7fd7a633dd : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/99ddfd3f013a4756b2409e259e8f99f1 : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/077dd86ed86d4a3b968a5e5d76e0fc73 : SUCCESS in 1h 34m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00b01d27996e47ba92d9af6673dabf0d : SUCCESS in 51m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e3570c7e9074fbfba8433b836c67634 : SUCCESS in 17m 21s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4690386c01a9438e88be3ef81949121d : SUCCESS in 12m 47s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/87260dcce9c24e108d3eeb591fe9d63b : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a586486d88a4aea9d7a854d8e8bd26c : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/87108dd1330e4b0a9ce46a7b021d8166 : SUCCESS in 6m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6d97081639ff4fd09fcd53337fff5463 : SUCCESS in 6m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4007bb65ad114627bd29f524c5dda3ff : SUCCESS in 43m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bff1b6ff0a1d466da79f1df576b572e4 : SUCCESS in 5m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8bb89786922e45b0a42b1e6f0b83c3ce : SUCCESS in 1h 39m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ffb4d89f2f7b4a8dba90979b0ff9734e : FAILURE in 2h 44m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9e510f71607405686e1565a6f32fe25 : SUCCESS in 1h 55m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2c1a1ff9bb694b659a733ead9f27527a : SUCCESS in 1h 54m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/81209996707f41adb5baada318ce0b56 : SUCCESS in 33m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbacafc9b34440fea6ef60f8bdd72f61 : SUCCESS in 1h 11m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/955508f3113d455c8cce338a44cf3846 : SUCCESS in 1h 33m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1470abed20764d69b641329ffe5979a4 : SUCCESS in 43m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb2e9333e45f430182298b04b261d09c : SUCCESS in 1h 46m 08s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c8d97e73fc1c498b8a58a16c1573499b : SUCCESS in 1h 18m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d204b780ff64e6eb9b40db9c3a180c5 : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0cbb673a0034601abc74fe2525d9658 : FAILURE in 59m 25s","accounts_in_message":[],"_revision_number":4},{"id":"58610ef1ea78b059e5681a848f58edf9c0840008","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-24 11:52:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a05d1bc5bd83b3fd2d497c38b907bd72e610d82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 12:12:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755f100ef97e4e64940e5c9118e24638\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26861837d1654ca4bf0b45e1d0083aa2 : SUCCESS in 18m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb935914cfae413a85de6fb93719cab2 : SUCCESS in 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e260ae56c11f6c2bac30d550243af2154a24320","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-04-24 13:25:08.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d2ccf1ea6be94733acfa3964ec8db22a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2c5e0e0395b2436692a795393f9ccb15 : FAILURE in 1h 26m 14s","accounts_in_message":[],"_revision_number":4},{"id":"da9a039c0bbc07bd0f3483d19a44d71f60089328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:20:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b7354beb9384bdab201437892f51f58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9fcffeac4946798d7ac297237ae169 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b10007e2b724f4592b1f069987eb92a : SUCCESS in 11m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45be6e35b1f54814b4a49b499ce8f8f0 : SUCCESS in 10m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b28e5f43ca774693be45c692059e821f : FAILURE in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3552696be7c4ef9bc1e35b8056259ea : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a8dbfbbf0ac94bc98f2c94c9018e2e90 : SUCCESS in 56m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d493b07d20943a2bbaac6f8021062e2 : SUCCESS in 1h 29m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df7eaee5f70349c0b9f0044740a66e95 : SUCCESS in 59m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f7c107f1925340d0b74432349b691d18 : SUCCESS in 16m 02s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2c97aca9ac5c400ebd952f2a047997bb : SUCCESS in 13m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/77d2ef65c06b40e89fe3763413c7bf72 : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5e358c44c6442b095496ccb735ae87b : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f1c4db8bf4e466aad4200b2472b18ae : SUCCESS in 6m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1be8a4bcde7e4668a0c0da8648696bfb : SUCCESS in 7m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/01ae86ee19bc446882384de9d8b17d5b : SUCCESS in 34m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0ed459e734a4a0883fb4055bfab2f39 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47712ebb775f44c9acfd7c87a9cf255e : SUCCESS in 2h 17m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/687c36f65698423da51510db446e2916 : SUCCESS in 2h 21m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/feb1a07509564f44a5335ec55e49eed4 : SUCCESS in 1h 36m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/28f3bf20e02b43eb97cb38959f4e82af : SUCCESS in 1h 58m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3b786a8a0dc24889bd05327c5d4851ef : SUCCESS in 35m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c21015d6950648e38234827ec08a8f56 : SUCCESS in 1h 12m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c2bcc070c11c4ff3b56427f932946c31 : SUCCESS in 1h 27m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/72749dfba6b34b22b4ac9b95b90021ad : SUCCESS in 36m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ec277dadca3d44e5882ddb53bbc568ce : SUCCESS in 1h 11m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2741d0f9ad074ac7853ea4646d2be3ae : SUCCESS in 1h 25m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57c367ad5b074437bf1aa69dd1ddd7b4 : SUCCESS in 1h 12m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d877b7af6ffd470a89443d9b2bb6b8a6 : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":4},{"id":"43116bd5746611e5093ecfa2547a90dfe9a92207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:21:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2a1c4c05248b16743e36c2b3503220b3c35c20f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-04-24 14:27:12.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_04_2024_14_20_45\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"925655d738a325b76b83a0cc442b3ebf9022181d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-24 15:08:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-fcp/c92915d/ : RETRY_LIMIT in 3m 49s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/e13dbc9/ : RETRY_LIMIT in 3m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-nfs/fc0e456/ : RETRY_LIMIT in 3m 43s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/909000/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1b0b043/ : RETRY_LIMIT in 3m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"0738171970bb641045d71888980fb64ea6ba1053","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-24 15:59:16.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_iSCSI/b08dbb0/DellEMC_PowerStore_iSCSI/1267 : SUCCESS in 1h 12m 48s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_FC/cb02dd8/DellEMC_PowerStore_FC/1438 : SUCCESS in 56m 12s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_NFS/965160b/DellEMC_PowerStore_NFS/957 : SUCCESS in 1h 03m 24s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerStore_NVMeTCP/e61e51b/DellEMC_PowerStore_NVMeTCP/818 : SUCCESS in 1h 38m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"923f872f3fc4e9692626ab49c7d3228eb5fc6142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:56:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c0b0bdbb5141f8a68621463caa38b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83651bbdfee3478eb2d4c3a275e0dbd7 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e92bf52ab4fd4057982ef9763c9e16a8 : SUCCESS in 11m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad5cb8a428a5422aa3314c16c98ddc06 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ce8f87d39c464eba7e9078ffeb1c3b : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/1dd7809282da46b498220ded51663e5a : SUCCESS in 55m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/946f7e230e9e45d5897b32d6497c631c : SUCCESS in 1h 29m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c6e641b348f4b7f8addd29302f460be : SUCCESS in 54m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6c7e1cdd4b5f475c9a2d3c544b08cd3f : SUCCESS in 1h 09m 41s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10bc4d8e35204ce1964934e963889aa1 : SUCCESS in 1h 49m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/777bc5cc29454d5fb3920222b8c729fe : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":4},{"id":"30dc216d9dc8f80dbd3c5795950dee7878d56d36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:56:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"822b200c71f31470e04e6d3e93646269920ce3a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:58:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed966b0b5de54261a02c33f6be511fd6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424ff20577014dd4ad31a6ced0d875a3 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":4},{"id":"57509476f14e69a288a77e45406a06a3ceab6a50","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-25 08:43:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_Unity_FC/bc48395/DellEMC_Unity_FC/1763 : FAILURE in 1h 35m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_Unity_iSCSI/4051a3d/DellEMC_Unity_iSCSI/1792 : FAILURE in 2h 16m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d8dbf2bb7ea8d924285bc9fa78419218625c9919","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-04-25 16:30:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/909000-4/Lenovo-iSCSI/ : SUCCESS in 2:09:06\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/909000-4/Lenovo-FC/ : FAILURE in 3:33:30","accounts_in_message":[],"_revision_number":4},{"id":"3cf03a0080c6f8af87a8eff3d5c3d71830c5c5c1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-10 05:36:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_3.6/524d580/DellEMC_PowerFlex_3.6/1264 : SUCCESS in 1h 44m 42s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_4.5/3ebe1aa/DellEMC_PowerFlex_4.5/199 : FAILURE in 2h 04m 00s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/00/909000/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/f514f11/DellEMC_PowerFlex_4.x_NVMeTCP/139 : FAILURE in 2h 54m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e958d746f56cdc27d8241da59cffd1723e04e387","revisions":{"d67940aa297773e17612eebe1537ab85de7679f9":{"kind":"REWORK","_number":1,"created":"2024-02-14 15:30:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/00/909000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/909000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/909000/1"}}},"commit":{"parents":[{"commit":"fa19d4a8662b671c5c75e79b775d3b4417ab8695","subject":"Merge \"Remove six from test codes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa19d4a8662b671c5c75e79b775d3b4417ab8695"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-14 15:25:31.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-14 15:25:31.000000000","tz":-300},"subject":"Mark nexenta_encryption option as deprecated","message":"Mark nexenta_encryption option as deprecated\n\nThis option doesn\u0027t do anything, was used by a\ndriver removed in the Train release.\n\nChange-Id: Ia241027e0fb31c00e80d088f11ab50c375407519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67940aa297773e17612eebe1537ab85de7679f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67940aa297773e17612eebe1537ab85de7679f9"}]},"branch":"refs/heads/master"},"a1abee72b551dc3bd09fdd995f208fc1e9309880":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-01 16:12:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/00/909000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/909000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/909000/2"}}},"commit":{"parents":[{"commit":"ba424aae9241551433f80c15555f815ba9c9d8ac","subject":"Merge \"Dell PowerFlex: Added timeout into rest API call.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba424aae9241551433f80c15555f815ba9c9d8ac"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-14 15:25:31.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-03-01 16:12:25.000000000","tz":0},"subject":"Mark nexenta_encryption option as deprecated","message":"Mark nexenta_encryption option as deprecated\n\nThis option doesn\u0027t do anything, was used by a\ndriver removed in the Train release.\n\nChange-Id: Ia241027e0fb31c00e80d088f11ab50c375407519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1abee72b551dc3bd09fdd995f208fc1e9309880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1abee72b551dc3bd09fdd995f208fc1e9309880"}]},"branch":"refs/heads/master","description":"Rebase"},"005992cc76e7bb8f044ac5682ed3153df6cf26ea":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-13 14:08:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/00/909000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/909000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/909000/3"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-14 15:25:31.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-03-13 14:08:23.000000000","tz":0},"subject":"Mark nexenta_encryption option as deprecated","message":"Mark nexenta_encryption option as deprecated\n\nThis option doesn\u0027t do anything, was used by a\ndriver removed in the Train release.\n\nChange-Id: Ia241027e0fb31c00e80d088f11ab50c375407519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/005992cc76e7bb8f044ac5682ed3153df6cf26ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/005992cc76e7bb8f044ac5682ed3153df6cf26ea"}]},"branch":"refs/heads/master","description":"Rebase"},"e958d746f56cdc27d8241da59cffd1723e04e387":{"kind":"REWORK","_number":4,"created":"2024-03-26 14:34:54.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/00/909000/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/909000/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/909000/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/909000/4"}}},"commit":{"parents":[{"commit":"0581e8cbd6b28724e75fc5b48c29c07e0d910fdb","subject":"Merge \"Remove six from HPE drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0581e8cbd6b28724e75fc5b48c29c07e0d910fdb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-02-14 15:25:31.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-03-26 14:34:14.000000000","tz":0},"subject":"Mark nexenta_encryption option as deprecated","message":"Mark nexenta_encryption option as deprecated\n\nThis option doesn\u0027t do anything, was used by a\ndriver removed in the Train release.\n\nChange-Id: Ia241027e0fb31c00e80d088f11ab50c375407519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e958d746f56cdc27d8241da59cffd1723e04e387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e958d746f56cdc27d8241da59cffd1723e04e387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
