)]}'
{"id":"openstack%2Fcinder~909242","triplet_id":"openstack%2Fcinder~stable%2F2023.2~Idec29ce596438a7969ea32d48bacfede84f5b0b6","project":"openstack/cinder","branch":"stable/2023.2","topic":"fix-cinder-backup-tls-swift-stable/2023.2","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-04-26 15:23:31.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-04-26 17:05:41.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-03-27 18:52:43.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-04-26 17:05:41.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-02-19 15:48:11.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-04-26 17:05:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idec29ce596438a7969ea32d48bacfede84f5b0b6","subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","status":"MERGED","created":"2024-02-16 18:33:45.000000000","updated":"2024-04-26 17:07:04.000000000","submitted":"2024-04-26 17:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909242-fix-cinder-backup-tls-swift-stable/2023.2","cherry_pick_of_change":907494,"cherry_pick_of_patch_set":1,"meta_rev_id":"3ee6d6ef178e59dbb90afdfce53313fd73fe9299","_number":909242,"virtual_id_number":909242,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-26 17:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-04-26 15:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-03-27 18:52:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-02-18 02:54:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-02-18 01:25:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-02-19 15:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-21 14:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-04-26 15:23:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 18:34:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-16 18:34:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-02-16 18:34:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-16 19:36:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 20:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 22:17:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 23:26:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-17 01:45:08.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-17 03:43:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-17 04:10:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-17 08:46:27.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-18 01:25:48.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-02-18 02:54:54.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-02-21 14:53:19.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-03-13 19:39:42.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-03-27 18:14:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"ac67ca02baa0ac65a3b6a9bb86d2e407a8b79733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 18:33:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9fc8c6e70b8c4fa9980ea9969ef9ec02645979de","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-16 19:23:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76497dadda0e229e7314f4d663d1dd5d9e1cb848","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:36:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba4e8d97f56a473a82559b960718b997\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c661bbf2afc420fa2709ad7c439a47b : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0df58db451b043c0b71e156b07f4995c : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62fa9cf811534d8082d3a12bbc7f4bc4 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b3dae125a0443b194c578d73a4a2715 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60b330e2672b96ac310a740e062beeae5929cf6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 20:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20ac975ae2ca483aa64b51ee1247ae9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da1fb2e1dd5f48d688e9957a6496e9f8 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a536c18b0a8f4ef79e15e9dadb969c7e : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a85a99a305043f9ab4d9e7b57e921b6 : TIMED_OUT in 40m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9302ad84bfff4ba890aef3552e7ae440 : SUCCESS in 10m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36bb313126b74aef81081bc021c0c9d6 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7876b075834b51ad6745f5554875c2 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/f000a719c41943bc896ade6cf4e63563 : SUCCESS in 49m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71b2cc7de598424497487002067a01d9 : SUCCESS in 1h 47m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8864bf9f67bd4f3387dad32a454f9919 : SUCCESS in 1h 05m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c3233884c6e840f69fd72a377951104d : SUCCESS in 16m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4dc800102b94819a3005909a2d50ccd : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a79e777ea004b08a356afc7131ed5ba : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b45211ca07342a0a3901d2c74c86cfb : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b32c00634694dfd820ab7f789216052 : SUCCESS in 7m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c3310cdbeaf4fadab938a2cd98b6cd1 : SUCCESS in 34m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e4ed3a8513e40ebad4c0bd0dac93f1a : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6962b13b086423da3dd4631af2a3056 : SUCCESS in 1h 52m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/18cf05b394384852be1b3324bbdc22fd : SUCCESS in 2h 00m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b905f9fbbd0f4cf78d5830fd8e076362 : SUCCESS in 1h 41m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d740086139dd489c99efd6a18405efc2 : FAILURE in 2h 06m 08s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/edf7aa6c1937486cb433a11095784131 : SUCCESS in 1h 11m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/514d15e21f114f59a1c50b0749daa22b : SUCCESS in 1h 20m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6bd8fef149934d04851ff91558a612b1 : SUCCESS in 37m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8fb956cfa7474da39155f7846ab9cdb1 : SUCCESS in 1h 15m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a89d3bc3ed9d44d2bda20ba3437bfa74 : SUCCESS in 1h 37m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/30d5052965e0449e8c7df345e8dcb682 : SUCCESS in 1h 10m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c6b51badde24a478312bd35ac2d050f : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bcd3da6fa970a00bad82c2ef02887802540956f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 21:28:40.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"76e74561349eab29ad31f1a9c443e05b258a8195","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:49:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d0eeacbd7649afab62cfd92e123f86\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d0969e9abdf432f8b18bee747f80d60 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a232cb5af3e47f79e3436387c5da62b : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1cfb6c6dc2f24cb9974be8555baea9b5 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8740c9e0c903429c88628b86c0edcbf4 : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d7714ad3982d87237d662425f7f34e9fb4f23e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 22:17:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"d24bd4d97499954719cb0e1abb28522c153cb232","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-16 22:54:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cb637b0c99c7a50fa97e7eaad86cf563a0fe5327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:20:01.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d664a13cd85ee3e09073ea7b6799d5278cd3d845","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:21:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"671eccf51edf91d0317886474791c13a3efb798c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-16 23:26:03.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"fe17d496e6367405b2466738b6dc889e0f0a65cb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 23:26:04.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"583516c43f403a96f298fabf0184c2dbd3e75ffb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:26:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"48b0d08570681c860738d58bb5ce5112cb60a06d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-16 23:30:30.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"d3893ffb7b56520384d7ebd4e85720f61daef4d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:31:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6332bd6f45d246a0a65af625f84311e4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d14ad52950534b038ccd0ee68a228b8d : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e9a3b8805824a8390312a86104c2669 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9dadd13e83a440e6a696560be11b3cbb : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8adef6fd8e5e465083c70574d3a2f5c5 : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8fdaf8e70378aa4fb0596ca6d139c1a8dbb052dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 01:41:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e050ae0c3cf4a86ac63963ec8512e33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5632d0abd6a04b4082a837335e8e1521 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f49c6e4851aa43fe8540b9c5f5ebdf41 : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0983d1ced5934289b1c67f3aecba2546 : SUCCESS in 11m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24357d8b4a7a44ebaffede2699c574b4 : SUCCESS in 13m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b57c84381084fd3bf55079792f4eff7 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6f2b5084754468a23649d6be8473f9 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/80cdc259ce5a4cb59d436a2518997052 : SUCCESS in 58m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7197643ea014494fbe662be4baaee73e : SUCCESS in 1h 27m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63087d33340d4a40a76ea2948faaf716 : SUCCESS in 1h 04m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e48f40dbe594321acc9807f61fddf7e : SUCCESS in 13m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cca8ade8d634421a8d6f6407a9d4c9ff : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/61e9a6fc326c4934b09462345a370b19 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/199670ac63204f12b230473b14b7c44c : SUCCESS in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4054978aeb584cfdb55cd92a661a1302 : SUCCESS in 6m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b62ebbd8a3924671a9f4d008c5584e1b : SUCCESS in 44m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b629ca82905d4805882104c9cea395f2 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc8352ffa53c4a718e15ae4309ff7994 : SUCCESS in 1h 42m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9145ae868ad14fc0ac49b67207641c7e : SUCCESS in 1h 51m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/375c565b983b48c8bf8b0561df9cf657 : SUCCESS in 1h 48m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4f02189c5984439199b3809a65a13029 : SUCCESS in 2h 00m 53s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8fd7d277383414aa8abda122ad5e000 : SUCCESS in 1h 12m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9f5a86200eed4b36b6baa1b88333087f : SUCCESS in 1h 29m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1a20d76da9e1473dbeadf3f60713e0c1 : SUCCESS in 43m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/285f9596b7cf44c4abe471a222554848 : SUCCESS in 1h 19m 35s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8caf142a95ae4fd4a4c9ad1d9e61ebc5 : POST_FAILURE in 2h 15m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7919a793334a42b5a74ddea3ac6fe62b : SUCCESS in 1h 24m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbc64b4bb2bd44f396720c99f0ee2d29 : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":3},{"id":"82611968e6e40784e47a7a53e2a5535ca4ea26ed","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-17 01:45:08.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6c5915235eb244e686fc71460ba051a3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1b4e3583ed7d4da0b26e81286186d741 : POST_FAILURE in 24m 11s","accounts_in_message":[],"_revision_number":3},{"id":"b64c7fd0b1be61603178c044524cc5c08565d743","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-17 03:43:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/730/.","accounts_in_message":[],"_revision_number":3},{"id":"e565ac5f8d601b9f7a440a2332c5adac5f17f8d8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-17 04:10:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/909242/3/upstream-check/cinder-tempest-plugin-ontap-fcp/df06ca7/ : SUCCESS in 1h 04m 13s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/909242/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/9548a68/ : SUCCESS in 41m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/909242/3/upstream-check/cinder-tempest-plugin-ontap-nfs/e11ecef/ : SUCCESS in 2h 06m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"a090d2f70baeaacca039e5b9541de5b4666b2b5d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-17 08:46:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/909242/3/ : FAILURE in 1h 15m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1ed25ea46cf10c17ffe72ed6ad5d63f8271fa307","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-17 09:10:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/725/.","accounts_in_message":[],"_revision_number":3},{"id":"9893105eb2f29b373d6079b3589dd03ae77223da","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-02-18 01:25:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a137c33f5981f3d76694b03723eb3fc476046279","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-02-18 02:54:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"de5f10eec41e817ca963722086f284f8763c7e0d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-19 15:48:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"001940220b9c2561f29709f5dc6afaa86d0f2bb3","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-02-21 14:53:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d170136e434df3a3e9942deb9dd364e0a5bbef8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-27 18:06:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1a0f04a555a73ff09b43f7175e009e729a53116d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-03-27 18:52:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be45e78e34c7d845d8d0df2cbbae678a4c172fd4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-26 11:11:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9dedbe8bb38ba333fe58ea80c682661baf8bbf24","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-04-26 15:23:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0a90ba16c5b9b9966cd180bd47d641a7bda71ea3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 15:24:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5f2f61c3533757ed59c382ed6c1ccae6315c4e41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 17:05:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef55063d28e4f39b9dddda048379714\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3675aa93af5048768360dfbeda3b45fd : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc07de603e184e5c9502ecef6e62c931 : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/160c73993256421e861c0709033ae5e6 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbfc831431f4493791d18f39b0d9abef : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c349663e64c842f5b6d003adec01f1d6 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/0ba43295f4ac4e529b4f6281233cec63 : SUCCESS in 1h 11m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/600db04653324f86a5098dc8d39782f8 : SUCCESS in 1h 17m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f11837b6b454cf090a62f122a858367 : SUCCESS in 1h 00m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/833e15cf56a441d98a71f9f98e3e7d8d : SUCCESS in 1h 07m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/181b7cdb90da4c8090235b928f3761b0 : SUCCESS in 1h 41m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2ca31eae91646cd911262d0781e2728 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":3},{"id":"f1c895b488a448a7312bb08d07d10303c6d79734","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 17:05:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3ee6d6ef178e59dbb90afdfce53313fd73fe9299","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 17:07:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c193d5953ae740a892c070bd2db16101\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc3a93832c54b299c279e9b01168262 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1e52602646c8b7cdc616659777563847fe53ada0","revisions":{"7abb15d226a1869d04d01c6f5a0201cc1b077e5a":{"kind":"REWORK","_number":1,"created":"2024-02-16 18:33:45.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/909242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/909242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/909242/1"}}},"commit":{"parents":[{"commit":"34338cf26f9d070ad8106fceccd657f613c4d5a5","subject":"Merge \"Skip sparse copy during volume reimage\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34338cf26f9d070ad8106fceccd657f613c4d5a5"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 18:33:45.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7abb15d226a1869d04d01c6f5a0201cc1b077e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7abb15d226a1869d04d01c6f5a0201cc1b077e5a"}]},"branch":"refs/heads/stable/2023.2"},"03fb86e951bbee36ff5e2703ecd67728fe2ed6f6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-16 21:28:40.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/909242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/909242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/909242/2"}}},"commit":{"parents":[{"commit":"34338cf26f9d070ad8106fceccd657f613c4d5a5","subject":"Merge \"Skip sparse copy during volume reimage\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34338cf26f9d070ad8106fceccd657f613c4d5a5"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 21:28:40.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n(cherry picked from commit 7abb15d226a1869d04d01c6f5a0201cc1b077e5a)\n(cherry picked from commit 3403639ba58c27a741e4855c507a0680f6a55a4e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03fb86e951bbee36ff5e2703ecd67728fe2ed6f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03fb86e951bbee36ff5e2703ecd67728fe2ed6f6"}]},"branch":"refs/heads/stable/2023.2","description":"Edit commit message"},"1e52602646c8b7cdc616659777563847fe53ada0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-16 23:20:01.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/42/909242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/909242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/909242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/909242/3"}}},"commit":{"parents":[{"commit":"34338cf26f9d070ad8106fceccd657f613c4d5a5","subject":"Merge \"Skip sparse copy during volume reimage\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34338cf26f9d070ad8106fceccd657f613c4d5a5"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 23:20:01.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e52602646c8b7cdc616659777563847fe53ada0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e52602646c8b7cdc616659777563847fe53ada0"}]},"branch":"refs/heads/stable/2023.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
