)]}'
{"id":"openstack%2Fcinder~909244","triplet_id":"openstack%2Fcinder~stable%2Fzed~Idec29ce596438a7969ea32d48bacfede84f5b0b6","project":"openstack/cinder","branch":"stable/zed","topic":"fix-cinder-backup-tls-swift-stable/zed","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-04-29 11:21:04.000000000","reason":"Change was abandoned"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-04-29 11:21:04.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-16 23:23:05.000000000","reason":"Change was marked work in progress"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-04-29 11:21:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idec29ce596438a7969ea32d48bacfede84f5b0b6","subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","status":"ABANDONED","created":"2024-02-16 18:36:47.000000000","updated":"2024-04-29 11:21:04.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"cherry_pick_of_change":907494,"cherry_pick_of_patch_set":1,"meta_rev_id":"b6c25fcc34c29f2fc418e3271ee054e380aa0870","_number":909244,"virtual_id_number":909244,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-17 13:49:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"reviewer_updates":[{"updated":"2024-02-16 18:37:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-16 18:37:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-02-16 18:37:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-16 18:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 21:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 21:57:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 23:15:24.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-16 23:43:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-17 02:34:33.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-17 06:58:32.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-20 08:35:31.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-21 01:02:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-29 11:21:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"0168418c715d804577d140e37309f9f8f401e509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 18:36:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aaa489496df735d4090cac5c0b35de385dd78c6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b626d0114dc5486191f397214e21e06c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f922f610be3b4e15909e8f04b50a6ac7 : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cac493124d4747aba7a041dc909f9698 : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a71c0ee2065dded3b63da7af1333cfdadf6d69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:49:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739ac85165d6496698ca5a24c6815539\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9b7cffe0794907b5f0eb3db0330c8e : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d01d6760b34fa184034f95f0b2b4ec : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8914a1fb88440328f9bf715fb1b99d7 : SUCCESS in 11m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f546de8cb10948898f0929f293a3106d : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f56c79ed4e4686b444a462969349bd : SUCCESS in 12m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/817850a9710d42819f6fe3025ce67b54 : SUCCESS in 13m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/855a23c956d94ce590610cd719140715 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d015237c67f84951b50dee5d521c2a68 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3e3a0a9a8f84fa19c58263e0498be49 : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eadcb2ae6e654477ad38fe6b68c95190 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b8103e438b114abfb0ea09b81342b3fa : SUCCESS in 28m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9832e1c535c4005b1232325a58c4f25 : FAILURE in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1d35d392decd4476ad4087b83230bbc3 : SUCCESS in 1h 36m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e5b4781d65464e8388ba1c50389c51fb : TIMED_OUT in 3h 05m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/15a0e6aa91974c7989ab16b3a8c40d8b : SUCCESS in 1h 41m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9becdab852b8467388b7a0448593b539 : SUCCESS in 1h 21m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cb719a3f80364c05aab1c2ebc274f941 : SUCCESS in 39m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3e950ae79cb64219a690d775d2b920e0 : SUCCESS in 1h 18m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b9beff1f0334fc2b769ddbf50f14fb5 : SUCCESS in 1h 15m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1446b0d448954d4289beccbec871bd12 : SUCCESS in 1h 09m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3ebe4a3d46410f9a48bb57f8a252a3 : SUCCESS in 49m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e48621cac63400cba256e552e0fba1c : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"605a06a79d6e4d727183548b22b035fba5eada07","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 21:57:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ec876b234cd3ae1448aebec9c44f7f9ff7a4fe09","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-16 23:15:24.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1352cd26ec8f404eab5a428135cfe4f4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c65b9c8e9dde4932a0c07a08d968c501 : FAILURE in 1h 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5796486c35cecc247b687cf6d59f58b718db3183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:23:01.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9ae5c6942cc41c3cce9c78f784d242e9a8720121","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:23:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"2c208d293d0dfef26e09413a1a6984066ed19d16","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 23:34:17.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"56d4f403127148d59ed62864047c945c10a18932","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:38:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a99e1e09555941f99c62efe77966b2ee\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5d0d1dfd72348fe848e9c52e78fa96d : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11b745fa6d9f486ab0303a3c6fc8adbd : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24602be5310863e8347093df1f16abbf49402b71","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-16 23:43:40.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"e1bef53047d156a877cc3516867746338076ee5e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-16 23:50:48.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"9dceab3dfae0daa3b6a625b413d7fee678ba2542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 01:23:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd04899f9d42499e9f0220bf8883f588\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befaaea01d334301bd1d972c376227a0 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d235049150e486bb7b092ca13c8858e : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6545c2e289245a0ab74e56106310909 : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/495def79b16c4ae7abe21a6288bd431f : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de385d31f4324cc79f42cad2c752e25c : SUCCESS in 11m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/205eebc6e4ed4f6cbe832b8afce90dc5 : SUCCESS in 14m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6d6bb547b7774703ab09e8e581e5bb7c : SUCCESS in 6m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76e02dbcb1544e268f812d2bca80cd93 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd48a71b191c41b6a9d472fa9ec1bbae : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f70e195ac1e443f2892a3d933136f10e : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e7ba2b5d40f4ecfb61c3aa96d871236 : SUCCESS in 33m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e22ff657faba4ab4aedd7b2f8e0fe106 : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ba3ff8ab53740238aad9d7aa3d6b8f7 : SUCCESS in 1h 19m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb9a47189c7e42508a0c09182ba332b3 : FAILURE in 1h 15m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a97f46b9c474359805ab9837f76a0f4 : SUCCESS in 1h 50m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2691e1adb5cc40de8afe2b1d491370e7 : SUCCESS in 1h 18m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3ae196ace2fa47fb97abe995fb196669 : SUCCESS in 31m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/95136157b08e4df58a896c72eb19f1d3 : SUCCESS in 1h 04m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79f2138e5a6e402abe272f235409f579 : SUCCESS in 1h 12m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e063ab27ef1146c195e0075896f9d71d : FAILURE in 59m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b1988b5173e4e1ba04dd649e1e5cdd8 : SUCCESS in 41m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/996d5a7dd0814e3b8bb0ef6e331b554b : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":2},{"id":"cde76eb95b833eb7419797216879a80d349df548","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-17 01:39:38.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/30bbcc1e7cde431f9345486229497cd0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b8b65972f7fd409ea974edf073bdce26 : FAILURE in 29m 33s","accounts_in_message":[],"_revision_number":2},{"id":"81279e2083bdf79dc6db713a1927d5a8fff884b5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-17 02:34:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/44/909244/2/ : FAILURE in 1h 08m 10s","accounts_in_message":[],"_revision_number":2},{"id":"11edce9b0a04237fe53125c0a009e746aaff12c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-17 06:58:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/732/.","accounts_in_message":[],"_revision_number":2},{"id":"9998310aa07375f045d599dc72ecdc50bdbaae9a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-17 12:01:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e649c0e9cff411984051f524c5bffa8b485a949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 12:10:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06af4868c68b4379bf4a67185f23654c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/646ccded831e4cf495c58b11fa1ec32c : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dd58ec45a234ba481129233956f6eb0 : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec219352b7c82fa83a69b67b18184d76fa74fbb8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-17 13:07:24.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ac4db86c1c5e4e1b88b60ac6f53a9089\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/80a7b1962eb64846bbe36edfcf5b7087 : FAILURE in 1h 02m 32s","accounts_in_message":[],"_revision_number":2},{"id":"465148238375e187c8ceac06a52f961b4389e9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 13:49:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f97ff88f964abca520d3a98d139f43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1802ab5e660a40dd80c977b57f3e3183 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c6b6d3d6b5d4bc29ab5a261c45eafa0 : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bbb8dfbdff34168a932d746fa655625 : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7948477915d4dbd973204a07cb97bc5 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e996b410a9344964aad7db7ca1348aed : SUCCESS in 10m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a669f5cb610e4f0098778521a72b5764 : SUCCESS in 11m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f463378a29b4799b6406596d3145324 : SUCCESS in 4m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb458535ad3d4a3586e310bcee32e577 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1556cfe9abfe4c60b0931aeeb9960436 : SUCCESS in 5m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a05891b20d29498eac4852d55698e251 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5cc28b78c2f34d95acd3a1ad53915fe0 : SUCCESS in 40m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/16cc863c583248d6afa0de2a6bae807f : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a380b3e259444649feaa27451e3ff06 : SUCCESS in 1h 32m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/60b4d5ce4c814abc8d6503f710d633cf : FAILURE in 1h 28m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/737cf6a884c146b281c97080d2141cf2 : SUCCESS in 1h 42m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a3c22366ed04643a4759c7359e78e97 : SUCCESS in 1h 15m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2630ad1b60bb47ed9ba6fc4eef2451d1 : SUCCESS in 27m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8050852ec350476ca42256c8831e9b6f : SUCCESS in 51m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4003ee7e6bbf4ea5aa2f8cf7aa00093c : SUCCESS in 48m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/233dd8129f22460e9d51926305f2b724 : SUCCESS in 56m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9534214efd63431bae6c74bcc8e79808 : SUCCESS in 36m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da588f796b564713beaeb27fc47ba4ee : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":2},{"id":"03321e35e3adb2252c0f6e23f4210ea052c7a863","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-20 08:35:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/909244/2/20240220101645/NEC-V-ISCSI : SUCCESS in 2h 59m 46s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/909244/2/20240220101645/NEC-V-FC : SUCCESS in 7h 18m 43s","accounts_in_message":[],"_revision_number":2},{"id":"f5346db11b068ba8af2f1f848ba7b5c9840b3411","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-02-21 01:02:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/909244/2/20240221060027/NEC-ISCSI : SUCCESS in 3h 56m 18s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/909244/2/20240221060027/NEC-FC : SUCCESS in 4h 01m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b6c25fcc34c29f2fc418e3271ee054e380aa0870","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-29 11:21:04.000000000","message":"Abandoned\n\nstable/zed branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/zed if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de3b56cc2cd54f8140cfce2221c6ffd807cfe21b","revisions":{"c39befbf4ed2825b4f0abb09016c20299f2535df":{"kind":"REWORK","_number":1,"created":"2024-02-16 18:36:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/44/909244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/909244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/909244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/909244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/909244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/909244/1"}}},"commit":{"parents":[{"commit":"bd2d192c6b76961ec665d922694a21b1f38a0f16","subject":"Revert \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd2d192c6b76961ec665d922694a21b1f38a0f16"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 18:36:47.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c39befbf4ed2825b4f0abb09016c20299f2535df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c39befbf4ed2825b4f0abb09016c20299f2535df"}]},"branch":"refs/heads/stable/zed"},"de3b56cc2cd54f8140cfce2221c6ffd807cfe21b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-16 23:23:01.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/44/909244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/909244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/909244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/909244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/909244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/909244/2"}}},"commit":{"parents":[{"commit":"bd2d192c6b76961ec665d922694a21b1f38a0f16","subject":"Revert \"Driver assisted migration on retype when it\u0027s safe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd2d192c6b76961ec665d922694a21b1f38a0f16"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 23:23:01.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n(cherry picked from commit 1e52602646c8b7cdc616659777563847fe53ada0)\n(cherry picked from commit 22e6866f66748df3dbf347d48fc4b6f55ded847b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de3b56cc2cd54f8140cfce2221c6ffd807cfe21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de3b56cc2cd54f8140cfce2221c6ffd807cfe21b"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
