)]}'
{"id":"openstack%2Fcinder~909245","triplet_id":"openstack%2Fcinder~unmaintained%2Fyoga~Idec29ce596438a7969ea32d48bacfede84f5b0b6","project":"openstack/cinder","branch":"unmaintained/yoga","topic":"fix-cinder-backup-tls-swift-unmaintained/yoga","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-12-06 11:49:44.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-12-06 11:49:44.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-12-06 11:49:44.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-12-06 11:49:44.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-06 11:49:44.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-07 14:59:39.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-12-06 11:49:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idec29ce596438a7969ea32d48bacfede84f5b0b6","subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","status":"MERGED","created":"2024-02-16 18:39:25.000000000","updated":"2024-12-06 11:50:50.000000000","submitted":"2024-12-06 11:49:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909245-fix-cinder-backup-tls-swift-unmaintained/yoga","cherry_pick_of_change":907494,"cherry_pick_of_patch_set":1,"meta_rev_id":"cd366f231dbdf76d715e2d34e9c2bbe24ae8a890","_number":909245,"virtual_id_number":909245,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-06 11:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-12-06 10:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-11-06 15:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-12-06 10:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 18:40:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-02-16 18:40:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-02-16 18:40:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-16 18:51:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 19:01:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-16 19:13:48.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-02-16 22:22:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 23:56:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-17 02:05:43.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-17 11:33:03.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-02 20:28:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-10-02 20:28:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-11-06 14:53:57.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-11-06 15:05:05.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-11-06 23:23:19.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-07 14:59:39.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REMOVED"}],"messages":[{"id":"085e25624b174ab253d52a981d741ff6491c9dcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 18:39:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6836927eb83a1ad2326eee3caec40cc3a02a8b2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:51:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daeb5b518c354477987f335513f70458\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb3154660d2c4e519ec5ae0fc5b288d2 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ec48776a4eb4d099b0245eb57f627d6 : SUCCESS in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bf408567d6931ab629a2753227a28681a8b4c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:01:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/662e5cd59b07474e90affae1bde0dbbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70b415314bb4bc59e3c3b71d51c8be6 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a733e7de5ba4e96a8c85569acb4083e : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3101c2f297f49fdbd73fcd5a2d27d9b : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c057d3e317a46b6b44415b21617a9e6 : SUCCESS in 9m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9b8ca99939114af7beae873b0dbe38f6 : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1ddfd0d0d83b4827aabbca984270a5ae : FAILURE in 4m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7eba65dbf11c4b358db61abc7f9861ae : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43e8c7aa79a74144885eb81b618292ed : SUCCESS in 9m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f148813d57cc47258eacb51cf405221b : SUCCESS in 8m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7be9f901cdd745d399c6fe35a214fadc : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95952cddbc6cfff46ef15d63c26ff37467929aba","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-16 19:13:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/1/upstream-check/cinder-tempest-plugin-ontap-fcp/a791998/ : RETRY_LIMIT in 8m 02s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/fbee336/ : RETRY_LIMIT in 8m 48s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/1/upstream-check/cinder-tempest-plugin-ontap-nfs/35408d5/ : RETRY_LIMIT in 8m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9fb0a58f30304f9ce96afa39be90e5a879abcb98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 21:29:40.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9faa08a49361f25e5ddea8ef2ad81df3ead3ba03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:47:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd10823d2066473fa9a3877042af0d1c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/231f9520eb4a4b3389e051dc170c36cc : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04391bb6a02e4096992011440ea568d9 : SUCCESS in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4b64e68f4cf57cf5ff7d457ea2e8602ba80d51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 21:53:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c6109f959a4765a06b48b9cb308fee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54544e8f9ecf416fa4823d90eb6a53c8 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ce1669efa89470095a1d629c4dee37d : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4095cb5e67748bd99f1298c6f82fe71 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c347903ff548d79201a43e6fc82ae6 : SUCCESS in 11m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df8104ae53c548839e4b2e335e64b89e : SUCCESS in 14m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/199d3d5f4ec543fca0a12b3c39389734 : FAILURE in 4m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e31a1c8926084dfd92a8a2dad62cf3e2 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ceb0d60423954d0791c0021584aebce5 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6f3e04a6b5154444b78ce0d43fd92ce8 : SUCCESS in 6m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/99817aa8a5a44fe9ba6db2b7ffd57a04 : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ab413fc43fb66021d895962e0932d15e110b101","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 22:22:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"be1de9c601f7b4f87c342ed443a6812d0db26028","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-16 22:50:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/2/upstream-check/cinder-tempest-plugin-ontap-fcp/5dc4270/ : FAILURE in 47m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/a55df37/ : FAILURE in 55m 53s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/2/upstream-check/cinder-tempest-plugin-ontap-nfs/20fb531/ : FAILURE in 30m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4bc091860956da8325a9b60ed6554502fa76bc3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:24:52.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3e703a0d8be191d252d40aadabb02de16735f11d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 23:24:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"7b94d8884bc521e2539d69deb8b7bf03367b7ac5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:37:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0356e64d209c45a69e05039a0056bcd7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1827e8b054d449a39283875bd5642247 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e43cb62286b40f49eeba191707817ba : SUCCESS in 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f2dd8934229110c586b67343b27926f5dbb51fe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-16 23:39:42.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"590e8071b843756bd5443bc0577e7a7ebe151858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:45:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0e6934264c046e99365e98c64cd4ba1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e6170fe4b346429d9d2c7279f646fb : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30eea98518184820b0f45aa6804309b4 : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b4cbc2f0ab048a8b6117f9a0ecadd7d : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb352f897ffe4112a61b8b7e902e480c : SUCCESS in 10m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e53abae627484d82b6ec1a3dcc51c8dd : SUCCESS in 13m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b1d2f25377d84dac8c7cb06cce525afa : FAILURE in 5m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c2ba5cc9922431dbeca5921303c0d82 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43e7e729c2cb41ae860939520a8c8a7a : SUCCESS in 10m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17738b8882724bcc884b672e4efdfb19 : SUCCESS in 6m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6da1d23f06ae4fffb8bed62f2a27034f : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad639f63130f2a30b794c26b3e34aec7ca97dda6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-16 23:56:01.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"f3bfe2ddfccc0a88bdbcbbf455badc161b1685ae","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-17 01:25:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/3/upstream-check/cinder-tempest-plugin-ontap-fcp/a7d9bf6/ : FAILURE in 1h 00m 26s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/3b2da00/ : FAILURE in 1h 17m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/909245/3/upstream-check/cinder-tempest-plugin-ontap-nfs/a791594/ : FAILURE in 1h 17m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"d28018d9da6d7deec2fa4aefae5b68f0044af291","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-17 02:05:43.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a19d2bb622a848098b78f0aa7e6da837\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/6273b8fb57954c49b084791776b9dc04 : POST_FAILURE in 24m 30s","accounts_in_message":[],"_revision_number":3},{"id":"42a05a16036049a3e112a35fa489856b03552d42","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-17 09:42:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909245 : FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9022fa2b52a0f69e780433a21acfebed748545ea","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-17 11:33:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/45/909245/3/ : RETRY_LIMIT in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"7d418ed253934c9a94c7026e6049b03aac6af02c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 20:28:18.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"27ad0a32a2634bcffa0bd54a751227edc9b9a039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-06 14:53:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e6dc6f43c64b3be9c5b2064c61a68da9cea5e377","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-06 14:53:57.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/873/","accounts_in_message":[],"_revision_number":4},{"id":"b6f067d7ea2023070779efc7fe6fc976ebf028e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-06 15:00:27.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/909245/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/909245/4/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"0f26f2269b849d9f84628e29dcf6a382897dc3eb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-06 15:05:05.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8240840b120f9b13523f924e8dac3f62e8ca0ad1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-06 15:07:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0bba3a0009925267ca46672c3f0d6fbec23d4ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 15:46:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5445279fbe340909e92dfed31972708\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a208a62e4bf495d88f8450def4eb5c3 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e3467062e2f49f8bc210d17c931be65 : SUCCESS in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7e41071f37e4067ba777b79ab90777c : FAILURE in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7d07a9928f4e29bb565302fb996933 : FAILURE in 4m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/41cf654e40044259a0a20e23084d5bf9 : FAILURE in 23m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73be927fd33644cb974d7c68251252f9 : FAILURE in 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cfa8f9684db43d7b65f49ed4272318a : SUCCESS in 27m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c34e9c26de443dc8f92b91ffebf7637 : SUCCESS in 14m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bc433ca99fbd46ffbccde37ad9c6a03a : FAILURE in 4m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/16d8d5347c474d70bf32fc384174be17 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8015cb679424495484ccd2d6ff14a485 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1674c412f4de40d2b3183ae9e9d043d0 : FAILURE in 5m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eab5b8febbd14c8abd7c4e3453472f8f : FAILURE in 36m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ae1dbed091e4121b897511216a21686 : SUCCESS in 4m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c331bfc7288a4fbd8acd1ef6ec084a5c : FAILURE in 32m 43s (non-voting)\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f6a0088cb95b44a180a7ab44f04d793a : FAILURE in 35m 23s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/493086e5858b40e9aafb480f0ef494dc : FAILURE in 40m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90212cfaee554da0bd80231fe51fbb35 : FAILURE in 35m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7e90234b753e487185e2f84f08dcc3e7 : FAILURE in 17m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/012c873710374de08a2447ae709d5851 : FAILURE in 25m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19d27204bb844bcd8d5b16f55b95a673 : FAILURE in 38m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de2c8c7322af46749e1e23cfe0720b1d : FAILURE in 16m 40s","accounts_in_message":[],"_revision_number":4},{"id":"e0a1e059797aa631481e5c81b0bcfd7cbba48341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 15:50:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e196234ea7404b5bb41606239eedadb2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6969f8af23284811ba0746eb94d6b9f2 : FAILURE in 28m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c36597f7a9e4a52a85440c0c6d6ff71 : FAILURE in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eb5647943db8a2f51d5fd245bc6c3a33dd7b79f8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-11-06 23:23:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/ad58056290884dc7a2830e6a3486a757\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/dd0fd1e7e2564305bb3937240af3d936 : FAILURE in 7m 51s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/47d69b49457e469a96589aea94aeb375 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":4},{"id":"3c19001f258c25f4521b60cebe8cead02eba229f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-07 10:05:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e3ed799db0eb1d96a896d1866e570c20fc3f116f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 10:30:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c22b85737e8465f9dee5079ae3fe21d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f278471341a24cb1b055439c3382d5d8 : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d9393c855ce4823b236756b377445f9 : FAILURE in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a4b8f1253f4cc6b34d5c4b8cf87a98f884e27a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 11:54:21.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50d5954bcecb40d3ab6667377236ccfe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ed14f8f5554a3da2530e6dfbc2a770 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c743bfe03284cd9aab08796edcfab90 : SUCCESS in 17m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77992ddf41dc4d82bfcadf13b7b3dd75 : FAILURE in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a8307d615348fd9315bae6ef358c86 : FAILURE in 4m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/975bcb98e6df46e0844781703d8d9874 : FAILURE in 23m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c2768ce854a4a9aa3e53b06126afc39 : SUCCESS in 1h 09m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e629ef8ebf3848bf864c251e80885ccd : SUCCESS in 44m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21aeeeb569024becba72cf65c5ccffc5 : SUCCESS in 30m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2de6686e4eec43f58752b31252885dea : FAILURE in 3m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34f784c810b146f9b8b9ae5cfc7690a8 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b88f552beac4ed99403563c0ab195e6 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b410fe9e0dd4de58896f31a795dfab5 : FAILURE in 3m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25b11621961f4a4e8294501e1189466e : FAILURE in 53m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/677087c68afb44ca8e6480b77619e3d3 : SUCCESS in 4m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52ba60967cf446a5bcc02f1d8aeb675a : FAILURE in 1h 19m 30s (non-voting)\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb427a9aa42d4c0089879a55c982e2f5 : FAILURE in 1h 42m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1ccef4deffb4aad84bd7996e0043137 : SUCCESS in 1h 22m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8fdb3e09a630477ab49ff635977f8e1c : SUCCESS in 1h 17m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a097cba096e04b8a951019a791c51674 : SUCCESS in 39m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34228f54ddb24f148f35eade2f03620e : SUCCESS in 1h 14m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d30267a7cca3450f98707aba3f942ca3 : SUCCESS in 1h 37m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dcc87530b3b4ec7a9bb3db3f8c612dc : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":4},{"id":"6e79c24ef017e86c05e84e22f220433b61387bcc","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-07 14:59:39.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_32553\u003e.","accounts_in_message":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"_revision_number":4},{"id":"74a1e2727297f0dd7b4803f8ff40455b4c8479a7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-20 13:07:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"499e85fd2dc9d28f39a62650b6bcdc64f9b86a25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 13:24:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5453c0c0566d4faea01e781d1cb2e032\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e326e9b1d04456ca37d31f4b67d2998 : SUCCESS in 14m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e9f160edad9400aa65d22773ab022b1 : FAILURE in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c100f0ff956ea90d766f653076af33aba148b873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:54:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/373dd8e5553b4920aed51e748539e8f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d23ff607eb84a3cb52f97c1d4368af9 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5dd69b669714e618a838d6d97fc492e : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1d606adffee4b368504008aa6cb8ff9 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0503870aba450590e0f11941eee592 : FAILURE in 4m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/9c19212d3f6d4b3cb6f29eeb5b33c086 : SUCCESS in 1h 03m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62faa28d023a453380c59a5d569e65f8 : SUCCESS in 52m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bed6d7163a9841c2bce24fa4887728eb : SUCCESS in 36m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7028afe870f745d59d2d4848412893db : SUCCESS in 12m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dea394fd36f94a4d9682d98d45805731 : FAILURE in 4m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76753214aebf49599057231e6bc99a84 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b386929795cb4bf2ad92116245ed039e : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8d95bb8ff92645f3bd62a7dd14fa4fea : FAILURE in 5m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28d892bf2b684311ab52e5a9a1b463ed : FAILURE in 7m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/877288097d6740b5bcba939a24a42bd9 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54f4d5a945144e42b3d0fded66bf7199 : FAILURE in 1h 21m 55s (non-voting)\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80d0b5ae4b584b2ab9ad64b6b5c18c9c : FAILURE in 1h 39m 40s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dacaa3ab4a564547a7dd4d369d538976 : SUCCESS in 1h 04m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/967be7babdca45368a24626e12acc099 : SUCCESS in 1h 16m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6d7101a9838d4ef0a77a28cf4efad6bd : SUCCESS in 26m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9936ca26b9714a6db7c7f26aa7058008 : SUCCESS in 1h 15m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/786d5e894bab4f76aab39a27c1c73562 : SUCCESS in 1h 27m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0946f2af9cb44cb978b00a98069cb1f : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":4},{"id":"42ad121837d9530368bd98a381470e9530029012","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-04 12:58:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11ea51ae6c0fe5ecbfb63213eb457020bd1fdc5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 13:18:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc67ecf24d2c447f8f637215e02e3926\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2e03fb6e05d47d39baf143ffc31dbd8 : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7ab8508d73b43e59796ba1fe08a5d5e : FAILURE in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e477933442856759622a0bdd8d0def0110ea0a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:04:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9290671606e34f9592300d3b07630c76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b524f5ae83ee4fd7b1519c537f816a19 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fcf37b9e2694fe2815bd7ffb72de178 : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc411dbea5224f8d94447bc1480776c0 : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a9e8d00d9c459faabb244a2c7b3565 : FAILURE in 4m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b3067455030d42b497f7340cbac306f2 : SUCCESS in 1h 03m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5561eacc882644bfa20dada743852158 : SUCCESS in 1h 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c826b2177db3474baf0372d895ca5ca3 : SUCCESS in 37m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/965a79f9f5d240cdad14177d40fd5e24 : SUCCESS in 9m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/607c0512a22b41a8b892378cb9648be1 : FAILURE in 5m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86064625e8f94451aa37389b85ee8e00 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fde9d6b662547f1bd41e43a24b0f65d : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f87dc7c190fe46bebaf6e5091ed2d88e : FAILURE in 4m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb25a2a54d894bfb8ba2d7558ad760c7 : FAILURE in 17m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a69be754a2341ef99fb541a43c03c15 : SUCCESS in 4m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/868cad87e44f430cbcbb1e980f0bc75c : TIMED_OUT in 2h 03m 37s (non-voting)\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81bea9851b07453bb503b0500e0990b5 : FAILURE in 1h 23m 24s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d1cb82a278294e3dbbf6e4aa6ae20b31 : SUCCESS in 1h 05m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0a51da74e80408dbbe15f50ac75411f : SUCCESS in 1h 15m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe44479891584600b0e2533eb42e327d : SUCCESS in 45m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1552b4b167944e3b929dca7cf5699577 : SUCCESS in 1h 06m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e0740f488f904d5e917a2efc5226c71c : SUCCESS in 1h 09m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8507d66e330a49738ce282b43995b3ad : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4e8404d75d160405a5122f26022a9690de086472","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-06 08:54:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5f3e92da1def3cfe6ed09715b49086955fc431f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 09:15:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d05c2e66f6e41c7b12182cd9e8d525b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/727c4841711746f1a8978608a97b5ae8 : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a365926e907496882aa6c8540705fbb : SUCCESS in 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"50ab9d0436ff9257ebda29ce934f727d3cb0d873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 10:22:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b46dd51b7a0486e923895e2aece84b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b0f431af8541c693c45e2d799d55a2 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0913a1216df64638af1a1db07370251f : SUCCESS in 11m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8b9c0f140fa45c7bdeaafa76770bd3e : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118622063585499bb0455d165ccb887e : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/a32d563ab46d4c8c87191ced09b9ec48 : SUCCESS in 55m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d78aecca89e64495b397faacbe1d4f8a : SUCCESS in 56m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5640ad2d97d4be3939e9edc9852a8d7 : SUCCESS in 48m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1b428960b28421ea22106774ce97a3d : SUCCESS in 13m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f1fab1a580d4beabcbcc97f4243a27a : FAILURE in 7m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79ea98c3183843048a6b58b2c8669bc6 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d3eb08f383443519b5d7fc54c397972 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8843755f85b04c2a8c5f39b440a56f54 : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/393e454cb5874ec191ba70282cdb1277 : FAILURE in 9m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0727e80a63f3461db7fc20887ae6354d : SUCCESS in 4m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54154f77f6364b4ca9b97b9a08553259 : FAILURE in 1h 01m 34s (non-voting)\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b65f9ca2416e436ea1584c93059cd43b : FAILURE in 1h 22m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80beea1184de45c68ae7822028b863da : SUCCESS in 51m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a0b6e70445734731b25e1f397472ca0c : SUCCESS in 1h 16m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/913e5800be04456eacc807fabe64b17a : SUCCESS in 46m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a0c5928ebc44f8aab9b4627c58108ac : SUCCESS in 41m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28dbe40c7d5d427797da8dbc6b4f829e : SUCCESS in 1h 01m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ffce7dabc674588b0bbdb73913ca3e3 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":4},{"id":"708b89b14bdcecc326884d607429da0d718fe745","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-06 10:35:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"71a48c7a29a29abfcaea3fe33c0c26a7ff491e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 10:37:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7f07d4a4400e6a927c3d448e77b70cdb154ff910","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-12-06 10:39:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/38cb7f035ca0458dbb463ab17b8662a7 : RETRY_LIMIT in 5m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"525fc4d861657771ff02b9a023331e10ac3d3b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 11:49:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d0bfd1df594a9fb697defba6113be8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e409763d725b488d892b5e00e0b954c5 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54517e60cc0b47e6b6f57de36585c60a : SUCCESS in 13m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd0dbbe72e50400f95f0866ef944c0e6 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763a0f71b4b645318ac5b180495f4254 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/37f572490ccd472f8fce44c1bf6e9646 : SUCCESS in 46m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/320950751a8643a5aa9b816f271a8631 : SUCCESS in 1h 09m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67ac6b7dee3248b585ff678ff0d4c826 : SUCCESS in 36m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce7082e6b88c4417b38b0501fd6af901 : SUCCESS in 1h 01m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/41f67049d6634513be8146eb8bf59e13 : FAILURE in 54m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31fec082712d4317ae23b6e8012f2a30 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":4},{"id":"48552013243e5f57b7e5960cd49cc574fdf3473f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 11:49:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cd366f231dbdf76d715e2d34e9c2bbe24ae8a890","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 11:50:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ebbe4203e5f42f08a8c11ccbb269630\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4209c8640c814be7baa6fd98f39cf2a8 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9a59236e72fdb52315afb6767cdb6fc0a0b76145","revisions":{"6f181f79617b37fec6b8c389f61251f21285cad7":{"kind":"REWORK","_number":1,"created":"2024-02-16 18:39:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/909245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/45/909245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/45/909245/1"}}},"commit":{"parents":[{"commit":"38f84a4efbb513235191e14cfbef58749f8b5d77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38f84a4efbb513235191e14cfbef58749f8b5d77"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 18:39:25.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f181f79617b37fec6b8c389f61251f21285cad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f181f79617b37fec6b8c389f61251f21285cad7"}]},"branch":"refs/heads/unmaintained/yoga"},"c66384084d48a0428c30ab99fc9d8f48f335d226":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-16 21:29:40.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/909245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/45/909245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/45/909245/2"}}},"commit":{"parents":[{"commit":"38f84a4efbb513235191e14cfbef58749f8b5d77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38f84a4efbb513235191e14cfbef58749f8b5d77"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 21:29:40.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n(cherry picked from commit 7abb15d226a1869d04d01c6f5a0201cc1b077e5a)\n(cherry picked from commit 3403639ba58c27a741e4855c507a0680f6a55a4e)\n(cherry picked from commit 03fb86e951bbee36ff5e2703ecd67728fe2ed6f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c66384084d48a0428c30ab99fc9d8f48f335d226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c66384084d48a0428c30ab99fc9d8f48f335d226"}]},"branch":"refs/heads/unmaintained/yoga","description":"Edit commit message"},"f6ff8ab404e54b5e2b5f41c3bed0c53f8f2b8598":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-16 23:24:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/909245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/45/909245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/45/909245/3"}}},"commit":{"parents":[{"commit":"38f84a4efbb513235191e14cfbef58749f8b5d77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38f84a4efbb513235191e14cfbef58749f8b5d77"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-16 23:24:52.000000000","tz":0},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n(cherry picked from commit 1e52602646c8b7cdc616659777563847fe53ada0)\n(cherry picked from commit 22e6866f66748df3dbf347d48fc4b6f55ded847b)\n(cherry picked from commit de3b56cc2cd54f8140cfce2221c6ffd807cfe21b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6ff8ab404e54b5e2b5f41c3bed0c53f8f2b8598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6ff8ab404e54b5e2b5f41c3bed0c53f8f2b8598"}]},"branch":"refs/heads/unmaintained/yoga","description":"Edit commit message"},"9a59236e72fdb52315afb6767cdb6fc0a0b76145":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2024-11-06 14:53:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/45/909245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/45/909245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/909245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/45/909245/4"}}},"commit":{"parents":[{"commit":"d670b60aa8b989064891b0876c0d1c6f9465b0f2","subject":"Merge \"Dell PowerFlex: Additionnal params for enabling self signed certificates\" into unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d670b60aa8b989064891b0876c0d1c6f9465b0f2"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-02-01 19:59:07.000000000","tz":180},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-11-06 14:53:11.000000000","tz":60},"subject":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled","message":"Fix \u0027cinder-backup\u0027 service when Swift with TLS enabled\n\nThe \u0027backup_swift_ca_cert_file\u0027 [1] setting is ignored by the Swift\nbackup driver so \u0027cinder-backup\u0027 service will appear \"down\" early on\nstartup when TLS backend used. This change fixes the issue.\n\n1. https://docs.openstack.org/cinder/latest/configuration/block-storage/backup/swift-backup-driver.html\n\nCloses-Bug: #2051985\nChange-Id: Idec29ce596438a7969ea32d48bacfede84f5b0b6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit e99afcafe8d33624b9a695b1f7576ad0e8e5354f)\n(cherry picked from commit 1e52602646c8b7cdc616659777563847fe53ada0)\n(cherry picked from commit 22e6866f66748df3dbf347d48fc4b6f55ded847b)\n(cherry picked from commit 33a166e4ea83195d71c9bedacb56f16dd3c5af0c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a59236e72fdb52315afb6767cdb6fc0a0b76145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a59236e72fdb52315afb6767cdb6fc0a0b76145"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
