)]}'
{"id":"openstack%2Fcinder~909251","triplet_id":"openstack%2Fcinder~stable%2F2023.1~I0d266a57f68221a3b1740a7376e152bb64cac729","project":"openstack/cinder","branch":"stable/2023.1","topic":"bug/1990136-stable/2023.1","attention_set":{},"removed_from_attention_set":{"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-03-26 14:24:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d266a57f68221a3b1740a7376e152bb64cac729","subject":"Dell PowerFlex: Additionnal params for enabling self signed certificates","status":"MERGED","created":"2024-02-19 08:40:57.000000000","updated":"2024-03-26 14:26:13.000000000","submitted":"2024-03-26 14:24:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909251-bug/1990136-stable/2023.1","cherry_pick_of_change":858370,"cherry_pick_of_patch_set":5,"meta_rev_id":"ec60c207d8b8a75b481c071be2664568f028e08f","_number":909251,"virtual_id_number":909251,"owner":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 14:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-03-26 12:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-02-21 12:36:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":1,"date":"2024-02-19 11:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-21 15:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-03-26 12:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}],"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-19 08:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-19 09:19:42.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2024-02-19 09:26:08.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-19 09:28:29.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2024-02-19 10:16:03.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-19 10:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-19 11:08:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-19 11:46:18.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2024-02-19 15:13:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-02-21 15:07:22.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-03-26 12:26:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"a993260b56821f1c31b57c49cdd5930b175375c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-02-19 08:40:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8aa37fe41466d4e63c4ce90d395a457e95f27c45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 08:57:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fae9c0d8b54e09813458a5d07d1551\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b64d569dc6fd410683ed2d421498bafd : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3e492de2bfc494788b99a1ee4d8c655 : SUCCESS in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a99b8519a1ea114b86572bb1be4d68ba8d37ec43","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-02-19 09:19:42.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e85de675711e46a7b1cd14ae4fa3fd89\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/bae5c72ee7e7434eb05e8183323f1958 : POST_FAILURE in 33m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2319e9ce3bc0ca654914e6f29f38f6eaf5e12887","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-02-19 09:26:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/909251 : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"611773645558251332e6bd6ed2872b4b04e6b08a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-19 09:28:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909251 : FAILURE in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e2071f9d6c979295a0ee31b99d0f4340e11f5a5a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-02-19 10:16:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/51/909251/1/ : FAILURE in 1h 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d5fe2e6426e18f4a5e0a54fe747b0d4d96538d9d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-02-19 10:23:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/909251 : FAILURE in 51m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d8e564750daf8e1d32a784abd199f3f08d19e786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 10:52:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b106aa9364a84926939414880aa88bad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa2840ca7e94eddb30d1ba57972fee1 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/131070d7dc214bf983707ce96ef35539 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095098b4608e4d27bd535ac2d7d87bc4 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/ae0efb4dae0b4a659a7e50b39e9cee11 : SUCCESS in 56m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5056d4c8b0ea42a7a0d84b95b4c54d5c : FAILURE in 42m 04s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a6b8a0b9804447ba1425812801c647d : SUCCESS in 1h 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c879624bd5104b00b182c997c2e2d4e7 : SUCCESS in 1h 00m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cda3969d20474041aeb399f8c512952e : SUCCESS in 9m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6fe666cb9c484d699be05ab745204b57 : SUCCESS in 15m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/816d4956cd5f442496c095acfb5e3349 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1b96cafdc304bd79d5a11e175f7fdb6 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8326d095db0e4853a6ef7677ade266df : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0059be012bdc44dfb35fa9b470b7b426 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/176d576210a84a0fb9d276853d9ef152 : SUCCESS in 47m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db763c409df9461ab421d5c18e021970 : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db66950ef51c4b3a829c4ee7d3a5c451 : SUCCESS in 1h 32m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f32bea7fd5bd4d13b27489055bc92673 : SUCCESS in 2h 01m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b23ab8a8ace14a92875abcbb043e949a : SUCCESS in 1h 37m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0329e654ea954a24a37d5a31c2e95442 : SUCCESS in 55m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71405cb589f647d2820922fbcb52de9c : SUCCESS in 1h 05m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38449a6f65af4b69a41df7c306f5edcf : SUCCESS in 30m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7fcaced85d5b4263902c62373af505d7 : SUCCESS in 1h 05m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3a9c53790184116b373e59abde31470 : SUCCESS in 1h 10m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb5cffcb8baa4bac9a2d2864f44dd918 : SUCCESS in 43m 05s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3241c6f3094faba4e20e3f08b63767 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e86bf62869429eec34110a32ea7e010ef74ba915","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-19 11:08:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-909251-1/FC/ : SUCCESS in 2h 26m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-909251-1/iSCSI/ : SUCCESS in 1h 47m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"cbd93eff8a8ba9f0ee8c09eafc4123098e4eca31","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2024-02-19 11:46:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2a0f4f9bdd39ea928bff7c5d21fe89dfa498222b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-19 15:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/742/.","accounts_in_message":[],"_revision_number":1},{"id":"ce9624ffbc3480e375d3fd1463f79d32ca5f00b8","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-02-21 12:36:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7842ed9e7e49615c92705d55f5acdd96315fe2f6","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-02-21 15:07:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34aaad6ade2646fcc6e499c6cebb1bbac27fcb16","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-26 12:26:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0306d3314de57574a1428d70c3d1be244849c443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 12:27:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12dd4723d9b0ecb5a066b5e2bdc54d0dee155e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:24:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b345e41e40d74e3a93dbff91ad93b49e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36cee8bda6d24b15a0045a6d918ee42e : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38804f96a6ad407f98a79b5415f0f144 : SUCCESS in 11m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d61fc22c244509955c6d2f2a076fd1 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/d4c135c723e04e468ba615f0f4e5cce5 : SUCCESS in 54m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e937d388c1f422b8d88067cfd58b0f0 : SUCCESS in 1h 09m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39b643726b5b40c38b2b743dbca14eb5 : SUCCESS in 49m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f414d99d133e4b1b8ef147afcb67c024 : SUCCESS in 12m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a0f24c681a44635ae36147240fc59ec : SUCCESS in 1h 11m 46s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/595f4a0e2622428b8081fc2502e1fde1 : SUCCESS in 1h 46m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7333ae01ac147e3b6641e1e280e4999 : SUCCESS in 51m 04s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f54279b3e3c94da982e8b6fcebc33f00 : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":1},{"id":"85ebc8a41fa23fcc23d7bd475019e39a4010b75a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:24:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ec60c207d8b8a75b481c071be2664568f028e08f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:26:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd2e35d1bf74d31970c37902e3cff6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd287553cd24b56aa773e8acc444a8e : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54ee2df4ea4841438afb979933c835b8 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8794f84e77435a4d86d0c08980fff18a80151dc2","revisions":{"8794f84e77435a4d86d0c08980fff18a80151dc2":{"kind":"REWORK","_number":1,"created":"2024-02-19 08:40:57.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/51/909251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/909251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/909251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/909251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/909251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/909251/1"}}},"commit":{"parents":[{"commit":"e848c7cfc75c549b22be2b02a9f8360f5db2b246","subject":"Merge \"Revert \"Driver assisted migration on retype when it\u0027s safe\"\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e848c7cfc75c549b22be2b02a9f8360f5db2b246"}]}],"author":{"name":"happystacker","email":"jeanpierre.roquesalane@dell.com","date":"2022-09-19 06:55:24.000000000","tz":120},"committer":{"name":"Jean Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2024-02-19 08:40:57.000000000","tz":0},"subject":"Dell PowerFlex: Additionnal params for enabling self signed certificates","message":"Dell PowerFlex: Additionnal params for enabling self signed certificates\n\nInitially before the change\nhttps://review.opendev.org/c/openstack/os-brick/+/810419 was merged\nto close the bug https://bugs.launchpad.net/os-brick/+bug/1929223,\nverify_cert was always set to False which can lead to security issues.\nIt has been decided through this change that this option can be set\nto True or False based upon security requirements. This change\nintroduced a regression failure as the value set to the option\nis not part of connection_properties.\n\nThis patch adds additional params during initialization so that it\ncan be carried over os-brick and get adequate REST API response.\n\nCloses-Bug: 1990136\nChange-Id: I0d266a57f68221a3b1740a7376e152bb64cac729\n(cherry picked from commit 82823ace4d714ac10427ea3c6fed320c27b56f7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8794f84e77435a4d86d0c08980fff18a80151dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8794f84e77435a4d86d0c08980fff18a80151dc2"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
